)]}'
{"id":"openstack%2Frequirements~907413","triplet_id":"openstack%2Frequirements~stable%2F2023.2~Ib45efee3d39c3d5b40c09adb2b5ffec5c023dafb","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"Ib45efee3d39c3d5b40c09adb2b5ffec5c023dafb","subject":"update constraint for python-manilaclient to new release 4.6.1","status":"MERGED","created":"2024-02-01 11:38:11.000000000","updated":"2024-02-01 17:22:28.000000000","submitted":"2024-02-01 17:22:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907413-new-release","meta_rev_id":"c9720baf77a6eadcd4173d5f19894649585483f1","_number":907413,"virtual_id_number":907413,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 17:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-01 15:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-01 15:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 13:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 15:44:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1b188f4450ad7f487d1846bb8e63b439510affe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-01 11:38:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba6eb470577e3b6be62c2d619bbbae6fd40aceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 13:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e68c243341f64ab5a728d673eb37007e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61f4a484d0443bdaa500c6452b71fe0 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e2c18b19ac04b27a87725cc31171cff : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef484ff70d6d49a2b6d9da2166d5b640 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9349578a9a1462fa56761bfce18740e : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65bfb173573a4dea896969ffce2e10d9 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0508f9e00da941728424d1c3f8a8e849 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86c7a746881e4728a5a0aadc65fdc2c6 : SUCCESS in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/011331e93e88400bbc0447340b480380 : SUCCESS in 10m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cbdaa2f4e5ec4d07900048674bd2c321 : SUCCESS in 13m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2f3e934b7f894fb292f3636e28e08390 : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/321987dcf27a43928c221d3c3ebfe4f2 : SUCCESS in 14m 43s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/57fad36aaefb4150a0b8ebf2773b89a2 : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/170c0740d27340cab99891897e0ca90c : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/20e8b03aefcf41a1891df0fc589aa266 : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1c52e1b1c03543af99ac2b5c8a7e6f96 : SUCCESS in 8m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aa5ef78c59d44cb6abd719a920f19a0c : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f3b451da9d1f4e8bb6ad62efb6fd16ee : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c1b7b6c04c12482bb43c68f16871fa03 : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/646461e546fd4a7083fc0d606c43f622 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f1eb4ff650e441a91f4222b74010e51 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7f9377c12cb04cbebee5d1fb835ad6df : SUCCESS in 8m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b0578c56563b4919af205f90622801fb : SUCCESS in 11m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d1f4d3c9661043ae9cbfd3f49fe846f0 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/60a220064cdf40bd9bbdec641b221fd8 : SUCCESS in 10m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1c860a1e66e54b0fba630c2615f102e6 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1f0d2b44e4da47ffaca753f80e719110 : SUCCESS in 11m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b9afb9c4425447ccbc465daa3fa4e32e : SUCCESS in 23m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fff8441989f04e988f4b6e1887878bde : SUCCESS in 12m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/134d70ee61f04bf38a0b17613f3835af : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4130f5b468d94e1588b230426cf7e221 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fab6ea92ab934f659698ca773eb28df9 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a430900479cb40f884ad1bde0459f27b : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d53a859670044708b8f7f3effacf32c5 : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/786bece2d85543ad8741d46eae72444c : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/412d6db6de604bf3bd161a3864a26a96 : SUCCESS in 5m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/906e648cc0a04ccc9d5a5f16c764f43c : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d46ea33bdc6444cdadb873f5ac6edd84 : SUCCESS in 24m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90926a0e2ff34d6b80bc4e492ba11206 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b207c5f44215f74b5c5aef4a6e68fe65839f38b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-01 15:44:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccb02e7d56b65d0bd2491945c8c8f283ff668ffe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 15:44:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5af57cd851330d28ad7b603b061c576025156f3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:22:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d3d8f8eb8e44848978e5706d1dd567\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a0cfaa318a4b6db0727bad357a5559 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4976a8a4577b42d98ec2416a078d3224 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b6e7da2556146ccba92fd58d1bbaf34 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1004b385346746d9bb6794cb77151448 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2833f991572f48b8a51707b27e47ad62 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75231b98544a4957aaf95da98fd913d2 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/53e0bfabfb5e4f8cacce61bc935ee06f : SUCCESS in 9m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38170f0c217b482e896e11bf03b84241 : SUCCESS in 15m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c5cda334205b4979a0ef463db605bc25 : SUCCESS in 10m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6237fd12096a452e92bf59abf4627896 : SUCCESS in 12m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/74adfd3b63ba4529a99cf94f69d23bda : SUCCESS in 6m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/215d5aa319054f8991cdd0d0f7ae5796 : SUCCESS in 5m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/229ef8dac9024751886eaf5c73ea8d1e : SUCCESS in 9m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/10a0febde7cb4aa895a98416dcb76f41 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11de6986bc6c4d1ba97caafd96a487e3 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0b3c6d13edab49d28c1b25492e7b3892 : SUCCESS in 8m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2cabcc6dfab9445f84f8b627c3b33256 : SUCCESS in 12m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b76271305f5843acb36e0100639e94f2 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1a24699a96cd4afb993149bfbbba29d7 : SUCCESS in 10m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/00427a7c6fc64680be7f66db9194ead9 : SUCCESS in 24m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e9044a7a24d8417b8333f695f71f2e79 : SUCCESS in 10m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/89f65c7fa4f4426d998dca2094dd3e8b : SUCCESS in 14m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a1b88b51ec574deead15b4ed1f0fddd9 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d1c973102ef43908520b2ca232f3049 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97dd6354dd93423a8cf0179f201017fc : SUCCESS in 27m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c46f44fcdc5d4bdf8429515f8b8fb81a : SUCCESS in 7m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/57ab3753227545b8ab89d430ea0da51e : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/437948ff301c4d7b8acfec34206b0a88 : SUCCESS in 7m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bfa86c7e005e4b1f8c76525e7b9c716f : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82f5b2b4e3ed492a8a4590fcfc2f0719 : SUCCESS in 37m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40680de12d4241a8a44f1dd7593f78c4 : SUCCESS in 1h 32m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c9720baf77a6eadcd4173d5f19894649585483f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:22:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7a5cb98f4c8f5397027e97fa1cb023f0b178928","revisions":{"c7a5cb98f4c8f5397027e97fa1cb023f0b178928":{"kind":"REWORK","_number":1,"created":"2024-02-01 11:38:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/907413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/907413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/907413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/907413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/907413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/907413/1"}}},"commit":{"parents":[{"commit":"09abefa91902f269718403634756020b581f4dd0","subject":"update constraint for neutron-lib to new release 3.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09abefa91902f269718403634756020b581f4dd0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 11:38:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 11:38:10.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.6.1","message":"update constraint for python-manilaclient to new release 4.6.1\n\nmeta: version: 4.6.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I053b73f44c7f6486d5969a5fad99832e5cc4e0c6\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+1: WangYun \u003cwangyun@yovole.com\u003e\nChange-Id: Ib45efee3d39c3d5b40c09adb2b5ffec5c023dafb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a5cb98f4c8f5397027e97fa1cb023f0b178928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a5cb98f4c8f5397027e97fa1cb023f0b178928"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
