)]}'
{"id":"openstack%2Frequirements~907452","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I2bf8f55360695155e497a295f4aea334ecb0d8d5","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I2bf8f55360695155e497a295f4aea334ecb0d8d5","subject":"update constraint for sushy to new release 4.1.7","status":"MERGED","created":"2024-02-01 16:15:48.000000000","updated":"2024-02-01 20:43:04.000000000","submitted":"2024-02-01 20:43:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907452-new-release","meta_rev_id":"e1dccf1059aafe2b7b8fe72addcdf6b4c4815562","_number":907452,"virtual_id_number":907452,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 20:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-01 19:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-01 19:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 16:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 18:19:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 19:21:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f251ef6e99d9e0f5116feb2bd39230279f74c861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-01 16:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd448955894985302cece95e2fadcde9c3311ee1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:23:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628d6077930f4927963ca232efdb5cc9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a345d4362fa74019a64cf567e7d984a8 : FAILURE in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6abafdb9c7c08fd362a816e85304a8171eb37751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 18:19:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065b68b979624bea80df0b2a3528c853\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2ce37659cc47f18f5e6a63f263c10a : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17dcf0b57463455ba4dd777078f12a73 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae48958719bc4014aae0392516f7cc0f : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa0657237f754b3fa1ec59e82cd1a865 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a62bd09b49f4245979b3a20c0e7451b : SUCCESS in 3m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96b923188c004a8b954a25be65fa01ed : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/23c79db88fca4de6adae05a13b6a7043 : SUCCESS in 11m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2769188425bc4603af09a8b5db65c725 : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ef70f72af04e4d59b1ae5d6d4b1b4764 : SUCCESS in 10m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6b7be7e1e84046278baab0c525b67bd4 : SUCCESS in 4m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a93aa621f5e4d37b1d57b54b1e28922 : SUCCESS in 5m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e00953f919a4c47aacf673044ac3e64 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/248961b8e299411b9af76b73fce95338 : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5242f17dfc1b40d19b16e35691117173 : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4da778f08f65456ba0f3598ab55352a4 : SUCCESS in 11m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/63586de8373e4f649e01524e14e056fd : SUCCESS in 5m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c45e9a89dea94580a2b617e266aeab95 : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/08c5fade483f407aa66ddcdb8b380b92 : SUCCESS in 4m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c11923d20f354c0bb2407bc6f91b61c0 : SUCCESS in 20m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/759ec090f0b945039c147ac10b607b4a : SUCCESS in 11m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3923e6c91058486daab9473339a02047 : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04232f30dd2f498dbb89c334a601dd78 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abbf7d0b7b744a089ef14edbbb89b602 : SUCCESS in 19m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cc096ba24ba748de8a4ae1fea198f730 : SUCCESS in 7m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/64b71cee962d4ca28f83c694f7c19e50 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4172dee4b74d4052a065d317d0ecfefa : SUCCESS in 6m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d1e63bde7dd1448781f8899574013b7f : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06de9fec09434f19a4c0bf6afd26eb60 : SUCCESS in 47m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c76c7b6abf94b2ebb788b21e2ce501a : SUCCESS in 1h 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"82a3c0b20b06cd20b35dcada1780c786c62f66c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-01 19:21:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"630ff35ae8d9a06f7c2b4a6e8abfb9f58717352d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:22:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9b81e0b39d676c531b282fda97787db23116cc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 20:43:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed127efbca6447f7a493ba12b3853b69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd811e0ff0d43b580a4589f4098cd35 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f79bdba52d84ad1932c347a16766768 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50100464834d4b40b945d2d62b21e82f : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d68e279858442a58ee40e45f95d541e : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8258cdeb96c14943ab5a3ab38f236c08 : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5274ca68c1c4fc78dc95e142c10717e : SUCCESS in 12m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af953389ce21435d8152b365170b38b3 : SUCCESS in 8m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dbbde53bdc3c472b871606d81e0d65a9 : SUCCESS in 10m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a4410024b9ac4e29a658abb70355dfff : SUCCESS in 4m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1686e0ba33904ecc898485c919e082b4 : SUCCESS in 4m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/17474d7243754de4889adc0f1523394c : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/272ea06c8c4b463b8852e5310c9b5db3 : SUCCESS in 6m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/500da8054ed3419d90251253b3047aba : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/10429f288a694855a9e920381ab7a807 : SUCCESS in 12m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/275a6070c31443d480abf660de82e41e : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b56022eda1a1480bb62558806defc317 : SUCCESS in 7m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4ea186b5396d4363aee1cacc7b69d8ea : SUCCESS in 4m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4a47f94be8f64db08b91b98cdebf5d68 : SUCCESS in 21m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93ba47c0830847eb83faa2cba00dd615 : SUCCESS in 10m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/23c33c80ca4547c9afec4c517be7bd94 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39cf475bc119460bb847868feac3d9b7 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2a49b666a9f4ec78c5ba1ec4f44fe01 : SUCCESS in 19m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/60655d672c3c480ba0c5a46166d79002 : SUCCESS in 5m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f77c510007f544d2b10dbaa869187255 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f989340a0e8b47e99fad3ed3c0c64207 : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bc3b1444ce93473c883fcbce503a4a5c : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40f0dd4545a6492c854ad4f1df97cf8e : SUCCESS in 40m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8279095c0a8b406488d5c3f0a3a173f7 : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e1dccf1059aafe2b7b8fe72addcdf6b4c4815562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 20:43:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7708c793ab67cd579f3576196a3b510e6c226996","revisions":{"7708c793ab67cd579f3576196a3b510e6c226996":{"kind":"REWORK","_number":1,"created":"2024-02-01 16:15:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/907452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/907452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/907452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/907452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/907452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/907452/1"}}},"commit":{"parents":[{"commit":"e2034cf2edd94a5962aa010b5e2a0654dbc61457","subject":"Merge \"update constraint for oslo.utils to new release 4.12.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2034cf2edd94a5962aa010b5e2a0654dbc61457"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 16:15:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-01 16:15:47.000000000","tz":0},"subject":"update constraint for sushy to new release 4.1.7","message":"update constraint for sushy to new release 4.1.7\n\nmeta: version: 4.1.7\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Commit: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Change-Id: Ie3ef85f94982ec408aa7a15a4a3c3d7648e9186e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I2bf8f55360695155e497a295f4aea334ecb0d8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7708c793ab67cd579f3576196a3b510e6c226996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7708c793ab67cd579f3576196a3b510e6c226996"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
