)]}'
{"id":"openstack%2Frequirements~907665","triplet_id":"openstack%2Frequirements~master~I6c31420e9dff496e4296b0d7b8af90a78a45873a","project":"openstack/requirements","branch":"master","topic":"xy","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-12-02 05:07:08.000000000","reason":"Change was abandoned"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-12-02 05:07:08.000000000","reason":"Change was abandoned"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-12-02 05:07:08.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-27 06:13:39.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-12-02 05:07:08.000000000","reason":"Change was abandoned"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-12-02 05:07:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6c31420e9dff496e4296b0d7b8af90a78a45873a","subject":"Fixes applied to automated upper-constraints.txt update:","status":"ABANDONED","created":"2024-02-04 10:29:16.000000000","updated":"2025-12-02 05:07:08.000000000","total_comment_count":36,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d6213a90fa5606bbdd9c16f0dd8ff3db0226f226","_number":907665,"virtual_id_number":907665,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-24 02:30:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"reviewer_updates":[{"updated":"2024-02-04 10:56:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-07 14:30:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-04-04 12:39:10.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2025-04-06 07:52:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-04-07 09:10:11.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-07-02 10:51:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-08-25 13:23:26.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"b8c2b08023e4f35e95a9b80a62f60a041e07206d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-04 10:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3999363e65123828d969d7fc4f18589990566232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 10:56:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff266af6399b4e1a8addaee2bfc1b4c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1890ef72244c75994800aace323061 : FAILURE in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76cb06e6c424b67aea431210668afdf : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/431abd69b37442fe8e1245e14ab4cc82 : FAILURE in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b5047adc8464df8acd09e30e9cdb8b9 : FAILURE in 2m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fcbd58847ee446680432ba31eb80010 : FAILURE in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e584917a18748f1a15774741850027f : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e635af6dba5a46a3a7760c483846370e : FAILURE in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/392ad9ba827d4a7fa780e37f6a10de26 : FAILURE in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/13a70b28f1c24e2295e77e4ca84ddd9a : FAILURE in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3c044d0804b1449db1ecfa9738d0679a : FAILURE in 2m 46s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f46c368a6d1a4537a586dbe933e56176 : FAILURE in 3m 10s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e23d6adfbafc44f185d08b5c636521ea : FAILURE in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/21c06d971b0e429c88f42ec1eede4e03 : FAILURE in 3m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/cb9cade7a33b464db866b66b0f2914ca : FAILURE in 3m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/dfd446e6bccb42b59327a4864649d29a : FAILURE in 2m 50s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7a4fbe9ba4d943f89b511b42d53dcc43 : FAILURE in 2m 48s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5000e55da7494b53ace7d2fc43cb6626 : FAILURE in 3m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1324ac4108a649fd863c95257a088f62 : FAILURE in 2m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9df4cb7b7604c228199de4e465d2750 : FAILURE in 5m 18s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d195f166d71d44179c226d9266623976 : FAILURE in 3m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fc95878f35974ee09539ab822a8c5f60 : FAILURE in 3m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7b120b10bfd149909d39ead194a2b67c : FAILURE in 2m 08s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8668f299943f401abe739f1ffbcae444 : FAILURE in 3m 04s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d8d7226a6d5647c997756b611345ab5f : FAILURE in 3m 42s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0a1f048251b44839a769b6e24a1c14b9 : FAILURE in 2m 58s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/32d9c32c403a46fba68d34076f3e6881 : FAILURE in 2m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/49c230a320c641849ef8a064fe73d664 : FAILURE in 3m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0ce5a1d7fef949cabd9917e1a6049269 : FAILURE in 4m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4757778c0aa949c6a758804f3894c6b3 : FAILURE in 3m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ac982af8f0e45f5819fef3df0e275e7 : FAILURE in 2m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e80aebdf063f4245af2049d224dcd8a6 : FAILURE in 3m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4095a77f812e497f9e0e271edab1d828 : FAILURE in 3m 08s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/04b1ff23b2f9462f90b84d3ca5014107 : FAILURE in 2m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d6e83df1f324ccaa115b24f56ee62f6 : FAILURE in 2m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/eaccbad57cf247259f4c4fa83854a3a9 : FAILURE in 2m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63629b278bba429e9119e867c856e605 : POST_FAILURE in 9m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/452ad3acf26843c8974e97dc0f8e595d : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"853a4360e03738e65a6b19dc9d294938a90998f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-04 14:02:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1f83e5927e3f9f444a891e3569e4a2ae87ccd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 15:19:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2a5b226c6f54a8b8da44f294b94cbf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9a4337ef2e4f008ea9bbdfcfe7706e : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7be0a9d3713467a87d0c0f03141bc3d : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18212f86c2f447f792d57c2b77289443 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdd78c70a87e4673919358065b94d986 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c73a708aab3f456481b65857b8b371bb : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efbe1e190b3948ce9f9666cefb337915 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caf4a0e71b8d4c32996eb13ce145cf28 : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c928aae468a34424ad301b41feb5a9bc : SUCCESS in 12m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7832f890d5384200ab2e7e6dd9659201 : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff5324daa7ed428fb8de6fa61be9332d : FAILURE in 5m 01s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/04726ea2e33c4623b06db3ff245b2290 : SUCCESS in 4m 49s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/93a56df7b63d4fe4be252da7be520a5f : SUCCESS in 4m 39s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b33ef5afaf1f42a38f8d7a72b15b6a39 : SUCCESS in 4m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9daf2e56add349f39b271de87ddcf80c : FAILURE in 5m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/73492f8d86304db9b26fd1ed15d643c7 : SUCCESS in 5m 15s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6ed1ce904b594ea79ec8d06da9941a4d : SUCCESS in 4m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f0ffb33058a3417bac07fced63479437 : SUCCESS in 6m 34s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1b63ba7b9cef4f1a80bbbfbb2963b351 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c54f248c4a7e4c4d9a39be8bd9ebbc5d : SUCCESS in 7m 10s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c7e1794a1bc7447e863bea5f708f2553 : FAILURE in 8m 10s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a77dccbcfb1f4786b63426f27d412514 : FAILURE in 3m 07s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c4db0919b96743788afb1db3882b3c04 : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6e709c4573854f93966181392f8c1a0d : SUCCESS in 8m 20s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7e2c9a2fc8fd461db6d061241cb0ab53 : FAILURE in 5m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/dd48eb3c469e4bd0b57c730ed7925f9e : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a7ad3cb76cdb465c83a6971da0f39ec9 : SUCCESS in 23m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/50cbcd64c2c44094900f0bc18d707383 : SUCCESS in 6m 28s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a5900184576b48e7a6727ceb5a26da45 : FAILURE in 11m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/155f1480ddf142baafe645c79ede6c90 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c026e3ed06f845beae37857cd9ce9d64 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7d903f251c04b318335d3a699a7e9ab : FAILURE in 25m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/9768ac60fa0d4db2ba4f551497781bb8 : SUCCESS in 4m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c730e5baa8da4c39a108e02e383890d5 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8eaa8c1a4a3c4e548a1901f7cf368fb0 : SUCCESS in 6m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8858d2d5696f4f43b7a6d7ac945aad3a : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/205aad803968476b87d163cf1e9eb7d8 : SUCCESS in 23m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30bf0486175447b883137dd6440cbf8e : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"0ff06daa7d646b7e9772166a30c7f6130a69ba77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-04 18:22:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d36c210b447f3732cb50f4ea6f12a869c4391e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 20:14:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/028e1aac87a447c3943b3c1788a311e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b9a8e60d3e449d9f7c06aa5ae9cad1 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d2497ddac74b48a39b4a6cde702cfa : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/217140caf8874ea79cb9657997d4c124 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fe762eb7c2a40fc943999b942c85597 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d697e4d906cf494d9b8ca59bd6e207bf : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/519b60d1d28a4c1db6a7d315a7f6cc91 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e7326d825f64a69bfe3856caa98aa10 : SUCCESS in 7m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/708da0124e0c490cbd4e9bd50c291200 : SUCCESS in 19m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/425bd2875a6a452f91c8d979d9471a0d : SUCCESS in 10m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fa50c28dc9be4aa8a464db28e00131bb : SUCCESS in 16m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/564c4e1d240547c8b60980edf1f464b5 : SUCCESS in 5m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/32e0d3274f394260bfbd9eada69a8f70 : SUCCESS in 4m 24s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6b0d2627479e46a290daec8c8f75af9a : SUCCESS in 4m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7cfbbb5fc53548e1ab37462fec48fcd2 : FAILURE in 5m 46s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/57b380b85134426d9321b7cf51cf2b28 : SUCCESS in 4m 36s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9d8b7db879a24c0386170840f811d328 : SUCCESS in 6m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2172341563ed4d779c2a2ba05ff8c00d : SUCCESS in 8m 09s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c8e4117a40a94c2f9ac558aa94abb422 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fd15c6a31d14551afc87342392045f7 : SUCCESS in 7m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0378807c044d416681217ef7d41ac98e : SUCCESS in 8m 14s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0e67871721dc476ba9f583eaaa544c82 : SUCCESS in 8m 01s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9fe673bf670b46ce935050a70a820618 : SUCCESS in 3m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b7a5e54d90984ec5a229e44b4cc890b9 : SUCCESS in 7m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0cd55376a9fd43758a95fb83e99344c7 : SUCCESS in 5m 27s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/56e4a015c10b450ba9a19001589d4d6d : SUCCESS in 9m 40s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f8a0b21c720b4bc192f197cc1f1efa96 : SUCCESS in 25m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/118192309b1748cfb11c629d395ff7e9 : SUCCESS in 5m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1e5f597065384cf1b9476b5d7238c3d6 : SUCCESS in 10m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c753072c9a21436695e6a8362140616f : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c6205cc94e64ea4aafbf1980f02e44a : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ac0f1bf0eee4f1c918a92d986ad49ad : SUCCESS in 20m 54s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dd5df6b9d5414a67b8e08b4ea083a4e7 : SUCCESS in 4m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/20d85ec5b34f43cbadd4eeb25449448b : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/affceae0604e41068c5c5fef27ac1dae : SUCCESS in 6m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/26c58d5c08f84b6a83ce13e1b009129a : SUCCESS in 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72f6baca05354a3f8d189d0502e0617b : SUCCESS in 19m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/969d8fc7df43475298126ad3c99fea2c : SUCCESS in 1h 44m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d29832944974c1c96ed697c59e846ef07561bbf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-05 12:48:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0bb3bd1988fcc291ecca58956ca1da9647d100d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 14:31:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f1b8cb315b4ded86cb01fd34924d52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136ee6b6b3864b5ab59e7392ce42ec4b : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a54f2007dd64dd1b8ac5464a7abdd53 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1255f1ff87143b5876177cc8cdc1cab : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/912797032bbd45019e0000ed6bd966aa : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cb133cd166945789dd969c7dc23ea5e : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb359e9e056946e495544e8b0bd64405 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8a0717934034af78d211fb793fe93d3 : SUCCESS in 11m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1e75b1050629416f9845adf7d3ffda5a : SUCCESS in 15m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a4708736bf6b4880aaf267e41ca36475 : SUCCESS in 10m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/925c8fdec76f465c9b2c43d311a23bd2 : SUCCESS in 16m 53s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c6e434e2b2b34ad2af68f2d537201b01 : SUCCESS in 6m 20s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3003508e065944059b808abc3e3fb0aa : SUCCESS in 6m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8c5af7feb3b941d595732e9e0215f73c : SUCCESS in 4m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4f40fce4e57d45f594548ffa9e71db5c : SUCCESS in 13m 03s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b0f31bf02b4e47339a382ab648dd5654 : SUCCESS in 5m 14s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fb0e93a7014e4663bfe876169bc93ddd : SUCCESS in 7m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/12d38636833b49a29070805ed271adf9 : SUCCESS in 7m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/50ddd7212ae8431c91dc9d83e91d2064 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/275e4ddab06843bb8c4f3678fe33d812 : SUCCESS in 9m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d205ec1d603f44248864dd94a1c5be4e : SUCCESS in 9m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/44a809aaaca5489aaed9d5baa639bbb0 : SUCCESS in 13m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ed3aeb9c2b7d4cc2b079a6f721d8baa2 : SUCCESS in 4m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3d14d7669fe94f6ebe37715a2a9ec177 : SUCCESS in 9m 15s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ca293ff312f0471db53aa2bbb9402deb : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e5519f388f884ca0ba178f6bbe640b7e : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/596f078daabd4461af43ec4f3b877c6a : SUCCESS in 18m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/90eaaf42a55840099894c704fe4ca0be : SUCCESS in 7m 28s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1c0aea6a5e9d450497728676611195fb : SUCCESS in 10m 30s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/cdee74e242844641ac7e8ad6b20dfc81 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74f2811e734941df805ed40d9415a4b3 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b1f3baa65f04b35864a51fe77304093 : SUCCESS in 19m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/27185089a18e49539cac0db59d5bb340 : SUCCESS in 4m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0327061f7da04ec9a915f3e49c07c63a : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55647f95afeb4924b1ea7646d914c7fb : SUCCESS in 6m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f3b0e542fa654c88978f96cddbde1911 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5530f223a49445aaaf5e66225ec26d2b : SUCCESS in 28m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b096a87a474df1896204446fcb14f9 : SUCCESS in 1h 40m 47s","accounts_in_message":[],"_revision_number":4},{"id":"12418084d411acd741502d1bbd6b402c76240c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-07 06:05:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"87a3d64588e7184c0540f971d369bb23faafbd35","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-07 06:09:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"57c8552decc644595b624672189c6fd68116b093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 07:42:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e8c869ecc74e019a89f69c27387532\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb94fed3b46423c89d6a53ecf133537 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d211e84be9854e2c920fbe6d405469da : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c7aa89c91a24093afc7b327e4a03735 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79b7b3ce69c94a7090fef3a6c49fd650 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4902bfd1b484ab3b144287eb8d3f9dc : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15644fe764884c369e069938cd7ebe2f : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/208b2bbf44de45a494b5b228bf097dea : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/40f22149b9ca4da5b2bee0b2ab15f83c : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c6f538caefb42ba943d4a3d2b41415d : SUCCESS in 8m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fa654ed81e3b405e939801905f672745 : SUCCESS in 16m 47s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/db3f58e7104c4e2398b9c379d37e8cde : SUCCESS in 5m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b646dd0bea1a4e1a97cd3189fce11912 : SUCCESS in 3m 47s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8bd140037fb348f69aeb7804dddc9e5c : SUCCESS in 4m 07s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/882b558b0d91459b8e40933e7d0d8ada : SUCCESS in 10m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d666c248abf342d5aa785365e634941d : SUCCESS in 5m 40s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/19f2787aa52c4336b0f596a9198b95e4 : SUCCESS in 5m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/227221d1d65a41eca9233f2f22cf0295 : SUCCESS in 6m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7f894847d6a5442baa45a3a3fa8b73cf : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0301ac4884af498e899181ae6e3322c6 : SUCCESS in 6m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3749ed094ab84f2d892e122e4fbc6e01 : SUCCESS in 8m 05s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/763462e778a445bca73d3d9dc7caf2ea : SUCCESS in 8m 01s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4167400e142242a8bbe8f7109d183425 : SUCCESS in 5m 28s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0d99821536ac42c1a6e91bf7fce358fd : SUCCESS in 8m 03s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a9abbe83e7cd45668532d40e5bcd9a03 : SUCCESS in 3m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3ba695b681404cd3ba99afbff5cae602 : SUCCESS in 9m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0a06bf59cac14a1e845733851cef173e : SUCCESS in 21m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e65a7fe4c5a4082a8518e36a09a45da : SUCCESS in 6m 51s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9863420b3403458caa715f0ae3699c8e : SUCCESS in 10m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/93f12830de884ac7a30475248393711c : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/896bc2fbe0c24462bd5010fa53ffb277 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/513aa438431d4592b3aa7a944d368d4c : SUCCESS in 22m 05s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ad53b6354d9d4640bad55d536a445a37 : SUCCESS in 4m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b3208ac039fd42fbbf1b6a94ab8eb199 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66b90155bb1b46efba614d0a8749140d : SUCCESS in 6m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f00e308381ea465b8922da442a76aa9a : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16070f25f105454c88733a7301b5e15f : SUCCESS in 20m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41818cf80e2b4ff6806a0f22c52fcfa1 : FAILURE in 1h 27m 59s","accounts_in_message":[],"_revision_number":5},{"id":"7a264827eb44aed2b1a9b8d3b6de23892f6335db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-07 09:35:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee61895bfbdb90aed5521013671dfab6aa74e026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 11:06:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a56d2f19ecac40c5b65938a86e4e85af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e4a1e9a2f54379888b28a833ee26f1 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c51a3bbdf914f4c9fc545ea42317c81 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b9c90348a2b4660a4ffcc2618e2d90d : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e1a072ab1454565916edb4dcb3c9d1c : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed3d81ac8e024ef59e445cc6a976252f : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4af7a626f9464ab2b0c2da67122affdb : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2af2905f553342e785754a7f7f50e24c : SUCCESS in 12m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/80057fcd57f44144be39a5571b07f0dd : SUCCESS in 16m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dc0258f46a7f420e86fe0eabd94f1eb2 : SUCCESS in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1355c6c6301d4b989bff9c60cdd87aba : SUCCESS in 16m 02s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/54dea26fbb894ba9a037cf92641dac9c : SUCCESS in 4m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/86a38d504986475c919d2c6be05aba4d : SUCCESS in 5m 19s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/fe0e2676743342ce95e1d86397e6e4de : SUCCESS in 4m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d534a94f447d46f59bdca741f112ba87 : SUCCESS in 12m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/31a6ea26d2da4fdf9e1a94b016c99534 : SUCCESS in 5m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d8a47038a6544dbbb18d1457544ae6ca : SUCCESS in 6m 32s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2c9aae1c39ad4c59885de0f73d2bee52 : SUCCESS in 7m 09s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a7f1df8cb68944d5bb730b08b528b008 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d38118a82234e508589ad4dd33109f3 : SUCCESS in 9m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/39b709d9116c421ca54334f10a0a147a : SUCCESS in 8m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/487719be1c2f4d3c99cca8c00d8910d1 : SUCCESS in 10m 05s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4dba17814fe84763b09825ac80341f29 : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7588aafd4d6448c3a16c8e24b9e8f667 : SUCCESS in 10m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5877cfd79eba4139a8d7847bad80e968 : SUCCESS in 5m 34s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8126ead226c04473aa41897e20822ad3 : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/af9628969f4b4bd7bc5b0470eb7ec6b5 : SUCCESS in 24m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc67bdbd24884e559a4b9214643a024d : SUCCESS in 6m 38s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a9819bde37b44ae3a457b5beecdc05cc : SUCCESS in 12m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b96a6de6711b44c9a339010527df915a : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22044c381f1649dcad574aad4b2d7ab3 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/515b50faf6fd4fd68ebed0fe81335b53 : SUCCESS in 23m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ef208b5672114754a28e821777cceb83 : SUCCESS in 5m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/36936a7a66ba43d59f5e86a05a6a78fb : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1059fa7aaec4df4b6c6b855ab575e04 : SUCCESS in 7m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b19c9c19484846e5b57d2eb1ee98eac7 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19368f9a58c445ebb0f2ffb94a138508 : SUCCESS in 29m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/622dedf16be34eceae4435860709b09d : SUCCESS in 1h 28m 45s","accounts_in_message":[],"_revision_number":5},{"id":"7a61cf7314615c8a61e9e2977e2cc99db9ea3996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-16 20:43:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"81ef2c60e91e1ef1364409dffb4db7091c77909d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 22:39:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba27e7fca6ac4d74800d049ccd65540f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd73036bfb9b4f77a90d0476d0f17882 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a650105b66431e8585f618177473da : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3e1d1fbea0c41b683592e50c8f07ab2 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1c92749717b469dafe28d9f5e3e6e15 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8bcb4497d1d4eee806037186ff72028 : SUCCESS in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/837c581fe68f4a0ebf327b49c293bb25 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f826b0be7f4847008a243f786551787f : SUCCESS in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/06393691c56b4fb6a4e4b7f7b2fc4fa9 : SUCCESS in 13m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/641225ad535a41019113c44cb1789aad : SUCCESS in 12m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6a9599d98144463ebad9566a9ad3e443 : SUCCESS in 17m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6f02b87d5bf24c988146ab69a3806631 : SUCCESS in 6m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/dbe45b5096904cc5b17382aac767c364 : SUCCESS in 5m 11s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7f5cdd8d27e844cc8735160bdc1487ca : SUCCESS in 3m 37s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6745c36da5ec43d4a486b4098d6b4a45 : SUCCESS in 10m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8c64612e98354237a3293fcee23900e9 : SUCCESS in 4m 41s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b75431f59c7e49fba159b4ff971b4fc5 : SUCCESS in 4m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e5b1cd3cf1d14f7084c483ff49e86360 : FAILURE in 10m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7c020ef764f84764812b5c77f6a3ee23 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea306d021f9940e5a59d05e4b828d94e : SUCCESS in 8m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9b747c63afed40829a09e0129cbfe111 : SUCCESS in 6m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b2d850fcc5fe499d903887a66906f179 : SUCCESS in 14m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/996d75f8f75a48eb832fbe53796495e9 : SUCCESS in 4m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f813fce6e8104c57be76b868634ef795 : FAILURE in 7m 29s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fd6783ad54844810b635ecea4a0f4397 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4ca6ef4ddc884dc6b4a3b9754716386d : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/32b1a4d4c5e74feaa7d06d40108ed500 : SUCCESS in 23m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/62ce7eca4ad14271a54dea1baaba4b7c : SUCCESS in 6m 09s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1df501c085cc4223be64a8aba1ece695 : SUCCESS in 11m 35s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9056c0cefe574151b791d6b197900136 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a3654ffccabc47ef8d63280301007f36 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cb97cd41a1c40e58b6e67d252d69690 : SUCCESS in 21m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/bd18307cfb5b47ff8f0cf841ad5098ad : SUCCESS in 3m 44s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7a3a8fc8494b493fb632abd93f23053b : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8eb2603d7e4462bb73d5e87ba6e5194 : SUCCESS in 6m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/94ed8814075c4340bff237a6ef0a0a9e : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b4f280b40bd42d9bd2edc0a1f9b4dde : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b668dcc32f0a4bc2bc9b396a0b96cda2 : SUCCESS in 1h 46m 17s","accounts_in_message":[],"_revision_number":6},{"id":"2d2fd09439f9c57d3dcc969f9d2fcf0803328d66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-17 09:59:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5ac126b39b9d741049462e689bbfed602d8d7c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 12:02:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/532ccf1ca4ae46e89043e02d517f0bf1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435047037e3b43a99a6dacee5c8e9151 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7daf9a484bd540b58a7cc982ce3ae19e : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22218625cf1b43d78c77e521dce686a7 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b9f3820a2af43668359804d8f39d02c : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5079daca204497c94a4b2c3931ceb0c : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33654f4b4f034592b22d8a1208ed520c : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/16265b7620dc48efb31e4bdec9956905 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/649d6d7dd92e4daaabc0858087c8f23f : SUCCESS in 13m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9edc953e85ed4cfd890e145c04c65deb : SUCCESS in 11m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5241e84adb7f4335a34a710b96692cd4 : SUCCESS in 21m 01s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c79b9c37afd241e0a5fa4abdb63180de : SUCCESS in 6m 33s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e871256e91f54bd0bcb4f576f13ad8de : SUCCESS in 4m 32s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8e09650029c943468178b836a63be73e : SUCCESS in 4m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f084d27019a74c2aad9e0f797caa6d5f : SUCCESS in 8m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/08fb4bbe16394c2b8430b0be76524974 : SUCCESS in 5m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/627860ef445a42f784d4abf211d3d700 : SUCCESS in 5m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8cfb13adad1a4b52b75c6c7744677ef5 : FAILURE in 7m 47s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/033d83b8622749e182cec04b77d4ff50 : FAILURE in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c407e902817645d282763d5bb8c648fc : SUCCESS in 6m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/680005bcd00d488e9b7bc11f50768202 : FAILURE in 6m 10s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6103c00b488d411385dceb5df8b68443 : SUCCESS in 7m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d770d5392d4a40e9ba91eefbbec057d5 : SUCCESS in 3m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c62e3b35106b42b493085c07f19044ea : FAILURE in 6m 03s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ee1caefc4be7413a94578885f56cd4f4 : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/bd98b46a19234d14b2a3fa9d259b2809 : SUCCESS in 10m 53s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8cfad48dcfb84be893fe78a1f965c5e6 : FAILURE in 22m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d7c897a404824488b68ce16e255ddd16 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7f19daf8abf644d390bee65942b066fb : SUCCESS in 11m 27s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e59102bc8fdf40c3953c8bba2b80afb0 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd562bf650c145eaa2d6e100d14fded0 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6fb51562b494abbb1bdad4d2b612fc1 : SUCCESS in 22m 27s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4ccef403327342e9aa691f93c63647cd : SUCCESS in 6m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7f3d78530f144342ae67587520e15b8b : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27e2d6974be54583b62f542771b586f8 : SUCCESS in 6m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a0c4cf386f2f4cdba03ba1ec2f98de2a : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd679ece0d084f76b0c2e80e655328d9 : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea72f9a2929d4b2c93f84263ab572ea2 : SUCCESS in 1h 51m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ea9ab20351cf4ff940ca194f56c19f820b25ec89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-17 13:28:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"3a16fbfeb6dd0362bb02d53a5ef9d78ffc5f6b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-17 13:41:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"efef5fbe3e5c6e766f56804b118996a3131d0078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 15:28:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e3b2085e6ac41bea193cacca7fc1b01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a5997b64d74f7e92bee37bf3d9a6d0 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbe90c1da634553ae0244d4c7d9bda1 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102f0875f1c94260b0c59a45ae03315b : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5307156a8de4fa09001ffe9263bf317 : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8e4cd9143d24a5e9cbec8292b62c7a9 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4c026940651490bbfcdb04d11db2ad7 : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b226972d8b2548559c25562cdd051b5e : SUCCESS in 6m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e7dcf0befe4e43abb5c3854c20ce527e : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f1d1640fea542eb81b4466286335759 : SUCCESS in 6m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b37d2267f5a49e9892426d02487d2ae : SUCCESS in 14m 17s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0280d235ab9d49908e2979e9450cffc3 : SUCCESS in 4m 16s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/68f90e90e252455e83c5b8fa7b25c26e : SUCCESS in 4m 24s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d55d75817d1a40d798df183509d9785a : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2de6732d329343f98d066d5921e081da : SUCCESS in 10m 49s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8cb75c3cae5b49bb9dd31492e7582035 : SUCCESS in 4m 22s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e53a4f09dbf94f129b1bba87a546ad33 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/125bbb835ff4485d839f82b94c99b4d8 : SUCCESS in 7m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7070a99941584c349b4d4aa619c61bf3 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74365798f1fd4184aeb87831b167f425 : SUCCESS in 7m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f04330da0f2b42b5b2708524dcc1d798 : FAILURE in 5m 54s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/007fdd6fe15a42fda33da5dd57c68404 : SUCCESS in 11m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/fb614ef484e64bb5962e0ece3ba844b9 : SUCCESS in 3m 59s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ea1dd2894f094be695873455a86089a8 : SUCCESS in 8m 11s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/960f1b014aa54931a901790e9b2ce0f3 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7e9c0b206e7c49f0aeec9bed5f9993b5 : SUCCESS in 7m 46s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cedf2af7d4804d54b938d7ff2775edfa : SUCCESS in 15m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/afaff1e0eebf4593a5e230a12e5a82e9 : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b4fa3d2208834a1ea9de14c3a33d1be1 : SUCCESS in 11m 42s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d3ceb68d16bb4a68bb2c9baa954dd894 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f786d1a72c664587a6de8d79b1c947a4 : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38e2fbd9f23142daba9e273836d8cbc0 : SUCCESS in 16m 13s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/179d9f4bff72469689882e147b5ed85f : SUCCESS in 3m 56s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a93592bf0fc94f7b819f62d11c52c8f5 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6aa59b0f1154d8793f0214cda41ed7b : SUCCESS in 6m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2623ec6e71a84be6ba77b7d1bc06a56e : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d3ccaa256ef428a98b28739b20b6385 : SUCCESS in 58m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9af6464c66a148cb96a7d81f30c6a8dc : SUCCESS in 1h 36m 59s","accounts_in_message":[],"_revision_number":9},{"id":"545a01c0ae5959dc34fd040f310123dd7437a208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-18 08:52:22.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"1b45de5a4b185df0e8bb7324b9c46cacef2c246d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 10:34:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf2417f505545f3a91947a4b15103db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2197a084bc4fa589daf69d87ca317e : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51873f4de08640afb562d124be040666 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/341b9531aa74496e9c925a18aa4b7a53 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c077954ca1f041929d8f42d0506706f7 : SUCCESS in 2m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc37c66f0b544b26aa6aba75024ed28f : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fa16b16bf914a8088569ec892e67e2d : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c00d965bb814ccda0f3991972ad3903 : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33125f91f0b148b09784304c7e631c2b : SUCCESS in 12m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b86854bcd0984ca09ff046f74ecd52e5 : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/967b40f71ff64a578011aa85dc4769b2 : SUCCESS in 12m 01s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5d47c5f97f9146f28fb37a6baa04a059 : SUCCESS in 4m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/90da782b1cbe453b8582e5c9aec86ea5 : SUCCESS in 5m 05s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4011f790d1a44bf3b352ad1defd5fbfd : SUCCESS in 4m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/81d3440a96b8407faad7309c0d3d915d : SUCCESS in 12m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/71892f9cdff44b4fb2aab6a43580e99a : SUCCESS in 5m 50s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/41825d8b9aa7495392b80cc68a64762c : SUCCESS in 6m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/04db13b5e59d43a7b55754820ba0fe7f : SUCCESS in 9m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/3699229637774ebcb6320e19e8b11279 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10a42e3515c64a068b3bfeb3ebb05770 : SUCCESS in 6m 33s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6242ddd489ec4c9ab59babf37c358f79 : SUCCESS in 6m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ddd19d4a3420432784b96b24b37b02ff : SUCCESS in 9m 09s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/59d31b36b8534977877c26fd826d0fdf : SUCCESS in 3m 55s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2bf265ad4db5466983dab2ea17a4c273 : SUCCESS in 7m 08s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6bf9ef6bc6bb4e0d9b012feba458d6a7 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1fbab02bb9f64cd892b2056656afece5 : SUCCESS in 8m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/36147a8b9a274c20886108c49ce462fa : SUCCESS in 24m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7f13d42fc3c7421eaac87afbf5b4ac0f : SUCCESS in 5m 46s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d812be6f11b5452da4c75f608747dffc : SUCCESS in 9m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/69ca4a3038ef470eb4fe4006275d6438 : SUCCESS in 3m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afa2058709be429f8f9b4b68fc710904 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3207d20e980243cfa762b7d824110dff : SUCCESS in 24m 29s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ab162c8281ed45f286f0e1920b1e3a14 : SUCCESS in 3m 44s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e980684998fa4c5e991db0d726a97d4b : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/134bec4abb5b4bcfb79dcdddf0d7b0a5 : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4520c7f821ba48c08ca166d4623208fd : SUCCESS in 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79c6a6f81121478890c839f0b9153e8b : SUCCESS in 1h 01m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f69c8790502f4cc3b019b3731c5d8d1c : SUCCESS in 1h 35m 56s","accounts_in_message":[],"_revision_number":10},{"id":"ac19ac07d88e2afc988ba6aee574c5f5cdb3f05a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 11:59:34.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"adf6f8540edd37c29575c3f4595e47e503670ea8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 12:02:29.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"47bc9c43a44096651d71294e6b803c34a6d333d7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 12:02:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"bde6a9d7941aae169689bc29ac18978cc301e477","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 13:40:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b9891fd441441a847e6195e57ca49f\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2819ea517e14ff3b7d8ccc40d2905e8 : SUCCESS in 1h 13m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/69a0f464f7a448cfabb9874f0cca9a78 : SUCCESS in 1h 30m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5961c72c1db94dd1bdd13cfcd9d2bf10 : SUCCESS in 1h 06m 12s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/a2640b783e564b548992158c5bebb433 : SUCCESS in 1h 24m 38s","accounts_in_message":[],"_revision_number":11},{"id":"acb07e9777815ce0a165cf62016954797051f7f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 13:42:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae0ec1bbfe8e4648816f6453e699cbd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c615c0abcb447079c0b9a9f2ebb60e2 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3aea5ebfcd84650977ebfc4d219807b : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b89ecddaeed420ca080c330ac211aff : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9ccf5b3f1a1461eb9783582165873a4 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cacfeee1132b4fbfb8097e05b702e646 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf242faff1ea4c5b862415b793c425f1 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7af6a3c6dce423d8f462a19f02d592f : SUCCESS in 10m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5f092cd71454c2eb03a95258833b686 : SUCCESS in 11m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7dd63e64a6ea472b96d17d9dee6c6f7e : SUCCESS in 10m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/614a017f1d0840f2b52651368dd12091 : SUCCESS in 11m 46s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/4a4f2b9b2a0d49f3b4a5ea78655d399a : SUCCESS in 4m 02s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e2adae8dda6a47ec9d6eb0af4a8d5f79 : SUCCESS in 5m 00s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e490f6c6e60d477ca55a88c8724fd9f4 : SUCCESS in 3m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6643fe4acb6a4b9c8459cc5c4ac49ec6 : SUCCESS in 7m 46s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e574ff7c461940e99e0f2fff1556b4a0 : SUCCESS in 4m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8e6d1b9db96248cf98efc0145833d4d7 : SUCCESS in 5m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/145e1734cf9149618cb082decbc5adca : FAILURE in 5m 57s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/234cfd85a59c471096b1465e6ab4481e : SUCCESS in 16m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85becb854800438dafbc79162d161191 : SUCCESS in 7m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4a9dfbc6f04a4e83a615996ed50c87e0 : FAILURE in 8m 54s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e6081d38ed974c37b090baf32116f19d : SUCCESS in 10m 54s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a91dec847b704937a8b618ee47b2740f : FAILURE in 3m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/992da6ba70dd408494f15ca99b568212 : FAILURE in 4m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3cf862cc8cc9457690ab832e96b81f0f : FAILURE in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/36b16377184b45c48f3f951288f242dc : SUCCESS in 9m 30s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/72310aab64044f30a13d9cc562504bb1 : SUCCESS in 23m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71c54e25946040fea99c0b7a6361705e : SUCCESS in 6m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9b4d3b8556ca410e87d6da61ce70fc2f : FAILURE in 11m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c637a3c96d4645e39a0dfff4c167f88f : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2673a6bf5ed242b0b75af4884dfa8414 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc5c826bafbb4272a3aaa4712b0b8bf2 : FAILURE in 22m 42s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1d449720473a4a169ecff268df5487e6 : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b2bee40c93b048e1961f05bef7b96213 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4cc1c00477f4fb391f0295235d8409a : SUCCESS in 6m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c6e4ab7ec59743e7bd5f210513808101 : SUCCESS in 19m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed1486f588cf4de39cdeb5134ed2f77f : SUCCESS in 1h 03m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aef7360c83a54b9a81f7f024b30ce3b6 : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":11},{"id":"bd4c1b608a710e605271d8521485eafc08f4e687","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 14:56:29.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":12},{"id":"9ef902663e8da8e7c46850f7e578d4f063e1386c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:20:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3a480b91477495b93be880c3a07ac11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ebe26415064dc8bd8d8bedc5223bd4 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af10a7a309804793a308461c3e1d6e04 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db26eee6702047df8e3781b07b88b3cd : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/376dd64084ab44229ab703b2aecfa37f : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7551cb989064e6e9d6123e9c0ab190f : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da5b558673c04c968e624c97a96953e7 : SUCCESS in 2m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e13a36dbcd1540b285d4413c686c2afb : SUCCESS in 6m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f4246e00bdd4b519d971c9414da4ab4 : SUCCESS in 10m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/12d12d4a75c54ab8bd20b6df5824b172 : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2ed4304f030b4d0f85706c44cba617b0 : SUCCESS in 17m 36s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/4e9c2ec0db1a443a87400622db30b64c : SUCCESS in 4m 43s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b72f69c412714fbd98ecccf478a34d16 : SUCCESS in 3m 51s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8751b4a7d7e44f8eab9c75c436023df3 : SUCCESS in 4m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4f5e4f27b93b4a8393f4949ee1080d63 : SUCCESS in 7m 32s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d2c5753e9a634b5a89f1b0a61fe9c9a4 : SUCCESS in 4m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1e66edcf0c3d4b3abf80a5934e1da54a : SUCCESS in 6m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1b4109c18dd242eeb60dae1a02a8e42b : SUCCESS in 7m 16s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a61630cb7c38453bb2697bd7cc2a1d53 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc018e5986a9432ab68d16f9ef5689a9 : SUCCESS in 7m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5d81205bccf1439eb48a7c141848195f : FAILURE in 8m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8381a03c21994fe99bf0eff02a276ef2 : SUCCESS in 10m 26s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7bd54b0b48e742019b0c0350578efe6e : FAILURE in 3m 38s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/05544ec123314fb7b95d94b85fbecb76 : SUCCESS in 6m 27s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3e3ade3ee4894d5c9181dfd892ea9496 : FAILURE in 3m 40s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/aef264c0fd764114a6fa599a51bcc1d4 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5f6c8f33a5d64fadabcd60af0f50664e : SUCCESS in 20m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/23ae88c619cc4451a16ae22fa77b5469 : SUCCESS in 4m 25s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/101ad938416548cd9622fa0b86288fab : FAILURE in 11m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/728b6b417f3549d8a06dbf0624a02c5f : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05da32fe50c647d58ade11bfdd75561c : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa49a78cc6304c2ead013089711ecf8e : FAILURE in 19m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5da8f49431bb4918b33d19ce0b1b4e35 : SUCCESS in 4m 56s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/292a10ef7f8a454cad8592777ba50a20 : SUCCESS in 3m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb85f02717b94894b175c2761265ff71 : SUCCESS in 7m 28s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9a88265c93cc4f41b188cbfde79d863f : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27bf4a946f1840bf9da0411c558c8614 : SUCCESS in 1h 10m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41d03d3d2e2f422da0314709d31a071c : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":12},{"id":"d351952383f45e0ddf968258a66d6d8f3946d771","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 07:07:19.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"f19c5a86652f3d41b5ea11f1ed62c13c22d22d81","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 07:08:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e3f7f61636a793fd265d4c7599fc6e88d1c9c316","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 07:08:29.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fd64f6c3866cf7d165d582fc31e5114bd8aefeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 08:50:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/370ac44b624a4726b8bf766d917a839c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0209bfa9cfe541b1ab63a5dae2f2c9d7 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e75d6d444545bab4a46a2595f324bd : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46411a96d25c422194ffd1b055af5db0 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e725793e2184834881d71f4a6bdc95d : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a16364b423754986bb31d2f0d96c248b : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f124b3efae64dbdb5a9ece56cf95c1e : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f43d8b8f95a4c788c92653469fab6d7 : SUCCESS in 9m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9c05ea5b3aa405c9f829767dddee185 : SUCCESS in 14m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2749870b900942a198df68714fdc94ca : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cede7df1b9a948fbbaedcb8393ae8a45 : SUCCESS in 17m 34s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8b6bc5d34bc640c396372eb1c9779015 : SUCCESS in 4m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e5d81116f53643c7a9b54b8b669aee1c : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/57a4d6dfcffe435e9f7503dbd502a376 : SUCCESS in 4m 01s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/501836a41c6c4bb4b5bbef640b6eca22 : SUCCESS in 9m 32s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c240485da300445f85a658533dbc3df9 : SUCCESS in 4m 14s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/57c90d5e50e8417a91cc87df6fc48b72 : SUCCESS in 5m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/14f568d110584b719f83b39932a223c9 : SUCCESS in 6m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7271dd85e7474694af0291fc7aea1361 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97b75eddb20d4dca835d4b9ef1bc2e24 : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b5b39fbd814a4d48a55daa6e8e18e00f : SUCCESS in 8m 53s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/417142c6ac0b46308ad6cc5dd1187853 : SUCCESS in 11m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9cb3a27e1e0f49e3b1f81482b0af0dad : FAILURE in 4m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/89b68abea8d94f358706185896b0060b : SUCCESS in 8m 47s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2094fa5e996645c4addf43ee0e8c0f48 : SUCCESS in 4m 02s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/290a1726a42d4f5a9c4ae73002ff7790 : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c25dcd263e4545ea9af8bc88130c84ab : SUCCESS in 17m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0867aa90339e442f8bf63006a1091a8a : SUCCESS in 7m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/856b645b54f64f19b9439e593f1f55df : SUCCESS in 9m 06s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6bfc974e75de4393a1ca99487d1494d0 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5e4b80ed08f44aa86063d0dc7547b51 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73f4c237e01148c9a284904b89fd37a5 : SUCCESS in 20m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/927c33d9e8b34f96a399124b983f4698 : SUCCESS in 4m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c50ecdadb6c049a88dc359c0bc60f8f2 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb134070b8d645b2a9200abee0ff0ee4 : SUCCESS in 5m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cee72581996c4f57ba6d5d219c844272 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/180d41f8e96f4d91b34e0232f92b2023 : SUCCESS in 1h 23m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5912608a555d41de9deea8c0aac6a0be : SUCCESS in 1h 36m 33s","accounts_in_message":[],"_revision_number":13},{"id":"5178e5c1dbc43e665f75d45d433ca6fd411ecb6d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:16:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a17d2016e82f418281c46e8cfc245b57\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b3d2f341e074fd490314d74c3ae88a5 : SUCCESS in 1h 44m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/7b7039d08c1f42dc916047ff307db763 : SUCCESS in 1h 11m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab9a152bd35542eb9fd5463d26687141 : SUCCESS in 1h 21m 04s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/2e721a0ef7b3491fb808f7da8ddde334 : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":13},{"id":"a5ef09b23621a72911747cfd0f1d0ae72c43661f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-25 14:52:15.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"6e2c56c7b9d39820f5a8d50adbacd5648cc403ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 16:57:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbe8e3996533472690298487b623de94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6dd2922cfb47db98ba60a135ae12b3 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e1ba680d4b43feb9d696aff7055dbe : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ffec87462f34be3a93e4bea0a46eed7 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39e22189635e4fd787d90a611734fcbf : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/886b1145f14c409aacea340c22f4b184 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9374be1f838a42f593be10c23d7bb4b0 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25339acc3c6042609b294d6e0c544897 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be72dae8cb504033bff1f3f0c9aa9af2 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7f2e512e2e584a3c92d8c50292c9c61c : SUCCESS in 15m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4b7a1d25ba3942e582438f84d4749ea3 : SUCCESS in 11m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d9327ff95ac4991b95297954eca082d : SUCCESS in 18m 41s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2cb60c34349041a598279f20ec951b15 : SUCCESS in 7m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d496f28ba31148ce83b172a9527a3fcd : SUCCESS in 5m 02s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4ec0cecb8bb545f889cd0669e44352b4 : SUCCESS in 4m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/827d2f3c9c3745fcaeb8b6320187b54b : SUCCESS in 10m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e85a42a5ac1c43e5aa8cee3c1faa29ac : SUCCESS in 5m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/57978b47d71d47efa612a9b87c67b3d4 : SUCCESS in 5m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/62768385c68e48b5a9932971ff6dca47 : SUCCESS in 8m 09s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fb6805b70745452a80edeb982b0593ff : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c820162fa074d26b9d65f58d9ee812f : SUCCESS in 6m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/34aba5c8061344d3ae70ce95cf960072 : SUCCESS in 7m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/30d36ee680764422a3c26c83f770ba00 : SUCCESS in 9m 39s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c20deeda00e5477daba02772fb3365f0 : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f211e5e72fea4d0797cdf0426e593174 : SUCCESS in 9m 09s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9deb564cfa6646a4856e34cef0e1e3e5 : SUCCESS in 4m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a2349b822fbb4c21b20d05e6269e8e61 : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/fa19c032b3d14b2e8c444690004d442d : SUCCESS in 15m 59s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2b1300a46dc24755b4375988fada9546 : SUCCESS in 19m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7d9201e18bfb4b6f8c86a0feac609641 : SUCCESS in 5m 58s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/297d7162ce404ad5ae69baf6b22c071f : SUCCESS in 12m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3aa476037dc34b0884bccbd8aa661001 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a636a62cba0640a08d137a1e9e41b2bd : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a3be09b81f343e3a0452f48317325d1 : SUCCESS in 29m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6f51437b60a440b486302eac3c35e7ef : SUCCESS in 4m 18s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/75280b0a92a04dcc9841718ceb9b74a6 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89eed8d5546148cdaef7684929225dc3 : SUCCESS in 6m 35s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/6eb65280837e4da68d3eba8373ce7fa2 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f322a06994140679eb103dfc877cbb9 : FAILURE in 52m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4134bdc60f284c00b6da34e6e3a8e397 : SUCCESS in 1h 42m 35s","accounts_in_message":[],"_revision_number":14},{"id":"4cda80bec360374dfda106f969db1b91fa058fc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 08:55:52.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"14735220a8523bf2b9e553bf12d8605a5b9e6854","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:07:47.000000000","message":"Topic xy removed","accounts_in_message":[],"_revision_number":15},{"id":"b3ffd787bb5e5dc36b06deb030a960bfd4154792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 10:18:20.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52a875afc44842dbaaf5e941a72a8b04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278c6458f1c3448cb043253588e4eeb2 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf07dc92a0bd44aeb65ddb47a764c381 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc9f86c05ee94684833ee86e463b4d8b : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c25c80b73a55466d888c308e043800ef : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46a2f789accd43968df6ddbef456da9a : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/583bb9937372403ea1cf25ddc0936d34 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4fb9799af49446bbc53cc36d82f2250 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f1ce7e8efe449089b9d43dcd7f31fb5 : SUCCESS in 9m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b78e98a7f6ff40f3a66554de34347a05 : SUCCESS in 14m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e35a7dea8f54a3187e2feef082bda4b : FAILURE in 5m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/65ba2bf97225483ebe518203851376c9 : FAILURE in 5m 35s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c9ead5fe1d1e44d99e811c1b4443e5f6 : SUCCESS in 4m 50s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d41dc3fe18f54aab8d20a03ac109f1a8 : SUCCESS in 6m 14s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c6ba67995f424c839c94859923c693fd : SUCCESS in 4m 11s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e8d9ba3cb8da45359a9840e225215de1 : SUCCESS in 10m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/937636987cfd44888b95d954ece332f2 : SUCCESS in 5m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2675a10dc91f4a78ba5022ed6e4f01a2 : SUCCESS in 6m 00s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2335d5df6dd640d08029cd84638c55e9 : SUCCESS in 7m 13s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b71cac78ad0a4ac9a4594255543af189 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fbb46211d124be4a1693222d934e3b7 : SUCCESS in 10m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b8535214494747fa98325b17cc3e2128 : SUCCESS in 8m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9052518584a84804bc6ae79f9733c9bb : SUCCESS in 10m 09s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f54d38ae47f641379931bb842bdf0aa3 : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1d4191e274bf472099c828043112f000 : SUCCESS in 8m 31s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/52068ad9384e45cf9aae1130a09141e6 : SUCCESS in 3m 58s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/50165e68154143d294d77c87fb2f2fcf : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/86a9fc1f461649aaaf0ca5317d2b8f3a : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d50f8e10d41d436d81ceccaa36a5ca21 : SUCCESS in 17m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9fefc8fd44d6439084783788fa092d58 : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4c49574cf76f4f30a4733a578775a922 : SUCCESS in 12m 02s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7848057023f549ca9a1b59dd541bea42 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/279be4ae028c4658978bd76495c84af9 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/baa451f9d56040e5b3fb44045cdebbef : SUCCESS in 21m 37s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e1fcd73a13554f758402236e239e7d22 : SUCCESS in 5m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/10c64c9c01414da0838edbd23396deaf : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc8f9216ea334ddea49bf33df8a3472e : SUCCESS in 5m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a43caba2638948519172f4cf8e5345c4 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce456351aec44a21b6c52e71ede48c12 : FAILURE in 1h 01m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/019e5a93c49d4315950474602b667f9d : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":15},{"id":"57a20ce2513bbdf0886b0e6d5d13727dc30d519d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 11:52:41.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"91d777927562ee942d1b5cf903c5cd0a1e6f2258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 13:15:26.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a11cae7188044dcba7589126e0119d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e92deaf37245b994d6638de25d8e66 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81d3d2256d54c11a8f37024fbf535b7 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aeac44bdfed41e696b81aa323535fa5 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e97d22eaea4461f80fb32ed5d528707 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b9156981cad45b6b81d72ad61b3be87 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75cea8a584a54a04ae0aeccfaa474967 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a4b9e41f40e414fb5e5f5ec3fa4f38e : SUCCESS in 3m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5cd9d1c4f31146ab84a0533a3276eaa7 : SUCCESS in 11m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9e33bff93e654df4ba887aa8494b4f71 : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/65a774d502b14904ac0c352415e39c13 : SUCCESS in 15m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/33462e110f4648ddafd58e674b3e992d : SUCCESS in 20m 37s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/bea1053c7309482ebd24d34e194258f7 : SUCCESS in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/205c63d04692439994384ed5a65c5511 : SUCCESS in 4m 36s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/33343465633d44978531abfb39038741 : SUCCESS in 4m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b3d31669ee1044bfb9cf9d4cec95eb05 : SUCCESS in 10m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/87aa9d7b8e9447469be0908df63c072d : SUCCESS in 5m 00s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4b30297d2a484d018e355945ff45a10a : SUCCESS in 4m 45s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/381ea747a9ad4fed84130305a3d215e7 : SUCCESS in 9m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6ef1069961b744bbbba1c1ccd2b13709 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37aac906d6ec45d690b5548f74e5668c : SUCCESS in 7m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ab6f78945cf549d395b0f15c1b8888b2 : SUCCESS in 8m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/55d059b83f4149fc8c277ab85120a92a : SUCCESS in 9m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ee6e9856bcf64bb38d197c5c941b2003 : SUCCESS in 3m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/91b665512acf4949beaa0730ecabc8e4 : SUCCESS in 7m 22s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d00c5992445f4de9ba2d05a8d5483397 : SUCCESS in 4m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/70833e64efdf49d38aa9c7bb682eedbc : SUCCESS in 5m 35s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8fc3b0282da84567909227fe8443ade0 : SUCCESS in 9m 58s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c8757e7e8c4d42fa8ef7f7ec81d99d0d : SUCCESS in 19m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/89e3b6e448dd4a6d86210a5dee237404 : SUCCESS in 6m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d7b4a48053e344b5a3d45a28ca28186d : SUCCESS in 12m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3639234c412a40eebb75b73b6b546a7a : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27efbb459d524879a405b408bf712a31 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35d9d2d004bd4cc092738c7e4c32fe95 : SUCCESS in 19m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b562e8142bfb444bbbddaa75428a3704 : SUCCESS in 5m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/dc3ef7924b524c22a606172ef5be73e6 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f4cafc8bb144e85a97426eddff0769f : SUCCESS in 6m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/98fa58db024940fabd3eced3ecb46a57 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b3195c01f3a4929a0e87fd697eaea4b : FAILURE in 1h 11m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1580d649dc3f4fcab10e2d924803ce04 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":16},{"id":"81acc44212fc50cce3f5ad429e999d554f58f82d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-07 07:43:02.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e154f65f3148cff533b871d595b07f82b5f90d09","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-09 15:48:29.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"89da751337289777359b9936db3f764abf6763da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 16:57:47.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5642eeae20ea4a1cbd22d424419ba1da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c2f40075f14201951f351d7c153678 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2844e5f891df4d8fac89ac1fbfc67990 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ed74c77818341e795bbe9ee768626e7 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37900668870e479fa9455b646d7adae8 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8819e81128324a2f8d890872094d7207 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e9aa89855564616b5ef9678d84ff8af : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83dddfabfb6244338cafc07de291b8c1 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48b5275df69c47f89276c5bc801302eb : SUCCESS in 17m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/57d96017f1d54b98be7a4eddab2bcd2a : SUCCESS in 16m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b1f9b9c6cda04bc381edfd55c0c3177c : SUCCESS in 11m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bf83ad4157e248c2a8652c037c39cc97 : SUCCESS in 22m 51s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0119683f61c5440bba2a53c40a444c61 : SUCCESS in 4m 06s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/73c073a038e24baa84546fd4a3475ea4 : SUCCESS in 4m 05s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b48806b5a58e45e89793940fdedc7617 : SUCCESS in 4m 03s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4e7ccd38108249408fb70d555e73b744 : SUCCESS in 10m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/24fea5065e684c71b8eebca7aa631c06 : SUCCESS in 4m 01s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ad99beac167b462f854d8860c57b46a7 : SUCCESS in 5m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a00105039c014af6bfaf4c3b6010047f : SUCCESS in 7m 35s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/986a8d6c94994443b6a6d750ecca46f0 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/608e07684a984567b8f377bc9daa378f : SUCCESS in 7m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b58cfdecadbc48528c2d218cbfad328e : SUCCESS in 14m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2e2d8ebd6b2446baa3d4b9cd886eec9f : SUCCESS in 10m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/adfdd9bdd04d46f89ff95c2a722fc704 : SUCCESS in 4m 06s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/83f5e2dc6a934c608cb03d222cbba123 : SUCCESS in 7m 44s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/34edb9833a144708a0646ea6d2d5b4b9 : SUCCESS in 3m 48s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/60a5e878648f4807ba80442297f58aa1 : SUCCESS in 4m 14s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/08aac2886b0944529524094926c864d0 : SUCCESS in 8m 53s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6d6a96cfe21a44f3b5c491287b45e714 : SUCCESS in 16m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/469dcc49b6884f3cbd61995b35772b92 : SUCCESS in 5m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/cfd34d70427648cc97b102c02cf41dcf : SUCCESS in 9m 07s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c4cf7693641d4815b689ec21fd17671b : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/799ef1eabe7343eba95389d369e66193 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76a9c5f66ef9437a879f97402e092f12 : SUCCESS in 17m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/131d54cb152e40c6a7b422333d2835a9 : SUCCESS in 3m 46s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8c52fae62a2144a3882bdb0dcc8aa6e3 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae47f77fbe3047609027fdad6a64f1ed : SUCCESS in 6m 06s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/676c8c5b19ba4d15b54b43f9240a2e57 : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b2091f7c8614433ad8f8d0460e9a412 : SUCCESS in 56m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c1f1bc4e48b48a78e3d4881813f789f : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":17},{"id":"7a3bf6e500b0b6f60c8c0796a5de71a5fd1c8479","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-20 08:54:46.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"47777e5b90327f411b9360e837613c8fe4359662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 10:32:05.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24b4899f389e49c6a7c83b460c12fca2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c863033fd384e60ba0d44da02786bb8 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe3769940384f028190d37e63513b8c : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16417638dba64739bded7ac9b5b45121 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8621a0776ff643ecbde728777171d9aa : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c02435fa7af5419288caad8bca6059f3 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c3502d0e6bf4fc18af439bc87ac8250 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06adc18bcca345f19dd7a6c0f88cc904 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0036e92587a84fc6be4182da6fe46dab : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/412c50403b784b2dbc0985adcb64ec65 : FAILURE in 3m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b2f386bddad84426bfa6f49532bf3ce7 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f7e29f5543e3449ca90d56c33e1c5d6d : SUCCESS in 19m 24s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5fd80eb80fb84639b554d7bf39b3fed0 : SUCCESS in 6m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/85e3948801384efbb438626c61c41073 : SUCCESS in 6m 26s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9a2b6a3c4c6b47c4942936ef433b1f22 : SUCCESS in 5m 34s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c31df3b393e2459ca2c6e116e299a29c : SUCCESS in 14m 29s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c100b9f565a5416b8f1cb168c54f0b9e : SUCCESS in 6m 31s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c777de1857aa402e8cd8f8124e8e2e17 : SUCCESS in 5m 34s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9ae6f0c58c1643ea90d0df40c216cab1 : SUCCESS in 8m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f2ee335da8ed4edcaab21926b866c825 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24aee56eefb74398a862037892fdf626 : SUCCESS in 10m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3e60cdd16bf94a3aa12ae6f499e24607 : SUCCESS in 8m 11s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/61c2bd5889244b94b7b06eccbc600dd1 : SUCCESS in 10m 41s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/451999780b8341caaf79e5ac03ddda63 : SUCCESS in 5m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ad9948854f844ce499d7984ffb1fb007 : SUCCESS in 8m 21s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/f40be3685e344c639a02307194a45d87 : SUCCESS in 4m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2608be50bfbf484a9c89beebccecebc3 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1ebfa5bee81640b9bb9e3c4bd601f3e4 : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2f3be4942fa840c4aa1ad65e6b9987ec : SUCCESS in 25m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/55069452014b4cf9926513e90c9d0888 : SUCCESS in 7m 27s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/58349526448941e0a11814f481621ce7 : SUCCESS in 13m 30s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/81f6eb0d3b094d34a98ac72e0851722d : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/553e5d498ed74d07b4e0ea9196879950 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b3fb9bfae78412e9cc1294739047567 : SUCCESS in 24m 51s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3b17058d2ff54ef68d2bf2ad5c49c25c : SUCCESS in 4m 21s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3130b352f5d649379ef1b444a73c9681 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02f9fdd7643b48999791e7a46325b2a3 : SUCCESS in 5m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b44380d7e5ae4edc87855e76b6c4121b : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ccd022c7024489a83972be4d25a0b82 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62c3c9944564472397e7019a7c000210 : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":18},{"id":"770ca784e1369ef39f16561e861ffc137c505af7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-20 12:31:53.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"fad7e85b56c86098eb041ab821f0fb5862b378d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 13:40:46.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91c18c3bd634f78a1fe394882a76aa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6242088a2d9408da7940731bd7399bb : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc3541f765247ee971bdd992713c8d7 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51332b9818724e3e9181397e523c87cd : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b68a7acd8534ec489bbf5fd8d7becdd : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bbd1bcd93ff4eb389f75c4058913f79 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f281fdbe9af048acb63617177bc6f2d4 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9e8a22617e748058a29fd51b01ecdb2 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fe29fa0aadc412d88452a3cda955bfd : SUCCESS in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/169b128de63d412ea5a050c0e6018ab7 : SUCCESS in 16m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/01f006fad2004d47b40e5db68e148365 : SUCCESS in 7m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59986572f85d41e9bf1a9b1ea36444df : SUCCESS in 17m 21s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/658c0c010a08474b951376c2f7307c75 : SUCCESS in 4m 04s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ee8761ebcf0a46d082e31c74fa3b5af9 : SUCCESS in 4m 54s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0449c8cfd6b64a33a4f5dab92b656c7c : SUCCESS in 4m 13s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4767f2e15acc477ba7e30b55a3c777e2 : SUCCESS in 10m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9a72ef76a37144b6a6f6107906501a99 : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9433ad7c02434244b54e9bebfa251baa : SUCCESS in 5m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f2806378eb3548ed976354bec4692216 : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7ecbdaa386174846be2f3d1538489178 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/610dd44fe2e1499aafd552ea4e0ffd33 : SUCCESS in 7m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c0dbcb6a503f4f90aaaa6c9c02be7c5f : SUCCESS in 6m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/86644e6afef0489ba28901a30131ad97 : SUCCESS in 9m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c3d63f07df0849649229208430dfa5ca : SUCCESS in 4m 27s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2bdeff8776a44f3dbd234fa0c9fab846 : SUCCESS in 7m 09s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/336f5d6f657747d98a98d4d7dcb19633 : SUCCESS in 3m 40s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0e571397a1bb471f81e66855cacca8d1 : SUCCESS in 4m 03s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a67fcd7c43e04ae0a1c9b485e0c5b5f0 : SUCCESS in 7m 42s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c7bb2248be964ad1b7bef7cdde4a59c8 : SUCCESS in 27m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3d0fc4200b5546e3a0479034246dd3c1 : SUCCESS in 4m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2de1d65ff1d648deb46e08112d4b3d9c : SUCCESS in 12m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1566fdd029764cd69cdcd9303596aca9 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4598453c8d7e48d68daee2eaf79ccc90 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb585a296a0b4b39802435130b49178e : SUCCESS in 23m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1d1d21594a394533a6841a60349a1086 : SUCCESS in 4m 11s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9167c177e9754cbca4ccebdcdc901396 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80ece961713b454abc1825da8a29da26 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1ebe9046cf16455fb54b414ea473ca2c : SUCCESS in 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56bcff6404d64bdab3c9abb89756974e : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f716586c03fc48ba91acb1086447e7b6 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":19},{"id":"1b988b7af57acb7f0c9d496c5e6f94a7aee9af08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 12:23:52.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"f173c500006f905f8652fac1a898a84a2c3916e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 13:45:15.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94f137968b8443398ac8b4b6d85b3463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3cf92d403c34dbb84be8e5789d68f00 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af50e3e8e6df4c51905316e312ec8125 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cadfac1634547eaaa61b53db52d62d7 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/691c32c2c8d64fbbaefba2b2d71a4076 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/527c1c793f93429d9976d74ba9dd07fb : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98df5abcc3394b30a0b03ad74d71d60f : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/583f1b707ea4482a9642707ccc575058 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/054e10426de94eba94f88644584aa3e6 : SUCCESS in 10m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/737768bb7cb84791aaa811db372bc3c5 : SUCCESS in 16m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/198bc5f6a5da475c8f2c4e79b3773684 : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a73d697516a04268b73d9107dfad34ea : SUCCESS in 18m 37s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e7b21db696ca47ed96a1d3be70e5ca2d : SUCCESS in 5m 01s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e3d007dcc23b43aabcbd1275bde58f63 : SUCCESS in 5m 16s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/67136ffa79714d33b37c43e4105f8b7f : SUCCESS in 4m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1fef84fa930f43049309815a17709d4d : SUCCESS in 11m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1e01ee20b38344a79461bc255462ab11 : SUCCESS in 5m 09s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/642f4c70bd6e44218784624ac9ae6719 : SUCCESS in 5m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d2941aeeb0224bdb99adb84fd18582a8 : SUCCESS in 7m 34s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fb7a8c8011e3486ea198c793865ebeff : FAILURE in 7m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03216a63779e4673a937c52ca96dac6d : SUCCESS in 7m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fd49e4fa01ed44d4a0001afe8d1e172f : SUCCESS in 6m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/955e2ef3eef34f85b6ab9d8bbcf24386 : SUCCESS in 10m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/aa5a585465684b39818c5a510cf24dd9 : SUCCESS in 5m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1b67bb8b55424d86909ff8329c977dd8 : SUCCESS in 8m 22s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/7b17919ed74840bfb31bcf68a2fff191 : SUCCESS in 4m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/948821da444d4dd4b927f0597868c5f9 : SUCCESS in 5m 58s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/6f4e25714db4407f98e889208bd062d6 : SUCCESS in 12m 15s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/85e83617bcd746b8aea4efa800373722 : SUCCESS in 24m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1677477c28294852aa702e3eaabf9bd1 : SUCCESS in 6m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2a87bc3163674dc8b30aa42fe86d9a8a : SUCCESS in 13m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3cf32cd573554bb8a20e40bc9c54220b : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2852aad94262440aab4ab1b03b819490 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82c78e407d7a45f4b099e1c5080c7626 : SUCCESS in 22m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2998d07889954333817119b8046b4ff1 : SUCCESS in 5m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0fbb57c1f9d54e58936ac7d951b7ede5 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dad67dc84bc846aeb5f7b6ad2fa05596 : SUCCESS in 7m 02s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0deadace7b4d4e0ba8aa7142ed951ada : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71bc046a3ed34948b23e601e09518737 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cb61ce9a86f4e20b86ab3ba80421800 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":20},{"id":"24d2926e23e3b912ff098ef946aa80525153e556","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 08:47:02.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"2aacd33d9534e0a055bf030261ebeff43165d8a7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 09:39:11.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"7fd5964a142dac1b9e216b7073519a8584178787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 10:40:54.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e995abcc0af4c0d861cc96f1ee9a344\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26465df1ae9a4221ac75c0079b374d04 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c8fe86ece74aaeb7d54f040ce4e2c3 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5056cda5c3eb40a48c25cdb90cb9cc1d : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18b1c024913a4ef68bae5780a77466ac : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c66c2cf7a3de462ab300ff9581ea4864 : FAILURE in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f091c60afe14d77b9e7ab490867419e : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad9ef4f2343c4e7d88547877e97656d5 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f42b6e4c239e46d6b81f0a0ad06683c9 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0885902f39dc4615aab4b07ef0951089 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7829a8a830ef4a7484dd098a4e967a9c : SUCCESS in 7m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e25695e212474b388bc068437ceb685e : SUCCESS in 20m 26s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e819de87ae4c41b281b449f532c06eee : SUCCESS in 5m 01s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/43ece268fd29478e88da2de0bc452ada : SUCCESS in 4m 59s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/579388861caa4e48bca410b56ac06098 : SUCCESS in 5m 03s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5524fb4f3d394d61ac789590300fa97b : SUCCESS in 10m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e898376215ee41bba09c4af1b3092d24 : SUCCESS in 5m 01s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dd1a9b6c17d145f887f07c97aeb72ad3 : SUCCESS in 6m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/87f75b707b3149189e9cd36661401aea : SUCCESS in 7m 50s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cef4e985212041bba22fd0713ec63282 : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2ad8f74dc084ad480eb74887de153cf : SUCCESS in 7m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4e0aa483129441ba911c077fe3cdbc26 : SUCCESS in 9m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5e38f3c6577e45c3b802ed2134a42869 : SUCCESS in 9m 41s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b5b6e35ed01b47e2989806a445b74f53 : SUCCESS in 4m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7bc954f78b674cb7860552512384c30b : SUCCESS in 8m 58s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e39b0a0e80f540308407ed88790ba432 : SUCCESS in 4m 25s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/8e4fff85e8684eeb9babb8101ed2716c : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/13ae6075a87d4e79bab0c15f2fba2482 : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6cbd11b62ebd400bb0399b36724d6b55 : SUCCESS in 23m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/84a9f78d20764ee68e231d2c29716b2e : SUCCESS in 6m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b1a0fdbb5ce447d498f5aea3a715afe1 : SUCCESS in 11m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/71c8a9ce7dd24478b4611d10f50d2314 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab2ff211caba404f86586c95f8069707 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4ffb07c3dea4b0b8573a319c0f77e39 : SUCCESS in 23m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/681465adf4d541edbd9d428d5b26d045 : SUCCESS in 4m 27s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ccd151a75f31420a9e5750135e110f2b : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f9fb35069b641ec81df73be50be5e94 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5c989de742b9474bb02cec8ec2ce34a9 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eacd3571e066414c8cd727f3d595cde5 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":22},{"id":"f92fc332c336bb4ad5886e181576b5d48401d7cd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-26 20:02:49.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"bd776483faa8e4b0d36645ad98846c287a3274e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-26 21:14:28.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10cec5ba48d743dd8bd7ee8e32a5a4f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c53dff2abf4f8a9dce633afe74c029 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a7ed83f4f24a339c3b9f21cd6a819a : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab862f36cc3f4336b3d7f18a5871c5cf : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/229cc4608e7c475b9f20b1ac7b6552b0 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f51d7df5d594ff7aabe1ee713ff1147 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa1d6cbba6c844378e0cb182559f96c0 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b77309e4e08d45fb9f344c0e8817dd5b : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83409effdff84266bb17e29a4642896e : SUCCESS in 11m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b478758064ad4d55bd544c7801169d46 : SUCCESS in 14m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b499e4e7f8ac49dc8ecf391c4d631e19 : SUCCESS in 8m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b023f0b7e9ef442ab0aab790ef498c80 : SUCCESS in 16m 19s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/db4cc9bf444d448abc04c0b51c9788e8 : SUCCESS in 4m 10s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ae70e9896ee64ceb9af87ee6104517ad : SUCCESS in 5m 14s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a89800c8e4074418908f47dbabd94acc : SUCCESS in 3m 30s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6f5423fc5d474d048ac6f3264675a78d : SUCCESS in 11m 29s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0396181f13d44b4cbef2cc5bbf1747da : SUCCESS in 4m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1e99aafac84243038d5ba37edc631b1b : SUCCESS in 4m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ed3621481cf1406ab199d8384103b5cc : SUCCESS in 7m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4298c4d10b564c019f79cffb44a9c818 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf1e8482de0140ff8ed0be9759f0c118 : SUCCESS in 7m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bdcca8a3b8864384a9b912845bec9059 : SUCCESS in 7m 35s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5b91924d8eff434481e36bb3878e3cf0 : SUCCESS in 7m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0e54b539119240699c8224513367f7ea : SUCCESS in 4m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ec2d00545dfb4927a00fdbe07d108623 : SUCCESS in 8m 19s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/db810133e4674953ac284def655399b8 : FAILURE in 3m 47s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4876ee8b3c7347c5994c806c31551a3a : SUCCESS in 3m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/28fd2f5ac2bf400ca583411ad78dcdd9 : SUCCESS in 10m 01s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/99c15d53962f44528723d478399578ea : SUCCESS in 20m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/52063105dfb74bd0b2dffde59383fcad : SUCCESS in 6m 08s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8e95dc392e8043b6a991b20be2a412b3 : SUCCESS in 10m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/20e6dc4062b34234844e718c73b5fa61 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6df446721f7e443fa569a8c4778ddd48 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ccdbdffd7aa4814a3c74cc346897f23 : SUCCESS in 21m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8cacaff02edc40b8bbf9a39ae0545bfe : SUCCESS in 4m 33s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/bf9be3989c5a4cf9afefeba19829f614 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0b057a0b3ab48d3bd07d84f33c0b576 : SUCCESS in 6m 10s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f09df351752842a8826392e2c349a0d3 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71257ca150d841379914abe4ed2a0e2a : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":23},{"id":"f8906b08da9326a5a94cfa95af3b1db624a12495","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-08 15:33:06.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"b8001e42aa34754326f75c30907e2ba23146f0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 16:50:20.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e16ef1db94c4cf08c443fd8fc6fd581\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5259416a534c2c83c2fdf1b27b609b : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939a1489e7e44d968c40b62a8dceac59 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91b104131f5d462da660d0aa5d3e8487 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/917ec5eb9ded4e26b14268bcb1b35cc9 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32691b17ff09465d82377184a74daa4e : FAILURE in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/502c458012f7417ab7c1c06307ee6652 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c23d684e259401c8cdb91f3f3882b89 : SUCCESS in 2m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b97305f28864995b8278b1471a566f0 : FAILURE in 3m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2130d59cba724672b3a1d8940c1b5828 : FAILURE in 5m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d82879ad89e244cbab440d6d81823fbb : FAILURE in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fef518c1c1cf45f99e1c83f977e6debb : FAILURE in 6m 30s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/27fae54d61314704be80ea2e0d669829 : SUCCESS in 7m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e2e69d561d4a4523920c9bd8e6224681 : SUCCESS in 6m 33s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/616038098c124395b84481db959d75b8 : SUCCESS in 6m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b7fc75455b1645d59e87616adb9362de : SUCCESS in 10m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0a9405a8cfb848deb07f8e9e784d3adf : SUCCESS in 5m 27s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9491ad48ad1c4f1d816a1fa905b0e71e : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0a71ab19c98a4539953e3b106486551b : SUCCESS in 6m 25s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/49a7e2d91d16499da465b4036297531f : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd91ee3491ca4103b24e53d4a15f045b : SUCCESS in 6m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a301c7b065674cf1b72b23408abc295f : SUCCESS in 7m 52s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b426dcafb121417db694f17762cf0db3 : SUCCESS in 11m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cfce2f886dc04542aa46529ef2e741d0 : SUCCESS in 3m 32s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/90359f335d6347db925c22bbd2b29ce6 : SUCCESS in 8m 10s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c6d655c0c94e4b348918ed83e48d9bc8 : SUCCESS in 4m 31s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3160898465ec4443b15aa3f90db6b587 : SUCCESS in 4m 16s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0b735836a3d44769a4da0f82db51ad71 : SUCCESS in 9m 15s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bf16b40d9bc3402fbbd6ce29cb6e0836 : SUCCESS in 25m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8c6650217e7544a18ebc8516e75f8514 : SUCCESS in 6m 08s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6549273b6a4d4a3c9cd23bbc019fcfc8 : SUCCESS in 12m 59s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a6f186d5c7214b1c8653571c4045a1ef : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc461220edd744cab21b57db65e0eec0 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f01a58c342b74cf0bb9966f7f06d9da6 : SUCCESS in 25m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5f3e110f414342e5af6d838859ecbd46 : SUCCESS in 3m 48s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/eecf09f26a584a4da31c60dcf41481d9 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d42b5e27031e40679703395565eaa0cd : SUCCESS in 5m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/64cf0b5309d1494eaec0948a153a57c8 : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/874417f308494883adc1db8056cfe287 : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":24},{"id":"16d5f221b28a141d666f1aded4a8c91c9a2d9dbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-10 10:59:08.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"105112fdd4a68bd64bf54c439e6e36377f640f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 11:47:59.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/644bdf9e1b30478280122cfdaf24f815\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da138fa426af4d4692cfd6c67d31e7df : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeffae7f09f04c6dbb98d9625c4eb952 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2200b3bffa9f40b19949de6929ecd876 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8031abadcd974889a074ce8846b2aad3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02973e5732cd499b8abd0874563058e3 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5bb91fabdda4b0b895903d5cebceb24 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1afd211404b4e74a6f0912929e447de : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/884607b6340b4e34a467b0c9a4bad96c : FAILURE in 4m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d33a384ad3b84a53b0f1461b394aeb37 : SUCCESS in 17m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ab3ca76620b14820b31cc5c4d4c3adc9 : SUCCESS in 10m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cec5466ec44848c599d124446c75edf3 : SUCCESS in 17m 26s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/37eb4b62038e4c07989ff3c0ba894fa7 : SUCCESS in 5m 02s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ce0fb9b32e664cc58fba048bdf33bea7 : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e44b2ab7b8ee49c2af7af777bebe85bf : SUCCESS in 3m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b7edc5b7def24924a0d62629073438ee : SUCCESS in 10m 22s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f9cfea03f39344f8a36fe519977e843f : SUCCESS in 5m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e682686d94624ff6a64a7ad48fc65e62 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2825801d76dd4cab9611e78ea5224434 : SUCCESS in 8m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2ced6331efcc4699ad267609809c1c31 : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3974b3e96d4843c7aaf6c652eefee1c6 : SUCCESS in 7m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/09cc404fa80a4a88b0d463f0f12ee472 : SUCCESS in 8m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/eb7325a7d58e424d973e5a457f8b2b72 : SUCCESS in 10m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/5165a0c27a7c4288bc8ef96f4d59b0d7 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f91ec38984e74f4f88837452b8b0f2ac : SUCCESS in 7m 17s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d1dd267234bb4b82985372e074557d5a : SUCCESS in 4m 39s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/134590d703284beea1e81af45e7a65e7 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/17912114fced4efcb57e971ca6d4f774 : SUCCESS in 10m 23s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/22833d613e324873978a12480ba3dfd7 : SUCCESS in 21m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/471174f9a18a40b68428b048fe4b7cdf : SUCCESS in 6m 06s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bb3d57318fa048d0a8e3a539e455ac32 : SUCCESS in 12m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b589383af64549108a2c7a6b6cfba397 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/624f80e8cdd54f40a28fc37b52a51525 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fa65a428b2442a680ed263dd1cfbed9 : SUCCESS in 21m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4699ebb76de74bf2b4f105c7b987bd65 : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d43e8085513d48d986ae20af626ad3ba : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/089efacc4630420e8925ad894b0a507d : SUCCESS in 6m 11s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f2624065e6334207a852735f2490135d : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e89cb574fa684833a68d69610bb34a5e : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":25},{"id":"3bef712c7695eff348c1eab354e7407619a899cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-10 13:23:44.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"f8c3d26ea509c253d9390dd9068a58090250bdba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:44:19.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a556bf29f4f246139f1792696c970840\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61bf28f618d4cf9a41db2f5a399702b : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfabf45da5e140dc8848ae5e3e76e598 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac601d56d230449d8994fd05e18affcf : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20d45f74c375402e941c4a33a8bf03d9 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9562830e6004c9ba33f5a86c872903c : FAILURE in 6m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b63399a3e10f48abb978572c8d9e294f : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/752703a4d4984087af4269c88cd1394b : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa1c11071f924ed4bc513a68314610c3 : SUCCESS in 10m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a124841e6764508b3e67437c93cc24b : SUCCESS in 18m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ca8a0177a5d5463aae94221e94ecdceb : SUCCESS in 10m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e92790104df6491e9cb2fb4088f245e2 : SUCCESS in 19m 05s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b78c21ed87c846dc83aee6e269a4b4ee : SUCCESS in 5m 22s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d088c166749242a2b9c80690616bcab1 : SUCCESS in 4m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6e5bd47277274c84b9b41c3966ecb0b6 : SUCCESS in 4m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0b7e2be02b13450a987326f6a70754c8 : SUCCESS in 9m 40s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/bfc5e2a68ccd4e4fb1fc55cc8548b6d4 : SUCCESS in 5m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f6fef01c772c4cc68057ec038492f0c4 : SUCCESS in 6m 10s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/4b002f64a3214fab91601b8fdddc3c69 : SUCCESS in 7m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0acbffda79a540ffb24a99645b0d183f : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f791067c8c9b423896093185668c640b : SUCCESS in 8m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/de5e07f3299c423c93eb49363b270bd3 : SUCCESS in 7m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ce870cf8bf2a410697bad06765a25700 : SUCCESS in 11m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/5d4d2580683b4afabad4edc0aa236a52 : SUCCESS in 3m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f61bb077f2984d99913a9b227d142fb5 : SUCCESS in 9m 39s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ed11668e83e94c0fbdef39f8ede314fc : SUCCESS in 4m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d6ea455bb813450e8a37b86795becf63 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/64c3602cd5a1406d9edf13ad4a47311f : SUCCESS in 13m 59s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6dccb9083f154ab782c9c168613e198e : SUCCESS in 25m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a9d183bdbf894cb6b55e859566033644 : SUCCESS in 5m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/07b84e2332d04ad4825970aed96d8791 : SUCCESS in 12m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/031e18a9ab0b4a91af7492c879819cfe : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a055a01cdb35469e90e00de9a3a33948 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f036acb801274ae38c5b5c41798d1e15 : SUCCESS in 30m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/03866685917044e8bd826a0bac4054ca : SUCCESS in 4m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c2ff851114354277829ff68e44fc9d42 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9894f0d6cf2440908b1ee6574010bb87 : SUCCESS in 7m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/262dba0a24374e83b0b70e3b26b498f5 : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d7867a83bde48aeb1ff056d3f72eaae : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":26},{"id":"74c78de6a2eef3b1a0584608108584cb6b436920","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-22 21:42:41.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"cb576ddbb6fc8b17d65ff39687f9d24440291b9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-22 21:57:45.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"b0231334e25a7773d873b86b48d2583e35c1691e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-22 23:11:51.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73e34583703f4413b7760dacd4b47185\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98108daad0c14bff8a119645e7445795 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac86a5aaae5949fe91c2bb8a46ba1e90 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12630842f6454a2bb81df03c3ce26ada : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec68d667ff3f49a0b335ac3bcaaa1473 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/031a09d98c004d2089f36274883b70a4 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/309a292d7ee94e2fbd090121d9e14724 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/665ac0b874d040be8773b2eee75cf7b3 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b218e52b84a4eef94291ec57c7dd83e : SUCCESS in 7m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/974ab587348f49769f4b7026e7337575 : SUCCESS in 15m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c1d29c5677ef4dfe8a21c18ba8994b1c : SUCCESS in 9m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a4100e6541244bbf8b0bf1ffdd4dbb31 : SUCCESS in 16m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f0fce696731a49f5af0f7bea62e2e7fb : SUCCESS in 27m 05s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/94e0f0cbf7e8416aa1d49a26c45dabfa : SUCCESS in 4m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ce8dbb1f8548491a96796a6287c45f4b : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/96a73874c7b4463da44ff06275d605f0 : SUCCESS in 4m 50s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bd4e568913b34bf7967c11dda9f2974e : SUCCESS in 8m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/121876fb1a9e483194f54fe76e0a9422 : SUCCESS in 5m 44s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2d6d2e3654984bd493202f889a130c05 : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d0bdd779bc2248b7a8aadc4522b6d3a0 : SUCCESS in 6m 50s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/acca40bbd6804b51a850eb97b4a9887f : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15356ba66daa4101b7d27549fddb3ad6 : SUCCESS in 6m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d3bf765aa84647cbba1d3479cd0e8022 : SUCCESS in 7m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a8da90eaed7a4ec0901edf59e0e4ebe2 : FAILURE in 10m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f3ac07d1dae24ef3915e3ac3c8a92fe1 : SUCCESS in 3m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2981bf6213164c05900cf000f8849697 : SUCCESS in 7m 49s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3ccbd3d5a89e4c66a3ce43ae7f1f4e6a : SUCCESS in 4m 47s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/47a38992ffb448e1a387392ad4fe0921 : SUCCESS in 4m 01s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b2e8308fbcc141b8922d75a394435f7c : SUCCESS in 11m 40s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9bf9570b9e534eebbb0fd95be9f1c3e9 : SUCCESS in 19m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/728124e050a94d17a8b80eb6a58e7340 : SUCCESS in 6m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/30dfd92249944984854e5af3fc9d17b6 : SUCCESS in 10m 45s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2a55383dd00e4bfcb68095b4f48ced57 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e95095cea3ad48faaf76d4af5a64c62e : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24d8046dd46b4b66ac98b90b755f5e68 : SUCCESS in 23m 55s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8788890ce04942a2b4664e7e431c6d26 : SUCCESS in 3m 44s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/72eaed46526c406493373a3ede289d54 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aca4462389f148969de6b5a2db3578e6 : SUCCESS in 5m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/831a99d3e1a74fd9a5068530b8c53caf : SUCCESS in 14m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b3328d853c44e6aac5814f8e20ae78a : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3efd1d0ba46f4fb88b08f0fa33484946 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":27},{"id":"3aa64aa0eb1c4fb0712364910756ce2b21db4200","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-22 23:46:28.000000000","message":"Patch Set 27:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc4ceb250a6407599ae7e339bd7e819\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/da85b5a3d4ca4076b64aa5926365c70b : SUCCESS in 1h 22m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/9304f9fce16b4b039d05516603ea41e9 : SUCCESS in 1h 32m 45s\n- requirements-tox-py38-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/ee6cafe00622410d8991dd0eb090edca : SUCCESS in 15m 05s\n- requirements-tox-py39-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/8b50280265dc4adbbfe6dffeb1bfdcc8 : SUCCESS in 15m 13s\n- requirements-tox-py310-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/15859d0ea8ae4800b7aaac57aae3ee09 : SUCCESS in 14m 13s\n- requirements-tox-py311-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/69f5012e7d5248c2b104925cf384933e : SUCCESS in 15m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b46f0dbf217948ecbbe2cb0aa056274e : SUCCESS in 1h 19m 15s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/2465ed713d6a491da9a116f160fe1c48 : SUCCESS in 1h 43m 21s","accounts_in_message":[],"_revision_number":27},{"id":"e3e3452a556d5b0ab560d4bdf6e3a6ad53419eb4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-06 17:33:41.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"840aaabc97b2644d2c7cf1c9d7e801c72ac8465c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-06 17:56:13.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"0f82bfb6e1197b710e2d62ae2a8dcf9ed3512d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 19:49:21.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e174c4dbec7b47089b5bc950a1218578\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01bd3bfba1845bca63a03489d53f77a : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab28069cbf714de29cd5491507e7d54b : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193ac1a1c9b74b01a8fb76078f49f485 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3281e81c0a5400e9ba035f56f2a88b0 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c19c8bae0ab474182fcb2565cab223b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d9e76a30e6e413da99864d64417561d : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f441ed8a61f84a0bba53edd6296f8286 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3178f8b3b69d462c9998ed524c0ba557 : SUCCESS in 12m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/59a4e6bde52b484bb64ad4413b499468 : SUCCESS in 14m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ab6dd23b3c7f4e5c8889cd2b04815645 : SUCCESS in 9m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4eb4e3ae7587415b9ecb21faa55ce3b9 : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a441ab2456584565a90c5459656774c4 : SUCCESS in 29m 01s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a2d883e8bd564e988a00331315f31357 : SUCCESS in 4m 52s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/01fb3c7348bf4b1fb4982e59b99cf7a2 : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/cc0222441a38403c889e8546301e7a15 : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/81918a1a32d64952a3ed3bc5e323a734 : SUCCESS in 10m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2ebc0ecc0316486ebd443f5a8f624bc6 : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4028d6341df94fe0afc428e7eae22726 : SUCCESS in 5m 21s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e921118471a8441fbe5464f809e1af99 : SUCCESS in 7m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9488e28a6e314eec87391a18d40e2708 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fd1d2b538964657bbf7a762f7507ff1 : SUCCESS in 6m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/742fbe4e21ae4d158a976dfee2c30b64 : SUCCESS in 7m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2a5aa23f1508489996117bcbbb8653f3 : SUCCESS in 10m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0c718980a3854dd78451f62c72cd38ed : SUCCESS in 3m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/cf89f76a62ef4bb5a4fd8d3ba0196e14 : SUCCESS in 9m 01s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d02062c27e8b46b3ba6aa37e167d43cf : SUCCESS in 4m 14s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/50cf1bab1f3442d793440757166c8ff6 : FAILURE in 3m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/bda9460c072f4beeb54b9925ef977cb3 : SUCCESS in 10m 54s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c98b47bf5a7c4d28b7f5270b4ab58d81 : SUCCESS in 20m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42a37ee418224fdbb90b5ea5e1581d95 : SUCCESS in 6m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2557b8a085be4bde999030af8df2d08f : SUCCESS in 11m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/aa1848da74854b758ebdc091d037968d : SUCCESS in 4m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed8e02f87f6d4fdbb828281286c12c6e : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81cb197957be4679901ecd241bf917c3 : SUCCESS in 22m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/29fb2b92bdb64918a92626fed37441a1 : SUCCESS in 4m 07s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4e627a2e708e40bf8e7adf02d1e0f780 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f61bd8e2e39b46c687244ad42d4d0927 : SUCCESS in 6m 04s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5e43e472665a4f13aea78f2bcdc19049 : SUCCESS in 20m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62001baa976b4f9b81e6add39b908685 : SUCCESS in 47m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2256063bf8084b52a14dceafc6ec75a0 : SUCCESS in 1h 38m 34s","accounts_in_message":[],"_revision_number":29},{"id":"99f42456e103d37a0e2c057369c0a7f7ae49e013","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-07 11:10:49.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"d8ea1014613e35eef099bd24158a4a57954b0256","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-07 12:07:49.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"5c6f8ea38bfcf1d6ee9eb4c0664461c9d85fd7b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 12:44:57.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91afbb1af9384a5198ed662b0e35c686\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b75e5daeb6447f699d96717a679b390 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c7ef608d6b40789a8187129ddf9181 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/028d2818344c4f8ca25ec3ba38135e4d : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78ef51455cc34af2b2099cfbe3dc0a04 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d78842c0cfd14c89824154c1f77a66dd : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a261725310842c0acebc9f9389aee23 : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba4c027cbff94d0d9a09ede4bfc31859 : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/604e4c319b1b4fd58fb798056ca5e4c6 : SUCCESS in 10m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/559725c6eaeb4fb0b015f361ebf9b045 : SUCCESS in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9515fb050004a78bfe37d990517e3db : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/09686db5d87e4674b23ab6f72d5e2869 : SUCCESS in 16m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/27d8be08b94d418286718b1c58a25f88 : SUCCESS in 26m 39s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/17b527133f464c679999d1d21d4bee0c : SUCCESS in 5m 04s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e8c1555b896a4e2f88e4826b39819c75 : SUCCESS in 4m 41s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/764c45594ee8456dbf915877eeb9e411 : SUCCESS in 3m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/177da43d79644f61838ff67968293c9a : SUCCESS in 10m 46s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/616ad34d9084453dbe518e8f536700c2 : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3cc16d745edd42e9beeca09072d47e54 : SUCCESS in 4m 54s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0f90d79ff17e43d5af0911c02a6458b2 : SUCCESS in 7m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/987c59a5038f41079934c8919c8920ca : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a153b66e2b049a3a25234aeed78577f : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/467bda51ee604fa2b2be96bbb7d6466f : SUCCESS in 7m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e17f02c0e66d439fa4f010aecb13e640 : SUCCESS in 10m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a5a09d2773d74fa0b7f7676443f37120 : SUCCESS in 3m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1801b5d23d56486eb0b07b632ce16871 : SUCCESS in 8m 00s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/eb40571e31584d1dbf636a71054ddc3e : SUCCESS in 4m 07s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1e75360eb8a044a8ac8207e20551d36d : FAILURE in 3m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/373f5c3bc6e14e2d93b5b3b2c3ae15ef : SUCCESS in 9m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1aef9401578545ef92319ac3e5c682a7 : SUCCESS in 18m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d6b4a0d1abba4400a415acfe2b58c920 : SUCCESS in 6m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0e7a8247c35b425ab9aba961e5019469 : SUCCESS in 9m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/754163f78103406cafaf02ef096c2813 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2eadb363863441aaf9e7feab0cb0097 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbca72664b4746859b0000217bad0810 : SUCCESS in 18m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/37edb1e2646147c597b3cdc9cb2f2e85 : SUCCESS in 4m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7c242b35c40a4ba08e09338c9f622744 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/224fc7e72ef84ab6b50c13ce934233af : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f24b680716fd403fac74d5416f9fb69b : SUCCESS in 21m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b01b17ddfc245399d40ad76de870fc0 : SUCCESS in 46m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5811003d59f349a08dd9b2b4d96e372b : SUCCESS in 1h 28m 21s","accounts_in_message":[],"_revision_number":30},{"id":"27c7188abf7fa41133f2ae777439c730f0df3250","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 13:54:59.000000000","message":"Patch Set 30:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24c8afa762449aeba74ea9691ac01e9\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/927126e3bd8b416d980300071c5416c8 : SUCCESS in 1h 46m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/d57b3d87f7e046eab163655224733c54 : SUCCESS in 1h 20m 27s\n- requirements-tox-py38-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/f9d3df168f7547b98f6ce3f358722f35 : SUCCESS in 15m 06s\n- requirements-tox-py39-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/47890e24c1384c57b36f268916087ad3 : SUCCESS in 15m 13s\n- requirements-tox-py310-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/a9123709a62e476186dc9ee6638c6dfa : SUCCESS in 15m 09s\n- requirements-tox-py311-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/9b1c65d4c1484b42852b389d4c5ff077 : SUCCESS in 19m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a434c71840a45e39dfd0b5efc1616ed : SUCCESS in 1h 19m 18s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/1245fd271b4f42c489718b511a1eac01 : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":30},{"id":"233b188f5c514a8c3062bc476e44efbdaea0b2e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-07 14:30:56.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"737925f8d902df0d278648c2f89222db2667e11e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-13 11:30:24.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"829b1b657b6b39c7aed397f25c79bfb49ce10e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 13:12:51.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/927b0b9ad3ba4b899ae48e35f3ef44a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2c6e6cc4344a3fa5cc37cc483ab4ef : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828543324d664c708205c9d80b591e19 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9ba47fde92e4e82961c7ea508aea449 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a540d9f056e4fe6999b2bb4ae12b9be : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cc2f980ef3c487682f8cee9e89a0d21 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e6ac53fbca740799c9155340d607480 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ba4b3a2c5e4404c94bb3a32fc82c414 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e4b1e70a99e547b398a1806f3ea71736 : SUCCESS in 7m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/029b677c605948ff88d66136f6676ce2 : SUCCESS in 12m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e8373c8bb9404cf79e6e577a38373cc8 : SUCCESS in 6m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/18653d1d1212469c81685e66cdd97e2c : SUCCESS in 15m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/59484849369e481288452ff24ee20ec2 : SUCCESS in 19m 28s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/84c5ff5ecc7e4cf996883b372c6971fa : SUCCESS in 6m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/7172c903cb694426bf9e8291c88f49ab : SUCCESS in 4m 28s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9b31fd12bd0c45018efa095e403d9828 : SUCCESS in 3m 37s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a519d5a95a934c08bc0d482938f74b47 : SUCCESS in 9m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2b10718dee104c179ddb91dfa2485527 : SUCCESS in 5m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e453ae94b21641809b747b4b1b1c1b40 : SUCCESS in 4m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b78c65c5bd7c48c8a9a9ab851fba9353 : SUCCESS in 8m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/227e6cb2d9cd4a64a8629ce46c92cad1 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06227c4807f94e0a8d09c1be889ae0b5 : SUCCESS in 8m 17s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e38c941a2a724a1e8e0dc7a3497dd7bb : SUCCESS in 8m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2a91d821ee51410e83ca4c2da564e8c8 : SUCCESS in 9m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e35985f2c15448d18767e963f8d31674 : SUCCESS in 4m 40s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2c1e566edef14acc8f2319b3ec63a6c0 : SUCCESS in 8m 35s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4284cecaa97f4c3295873b4a6bcbde28 : SUCCESS in 5m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5749eaf0ff60403584e5d33745ca470e : FAILURE in 3m 57s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8181f531a2714d1aa3cd6aa873eb2650 : SUCCESS in 11m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1aa2f063a0ea4f27ad848f3567bd43f2 : SUCCESS in 17m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0dc116155c784c4cbcc15b7b3f6d8c4c : SUCCESS in 5m 58s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/64e7b06ae46841c6b8d0b5a99eb51eaa : SUCCESS in 10m 42s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0e7e4201ec94429da321023158060ecd : SUCCESS in 3m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dad7ce5237ec46d79cf7e3bc472e3fa7 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50f2c49f6b364458aad59a3c5da0a574 : SUCCESS in 19m 05s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/72ee39d4023c417bb8e97d26fe853d86 : SUCCESS in 3m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c71c8db7d6474a80a3178f388a477590 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e16d162c919f4d8f965281d76b2825e2 : SUCCESS in 5m 07s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d9537000f16247c8a2ad69eb69313749 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58de96f3983c42aea58e946fa85f60ef : SUCCESS in 49m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33a419455547478880c10ddfab7af58a : SUCCESS in 1h 36m 39s","accounts_in_message":[],"_revision_number":31},{"id":"5bc0af832e775faba4dde8e565eb9411c6e10381","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-20 09:51:58.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"9274977322419ff594f0c31dc4bc1b7ef393ef20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-20 11:36:18.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd5960bbe0f4f59a9e0b025d2aff96e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4777a86a311846ec85b3beb4759ec71b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643e327e4040482084b7ac4017e4ef52 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1459da26024244eb91b4c0004dd5e031 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/676af2ff150a48e0bb5f6abaf52adad0 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5ff7237c64b48c1838f1aff3690b6c8 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bda69185ee94bba80a45a765f528fc5 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3673103297ed4ef0969b867325da8242 : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab56a3ca11764060b846f0c2d1d47f90 : SUCCESS in 11m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1591dfa5dc3c4960bf65c1ceee10d153 : SUCCESS in 11m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20f2a7c4095e41efaad32965f981c873 : SUCCESS in 9m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/016f760704ca481e87f8305e6682e253 : SUCCESS in 12m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/49c261bf54ab4e758fff7b2466992a09 : SUCCESS in 29m 28s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/4a5ceb112926404094b600cc1ce9934e : SUCCESS in 4m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/f7ac1b666f09418b905ecc9099066fba : SUCCESS in 4m 42s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c027c02f79d04164b755af13943b3b20 : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8afb28262bb34175be6a6bee07214126 : SUCCESS in 10m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7f8c6565920d49e394f82613418573d0 : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1152f464d488414ebdf4c511473ffba2 : SUCCESS in 4m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f0865e7b29ca4d1fa6f76052afae1752 : SUCCESS in 6m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cb913f0c9091428b91f2994729d972ef : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a4abbdff93a4a61a262245cb147dc62 : SUCCESS in 5m 40s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/23f4ed9eab3c4d168300205ca1eebaab : SUCCESS in 7m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/437a05237a2f4edfa6d2af7ca6110acf : SUCCESS in 10m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f905a6509ca147b099c60ada3149e22b : SUCCESS in 3m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8c7c4ee8a9f949fb87d7ff6c2af5b6a7 : SUCCESS in 8m 52s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4a6bedbd15a94cb6a418c1f7522773a1 : SUCCESS in 4m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b3537ba14a0f45c290bf331018fe3fa5 : SUCCESS in 3m 59s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/69399c97011f4a9a925d25adb3877a9b : SUCCESS in 12m 53s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2304dd21d0d84087ae3fd5fa4d2c0fc7 : SUCCESS in 14m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cca94067ad174c10bbf4b772b3152af9 : SUCCESS in 4m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3ca295e38838402ab87721267f3fe64b : SUCCESS in 12m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d95082879ac94060bc9639b4ad1d53a5 : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/913c59412733493a93da71a700f3a7f6 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f1ec39439fa41d6bf05d8f37003f6b5 : SUCCESS in 17m 48s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/37862e84e0964864a31c37bbb85cf469 : SUCCESS in 4m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/940990670c90485ba5b41f742cfdf761 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ef005c670d84fac8a27dc05038287c9 : SUCCESS in 5m 45s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/eed82573a5ea475f87c6d29673767044 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a83d0ad8777a4427b99012afb7463def : SUCCESS in 1h 02m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbc3c534f4d84bf48d777e42568d1379 : SUCCESS in 1h 43m 07s","accounts_in_message":[],"_revision_number":32},{"id":"b13c9168fba5abc30a89bbf9b009c68283041f7c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-27 09:47:23.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"ef88a4eb3a5372b22cc06bb10ce6f46036045ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 10:58:38.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f40abd8287549bdafa9f0899f79e4d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7520a814ca48ada33bb265147d76e3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decdc2a06fe44ab09f8ea48ca84196bb : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621c1d7bf70043348f6d8fa7ff683be2 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa3d446c7bc048f49972fb101dc474eb : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/736e4567c3444427a2b7a86cd7b3f947 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/440619dc0b38406d87c20188d52783a5 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bc33be35875491889c86168fe9d4f85 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/959dfd8aef344f3e97ba1308a88d9262 : FAILURE in 3m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/899722c54f6142e183d06fd22e5de5b9 : SUCCESS in 16m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8403f7f7c5ad425b823c288948ca88f8 : SUCCESS in 13m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b493bcf997249c7ba99e249a34a5b3b : SUCCESS in 11m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/df3749f1549347e6839b9dc4e1bdfbb2 : SUCCESS in 29m 56s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5cb3da93cdc443e88032c09ad6e2a030 : SUCCESS in 4m 44s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0eb46365762f4b298a03be1194c0c74c : SUCCESS in 6m 02s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f85a9913243b43df87de836c3701b090 : SUCCESS in 3m 09s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/213a1d18879743edadda0e7911dbad6d : SUCCESS in 10m 52s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/fea374a7089f42f2bc2979795dbb98a8 : SUCCESS in 4m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/125415de0bb243cabd6ef5bf6b3a94be : SUCCESS in 5m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7fbd125242bf4d0a9734dd218cdad6e9 : FAILURE in 8m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ddb4539bb01c4d139b396ffe5e082d7b : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ddbf5b091cd469fa8f5a2fd75d87486 : SUCCESS in 7m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4008562183214287acc7b778ebc274c1 : SUCCESS in 5m 52s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0d485847664e4a858a3a38296a418c42 : FAILURE in 10m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2526c8b1270c40279ceda0d0fbea06b2 : SUCCESS in 3m 40s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6bc61889e26a4fb5a565b57743ca3dea : SUCCESS in 8m 18s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/88083784e2e44f4d932cd32e0393b4e7 : SUCCESS in 3m 51s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0bdbb710fbf9489995915d8f1247ca6a : SUCCESS in 6m 10s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/14946840c8ce4758ab3ee94e43f2df7c : SUCCESS in 8m 34s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/69b6cb58366942f89c74174740850132 : SUCCESS in 16m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3d741c2b08a4372bbfd50cbcad2bbf2 : SUCCESS in 4m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3333fe9cf2ef45f9bed363c7e6d137a5 : SUCCESS in 11m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c3d1b05aa4124b8aae7ab6af0cd6de94 : SUCCESS in 3m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecaa1ee6aec743f6b8a3b7ac9ef2f3c2 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b40880f6d864c5b8d9f58d2e9ce23ef : SUCCESS in 21m 36s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e4f087cab75445769775ec7b3c734cd3 : SUCCESS in 5m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d3ffeecfd4af4443b9ef804822e0e3b1 : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/defa3775051042b986f90966da1ce08c : SUCCESS in 5m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/6b9e1b4cae6d4dd7b348c2fc579f2903 : SUCCESS in 16m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cb51a047f094def96bf8db3d184f6eb : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/129160ba6cab4ed1989f061c6a5309e7 : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":33},{"id":"696e87375f891110843f17826ec5a3f0ab60ca24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-27 19:38:23.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"f397abde43f0cb799fb06e79e9ac3b2a019a89fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 20:48:43.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d941de6e22d14f16a410fcba006049b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589a6b0c583e4c4cb5d831f9be63eefa : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3d82a90d514e41b7715a7543a1266a : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7da362c6f9134e3ba091c7fc9d31c144 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c733a225fbf140238b74148a0b14ec93 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36018742b05c4c7fb2cf8bd4275dedfe : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c48d2b5b4df4c488c38491ce9254dfd : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52f25337d1734f59899775c8361db3e2 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef2ffe2ce0344ab4bfda3b59eaf3a3f5 : SUCCESS in 11m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5da5711140594aedbc7064909caa15da : SUCCESS in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0bf11f5675224a01916e37c48d5c7281 : SUCCESS in 11m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1bfc598ac0cb47a59920d0e5dcbe4ed7 : SUCCESS in 11m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aa8bc5be5f6e48db825e17106fbb6426 : SUCCESS in 20m 18s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/cbdaa96035a049799050b643bc74e7c6 : SUCCESS in 4m 05s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8be03eebeeea47f7a737e922d0b1dd25 : SUCCESS in 4m 11s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/16844d5d91b34cbd9e57fb657343ffdd : SUCCESS in 3m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/05ffea42f9c448ecbcf1ce568630fd55 : SUCCESS in 8m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4186b28ab6cf4d96ad24e78b2ce799af : SUCCESS in 4m 21s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0ee2f308476d46a48c4e59f231d9732f : SUCCESS in 4m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/62d2bb1f919442758d396c627b82d7bb : SUCCESS in 5m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/644e04a661bc441991c991197cae306c : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00f71b806a3149f6894317568571991b : SUCCESS in 5m 18s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/47617ff08471430899248516b6a66290 : SUCCESS in 7m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d474954fafae4ff9a82e521be2d6a70a : SUCCESS in 9m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7205b3018c424800ad260963193e9321 : SUCCESS in 3m 32s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/eaf3d4f9a5a040908e0ebc8dcba35406 : SUCCESS in 7m 39s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ea16286209844cc0964d482f2ddf6786 : SUCCESS in 4m 36s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2509a86f563b491aa917675a64cea832 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7e72aee231b84c43aa52a5bf64734f15 : SUCCESS in 9m 17s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5eb96dbb76404f81a6a86526af3198fa : SUCCESS in 21m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0a5cc65dcf2649ca9268dbd4d82a9181 : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/78e34503ba5c4288b82bc36bacff2500 : SUCCESS in 12m 42s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/27be759f5b3b48d59be6a7cd54ffb1e5 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/915a410357964dd1b53b5ee1f34c9ad6 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34dbe8cec9c34d87b1d82d5f907ca516 : SUCCESS in 25m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6d5e7ca6ffdf4fa6a382ff1002dc0856 : SUCCESS in 3m 48s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1a7855e9432d4a37936889aaf9b2a9f1 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90e92a70912340599254f7f099889475 : SUCCESS in 5m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8d0a7f849b9b4191b20531a6ce5adfbb : SUCCESS in 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69563978436e4219b71ec0f36298a6df : SUCCESS in 58m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad4d2fbcd48e4ded88dcaba6dcb5bd3b : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":34},{"id":"6c544f31e256a47bea1818cf2e082cc7c556305d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-05 11:41:06.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"979fcdf273c7cd1a769b87b9d8ad9a9c80ddef80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 13:06:24.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9110bc0de0754666a963be65e43b7226\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c8daeb51314a54a2bcfb4166701ad0 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d443d1f7c2a04dfa9a7d41cdeec4ea47 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f4d43aacc5d4b0195d4e81383c4dbdd : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2319495360e34a6aad183b7a54511a97 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5949d294ed1648b5bda70e5559187deb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7f5fac17302473e95b51bb2e8b7a9ab : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca0845df001942938108582fc8fca8b3 : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7bae4a9fc2ad453c9383e6f46b022674 : SUCCESS in 12m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a571b45288b94e348b3a3058d84e12b3 : SUCCESS in 21m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f3275480105643f7af2f6344e6990aac : SUCCESS in 13m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7257c869b84f4c458b6cf2ce328b00ae : SUCCESS in 15m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/51dd397a590b4ccabddb292242dcbe8b : SUCCESS in 30m 17s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fb90206b55b64982976a913d0ca59a13 : FAILURE in 5m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/39965836f8cf442e96b90068841fa300 : FAILURE in 4m 16s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e1747acb7dd64a198c995b86d0ba5ac5 : SUCCESS in 3m 46s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2a50c0baf62f4efa91bf164c55f0a25c : FAILURE in 9m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1f183776070f41c095f03dda3079e55f : SUCCESS in 5m 03s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c772025ed5b541fa8ba513d754029ede : SUCCESS in 5m 16s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/08b5a6e940cc43849b3bcf226472a020 : SUCCESS in 6m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1a6d3a40d3f04b29ba88d70804f4cb91 : FAILURE in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e41539b16654015adad14e785c59b08 : SUCCESS in 6m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/185f06c5028747e48777ff23aec8342c : SUCCESS in 8m 55s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/416539bfe96c4d98a4811def499479a7 : SUCCESS in 10m 53s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0c1941a115434b68bac6d0e3a09137f6 : SUCCESS in 3m 33s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1b1c963e780a4936ad15c00dec23b074 : SUCCESS in 8m 14s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d9bf7eb1ceb249a0a5cfb88be774ac95 : SUCCESS in 4m 14s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d2fe3ffa5221421ca043eab83644ae6e : SUCCESS in 4m 16s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/af9f30ac922d4efe81c91fb355e13667 : SUCCESS in 9m 20s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/30c0e29eeca048809e9860bb38fc5534 : SUCCESS in 20m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b28dd284452d45129d8b303a9b04a194 : SUCCESS in 6m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c5d26dac655e4612978e0f04cf038ad4 : SUCCESS in 14m 09s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f0844f3e673d43c8ba4dad283dd0a0d9 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf5304e6692a4f0fa4eb932e58ef7637 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc4ea562077d4c0c9430ec3d2efc0013 : SUCCESS in 21m 31s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/aa7ed117433946b6bb308622b889e6ba : SUCCESS in 4m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1b6ab83c8ddf4328a60f078ff7d0c8a7 : FAILURE in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba9469cc3a934a70bf595aba41a932a1 : SUCCESS in 5m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/645ae0846f974936b45c7a448ddb7ba8 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5574fef2df444e62b9c19aac3c8dacee : SUCCESS in 46m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95558e173c6c402f83a296d609aac75c : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":35},{"id":"2e0ba0f84c9a4698e4b538e8ee7d2a089d11b409","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-27 12:31:52.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"4c25dc0154920c565aaf91680c96026b232978c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:46:11.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b7cee942c534673923f555315a90d6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1842e37dd6e841399020b0e5119ad333 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8600540db05466faa1ebb8bd095e0e3 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f199470dac2c4cbf84c0d3ddcf14eeb9 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb04c58130844ceaa4d8ca295236b17a : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5abaa299fdc142489919228edae95a9e : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6edddfe4313f495f89b0fe700c44a922 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20763383842949df8a1f64f78ca3cdad : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bc5624bf73e4463bffd9c109e45523f : FAILURE in 3m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a770213c39e34628a4a8cd636c83083b : FAILURE in 4m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/98d16226feb54e7e909bd5d0a233e51f : SUCCESS in 12m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9da0ee2a4eeb4a0e91d5ae6a8f2520aa : SUCCESS in 16m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f3038c2c0e314bf78d057687eadc7177 : SUCCESS in 30m 10s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8f3fa850497b465281b8c815378c932d : SUCCESS in 5m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/715a6bb8df7c42a8bda3c5664228414c : FAILURE in 4m 32s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ebd31e7712444b1eba3d5ab636f613de : SUCCESS in 4m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/faa25ba4f0484800b1e483a7ad647d89 : SUCCESS in 10m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/76a28ca3a025419bb172714062204bab : SUCCESS in 5m 27s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dd081985af9d45a3b7420fa44737c5c6 : SUCCESS in 5m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/45bb0a526fe44106b6ba6ce4cb54ec75 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/632a444cc66344b38a85e1145c820e97 : FAILURE in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8eef40c826e94b878cfc943081cbfb7c : SUCCESS in 7m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/545098fff70a4c0f87b44eed4aab4986 : SUCCESS in 7m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bb0cabccebaa41bf9721fb41f6e8cb60 : SUCCESS in 10m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/dac776d28f5347cb9b90b9bbecae4d60 : SUCCESS in 4m 38s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8cd6d901ebc643aeb769b6c49e47621c : SUCCESS in 8m 45s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/0f4b41ddcdc948b48b41b3f436e10fe7 : SUCCESS in 4m 12s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a14fcd8186fb4ab8bef558a0bf41c501 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/79d823ea16c7431c8ed3d043500ba949 : SUCCESS in 10m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e239d94b7be241b8abab2b306917265f : SUCCESS in 22m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a9ecdbe23f9440a9802cdecc8de0f5d : SUCCESS in 6m 57s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/48ed8d99db4f438592d5679b28299c1a : SUCCESS in 12m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0745781f36c34a858628b581c5121ecb : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c723b98a39140108a587a75c47fe7c9 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/887cce356d1343bda66f1eb650f5b269 : SUCCESS in 26m 02s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e4eb6d827a974e88a0035d91df2ae8fe : SUCCESS in 4m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9924a90ba5154807ba85594957ad82fa : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/20ca0615cf6e4cb8b77e4f195752a225 : SUCCESS in 6m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d870936bee8b404ba40439967827dacb : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13e8248fdc25459bbbd837359bdf4698 : POST_FAILURE in 25m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/247acdb268744124823b747aeb9a61b5 : SUCCESS in 2h 06m 10s","accounts_in_message":[],"_revision_number":36},{"id":"ea025402e90c2cff2a3dbe509b60f911081a3628","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-27 15:35:06.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"6b30f39008efa24a765d712bc27bc2d40a97f6f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-27 17:42:07.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"bf7a4679934cebbed8d4847ff0d9ed77053ce3d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-27 18:28:09.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"75a446deac695797cb1ff47b5736e87180d6de4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:51:22.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a3f9edd41746398102ddc437eda7b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52c9e0412a14d62b36b5d2a40133863 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4df17889504068adb2b9cd3e78bec5 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbcbc114fd1e4238943121d228f0eaff : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6edd9b67d4db4162af5271fc4e37778e : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fab826a6645041a681ad14cda0ee5d68 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7ea18ebc9884e09a47eaf1bb02610ce : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbb162758546434b8e9088011747b21d : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/49803129d6484e9595dcee6f921a1b36 : SUCCESS in 12m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c453ea7b5b6a4448b0a5b9e6a5b19a90 : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d1ecaa9b860f481aa1c0552fb6445f61 : SUCCESS in 12m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/add9bae7dd984406a5b86acb33fdd77e : SUCCESS in 18m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/75f7c7713aee4998a909427dad13373d : SUCCESS in 27m 45s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/02f87cb56c234d24a2834a0ab069578e : SUCCESS in 6m 18s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b155d146cdf644e08cb8cfc3ebc5efc7 : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/134f3690e84244cdb8031d31c439b3cd : SUCCESS in 5m 28s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/91eed137e238438e9b28f02306c4139e : SUCCESS in 10m 53s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/adc8e8d43f8247b38db098763e393562 : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e3c83d56bcc143b099d9575cf610904c : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fb9e740ed0c048538277747b196fe70a : SUCCESS in 6m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/89ca68c305b94d0a986f6cdd350d1adf : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df2b9f59fcef4c41a5a58612ea4ea41b : SUCCESS in 7m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c07b22593855431d9ad2e432af54535d : SUCCESS in 8m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/31210a1dc9184f0d832a3362224b3ad0 : SUCCESS in 9m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2dce3ae8024843609375817d59ea267c : SUCCESS in 3m 32s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2598318ed768402f99ea0e3202487e79 : SUCCESS in 8m 42s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/de3cd2aa0ca8457985ed04d9906ccf4c : SUCCESS in 4m 13s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/94d5e70599fa44e08f4f0f654b6219ec : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/976d6b02f77346b989099b68bd9942b2 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/10ad4877ee8d41c7a3cc674fef6fa61a : SUCCESS in 17m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb5243736df7451a87082c01c5209506 : SUCCESS in 7m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ee57844a67bd49e1a4eeb13e45540228 : SUCCESS in 11m 31s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/50b5eaee097e4820b277fdaf6a21631d : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93bb598132904279b61981d917eccbdd : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87a27b5e90b04ae39b73e8c2f52e5ab7 : SUCCESS in 23m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b40c5dbf0077425781ab6271ab698c23 : SUCCESS in 4m 26s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f8604579ec98417188818a1ef2cb52f6 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e05d301acd544d6af273adfa02fe8f4 : SUCCESS in 6m 04s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/69499532b63a41068bba5109cf95b543 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a2d7c17996a4dc8aac210cba58294d6 : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70335f05bf314368b703bb8722ba66f4 : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":39},{"id":"a27ee087aea33e0b0c50cd833a0bbf0dcec13d3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-28 08:29:20.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"e54b4b43302fbabc0b6fb1223cea2412010e3cbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 10:14:57.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11abb20ae06b42538101daa26345e498\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b2298c9bd74e21aabe85120fe156a5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563307815ac448238d5afd25010c310c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c89f62bef16e487e9456eef9a0f8d6ab : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc8cf0c382a34ace85722320fc598855 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69a4fb1d27e6495f9152e10c7462d783 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d78013eaffe94668a614cba7ff54c950 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bbabcc647ab4c34b0da053a8638e2dc : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63dbcfbd787649fbbfc74a66f526a93c : SUCCESS in 13m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ec690314ac343aabe9aab2451e86da9 : SUCCESS in 18m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b84e9492e6f440af825b5b087cc70f0b : SUCCESS in 12m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3109372755334dd3adcf05cf3ee4c5ee : SUCCESS in 18m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b1b4e813b584445cbfc0fb71b338abe3 : SUCCESS in 36m 26s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c068a18619da442392410a1e057d3929 : SUCCESS in 5m 06s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b947cd09f7ae420a9df0664522979c36 : SUCCESS in 5m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5f47e2b38d154e7992e4c7004bfd15b4 : SUCCESS in 5m 50s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/91f918506b114ac58230f6823cb7c8c4 : SUCCESS in 10m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d5f209370c604a2aa2b7dd48d5fd281e : SUCCESS in 5m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bf2a07b324674fc6a26df929ca05dfd9 : SUCCESS in 5m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3983fb60476a41428a1bbdba21d37edf : SUCCESS in 7m 09s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2b2587e5fb6243a3b6b0efa87eb1c51d : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d787be797db4576b5cf0f47f44ba6e1 : SUCCESS in 6m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3f93d97443c043a1b433e7265884993f : SUCCESS in 9m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/97d36667d24147bbb215d355a9cb06e4 : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/85534cea600747d0a9189af2be243ecc : SUCCESS in 4m 25s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1bd51a8989aa45e891dd3272b80c4571 : SUCCESS in 8m 32s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/6b8d184d86ae4563948960a6b0f46ef5 : SUCCESS in 5m 52s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ccee197d88a64a03a59023b11ddb1658 : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8be48ba1855c478d8c52e9f19fd94649 : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/30700f3c2d7d4915bd5545f7c0050197 : SUCCESS in 21m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7651adc6bf114d2db58df097383c67e4 : SUCCESS in 6m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3dacc561bbf442c7a18a9c68a978b1a5 : SUCCESS in 13m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/97d3bb60612348c7ba49c6204769ad5f : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f315ef7d6ad0462e80fb4b574f5a09f0 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6107444137d54ddcbcad250f8823aa75 : SUCCESS in 22m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/57fca5feabb44beca25da37b2be579ae : SUCCESS in 4m 12s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e68a40c4c0d345e1bcc4e4af277e5c84 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f69af7dc9f149fb90f2dd441830d7dc : SUCCESS in 6m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/141418b62ce5406f8b1daa57b90a5297 : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b2dc3eb52fc4108b83d4b53532e6aec : SUCCESS in 58m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e58761ee59804ebca4ac264e18acafd4 : SUCCESS in 1h 40m 35s","accounts_in_message":[],"_revision_number":40},{"id":"b177cd8c285e1b630bcaac8ae8df4fae6e035662","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-27 11:21:07.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"3000a78c5a8452a68b89cbe57a4d57893cdf2482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:15:53.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d43f55dbfba948ef9989f6ea7ce8e10f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457ead831ac64fe7972aeed9f8d3a787 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5d852b631a4ec7a552a0b9e2d20264 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c70ba290c9e8409098a0d24a276acf8d : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01c715db6a0f47cc9e2633ea4a71a8df : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db5d881bdf2d4333a413ce9a780e2513 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb7d8f0fd27b4237880fd3803cd26f05 : FAILURE in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90a3dffcad17468fa2bdc848c9df6d71 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b0ddfba8a6442509f1b21c5dcd72116 : FAILURE in 2m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/244302798570440d933ea8aef8801c9e : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/44b051d931dc48dfb6689b9f483c46be : SUCCESS in 10m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/856e8ea8966b45448cededfe19714ee3 : SUCCESS in 16m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c5c8307ccbb046e7963ee4d1f5999b15 : SUCCESS in 12m 54s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/d3503dbd317f4a20a41d195658e5bd91 : SUCCESS in 4m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/69834d7e72f749299405232b19e31d0a : SUCCESS in 4m 59s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9e82fd0b5a0947b88e943a655b4654fe : SUCCESS in 2m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/92f7a930cb4c4748930e704d0d4ad194 : SUCCESS in 11m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c78264fa56694501b18b19cc478ebc63 : SUCCESS in 4m 19s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a0cc3a9f55f14931affa6b53c0745804 : FAILURE in 2m 51s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9568b61190174ddc8e6f5171d53eeed7 : FAILURE in 2m 17s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/495c7ca6038741eb82d98e4d0c3d35b4 : SUCCESS in 14m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63a481b7a4c14d5c805cf99b5092c773 : SUCCESS in 8m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a3d8fcf549314a43a1237885c41fb0e1 : FAILURE in 3m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cb7711086c5a4c3689132947c73c7949 : SUCCESS in 9m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6091e5322fcb447a893a777214e3e050 : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b7c49efb0a0643f085b24222abc37cb8 : FAILURE in 2m 52s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/32a9e9ef13b24848a0832a125e9dcdfd : FAILURE in 2m 12s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e39a80545d3a4808b208fca2cb95a9ad : FAILURE in 4m 54s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/5226cb5befde4dc98c9589eedf3b825f : SUCCESS in 8m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1b2b54551ba947d9ac3656bdca5c7698 : FAILURE in 2m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/63139ee1e8154a07a9fc6284cbd23878 : FAILURE in 3m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/462a6f371ecf499eade569c3d80d43a9 : FAILURE in 3m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/cfc72df20f0647918a728a27ca9d2730 : FAILURE in 3m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7df98bd948a946a69b63c03ec0872dcc : SUCCESS in 2m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e96ae89b3ad46e69932ac985fc5455f : FAILURE in 3m 40s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d22d2a9b6ea2457a9e862b64ad5991cf : FAILURE in 5m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b3971e6922564534befc3a4b1486f665 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fd1a773d3e349909f7442272728b503 : SUCCESS in 5m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1baf4a77d3464adea12c96e656d68e56 : SUCCESS in 21m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe1f846d7d2648f6b540fdf8dc0f8082 : POST_FAILURE in 10m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56fe22e0b4294be7a5b7fc3de5042256 : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":41},{"id":"92013d50458d1fe6869e0bb353d4ff8112cc6d98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-27 12:18:59.000000000","message":"Patch Set 42: Published edit on patch set 41.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"0114f01bd793db0acd18f4e354c1732276bbbd9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 13:02:02.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b4d6cadd15347138f2fe70b0f01e8aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fe886a51a94f538edeb3e30cd8af39 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d543b7c474d1443a94531123a1c93e07 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9366298452f64fcf845b349e6fa789d2 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb02cda8df6a4c339a9a4e33ad4eace1 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81495de4d766422a87b67bc0a7eb4fe8 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bc35d056d404e9abce17218a1b62a4b : FAILURE in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d876e8ed788e459fa2c09277c3d8bfad : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f9f65451eaf4bd9896dd83f9a824147 : FAILURE in 5m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/27ca2dacc3474fdba1e1a160e98eff60 : SUCCESS in 16m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cf41ed3e8e494409af41ba0cfe8c5906 : SUCCESS in 9m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3863bfcfe5de4f05a569248c5d1b3f5c : SUCCESS in 9m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/617ed2c6872e46919acdfee05f7fbfc1 : SUCCESS in 22m 14s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/674ec3fb8a6244559beee73ddb4156f0 : SUCCESS in 5m 05s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ad9345d22d2846dfa26fd7e41b96dda2 : SUCCESS in 3m 47s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4d246179920d4a2283a1fa9e2ad491ef : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/90dffc451d9642df827646f2beea56c8 : SUCCESS in 12m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c6d127d84c37489a9285fbbfe2e6f201 : SUCCESS in 4m 14s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f5e51db960f142d9b16a7b25c3aa0df1 : SUCCESS in 6m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/778a9c6951e145a481f9ad1566e57a42 : TIMED_OUT in 40m 45s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0e9aad490a0a444da646b6c1bee29fd9 : SUCCESS in 14m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f163dc21012c4bcdbbda88a4f1d97de1 : SUCCESS in 7m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f7f216a207434ae8863de66da397a91a : FAILURE in 6m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/3a8987a08cce40f3ad587c3c55683df2 : SUCCESS in 11m 46s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/bef8d7e0c0e34e9c93fa3c56b2df00ed : SUCCESS in 3m 09s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/12aa621788904023bbe2da79acb24da1 : SUCCESS in 8m 49s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e49c6dcd7a5d4fe8a3b67d62a855cab2 : SUCCESS in 4m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/88fd31a53ed5430fabf266076b506c7d : FAILURE in 4m 17s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b21eba46de3b4d778a859baf036c29a0 : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/df810540f1d54d19ab62488244125404 : FAILURE in 19m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6479078ebea0413c917dde2d9e0cde87 : SUCCESS in 6m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2e16023654644e3abb646507afe84696 : FAILURE in 9m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0e8c7104f96b40a4aba886bc377a9351 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8249afcf85cd40eea8b03f029455be32 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2f173be0c5b4519bceb9c278445c88b : FAILURE in 20m 00s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d372ef4787d64e9692bd1b5cce77ee06 : SUCCESS in 4m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7d942fecc3644b1ab201338f98d810c2 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7d978de0c1a477a852becd9a414047f : SUCCESS in 6m 25s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/bd8742418e174dabaeff56009a5cb068 : SUCCESS in 22m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a433dbb458eb4f04a3cf7e7189fa4221 : SUCCESS in 17m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/484a93c3b0414986a7918a96125f30e8 : FAILURE in 24m 23s","accounts_in_message":[],"_revision_number":42},{"id":"fcde151b659fb827e08ed9b54dc2d8dd9f097e15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-23 16:00:16.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"2a91da79cbb43b692f6d2b94e05fcc5117121739","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-23 17:03:16.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"dbe88f3db69b9fe09a0d85880cbc8a1a3868365b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-23 17:16:16.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"c7fb837e0654030a3aa04ba39d1d6c91d53362f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:17:22.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6be081cb421400db3c254d9d9995482\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f7912afe444369a95a6c8ff1dd8386 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e5796de0af43fca3451f8655b48da3 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/053162dda4f24fbca76e152fe2ce9e4e : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49e49a3c8e1040dfa90a69831a82e957 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e036f19c34834f4486effaf0725ceaf8 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d3f0a34d37943299790ee4299c441ba : SUCCESS in 2m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3158b8fb922a4814a4b39925d753445e : FAILURE in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e5367f79e7f4fa2ab1de50c5c8d5cd8 : SUCCESS in 9m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5567da71fb264357a0d6067faccb4665 : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/40414d78c9964682858301649ac74030 : SUCCESS in 16m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aa85a387295b48da869600acc64f1ec1 : SUCCESS in 18m 15s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a33dc514d3634bc88b460a2ea9ffc8d7 : SUCCESS in 3m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8a9f2aa92de64ab2ae91c138b2463b98 : SUCCESS in 6m 15s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/29cb307d6a3c42c1b15fb172ac3fe545 : SUCCESS in 5m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7635a2375a5b4d7bb6ff7bf8ba3ed02e : FAILURE in 11m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/30a1df2f4347433ea01dff82e80488ae : SUCCESS in 4m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/42552e59790e4b09a14b51296f44ef6c : SUCCESS in 4m 44s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8e321e5d9c4f483ebe6386c43c244a1f : TIMED_OUT in 41m 01s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6d0da5946b7b4df0beb8231b9ad5dc11 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8d459fcf5274786a9fc475335ce62de : SUCCESS in 6m 37s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3ad233791eaa46ea8a26f96344839b36 : SUCCESS in 7m 27s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b4f35032b0f74933a5d89f36b54e7c9e : SUCCESS in 10m 43s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e274e5f748d34facac3cde55cc951def : SUCCESS in 2m 08s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b1c55334344846bfb4b17600e54d832f : SUCCESS in 8m 42s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/66ad354d5f8e41ffa43de8d5692488e0 : SUCCESS in 4m 41s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0dbd9307bbe74b9abb7b1e069d5d71bd : FAILURE in 3m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1943e211b7724be98568e9b05bf276ab : FAILURE in 9m 06s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a2b5f46546d34d109414ac8fde3280a3 : FAILURE in 16m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/08e23f00c9b148eba9e0bfbf6e40639c : SUCCESS in 6m 17s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/70eb8076d8134a3aba38cc74e5c161ac : FAILURE in 30m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e81166085cd64d48a6011bf8c49280da : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1df77df0edc1486e8d5e4fec4ff7d2d0 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cb45f3b47e84688ac6fe3e7a44cbd73 : FAILURE in 21m 08s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/06997ea032424303be30a2288c23edc4 : SUCCESS in 4m 27s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/79890249ede84d04bca9f9a51e914e16 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0ef10218aa24271ad7b11f347ead4a5 : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c5b5ab91932d4a1b93cd4b61e771ebe6 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cebff5b46355465ebd589d13f0cd68cb : SUCCESS in 58m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9808b186f17b4fb4bd0d629d8ab2d252 : FAILURE in 15m 11s","accounts_in_message":[],"_revision_number":45},{"id":"9927e1efe8cea01c656580b19bd48986f8e56b65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-24 14:53:58.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"70b62e47a9ef45c013c83b1288c5e5b49750763a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:23:29.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1036a772794d4b07a0319c7e0d3df7b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d30e92198de4a34b96692fb9194d4ea : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9e4ddeaa134ad094a2ef1f0fb73dae : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0fb39927d24b19a32ea0652dba103a : SUCCESS in 6m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2718f61fd91428d954199227221eef2 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a72c30c27136405493c31d5985fbddda : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f5ed10c35eb4fc295e3b073a43f9053 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39b55ddce1f8479f90d2ed3791760b04 : FAILURE in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fada7b92abab4685be217ebb29d05240 : SUCCESS in 10m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bfe8ba24b030484b8274a3576e128bf3 : SUCCESS in 15m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6dfc54003978431094990dda1a1b3320 : SUCCESS in 19m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/171daf373fff4846b6f3b0c99904cc69 : SUCCESS in 27m 10s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8681ab6ba8184774a640fced4774806f : SUCCESS in 4m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/9f60210b1ea9483d80ea68560fcc1142 : SUCCESS in 4m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9941907e24d04c09adda3b0bf03c48bf : SUCCESS in 3m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/011cdfcb0a31495eaeb7dfcc358929e9 : FAILURE in 10m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/016a027150c4465794855dc166e60aaa : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/41920e4b64f9487a974fab2661d3b380 : SUCCESS in 4m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b16167d8e608466da70f9eedb16d00d7 : SUCCESS in 7m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b4a63579e97941f3b83639e34c74fc93 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b77ecbd3b1f545b18d409297369894ed : SUCCESS in 6m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7d2be96cfd8d409fb08ac57e90d8f8e7 : SUCCESS in 10m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e78121051fb94d09b6f982d98ca66a32 : SUCCESS in 11m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8fb6db8f86c3499a990882c868c7f096 : SUCCESS in 2m 07s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b6dfeeef2f3240a18c8b877107467d9b : SUCCESS in 7m 14s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/793c66fb7d2a4825bd54d6bc9ff0da09 : SUCCESS in 5m 49s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a435e8070e14476e9278ed9e945dc9c6 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9a7e926924f449b5b556f05f9dc5743f : SUCCESS in 9m 52s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b80352c7899e44b6a1444fa4499b908f : SUCCESS in 23m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1a553522d7c544df994ccaad844af969 : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d9cc1bb059a142c695a7846c71585bd5 : SUCCESS in 11m 07s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7080e9123253430d95fed7723997903e : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ced075aa72344ab4b3560b57a32af12b : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/256a5a9fbfd441ada618d4f8dd6a6365 : SUCCESS in 20m 13s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4d4735d8408548bd9b12a23c3043918e : SUCCESS in 5m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/80905dca854644989494d1394c49f923 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b443a324215416ea6df015db348aa6e : SUCCESS in 6m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c107d117b45b4ef4a4b98e8356a8bc0d : SUCCESS in 17m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f52a5fb034a54fc2be0726d7dc3d7384 : SUCCESS in 1h 06m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28885799b1e64ae893943765dddfb2a9 : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":46},{"id":"e21198d974e27ac60f6c98bcfd93350e06de337e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-24 18:06:24.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"d0748929f43367b93cb5f4c3194b7e06497a0b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:41:10.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e56518a17e5c4e17b256db69fb83bcdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ad1303c1c84ea78f2df38a90ad51ab : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a12561f3e414d3eafcf4899359c1bed : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbc14ca3b0d49b4a038f955b1938655 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/615139ac026144b597f723881fe570b1 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cf0ec120bb3405b8660de000598df3f : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58396f19cdb046ac9d1580dc253b3dfc : SUCCESS in 3m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1c72bbc4a5e3411fb4aa1aa956613bed : SUCCESS in 17m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c339863df8a4e69af8bf3c12b4290eb : SUCCESS in 13m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/83cfaaa5758d475f95d97146b2413a75 : SUCCESS in 16m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/897e2ce951d246c0b7655cd75c6aeae3 : SUCCESS in 23m 10s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fc9a86d2a49b4080be06de20687310e2 : SUCCESS in 3m 26s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d02c04708cd5432198404cddc71b02dd : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/10a7213d69db45d2b859600daf0efcd3 : SUCCESS in 3m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7a2e45da77b34397a3c9799262a3a8da : FAILURE in 12m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/cc40c74875694fcdb7f6c3559b3533c4 : SUCCESS in 4m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/94bc96f7579c41ce85f80b018d7939cb : SUCCESS in 6m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ed08a7a406034109a3b0dbf3202f2c0f : SUCCESS in 5m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a7a33491175b45089623101e2480dc4d : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28616e9a40b14353827e0b309cdfc053 : SUCCESS in 6m 28s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/94a2680c319b4fe78992af82c8ab514f : SUCCESS in 10m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d8ccbdb76c2f42b3995eb657a3b8bcd1 : SUCCESS in 12m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0ce581ad0c2449c59a38067329682cbf : SUCCESS in 2m 27s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4d201461c90b436b88d50747b9143378 : SUCCESS in 8m 19s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/8bb484532ed347d7b6237b741788cc1b : SUCCESS in 4m 08s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3fa7b952c91b401fbf38fe1094616dd4 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/29822efdcebf42bd8519354c68e487a4 : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/82848a13210c4935ab4c7dfef4de0977 : SUCCESS in 18m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8dab978d36194f2bac5ef36e11ca6249 : SUCCESS in 5m 35s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/479109302688480bb9ca574bea82bf87 : SUCCESS in 8m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/428fa26e2036491998561b6e252ba043 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ddd70dd553c944b5915c40f2e86717a3 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4ed5cfcdf3147b1906cebe33667eb23 : SUCCESS in 25m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2b1ec865616d43529f1b74b5d7d4a15f : SUCCESS in 3m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/62598227b2d54590accd71f9f84b2d2e : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7515af0486654cf8abb56f9385c0486a : SUCCESS in 4m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0e9ade61d94345e382ca9178cea246ef : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2249f8b357b54e1b98f320b2b6527967 : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc18070844d541d1be9df9d4a0f05eea : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":47},{"id":"db2654943bf593b110f6b6b50a4ce492a98ea626","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-25 07:44:21.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"dea6966f5f0b5f91bee2f5f50d9a8f520e4a4651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:12:32.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec9eb221d41049d48a86b1daaa46b7e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5dcc22bc23c4175adbfa7cc629a089b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95836aa9a57e4e50801f785626f9ac89 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5fd925026de47f8a83794c341bb512c : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/468020a4975d44baa25aba21a56ba396 : SUCCESS in 2m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60b890c6009a4d598721be2d1e924f79 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c62c95974394409c8b831fb4f5df0dee : SUCCESS in 3m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bfc5c916483945069bcfe11bd030e8fc : SUCCESS in 15m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ad848a04d2bb4c22a486b171581fd24f : SUCCESS in 14m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84bbee4f45304168b80ee1b18d24cd28 : SUCCESS in 14m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d5ff179a3eca423486facb20660b99f7 : SUCCESS in 34m 43s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f2e46f34dbf843a4afdd07a780593d0e : SUCCESS in 4m 38s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6dee638fd6134a0b8cfbb3a771ec4abb : SUCCESS in 4m 11s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b263584b9dfb47a5afecbbb9b777db33 : SUCCESS in 3m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/52d411cffdc44a8c8d768090aad72a63 : SUCCESS in 10m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/51513f32f33d43b398377f52252760d0 : SUCCESS in 3m 23s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2b2e4de0ca614e4bbee87616b1cdf5e4 : SUCCESS in 4m 47s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c477631741d44ceab2d151ee32ef52ed : SUCCESS in 6m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/35d6e2c283014d81bee4b9035671960e : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29150803c39546c287c43d742193da98 : SUCCESS in 8m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fd389e3bf9b84230bd46a263a289fc4a : SUCCESS in 6m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/93135d6c60ff41c7a39d0536395a749b : SUCCESS in 10m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2266108d8b9e4b3bb4048e2fdb957150 : SUCCESS in 1m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6f1489bbea914026bb4dd894981398da : SUCCESS in 7m 43s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9e74c6ba239d40149bcad50637184d7c : SUCCESS in 4m 05s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3a9edb376dc44db99c97682eaea080f8 : SUCCESS in 5m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7a2a46180bc8496db196efe9e1af9c02 : SUCCESS in 10m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d0b3251a4b0549d6b0f6ee818a7b55d6 : SUCCESS in 18m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5b5e2ed8d6c0450a9bf92c1a209b2574 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8c5cd358e7ae4ec2a1d0539b28ce2306 : SUCCESS in 10m 02s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e4f3cd66f1b14d02b7b222d14a3e3590 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c1ed66a912641289d7b75a772c7d624 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b913c004ce54411b4ac01849a0b5ac3 : SUCCESS in 25m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c2762a22a19841c2812aae97736ba1ab : SUCCESS in 4m 15s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e02df9ecafb343dc86822047695adac4 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cb096e7db3e4f18825cb15f8a46b60e : SUCCESS in 5m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c8889d30f15946859056e7eb1225848a : SUCCESS in 7m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b1826e5a5d44a07a789a3a60604d42f : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9408b10ae9844edb9b685f6a91fa1120 : SUCCESS in 1h 26m 00s","accounts_in_message":[],"_revision_number":47},{"id":"78b1f3c951d7bdad7a9accf64e3adff98441669b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 06:19:43.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"296b1cf5fd324aa1f95f03e78ab223c6eae55638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:36:22.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea3abb914b4f4bd4b8220cf6dcb53315\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bad3e30da040619d105e494d016a2c : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb115492e534e0099a88d59158e75d2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53ae98163ea5483bb937d20a309a0633 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4b5b81b6c624bb3bb07cdaf799fe0d1 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c74bf189310f4612968f22f0f4c5a9a3 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a38051e74e448f9b61fa32efced415a : SUCCESS in 2m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/66d3a6b576894e93ad32126d9dcbbaa9 : SUCCESS in 13m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f91df9d9591a413d9af0775e417ebd19 : SUCCESS in 11m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3576db2aae743bc8b252e4f8f7c6745 : SUCCESS in 13m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d0d1898c5bae45bfbd52c31b7eb8d1ff : SUCCESS in 19m 12s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7ab6029055054585af93f2a0860978b8 : SUCCESS in 4m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e608d24712c34d8ebd18eaf147c59152 : SUCCESS in 4m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d8fecc4333834d0d90944d457f6daa1e : SUCCESS in 4m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/01525ed1a0ba44c68e96f870f164b709 : SUCCESS in 8m 51s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/56dbd7559e8a48f28afb5b79af15533c : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/89a7a3cfd698433a9f32864efacb092a : SUCCESS in 3m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e0029b986b64462589105f73f7f17923 : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/237d228dcae045e1ad85ef5bfdeb10d2 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9754ca4a2c74004b4e02223ebc56410 : SUCCESS in 6m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3a8db4672ce341b086cebaa1dc56ca8b : SUCCESS in 6m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1f8bd4f2901141c7aa32f29b2cba3a5e : SUCCESS in 10m 29s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ba53809c325645adbba9bfc1d11f57a2 : SUCCESS in 2m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/470981a160c9409595a435a1e7c8de6c : SUCCESS in 9m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/af628f5ba8704401b3494db8522efd07 : SUCCESS in 6m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bf2348b4264e497fab8f16c1f3457744 : SUCCESS in 3m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0b8158fc78384cc8a6b1f8d203a6518e : SUCCESS in 11m 27s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/612f7abe572740feaff82bbab8a55bda : SUCCESS in 20m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d951029a044540c49436a09be8703e13 : SUCCESS in 7m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/15330642eb2c42769fff55e668901cc2 : SUCCESS in 11m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c81ec42cb2d74d8b967018c9b1306aa2 : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d09890e71504b77a5775a57eb8db8e2 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b905675cd2b456692d40bc9fd1a21b2 : SUCCESS in 19m 20s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e1a9a160190245e19bcce5ba97058c8d : SUCCESS in 4m 25s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0b15a576d99f42d7a90bc0b50a20bbf8 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e10613f48dd4ab8bb2d03e7edb6e7b5 : SUCCESS in 5m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/87378d71d52b41f69daa1a3b4164b67f : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dca0bdc2bddf4fba842c07b97f9804fa : SUCCESS in 50m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39502a04ef3742ceafdd70923f620457 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":48},{"id":"f8c80bc2ecd18aaac8ff2596f6b1b6d48bdc22fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-01 04:09:47.000000000","message":"Patch Set 48: Code-Review+2","accounts_in_message":[],"_revision_number":48},{"id":"8eea3047a565b5e7d9f626bc1bfe993fb991ba06","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-05 09:43:15.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":49},{"id":"4c271e9ce62265fe86ee3988978879d3c5064418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 11:12:57.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0efabb65cc38474aa1df34baeced0d49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09583803f89e4026bcbed2bcddbbb921 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ac774a6380411f8a31980a5a435489 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5945111dc6a4e8791459ea78b3183eb : SUCCESS in 6m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03db4b1b127d4fd4a537c631a72dce99 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/344858c44ec84c8b9a2fe60b0c0a57f7 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/592dd13e4f3649d4b79fd0d4ed9ace8b : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d954e8bc59684599b51604034f51b94c : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3710634a93c340b8b748b992f8ba227b : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bf125454a9f44382a580c8fc59d9c38c : SUCCESS in 16m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d3bc2984d9284eeeb645bec062f92134 : SUCCESS in 22m 39s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c771d89b002c424a983ce15ff4fff3a2 : SUCCESS in 4m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/28d8ebb551ce489ab526603234459987 : SUCCESS in 5m 20s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/49197fb3fb7f48e4beb495f6e13d2ad8 : SUCCESS in 3m 46s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c974693aa29c40c3a8ddc00e26943022 : SUCCESS in 11m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/59a125269d154e36a960c533b2d1f77d : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a3d17664956a4e9888779324ecf887d7 : SUCCESS in 4m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f168bbc5f6fb45319e1cc839560bb454 : SUCCESS in 7m 35s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/dd14008133cb4857ba4bd2b45c893095 : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f18a1e06679945939c376198792705cc : SUCCESS in 6m 34s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/aa17cffcf3af475d9aacb5d5e09b7158 : SUCCESS in 7m 32s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e9dcfd9669194fb283d33c8fac3c634a : SUCCESS in 8m 53s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0755a73c362e407eaf30e8620fc8565b : SUCCESS in 2m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ead03190d4de494c8283e14f6c07e719 : SUCCESS in 7m 49s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/18001154277740a8acfe78c8c8665d32 : SUCCESS in 6m 40s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2cbf8c5e39ca40dcbd8f8282adc0bb9d : SUCCESS in 7m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/48a51f54cc074efc97ac736ab08ed0f1 : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/fa9e41ff285c407885defd9ace3881a5 : SUCCESS in 17m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8a73b4ced10048ad89ad34ed61d76456 : SUCCESS in 6m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5a522d7b9cfd4d94be8934ecdb4ca936 : SUCCESS in 11m 09s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/82cf7fc5454843888c8fccf2552d7f41 : SUCCESS in 3m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bcc9008206e34152a90ed6afd89e4b1c : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a60e02fc493e43fb84d6b0537d4f161a : SUCCESS in 22m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d61c5d6471ba476c81df47ba19352da0 : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4708202545724a67911cf7da5fc5b17a : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/102e6acc6e1347ea8d53327887717a5b : SUCCESS in 4m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/58d310aefcb94d8891c069f595a9977d : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a1e59fd093e4af98aef141f19aa514a : FAILURE in 34m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb56eca55f2d42c6beac7ba8f659179a : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":49},{"id":"1b72b7e4db19a17b8456189cd6c78e561913b803","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-05 16:44:06.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"322821e9ea6198c42651ebff5cfd5a1062b7f205","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:44:24.000000000","message":"Patch Set 49:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 933484,3","accounts_in_message":[],"_revision_number":49},{"id":"76783efdae446ab3f2a3bca22ae68aa67f0a3ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:44:24.000000000","message":"Patch Set 49:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 933484,3","accounts_in_message":[],"_revision_number":49},{"id":"6a1549d9acca61226d9e183b065bec8d8e971cba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-11 13:58:54.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"1c18c009af78c2a663b945d3306580b6392f987a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 15:54:46.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14f06cfe7a044303b6efe09ade19190a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231e21f893b74b32892e6c26f195b4bd : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca571cc883d4e31a66a8b4545c8de5a : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d48b56e5f24d0fb749fd1e372ad9d4 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64ee066a41d439d90683f5c005806b2 : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0b26b9b402f4e2086edfbf28714e637 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a4ace57f7ad49d99522ffe0f0aa6ce8 : SUCCESS in 3m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67af84fef46d432a8846040399ccf107 : SUCCESS in 16m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1983407a4091492696afa048904507ef : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/34e06b4b0f824eb28a6b90b6b5c177a5 : SUCCESS in 16m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db756b1e431544bf8af163e324e90db7 : SUCCESS in 25m 25s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3f544a2bac444ba682d82889fac4baff : SUCCESS in 5m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9bd1e40d304349829e5d0ed5f571bb52 : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7d48de4adee24cdd9f5bd71e9a5c14f2 : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2cff2b453f9d455c96d3acd79a43b7b9 : SUCCESS in 10m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0579cd328d504f23825cab639bb2c26d : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3f34356360db4090918c2ddf1089edac : SUCCESS in 4m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6862619a814842f899cc8b497e04f3e9 : SUCCESS in 8m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8506d4899d2043268a5a4d6cf45b997b : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a2763a4902f4d419c7dfbfd962b097b : SUCCESS in 7m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2bfcdc9f38a84d3e861044f238cc3676 : SUCCESS in 7m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ccb8775264294051b542078ff9d4b979 : SUCCESS in 9m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d1dbfd1f6efc4666bcbbcf76b333418c : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aaa2d207854541fd995aefa68af51667 : SUCCESS in 9m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0d0d6130079345aaaf7cec38ceaf6e79 : SUCCESS in 4m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f20124e2a2b247f5a557fb998252661e : SUCCESS in 3m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9d177bae3f114884bbc2dad209a497df : SUCCESS in 10m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8330e1d5228c43d6801a16e28a41c69c : SUCCESS in 22m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8456a3db73634d6185cbd6859caec179 : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a67efa590ed40b49779196cc1528f7d : SUCCESS in 13m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/94ab293e0d8845dcb6a8fa3362435cb5 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48d9b0efa2f34f7abfcb46a4f74c12a1 : FAILURE in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/004d47a67dd648e1ae3d09d59bd146ba : SUCCESS in 23m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/470a5b15af5c491c813249e1c0ea1f83 : SUCCESS in 7m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/81c764f322a64bbcbe4870c65b37ec53 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/786463379f5f4cb484a2640a6407222e : SUCCESS in 5m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b9e3f08899144f1597068cce717011a3 : SUCCESS in 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/121023d247c1492aac26086c52c75fc7 : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e897dde032dd477480993a8981179d93 : SUCCESS in 1h 43m 08s","accounts_in_message":[],"_revision_number":50},{"id":"5e5bac26eb44d21a9ad8aa98647d2a61c5a087af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-16 14:20:27.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"fffeaed700ca729280b25b53d1ff6fa9dea372d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 15:36:24.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d15bc45234b542a0b17f543f3cd4d99c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11d1e4888434aa2ab63e17fe500a9db : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3dccf03dfb4b1ba4e12399acd4e60b : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e95a86145114bfcb29eea9260c1f2fe : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a89418aac9244173ba178186478464cc : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f48ac8bdf8d44d9bd16be362028dbc5 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/997b7a8e7eb8495e8f1a037f31b80ffd : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb14ed0cfadf40c2a429c3970d692e78 : SUCCESS in 9m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c7462627486546de852c77701e0662d3 : SUCCESS in 10m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/27757abc512e46ecb087801817db6a29 : SUCCESS in 9m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/63f9c7dc07af4734bd8f320d9e4cee44 : SUCCESS in 24m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/81cee950abfa4e718ad3790be58a96ca : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c1895a57bc124b4a8fcf0397602a3694 : SUCCESS in 5m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c0899ca3be3f46bd8d2990c7f3177574 : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3f56799a7dcd4f59a633b7363b4183eb : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c444f8b886e1444ca3f40bc442485c53 : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bffff080c3f64b168c2d6ab0a0c2d0f4 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/131f60a3cfca45a69ab81c5e77a808d1 : SUCCESS in 7m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f86183cd1bda4a4bb8b38200c76095ac : SUCCESS in 15m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b097399962845729be8b27c2083d95c : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/457e9f669c1b494e9e1770d7a98a61e3 : SUCCESS in 9m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5bea772b612544c3962f6ac83684ce99 : SUCCESS in 11m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6bb7b1e161ae4d59aacf92161bd906fb : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1d4db190d30749c3acd5e3ea835d8ddf : SUCCESS in 9m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/af137985903c43f7a269db306ed4b1ff : SUCCESS in 4m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/814cc12a364c464d9b4d2f76ae0bf243 : SUCCESS in 3m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3384dd460fe142f19a61d7063409344c : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a951b7b8869f449c9c9dbdcfd23bc81b : SUCCESS in 14m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4fb61188cd9649819b1f635cc49859f4 : SUCCESS in 5m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4a8398dd3c8e4b54a4e39030927bfbd4 : SUCCESS in 9m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/185b757a937e4be29b1ce7949a2aa9fa : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34fcc847f4b8435c95bb55b1b8b7a1e5 : FAILURE in 2m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8098f42c5e974a5099bac651faf2cdd6 : SUCCESS in 20m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f2146d34f1174044ac97e094548f658c : SUCCESS in 4m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6933ffec48b44db1b094396bc812f1c8 : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84e9e8ae07c644e3a4a29d2879285f70 : SUCCESS in 5m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cf2d7b80ce1c402aa620887750cdc3d4 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/576cf0064db8401a93a5953da27a899a : FAILURE in 34m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bab3068125d43d8a179da4f4f2af5df : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":51},{"id":"a67f1573529f9fe6adaa87b64c2f9d49f67665a6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 20:13:13.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":52},{"id":"3652055717f84fad68399a5e2c9bf54bbf8815e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:58:41.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f68d50b14e854ddda55d97a7fa8d9f98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b93be609f1434d6d8c4e9c2f38f14633 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6589a581c59f4ef8822ae417eeaef888 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6093686ed805473dad910b7da78d75e4 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ce6cee38a3d4865913d3e60c9d7500f : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d134eb8cd6d4307926e42e2321f86d1 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e7fea30c1cb42d5a8769d55f9755203 : SUCCESS in 5m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/494f493ea66d4d5da7aecd44283374f4 : SUCCESS in 17m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b735a528bf2f47c8b65c3012e6e97703 : SUCCESS in 11m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/226ae1dcaf624ad89c7854ed7ff26098 : SUCCESS in 13m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e9229535d7e94cf58ecdb5a98adb4c0e : SUCCESS in 20m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4840ab1bbb394487badcaaad17a12223 : SUCCESS in 5m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/abe6836168fb4dfdb528a460ae68fd4b : SUCCESS in 6m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b232066243ba475589a17d5751d0765d : SUCCESS in 6m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bf0facf5817f4d63b18167d9de3c98b7 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ce16413f0d544cff829aa1eee9c34b25 : SUCCESS in 5m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d1d52d99eb85451a87891be40d7545dc : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0089f02827f04d2f9184c44bd1b272df : SUCCESS in 7m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/aa497f78833f4f3bae82e6c5a4f8f246 : SUCCESS in 16m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9458392cd3b470994073b5b254b75ec : SUCCESS in 9m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d89a948f534d48c18d4b931c81988c8e : SUCCESS in 7m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/15186c91b2ce42e191994a9707b465eb : SUCCESS in 14m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/78bab2ae625c444c954e8ba6b57dcb84 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8491dbd1258948d4b3bf3638498ef0d0 : SUCCESS in 11m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b6b5f80a53434cdc9e3bf6cf70ade7ef : SUCCESS in 5m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cb9e2bab85f54e49b16560d0a0a7c51c : SUCCESS in 5m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6df87ed3174644a4bfdf7ea79fbf4ed2 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f4088218f71648999d9cadde40e5e23a : SUCCESS in 18m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ee60de89c9814a8dade05baac3bba232 : SUCCESS in 9m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/163936ff0fc346079c7ac8ded0ff7a81 : SUCCESS in 14m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/161b7e0f5b1b4546b1f6811186c6ef1b : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11ea98cb54c74271acc6473c4ac7532c : FAILURE in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7031fa0602d465ba3d6c89fbd708840 : SUCCESS in 27m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/576449ca291d49259f32a9373ebe0e41 : SUCCESS in 6m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dda0b88184c4424bbe056444eec33368 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55595a4e08d1499f86b751a0094cb9f4 : SUCCESS in 5m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ccea51536f0c4809b13de769e4972460 : SUCCESS in 20m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e9a451a58c54dc59084bcda84beb734 : SUCCESS in 1h 01m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaee9dabb51840518facc357a7c1c136 : SUCCESS in 1h 37m 40s","accounts_in_message":[],"_revision_number":52},{"id":"34bd54216b7e0f71a3463806fe0afa84b1b14227","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-20 09:58:38.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"a20de14726ccd44aa557d33cc56ab36d0f9f83ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 11:42:09.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a40d12f1319845aba16061928440d227\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a89eea2c8e4a64a013d15a2ace1318 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a3c3a5b2c545f7a30129cedc7d1c09 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b57c5fb9eb5a49c59478aa99a639826c : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13640c190ed747c39102b2c893c8d302 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d34ff651a364888b2ded842fb68d8d0 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ea4ad57323e4a38b4e5c7cb81c3265e : SUCCESS in 3m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/afbbba9edec54860896c47f3f633a20e : SUCCESS in 16m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cf1a6ca8014e4824b0679c1cd6594d9b : SUCCESS in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/415396db8a3a4c71bea7d09373861c21 : SUCCESS in 16m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/135292df38984175b4eaf47de123db14 : SUCCESS in 31m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5409c75277eb4eb2bde639fea5eb13b1 : SUCCESS in 6m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e577b2b42f36421bac111694ce0b0c2d : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4796d52e789b4397a60d299ed59d3a60 : SUCCESS in 4m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8e4afadb45244726a8d52793342e87b0 : SUCCESS in 14m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bd9c9b2a158949748a7b2770d7b4e3d4 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/349d65e3dc1641298a0c435f03ae9655 : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/50f885809ad24eda8b49e8f97c8e8359 : SUCCESS in 8m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/05306d77530f47c5b326d3a05f9fb601 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b15a8c33f82245499b7a1072d616e624 : SUCCESS in 7m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dec6ce39bab8411e8ccdfc4ec1cd5b33 : SUCCESS in 8m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/38dd93b4df284edbb99af36f8868aeae : SUCCESS in 19m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ecfaa37f648640178f84e54e3398ce1b : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d955ba670fca439ead8c0455467edfff : SUCCESS in 11m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/22099dd24cd54f8ba19dce3c07b0c9b5 : SUCCESS in 5m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7761d0d1ec6e46dcb79235341a86389f : SUCCESS in 4m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/78c77044f9da46fa9314541ddba6dd08 : SUCCESS in 13m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ae8dfdb685ff4a2cbe7c3029e850cdb4 : SUCCESS in 21m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c172c47351844d4b7bcb2e93c80550f : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6e0d07c677eb4e79aa17dd4ea554b495 : SUCCESS in 14m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9f7092f63a94548919dc2e69ddc3eab : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23e7847f72754810b06f9e3b7f66738e : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/faefb25f3f384332a0d34db05b101ed0 : SUCCESS in 24m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/912d312676334ff29661853b79fab980 : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ce8e320cda2c46d8812b5da37f76b8ca : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30f8863cc9ec4e97af3fcaf51f35f254 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cb54494e9f8843cbad5c65813a98a5a7 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c3bb1eaf3f341d580641304c54a4ede : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abe47ab9a9444c88a7534b86037fa5f5 : SUCCESS in 1h 33m 51s","accounts_in_message":[],"_revision_number":53},{"id":"d8b805214cd632500bac34cdb2dd13d23e1baa9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 10:41:50.000000000","message":"Uploaded patch set 54.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"88c1ec4ea5c39bdc7b239b64266578083d53f5e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 11:24:03.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"d235fec5d983125555256e0fe28a33caf6b5d31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 12:45:37.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df0306ab2cf2476d9be8f396513b9c05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700681d8987b41b681d62ba94c10c723 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb44c19b1ca4464d8165df3bdd62c16e : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3537fe4d9a6946ecaac75b1da35e523f : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16b9757625474d29985fc37d3fa0e713 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d5a3b036a2348a18eefa34c653d702a : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84655c1df7ae4156a2aeec15f60d221b : SUCCESS in 2m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/64fc91e87f3a4d2f8058f7065d45508d : SUCCESS in 7m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6fbdd4ca3f62418bbe88273abe980744 : SUCCESS in 5m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f01d7a43daf741c8976628f4d4320d89 : SUCCESS in 17m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e0f0252a0dc747a7818f881a1bc94cb3 : SUCCESS in 10m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0623e97eca5546a2a09588dae7933d4c : SUCCESS in 5m 22s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71679fd89a8a4032a849d43013a580ef : SUCCESS in 3m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e36b31744a8946008045ef4ff71de630 : SUCCESS in 4m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8149079f04ae453a8cc143ed8cf4bc1c : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c7b91bb883db40a1a1ded9a22d531827 : SUCCESS in 3m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c461b41da39f444198e83b87670dfc19 : SUCCESS in 6m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aad11c9c67274f649ccd56dc896dcebd : SUCCESS in 5m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/04ed2c585a06486d83bbee495a996f03 : SUCCESS in 15m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6aa4e6031d234cccaa06a89b7f0af278 : SUCCESS in 4m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b64d427117ab4550bcf2685bdd79ec2b : SUCCESS in 6m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3f55706293cd42aea5e0a732c1eb2850 : SUCCESS in 8m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/93ce20ec58d54a00804fee195b994c01 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/643134f3a52a4b4384d88d9c31f20499 : SUCCESS in 8m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e93c471b754740279005f61506007ed9 : SUCCESS in 3m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/509a3ae544a04251ae84225b1b77438e : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6f235dcb05fd475c96ca1785e81434e4 : SUCCESS in 8m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a65f5f9982b049bb94b1361efe8d77ed : SUCCESS in 14m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/30d2167432a34fff92d343d05234e244 : SUCCESS in 6m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9daca4a3df264234abe1777b638866b8 : SUCCESS in 10m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/be908d83bbac4b5ba9c1a81fc84f331a : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f37391dd4164f769f222d524fcee74e : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/703699234ae04f5dbf4b1cedbf6ce7a7 : SUCCESS in 18m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9bf4ee526f8e4438ad1cfbecb9bc9d0d : SUCCESS in 5m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6a02889957e94c75ad2d9748a3d0f6ef : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55bcd13d4fa2456d9933ffd3a213e24a : SUCCESS in 5m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/781fd84d16414c4dad793edb4689196c : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f18ccf51cba4127901845e53100b427 : SUCCESS in 43m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85591822cac84f40a5418ee432eac49c : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":55},{"id":"5508bff6a50c172af0f659d477025674ac25bfe6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 17:50:48.000000000","message":"Uploaded patch set 56.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":56},{"id":"edd8bc610ddd33f4dad3a385b3360f6e77237442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 19:26:06.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/263242d212ea4e39a5fbd242c56518a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5594f0cfa34c8cbee2389dbab481c5 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3376774195574482b76f43987b4a5ad2 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/556f1797f9bd4cea8fb6315e1ffab2b3 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af32e14d90d9444fa7e32703cccee3bd : SUCCESS in 2m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4b40f3552a34354a666d0717f09f72c : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d84f3e375c33465f924729e0c2916c2e : SUCCESS in 2m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1654f7dcc38548878eecdfdf7997cd9e : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b2da632f168c4ae4bb79753c9f6a965d : SUCCESS in 6m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/47e97b9bc16f4a028dea8740b52641f8 : SUCCESS in 12m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/752a7632e5264296b87b823a4b6619ae : SUCCESS in 15m 01s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/02969bedbdba4793ad24b4522109c262 : SUCCESS in 5m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3ef3d23ae287469b9410d83cf4b179dc : SUCCESS in 3m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7623c6b497f44b56adadd21aa6d3ea81 : SUCCESS in 3m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dadb7349ac56466983fbdee07cbf4cc1 : SUCCESS in 8m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/563977733a6545c1a9c9c30ccc5c1aa3 : SUCCESS in 4m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d7f4f509c7ab4c82b66e101907aea620 : SUCCESS in 3m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2a8f7f594d6d48e1b8be3451ad8c1961 : SUCCESS in 7m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ede1590e8c7b42ff95b4e5393dd4fd7c : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a055bf400de4465c9db441f7de7bdab4 : SUCCESS in 5m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa9bc76e3e0945e79549ced0725cbf0e : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd6b6ac313a249a299b975a91e5d2049 : SUCCESS in 10m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d20ad54bdcb94922b1697ceb1d13913d : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1959418d869c4523870155febc5e1e51 : SUCCESS in 10m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6892256c193b4560acf72d9d4ed653b6 : SUCCESS in 3m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f3fd7a82fea8454391e0c711bcd05bfe : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0bbe81013c144168a42e2ea99f20cab6 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f13a8dabde9844dab1ae612fbf066677 : SUCCESS in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71351e36e10c456182585ba9f8cc5be1 : SUCCESS in 8m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ccfacbd9ce1541a8ad54e34a033a466f : SUCCESS in 11m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f23827b11c0e4c1a9f41694e7aec4d48 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7279b8f8b0134133a6de9226157b3d17 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0dce62987c94434880be56f13d6c94dd : SUCCESS in 18m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2bbed8a0d75c4dbdadc426dda03f43ff : SUCCESS in 4m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4f48f94c037b40d489249c79c8b460dc : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e92304c4c6954fd1b03968af69d599ff : SUCCESS in 5m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b85a0324f33243a8b4f0708a0e2e1d82 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baed30b97d7a4d35bc7dfde353bb4a0f : SUCCESS in 35m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e9f4c03efe54b64b5e09b34cee1bd75 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":56},{"id":"430fa1a7d081dace56f2fcfc451dbf1b24717e41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-30 10:28:19.000000000","message":"Uploaded patch set 57.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"1222fb3a0712a0e7b247cbbb120f85554ecfaea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 12:06:45.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30bf2eb50d5846cea8f2eb683f19b62e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708eab8e0afd4a258df5df39a6f25822 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551543c94cf74d5f8d2e252fe077fc15 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6202bc7244eb45f7be732b8a937a3ea3 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95648fc621d94947abfe46e1bc012159 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f303d2793cf0486592ba0103e7a6de59 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1761e2db38ae4a479fb24f581d863298 : SUCCESS in 4m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d421e80fe3f4442684652f5f2ac3132c : SUCCESS in 16m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cbbdda5c3c3c47eabb4993fcf1f381a2 : SUCCESS in 6m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ca2179a44c045a5bf1aafa4d571697e : SUCCESS in 16m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d3acb7d57d144119aa1a0046ec303e35 : SUCCESS in 17m 19s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/50e0f4dfdce543d7ab9f1a527af586b1 : SUCCESS in 7m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b9e3a12d49964fdeb8fc78c64933c7b7 : SUCCESS in 3m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f1395e8ec6fb4eb2860a2f88de483cac : SUCCESS in 4m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d9799e8d8ac54d1aa08d47ad15c602b3 : SUCCESS in 9m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/81d96d5b0bc24ee78bc8ae862c7b38ca : SUCCESS in 4m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/83e280adabfc4479a3df5a2e7eb26305 : SUCCESS in 4m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/027af725cee94ec7883ba2f4d4458f0f : SUCCESS in 4m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/be7ee063037a44c7bd70166a46b56d8f : SUCCESS in 7m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ce1d93d4bba475eb49e4ad43a207f7b : SUCCESS in 4m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6de10ab66d4649f4b1fb0f38411833b2 : SUCCESS in 9m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f0a78b395d75478097d359e49da386b2 : SUCCESS in 13m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/71154e75cb9b41d88a8d1b682c795f59 : SUCCESS in 2m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1a9cf9f6978246fdb194be8df11fbd75 : SUCCESS in 13m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/db46d9e1ce324760a365297dcea05232 : SUCCESS in 5m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/efca6e99f3cd432e9970900e4a9a981f : SUCCESS in 3m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4a60dc8262e34e07a899876bce70fef9 : SUCCESS in 11m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8831c508f23641c398ea05c3d896af08 : SUCCESS in 22m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/668b8c229cc7450d8534d6e517dedd79 : SUCCESS in 8m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/35d24ff9ae054b89a7e4040e4c3cc2be : SUCCESS in 10m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bdd4475cfb6d4bf7a531f4246bbfc78d : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b1391da6d15405690d8d47d89fbe72e : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c13518b7342d4933ac0ddcaec4307f3b : SUCCESS in 19m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8c68558c4b74404fae01de76ac7bfb6a : SUCCESS in 5m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8c2402a353cb48ea9308835d92a57e79 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6146aa4cdce449e0995d5f339dd54b96 : SUCCESS in 4m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4e759e1aad3547ca8632a84708430089 : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb0ad9a59c9643f3891d2f77a353583d : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b65a62577dbe47d398ae816824b39d58 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":57},{"id":"1ac37c35e970930ea42c1e69eab1a02c8ddc08eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-12-01 22:40:29.000000000","message":"Patch Set 57: Code-Review+2","accounts_in_message":[],"_revision_number":57},{"id":"59f8881d5cc367fe430d21363e7b094b103bae43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-07 08:10:38.000000000","message":"Uploaded patch set 58.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"56f06d4feb467d6a5572d610d426029644de33ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 10:10:49.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08390135f86642a08c7a16414c960d30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d688b3cde1474af28e423ee14cc81419 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5769a4b1fd224a8e9781f598b4a91e1b : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093426a368434e0ab9693a42a82089f4 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b292400629a4449eb4675432f345b3f7 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89425c85615143a9ba65b94180f73e43 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21f78cfb755249faab6572876c76b114 : SUCCESS in 3m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb7764e6461c423da42c4594b24d04c7 : SUCCESS in 7m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b64ea830139d44729677b10b65905c80 : SUCCESS in 10m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2aebe0e06e5d474bb674f65e335db6dc : SUCCESS in 16m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5a4904a0a7db46e19f3a43e94dcc401b : SUCCESS in 31m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/856bcf6c8b974e62b84f65d7126bc783 : SUCCESS in 5m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fea5f4ad181e4380ada9c6e3a7dab914 : SUCCESS in 5m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/19d6a53ba3704755839b58c27a1110e7 : SUCCESS in 4m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/75b17af5960547f3bc24fbb2af535e87 : SUCCESS in 10m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a620b5fb8d704d6ba1f6abf5080dd222 : SUCCESS in 5m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/769e6ae5513c4d00a7529089cd77fd02 : SUCCESS in 7m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/235f4b10ffe94ff59b3e1b2db5bdb82c : SUCCESS in 9m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9f43aa19b24b45908b7f8865db606514 : SUCCESS in 15m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49d9d58f705e4590bdbae1e3d252d326 : RETRY_LIMIT in 2m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cf366aa60ef04af09bc480126b8727a6 : SUCCESS in 8m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0d451a25470048268554b8293bc4f684 : SUCCESS in 9m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cf51fa912cec4fb1941831a4c59113fe : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/adfb7c43493f47d2afbd3f3f1502e44b : SUCCESS in 15m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/82ce4f041e6541b0a614d1b58aefdf07 : SUCCESS in 3m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c62bb0a5c168418794a6a87c14fc9e9c : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/514ff7cba9654afdb1c60d57274ebbd9 : SUCCESS in 11m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fb8efd93e7694c93b0474f59eba2b1a8 : SUCCESS in 15m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e179c38dde614f8794e57dbf43a9ab9b : SUCCESS in 8m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/56ae663a6f2a4a4d83e722f1e00ec689 : SUCCESS in 8m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5e26704a12c944cbb25cf9f5f9f5a359 : SUCCESS in 6m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30e737e0438e47c580c49cf0a777cf2b : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37f80aa9e1894cd4a510c43e1ef2db11 : SUCCESS in 32m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d48b0d25725c4d85a66200a532e5caab : SUCCESS in 3m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f415c90e2c2c4c6bb324dee91497d3ce : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b8c2d258ee04237991cc600b8e9bb8d : SUCCESS in 8m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/42141a7329864b108fa06a242641720a : SUCCESS in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de5cd63da97d4f80a95d9f94ad7b4d0a : SUCCESS in 38m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/611de31cba2549d59b59c12ef16a55a4 : SUCCESS in 1h 52m 38s","accounts_in_message":[],"_revision_number":58},{"id":"631691dee55598fd5ff1534ce64e47030a908606","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-08 20:33:56.000000000","message":"Patch Set 58:\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"9dd79f8aea709b3b70d6fe880e3ababbb1c98a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-08 21:53:27.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f775dfe154642a7b0476688515f5047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a775607d4caf4bbd8f283de594254f6c : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334f6b3768c84184bc9bdf6d4b4799f2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/181d9a77a09b4cdfb2bc413f4aed73bc : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/781f9c6872bc47fd81ee7ded01da5c88 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8b9cf61602e4348bc2d29a89a0199d2 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c8b91f23d7a4b989990a66b59d9b3b1 : SUCCESS in 4m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1aa71d6334f7454594523d76c83365e3 : SUCCESS in 13m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1fbd279d0a34b198a0895d2935b0722 : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3787261c64594e92a26c40ee54971afe : SUCCESS in 15m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/92b4281374b947a7bc9b0fd74d7dd3cb : SUCCESS in 19m 10s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/780f3bb2f7a24964a25f56bd3f8753b3 : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d13834550b144a3d80c3cec3c2d67fff : SUCCESS in 7m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dd7a5fe86f2b477fa47437fe15558def : SUCCESS in 3m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5674dae760bb47dda742802a12f2fbcd : SUCCESS in 10m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f131c9e56453443f9a58db2808c66246 : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d74ec688b544c74ac9e9f9c5a087d52 : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/84e8b091d3fe47b386c3bf8a9968c0d9 : SUCCESS in 6m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d0e831fdafdf44508f1a092dd53fab06 : SUCCESS in 6m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b807ad9464d4337a63e0a64034c7f14 : SUCCESS in 7m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e0831e53e11d4c319280890777b388e3 : SUCCESS in 20m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bb9618e5754341cdb9b117e6ac1e770e : SUCCESS in 11m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cdc825b290c9480988f7311e0b62503d : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c03893e2d90f460bbe064a99454e29bb : SUCCESS in 7m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0a17ecae7eb64b7a870ddf2d1a111b5a : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6df131df74ed4da1b570d6b50eb19cdc : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/029910f99a2c4a10a9f921e390661f5b : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fd36e1c8dd6c4d49baae731bf315a1b8 : SUCCESS in 24m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e147d13a3fd24e6eb7da604d82dd03c7 : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/31ce53ec343841068b4879a1a8ec00e0 : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1d5a89f1c41145d7a957c0b4bc23d2ad : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d547e10243d2446fb654697294382096 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39793f98f30c43cc94750fa96d551c7c : SUCCESS in 25m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/09c3d9dac0134164bacbebbbcf3958db : SUCCESS in 5m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/23bc3e7c5436442ab793aaf32cd57d2f : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14f86fcaab8a465f839d5781121d6ffc : SUCCESS in 5m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/27a4c623169344fc9c41668cd7d92e12 : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47290d1b04654d1d91d3bebdc883997d : SUCCESS in 49m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfdd5a3b850d45c3a7f5f95887dc0f3a : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":58},{"id":"fd5d5e02f519b1e0c3ff9cfe23fa000eb011a729","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-14 09:45:06.000000000","message":"Uploaded patch set 59.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":59},{"id":"f1f9f1ba2460d74e3ce315dd4191241cdbe58783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 11:29:37.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c751bbcdef148c8af86fcf44b363d97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fa8919bd81483b9bda6b38f71a9d89 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2e4d98242e4451906989ad1abf4ddb : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8425b786e0f44c9e9bec001a82512d68 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ed9dc29caed4f159f4601447f67ead0 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac78718acf8e4398b5179f7ccb20e129 : SUCCESS in 2m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1e4a28bc769405f96cd72e1caeaf1ad : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bea7e550c9fc4bb0a521ce43a3e8f6ad : FAILURE in 3m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3f8c01bdb8964997a2a7ff3339bf7018 : SUCCESS in 10m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c3ceb4f08ac34e1980f795a458436d0c : SUCCESS in 13m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f883b970b87e465886ddf57dab88dd30 : SUCCESS in 15m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/dd3327e5eaea459e9430daafd5d478ed : SUCCESS in 5m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3270c355aea24341a7a7ac48c3f37202 : SUCCESS in 5m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dbb7a221b11d4f149c335a13bba7fa01 : SUCCESS in 4m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/03cdb1fbefa84263a74e082d66824d6f : SUCCESS in 8m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c093f4eb704246e29ae94ad718802729 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2d017e8aae824ebdab85cf8856ac0d13 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/57a6d90de09043e79d97e4a4be8b80ae : SUCCESS in 8m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a514b3e07f764f1d9326222b75a4b3cb : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17742300b95d47a6af047adfe18d683c : SUCCESS in 8m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f3be5b2ed07c4a24953060eb3b437faf : SUCCESS in 6m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/112c1d51ae91496d8edcbea8d740f8f2 : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/28747c59879f4bd18b69039d06eaf486 : SUCCESS in 3m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8d6e3dc4665647cc90912159c5e39d8b : SUCCESS in 11m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/69d1b6b85980471c90a9dba11538985f : SUCCESS in 6m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2fd396ade66742baae0b4a44acdb9a23 : SUCCESS in 5m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/af5ab10dcf5c41ffbd03d68e8f0be642 : SUCCESS in 9m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4cf645a557ba4a25bd49937bba392ccc : SUCCESS in 23m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8fb2726feb1f4738b482dbfc6e54ba4d : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e9e6db3d3f7a4db080c3599799bff323 : SUCCESS in 11m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e6461d76387c4719ba7b045240a8286f : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59fb7150f12546df9f672ff3a1abc393 : SUCCESS in 2m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a547aa951cb140b0ad3c7014492ec28b : SUCCESS in 24m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/12ebd683d352403fb3c8f284eb8a8eb0 : SUCCESS in 5m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6546e64f777847f4a49db75b40c5500c : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32906a0dc6f64bda93e5140c43c8f176 : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a5604012ab7b422c93061ff0f6c38fcd : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d95dc35c848148f4a11bbd5aaede9d63 : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33dfe42446374aa18c90b0fb1c7f8d7d : SUCCESS in 1h 38m 56s","accounts_in_message":[],"_revision_number":59},{"id":"7348655f146520f9555c5ab1d6d0e50c8bdddc3e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 06:24:49.000000000","message":"Patch Set 59:\n\n(1 comment)","accounts_in_message":[],"_revision_number":59},{"id":"cd5b4e263d9c56a7ee2110f1912a73009b013d5f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 06:29:20.000000000","message":"Uploaded patch set 60.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":60},{"id":"7096aa357738b15d99ad88e2d70d2ff35797e26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 07:48:32.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1905d0574baa415e92b2240741d124a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9056c2467a154888be22d15b96d68d0f : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1068d7e7a8e64846b0dea3a4d124692a : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6aace8debd849558a13a48f342243ed : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/016e171162db49b2a5cdde478f2a7c19 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dca2229c3a2547d8b7e2ef0ec8b4916d : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/221d2010dfb14bb998cea6933fd470be : SUCCESS in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce2a839152134f6eb12fa8894d5c370c : SUCCESS in 16m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e9426d727d6e42448c1619f4cd1f2fcf : SUCCESS in 11m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d4da0b3593d437c8122330ae649f329 : SUCCESS in 16m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1644d45d00874bce988088f16f9afcfe : SUCCESS in 26m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f2cd019b5d564415ab90098122fa59fd : SUCCESS in 7m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/84de348a370840dcaabd56c16b80860c : SUCCESS in 3m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f611ca8ce5c349f29e91a104b27cf885 : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/81245afe539847e0a9a796ad3e90e329 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a4a68f3c1d56450c9562749892bf2a97 : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c1b7be91f35142ac833a6110fc7cf09d : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f30613be4f8b412096375782b77c6405 : SUCCESS in 5m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4e853491e0a44508a3fa9a50444495ec : SUCCESS in 14m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19fee5c3aa914904a337d7d921b660d5 : SUCCESS in 7m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/40826697677c49138811506db5f93dbe : SUCCESS in 7m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8243e21a71ca4f8fb32b983c3540d341 : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1e7f867e2c294d57abcd3bf09d03b1f5 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eedc4d2cea86408e85a3b426d23b5781 : SUCCESS in 9m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c8f75810fd4a46fc9075cabdc9ef39b8 : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6234e11298964b8cbd64ffc70b6ebfeb : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/34c1299874884a68b0540aa6c8a6ba0c : SUCCESS in 8m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c70665b0654e4a5ba9a1b8fb3f769d42 : SUCCESS in 15m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/09c54790a2f041c78d282fd32c34e0a6 : SUCCESS in 8m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c97da5cff6ec4166a2438f6507efb822 : SUCCESS in 11m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3202f5349c7d44e3b04d3b356ff82f2f : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/042ed7c5dae74d1b89dfee715ad3b754 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3681d210ff3946dda7d7c2f411ffbdb4 : SUCCESS in 23m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8b50edfe0b45464dbb0d9f98eb63af40 : SUCCESS in 5m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9fbd8ede5d3f4ceaa4e5e062cc57c722 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dedc68faa9364313891d82af9251f6cc : SUCCESS in 4m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/98a51831c20b44d0a85494bf4e213135 : SUCCESS in 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/403954ee3bbb49eeaeaa55ec922d8699 : SUCCESS in 1h 03m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b04b1ac93c644dfa360e833f91fcf4b : SUCCESS in 1h 17m 44s","accounts_in_message":[],"_revision_number":60},{"id":"50d8492f61ef2f3419b5395d421e2d2f9b3718a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-21 09:06:27.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":61},{"id":"0b2176859e12bbd6e3047759c1ef1fbea71f4342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-21 09:57:26.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/112a49eebef046a8a96cdf244d45de11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b36e9179e854ceaadc512c106e33237 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a064084777164cc39a5260d770112c63 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f82fcc8d9f94c199384b0f8229bd91c : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee6e63cc60ce46459dabb54528c777eb : SUCCESS in 2m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bec67f5713b42ce911cf4255a0fbd23 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ed0638a30194ba59e0a8b838aa8e70b : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/950eb2ced5ad4aa6a985ad64dbb0b6e6 : SUCCESS in 14m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/72fbada09d8340d6885309a0f106c0a1 : SUCCESS in 5m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/85f3b68ad28d4ffb858d3c7f6c6d7d3b : SUCCESS in 12m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/65febe2fff1d4388937996022920accc : SUCCESS in 16m 07s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2ef0d69780b4449cbc6232664f3499e0 : SUCCESS in 5m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e05306ca5ae645debd2c215b578d496c : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c1e2bced8eac4418bbbfa6c656a338d8 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/146dfb5a8b364081a14d76779b1d424a : SUCCESS in 9m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9807c310cca44a96b70dbe05da7c3be1 : FAILURE in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8738da980d254a61a7675bd60fa26a14 : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9007b010082a4962a9d51d7ca471bae0 : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/21240a9fe2684d8ca71ef920ba15a2f4 : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6dcd12bbb8ca4105bd0431ecac40efac : SUCCESS in 5m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/53e7cad863c54840b9772833278c6527 : SUCCESS in 9m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/58cae47e03944553af5038e47ba23067 : SUCCESS in 10m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/90fdaf8b7b964487bda23af4ce0d0208 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0c10cdbd21e645389508fdbb7bc97afc : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fd751fbe5e934d87b508316d5e7fdf7a : SUCCESS in 6m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/629d7277248741449a697787222d0598 : SUCCESS in 5m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5c9b0d6607314101bfc165031aab9610 : SUCCESS in 11m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1d99af0d907d4bce98a33e2c101af1aa : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0c44fb44508e4fc69442cd3b605df5cb : SUCCESS in 5m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fee56eedd88d415d956aaeee68eca7e7 : SUCCESS in 11m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/19176b9b638b49bfb24a86f572998b10 : SUCCESS in 3m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/571bea94427247989555e990362ac1e7 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53b10cb1cccc4315a6a46c8f6b768b61 : SUCCESS in 26m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/650b2b0e3d5e4c2daea46455a98db88a : SUCCESS in 5m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5411526fde8a46a4bf4d9bde34080500 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb2d7d7793524211afd3aa78ceaf587e : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2e92ac01874f414c817edcda30f14e00 : FAILURE in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fca06fbe43d0411ca86896459cb1c29e : SUCCESS in 46m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30f024d906ab463593eb75253facc1dc : FAILURE in 49m 18s","accounts_in_message":[],"_revision_number":61},{"id":"a95ce2fe50e00733b35725aaab1ed87a6291a619","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-22 07:52:00.000000000","message":"Patch Set 61:\n\n(1 comment)","accounts_in_message":[],"_revision_number":61},{"id":"e86b621b729c1c24ad42d3e2ccbb04dd42b2a755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 09:46:37.000000000","message":"Patch Set 61:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9c02f3e76fd40b0a0a1be92c82357d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17c3979a3954b4cbaeaa3b68f83328e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d0887b645c4ca7bf420ef633b12198 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bcce24d468a402ba25443c85153e53b : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55fa06633d7443cba193e5842d8fc5f0 : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ff41d41eca54dd09ef34822dd4c62cf : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/982f73e5ff4e4fac8c8523df18460ce2 : SUCCESS in 2m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c7e9a6a758549eaa849a8306f74b5ab : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/33e472264904484a9dfb782c0314347c : SUCCESS in 8m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7caf38c1d1ff43a2930cf85e0e8b0f68 : SUCCESS in 12m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cff0f3af84174d60b90ae85f83bea33b : SUCCESS in 16m 12s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b6d01b5b31f24bc4988f03bad56858c6 : SUCCESS in 5m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/93fa85e9db244fc0a4ba1c267a4ae8fa : SUCCESS in 4m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b93c51ca3ce44c7186ccaf93c001c18e : SUCCESS in 3m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/72de0b9d23e2466eba68e08832afa120 : SUCCESS in 9m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a88802cfea8a40c7a0f6ca64e1ad854b : FAILURE in 4m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fcd0ed4a5c774b2d806f02d23d33bc0f : SUCCESS in 4m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/af19b58bdbbe4b6d899cc341d7ac0bcb : SUCCESS in 6m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/379c6a77523049eba70c4ec35454bcb5 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54251af223a04acf880109a73557db2c : SUCCESS in 5m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9247307998b349a38cd64cea6ab680f8 : SUCCESS in 7m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/25eddc0ec6eb4526ba6194e1f2d26d9d : SUCCESS in 10m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/42809b8d5459415b8f1421ad596511c5 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9c0c8c1cd9294f6d941ee045bc2ac8b3 : SUCCESS in 8m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eddf3adc9b574382ab5982363f4e5e57 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/087037b01cf0446d80971e131e77702c : SUCCESS in 4m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a9df34d8001440f8af0953c9a3dd97f4 : SUCCESS in 10m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d5af719ffdd49f38322e16afc42d7dd : SUCCESS in 21m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eacd647e7c4746e9970e5bce2cffc0b5 : SUCCESS in 5m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c29de00b06b945988fcc18ce57fe5b36 : SUCCESS in 13m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7fbd01425e145ce956188d6862841f6 : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/285ca1498f6d4e3d925607060ef6f124 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7a471de12874b1dad0cdca081029bf0 : SUCCESS in 27m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/51477df2e8a540a182102efbd335dc6c : SUCCESS in 4m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6f8658ecd884b33b32853229ccb50c8 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57911ce6b3bc457a9a09eb3f503dd908 : SUCCESS in 6m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/95b0ddfa2de149c6aebbb994aeb496d5 : FAILURE in 8m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3be4444c6d934523b0852cef2ddecec0 : SUCCESS in 1h 05m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2efd4f798a2147fa93c0dd4adf2e280a : SUCCESS in 1h 47m 02s","accounts_in_message":[],"_revision_number":61},{"id":"d87a6c1253b71c4c4a34e49b0adb229e9c777fbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-06 13:18:58.000000000","message":"Uploaded patch set 62.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":62},{"id":"88758279a902913c26ec359cd0a316fe7107adb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:43:46.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95264ab4f6bf47aa88090ca339548248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff1dea5d2d24bcd9c265b430c3eb945 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b169fb608498444999ddf2c013b57f72 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b9f489664f0475295e93ea49179f913 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39b26cd33440461abe567b9dc9dc4707 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1dd73981b10349d6bcc644639047e58a : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52c3b9c2d98542dc839e2ef1d7493a1b : SUCCESS in 2m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56a19db28c8f4b53ba3ee923fe83cbe5 : SUCCESS in 13m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d84dd4d1182043cea705918dc5c43fc4 : SUCCESS in 7m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/786ccf09fd3741ad94d1ebd40806e61a : SUCCESS in 13m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8f0840b7c09e467195c9d25aeece2eea : SUCCESS in 16m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9b032f39e88545be931d1814f4fce2ca : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dd0050f6775c47e3886f7d2bb1321755 : SUCCESS in 5m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28489251592644e69e9e0dc9a8d4ab98 : SUCCESS in 6m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e83b28a1674d49a7a9dc8cf77910492b : SUCCESS in 9m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d177eefe0d96479c907b0a256ba8c8a1 : FAILURE in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5f1e379fd85643599aac5717e53f078b : SUCCESS in 6m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a0ec679dd269411381fd57831a13bbe6 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dbc8ab14c90348ad8ebc14d12727e627 : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/567cfe1ad68241cd89eaaf8674946b7e : SUCCESS in 7m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1b7c0b92726a4fed954e7c3baaf92d0d : SUCCESS in 7m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4d2a8f70563d43bb9868ff0d41d12e5f : SUCCESS in 9m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd4b76f3c0da46f5bab5136bba68fdd2 : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2fee597768cb48dc9e8cc4c0c7b99a65 : SUCCESS in 9m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/86fbf77131784bfe942a17121bb9b0bd : SUCCESS in 3m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a205cca171bd4485b404e7927e8dce99 : SUCCESS in 3m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5e230215aa3142d88f0c6e11b3886fe0 : SUCCESS in 9m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e204cb62e9b849ce88aaebf5a410eeab : SUCCESS in 15m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b73cbe4472d4993a02602531e0d3596 : SUCCESS in 8m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7d0ab6e7646f4add91d07a162b6f8c90 : SUCCESS in 10m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1b093e69d09e4c1c8844d2d9726e6817 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c8d573f889844d3a78c29e83e38a742 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03b0cb61144e443897923bbf97fbd23a : SUCCESS in 22m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4a3fead3f85d4b2aba498e7638dbde5a : SUCCESS in 5m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b79c9b1c59384599a9689ee86034761e : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66ce3339693445c19a959d114cd220cf : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ee2997a7a7cd4ef2b3a41abab878c56b : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a84cd865eb841f3a1b19770f9821efb : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7df400f7b9b4515ad9ebc32e0caa29c : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":62},{"id":"ddaff0fd181fd9e3bf14825fe244a2db8092dc7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-06 16:26:00.000000000","message":"Uploaded patch set 63.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"a058839bd2a35cbfc115309360b656970dbd3a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 18:03:34.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca7b79194e64b239bf6d176ba3936b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241f091be691441c831cbd84068b4e94 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e290d65f61442a907e46f2d8016bfe : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3aea92be1444106a6e9665226c70ce5 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7c58cb8fdd34687a9d205918abf8570 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/beac7a6f3ecf4977a643d03655cbad48 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b4b1cea17b54376be0016a8390b2d8e : SUCCESS in 2m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ecc414c3e294a5b908a32567c23181c : SUCCESS in 12m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b6134baf2cf249eba59eec54e25df9e6 : SUCCESS in 10m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1f2cae7bcc6e4f5493b69c3cb52ca562 : SUCCESS in 14m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a696cf9370824065acc352ef97bc1a37 : SUCCESS in 20m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ea8601641678496e8dc1176b6aec60e6 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/39e263924da54557bbf6412e0b69993c : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fb094411c2bf4fe99d1de920d0a38919 : SUCCESS in 2m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0f9c266cb00649c29c7322c669343bba : SUCCESS in 11m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a2a3996acaa4406901af09347559203 : SUCCESS in 6m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2de1b91e39fc4e419c54b100f4a6e75f : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fa83a751acd64a2781688b782729c485 : SUCCESS in 7m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e4a3bb52287c4c96bef58a736517af15 : SUCCESS in 13m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/749dc0e8dd094e44afd796a167380e09 : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ae1cf87e3e88497194b536a289dbd23f : SUCCESS in 8m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/297f1fe1e47e4370b18d37df707b6b3e : SUCCESS in 10m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/917b4cb815be45de9816af1bf12f01f5 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e65adf28b5d84e5881014cc30d85250d : SUCCESS in 10m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b877d459e2474f41ba3bb66035c3a6dd : SUCCESS in 4m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c8d0e790fe9f49e7a2351956884e9120 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5f2b6afb0e4e437f985fd57a01b65aed : SUCCESS in 11m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ece4d26acad94d41bad1c839158104f0 : SUCCESS in 15m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e3e30b706cdc4447855d11b2bbfc2a0e : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/370465eff5914d139725514a1b731f9f : SUCCESS in 12m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f409152d42a84b33a3e15e7e7b8cc267 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f78a1f3c26ff4985a00b9b4b95bed3b9 : SUCCESS in 2m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddda997df64146868c548a30c6070c98 : SUCCESS in 20m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fa55c003742247c8ab8a984047c2a7ad : SUCCESS in 5m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e5aa1ab8a7b4448eb3661cb7d894d48a : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b967f9c91a00402ca5467a2aa68f33b4 : SUCCESS in 5m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/94c389e932e74585bc101eeb1bb79855 : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44de8ff5ac944601b9bb9962a44972b3 : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28689cdf73c4461a8e86cd3f302a62bd : SUCCESS in 1h 35m 16s","accounts_in_message":[],"_revision_number":63},{"id":"e9ada96fdc9471ba97b1341870b8377558746215","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-11 10:36:36.000000000","message":"Uploaded patch set 64.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":64},{"id":"4c4fdf45a307d6c148af9de92b0f5096dd556d7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-11 10:45:14.000000000","message":"Uploaded patch set 65.","accounts_in_message":[],"_revision_number":65},{"id":"34a02d64588ff69f823c3a2c7cca502bb0fbdbfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-11 12:27:11.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c970fdc29bb4c8cb0690c66bf5025c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa55512dd1094682b2a7cf3cc24c834a : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21307b29d3d4b4dbd8c3ba3a7191fb0 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4842d146877f40649602bcaea14dd6a1 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b751cc43f04bc88174c308ef5f1113 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/649a5d4339774d0d93184435d0cb770f : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/853d5057a77f4563b59585ce082b2f2b : SUCCESS in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b89e1403cf5448e8b4b81970a63f5fd : SUCCESS in 14m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c3ef8d2ce9c341739b7a6fb3b3fe99b6 : SUCCESS in 6m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a89b57ada974917a5c8ddf83561ff2a : SUCCESS in 15m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/34abf9b546ff4b8c94bbafe07209913c : SUCCESS in 13m 59s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ebb2943fe5fd4a98b622f48495098039 : SUCCESS in 8m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0430a9bcc0e74bd0861970009e96ce20 : SUCCESS in 8m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6ffd67b8dec54a868bc4d56d8fe33c1f : SUCCESS in 5m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/152a2171c22746dea4819164d543ab20 : SUCCESS in 11m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/40f0beb2cb2d4166bff1d5572e18502f : SUCCESS in 3m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1f85cbd3015f454face1403b6047ed61 : SUCCESS in 5m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/36117e59e4444b41812cd49f712b32d2 : SUCCESS in 7m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8c9d462c7e0b4d65aec2976b86a8e26d : SUCCESS in 6m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/639570bd0d184a319e044a2990f0fdb3 : SUCCESS in 6m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a9bae6a88b445b5a40ef64d4b5d4f52 : SUCCESS in 5m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2e42e4b85548459abf521088567197ad : SUCCESS in 11m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4304596df1b845579fe2bb022ef80c5d : SUCCESS in 1m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fd2309faefde4296bacc1ab6ed50012b : SUCCESS in 10m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fdc01ee3acff4aa5b35032d762de4bed : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2297d7e5ffa14488a26a09dddb11f7a5 : SUCCESS in 4m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cb94db711c0646e39534485f1c7e2839 : SUCCESS in 8m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fb8f5453ccbd4a1db368a7e67c18a83f : SUCCESS in 19m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d9cc6da68d874dddbb19c74901af472b : SUCCESS in 4m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ea3a7884743f45ed8da7c87afae2ae85 : SUCCESS in 15m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e260208092964dee90de0146fa0f1c43 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7bc47e7ea87840b1aaf8e9ec2e1ddc36 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01e67f8539c94ebf9a53ca553113dbf8 : SUCCESS in 22m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/59bc98b75d80445fbf7f8d1fe72184d4 : SUCCESS in 5m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/39a287da07fc447a873453dcffe5fcf7 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da74305257144c34b0b6f6557215edbb : SUCCESS in 4m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/45d5a7bc6fdc45f683b45191d59cad40 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a6fc0d11e4f4121ac92769057a70542 : SUCCESS in 1h 02m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6627bc441a944af5b4b21cfa56866f6c : SUCCESS in 1h 35m 38s","accounts_in_message":[],"_revision_number":65},{"id":"83f89b2eb1efb47282e4e22c9cc25580167f0081","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-12 12:40:46.000000000","message":"Uploaded patch set 66.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":66},{"id":"d22ba8c7c06149eb84fd33b4efe2156472ca0dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 13:48:55.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930d8b0e2361408e8ac91912f15cd0ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f34869db87d4659921853186c106125 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fde2e4db5fe4e398fd492c2137aab06 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5783275e95b14e2cb3d74abd0ee76cbc : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/980e11be49854199a64e767442ab5775 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec7ea65dc9e84604891a6a28d47e7591 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/baaa6310cc5c4d8eb41bf8950ec18c49 : SUCCESS in 3m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a1f1fabcc5345f7ac7ee697b2550b7a : SUCCESS in 13m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9deb4ddf459a425a850d4a76da11d28d : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/552fcf73830949e6bd02c00675caf622 : SUCCESS in 15m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dfad69280ae241a08782b146364664b2 : SUCCESS in 15m 44s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/93cad5fa7fd74379bebe1004c8a6f670 : SUCCESS in 3m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5bb0b5697d0443398f3b8a88dccb121b : SUCCESS in 4m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bd7650a50b204ad4acb24f973a2b1c25 : SUCCESS in 4m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b49ad12867ba480bb8538f2ecf0dfbd3 : SUCCESS in 10m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/12e506b86d1d44d49660e9d095150cfc : SUCCESS in 4m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c1553d84850c4a169e6ece1162e172f6 : SUCCESS in 3m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cf1af504512f424797b5aabc598fea65 : SUCCESS in 7m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2fc04e01f9df46aaa6706dbb60072e4c : SUCCESS in 14m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d2a2f6c547d4bbf9bd7c385b7dcfe6c : SUCCESS in 7m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c525e89ea56546fe87143473f65bc46e : SUCCESS in 10m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/42cfab833ee54a809dab5656101609e9 : SUCCESS in 8m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e8a58c8c99ae4fed9b993e00b05687f8 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/58ac7a55b4a94641b0f15f7c153374de : SUCCESS in 7m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/60dc1113dd944af6931965ccb8e4743b : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/41d35693786d4b3d940e122dd70a53fc : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b30b4c87640142a2b72b5be69d7c8329 : SUCCESS in 12m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/19661e9461454ad19e1f816bc15a4cae : SUCCESS in 15m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f8ccdac50bb46db9d4f99feee86a0a1 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2d20999ebda544898894864669d0bbb2 : SUCCESS in 12m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7e113fc9148e4f2a8b107fb91ec1a2a4 : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8b9f12a190c48ca891cd3ea3d47f8a7 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/858213b04cd742118eb7b29857388c9b : SUCCESS in 30m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fd678957ee7d4136b468d387d0a5e234 : SUCCESS in 7m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b9ba203e8c142a2a6a6f796820854ce : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/755fde196fc24cccb7fc70cc8d378d8e : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/836419849789412abbe8c4474426b855 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5952dc3ccc5d4bd9af9a32d03bba6139 : SUCCESS in 44m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d238dbfeb064c81b5f3099e33f4e544 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":66},{"id":"22dba4573a251b7519b415b37cfde7f3ca970162","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-20 07:59:48.000000000","message":"Uploaded patch set 67.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":67},{"id":"5abbf2942673c4cc2084a9b000bcf33eb7e9c03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 09:34:27.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba9efb56deb1441b99cf0a93f210ab43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7e8c9e3a714bca93bbd872052cf763 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f6019bbed743a8bcf70dba88f3ebf5 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd1bad9673a4546ae32e8d14f69e7e3 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b3aa6fc9b6b497eacb462bf2ca52942 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0159e750077548bf92d68d0484c42d4f : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2825d24252644ab5a88a81e1b7df772e : SUCCESS in 3m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/311baf64251049999cc285c37b595b88 : SUCCESS in 16m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d3e895c951424174a179c6a409544b82 : SUCCESS in 9m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/781d197da97c45569f768c4cbe4ff670 : SUCCESS in 18m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3cf81cfd86d946b2970af2945063af36 : SUCCESS in 22m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/db520be318af457982e5a3234b07949c : SUCCESS in 5m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3f1dce2eb94c47e3aa820e64ada19dd0 : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e6b191c144e9494494da0370974e2bfa : SUCCESS in 4m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1423cf38695d404ba170e47f5daf34c2 : SUCCESS in 10m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9605faca9c5d4ca0979086058fb375f6 : SUCCESS in 9m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/20930cf3bdb04c6e860f519e6178f419 : SUCCESS in 4m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/244de70ea8614084a3d48a5e0c1feef3 : SUCCESS in 5m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/61abc800b88c4051ad6d693917b9cc5d : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c71e368a200406d90d02727923ca34b : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e2af0d16122b457c9404b1a49ca2bc56 : SUCCESS in 7m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/756555e097cc4f98bd8e43cb04e43956 : SUCCESS in 11m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a6a5697c85624e22b93d7ae9090d8742 : SUCCESS in 3m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba05679159bc401181a24ef786f4def1 : SUCCESS in 7m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7854ffb2b470408dbc98dac1d2979327 : SUCCESS in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/64f2f1cfbfe547219c36b3674c635aa2 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cc0ac07ecd3b445cb606f34f8ec75998 : SUCCESS in 10m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/103a51bc4ea34fb98c1b3435a1129bb2 : SUCCESS in 20m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f1a11bc4dcb425f9c42cc0429912f8b : SUCCESS in 7m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c149aab4968f47f18c14f83af6d65ea0 : SUCCESS in 13m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c990ffb4988c43e5b3a28050c134aa7a : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0285aaed70024c24a3770f6dd489215d : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dbd373f62f442fa9f81f7d93307829d : SUCCESS in 24m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2f44123145a94239be874ad9becc0f68 : SUCCESS in 5m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9cf912b692b14de8babbaabe75a14404 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/962cd939ac6f4bc3bffafcd94b9c64bb : SUCCESS in 5m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/348a40595e37468498ca50aa47f78371 : SUCCESS in 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af133ef20f1a4da8b569f1d3e603d629 : FAILURE in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57819e42027e4826bf7749673091c513 : SUCCESS in 1h 31m 29s","accounts_in_message":[],"_revision_number":67},{"id":"65f5a783e910311d4696f9a915d164069a6bc29b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-21 07:40:34.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"8caf708f04ed79c8c4961e6e3cce1ec207d08458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 09:25:55.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c791ee6321cb4f9a8bca91416bf9c551\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5d2f9836a042a487fc699fe8ab3d6b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c975cfc07341119683fdbf709771df : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8f4e1f1a7d46a0b28c30438114a52b : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/179da73e79f24199be28903a2e3998bd : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e062f7cdbd1458896f81124479cf3b3 : SUCCESS in 6m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc54d6b02da14a52aa7af1cdeb115791 : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1ce24b811d3413ea064e0e0d796b80e : SUCCESS in 16m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e096b45d6ed94d37af1f28ada3b3a28b : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c23229e278ca4517aee0e8b04f59b471 : SUCCESS in 15m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/34c256c5775f40c8b1dbb34bd7b92c48 : SUCCESS in 29m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f4f17dd876524899be46b6dc697905a8 : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b1dc004b4c354bb5a0d2a09b2336d288 : SUCCESS in 3m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8128487f545a4c91aa350f742c218db7 : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5888ff021a644e928ea92c7aeb76968c : SUCCESS in 11m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0ef8a631600c4818915118a452c9ba74 : SUCCESS in 6m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c16a421c8707480c96c646a0867314ce : SUCCESS in 6m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f8e2e017c7fe4c4cad5d5ccd1612f6c2 : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c4bab4f813bc47489462f1daeeaf0936 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f244590d9604fb98131494fc15f5522 : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7b79fe76a8e64ffe8faef2b3ce7c122a : SUCCESS in 7m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3985c7de26024a3496ce87dd36c244ca : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a12fe8f5bf114792bcca6fc7e8a60c89 : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9527e1ca21fc4688b20800fbcffe730b : SUCCESS in 8m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fc3fc8a875114ef18aa6f1a2d203cebf : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/182218f589f84964b15927151a9bea94 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/394b4122247c403ba8889fa81c91cbea : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8e45d3f4865d4a68a1bea76b5200caaa : SUCCESS in 18m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e1d82e161302471c89a23be86a00ca30 : SUCCESS in 7m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f46718c1534a4fcf84fee55203915bb2 : SUCCESS in 15m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b430212cd24e46b6a7daa0a8f5fbb5fb : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f455963284c14171ba818cbeeb498601 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e07a912ddc7d4a10a4771ec06f58c153 : SUCCESS in 26m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f382369965864374889766a7f561c2aa : SUCCESS in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/64fad4bd8ce940cb9c4157c700177ecb : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc917be4e6134c90b9240bf4618df1b0 : SUCCESS in 5m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b7880aa0f4154ad28acc2736ca71e699 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea8cfeacbb094c3c9d45097647c9aada : SUCCESS in 55m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcd67a8159054402ba83e0f42749e9da : SUCCESS in 1h 40m 42s","accounts_in_message":[],"_revision_number":67},{"id":"21e032009e4c535898f04fac707149ab255fde8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-26 14:15:32.000000000","message":"Uploaded patch set 68.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"a6b1fd34dd06a375c2a79f7ff93c11c5a41b6bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 15:18:43.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b74285299a24c87af71d1691fdf4f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294b8ae934354dc9a4a660dc2e870985 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5d62242f364fa38a790bec5d334ab7 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b617e79919ff4f19b4d33355439c652c : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6a8507263df4a1e9309c9aab070398b : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33270c76d6114c79b3f99ba681acfe3b : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a409b304435546adb3a00447c4f8c6fe : SUCCESS in 3m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9dc189df2f1d4bdd86e13d0e6260ba65 : SUCCESS in 11m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/80c307c6b8f043b19fad0f3f9391fed4 : SUCCESS in 6m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/78c989d745994955b03b2bb2be313598 : SUCCESS in 16m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/67eaf8bed2ad4b179e637d579cd140d8 : SUCCESS in 21m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/76cf7679123c4839ae299021614783d6 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/06507e01be7f4455860a1a0c2ea85624 : SUCCESS in 5m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1074111dadb14fd2a941b8160db99df9 : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3dbb77c52d1d4e93ac7137e1bb895d03 : SUCCESS in 10m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5bf1de117ba346a5b6fa124f9cfcb7a4 : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2994420698c6497a8633dd1ba39afa22 : SUCCESS in 6m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f056d45cbc4d4adb8ca0558c6134ab9a : SUCCESS in 5m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/126eca7d2ffb4631a1977e185ee0c357 : SUCCESS in 14m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b8d2040936348fb92fc43ea1ccd1d30 : SUCCESS in 4m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/49537a8dcf3f4c15a5aef9f982b3b642 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/87df926597494382ae1f62c861e821bf : SUCCESS in 8m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/328c11a1134e4b999169bfb069e4c5b2 : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2e727dc12ca84fb382cbaa73a0b75a9b : SUCCESS in 9m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/10d3642511eb4bef87997adf75e03a2a : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3caaa65cca8e43428c56a15d1eb1b5c7 : SUCCESS in 3m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/50f5746333ff4c47b0a5f1ac74688b7c : SUCCESS in 10m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/08c77ccf73db4e2fab1f976cf9561d1d : SUCCESS in 21m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/73628e46565f4f4696cc3834af257d74 : SUCCESS in 4m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a5e02c46b6ca4ecc89033c878239def4 : SUCCESS in 10m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ed4f68c38a684212bd3c2db175d42d6c : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/584376142f214c859d0691b06ace8564 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a57c111c2834e4cb00a9416c9123fd9 : SUCCESS in 28m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4dac59342bc84e53856d40b3f2fd9e9a : SUCCESS in 5m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/718ce38017de45549e6b00fc6ab50792 : SUCCESS in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3db76fd034934174b9e80c7600ea5fb5 : SUCCESS in 5m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2677e6844b4d474e82fce123fb8dbcf5 : SUCCESS in 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c12b41d6b774a4eb5121580e8d682e8 : SUCCESS in 44m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/808d442457b6402598e17b6f4e2fdd24 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":68},{"id":"e92a43db8905942ea9cff24bc1355ddb5daec9b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-01 20:28:45.000000000","message":"Uploaded patch set 69.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":69},{"id":"25cfd46298cbe26e30510882c94ff008c54eed55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 20:54:03.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9d1d6ce16524ddf9673ea1648937c02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c771721327d447eb7771a8e572f1f15 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/087dc34ec7074700baa5258879f40eec : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8352260aab43cb87ab5c9a1583c597 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa0365245f44acd99511d6a5d32e901 : SUCCESS in 2m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46ad10fc781247f6b16bd9cbae031b67 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88343b2812c34c30a4271906a53ab3b1 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f7aa4872453c4320a8e89fde4ee02bc0 : SUCCESS in 8m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1516c519dffb483580cffde38481e692 : SUCCESS in 8m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9616a454618545c09d2d11c9a41ef2bf : SUCCESS in 9m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/338106ac23224d4db9028740105ae7ec : SUCCESS in 10m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8482b09f62c94a2c9ccd273de58399af : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cdd9b2e06b8b4a0588a1ff39f852ffe3 : SUCCESS in 4m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/870c682dd38e44aba7703c2f7769f92f : SUCCESS in 3m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2fd3216c3f0e46be8cf42a8612cb2145 : FAILURE in 4m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4cc73630af2249f58fd3eade4ab381ff : SUCCESS in 4m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5ce781fe2b454f3d8aebfbeb733b572e : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8fd4536c3c7441e694be062b947a86be : SUCCESS in 6m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9813edc9216742dc8f244ebb2b974c3e : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98ddd6952df44e18a45134af53b4c130 : SUCCESS in 6m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/370cca256d8342b1ba75569e51dcfc6b : SUCCESS in 6m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/19eec5454a634a92b90a6db87a4a3717 : SUCCESS in 10m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e1147db6cec64ec1ad3b2237bc83e164 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a1bb24ce612048808fc730b1ae5fae7f : SUCCESS in 8m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/16ce9f39a2e14d03b10bc4a863482ae1 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cf26fbfe55bc4efd87871d36a5298299 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0e0a4ee15e6141ac904e8da22c15cd12 : SUCCESS in 9m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/821f7c9ee1c04413839a533fd5bece23 : SUCCESS in 15m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e203eb78318d455888fdb1c461e483fb : SUCCESS in 6m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b83f3c46d2a343c09299922db9911856 : SUCCESS in 11m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/89fcf497af6b43c3adb5b24838f7c402 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0040dad118b445e8b1b948be62130b9a : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fb8a26e422a429f945d2a5aaf827c23 : SUCCESS in 20m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a571723c867f49c99ba974d645d9d8a5 : SUCCESS in 5m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f92908082a98473285897dd8f6760227 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/139229a3a96e42439ae828d726f03dcf : SUCCESS in 5m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f682adc1153142a9865d86902e91d3b4 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e96e45cd4ad4dd290f6eb065e5e0267 : POST_FAILURE in 15m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef1e132fa3f242fe895b4ead35f1524b : FAILURE in 16m 13s","accounts_in_message":[],"_revision_number":69},{"id":"8eafc3051887c170a50acfc19c3a6ae8587f187d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-07 08:59:02.000000000","message":"Uploaded patch set 70.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"20724b8bf5e27858c55441bc418d18f028625719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:36:17.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9b62a2db0d41efbd72b011341346a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13b8d7180474f22996366a94c338ff0 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/426779568f7f40cbb4e85467fe3828be : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f79eabdc919d44e9943666d7990d8211 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dca25b33a5f2464a8d48db1c364d8a03 : SUCCESS in 2m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5006d56b8a9e43d188f8f123405f4f6c : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba1e8470072a461898050c5ee72bd30c : SUCCESS in 2m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/13b014cabc444971a3f81e7186cb4cfb : SUCCESS in 16m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/653cbc5e503f4e8a9e1bf57be844af48 : SUCCESS in 5m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/854c46b234ad4557b1a3231c674ee3d2 : SUCCESS in 16m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9dd865a29a6746988634ca08d9ef0d55 : SUCCESS in 10m 19s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/afe25202d32e476690c869158837922b : SUCCESS in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a3b2352233f947849dffcd4e2ad1cbaf : SUCCESS in 5m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c98e5d5691a34334a2acfc6aa2c2bfcb : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/647dd50322cf498d90d60c6d7008ef20 : SUCCESS in 11m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/990da257362343e6ac783ba65aec01d1 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5b422840a46142cab234d3c8298fe914 : SUCCESS in 5m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b5c4edb657b8409298f31d55b9d3437e : SUCCESS in 8m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f31848888bb243b7b82ed7234918e59d : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45a35a5d2a43476eb0cf846e1690ae69 : SUCCESS in 14m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5199dcdae9b441c1bf5d8fb9ccd811e2 : SUCCESS in 8m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6d2d347c7d334a21bc252aa08e0ac2f7 : SUCCESS in 12m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1fcc4cd125544c7e974d03a640f8b22d : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5dd838fadaf64b9cb5c5f6033f0bc4f3 : SUCCESS in 10m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7f7bca1712a445149e42a20159fc1950 : SUCCESS in 5m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/84ce0ea829d448d2a2006d1a26beaed0 : SUCCESS in 5m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7ba3635885504f89ae985b6515ee204d : SUCCESS in 10m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c51e6ffb1f254aeba7c9c49d02960651 : SUCCESS in 18m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b24439ac3d1c4f03943890f7eddbdb52 : SUCCESS in 7m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/55e6897cbb924275beed9303e4eb4cc4 : SUCCESS in 10m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea9d94e73ec743f19d27edc8f9344c7c : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f357a07cc244a778cba371056018901 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a510f581e5034fbca3e0e05b35b67240 : SUCCESS in 26m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/73a987939f6e4f1db799000431a00790 : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aacc1d86c6fc49c8bbb698ebbb9b4422 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62601d128d2546fdbf3949fcdb61d58c : SUCCESS in 5m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2abcbc1a6213474881b9fac65d55ea29 : SUCCESS in 16m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f811e35078404da04757cb82f5fcfd : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d70c5688e9e1471fbd580f9ec24695a3 : SUCCESS in 1h 35m 58s","accounts_in_message":[],"_revision_number":70},{"id":"6fe592423775f821930944cb44a8a31bd0ec50cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-15 12:46:48.000000000","message":"Uploaded patch set 71.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":71},{"id":"eb62b93c627d74aae7cdd0eb14ee28fb66ef83f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 14:50:07.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df77406ed694ff7971105c31dcc2570\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a180c7041e4a5e94dba08bbfd1476e : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acb842d84394cf69c85956c6bb41264 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e9a75e248b4992ba89dd5b92a4e71c : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/412f300462f841b18c5249a1faba705c : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3086c88f94e44aab800a93a604740f1 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2259023830e54f6389c58d6f79ff05bf : SUCCESS in 3m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/39d501dad9df4ef8ba8841f56e729fd7 : SUCCESS in 15m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1095782d00c84ff7a93aaa7ec4d1476c : SUCCESS in 11m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e2cc527fc70b40bfb8f2f891fb56b093 : SUCCESS in 19m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ee1d4b00c5544bfb89651f74dbc3563d : SUCCESS in 26m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1f20e3b1895242a9a062de49d5915707 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71454c63386e49e792c0ed554d490cfb : SUCCESS in 4m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0cef4a6daf584532880b13ecb3ff7a24 : SUCCESS in 5m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/38f7559e8a8b47d0ba8ece6718590f44 : SUCCESS in 10m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f7712126207041d3bb407e0f0688848d : SUCCESS in 4m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/81f3a0415ef04df292664d6aa605f2fe : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/be8f4352540c4816824ca6cc9712059b : SUCCESS in 7m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c0d138fe31c747878071f24324da6c96 : SUCCESS in 6m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b050f1f47df44f41a5e423f40e2b2cb5 : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f0f47897a89a4b91a017f92b21cdd081 : SUCCESS in 8m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5dab3048feb741a485c857b3a480000e : SUCCESS in 9m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5d8c6a8522ac4a3586c3a2626a937005 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2e3c10099a484daab76ed3b106bb5165 : SUCCESS in 11m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/84b9d46e4b11467b991c4659d74174cc : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e3d7453693ec47a5b491900b49fd7f29 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b3ce5b37b675438385e4d0dc7c667b75 : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6b10b9ebde0a4878b41537c811d152e2 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c51cbbc84a5143b896c4443b24e2f550 : SUCCESS in 5m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/388d0443763042ac829994c0c42f8fdb : SUCCESS in 14m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d95e99638bac4bedbaa8b683639256f4 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b321eb46fdb647609a5b2c04cce1a5e7 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7cc75af3715487ebe746b6c6277ef15 : SUCCESS in 26m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/52a7bd2fbbe543f0b0f9ac7e95b6f608 : SUCCESS in 5m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8b30bc856a6f439199bafaf51a4655c5 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6285bbb41567452ea550f1fe0a6e4a9b : SUCCESS in 6m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2d37a969ec1a4b0db41cd99a05f20ba0 : SUCCESS in 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ee62cba92004a1e9b0f096e6820e1ce : SUCCESS in 49m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c1415175b5942719e9e24436ab8ca48 : SUCCESS in 1h 57m 30s","accounts_in_message":[],"_revision_number":71},{"id":"414492772d7e247baa9ee925d22b8b4c294c3674","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-22 09:45:35.000000000","message":"Uploaded patch set 72.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":72},{"id":"752f478bff9b75111fe7494a328e7044ae22e6e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 11:29:51.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9237d3efba84d3a9219210b3a5aa6c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24ef003d57047e39ec194755238d32b : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b8b2e57a93245a69f67f7c073aab29e : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b22a8f2be86b44aebf72262e324d23d8 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86ce377819f3443aaec28251cfc6d5aa : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6035e32c1de402db4606261b20846fd : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3613fd0b3e7e45be824d7dc57c2559df : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/81024cd0fdd649478f6d69a7f6922cc4 : SUCCESS in 16m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b7a872921d9448e2ba91956ffb317336 : SUCCESS in 10m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/544fda473a4c4284bb5abd720a213a6c : SUCCESS in 16m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/79b555a3667340c391aeb8958c550763 : SUCCESS in 15m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/606a52d7324e4f439be2ee33c51a9704 : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c94d0965c9e4441288bf0415b2c44724 : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/76fe8e6326ef4161bcb815fe44dc50b0 : SUCCESS in 3m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d0233e0891b744cfbfadcd73a84053c9 : SUCCESS in 9m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1b983a4e2bd643bea579d90ae0b763ba : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/86dfcbf8bc2a4551935435eded345acd : SUCCESS in 4m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fab249978e684322aed2976cbfce86f9 : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ad3361436efe4ff885e5833640fe8747 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dab67f835d7442ccae137a1a3167afb4 : SUCCESS in 6m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f90fe7540a28428d98ea37c4bc383855 : SUCCESS in 8m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/650ad87932144f93a072fba332810e24 : SUCCESS in 8m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1ca257ff8afe4ee1a191f512b14de7d6 : SUCCESS in 1m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/39536cb84ea54925b426050b472e646c : SUCCESS in 8m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8bf3749aee5d41bda69d6cf461a267c0 : SUCCESS in 2m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4f96145bbd8f4f5da69b6d5d5753c806 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0d336ec66c484e1ab2d7e156d8601e4d : SUCCESS in 9m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5dc99e7240c547cbaa57349917896e52 : SUCCESS in 15m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5458bb6236d847a98bc53f9d37923bbc : SUCCESS in 6m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1721dd4923384a629256c14c12bf6472 : SUCCESS in 12m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5beab873347c49ff882f6365baedf4a2 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc7c5a34368346c98eeef75f47e85407 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f01aaeeaaa2845a5bb275336d19a4dc7 : SUCCESS in 21m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8fd368cb63e240cb8b59ebbe2cd2f2a0 : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c8ca0f27ffa141a3a7e8c657133d6416 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ca29c5358f04be3bd82331e920e6972 : SUCCESS in 5m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bea8eb582a794bef8152a25b0c0a9c0c : SUCCESS in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f98d23a0a3d842f5ab886057289e7fe8 : SUCCESS in 47m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97c809e12ffc4f349d2a7658e3f10c71 : SUCCESS in 1h 38m 26s","accounts_in_message":[],"_revision_number":72},{"id":"0b1eda4094b81e5a642a773f301a2c4f3e43a063","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-03 04:47:06.000000000","message":"Uploaded patch set 73.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":73},{"id":"33c374aeee907c6fbdfe5d157791c0be7a085179","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-03 05:39:02.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"933767283b980fff16b7be571dd934015a9ab16b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 06:37:28.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0954fdeea024471aa4730c9059b5db66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f365b151d8cb42b78919ac2d9a4b6c4f : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15c3a1c94b843b4b99893776bb398bc : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c204048a9d034716818bc63da8c1d3e8 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c331c5c7a7264dd1b1505b0e6ed47033 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db6e082358d042d8a66faa4e136cf666 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbbb4eb60b684d6f987fb626d070b53e : SUCCESS in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86c3f5e2904b4f67bef0f313931ee303 : FAILURE in 4m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fd2380e5c130400db61f4c34cf5baddc : FAILURE in 5m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0056ac4936ba42779fc2087b78981a63 : FAILURE in 4m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/81f31873f44542d9b9d4cd00e30ed0b2 : FAILURE in 11m 25s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3707f1bd7fac4348bc6a861ff5437749 : FAILURE in 3m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b0761eb6f26c4289bc26342a07bc45e1 : SUCCESS in 4m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4f3fe206dc5c4d2ba631ba61bf9e39b4 : FAILURE in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fc5d86bea9254c028705c20773e215ac : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/784b6ef939f344e19bc56f5ac472fdb9 : FAILURE in 3m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a5a1d17d9c164eff8d0ff6f5fc7bfdd2 : FAILURE in 4m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/846877499bb742aca6338a6186980de8 : FAILURE in 2m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9b85751586a944c1baf1552794fe2c8e : FAILURE in 2m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0fe659127564f2eb8bc02b238be883c : FAILURE in 4m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e473c0bbe9f94c23955753407387ed80 : FAILURE in 3m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/72953755d0754b7b9a69f7bf53e4fdd6 : SUCCESS in 10m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d372a6ac17f84e3186e7dc9a6fd3337d : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/acf8f4e0a0e74341b057b1d6ca6c26ba : FAILURE in 3m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3a7b5b66dd904d8ca6b12b524ba76e4b : FAILURE in 2m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/22eec9a54c414f3aaec136c4b852c363 : FAILURE in 2m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4fd00d035c524e468302b278059b90bd : SUCCESS in 11m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c8e9a9e11c2f45fdb1f9190f0b2a2b2b : FAILURE in 3m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2926fe499ab14d6f8f42798ea94ffd17 : FAILURE in 3m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/40de442c84f24ab8b7618674c8b21443 : FAILURE in 4m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d5c297ee73964e29a322d55142a270df : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ad4e42be9e947cf9a333217d457641c : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57a84755933046c3bdfd79d78cffe87b : FAILURE in 4m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4d062a8736d145fd8a0d402c74cd0b1b : SUCCESS in 6m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d74907b05c8a4b94b9e46ef814c7bbb6 : FAILURE in 2m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1999d11a7cc144fd86bf915dffce091f : FAILURE in 2m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ede84413bfc84a2e8e59f288ddda73b7 : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4f2a796a538473180e4c299f67bb3c2 : SUCCESS in 56m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bcc45471ad146c08836806af6d457ac : FAILURE in 23m 59s","accounts_in_message":[],"_revision_number":74},{"id":"8c75beec8167c8e4c1bfb8a95671210970dc8bbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-03 09:35:39.000000000","message":"Uploaded patch set 75.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":75},{"id":"10805d92643facddabe57488cc51249784bcedda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 11:21:32.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02042895233d4337ab571257945265a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2103bfc954d4adabe47528c0f6fcc2b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1604b67b3d294693aa13af8758c6ae85 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cd787963024478c8fa4f6e045fe7923 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d951eedb2f1415fb25e387719844855 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f4c71eea6434d80925a017134495b68 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99557f26a4bd442086c32efa48ef79fe : SUCCESS in 4m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/378d32458bb940e48429aa08db1ed91c : SUCCESS in 19m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b3830b5744e545b7a68a9d6ed96f52d0 : SUCCESS in 10m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c9545dc73aaa4b09821dfab573eacc2b : SUCCESS in 13m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/628e7084a6674bb59917746d77c7388e : SUCCESS in 21m 08s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f205c9f91bff44d098de3126092ef51d : SUCCESS in 6m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2bce2d7c2183409fa8c6a9137d19663b : SUCCESS in 5m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/63d6c03628c94448921a2b4552bed54d : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/950343959da1437188395ff2456c9dbd : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1d1cef97724649279e19bf320e439ad9 : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c784b8170862442ea90e5b44f11e68bc : SUCCESS in 6m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3d4c894652fb47b788e0aa41b9b9e831 : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c80c267f4af14c98abfdad792b3cfc2a : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3383dd0e5a16476fa90cc48ae5b43b03 : SUCCESS in 6m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7b42b11a5cc244e5a51061e43aad76a6 : SUCCESS in 9m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/80abccb666b04d3891157aad09479b33 : SUCCESS in 10m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/190e3f6098664dbdad119907eda7ea4b : SUCCESS in 3m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b8c2093a360248f2b0726367adbe3d94 : SUCCESS in 11m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c818d06721d645f2bb012b26df2423de : SUCCESS in 5m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/650e3d376d114e09b29d7bcbaeef29e4 : SUCCESS in 5m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3e0ad7e3890e40588a0f08b1f8465ffc : SUCCESS in 14m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/636eed63bd4347198d923053b52fa102 : SUCCESS in 18m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/52ee8fee7e3a43b995b4511f77088c9b : SUCCESS in 9m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/11990f74c7574656aa5a9c8f9884b826 : SUCCESS in 15m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1694ec46ec974fcf8034028e53a43ef8 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/542f42c7e2b44587b9459a0b8b47bc72 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ebeebf444e74c739abfb88895aa6eb3 : SUCCESS in 23m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a66ab8db9c9b426d9917e7f417dd4150 : SUCCESS in 7m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cb8041b894d041fdb551bf85ef2c75c0 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48df377e8a6e4db5afd30b0ca764f378 : SUCCESS in 6m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/26eb8aafba0b4b5086a4020837bb2134 : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ccf78943ce5436d89acbdc8063e1bb4 : SUCCESS in 1h 04m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1359bef2f35a4327956f9b9742bf61bd : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":75},{"id":"e885a45769e68169d6321b74d92378df12a566ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-08 09:18:58.000000000","message":"Uploaded patch set 76.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":76},{"id":"ad201f5bcf54f68c9b1e252019faf1b1f60f84f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 10:41:36.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0452223504345838bab9376e86083a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e296ca895d445bb984ef7a7fe281b8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153b33e7aa5847c394a3aaa421ff2900 : SUCCESS in 2m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91f6586a2beb4c76ae80190a166f1b9d : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/766fd0582c49446780e0e97b81d8ead3 : SUCCESS in 2m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc883c133cd34c1ea4b4c0d7ff63042e : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a05e0b729fd94c31b9d2016bc55d55a9 : SUCCESS in 2m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b9e945e062e545d789275cbf3554c7a5 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1de408f631e74175b783485646ac728a : SUCCESS in 5m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ec5d4054183340ad9eb7e61b33ffd5cd : SUCCESS in 13m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3b276fb61e9f4a06ab14a0f28dc2af5a : SUCCESS in 10m 38s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0721ed5b142d4d3398f6abb1a399bd1e : SUCCESS in 2m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/859dee166fd04b86a26bff1116537a1a : SUCCESS in 5m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dcff52deb7464c52bd9e16884d6f263f : SUCCESS in 3m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1efc5052dd2c4ff19bc3333edeff2b9f : SUCCESS in 7m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6e22700eeeca473eb8de2302c1c17959 : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b3f2f5c253c4cfc89822e998d46453c : SUCCESS in 3m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bec8ca99206c4bbabdd5f42bb8f0b9c9 : SUCCESS in 4m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ddb5bbc06599457a9527a40bc76c73ea : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c139b2a9b9e4ec7b078f2b807b76997 : SUCCESS in 4m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fe58e75a21fe482d8c067f5d1ecda9b9 : SUCCESS in 4m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/514674d635b74ff2a275548a73091c40 : SUCCESS in 11m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c4a4579fda86458caf0fc2b7a7bdccd9 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3f74204c20414a159c9b88bda229e043 : SUCCESS in 7m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/749e5a5da0054ed3b3cf7306dfa0db90 : SUCCESS in 2m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/db1d14ef8c9c4459b969e9ba3f6d8fd0 : SUCCESS in 3m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c7b58b92b854414292020f1a37b26e80 : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/490d5dd6b1e7405f93cc025889a1695a : SUCCESS in 18m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e24bfb699f448a09dbc032cf4b5d8da : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d6d3fe4179b1413ba72d8464e56aafd8 : SUCCESS in 12m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c5fa5c99fa1847a6a0b71f4a37c95dab : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c5146dd49d142df9b2882dd329b7c6a : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc7aac6fa1be471388ec75f8834a4561 : SUCCESS in 23m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b691e70459b9434c853ad82f87aaee89 : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/380bb43113664995a23e0f6ab709bb1d : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0fce5746d284bc5a0012cf3dbe7a41b : SUCCESS in 2m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/990fdf0a6e4d422887177a2b372053da : SUCCESS in 19m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfc37d13294b4cef81e4b47ad20179f8 : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b45b2e688b941f48d16b12f5904b078 : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":76},{"id":"43db79ddab572c21be07d87eab2dec2520a82d3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-15 08:44:57.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":77},{"id":"d6826617087fd9841e9116ddf1f70215ce81496c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-15 09:54:23.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264cd143aad543ff88a019f348b6042c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035692c4258945fbae1bd7026f96e6cf : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc83cf077f34c2f8b9c2a6914b05b5b : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbfc88df18c34192b2b63ea519116507 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/160670c372244aa7b980868266b5912d : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/317fdb62c66a4a53940400845a45d0f4 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aaf1ee5000b4409dbb8dbf425b99e526 : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/87a58567ebd241a5a9b0f235e6e00d07 : SUCCESS in 17m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d7debe62ebbe478ea4617b8db7368807 : SUCCESS in 12m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dfc8197dafd247a69a8e711d29e90c27 : SUCCESS in 17m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a372d70dc9a941a29847620bccaae043 : SUCCESS in 21m 14s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/96698845d4ad426eb0378f2142878b28 : SUCCESS in 4m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/574423ae83064acb93b7a19a46ee21c3 : SUCCESS in 4m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1ce8091e10c74b9e9751121ccd62b41c : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bca59b06675c4e6c8aedac7c692507e2 : SUCCESS in 11m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/57531c88db674130883662a340fb11ad : SUCCESS in 3m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/08adc323562f420cafc23217ba928780 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/16dffb3c2d7a403ea9fcc760e254a8f4 : SUCCESS in 5m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5789ae465e0b47a69df3793a5d268ce9 : SUCCESS in 14m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ceedbc9483a43a1b61ce4588e706482 : SUCCESS in 3m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/09da1bc686474a159712738adf3902a0 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5b70a30b80af4d7b9c6984230306ccff : SUCCESS in 8m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/158b71ec459544f682d23591d150b2b4 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1a1808a928024179a66b6e0f417542b8 : SUCCESS in 6m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0c045045e47e472f814950923a91b186 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/81a0117753424fa1896c648ba678de5b : SUCCESS in 5m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ad8a0a525f0546158df92688d7614511 : SUCCESS in 12m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/22c06b415b37442bbd86fcd095a343d1 : SUCCESS in 18m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f82aafe6e0f840adb50a9530c408ef2f : SUCCESS in 8m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/76650b532d774adc84d182db20b915c0 : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e65783c1422c4f2d9dcf7e6154120f57 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e29d4757004440da4ba3840bc13a3de : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/935e03fc6a634bdb81a191cc60f05180 : SUCCESS in 26m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6bf7b679ef474a90bef5444af37e3221 : SUCCESS in 7m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/de2edfde948d4ea49f61af1892d264c9 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3ab231b2df54743ab26ad81a7a573b4 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ebf92d528b2c433d89ff67df133e17c2 : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74bdad527e7d4735bc1907f40885f7c2 : SUCCESS in 1h 03m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54b9f59e251a4977ba18ff8e24545afd : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":77},{"id":"7d63510bdd78ec80f6703d05f3ca600def5b37ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-22 09:51:44.000000000","message":"Uploaded patch set 78.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":78},{"id":"c067548ecf1f03d4e41269092a712b9eedaa087b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 11:43:42.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8151a4aee8564ae5874fb83f4d88ad41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b12a565b6874d86890d282749584c02 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20878c2c934f4caf80fba475ec62d6c7 : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc3b87c519240d58cac9e908dd5db32 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4df732c5fbf14f10a41e9325058b5535 : SUCCESS in 1m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/750f95ac3ae74a469869ccf787c5a8a9 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df818a5aa4914441a4f360aa051fd657 : SUCCESS in 2m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e58b6b42a7194b9cb2a3c6a1a778403d : SUCCESS in 13m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef1a642235eb42dca9b0fa1932be54d4 : SUCCESS in 8m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c491706627b9404398d5bf6fcbe812d2 : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a75c7ff6a22a43a29c09ce5f782e65a8 : SUCCESS in 20m 12s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3fa4cc2732374909b97ef3dc8f19cc72 : SUCCESS in 4m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/45507a130f2741f6848440e58c00cbf0 : SUCCESS in 2m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ecc551e0b11b472eaa70a0bfa1ff3c14 : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/296ca3a1dd4442d48b0aa900d9f56705 : SUCCESS in 6m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/28beba8dd04e4d4ca1f04a454cdbfea5 : SUCCESS in 4m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8f68fa4dec284d47a0252e1f05ab9189 : SUCCESS in 4m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b38bed9cd16a45cc934ad4e181149921 : SUCCESS in 6m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1ba82d413a6a469d9e93947be07218ae : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c359bc35546643caa2cf7266c12a3736 : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/88e62ff5bbbd4a69860060442dabb58d : SUCCESS in 6m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0ac7f34faa7a4846bfb6aac7dad8ee0f : SUCCESS in 9m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cf470dab3e8e4b9791cde64e0434f54e : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1f74dc95f0f54416a091b59806f5465a : SUCCESS in 7m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/450fe3432f3b45368dd51ace88e9f0ab : SUCCESS in 3m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e4700c9e6648467d894d6a1ca264fe64 : SUCCESS in 3m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c7157b09629f493d95c3ba742bdac821 : SUCCESS in 12m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/69257941f0cb44fba0e7c31e03033e0c : SUCCESS in 24m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/26c66b2865b1410ca3e7994f13dfe923 : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d1927782ec1749879631dd6e72bb0965 : SUCCESS in 7m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3a8460be83d1411982684e1e1bdf18e4 : SUCCESS in 3m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4de7401cd1454dd2ac8ef46bea0b5a69 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a11d5d2622d486a904230eafcdc2e42 : SUCCESS in 20m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c6030ba53de6407f9fb84b65257bb8f3 : SUCCESS in 5m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/beb9e50292c14378a81264839a7d04b7 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efe78abe32a947adbac12a16e7fdc1ed : SUCCESS in 6m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/39cf145c47354444a86704fbe4235ba8 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aea21481bdc04ae08a37ef2c3a080c58 : SUCCESS in 48m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8544311c5224afb8f25f9a09d8441b9 : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":78},{"id":"9f43b8111944e0b6f475fb304f0404daa2b66218","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-29 08:45:20.000000000","message":"Uploaded patch set 79.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":79},{"id":"51df68058e4fb35f7788a19b003874c921ddf5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-29 10:23:19.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8458846964148c09dbc21125160155e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84bc01feb8c4ecc87ba07ba8c1fbefa : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee910f77bf84ece87087f3b0157d2ae : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c48aace7c51469cbf23ddcce753792e : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2fc913d242c429d9022fd81e5a97b34 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa25fda74f4d4a43a02e5f6e8f466058 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd95c19f635547298d215117cbf08b0c : SUCCESS in 3m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be6eceab4cac431a90c7d79d0dd0fbaf : FAILURE in 4m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/33d919b24a174a7ab8144c763c7ba6aa : SUCCESS in 13m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/41c37ed7b4bd484aa0621736b1c29150 : SUCCESS in 18m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ab220de2e8a942ea9924b7d656a21117 : SUCCESS in 8m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/68a31500ac2645aa94caa24681c4fed7 : SUCCESS in 4m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/455fbd0c74ea4760b35138818037ce61 : SUCCESS in 6m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ee2efac31241489ab8e8bc8216fb2c88 : SUCCESS in 5m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d2004a74e7c94c588027bbea31d29822 : SUCCESS in 12m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/82e999473b034b7a847e3bae00a134c9 : SUCCESS in 3m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f04e34ff60d40aa8db46875d3370e9b : SUCCESS in 2m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c6cd130e41c245e2820aa55d53cb0440 : SUCCESS in 8m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9ed34448db054bc5b19152917a401dab : SUCCESS in 18m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/247e6697b5c74f7ba1b569c640374558 : SUCCESS in 3m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8ed6f74b8e8b4e4ba3508ba2e0a11649 : FAILURE in 3m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aa13dcd0152b45f29f437244e634df95 : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/78bffc23192042928b191b0a2c933f78 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8018c52a0d674c45a18b4d23a9eb4399 : SUCCESS in 9m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/98d337149aac4104ae67490060ad30a1 : SUCCESS in 4m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/635a44fd06904b3dbc1b555f1382a0be : SUCCESS in 3m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c1af8a0b34e74ad3b3f1dff1b9b4fef9 : SUCCESS in 12m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6737026c8ab84f668b5d23af4daf3e22 : SUCCESS in 19m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b72d8a51d3c408fb15dac4a5a0d998d : SUCCESS in 8m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6a870e23a0b6406e834a0bf280b8d559 : SUCCESS in 12m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/31bf8d58f7b3474cabc957c5502cb830 : SUCCESS in 3m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f1fb47851b049efa4038ed118ebf59b : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e09e28f379e84a01b941fb194bbe9caf : SUCCESS in 24m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/101787afb361404fb518c0580f5b8421 : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/900b22ac9dfc49398830472a5725c6ac : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7d1f95f5ade42babacacd1ce0a64637 : SUCCESS in 2m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2b54bc2ea35e45ea907ab46cd8d96a50 : SUCCESS in 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b49bdd43b774a0aa05e303668518777 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f7ad17512784b649cb3b9e85d662199 : SUCCESS in 1h 31m 44s","accounts_in_message":[],"_revision_number":79},{"id":"4b2b0434b540aa1ad53b232d1f8221a59c2b399c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-04 12:39:10.000000000","message":"Patch Set 79:\n\n(2 comments)","accounts_in_message":[],"_revision_number":79},{"id":"2ac33bbff0a5be8aa8af0b58cc85db6e618eafde","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-05 15:32:10.000000000","message":"Uploaded patch set 80.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":80},{"id":"ef0c4970bb11e907c0a5eded8707020a5d339ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 17:05:58.000000000","message":"Patch Set 80: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40693fb676454ae9931a31b84cfdcd82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d880cdbca7346c096230701513c1cc0 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9a31191c6d43e288dd8ef37e798809 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66d96ec50f6e41089b9199279c70b2b9 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55ccd521647b49d58e3ec07a46c9a4cc : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/237457572e1a475ea96d8426b6bb2526 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e441e9e9ba74984ab05629caf776e23 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d32576492f8a4109bed90e1c1b770fbc : SUCCESS in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b5c67c46a084dcd8204b4f468c55268 : FAILURE in 5m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/86330063139f43288f499ef9e510c3c7 : SUCCESS in 8m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4bbd9799b7734c7f9c7f26286e0e40a5 : SUCCESS in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9a9be142ac0241c7840f404cd449357e : SUCCESS in 18m 09s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/acaa0822121c4656b2961cbff39a1c0c : SUCCESS in 4m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7c8b1b290e954da1beb0ab966fd01872 : SUCCESS in 4m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e40ebcce1b844b17bb09423e218385b6 : SUCCESS in 3m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c4b828218e7e4a719d3a8ccde7fb10c1 : SUCCESS in 8m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/08c3723b27cd401b99a74190914d5367 : SUCCESS in 2m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/96cbb9f2b15f4d5991ffedded28be50a : SUCCESS in 4m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/251f2c7d86ae4baab402c9dda490e7f6 : SUCCESS in 6m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/16a8afab54b64d37ba7e1d81b8681941 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97caab16469146b98f366891ac8cd2ba : SUCCESS in 3m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/127d4f4a2ecf43ca89660df1b9b80812 : FAILURE in 3m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/072e75b53a664d99a487ed15325aad03 : SUCCESS in 12m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/05f756bf01f94c018d53e0926811ecc4 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/90eab3325d2749c4bfd32f91b6a08b9b : SUCCESS in 8m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/66c3feff6d5b4c47935f1c4af9799901 : SUCCESS in 4m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dfc4bd7f5e4740b1b0a8ec7cc41ad312 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f2014cf3413843c2a8a86548e44fa73b : SUCCESS in 13m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/feb39343e93d4dce9d4bfd6676148c9d : FAILURE in 5m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9af1ff5f7c134af88af681b14b042284 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bf32ffd85ae84a9da715aa1e125b0d79 : SUCCESS in 7m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0b4c876a30f74c1f966075f8aaf8fd2a : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f799961b266d457c8aec9156fbe1ee3b : FAILURE in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1885f7812c134a2b87a4a29ddd457c4f : SUCCESS in 28m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6014a09b654f461689939c8b01b94234 : FAILURE in 4m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6acb22c667b24947b69367215acbb7c8 : FAILURE in 2m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c9e60a96da14011a815480d3dc674c3 : SUCCESS in 3m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/581c69b35aa444228d54cfbfa9b3b183 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8ddad834c644642b0325a67c1525ce4 : SUCCESS in 46m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/973cce1e40e046cea31e2ec9bda2372d : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":80},{"id":"58ad4f2748321211f86e39be6ebe1eeb06771674","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-06 07:52:40.000000000","message":"Patch Set 80:\n\n(3 comments)","accounts_in_message":[],"_revision_number":80},{"id":"a12f177918dac2699dbd1563b3eb6efc812367f8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-07 09:10:11.000000000","message":"Patch Set 80: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":80},{"id":"1608f019c79d6d4b8a650f932e49f25b5c899aa2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-07 09:17:09.000000000","message":"Patch Set 80:\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"d0151d5a657afa3542148752b2642edd3d02e97f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 06:37:39.000000000","message":"Uploaded patch set 81.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":81},{"id":"abaed9470a09280470e642946af42554ef3f6f5b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 06:39:21.000000000","message":"Patch Set 81:\n\n(2 comments)","accounts_in_message":[],"_revision_number":81},{"id":"ef3fa71c15b1764e8bd12d51dd052742f1d9bf24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 08:01:12.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"d0208e113d32597a3d2ffdbe1fd8b8518caaa80a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 08:32:29.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"20eec6b396f207f609f69e7724944c5f8a2be0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 09:34:08.000000000","message":"Patch Set 83: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6ed9f7b16d4a759132f56b2e857914\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7171c6823941a391fa3148d7e6e2cf : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3292962ac12040cca662b14b5a3b2b11 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4eab925f2b4b4df78ce3d34f3d76e1e6 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ed956c5a0764a10a7b142443429fb18 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1eeac3f05b5e452d917495b6c22ccd11 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb2f2f23c6c34a03a2d022b283d8eebe : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b14c189db8c4690a6e2a1cbbd242d80 : SUCCESS in 3m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f37cbce2d4f64f449e41a6388d555665 : FAILURE in 3m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/591a4f45acf6416b9ebe65c3b8676d28 : SUCCESS in 5m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5bcb2fe601f545ad9a9b2a097a5413a7 : FAILURE in 2m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4a9b02ee2deb42619aa49d0e4ebd49d7 : FAILURE in 5m 20s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b387dbbea11d41d2b0f25f2203ebfb9f : SUCCESS in 2m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/580cabdc8c8c4b5790010e69a363498d : SUCCESS in 4m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7607a3a670f0442baf2c9dfadd4fcb56 : SUCCESS in 3m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2ce6298340a24080a63f3664e931c56d : SUCCESS in 14m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/576dd7fcb813479a97c5872b139cd89a : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6374b7a72f404f68bba31d824932215f : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/13c32e81dec7414caa347bc7be7c6de5 : SUCCESS in 7m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/de1e28acc31c496f8806fe5dc48c0f41 : SUCCESS in 16m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4ace1e545bd4972a839bb01b59b86b5 : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/37bb5258d7234c2fb3753796187cbd17 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dc89f63209cb4018974296c5f9a429db : SUCCESS in 14m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/21a6ab20689b4577a507ee8e7aa9c055 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a82daf8c1d5a48bd8cedf54acf7e8a5b : SUCCESS in 10m 55s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bebfc872ab6b4e6592b416dcf40606ab : SUCCESS in 2m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0ea4a77ca89f4958ba42e340bf08217f : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d958f053d9fa4dcf9d808a738ee4d4f7 : SUCCESS in 9m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/15c409b2dbc646cbb7633da2d8e0ca63 : SUCCESS in 13m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc4160119315456a971c0d8b0047044c : SUCCESS in 5m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0df8fafbb14841ba8ee16f664a6aa64c : SUCCESS in 11m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/45fe79e82712440080c32dcdd56b6ab8 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f1ab59ed5c040bc90d74b84a1a20346 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/754deb0a39104b209d9d1ade1229260d : SUCCESS in 23m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5f00c77e4fda43b4a4d49bbac3e0b86b : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/29c38f15883a480fa20c56f923c85ec1 : FAILURE in 3m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96d22c2b89524ed9b1dd03bc620f2466 : SUCCESS in 5m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e931d4646b5146a6bef16adc215d1a7b : SUCCESS in 6m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b17a54493804f35aa8633b66442a842 : SUCCESS in 56m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1cb07cadca3495bb1bda47bae9de000 : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":83},{"id":"c637c16f704cbef8f6cbdce47990f436d5c95aad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-08 10:47:14.000000000","message":"Patch Set 83:\n\n(1 comment)","accounts_in_message":[],"_revision_number":83},{"id":"6e963c2ee78b1a16ee51282fd5371a2ab84911ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 12:10:24.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6006806836f04668874ee628b6e43460\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c608f3b10e4609b96169b57a364e77 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5645d760f9a499cb41e70fa84d323be : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6aaea79aa0542c1b8d972f48817106b : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bc5948471e14666989742d80d0893e0 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac0e1fc10544e5da07060db96c11e4c : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61915c4597ca4185aa630bf90d517a98 : SUCCESS in 1m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e44ee4a692cf443381c3b4bb1eee62a3 : SUCCESS in 2m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bf1f1dff2bf7427982d1e8178914bc26 : SUCCESS in 16m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/080ad28787424a399b0eb5eaf2499953 : SUCCESS in 11m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3241d1fff1b4e8e8205662dada5ba8b : SUCCESS in 12m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aad9462d8b304cfda4f26f983451498e : SUCCESS in 25m 24s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fa4dead884a14d7885b784656c07318d : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b875b43d985a4539959a0bd8917acb9b : SUCCESS in 2m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d586feca5f45473a9db622287ce8e7a2 : SUCCESS in 5m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3c512b8d56d445a9850f4ea9a4530f86 : SUCCESS in 9m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/86dca8a66ead45fead92adc402bdb991 : SUCCESS in 4m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b6c593f8612f43ada084c8d2be6c515a : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f438962c4b0e4885af6d08fa9cd38437 : SUCCESS in 8m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/55985e944c654c3482a61ff5d4602920 : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac33eae66e3c4303847e63964efc0b73 : SUCCESS in 5m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/95715b7c1ec94b0f97ae981499460d25 : SUCCESS in 6m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a9d2f0b76583419a86edff337ddb14f8 : SUCCESS in 9m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/073a617e4567485b9afb2e0d903d323a : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d81bf74feb8a4b29aa984ccdd4fa78d4 : SUCCESS in 8m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7d0f333fdd954a5dbb5c53c3c5605d60 : SUCCESS in 2m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3b2863b3f6a1438ea6ef1b7f556ca4e7 : SUCCESS in 5m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/073c4ffe2f694b0fa0700d810a4b247a : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d242673de96d4755babc1c01b76be9d5 : SUCCESS in 15m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7b45336c2d284fb3adf56396989abad2 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/367c95d75fb54834be8611fc5b6cb5eb : SUCCESS in 12m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/39c59ff91eeb4eefa289c5be889745d4 : SUCCESS in 3m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77c87fe82c354f8f842e1a01a46d8a5d : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db449ec9f9914e66b56ac02ba536a53b : SUCCESS in 28m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/542071d04f144666ab37d2532acf080b : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d44c986f268e4276a53603c96e742b23 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5528fb046d7446f08578c9d222fe4a59 : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/79052c7fc3e54574b97e7df5db3bb9ef : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f2d9a68e324a0ba8beda93ed23f117 : SUCCESS in 51m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/653a628bc38544599a2cf036326e1a9b : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":83},{"id":"713fe801a0af331e0e676defd88a9a7d60ce0bba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-12 09:28:12.000000000","message":"Uploaded patch set 84.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":84},{"id":"725956ddeba83fbbb81f32fb00bae8a603ae5d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 10:46:36.000000000","message":"Patch Set 84: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6476bdcb4abb4b3ba6d602f463a431f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071b64056ead4dddbd188a97b6127c63 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499388fe019b458fa0c8e9d840a03d5c : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8de12dcf64024759a14cec14459e8cea : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6be686c28b14495bbf59d4ed97a01ce1 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b101e8bd4ca4d1599795ce69e070157 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fb0f8456a2c439788f830341f0334df : SUCCESS in 1m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/172a60f63c9b4f11af4b800d83178416 : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8bf3be8fee2b474cba7edacf4c385c7f : FAILURE in 2m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0703585cb16a47edba2aff0c7ba44264 : SUCCESS in 5m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/440366d296fd4134b98ffccf49df8ad3 : SUCCESS in 8m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3b65e113ef5a45b5b1012e22bbb37410 : SUCCESS in 21m 16s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/da42f5fd2b924c4fbb80e66c9baa2c33 : SUCCESS in 5m 21s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6675b59d286b49408bb1ba8edfa627cc : SUCCESS in 5m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b89976049c1a47c49ff8541fd04a9d22 : SUCCESS in 3m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fcb1839f8fef4b2dbc44faa8c7879257 : SUCCESS in 9m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7a7d33b880ff4517b1b702d98eef9236 : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ddd8a4fc7ac04fbdb1ab721fcd9c6025 : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b7f0fcbc84ed4ee1bfcc40b697da501b : SUCCESS in 6m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/24dec42b601944fd9055c5b3048deb36 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13d15647852e432ba8f86f0e6c0a22d7 : SUCCESS in 6m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/04709730ff0d42a7903106477103a197 : SUCCESS in 6m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/118103ad68204aa5b16adc700a823e33 : SUCCESS in 10m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/61b3cdde64154e4da5262bcd31ac534b : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4a068ffbdc4d46bf8886a15074658f5e : SUCCESS in 9m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1b84bb6b78c34fb1a044378e9c15e703 : SUCCESS in 4m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7fa3789866024de6a431f221016c3e1c : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/891cc60a2cc5471a9034d00d28a5936c : SUCCESS in 10m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5c67c0b4d3ba4efc855748fd0105f1a1 : SUCCESS in 19m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3221e50a3104c95bcd788bd3c4f5e7c : SUCCESS in 3m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b11487b8e3c44f6ac8b5e38dbf74be1 : SUCCESS in 12m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b80702ca034344489c66c5862ed620a0 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf1c21c8b7a24e3aa8f2f4f7a6eb69b9 : SUCCESS in 6m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6186ce5f3e248bba3a01fc5e7ece9e8 : SUCCESS in 26m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/166039ec7d1b46e282774eb056fe4de8 : FAILURE in 2m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b1ee0c728f584058ba2673c0ffedf0d0 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bb3fe7ccc8646bc8805ced355ea4af6 : SUCCESS in 4m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/07d64144c4514bc4a251d79b174fccdf : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b8ca483b43a44c380e913a38fb9e0fa : SUCCESS in 46m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4441ad06542f4f4ab43d8b37576fc895 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":84},{"id":"ebf7782350053469e86cd8a22173ca9d8c7418a5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-16 08:55:31.000000000","message":"Patch Set 84:\n\n(1 comment)","accounts_in_message":[],"_revision_number":84},{"id":"2f706258ebc2db01bbd1e01d6e40031c11c3839e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-16 09:46:20.000000000","message":"Patch Set 84:\n\n(1 comment)","accounts_in_message":[],"_revision_number":84},{"id":"4f31ae48bd0363146d37fdaa9fdf88eae6879fa6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 13:04:33.000000000","message":"Uploaded patch set 85.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":85},{"id":"6752a7279f15ce8b672891ca9bfada16ed38b2b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 14:06:51.000000000","message":"Uploaded patch set 86: New patch set was added with same tree, parent tree, and commit message as Patch Set 85.","accounts_in_message":[],"_revision_number":86},{"id":"87ef29ffe7557dd485db6779c888f82e2d8f742e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 15:48:16.000000000","message":"Patch Set 86: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adaa4e8f64104278b574a3d0559e4eb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecea95b551e246b3bb2732b4412396c0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7baefd658a41048c247407003b0cdb : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2deef09d02b543419d9fed39abf1e755 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5018b4fb430c4baa998a0e42408fbd43 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1a6b69049cf4552a47379bf7e22b38b : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de42e5051c864325b23b7054f9e5adf4 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3246816fa55641c39751680fe6b72d09 : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0b1a88d4bf44a4db75b35d94cdb1ecb : SUCCESS in 12m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbd3296e11f14e30ae4c2b67808ea401 : SUCCESS in 19m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c5d317b6bc5d4aaf931dacdd8975892d : SUCCESS in 18m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9d339b90b8d84958859f73e0174f939e : SUCCESS in 25m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/559033167a774ced90e2fa4897d09c73 : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9a0a217bed5042e2910f99b28b7592f9 : SUCCESS in 5m 32s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/73c8aaddbc4843a5b4aeaf55b99472c0 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bbeafbdfc76746a785ec476e8d6c8eb5 : SUCCESS in 12m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3a5cdbfcb4da417fab872680d01d9a44 : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6f81302838774fcc83125c508b9a9028 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e494fce108e940f59d924a70dd74ea35 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9ed99a6eebec42b3add893496be0a224 : SUCCESS in 14m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6dfb75565d54012bbe3bc0f8c4a334c : SUCCESS in 7m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/71ba6ac187964090856366222c50e6cf : SUCCESS in 8m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cc01f64adf4245a79f5577200016d550 : FAILURE in 10m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b73a856ce74b4fe8b3998d7b8e624ee1 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/34d55582c76e495c96ee665065183797 : FAILURE in 9m 33s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8bfc9adff7524956baeede04f9c218b8 : FAILURE in 5m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/af226d9f97834328b392d5692329ef2b : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7d397a4c916b476f92022a44c82b460c : SUCCESS in 13m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bac828ef3b9e4d12bb60173816abb9d1 : SUCCESS in 22m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c424eccb47494847a73878209ae1a7f9 : SUCCESS in 6m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/08d2b96187624463a844571e1a81fcf9 : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/539151ada21b443ba1904d93a7073737 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1f33dba507343cca79e5c2b47b9a971 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/262e417b17084d9fb4b731ae8993f75f : SUCCESS in 38m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ace65e3897e64e048dcbf3fe039838db : SUCCESS in 6m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/50a12abf944346e5a11eb4fac3cbe8be : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec05b35f52eb4a39881f4e31978d7141 : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d5b1c22888fb4180a12de0e53f3dc073 : SUCCESS in 19m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb9774fdaeec4f99984df8aed0b35c76 : SUCCESS in 59m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8389adde609a4935a03f61ad252b3398 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":86},{"id":"849c33894e46ecf31f22d4f870facc167d011281","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-26 20:17:51.000000000","message":"Uploaded patch set 87.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":87},{"id":"453034119b70154d91c8250cec7aff3eb5eb31d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 22:02:12.000000000","message":"Patch Set 87: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3739005d4c45e6bf87ec7c53c34aa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c38f36919241b3bdeaf429dd57cd21 : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c36c90b30634136998ace4c35eb3328 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9199c73c31894a00aee61f055012212b : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb7bfa3f2fcd4f489ddd6e2533cb3f63 : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47be75986cd2459f9734c8f76d02f9bc : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/589bd155d5764b04aa0ac11deba2f9c4 : SUCCESS in 1m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c83fda0d56a943f6a5ca1c4f5276a405 : SUCCESS in 1m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d1c5048dd9c44ecaf7cc255f2f5a2b8 : SUCCESS in 9m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2d1fd41c2342421ca62c353cbce74c6f : SUCCESS in 17m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aeee1fd2202c4edf961b1bc5e5632e3c : SUCCESS in 9m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2c7ca75b04c742858ebc1221b20396f7 : SUCCESS in 24m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7b7ff980d3ec4ce5b61e0f02055f26f8 : SUCCESS in 2m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/95cb538944624a499687988b5ae554ea : SUCCESS in 2m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6c6aa9a372fa4a6089dd9f8ee0ac233b : SUCCESS in 2m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/20331662aea54b7b9f3cfdea72e4a248 : SUCCESS in 9m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/05467aefd59f48e69a76704f27aa753c : SUCCESS in 4m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/af00e3ebe7aa4bdd9b223d33f6723f3b : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/850f23f8b6a1493e9313ce0be1277ec2 : SUCCESS in 6m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a2892fdf93aa47a39d82d216878bc7e4 : SUCCESS in 5m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eff4494d5ca74cf392ccf08dbc13af10 : SUCCESS in 5m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2368cba6a518438b868bc38fc6a6f729 : SUCCESS in 4m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/280eab76420a426f9027b045de1b07c7 : SUCCESS in 10m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c4fbb444365a48e18cc207849550bdf0 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/596ccf4aabd94d0c8202bd46a9ebf377 : SUCCESS in 7m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/55439a03ddc84c7ab56cb80e6a1799da : SUCCESS in 2m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5efbe652c5624b168dad42521390a808 : SUCCESS in 3m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8ec6d72974284e4fa0e5a4a72a158079 : SUCCESS in 9m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d22bbe5c1308401b9bd8064432038593 : SUCCESS in 15m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f302b5c8c2b34c1186ffc3652cc42198 : SUCCESS in 6m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/553e923298dc4e67b3476f641b6bafcd : SUCCESS in 10m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0003a7a1efec49828479f679588cfc0e : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f11b4f4d411d4c13815eb1d08d893026 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d76b3b6d36ef4132a5f9f54cf850c607 : SUCCESS in 21m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4cc90bd2a9554ef4a8954001296faf78 : SUCCESS in 6m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/46c9980f7e9947ef86e2ce83dc89da80 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37d3e9030201406a9563d61f261e6776 : SUCCESS in 4m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8a8c570ed4754266b325d45e6dcd7798 : SUCCESS in 18m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0476bf0f53a048308a163519c3a23542 : SUCCESS in 56m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf2fbdad5852483e88cc95f285185600 : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":87},{"id":"d0e35ccac34dbca427f6f2bec38195791d85fd26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-03 09:30:44.000000000","message":"Uploaded patch set 88.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":88},{"id":"0053a4c819bca90369a9b9ebd9e5ed0fccf1dcb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-03 10:09:52.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b343ef812874831a0b5622911582f84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526cc42405f7442781712fee03bd95ef : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd126149bc7475d8500ac37dbc7f4b6 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eae143495f1d40bb85d275bb14d87d6f : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bf136c148094ec5a56f12b16d425d2d : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1e68e212d6a4352bdbbdb60e9c17a71 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f69e4c7de5143ed8cf97aaf7125581c : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4ba63cc018249a2a0081d3056039710 : SUCCESS in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f62dad08d1ed4bdc8593e80559fd1ff5 : SUCCESS in 12m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d1aeb400ad0c45eda6c4eb60355e011d : SUCCESS in 15m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/016dfec5c50e41e3b1052238ba23638f : FAILURE in 15m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7acd993ce94f4cf5a321331de05b9b46 : FAILURE in 18m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1aad55ff6ca44f61be3185bf8db23df3 : FAILURE in 3m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9e01f9da41c84bd998f5a3f5dc939e5b : FAILURE in 3m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f1bd7006bf1f450ab440016c0f0f0815 : FAILURE in 3m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1c435c32c6644a5d8f61af8f20fc8256 : FAILURE in 4m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/afd6034b99bd4280bfa568e7642a2a4d : FAILURE in 2m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2cb42c02b0f54277a312e09967844e65 : FAILURE in 3m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/95bc1aab409f4f88b875bde2d8b138d2 : FAILURE in 3m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e6457ee625e84bec9dc6a41acb154169 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb387bceb5a5420fb1081d8edfa7a051 : SUCCESS in 6m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7bf94aee3d454c72a09953efa52daf6f : FAILURE in 3m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7a5716fb80134dd58d557d50f314623d : FAILURE in 3m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a3f615d958134868876dcf09f3e5a490 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/54201cd3f7f74106990750a486084d94 : FAILURE in 3m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/755d6602fcc04c9cb262a31cc1bfd172 : FAILURE in 3m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/71e7238da6814d3b958cb07983f79fd6 : FAILURE in 3m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a847c06c438c4897a315623aa5d08ecc : FAILURE in 3m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/06da359c6025430ba21f953d0e36546d : FAILURE in 3m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/468bf205e7b948078473e0ce4b976894 : FAILURE in 4m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5504183ae8ec4bbe9f134326a2d522fd : FAILURE in 3m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/779e7ee106a54d9385b8567a164fd987 : FAILURE in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eab395b2ab4b4cdfbd2831008778cf64 : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fb16af2d24b48c499cdeab2cf3764ce : FAILURE in 4m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fdeecdedab3f4dd0b05580d23ffd5c08 : FAILURE in 4m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/26f5f4fa7c6e42cbbc33a81f818e8450 : FAILURE in 2m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6ae32e6b97e44029a7b65ed219ad8dd : SUCCESS in 5m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2ac4a6b3713f4f4ba9b493a4cb7f2d45 : SUCCESS in 17m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/986c740c35954d2d9ec1661dab459829 : POST_FAILURE in 9m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/800e39f71074452fbcf567429866f087 : FAILURE in 10m 29s","accounts_in_message":[],"_revision_number":88},{"id":"301477b1dccd819c2406a2259c0a8ab81b79adc5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 08:26:58.000000000","message":"Patch Set 88:\n\n(1 comment)","accounts_in_message":[],"_revision_number":88},{"id":"1292d98b26f2b22b08e0d8b490dbd1cbf74a3449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:40:38.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4443051f2924846b5a92e6eed9aeb05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f773c4dbde774f7397e069a8e003cef8 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37aa5d6afa654e238f20f18ded089d3d : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/685ddce8d5d24c04897b78a521b11227 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b8954df5a674477a3dcb470097db88a : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f295e86a70c42519d48c8d895427d99 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3423988c6c44670a87ae3edc16209a7 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81b78e49a30d4d5c90e3a1eaaaeb4207 : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ea75284d4bf142f48c0e59a7f3a2720e : SUCCESS in 13m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0dd6ed5f912a4ed9b8fdaa78cca57912 : SUCCESS in 17m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/38b10b63949849f3be13e63d164ba70b : SUCCESS in 21m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a2c6b4fd40b34c03899896410321e12e : SUCCESS in 23m 32s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/93269e2710bb40cb954d0a40fd4ac6cf : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6ceb87a64ff943dfb243ca90a1369e06 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a99d293a7cf94eba880a8203e58e38d7 : SUCCESS in 4m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/015244846ed74304b20434dcaf2680c3 : SUCCESS in 10m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/70131233532c4d578cb482d4307c2563 : SUCCESS in 3m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8ff2e4d1f16748dbac0b997cee0721eb : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bf12c61817364aff84603e7febc19dda : SUCCESS in 6m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cc3bca4163a74d1eb52631e29606c017 : SUCCESS in 18m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8b84116cb7c4061aa2e768ca14f058f : SUCCESS in 3m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bd87550425cc4205baccce352bd7785e : SUCCESS in 7m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9c5bab4093144270a28a0bfe09633f20 : SUCCESS in 13m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f38c879613fe4a6085fb077fa7981daa : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8f751e68bc6d474589deaed4831f6ac1 : SUCCESS in 11m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fdc2d0de220f43669b1b4e9589f51366 : SUCCESS in 3m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/87bb8db88429406280007d2f4a666223 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e58a3414ed2a45a2940aa38bc05c3a80 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0f9e5e468ce44bbab598a787ecbc2dd9 : SUCCESS in 25m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f9920f214eb34b17a10bc09432b6c740 : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1efe460600ee43168c03c0aed60ddd08 : SUCCESS in 13m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a9a8d34392d24e85b03ddda5bd35f44a : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58bbf640c29243db942b8d56c9c934d3 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/417cf042ede44b18bfdee8259d3cf5e9 : SUCCESS in 27m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6ff2852c9aae4a418bdf32c8962c9ca4 : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/be997073f08b4125995ef1921366d969 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2771ec68e71e4e06916df24f0335eec1 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3d93af9c8059414da413c290108a4ee1 : SUCCESS in 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a7f271702aa4e35a427c737c6d35f2e : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88403fc28d5740e5a3ae353ad5eaf319 : SUCCESS in 2h 05m 51s","accounts_in_message":[],"_revision_number":88},{"id":"e430654f91449bf9dc3f6d76ada3ef93afce65d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 06:28:50.000000000","message":"Uploaded patch set 89.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":89},{"id":"73734e9ca53ab4c760fe7df4e917ad31a3120547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 07:32:30.000000000","message":"Patch Set 89: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af1f91ac607246988508f5b6ff5433c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6509387d3894406bb0c582e3c905088c : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f8c1561b904487a9c7668eec1011a2 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3475fda9af644e9abe0dd8bf1736efc : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e7e337c1d3e44e3ba9e848e68281bc4 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1c221b67a124fe2a0907f7e46ff271d : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2e77eb00f424b8387a3fee7e320da5a : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/923c1f1cf4c34ba6a7cb40631f0c3a28 : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b9a36cf2d0284ba7b337a475c512d237 : FAILURE in 5m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/136dcd173fff4546833b92736d83b01d : FAILURE in 9m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e5b9cb76e43f4b5694ba82bdb0500aab : FAILURE in 7m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a016e42f7bad42639242fd37390e28f7 : FAILURE in 16m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/07823526e71644c39755fc31cec2d911 : FAILURE in 23m 10s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/adf84fe04e4b42f496a288f462525209 : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c38be56e00874ea7aa85f23fb19c489a : SUCCESS in 4m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/08767692b7174c5fb401cd1a76e56c12 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4607912dbeac4e99b668562b272cf778 : SUCCESS in 11m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e3ec37be9fb140a7b099efd8fbfee886 : SUCCESS in 4m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/216b7b03616c425999b5b7a25900c917 : SUCCESS in 5m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/837e752c90514e91a13a17aa29cebbc2 : SUCCESS in 7m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ff1e4c150ba4d9e851fa1b356ca2f0f : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f940ed72b23b4129a69712a5b0b15926 : SUCCESS in 7m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9f9ce3a4640548eb9f4587ce404dcb86 : SUCCESS in 8m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/614e9a528a804d739f229ad3c3195cdd : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/95debfa9a8384a5e8e408c173e4c8129 : SUCCESS in 2m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/29554fc3614649269643486e01a42c7a : SUCCESS in 12m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7883edc0be1c44089b123a95b99a9d48 : SUCCESS in 5m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/178c44e60b2f42a2bdcd018c0bba1882 : SUCCESS in 6m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/70f457834c9a467c8a9c29d2a01874fb : SUCCESS in 11m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/38df55184db142249a7b2e02ef253077 : SUCCESS in 20m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36a02d1b452e496990ec042d21e6538c : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ed35a3f298134a85ae0cda114abfd79b : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0bfcebf910324ba88c306f1d39332ee2 : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91547463f43d4801b554ed03f87ba80e : SUCCESS in 1m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd39418b071e40d2a8d970763f57138a : SUCCESS in 21m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3351913ac7bf4c65965063c2c92a652a : SUCCESS in 5m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cc425a6c181c4f4dbfffa9a00c79c72f : SUCCESS in 3m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58724000f33a4b87bda8a991c68b9dd0 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c5fbe1c5fab44e929d4b145ab8651565 : SUCCESS in 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1095776c1dc4d96b609d768c854cb3f : SUCCESS in 49m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ba1e904439e4d948f5a3fa831b29913 : FAILURE in 1h 02m 20s","accounts_in_message":[],"_revision_number":89},{"id":"65ca37433576cc205da3f2417d9fa2b87a24d8ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 11:55:59.000000000","message":"Uploaded patch set 90: Patch Set 89 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":90},{"id":"d1908590dc095b100758e80992c594ab500d7c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 13:37:07.000000000","message":"Patch Set 90: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad012d8bfdae4ed6a3476b1424983cec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3697fb1cb646d691eecc3ec0889762 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cce5bb45d9f450c8ada4c1f2ddf52b8 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad49395f6b344d70b4bf50bac4ab1b3e : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc92e1e7c9de44a2b5290d8050823b5b : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7dc6d8ed8854f788e335d5e6dde76f0 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95ad22f6876c4c3a92bc4134221e2877 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f533b6b5338046ce99dd63dd072bbdad : SUCCESS in 3m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b15d88eeb63943448077a5675d8d64a0 : FAILURE in 4m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/621a32081e63459788b590843713a958 : SUCCESS in 10m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f1c661120beb4bf2bda2fd8939424ba7 : SUCCESS in 18m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c570c117aadf4ecfb7646c16c50d0e9f : SUCCESS in 19m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8fa2f5883b3a4dd793308984c79e57fc : SUCCESS in 23m 34s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/341780ee253c4d7586372cbfefba9bdc : SUCCESS in 5m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/10b6da020158433a9bdef56ba304033e : SUCCESS in 5m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1f5203afb4984aadb61a56b732af7e83 : SUCCESS in 4m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4a2abf2a4f014a389a75cbabf5a9e212 : SUCCESS in 11m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0dea932a59394d63af6f93e5038bfbfa : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f9955995e35044a8bd737df551f5058b : SUCCESS in 5m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2b9db0cee94c4800bc73487a746c7a72 : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5144e2ff6fe44e75b310640f9808ee2e : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af6f30815930468584e606c77fe16286 : SUCCESS in 6m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d39f6457f14a494da2bffe0a4de38743 : SUCCESS in 8m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6692747fac741ff842126744538a228 : SUCCESS in 13m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/809c210a1abe4018be173a388e0ce624 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3691e43216c84d5e8fe62f460225536b : SUCCESS in 8m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/336d625ab8fa47aa828df6dff6add950 : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/808f940d63b8442db0f0dd99a06fda80 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b03a2cf1da864f908937a3230fb4d83d : SUCCESS in 9m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6ffb4b1d771e4cb99399615637ecfb5a : SUCCESS in 20m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/39a914ee0fed4679ba35045a4b475a4b : SUCCESS in 7m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/903ec26b483a4853a5876402a49371b0 : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f9ecf30742be49829ff1c6a8bcb5bd89 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55b04239d32a4806b0f552d33483859e : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39838ac84cc3418e9d22ae8317cca679 : SUCCESS in 23m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0187eb1b93224329988258103e99b1ce : SUCCESS in 7m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/35820222d6d4418e921edc6256cc667c : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/459112fdc216445ca8a64b434a4af732 : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0e693adf8bf240038dfc75e2265bae63 : SUCCESS in 7m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e5d008d1e7d48c7b96e41305e00d55d : SUCCESS in 1h 09m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c012f34e81674432badf06069d028141 : SUCCESS in 1h 38m 15s","accounts_in_message":[],"_revision_number":90},{"id":"ebb9e1b06236b1d4d109e6553d3cb695353ad55f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-20 02:11:57.000000000","message":"Uploaded patch set 91.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":91},{"id":"fc9e2e79c11692c48b0125846228a64625db847a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:07:05.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fc894f858d400796aa26a98c3ff340\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a12d48126f04c62a4fa14d6f02f98ed : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94709e021fb944f0a10853467806b1e0 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a5b8c5f07c143f7ba5566b30ff71fdc : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9490052b0ab44bf19fdb0f69ee4569b2 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b5acffccc2947479cb37bc4f96ac6b2 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db1daebedc634a738d28907974ba808b : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99bc838e20134d62a1afbdb768e684b4 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7267d1f96646421ab22574b5960f4188 : SUCCESS in 16m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4799b75a91ec453a818081dea57c423e : SUCCESS in 10m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c98dee06a0d34e959de5b79daaceb56a : SUCCESS in 16m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e4292c47388b414cbe0acb6bbd2e5be8 : SUCCESS in 18m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/994b7c6107ee47dbaf60a099798cb8c3 : SUCCESS in 10m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b6e62383a1b94136bad763c0cf98ca9f : SUCCESS in 6m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8280e28ff78a4abc97e2e7e323244f9a : SUCCESS in 7m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/30a0d2ee2b384cb0bf71efddaf7f8756 : SUCCESS in 6m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/496f8397ec254a90b61f3f15fa438566 : SUCCESS in 10m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/55a1b049b0ea4fdca8ca4bdf7c2509e0 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/59d14b0eafa24e8f8369bc109e413a8b : SUCCESS in 5m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e8c3bfb98d654c2f9965b65a0da7e65a : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fa83fb588aa24132878e4447442105b5 : SUCCESS in 14m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc17ec8df0484063b0f50584c5ca15f9 : SUCCESS in 6m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/73df2b2aa17d43d3bc0229e7816c551e : SUCCESS in 8m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8a73c29de3d3417aa9e3b2fac78b6796 : SUCCESS in 12m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ae12b63344804575a71f041ba7d132bb : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/955d96f0db404376a4a5ad81c8318a88 : SUCCESS in 10m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/af35f9c73800469ba946344134676af1 : SUCCESS in 5m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e38e0acf20224a829b5c332378536dbf : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/165c837c50af4c0ab25861346825975f : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d004ecef571e45d7a560d152dda396d4 : SUCCESS in 19m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2d371622d5334332bb149c6eff24ee59 : SUCCESS in 9m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/98bea019e9d64aca8441edc32e946b81 : SUCCESS in 15m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6596f2003fe44f16bd25d9498b66ddcf : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/145f0023218e4e0983003e17fd7ae7c6 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/757916eac31d4788843e8149e4628bc3 : SUCCESS in 25m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/06d8964e66ce4b28871ef0e9d88081b8 : SUCCESS in 8m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/91ef53abfe5444fd81f2b5bb8458a850 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b3b2a88216a4e66abde838fb8a4cc17 : SUCCESS in 5m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5beeac8ccff14874945beea25c00ce70 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b97ce8deeef44573be64fdb473f1fbd3 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46c731355d734e918ab44ea0be82afdb : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":91},{"id":"2a3679750b792f154b110fffbe616e1007eb2791","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-20 09:26:35.000000000","message":"Patch Set 92: Patch Set 91 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":92},{"id":"cb03c6861262f66cd585aced50f8d1b77ca81da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:50:01.000000000","message":"Patch Set 92: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3930f1b1d8e040f3aa52f3fbe7ccae05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4be207fbf514331a38dcea0b90e3328 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6f2b534e3e42ee8abe0df03b960adc : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da80e3a629ee4f70bd9dd08eddc97c0a : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f4ca186dff9441f8086b14fed82a858 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac12d2c884cb4bc98710ca2189ae39ab : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/538767ba06424acab7a6d41acccb34ca : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3805acace63b4743ae84eeca232fac2d : SUCCESS in 3m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/42fde5aabca84b89a14a881bc4e03a69 : FAILURE in 15m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/60d13fc9be0944c5bb782443e6208293 : FAILURE in 11m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/847d2d4cc9be4be2bf9e2c7b9aa065bd : FAILURE in 17m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/230205f806d2458082fec2dcc023bbcc : FAILURE in 15m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/523c0aee2c6a4a63b45a80e0bfadf98c : FAILURE in 20m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5849bc16514245f49e8bc506aaab99cd : SUCCESS in 3m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5cd2373778604c2b9f17213935769998 : SUCCESS in 4m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a2c7ef6653b045d8a8ba88fe2e0aa75e : SUCCESS in 4m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2daff3474be24b88bcd500172b30438b : SUCCESS in 11m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2f94dd08d8774ede98ab33d11e67396b : SUCCESS in 5m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5a97a8855fd84b1c81d05be2a1e7858f : SUCCESS in 6m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6ab64a08a65b4b3ca41636bea6d6345c : SUCCESS in 6m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/decb53241c2c4e64a89796fac8ba9278 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/702274e2c6b54d66a1da73c2cfef38c2 : SUCCESS in 7m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c19dc9b633914e3889c70dd6571837ef : SUCCESS in 6m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e7f53dd7b5af462d99361253f5a84ba6 : SUCCESS in 12m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e5d23602029f495b959cb6c8cb0a56d9 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/818e9e277eea4bf7bc3b69cea3853b50 : SUCCESS in 12m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/249694f79d484bde825514b9e535907b : SUCCESS in 4m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5968543d909b49d8b80d451abff4a52d : SUCCESS in 5m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/905b24dafb4a45f3b55493d802aa00ee : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e1856b6c0fd4818872939a5713efbc6 : SUCCESS in 17m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5784c75ebb3740c4a8be9e69c3284233 : SUCCESS in 8m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e0360ccbd9644f298605e6ac9aee09dd : SUCCESS in 14m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/672625e519754dea971b6e8d2370e7b6 : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/478cc9e923054407beed5a350c2755bd : SUCCESS in 2m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6b0e37a57dc4c1f92cfa2cbe5244581 : SUCCESS in 23m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f42000b1831f4ae4a91ecb85f26844a9 : SUCCESS in 3m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/10cee973728540f6b3e80c946e86cbe0 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59b413ec48b94bba9dd14cb5d609b5ea : SUCCESS in 3m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d36cb1867d3643c88c215fbf5cf204b6 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be90327ed94f4282ab80f51155cb6baf : SUCCESS in 55m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/329d747325314f8abd4aa5fd71244082 : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":92},{"id":"7388de222fc10cf64279ffdef603bd23842bec58","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-22 07:57:39.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"db0ee090879fd024d4cc40f47b3661e75c7faee0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:29:45.000000000","message":"Uploaded patch set 93.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":93},{"id":"cc65b806604b27e344e570e76ae418696b82c61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 15:11:02.000000000","message":"Patch Set 93: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c585639355304492bed448b0da1b14c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd943c8bffc4834883c4ee23538ec88 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aec1c1587504bdcabdae13acc36a3fb : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7930cdfc452d412ba91cb322b2607e40 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/575ebddcd13d405aa31904b23ef66d32 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef6d560e64ee423d8c6812c641f19f95 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/639d031ae0744a28b2089c967d4c2551 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/449da0b3bbe54d00b461a6cd190d3225 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/310a554e944644ada03344c55f0157cc : FAILURE in 3m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7f7b48e3c1e846eb8b0d82e537ea3bdc : FAILURE in 7m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/56487ad487ea4259a5c5975f115b59cb : FAILURE in 12m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/545ad2b7224c441fa5832f2871a6201e : FAILURE in 13m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/581a3078b8044008856888dc9492fb9e : FAILURE in 17m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b6511df6b901402eafd57a3e7558e316 : SUCCESS in 4m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9d229c1306314202a922c0dfab3915cf : SUCCESS in 5m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9c2ef05b6794489c921bc0b4dac27711 : SUCCESS in 3m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c02352fc4d9a4e42bd8472a9d7fd0fce : SUCCESS in 13m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f1773a00f4f845bca9fc55ea582b81a4 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/09d973979a7e4acd9ef4055d18a041ee : SUCCESS in 6m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b994f45837e1488f9a4c1e034df6d1fb : SUCCESS in 6m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3c4011b280ac48b18bb8f1f1e59c9d9e : SUCCESS in 14m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/513c09f943654e088b2adc519ba8b697 : SUCCESS in 5m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f9b8979fd0c94aafbb65f87a0d7275f3 : FAILURE in 5m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/275d4af5bc5b48018e8e370d4f555112 : SUCCESS in 12m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/396513ac81db46a8ae187b7ec6f0d60c : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/08074334cc3b48e99f14bfbcd90511af : SUCCESS in 8m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/890db79f828a48e59eebe53f640d2784 : SUCCESS in 4m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9ccb281cba4e426488b641780d7014a0 : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3d7e4935022641049b246def88b436c4 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5592002437684aa5adebae8d3a98af26 : SUCCESS in 18m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8437eb3dc2984dacbedd3eaa01db07f0 : SUCCESS in 7m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/21bfa27588794d5787870184baa5c3f2 : SUCCESS in 12m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cf3789510e1a4bb4b95187faa136befb : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95718dd092974935a2f160def89ee2ff : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f7dbb3297d5458eb5ddeb5d7644b492 : SUCCESS in 25m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9dd8c205e0b94e12822297eab9c8df01 : SUCCESS in 3m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c3a77ad267b3408496878e174f73b62e : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78403a4328014f4199b93f315d589068 : SUCCESS in 5m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/07b77b7c1e6444e098b42fea57b50f69 : SUCCESS in 17m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/465eae3c01a8462e8c998d7652b3edad : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a700dc04c150495482f61b983a06cbbb : SUCCESS in 59m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8a8fe3030f84ecfa00f2848fd027040 : SUCCESS in 1h 33m 28s","accounts_in_message":[],"_revision_number":93},{"id":"b6c1a7a77ec6bedbb8930abeff4aba1f0c388544","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 18:30:47.000000000","message":"Uploaded patch set 94.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":94},{"id":"ca7fb3af4fa8178def041164080befdbce4a40a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 20:18:00.000000000","message":"Patch Set 94: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5cf23efa1d44bfebafc5feb77a5c03f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33c8f9f5e7c49868bf88b63eea66856 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b484592fc7b64f278d3afd716808c442 : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e04a3da28cb47fc9c1e0a76187c0b7e : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/280a5a0e073b44949ebead76fef4d34c : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bb21bd434974e15b4d1564742e7ddfa : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47aa71bba5c146fc8d986779b5aed702 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee6663265fde4e589d1626489b679fbd : SUCCESS in 1m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f4110eff98ac43b09a55ea43023ed8d7 : FAILURE in 6m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/677a28e0d485481493328caea152ed32 : FAILURE in 15m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/de0301a2009f4a468914df38452dcf68 : FAILURE in 18m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4e24a6d731b84c0094698e51c020d393 : FAILURE in 21m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ebd08a11a7a546c585e086b8f1ebe745 : FAILURE in 26m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f0e3754832cb4091a0a1671ac22abb1f : SUCCESS in 4m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7020b0510bc044dc8e91c136ae894c62 : SUCCESS in 4m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e9403779b88043318a91107e67b6c27c : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e9451855893491d9a19cb9d4f4bc47d : SUCCESS in 10m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a7dd9454b4b4402da9574324421deb29 : SUCCESS in 4m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4b121cfcbd6347f1929b487a2f06ad8c : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b568195cc8564f47a53726180c1cd320 : SUCCESS in 6m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e72b1232a0534079831d7de5fca4aac0 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b8ca7bff7654ff1b59efc6be96f5832 : SUCCESS in 5m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7af6045e7eeb466fb89ec77b0a134d03 : SUCCESS in 7m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c272242155c54bada1bd6cdc1f2c8dd4 : SUCCESS in 11m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b7750eb33eb24f0dab0fb0b718d976ed : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f019b8422f4649fba41e87f76c8f5019 : SUCCESS in 9m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b49b2b51a3dc43979fb704b00daedaeb : SUCCESS in 4m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0f8ec6b31a994667ac42ba621e9eeb21 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/897cfbd2e892449bb5af2796b8a53e5b : SUCCESS in 8m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/51fdb7c57f9343c08e43b3bf2912a279 : SUCCESS in 14m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01465c0feb97494d999ac5ab05757b00 : SUCCESS in 6m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/30daf754aa504136a3d5bfa65fd901c2 : SUCCESS in 10m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3aa5c4501a444e71967531b6e3f1a137 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e4a92744f7042d7a2fee989de001366 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e380bcd31e0645c4bfe46bb3aeab486b : SUCCESS in 23m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3a377de6170b4e3e8433ba05fdcb0b91 : SUCCESS in 5m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/54b345967aa14d27b96610d987e3fa35 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18397d4fdf684d3cba5cd19ef539a4e4 : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ea71c18daf944c5f95deac0c04d75abb : SUCCESS in 18m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1ffa77d7f2bc49499926d39683b56464 : SUCCESS in 4m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/926bb4ad2f664dec84fb2a49e1555884 : SUCCESS in 31m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/099ad6944e294459b3b4091679c455c7 : SUCCESS in 1h 37m 03s","accounts_in_message":[],"_revision_number":94},{"id":"8cd4fc70642677e4d2951fb23adea31c8f328512","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-25 09:40:32.000000000","message":"Uploaded patch set 95.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":95},{"id":"86f3bd572975d4b3fb8fdd7f32e66ec074e9f81e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-25 11:21:35.000000000","message":"Patch Set 95: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b39278cdec422eb800eb8b05d8d992\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d196936e9c444569d26eca79e96a7d4 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a8ccac08f546fba0fab8b5caa3520f : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcd59287cfe64b97a134338ba91d7a59 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/019ec204d2cc482d86a27979474da53b : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3aa79185eeb46c0806691c43a6b2849 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b27fafd1151475b9796152346fc7bcd : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a580b59caa554dc38936e86b947753f3 : SUCCESS in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b7f849bfa7cc4dafab8fb1032ec477bc : SUCCESS in 11m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6ed084f27cf44be48081954a1cfa5d7d : SUCCESS in 7m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/709dc9c8f53e45cc8d1fb17d7744be2c : SUCCESS in 11m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b08ed54d1f7d41869849136da6d38bd0 : SUCCESS in 13m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7e879cf7c2444eb9bc48a11e43297263 : SUCCESS in 15m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/377482c3e4724209ade9be376dd8976b : SUCCESS in 4m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3087f89a59a348efb8a6751c54a806b3 : SUCCESS in 4m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c8541c8072f541a9a5b2c1b36708d3da : SUCCESS in 3m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/84a432f9e66c496481f9a69d425a9f65 : SUCCESS in 9m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f76ec08a2ef049fab23c839d44c25155 : SUCCESS in 4m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5f71a03497c64c43b0d0aed5279220d3 : SUCCESS in 4m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9c0696c5b14b4eb1b5daa2dbb13ac239 : SUCCESS in 4m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6b55e9ed3d4642c2a4722d3a4993bf5f : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4efb4c4e13704c8993c594d32b24ae79 : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0de60f80f7dd44a69400c7ee1a6527f8 : SUCCESS in 8m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/09338204e0b04a16a17510683c0efe32 : SUCCESS in 7m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5dc0fe404aa644a39d1df03546ef1e5b : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bb1071a6ca744668b6f798897f2f317d : SUCCESS in 9m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d5f4d91f254443359ace0d30aa67492a : SUCCESS in 3m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/95977354d5454a9db5b6b224d1f8f459 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1e842ec384274bbbaa40ff2670db943d : SUCCESS in 10m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/457db2cc499240f6bf83a8ec65509042 : SUCCESS in 15m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a42b7bcffc5f416a8e9d39ad9b0cc73c : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ddbe8b60394a47c9889618ce4d0a0648 : SUCCESS in 13m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f854bbfbd9b447d696bc795c51516f59 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4a86e02380914c60b719f2702355e6ee : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a2c97ee193745bdb9decd0cf0f7fe85 : SUCCESS in 20m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0b67cb343e2b458cb68af6b1cd070171 : SUCCESS in 6m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0b751bdf7269424bb3b4b99fe619e9b6 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/363a54ada430411da97fdcbb85a004a1 : SUCCESS in 5m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d5dea12cc5794d6499877d6f100b3fe6 : SUCCESS in 11m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f4e9cc36950c4287ac9609575ca64507 : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7d04baf0510450e9821a95330d31fd3 : SUCCESS in 1h 02m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef9634af97644ee0b9bcac8f1de350c4 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":95},{"id":"03bf46dbf7e5b0e48c61c162e0fd4bde5dda8aa3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-31 13:40:58.000000000","message":"Uploaded patch set 96.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":96},{"id":"18b37a6f5740f3d6bdc3265356d65869cb8d6402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 14:52:44.000000000","message":"Patch Set 96: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea346b7904aa4ab0a87db283393d4609\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee32567e028c4a009c80aeb55bc1ddb7 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e32b66422b4d53ad69976a57056dcd : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b089bdfa9afc451ba5dbf43e05519c7d : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bba44de827642a084547f6b4dfe5b0b : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8475fb31aaf34bbf83d72fbddde4145c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fed533ad7514ad1a58759df47a4aeaa : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6551de9e18d94682b6ddbf33e3b442c9 : SUCCESS in 3m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b17d73c3683418190ce6113f5272b87 : FAILURE in 4m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c3555d05ae694f31bf480802e727b155 : SUCCESS in 6m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/79b2ae7ce54e451587d162615419695d : SUCCESS in 12m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1ad23300c3d646718799ab25d03f382a : SUCCESS in 14m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3e9f8618cab94fd0bf603630cc7006ee : SUCCESS in 17m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/019e42ef19cf43a89a92b23c62225527 : SUCCESS in 3m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/92cd59052eef4095af99a072208728d7 : SUCCESS in 4m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cdc5b121418b4430a19caf853d901c3a : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c8a1a38f297f4fd28c58e5d05fb3f365 : SUCCESS in 9m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f4df715f93e8499aa4d37b8cb2a02a00 : SUCCESS in 4m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d9af2abb7741488fbc6ad744f55018bd : FAILURE in 4m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f9cd4043af6c4e59abde71a52939be36 : SUCCESS in 5m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d64441e5f0da48d7891dac62877cd9ab : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c63e628035884fd7b7fdc83b7696ac3b : SUCCESS in 5m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1ebacd0044e543529e54173f78484476 : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a781b9704f2442709df630d996e92758 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e9bee27879934805a017f8efcc04c56e : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ea4c69d5f44e4a9983e64b1441ee7ec7 : SUCCESS in 8m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a788f097ffd6453aba36ab7857a9853d : SUCCESS in 4m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f0fd8bc5958c4c12805b0353a7ed2be4 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/641532d502d441ca888ca0c3d1bd29ca : SUCCESS in 8m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e992e803f28844f69fdbde2189e02ebd : SUCCESS in 21m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c26d5875d1db46dba7628c1405df3cb4 : SUCCESS in 7m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ecde7222f6594332bb54d0ca28b805b2 : SUCCESS in 11m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/060ec08a9a2643afbbbcb49ededf6022 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3bddba7131247319664bc62a787e758 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6614a73a94aa477ebe71379db15dab58 : SUCCESS in 19m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4c0b382a19174569ba00bf9b27d6430b : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a8bf3fbfe4a941968dda4968dd79c30d : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/990872abe418420489dcd6a11cdf388b : SUCCESS in 5m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7fc556b912634b2280bd28d4485a9936 : SUCCESS in 12m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5f409a263abf4fefb50d8608daa511f6 : SUCCESS in 5m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fd6c3639b9744b3b4fea3a378fe57db : SUCCESS in 46m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fa5b5021f9a4dd7a23dd1d5f7f68f51 : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":96},{"id":"2e37d77af2c9361ac38cd064ce5c9c802ba9dedf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-01 05:42:41.000000000","message":"Uploaded patch set 97.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":97},{"id":"8184cb41460e8e67c9e4853c9c863d3afcd3bb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 07:25:50.000000000","message":"Patch Set 97: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76446078b1d94243b8022f179cdc6f87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4125d5eaef5345a29f747437cfd00ce2 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5003e2f98c1c41a2aaa84a6a223fce87 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b12037fb1e454f398502a7e6a4250314 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb0e61b14b0443648b5a3f6e429c7245 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bb146d32835434b889e5406cd347810 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a70d4a7875a848868f779642bc1004e4 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/616b88c9d0494b24b9dca2cb9f2ad349 : SUCCESS in 3m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7fadad4507934ee1b383d53d324b351f : SUCCESS in 13m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9eb148d82584464bbce886d592cf9bc : SUCCESS in 9m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/87ec5afad4f74a8ab4ca121e86597b27 : SUCCESS in 12m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/25c70433f27d46ccab66134ff0551ee0 : SUCCESS in 20m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5508f31a62b44d6988c3fed787fff263 : SUCCESS in 15m 31s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4b0d57784169408dac165401c2cc846c : SUCCESS in 4m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dff0e5379462409f9c19860c6c88dd9d : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b0189c3740ec4437aab8af340199e40b : SUCCESS in 3m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/060ebba338664bdfbc1e68c045aec76c : SUCCESS in 9m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/659b26b4fee74e12b9caaa4f88c8c85f : SUCCESS in 4m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d93686cf7eac4872872f5343f86b6e65 : SUCCESS in 4m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cd332fa2498e4e4ebe008994184109a8 : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a307fc93089e421baf435322e4b04697 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84e09a3459754a22ac1f77c41ea9e5d1 : SUCCESS in 5m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/79551c7a64e34745aa038d6faa53baca : SUCCESS in 5m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8759e05d654a4271bc0ccaf047e04f23 : SUCCESS in 9m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/10993b7a05df48b9bcee409fe0433793 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f59321fadc78407aba9a6563817f7640 : SUCCESS in 9m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2124f595886441a4b137c2c636db5060 : SUCCESS in 4m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a9e1718f3c5348018d31a728414f54c6 : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/10ce8d8c2717436a8c32f3f858eff049 : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/73f53245747344ccaf4801b8ffceeecc : SUCCESS in 15m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4ed90f12708d43a491cb0e68893ff51e : SUCCESS in 7m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fe91462f39a940de946b7fbba9b3f2bc : SUCCESS in 14m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/779ed9fb594f4f2892984d7c457ecfa2 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7914e0da09524963be1865be88662596 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9c53bebb025473d89fd3e753b11b0b1 : SUCCESS in 25m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/852bc6c6d50a477e9d7b588620d61bd9 : SUCCESS in 5m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ed795b5e533e4013bfbb36257306cf6b : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4ebae0a26b74f94927b3492a5ce5465 : SUCCESS in 5m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3631e73f4869416c8e31f51842b344f0 : SUCCESS in 17m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bda21959837d424b83b196f06fb44d66 : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb60179eb9246628b69c103f81fee02 : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11637747b8b742c9a360d0dff1dbe4c3 : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":97},{"id":"65a1820654d28c9409c806c32a0f3e9ab6bb1bbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-07 15:08:33.000000000","message":"Uploaded patch set 98.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":98},{"id":"35aa16752252c7b9ce89ba87e31788e75aff67f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 16:37:39.000000000","message":"Patch Set 98: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74283047ecc548408abb424119933417\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9831c78e2546eea23db3e518f33908 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d467569bea94053b0fe71a1ee8acac7 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e3b0d7e1dc349e8bb81572e44bec2dc : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be28bb5b4b834696a8e34ecdef5795e2 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72d32b5d46184ff8818c038f6654903a : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/feda1ffaf7b64c16a6121571cf09b076 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f6e57c0155ed4d82ba1ca9ada74d1caf : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a02e8bdf1eb4ab085c419f008736710 : FAILURE in 5m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/df17395fa17c4ce29fc753d111f7ca82 : SUCCESS in 8m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2bd988af0fc74085ab169f0e0585fbc4 : SUCCESS in 12m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7e133178c079401eadcfe5441ea0eed6 : SUCCESS in 8m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1ad1e4a10054c469c56765a5d21ea9e : SUCCESS in 21m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e782e45004014b12b3177c4d14ef14c2 : SUCCESS in 4m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bc6df2c9f588481ba2261f78b6f832dc : SUCCESS in 2m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7c849de4a2e445a69831fc68534d70b3 : SUCCESS in 3m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ce80fa94f8264b089373eeba04396014 : SUCCESS in 6m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dc9fb1ffa5214ba1a1b242221d8e72a6 : SUCCESS in 4m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ccb407ab38b74b9d8d34e0057267d727 : SUCCESS in 2m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1011852defb247cbbb7ddbadcfe1607a : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fb6008c0bb8c469692b6c8a38a1cceef : SUCCESS in 14m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebc42d5394944e2a8da9b7934bf6a337 : SUCCESS in 3m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1e993951d7fb4f06908b566e11030592 : SUCCESS in 6m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b8414326d0274497ad963214c7b1f469 : SUCCESS in 9m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/09b8400f28f84218861ba8ba0881b86c : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f36c038b8da24c5e82ff8c0e8426716e : SUCCESS in 7m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8fb7dc42f7de442f957e67bdcd671f6e : SUCCESS in 4m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/abad3fd9ef074707971cfa3381211a04 : SUCCESS in 3m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6e13fb2749f04d8a9e5e54543f34b23e : SUCCESS in 11m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/be4df63b6bf24b069622d53873f777bd : SUCCESS in 17m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c190c4548b7e4d559507b4c89d2b5982 : SUCCESS in 5m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7284a9b7fedc447cbeb2f39a6c9e55b0 : SUCCESS in 10m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2b50320e6d5e411f989fa0b1fdbd0977 : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60286434387646d89454980210eceaad : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6f573d977b74041b64e4f540b664c52 : SUCCESS in 22m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/28871c1fc47b41fcb63295ba85b50463 : SUCCESS in 5m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/182e942b9fab498090eda0768dda14a7 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f8b2f7cdb3d343a78822583f7923bd51 : SUCCESS in 5m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd8fc6101a154543844f1c12651c968f : SUCCESS in 12m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f30f639cdb1241dd956ad9e6f6ca7417 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0041fcbb1c8402bbe8039f6ff337c2d : SUCCESS in 46m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb1621e62b86436ba6286fe9000abd1e : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":98},{"id":"14e16dabbeef43bf1cfea2970722deb29c0e598b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-08 11:40:54.000000000","message":"Uploaded patch set 99.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":99},{"id":"a9cdad1b1336affd006ef99fc258ec9ba32ccf9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 13:35:03.000000000","message":"Patch Set 99: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe653ac9883a46fa913830a1a229dd86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fb96a5dc294374ba79cf932cbc5e1a : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891e0f7f749d48af8a39e78dea5eeee7 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69eba89304f541d4b2b5c5644d80970a : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c2bad81f58d434ab4cb62853792d241 : SUCCESS in 1m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4aec80b720124baa92c75a5b468e43a0 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26290112e9324aa788596b4e62ddcc9e : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef0389c38ae14c66a4e2e0817e93d8f1 : SUCCESS in 3m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e6b9dd2184e4be0baa76b797dbba320 : SUCCESS in 12m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c60c0551715442d4987f64fc41b83b27 : SUCCESS in 7m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8d9fb8a0d0614be5aadee7d930a48fb7 : SUCCESS in 11m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb4ba2866f5f4893ae708a5bfefa1763 : SUCCESS in 13m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/11d8f0ad904c4c8793d0816f4cf19443 : SUCCESS in 16m 15s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f3377891913e4e16a254b063796e4b12 : SUCCESS in 4m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4e34b2c1524b4d47a51c32d8570c943f : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/332bff70ab4f4923899999524fbfd1ea : SUCCESS in 3m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/21b2b9c21d48462d9842205ffb2074c5 : SUCCESS in 9m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3d8ef0f0a26f4ee18e653b1a83b59271 : SUCCESS in 4m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a092a61f5cc144f09c7ae77bb7489604 : SUCCESS in 4m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2e9bf24b33184fd18e4ca54a78421a1d : SUCCESS in 5m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f3cf66e5294843a89b534bd7bf1a801d : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d0b10957e654601b62ab5b25b703925 : SUCCESS in 4m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/85558831728f4327b2954f8b6e627373 : SUCCESS in 7m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d50abb85eee84dbd8141a7a7f4a5d0a2 : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3510a13c2bf94036b37374219c481411 : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/133b825314f84959a82939915d6fc428 : SUCCESS in 9m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6c9a536a229c4dd2a27a74b8c6c225b0 : SUCCESS in 2m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/738c57eb872940cc9a1234b3eed410e3 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7c18237e91af40909a2acf0caa06fb55 : SUCCESS in 8m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a64eacb581514e90995574cf01e4e3ac : SUCCESS in 13m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bdc61328766d48f7976b0b2ae256167c : SUCCESS in 5m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/019e4f70be8042c5af20944edb3e8aa5 : SUCCESS in 13m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/65b491cabf5b422fb5ab5d11b769398d : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7478b6e43f3348509238c305d8746a39 : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c40ae939f8a5404d9b98d297df43b09e : SUCCESS in 22m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/85df96e53d974b97bfd5af87a4ba9125 : SUCCESS in 6m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/946122f3095946f6b3a4902e95c07c97 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/865f39590d114aa7adbc05e5c65b5cc7 : SUCCESS in 2m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1c9a9a77bfa94db69a4a165a9e981a66 : SUCCESS in 16m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1d3dc4df776c4805a224c62c1bea75e3 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbfd39a90eef44199fae2eebb7a6a552 : SUCCESS in 36m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76ef42231f774989972b2e2620c64190 : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":99},{"id":"29677bce4292f81e5dd785997fa21631152a79ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-14 09:14:09.000000000","message":"Uploaded patch set 100.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":100},{"id":"ac3e70a65868860c163d6277e0e5e362be35cf1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 10:22:08.000000000","message":"Patch Set 100: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8166d32820014180b68288cbaa2bd475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0a1e4918e8486d997a960698d0690e : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf6e862ee0c4d75b2d4d890a5bf92d5 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/284d9fd031764500a14ac9b63fff873f : SUCCESS in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2407bd7bc04f44409ac0153ff3d61b0d : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3af0d97a57042bba66d5ac522c3b356 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c3a9524463540fa9f9c4571bd5b859c : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37f16af14f3143aa9bd9d67c838dbc11 : SUCCESS in 2m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2174079d918d45f8ae0b95469c60c2ac : FAILURE in 2m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c3758e62ff542ea8d5f0ba3787ff1e7 : SUCCESS in 4m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f4ad6e6b72374934862f5582428b0eb1 : SUCCESS in 19m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8f6f89dc621841a6a195bdacf781e061 : SUCCESS in 17m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/749a7c91eb23475a90755337d752be39 : SUCCESS in 26m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/596bb48c5dbe430399bb62b31ebd1cf6 : SUCCESS in 3m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fbe9d01b8ccf4b78aa7e83be8892eefe : SUCCESS in 4m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/35655ae562a045249eaa3149088a939b : SUCCESS in 3m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c9e0b855014d471fa34fe819edfab47c : SUCCESS in 11m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4874677c14ad476e8911d916918adb20 : SUCCESS in 6m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8dc27b74c87545669c09fa4a76a923df : SUCCESS in 6m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3265c404bd5e4aeb9f6fbb3dd14436d6 : SUCCESS in 9m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8105ed36ad5e4aceb6c4dba2a47a1d3c : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e23ec6753c042959723cb950390ac3f : SUCCESS in 8m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/768d6cc840824aa38f03763ad2aaf620 : SUCCESS in 8m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cff89300ce5f40d692e2a55596624216 : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/deb22523526b44b4b6a33fa70bcb267a : SUCCESS in 2m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3b3cc2c2423a42d58232443a7e2a655c : SUCCESS in 10m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f8bdd5a0748c48deba90752ad562a0e2 : SUCCESS in 5m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/67934c5dd2ef40089473c8571a715bb5 : SUCCESS in 6m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/86e864d4ed5e4f4aa3407026faa44209 : SUCCESS in 8m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b3f3e64eb7bb41ce838203a6ca6578bd : SUCCESS in 17m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cd24fe3806644190971f55f5067c174f : SUCCESS in 6m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9e069c1a24884b5c9fd9ab23d2936f15 : SUCCESS in 10m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/44a401e3dad3461aa3630b8fe06f775c : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1915598ae1934c95a97773c967c7890d : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f79e199a1e4a46979e9d883084d1680c : SUCCESS in 22m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0f9fe13dee0944c0b35f3ea0b2c9cad5 : SUCCESS in 3m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fa52415ce3fd43dab14efa4a7f419922 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a100c4ec1dd415b9845e409b649519f : SUCCESS in 4m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5bec74ec3bd34842b56310da1d0f06d5 : SUCCESS in 13m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4b6cd45b82aa43f9aba9779e9907bbf1 : SUCCESS in 3m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19b4c36e3f634a7c9d9109fbeb4acb4d : SUCCESS in 55m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9002bca23bf34476bdcaa7ee2b79c981 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":100},{"id":"df711e25ab79b901b968e8506191dae210dae1fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-14 13:38:30.000000000","message":"Uploaded patch set 101.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":101},{"id":"d1951e24ae92e20e2f820b3dabd858a6f1935e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 15:33:13.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3fe025b0c9e495cbb542f75b4160085\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38d59f9586248bf857629347c77ac43 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/392e3e67ba9342af8d6efce91bd5de47 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3657d6a21e904eb6a7909521fdec60ea : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3ec1ab9801d4f9f8cd1fb38e0e05e87 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3a9004906f546e3aa3de0611c6546e1 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07687d7a55494652a79b37727120cd70 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfd2c07886994b00b20367272f5bf3a6 : SUCCESS in 1m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/94265f36b54c4fd79ac7d143cc060179 : SUCCESS in 12m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f6b86c9a59045318b986b2b4ef9dd94 : SUCCESS in 12m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ce291d73c1ee496fb40fc6f7a1baad7c : SUCCESS in 18m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d122eae798454c9c90380b6cd8ab3a32 : SUCCESS in 18m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/adbab0ac8253442684338f0484c0794f : SUCCESS in 27m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d26df1193c90443692f8371c057ca945 : SUCCESS in 3m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/afa0c3e0aca74c2abc70c7b392b7f725 : SUCCESS in 8m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5bced3db312c4ab68de415ec2b6bf641 : SUCCESS in 3m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e9cc3850f3ac4191905b8decb70897c5 : SUCCESS in 8m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dd7693a7529d4003a60bccd172942fab : SUCCESS in 6m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9393d72fe0a34694bdc6e94271dc867a : SUCCESS in 4m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aa4f9d5741054e7ea765309637342ee3 : SUCCESS in 4m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cb9aab375c56417aa1d6dfe2409bc279 : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60b465a1f58c4681950198e49c04f31c : SUCCESS in 9m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8ce39e78a6544d52ba6e3f18865e6976 : SUCCESS in 7m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/78bd8500efab4dfeb0f37c0f2836d090 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/50d3d041742949129821be1da4b2e2d5 : SUCCESS in 2m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dddcc733ffef4006ab598e8b77798781 : SUCCESS in 7m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4f8cb0b651704cd0bea9b79a9bf8cbfa : SUCCESS in 3m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dbbf7fc67f784d968577d1ce66d348d3 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ca4cd9d89f7748c891270e8306568c0a : SUCCESS in 12m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2b1f688c7b204ef3a05c316349f49aaf : SUCCESS in 30m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71107caa98184040a809e2ae7fcc3261 : SUCCESS in 6m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5adca1399f824f7e8b44f01114e283c2 : SUCCESS in 12m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4506cc051a4040d084b07632a2f4ff44 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e3c889a846f41e1be56f79f842ec06c : SUCCESS in 2m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c2dd147e45242539f7056c39dceb8db : SUCCESS in 20m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/427d26cf21594aedba226b9331bbd10b : SUCCESS in 7m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/63886d3c1abe4fef8661f478507bf113 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/873d797263514f0496f016b37d94b9a3 : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9b237716e3734927824152d931b8b6d5 : SUCCESS in 10m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/41e4804d475644d79e01e44d7b5fa89b : SUCCESS in 4m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7309a9b77cf4e57bffcda9179971a2e : SUCCESS in 57m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3aef4d58106424bb2260be633e6e559 : SUCCESS in 1h 37m 59s","accounts_in_message":[],"_revision_number":101},{"id":"5315555856faaa67a01dee67e71c0d5b6980363b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-21 11:32:25.000000000","message":"Uploaded patch set 102.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":102},{"id":"005e9d2e7beb68429834f61a388f8986fb89af54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 13:08:57.000000000","message":"Patch Set 102: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eb4d347c2b4a6197bac05d0d836b04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab070ed9befb49c9a24ad9eda8d36156 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c14a1f619db42f68e1ee86aca28019d : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddf1b3dcd26746d989dfac0be08f7021 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d34e4861f674de5bb5bbcc61377751f : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc114220f9bf4249a55e6b45b581659b : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b6255fd4d584572b321f2093609ee7a : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f950f9f327d14e3cb480557c5f82ad5c : SUCCESS in 2m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/81ae4dcd5a9d40c58e0121b6de2c9650 : SUCCESS in 14m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bbfcf3e6922447d9995207244ab09a68 : SUCCESS in 8m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59ffa075e63f4b429ebc5a47da70667e : SUCCESS in 19m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b174b08160924aaabcd8de44953b65bd : SUCCESS in 13m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad98a5276afe4b938c632418a7d0e2be : SUCCESS in 15m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1362e926227347e7b6f878c26e137c41 : SUCCESS in 4m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fdb0f0579776442a8aacd2c9970ebc0a : SUCCESS in 5m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/94f6e48f9d054802a62179ec02cb1721 : SUCCESS in 6m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/37d9758775544f0ea5fcd5c99b8fd789 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/55cf1e062f4a4f4aba55522ca299239c : SUCCESS in 7m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/41ed49230f84475dadf6622c4dcfeff0 : SUCCESS in 7m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/58ee2ac7f1ce46e68d95566315eeeabc : SUCCESS in 7m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/691f3bf1b308411689e54930bdfba53a : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aafef8c061d64b44b3cb14d75dde6d65 : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d7aec556ec743b78e2d5b99715437d9 : SUCCESS in 9m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/39175222f01c4582ac4bb823b5cc27ba : SUCCESS in 11m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3e990b465c7b4d549788af3456912931 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bad9d71d092f44e4999665bcb8b7eaf7 : SUCCESS in 8m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/041558e340c54674988257feec71673a : SUCCESS in 4m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4e621f33b78346ae9bd44f8e8fa6b671 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/548e9cd45b774633a6724c8ee05229f7 : SUCCESS in 9m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7a88f28dae5c46bba607b4afb932ccc8 : SUCCESS in 16m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c38ea154a4b846b29e40cf51a3faa7e2 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/160bf34aa73c42388220c86d82b97f63 : SUCCESS in 12m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/40d120df62414803afb9d4dbaa434e49 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b1b2b724af342f38f5cc4472840e2da : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/035a8a699efd438b88c9be545e2db96a : SUCCESS in 20m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/77a00e8d85bc4a599014d7337848347a : SUCCESS in 6m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1510082b2f6b40d9b209e12a3a1eaa5e : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40541efa13a042809cf568a6181c286c : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b18fc28dda0540b4aca87f762c9362d2 : SUCCESS in 12m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/738467eab4734dbbb426bc6416c2420b : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68014b83b47148fbbb43eb35cb9eeec2 : SUCCESS in 51m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/775be6c8773541929bf089db0c766b02 : SUCCESS in 1h 34m 08s","accounts_in_message":[],"_revision_number":102},{"id":"8c9b5ab8100b189d62f85400a89492806a45f2c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-28 11:20:50.000000000","message":"Uploaded patch set 103: Patch Set 102 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":103},{"id":"7a7a338d41800da0c6573d7189a963e4333b6c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-28 12:26:13.000000000","message":"Patch Set 103: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7dc7d8880074135adf5e8c873113526\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d748b8c7e88341158b36e4fa940b29bf : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d28cb87853c4783af43fc5ed5a7ae3b : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a43974aa6894995bdb823bea60c617b : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1eb38876f064a599496525dde7071a4 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b19dc1e599f64af58a8e4033870fcc0b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c8ff8ea847d49d282980d75ec238eb1 : SUCCESS in 2m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9df03998f494d8492ba42c5ee883a64 : SUCCESS in 2m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4a5c1af9b39e4121ba4f96016fd00b5e : FAILURE in 5m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8982be179fe84a81991dde3928006c15 : SUCCESS in 6m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eaa1990fd8284e4483f38aef4979e431 : SUCCESS in 9m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b192ac9013324c1f8976ce25cd99c066 : SUCCESS in 18m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/058011767abb4484b11d94f938846c34 : SUCCESS in 9m 23s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/58d8e7c092814b32846dc4e256b7a632 : SUCCESS in 4m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d95f3feba321444f9af335a190de4e95 : SUCCESS in 6m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/30e66ce266e446d28963322cbe99f341 : SUCCESS in 2m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/48250cdbd17d4ea7ae826d9500f13c15 : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d04cd4a78b0e4a62aee5222026c502c9 : SUCCESS in 6m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/635d13b9f46e4de59af8355e644b5f9d : SUCCESS in 4m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4759cc3fbcb54fbc8cd66af0622a2bdc : SUCCESS in 5m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0e9067cee06743afa86909c9ecc220ee : SUCCESS in 13m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2cdac0a025024635bff288130476dbc1 : SUCCESS in 5m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ba2a0483fe7149f3902a434a93305d42 : SUCCESS in 7m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bf5a02bbbdb043c98da6451ff7b64fba : SUCCESS in 9m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ec659ce30dc244039782aa9e8daddfa2 : SUCCESS in 1m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/82a25ef92c8845c29f487087ea26c983 : SUCCESS in 7m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c56639013f38410996285225332f572f : SUCCESS in 2m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d0047d581b94604bcb9f4ca3232de95 : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bb81e147615442e2a32833911836c87c : SUCCESS in 8m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c1f943d1e61543c6a5815efa6b58fe78 : SUCCESS in 22m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/824be48aa7e94f21976eb9c098d17c0a : SUCCESS in 4m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2e7228ae45a04914b91cf948639a2e98 : SUCCESS in 10m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/83d4b66ffe824b4a9f10cca938e0184c : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a89fa56205242158c71ed5e4c04e342 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fcc9fbae4a64c3ebf18c989be334af0 : SUCCESS in 23m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b05e67607bad41a6ac0407c90078406b : SUCCESS in 3m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/71493e6217774550a137bb4d617a2b3c : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/687d0ad15dd0480b91def075f42d6bbd : SUCCESS in 5m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cb2fbbe61cd94be6975ff62c623f4e88 : SUCCESS in 12m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ab38bd233ca5422c90d25f689551d319 : SUCCESS in 5m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/206f5883539a441b893350a8747b1752 : SUCCESS in 1h 00m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5b6377bef1e4c428c4f1845f0ba6aac : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":103},{"id":"c232e70c6f8705ecda407ebf7fa5577fad602040","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-28 13:14:22.000000000","message":"Uploaded patch set 104.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":104},{"id":"897ef02b9c58663f7823325c2fa99c173249f58d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-28 14:39:49.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d88f71947854489d8fa4f8bd64338872\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfa3167747749bf93f2a900419afc10 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2515b6da5f4c2c8dc85a44972c0103 : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc7e52fa46547b9a4960abc9e816e27 : SUCCESS in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02b9748dac0f4da28c868ff152c1a45d : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3226482a42db4bb490723be6e7426681 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bcfa6e7831c4f988d8631f4620890cd : SUCCESS in 1m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/458328c51fd842ada4b3115862c49fe6 : SUCCESS in 2m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b025576ecc1c4d6f94c4999e26c84166 : SUCCESS in 16m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d94aaa9d8c5d4ba29106480bb555f5b0 : SUCCESS in 4m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a816617e0f11472b8c0f8d612a339560 : SUCCESS in 6m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/36c844e0f9eb4842befaac67837ad6ba : SUCCESS in 7m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2a9f75bb9313421baf6163cf01846af3 : SUCCESS in 9m 09s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/99dc980db08f4f2d8549105d85b02cf8 : SUCCESS in 3m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4fd949a7cd224855bfb49af2b12a97ce : SUCCESS in 2m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a5b6d2aecce040249713b5b670dafaab : SUCCESS in 3m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/85ef5c2e0deb404999200a55435c4758 : SUCCESS in 8m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4022a88ca22a411e92e75a31cdbe0334 : SUCCESS in 4m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/07fbca558737458fba656cdf4af2e75a : SUCCESS in 4m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ed774fa6af734369a70b870155cd8ef9 : SUCCESS in 7m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/13aa92fd7e24414c8564983752ad508c : SUCCESS in 7m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57dc7ebafbc64c6faa19ff1068a1880a : SUCCESS in 5m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0400bc78fc9f4839814f4c65a761a86b : SUCCESS in 6m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cea84d07305844ad90b084ac7570daab : SUCCESS in 10m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/53165da88f114fe18cc813778378357b : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cc9a1bae8eda4d9b8c68e6c8c5752ddf : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8bcbb92a5ea24888b119834e8eec0829 : SUCCESS in 3m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/13ea81ab5a644bdebf6f314010918fbb : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/844377d812ec40eaacc2ee2c8f3d523a : SUCCESS in 8m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8666c5bfa9fb455b9bbbe7667d25cab4 : SUCCESS in 14m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7936c73b6f8b4ae18c9cc455d10ecbba : SUCCESS in 6m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b3abdb87366345029dbba5be97930ca7 : SUCCESS in 13m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a98897f0bccc4da8beaa944d7786e70b : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f4541a87d5fb460ea86991407900de48 : SUCCESS in 2m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7883b317501c4f6681a5707512c26c55 : SUCCESS in 26m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/19b4017bf88345f181e670560579f547 : SUCCESS in 5m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a3f57f4fa4c343c89f0e957d5d677cb4 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44c703c9d92c4d93ad0dd97518f1fd7c : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4732ac1e8e53428898fcff3211dea41a : SUCCESS in 7m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0c50014119bc4c968153bd408e6d3861 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce67fe46515b4093b4360a974701f0e6 : SUCCESS in 46m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efdc523f2411447db1036ee9ca57a8aa : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":104},{"id":"4d93e07d2cfea598f0dadb435ccba35b1f7a29b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-02 10:51:11.000000000","message":"Patch Set 104: Code-Review+2","accounts_in_message":[],"_revision_number":104},{"id":"ebadac897a637703d2486ffe8c8deec8f5017d9c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-02 12:25:26.000000000","message":"Patch Set 104:\n\n(1 comment)","accounts_in_message":[],"_revision_number":104},{"id":"b8230b4f00abae946a3ea32ebd9512fd760b99b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-02 13:40:48.000000000","message":"Patch Set 104: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":104},{"id":"2db4fae0922bd8b3ca73bc50bd31b3a2a639a4be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-05 13:00:15.000000000","message":"Uploaded patch set 105.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":105},{"id":"3e138ee9468f686912bc2c374a0194e72e182b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 14:16:11.000000000","message":"Patch Set 105: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fade6435f67249d4933c0d04ce4444ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc00a06be83f4ecf8959fd008373504e : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96199a72956431fbd682978268167ef : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/059dc68f9a9c4caba4e8a72e300f94a1 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/559686d7960546deb4480e37441b5487 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44d68e4260994613ae6818d93b0d95d2 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b702a55a03374f20b41c9b42fc237e5b : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73ce983e7d6f44feb82a6d6beb5c86d7 : SUCCESS in 4m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/22a4a223d93c42e68c5617c860fa14ab : SUCCESS in 6m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/732a1cca4b3c4bb09154f35bfaf9e6da : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cea5e396be21404b857abefc329bc620 : SUCCESS in 16m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a3ad31b342b74d6ab03b197d98cfbae9 : SUCCESS in 17m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9dfe647286c34241a5d7e9880d702e73 : SUCCESS in 22m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c6568306d5b44c1fac10a550023d4162 : SUCCESS in 4m 22s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/94c6ed3a022948228138647e64167adc : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e652192abcf649d48783787c4e3f22df : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cbcf47c13dcb45d49e6b6015b6e6b061 : SUCCESS in 6m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2b6aa00593dd499fb7adc5589aa15f02 : SUCCESS in 4m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/86ba3fd80b1c4d8498c61d4ef2647ba2 : SUCCESS in 4m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/83d9eba1a9af44d1bdcc81fde91b21a7 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/220cb7198d4d4ab8b27d7903af0cd74d : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85d177acc1e446209fbf59a62455ba54 : SUCCESS in 6m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e2a0efe9c00e4abca03ceca439dfa9c6 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/358f7252c12f401bacc96d8e7c96008b : SUCCESS in 9m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5e58849efca64fe6b44f56e940481f8d : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2ee81fc32ee245399e4a0b3fc828e0de : SUCCESS in 8m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/36699c081908456b891a9b0feef8d732 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/327d8cf11ac941c1a9195553eed59f8a : SUCCESS in 4m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c10aa6ee83cb4e53b2a99f85e0c4cb4f : SUCCESS in 8m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9db545f967e74a35a9064fe9d0328e6e : SUCCESS in 23m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c6c74a0c521a4d96bd749ba801132bbb : SUCCESS in 8m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f68da3a420cb485798d7f5222b7d8bfa : SUCCESS in 11m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/14e8c295db074b6b851ab8af03c0486a : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc9ff460331e48af9cb0db8d80cad428 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6015ef531dc2476d9e629b5cb96e8b04 : SUCCESS in 21m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a0d1bcf44473409184553182dd54aa29 : SUCCESS in 6m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bd9ce16e1de548c287b68bb38efcd63a : SUCCESS in 3m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d09b63e85bb24fb48cdc32f20ff5d063 : SUCCESS in 5m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/842345cc70eb43ec871825f7738f4693 : SUCCESS in 17m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fabe6254706c4b488a73f168cf6a0923 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9352a5aa5b8a4a34a88a4018eeceb4d4 : SUCCESS in 32m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cd94c632fb145ce95afdf0e54d0c279 : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":105},{"id":"7827773cb7f6b49fa783f8698a0134fc7b028b57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-12 17:08:34.000000000","message":"Uploaded patch set 106.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":106},{"id":"945b9e03bb31fe98d7db5f2490ed3542e1981f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 19:04:06.000000000","message":"Patch Set 106: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c05455a103a4a2cacc19fbf899f7efd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc0c61abd08474f86b29ff45554dfd2 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ed055250cb46969fc5df46897cabb1 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39ca9b950da148f6880c54958c3e5401 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/250e03effcd7475a8ebbbe7e1a983a25 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62353f9afcb14c6e87a4631857c602cb : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2ae87cc5d83492c9fed99013cb2119d : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9893107a738045f8804c8f70e3f95077 : SUCCESS in 3m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56ca2b20c31e4500be85a6d967ef1705 : FAILURE in 2m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/63d2f03ad3a04fe380b716e03f32e8c2 : FAILURE in 6m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/18ebb15d0c40450fb4ac04788f2797f0 : FAILURE in 13m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3dcaedeb9a7b43d49b8ddf809172ee60 : FAILURE in 9m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0482e6801bcf414fa5107650ec411f52 : FAILURE in 9m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9eb92724a850457a984249438fd0b00b : SUCCESS in 6m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5546beb15c1b4f93af1038eb16ca96b2 : SUCCESS in 3m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/161fb065eb31411fa4d721e6a6cd50e8 : SUCCESS in 5m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8498c9cef3a9402b99f9d417907b4175 : SUCCESS in 10m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c299b8ab48e845088f42e382c6f9431d : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8ea2ff1b3e1f4a27ae7aff77f82f1a71 : SUCCESS in 7m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4c4d85ebee9442e3a6c635331a8f10e2 : SUCCESS in 9m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7274d5b5cd064a2d9a0ad5881800d056 : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2165edef3fc54f3fa414188919203fa5 : SUCCESS in 6m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/15133655a4184ae6812933b81258b0fc : SUCCESS in 6m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8ea950d3778c43e8bb519cd75afdddc7 : SUCCESS in 11m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d3ed84704b4c4d3db260e9a88e3e8a56 : SUCCESS in 3m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a088fb66eda744cdb60f6c002a544b84 : SUCCESS in 11m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8c105ed9587844b7b472107da4adfcba : SUCCESS in 5m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8a4d061f7db24228824d3ce224fef43e : SUCCESS in 6m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f2142a486bf74a7789c41b2608df8728 : SUCCESS in 11m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/94b573b47ad642339b1727788998a37a : SUCCESS in 17m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d20137a4bbd84deaa23fb1eeb5229ef0 : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9dd3d0234c604c26af89762b28c2ea51 : SUCCESS in 14m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ba1182e919ca4a7ea4dfdb265aad2199 : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bc50b4e81d846849d330dd54f27b1c1 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c54457ece8e84b6486fbce71c8deebd4 : SUCCESS in 25m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f226cbb53d6044f0bbe5bfd9cde5881a : SUCCESS in 7m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4234a4496b4b4546b69e2cac94ed2894 : SUCCESS in 3m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e1dc1bce08a40b2b3f09c34f69dc171 : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/405b2f83d6be43eea35241b15536eb81 : SUCCESS in 10m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b8b6e3430de047f78fe38374fe239d1e : SUCCESS in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ae42a67213d433cbf532b8df2570b04 : SUCCESS in 58m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efc41277d50d4735a115f9a984d18f6d : SUCCESS in 1h 46m 22s","accounts_in_message":[],"_revision_number":106},{"id":"67eb4e6448fed3737ba90776305fbb58dfc788e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-16 07:01:07.000000000","message":"Uploaded patch set 107.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":107},{"id":"d2d9729969558655ad6e250307ea163a0052e573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 08:17:22.000000000","message":"Patch Set 107: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2815a59a7bc745b9b349ee1621d749a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f3fe7f373e4725b761b9c1e92c980a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61ba32da0bf4b2389b18b7117d098af : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5313f9a503f846b9a42c96d7ebd1ce22 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69162e10d3204c2a9d4b0970d51626dc : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8624205085e4f0a945cb2f7e25742ba : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e0be27f2e564509b58ec50b9aeb3848 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44d6d726d63c405eb96e80c818d0ab01 : SUCCESS in 4m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ab516702e5f4081b108fabcbefb7157 : SUCCESS in 6m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5ee9644551f140e78247f95a338bf99e : SUCCESS in 15m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/251d8d55459f4a24b37489f3d16a6630 : SUCCESS in 17m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f34e0d9592574ca4ae884e8eb98db947 : SUCCESS in 18m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ea793e2c6a2949148ab6e4a20a316780 : SUCCESS in 27m 35s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/302c861e57074388854db889a3ee8b2a : SUCCESS in 2m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9b21575c4ea44a7dae49846fe89ad9f7 : SUCCESS in 4m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/625f8f2012c64509bc49f8fe9321795b : SUCCESS in 5m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2cc07e94912f4e31aad38f48fbf025b0 : SUCCESS in 11m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9fbb641201c5464ca6b8c144c65e1b96 : SUCCESS in 6m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eb2cf4ff621c4f96926e5b1b3762a8e6 : SUCCESS in 5m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/00aa03f1fcd845bd8935219c9977ce11 : SUCCESS in 8m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0414a864e66c48318b4ea74946d2a9dd : SUCCESS in 16m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/064606dccccf406cbadc3732c5bbbc07 : SUCCESS in 4m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/211c19c965fa46459c50cebe4aa5bd9c : SUCCESS in 5m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b8a84b5e00cd4d05af31ae6fa366cff6 : SUCCESS in 12m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/427b9fb73f2b4d76a95239ac7df2841c : SUCCESS in 1m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f2e33076c5f04294944c857ca59f9de6 : SUCCESS in 10m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f6d65c9fc35745948ca273c4086bae7b : SUCCESS in 2m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0a7e18d87b214e70a6e6fdf2c3303da0 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b066e4f6fb7d427d80a3296c8a24135d : SUCCESS in 9m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d66fd1374ea4387b34576a877a60c24 : SUCCESS in 13m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/89a6831044a04e1382ef3684a54c4681 : SUCCESS in 3m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7764285a4d2843c7a46b15df4ccaa94b : SUCCESS in 8m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/420ee2192a4846448bc2751ba3ea8a0c : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3f4dc8c7c7d42aba6f89f4735ce9ef6 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/026413a4e94d41a38545d526fdc7820a : SUCCESS in 26m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/092499f3703840549e9eb33f8f3241a9 : SUCCESS in 6m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b58ef602cf9241c8b4b83b980e961b31 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d4deb8bb5d54e25ac738ff303bfc393 : SUCCESS in 3m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f4a6f088070349bf956d14696f2ed484 : SUCCESS in 17m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/354b69e19af249a0811e3a6dd6fe5cca : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d271d9b58034410584801f130c367774 : SUCCESS in 1h 03m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cba1e13b656415dbbf4a9e0f1961a04 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":107},{"id":"219511ef3cd0d7abf16eeedb675e101f47954ffe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-19 10:15:00.000000000","message":"Uploaded patch set 108.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":108},{"id":"3dffa16188b0d6936632ceaa986347320345458d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-19 11:49:10.000000000","message":"Patch Set 108: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/062339fdf5e0480abcfcc2e4c828e6d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7676bfac02e4de4b3038a70c14a52fb : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c284c9ecc3847c593baedd80b6ceb31 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41081119d7364e5f96a4ddcd28d766c7 : SUCCESS in 1m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35779f8175064884ae1227e4e6494a30 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c54bda0b8614c51b8c99c9ca8cd469e : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb49d2170db54973870b729261a39e95 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c9aa464daca4618aba466a4e1aa00ff : SUCCESS in 1m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/edabbdd605944239956b910c1ac66b83 : SUCCESS in 16m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/59da06f5d7074cea9fbd8ea9ba8dae02 : SUCCESS in 4m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f54e114758904d5fae3d9e905b2c4858 : SUCCESS in 11m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7c97848b28a44906ba066b3065951e48 : SUCCESS in 12m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/63faa48d4d5f45019b1dd600346a8507 : SUCCESS in 23m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6da5bbfc93e54a63900fa2666121ce06 : SUCCESS in 2m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/26e16bdb9ee847f0a94d8aacc3602b43 : SUCCESS in 6m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/16556391081b4b23afd8736b24387051 : SUCCESS in 6m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d4698ca6bf194be8866170cbfb8ef5d5 : SUCCESS in 10m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6211245b45bf4276becee20827c69b44 : SUCCESS in 6m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/95d0f760d09a4b48bc4e5662507fc984 : SUCCESS in 7m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fc22ad41387a467dbadb715c55643b16 : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e8737d70c6084aa1b92c18455e0481b7 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d43583cefce4eb79348ebc44b6c1d5b : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6522e4101d3e4b2fbd0e6247ece4bac4 : SUCCESS in 8m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/47d6610f98e6401a83ae05185d803285 : SUCCESS in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d637ad0af473477c84ed928dd2c4d524 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/09e687643af249ab91b0f28cb5e0cbab : SUCCESS in 9m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/69c61d693d264f7b960c69f6e902651b : SUCCESS in 2m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b8f488f000634a6caac68e87d24bd339 : SUCCESS in 6m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d6b801a7b2bb4f7f883b6f349bea2b0c : SUCCESS in 12m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f457ec744fe343cd8a9c7ae6401e45d6 : SUCCESS in 13m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d9b5f9ae59f4a57bd6972d39beccf3b : SUCCESS in 8m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4eed83e51abc44739602db76278670aa : SUCCESS in 8m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3354fd12666a40d8896a40d8b618032e : SUCCESS in 2m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b1c236ee0c94992a84be5f13aa37ad5 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b34993f675f48b8b1eb1095c45ec3eb : SUCCESS in 22m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6390cc7909154538a4834652da4aad88 : SUCCESS in 7m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0cd65370d1584d9799274062ef981ab7 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26d871655f5846098f2a298a1ad334c0 : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4c33e60551c04257b063bb61ffa4a0eb : SUCCESS in 17m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/15b532e37d4943aebc2c3e37b0a0c5f3 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/116edb34c73c4acbb6dd926f707e205e : SUCCESS in 1h 03m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3aa7ac4832f2456198ffdf7f5a5376f0 : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":108},{"id":"ed5ac7b5688796126dda2ba1b2047e06c719216e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-28 19:12:27.000000000","message":"Uploaded patch set 109.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":109},{"id":"e6ff52564e3f4f329c98d77b9c58de083be568b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 20:52:32.000000000","message":"Patch Set 109: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/665cf9e55e7e4e9f8323cbdef3c672e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e3fe6e4426460d9c563e23d269e755 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309f790e0e6743b8b70b3040fdfb8d2c : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/704e2d50d3464ac38ef9fbb087702f83 : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8dc05409f0c84bd2a4d55cb689a389b7 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27c7c2730e3c4471a65551e835435d5c : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a629be51aa764bada12034d8f5aad2d0 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d386305420294410bbe73e71c33111e5 : SUCCESS in 1m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aecb85cfcc304e3b9c721582230def3f : SUCCESS in 6m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ff1e16d1313441f8980f083e05003e1e : SUCCESS in 4m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/78aed7a53941416b8ef4113e4ccc530d : SUCCESS in 11m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a197762eb2344cd7b1da149a6c1869ef : SUCCESS in 13m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/daa65d1eb52941c8bebced2d3a7d1cf2 : SUCCESS in 26m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/720afa81a97c4bb39a4fe3a842b52acf : SUCCESS in 4m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a183a18a3df1495f8d80516aecf2c217 : SUCCESS in 2m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/26f6b2f4d28b4bb98689f1bd2d4909de : SUCCESS in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/597655b5596440229e844ae9adf3ca8a : SUCCESS in 9m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c6c900d453af4db3a349cab2a6452b45 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3d8e30103f604caf982b56392f981822 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5e2a709bad55421ca8743b16d24f640b : SUCCESS in 6m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cfc0665cdda14a15bb1aa3da29677ac1 : SUCCESS in 14m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd144fd83ae040bca40d08820f6d6baf : SUCCESS in 8m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3e3acf77366144678319e83f73a6048e : SUCCESS in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ece1e8d3a1c8441e8e5ddace0289bc02 : SUCCESS in 10m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/66596c8525034c82a4736de2d75f44b2 : SUCCESS in 3m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/377a6cec379148b59285146b44160c46 : SUCCESS in 8m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/26f9302833024fb6aff475da9b299502 : SUCCESS in 6m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/16cdd8ad68b54087981d79e2bc8dba27 : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/888b74bab9cb44a7aa978a9487a55251 : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/591f169acb074e16aa4d5d583116b804 : SUCCESS in 18m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/84c700701e79483f9abfb615ceaefd20 : SUCCESS in 8m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/34198237d9fd47a181dd94f6184e462e : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/553cb1944b924b48a4eff826b5da8c0b : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7da276dacad54d0793d85f1879d1500d : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fdc918f872b489e955f9f2b16ad7578 : SUCCESS in 24m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/410bb7841c944231b9af43f63134dc02 : SUCCESS in 7m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c1dfcaf445d04c3e987a2866f58a0a15 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e4cf7d853a247cba10026ea897801f5 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a4b25f192cbc45aeb1c316726e27a9db : SUCCESS in 18m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e7e0b50100c74e0e8292b4035787a12b : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14a7bab978b14b9b8aeefee7a7d83170 : SUCCESS in 1h 00m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/313529b412a0442eb8510c76e79749ae : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":109},{"id":"bf239d43b783490dc67ea7d4b34e4b7c4c3e3f88","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-29 17:13:16.000000000","message":"Patch Set 109:\n\n(1 comment)","accounts_in_message":[],"_revision_number":109},{"id":"98d7164212c150ee4d44ed6d75ac7b7307dd4fdc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-02 08:44:12.000000000","message":"Uploaded patch set 110.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":110},{"id":"3761f1326ac19ff5ab1ae37681758066818315a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 10:05:03.000000000","message":"Patch Set 110: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55ed9a025dbf45b8a812d8abea961397\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d901ca1e45544e37a44728041f4e59e7 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4e85d0d8df4a04be1b69b55079e512 : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ded058b851e4aef8b23dba6f36c07b1 : SUCCESS in 1m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fea11fda814c4988a1b640d2cc9e08ed : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39a7595a7b334dacb89b3e2bad8f6bd3 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95201a7ec7e942ba8b3fda8735b87e7a : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b11bfa282a145b0acd8a612a153e256 : SUCCESS in 2m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/790bbf63641b443bb2bf9e015cadbff4 : RETRY_LIMIT in 1m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/569412192fd74e3ba38a33ed332ae704 : SUCCESS in 4m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/160561d36fa04f92a8910c51955ff942 : SUCCESS in 17m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6067b760904c480991746241fd78be96 : SUCCESS in 18m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/473d1a2cbe5a404e82541d9aa3a8dcc8 : SUCCESS in 23m 22s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9fbae9e0a12e4c1494b2fdaa2d908c29 : SUCCESS in 5m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e533ea7472fc43f4ba93e100de3256ee : SUCCESS in 4m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/956045b59aff4421aff1ca41d4c453df : SUCCESS in 4m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d1bc83a55dbc442abef094324db218b5 : SUCCESS in 11m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a5c6dc1eed2142628d204d1bdc8f70f6 : RETRY_LIMIT in 2m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/397c7a22b6c24c57850db2d15ea6950d : SUCCESS in 6m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/88a4f55b6fdb4c94bfdbc22081db5557 : SUCCESS in 4m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bc85e51cca0447039d7841c3979fed5e : SUCCESS in 15m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38f3bfe2d26e426fa6095920a63836c1 : SUCCESS in 7m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/75b22ab26ebf43e2b58e55be5481db21 : FAILURE in 4m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3faefc98ade94366a07ceb24c047d548 : SUCCESS in 13m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a049e46f83bd4f33b49c256561547ebf : RETRY_LIMIT in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/556e2735089348539d18ee7a452a02a4 : SUCCESS in 9m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f36d50aa00e24389a663eabb93f27e1d : SUCCESS in 4m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dd1764e80f1c4ae19601c2bf55a19922 : SUCCESS in 5m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c2a881f892b24796bd046dd262427fcc : SUCCESS in 10m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f8ef9a87876a42ddba45333e2a3bea3f : SUCCESS in 19m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4c88466ee18f4d68bec5fa435df43bfa : SUCCESS in 8m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7be92c5f75994618ba1a1c5af06356e5 : SUCCESS in 13m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7889974b6e04a45b7df3866b6c2a379 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58be3c23ece344cba275fb51cc13e52f : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1373957f73594d6a870315b8ce859281 : SUCCESS in 24m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/02763af2e1da457cbba8390bc32fddce : SUCCESS in 5m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe4935bc21784bd8b8bc89737a988851 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a57f352f306d4cabae0f2156898e5709 : SUCCESS in 5m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d04bd45a1bf14dc782f0c9014c9f7b8b : SUCCESS in 16m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b32dd1d3536446758d790111acc0c222 : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9435872ac316478dae007a4f30da5fca : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/830d6c52e517462cbfb11567c930aac5 : SUCCESS in 1h 19m 05s","accounts_in_message":[],"_revision_number":110},{"id":"0e482e24dd96a72b431beb2a52392c989c94b550","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-02 13:53:57.000000000","message":"Patch Set 110:\n\n(1 comment)","accounts_in_message":[],"_revision_number":110},{"id":"56fcac6f8cbe7ff0c93f86e2cbdd08edd5cbcfb1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-02 14:34:15.000000000","message":"Uploaded patch set 111.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":111},{"id":"5893a739fc2fef9237c93482a19bdab65914dc9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 16:07:50.000000000","message":"Patch Set 111: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7eb56f9ddf949f4962dc7edfd967dc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4313bd099c34964ba78042857f15982 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64f6de902604a29859eff8dd7dd652a : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cf721c87b6843439051bf43b3a85fcc : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b4fd63f1695490e95db623a8c263b57 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b657c358cac64a968849b4db5fb3afb0 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b820b7f58d7842a1ab03b08d2c176e8a : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/006d999f37c846a495d587dcae523be7 : SUCCESS in 4m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cc8b398cb9f8479f9ac20f5de5ed820d : SUCCESS in 15m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e70f82ac8db94c608b79fde3bc07e3ac : SUCCESS in 4m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/99f6ab32cc3045fd9d1f427b35907178 : SUCCESS in 16m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/06f98c4a0ec248d698c731f7f3bafcae : SUCCESS in 20m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f65e81259ef442fc8dd451d49a2b3ac5 : SUCCESS in 23m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ec1b9fb916c443b890ddad51fca76745 : SUCCESS in 6m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/82612b461a7f40d2bdbd3517bdfafc4e : SUCCESS in 2m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2a8b431da96642c4b35c96039c74ef4c : SUCCESS in 4m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/95c54edfc519424db0c3c30ce4e49aee : SUCCESS in 7m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/19ad0aff13a64655a757a550be6ee86c : SUCCESS in 4m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5d5451c0fe524975ba85564cdaf42f10 : SUCCESS in 4m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d7eec55b308e4fefbc6eeaab00849180 : SUCCESS in 6m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b9b40726156a4ba1af59efcee9cd4cf0 : SUCCESS in 15m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e2d69d6ee95481fa735e19cca27cdc4 : SUCCESS in 8m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0a0c348ae8c845859fc2742e6a0f71ca : SUCCESS in 8m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/63f9482ccbad45a09976e8ce3d9aa7a1 : SUCCESS in 9m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/860c9fb60d224f4f8f2fd1ac3320b929 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/415bb38a17a94b9481ab7769a6eae1c2 : SUCCESS in 11m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6c408c76b96d465e83cc12c5d1880fae : SUCCESS in 5m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8e3700bb5b9144aa8d7296ccf66a46a4 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0ce7e1a9cf424ea08b4105cf97663cb1 : SUCCESS in 11m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d3f04b4a07564684a873c5e35fa2338a : SUCCESS in 13m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc136abec3ac4aacb4d14d06728ddd4a : SUCCESS in 4m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a089d201b4aa42cd833f9f25c1e15337 : SUCCESS in 14m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/378f7fe9e5bc46a2af13f845b8a061af : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b67bbe770318482fa098daa17d3688ed : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74e9612380f14e85b563962508b4c8ad : SUCCESS in 28m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/af36225b74094bdda8e1934d4d6edabe : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f7167008b5b84e249e515d72b0e116a5 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe6ac41f3a77419481cabf4f4d06b836 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b0de82b4e1de4aa09fc98809ff71f601 : SUCCESS in 13m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e04e1a3f69044833b0425d37366cf5f3 : SUCCESS in 6m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01a6c1b2bfea43728e820fcdb8337128 : SUCCESS in 45m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f72ede0c5f694eb68eb36962f66f6962 : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":111},{"id":"75e83297029a4d8f28b07dff75e58980c83c8253","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-09 09:15:27.000000000","message":"Uploaded patch set 112.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":112},{"id":"eb574e4d7025391addfdd36dd247d051e8b35f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 11:05:17.000000000","message":"Patch Set 112: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b099bd3ae499443e922eb43af0fb76d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e59bd3b0f7f4de49ee813491b019ec8 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cecf01f4f06a4258bb85d08f6fa0dce3 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b999db4deb64895b965939a1981f866 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b55f48c812f42ad8c40bc847b53b459 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/824d713157714d10a26b3180c84bbbb5 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b78a7abe5de4e31ba1c6a19abcf194e : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bd54d18a7704667aaac458b958c10eb : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/91dc3ee42ee7455784d6cb1d0eecd20b : FAILURE in 2m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/86cf5676348843ea8a0584607e1667ef : SUCCESS in 11m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4c263c38023c46eba922365a324ddeaf : SUCCESS in 16m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/14e42bf96d104f8ea05a1a6b6268c650 : SUCCESS in 8m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/abb956fe4ee14f9aa109377fb51c8cba : SUCCESS in 9m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7e3bb29dcdd344bca7633f0f4ee6b9f0 : SUCCESS in 2m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71e5f20b62ad44719cf0657b2edb49a1 : SUCCESS in 3m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a52f10efbe8c4481b6ae06fc2be2321a : SUCCESS in 5m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/df691f54db004927b37f46ba3f7cb248 : SUCCESS in 12m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/888d74fc19884c7494947b364aa884fb : SUCCESS in 3m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/af9a4ecf65ed4309b3e92b7d82445769 : SUCCESS in 3m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8ccba6589ada425a8249661dd77b1ba9 : SUCCESS in 4m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9e2dfb1dc9e94088b72a35b082f9f4a3 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b0b98623b364f46a232d8cfefe4fd3c : SUCCESS in 3m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e0ae6d465169484e993e3a7722b95511 : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f8b6cd7c5a904635995eac76d72f20da : SUCCESS in 7m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/97fd1f02818e4a5aa6759c34fb7f78f2 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fe159af33c05427c8b89e40407521f1f : SUCCESS in 9m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1b182813dbee419d859792b3562f8530 : SUCCESS in 2m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b261947e12ac402cba055339406a5139 : SUCCESS in 2m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f4605d0e8e7b47a993bb1f99f3d29213 : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ce1fc9815d8e465d993e4f8313798311 : SUCCESS in 20m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3283908e797242efb39c1aa881e4cada : SUCCESS in 7m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/46cf57dc1fc3493abf36841f13b2967d : SUCCESS in 15m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ae71bed47b8049ee902e9554840c857b : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cffe610bb73f470e9ca991cd0813aa2d : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/918315d678af4d05be93f4de8c4679ea : SUCCESS in 32m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9d18876259c946bea83a49cb12eb9edc : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/654ec860dcaf4eac85e8b0944a0c6594 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ad124fd8a7c4c50b31c6be27bbf51a1 : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/296638d8378e47de9076455a2025a4cd : SUCCESS in 7m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8c5d8e963e274ad981cabe48e812bbe3 : SUCCESS in 6m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64428cbe63c3425c82098acd19dd81d4 : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df83baaa31ba4dcea45fe3b82fba9d43 : SUCCESS in 1h 44m 33s","accounts_in_message":[],"_revision_number":112},{"id":"3da760562ae646ae20116de4b15141d321be004f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-09 15:49:06.000000000","message":"Uploaded patch set 113.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":113},{"id":"3a13c0d47709014c9ce091130b7b6b1acc999fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 17:20:04.000000000","message":"Patch Set 113: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc325aa08724a7ea873b65baf78d343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f34288eec1481c95a948b04792f932 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184695433df746438cdc55aa18599c23 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b37fa888dcf426686f6798549bb7cba : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c667ca5530448a9ab48e66e8c756d18 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7031416a28a46dbabfd99747f4e2f27 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcec23948a674fe99f9b252006b46986 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b348c6c89e684f86b0d339397eb59f86 : SUCCESS in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d3f4a8f0821443d5a20ab2bbe97cd073 : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/06b976ca34be4339b912c27860186401 : SUCCESS in 11m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9b03789b44bc4502818c86446dab2852 : SUCCESS in 20m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc51ae7da42043bb8df4a7a6027d583d : SUCCESS in 7m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/39f98f4758a84fecb6ded5c4f7f231f1 : SUCCESS in 23m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/357406e9cf734acb817d0e6e62725119 : SUCCESS in 4m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1afb4a829e15467f9832931a615fed6f : SUCCESS in 6m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/01d9015b1800440bb594687a85f395da : SUCCESS in 4m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/17e8d7fa2762496db4df426027a4fd75 : SUCCESS in 7m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f27263ea34aa47b6beb43411e232514f : SUCCESS in 5m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b1b6c3c9fc76428d964a00c145cf9869 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b30e8fc1f05c4815a0afb14533602fb8 : SUCCESS in 7m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a547b5b2116742b99ae1f41dc51b26ec : SUCCESS in 16m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20134fb06db24a099be60625ee594018 : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/21b7fd05d14a4d20a8ead98255d1ff93 : SUCCESS in 8m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1269a2ceb8584865a55f98d17fb628c6 : SUCCESS in 11m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c7fa9e23d7344d2bb113248fe10476f4 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8f770e04cbb34242ad76747296ae5f7b : SUCCESS in 9m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/55df2fdf5ed648949641fe53287a4795 : SUCCESS in 4m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e186d9fe9326401992dcbf18c8b4f163 : SUCCESS in 4m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0f51ac7b9e314784b3faed6b9fac745e : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c9971225651f480ebee07883dd4801be : SUCCESS in 18m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b6c3fdf4c904fe7b0e11e55778cefd0 : SUCCESS in 8m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b49f6db14bc848688f602284d4ceef35 : SUCCESS in 11m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d7ee3aac8e63474b835d8d9a55e22f17 : SUCCESS in 2m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86e4e8c30793474789f2a8df2f2a34cb : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a35afaa36cef47a1be4a77dd780fadb3 : SUCCESS in 22m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/17c525e323d94b0ba90e95b7e178c702 : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ebea893fb6b54467af26ac36f2a18c7e : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b4f08c541354f438b1f5df98d10562d : SUCCESS in 2m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/eb7497ea7e3740dbaa20b8fc2994e0a2 : SUCCESS in 18m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ba92aaf9878498e86c44b2834103cf1 : SUCCESS in 6m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f4b704ded4e4be4b8afbd602f9b5724 : SUCCESS in 31m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50d513cbc2c1455ca0992a243eb93813 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":113},{"id":"5a58fde23df910032400049e292e7900d835a797","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-13 13:41:29.000000000","message":"Uploaded patch set 114.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":114},{"id":"aea17f83a18150b20154a673ae606f1f01aa2134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 19:26:28.000000000","message":"Patch Set 114: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff770a5fda164e05892783dfabdeee0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04878db0d29541699b66428804d6e6c6 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6f72891a9044479bb87ec22f07ca96 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66bda9f1ced4461f87a4fcb14fc2ea03 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cf5cd78e42245f896ed4c9b1f741c73 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc8317c3de4c44d2a2a6a61608086763 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd559524bb484ece80aebdf15f5564ca : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4387b78faebc43038e9b69025d2fbeb0 : SUCCESS in 3m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fb6deccc2fbd4828a8b03376e9fbee28 : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c887c53d3e042ebad542b435cfc1af8 : SUCCESS in 17m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5b7b222c8f6a437fabdc5270ad60098a : SUCCESS in 23m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7b91a1d476e9476287d1ba2a06658ed2 : SUCCESS in 24m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4471760cf8994bb380306194f6eba9c7 : SUCCESS in 3m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3fe201da76094ae59894cebd7f31ec09 : SUCCESS in 2m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/093e5b65af79467a9cf280fbae3b72ce : SUCCESS in 5m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8c6bd528f9ed424fbb93b1fcbf2da803 : SUCCESS in 10m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ec47e586bc674644a8bf4a57a443976c : SUCCESS in 4m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3e24caf560d1451fa6a271671324d494 : SUCCESS in 2m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/78731176bd4f4489b0302c1369f37e59 : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/48de6f0fc0484bb4926b19109e4306b7 : SUCCESS in 17m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b81434612f24f3b90e94e444f2627f9 : SUCCESS in 5m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/62917f38cc5b4e308388bc2f07855a21 : SUCCESS in 5m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1835b2f579a04fa09c3de28018ec7ab3 : SUCCESS in 7m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d9ad5c41f7b341358a5596f5e4c77f15 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d9a0a81f720d4151abfceb08be880ffc : SUCCESS in 6m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9b9dbcb4d13d439f977f44effecdd723 : SUCCESS in 4m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0196ec7b07554cd9ad3f5f7f28269669 : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f9c0faba56bf48d185e1d09af814bb00 : SUCCESS in 13m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0dbc7cee724c4425822b5629d46c88b0 : SUCCESS in 18m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/06d5c1a0d3114b8d841e4685f1ceb09a : SUCCESS in 8m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/64c64b29023746d3bf46e265623ac8de : SUCCESS in 8m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6ebdcbbee5e84f01a6d4c4dd57b27410 : SUCCESS in 6m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e8651a4d1234a9b91d5a7a7378d1dfa : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98482ef2fb104eada8359b6bf22a8791 : SUCCESS in 27m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/45ff9ab6b24f4d6a953ce5a1e3d561e3 : SUCCESS in 5m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9cbeb8fea6fb4ef088d532878398fc2c : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f2c5be6f203467d809c3001b7be65cf : SUCCESS in 3m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/045e03b4839c4885bc9a291ed0e1d98f : SUCCESS in 6m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5cba56f378d34bd193e0c5ccc45972bb : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f3b00b70bdf451183cea5649099f494 : SUCCESS in 31m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1926b7ff9985444e91578aefc9cb42f4 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":114},{"id":"c11662e8f2e2654b7f688ac80a074b2408cdfa87","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-25 13:23:26.000000000","message":"Patch Set 114:\n\n(1 comment)","accounts_in_message":[],"_revision_number":114},{"id":"3dad0bd94f8b8effca17d2fa3e787b3b1d2a1c0c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-27 06:13:39.000000000","message":"Patch Set 114:\n\n(1 comment)","accounts_in_message":[],"_revision_number":114},{"id":"d7cb243dde19c7b88a7984adc476818d57734e13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 10:37:07.000000000","message":"Uploaded patch set 115.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":115},{"id":"a1e4dfef2f60529b9169379d7d3e1ea0150c4118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:04:11.000000000","message":"Patch Set 115: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc84e76754d8471abb7a84b5fb975f07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1e56e765ce473da8b609cbbd98123c : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754cebea107f42f8955a083c670bc461 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5de0884e71547eb882329ca0a4e1e85 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06c9279366c34353a6d98bd0b5d07221 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac6991d946804ee0bbf2fbb738d5d945 : FAILURE in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5eb5c405b035482e886fe7a0237cb31f : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7803f29d72124f4586fc884d5456e83f : FAILURE in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2499b77d3d034c229de9cd347b79a401 : FAILURE in 3m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dafeafea0d524bf986a8af1477e3967c : FAILURE in 2m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/07b48a1bf86942019e26bb88f65de33b : FAILURE in 6m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/37b7aecca73443bc9e0733ee92d2938e : SUCCESS in 5m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c72729d68100429d9a09ce8f8d81b9fe : SUCCESS in 3m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d31be52e6da2434b9a755b80962f6b3c : SUCCESS in 5m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/64b9392aa2b64a0397066aba3ba9adce : SUCCESS in 11m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/090216ff4abb43eab1bad785e3267402 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7e689d996608471683894d07f338367a : SUCCESS in 6m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0c87ac080087467b9cd39001995663f6 : SUCCESS in 5m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/29f7d4f2eab34a65ba235718970580d8 : SUCCESS in 16m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b17d4530c6b5460c85c560021081d5da : SUCCESS in 7m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/149ba9510170490b99646349c13ecb35 : SUCCESS in 10m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b3f379cc23f1409cba879ed4bc4df95d : SUCCESS in 10m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7f1f5e46bd8747e0a07d62c83617083a : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d083478f9e2b4b0181d15b0bd977dd6e : SUCCESS in 6m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/06dfc1fb2a3d4acdb76508e62cfd063f : SUCCESS in 3m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/65a068c36249463fa4421dca6968f0ff : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/66d720e844de4ac38e66129262fe9dbf : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bc34f92a4dc24e678fb4796119ca8b58 : SUCCESS in 33m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ec0f616b3e04dcdb067041ca85cf2ea : SUCCESS in 9m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/34cf99b098c7451dba6a6ff65d4a74c4 : SUCCESS in 10m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/886e8096f2e64fd9b611d5d3a7005e75 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6fa1c98d4084872bb42ad1085555e3a : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2847bd7513a4b5788c253817159d7b0 : SUCCESS in 29m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5bca11079d5d4d288bc1f647932b7f0f : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7042fe7e779545a4bbf3628a8cb8cae7 : SUCCESS in 4m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f00f712c3e04486b83612efadaf4d788 : SUCCESS in 6m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fc65010ab0cb4cd19d046eb1be16c6a7 : SUCCESS in 7m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/104a4a8757a4403b806eb0d678ded584 : SUCCESS in 5m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d8c416ec2174261a2f3f5e2cbfe7f22 : SUCCESS in 58m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4458382c5514b37be21ec37d28fbbca : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":115},{"id":"e06ff2af1fb377d5a4c873089c2b0963e3205fa7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 21:31:09.000000000","message":"Uploaded patch set 116: Patch Set 115 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":116},{"id":"ab521b58fa68d94f529733caa40acb3c9a60ea36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 22:54:05.000000000","message":"Patch Set 116: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4b67c881a9f4520bd42ee82dde3c047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8224a51b4b7b40fcab84e3084ca95eaa : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8892c923f6374414948747c4d85e650a : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3a8652b0f19448e98cc5aa14692b53d : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8439da5d50ff4b6fb9b1e9b5236b8edf : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa8d7ee269fd4e7dab0d119c4735745d : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/810a79da36ce4196a59493910e47c053 : SUCCESS in 3m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/27b40f3997a0431499e5502e0188ddb4 : FAILURE in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/549f8b1390904e8f9e8dc8ca31b55385 : FAILURE in 3m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3bd6b5fafb5a4fca92a88d65aeee70cf : FAILURE in 2m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dbaa54be662e4714810d885f65a5f979 : FAILURE in 5m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2d2a4c7848444905b8d50c9e8c925624 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f2173f5e27904b018b07294dcf1812fb : SUCCESS in 4m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/59696ba87ffe43a4889aae30673ae69c : SUCCESS in 2m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/23067651eea549b09e4967e229a1bd63 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bca26ccaaf1d4873bdafec588b787dba : SUCCESS in 3m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/31b6a318e6e4404faf053e33489a28cd : SUCCESS in 3m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0fa167d9b4c143608c9449f3d6ccf19b : SUCCESS in 6m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f0e86fc8f55c4d2293209f540441a724 : SUCCESS in 15m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e88b64fe7b54b49b0da132e6e5dcf54 : SUCCESS in 4m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8f9e836b80c14774b817ae4c3f118ab7 : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ff14df2fa75d45b494572fc22c351398 : SUCCESS in 10m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bab52710a5e64e6e949234378697b071 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/10fe922d7642452386252ae94ab8293c : SUCCESS in 8m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/498028198db2429b91897349493bc7d3 : SUCCESS in 6m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d6d2ae2f6d7c45399de6b41f5e9ee515 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4122e5d1455b4278999fc14a5d17510a : SUCCESS in 11m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/99863c5888f94f3797de7ec9959b5eee : SUCCESS in 28m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b2be12c179c643999aad50e22209518e : SUCCESS in 5m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/81cfe678d9e244cc8cd5b57b0b38f148 : SUCCESS in 13m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/55a25dab87e54767846bba7f5a0e8016 : SUCCESS in 3m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05c33cbba7684b5885bb84415a87cbe7 : SUCCESS in 2m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60290034b3e74808a41de7f62340fe07 : SUCCESS in 23m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/31e858ad75204b38a85acde9dc307ac1 : SUCCESS in 5m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dd98b64d85f94b0c918d1d7f8ad4ac80 : SUCCESS in 3m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cebb7e2711f423db8d19082da645c85 : SUCCESS in 5m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/aa3c2c8151b04251b4cfd72666c7382a : SUCCESS in 18m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/feee1023405c427a8cc4c225d02ef266 : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ba397e937f648b0ba7a29ee3fd35c5d : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb09746facf14f5f94d2db09c3895d74 : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":116},{"id":"d017ec723082168e88d6a52c6fc0c755a625266b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 23:34:30.000000000","message":"Uploaded patch set 117: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":117},{"id":"951e463ed5045bd855aa7e133070c31f7d5420a6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:28:35.000000000","message":"Uploaded patch set 118: Patch Set 117 was rebased.","accounts_in_message":[],"_revision_number":118},{"id":"3293386ee60042ea7cf3cacb79db20b07108c5d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:49:02.000000000","message":"Uploaded patch set 119: New patch set was added with same tree, parent tree, and commit message as Patch Set 118.","accounts_in_message":[],"_revision_number":119},{"id":"b26caa9e74e5725dc5369517e7a5eab65149625d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 02:30:12.000000000","message":"Patch Set 119: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2624808dd0946b3910cff43353dc1a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d10faec197461ca5b75dfaed55ed91 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09c5bce70881409aa8d729aa6195117a : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d3c8731b0354f2d8acb97c277267752 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/091c512f947d4b79b72944cf7c7f9d24 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a707123509e43bd8796d547eec8ee8a : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a6120979dd84208ab9bfea752eff52b : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1dc966b442274c2da3cd992dd2221321 : SUCCESS in 3m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b02f92e17d5545798ffb84c28edb6b0e : SUCCESS in 4m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7e34c5d37e3440df84e1050a8eebc260 : SUCCESS in 5m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9625165e1ad94ceb8ff0a889558aaf3b : SUCCESS in 15m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c16bc68657144908a8fe1defdc7d1b91 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1a6b1843c4364d0688ad643e8ef06df5 : SUCCESS in 3m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/788e3b53458447dfa53eae6071495d6d : SUCCESS in 3m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1dfb22476c9144d4ad0bbaab10542df4 : SUCCESS in 7m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/55063f27d023488792454588a1c56920 : SUCCESS in 4m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/954e5f2015d249c7b20a0a0d325ab732 : SUCCESS in 5m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c0d43ea82a1d4c3fb51b340769ee13ad : SUCCESS in 6m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4a734abac5b44df9be5de888825e70a0 : SUCCESS in 8m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/269b4524382b48c5b63c49a5ae2f4a91 : SUCCESS in 9m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/810cbf36a2914b4784cda022f9598f1f : SUCCESS in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c8ade5a9c37c439caf5a09630a18587c : SUCCESS in 7m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7ae38c477ff649eaa518f70d7a4250ff : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b0762ced152940ffbe77d957b12b3237 : SUCCESS in 6m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0a4c19bc7c3149e3ab296a54cfe92e69 : SUCCESS in 5m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1cef0edeff544fe8a2ffb0bd0fc9e7ab : SUCCESS in 3m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7cf31d5d2ff74eddb890a5eea71c3491 : SUCCESS in 9m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4c4d4ab8fff24f98afc0221afca141dd : SUCCESS in 18m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b6d71e9fc7848c196c2daf22712a25b : SUCCESS in 9m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b2b2ceb3bb0744f2971524550333bb79 : SUCCESS in 14m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/deedcf9f187b47bda14bf268f35c9e25 : SUCCESS in 6m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b481f48e8884cb3aa6b1c57e2429a82 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9976ec2a8594058bc96324887c39f4a : SUCCESS in 27m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/aa9dd78289164ed9be968c517ae4467c : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ba7a4aba82994602bcfb1d39069011a0 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d54e5b70519447789a0a98d9fd60243 : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f073018db65c427caabc612ef4809336 : SUCCESS in 18m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bc0b715532e84ae0a3f6b3287c4a6871 : SUCCESS in 7m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44d892bddf12429393291ce1ac996d82 : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0aedfd6b48f461bba1bdc9c2ed11c79 : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":119},{"id":"d6213a90fa5606bbdd9c16f0dd8ff3db0226f226","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-02 05:07:08.000000000","message":"Abandoned\n\ndoesn\u0027t look like this is getting used anymore. feel free to restore if needed, though likely starting with a fresh change might be better","accounts_in_message":[],"_revision_number":119}],"current_revision_number":119,"current_revision":"f7822852a2e9b8a564d2890cdebde0eebac9c666","revisions":{"826c8e853bbbc8c63c2b065e663ea9381ed4ae3d":{"kind":"REWORK","_number":1,"created":"2024-02-04 10:29:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/1"}}},"commit":{"parents":[{"commit":"bd9a8b75fdc692d2e0102d64f775388ecef89172","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd9a8b75fdc692d2e0102d64f775388ecef89172"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:28:13.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/826c8e853bbbc8c63c2b065e663ea9381ed4ae3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/826c8e853bbbc8c63c2b065e663ea9381ed4ae3d"}]},"branch":"refs/heads/master"},"90935d6286263a532a1eee9e5aef8917227c4f0e":{"kind":"REWORK","_number":2,"created":"2024-02-04 14:02:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/2"}}},"commit":{"parents":[{"commit":"bd9a8b75fdc692d2e0102d64f775388ecef89172","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd9a8b75fdc692d2e0102d64f775388ecef89172"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:02:32.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90935d6286263a532a1eee9e5aef8917227c4f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90935d6286263a532a1eee9e5aef8917227c4f0e"}]},"branch":"refs/heads/master"},"81af07e51dd0f256017ce0115bea7af9f9ccd90b":{"kind":"REWORK","_number":3,"created":"2024-02-04 18:22:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/3"}}},"commit":{"parents":[{"commit":"90a3d1b203090d3bf98f6e2618c5ae38f10e6c41","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a3d1b203090d3bf98f6e2618c5ae38f10e6c41"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 18:22:15.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81af07e51dd0f256017ce0115bea7af9f9ccd90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81af07e51dd0f256017ce0115bea7af9f9ccd90b"}]},"branch":"refs/heads/master"},"0ab5a1df0eee69dd6abd053c3a160920b5485d6e":{"kind":"REWORK","_number":4,"created":"2024-02-05 12:48:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/4"}}},"commit":{"parents":[{"commit":"90a3d1b203090d3bf98f6e2618c5ae38f10e6c41","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a3d1b203090d3bf98f6e2618c5ae38f10e6c41"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-05 12:48:23.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- moto\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ab5a1df0eee69dd6abd053c3a160920b5485d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ab5a1df0eee69dd6abd053c3a160920b5485d6e"}]},"branch":"refs/heads/master"},"816edd8b3e1fe5d9800a78ca996b6737c277563b":{"kind":"REWORK","_number":5,"created":"2024-02-07 06:05:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/5"}}},"commit":{"parents":[{"commit":"90a3d1b203090d3bf98f6e2618c5ae38f10e6c41","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a3d1b203090d3bf98f6e2618c5ae38f10e6c41"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-07 06:05:04.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/816edd8b3e1fe5d9800a78ca996b6737c277563b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/816edd8b3e1fe5d9800a78ca996b6737c277563b"}]},"branch":"refs/heads/master"},"4acad5289076a6c85e52dedb87cc36d909e9a83c":{"kind":"REWORK","_number":6,"created":"2024-02-16 20:43:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/6"}}},"commit":{"parents":[{"commit":"605bdc33e508567bb9c838ef755743036edb16b7","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/605bdc33e508567bb9c838ef755743036edb16b7"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-16 20:43:16.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4acad5289076a6c85e52dedb87cc36d909e9a83c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4acad5289076a6c85e52dedb87cc36d909e9a83c"}]},"branch":"refs/heads/master"},"72d33a388d8856ea67218a843910b01ee7f5dbf0":{"kind":"REWORK","_number":7,"created":"2024-02-17 09:59:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/7"}}},"commit":{"parents":[{"commit":"e493017dcccf6c1a28666d1cea07b40bc16c1342","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-17 09:59:07.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d33a388d8856ea67218a843910b01ee7f5dbf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d33a388d8856ea67218a843910b01ee7f5dbf0"}]},"branch":"refs/heads/master"},"bac0b08390ba2d84aff6ae388c8728a2471d748a":{"kind":"REWORK","_number":8,"created":"2024-02-17 13:28:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/8"}}},"commit":{"parents":[{"commit":"e493017dcccf6c1a28666d1cea07b40bc16c1342","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-17 13:27:35.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bac0b08390ba2d84aff6ae388c8728a2471d748a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bac0b08390ba2d84aff6ae388c8728a2471d748a"}]},"branch":"refs/heads/master"},"fd2efdc8c36701d40846b92a1a86ea0234fe6561":{"kind":"REWORK","_number":9,"created":"2024-02-17 13:41:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/9"}}},"commit":{"parents":[{"commit":"e493017dcccf6c1a28666d1cea07b40bc16c1342","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-17 13:41:01.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd2efdc8c36701d40846b92a1a86ea0234fe6561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd2efdc8c36701d40846b92a1a86ea0234fe6561"}]},"branch":"refs/heads/master"},"78919460590f10afcd145c3269e979a6b5249574":{"kind":"REWORK","_number":10,"created":"2024-02-18 08:52:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/10"}}},"commit":{"parents":[{"commit":"e493017dcccf6c1a28666d1cea07b40bc16c1342","subject":"Block scrypt\u003d\u003d0.8.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-18 08:51:47.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78919460590f10afcd145c3269e979a6b5249574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78919460590f10afcd145c3269e979a6b5249574"}]},"branch":"refs/heads/master"},"a2ded5fa62830b4a15baf5d4f39fdba3d7eb712a":{"kind":"REWORK","_number":11,"created":"2024-02-24 11:59:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/11"}}},"commit":{"parents":[{"commit":"46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-24 11:59:16.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2ded5fa62830b4a15baf5d4f39fdba3d7eb712a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2ded5fa62830b4a15baf5d4f39fdba3d7eb712a"}]},"branch":"refs/heads/master"},"b447a33a5e6b7ccf4d0ecf1a6c04696481c561c3":{"kind":"REWORK","_number":12,"created":"2024-02-24 14:56:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/12"}}},"commit":{"parents":[{"commit":"46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-24 14:53:17.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n\nPin versions for openstack releases that break stuff:\n\n- oslo.messaging[0]\n- oslo.db [1]\n\n[0] https://review.opendev.org/c/openstack/requirements/+/909906\n[1] https://review.opendev.org/c/openstack/requirements/+/909930\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b447a33a5e6b7ccf4d0ecf1a6c04696481c561c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b447a33a5e6b7ccf4d0ecf1a6c04696481c561c3"}]},"branch":"refs/heads/master"},"cc5a24fdf43f2ec0d2928e58b7741cb6bb7f1f3a":{"kind":"REWORK","_number":13,"created":"2024-02-26 07:07:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/13"}}},"commit":{"parents":[{"commit":"46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-26 07:07:03.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n\nPin versions for openstack releases that break stuff:\n\n- oslo.messaging[0]\n- oslo.db [1]\n\n[0] https://review.opendev.org/c/openstack/requirements/+/909906\n[1] https://review.opendev.org/c/openstack/requirements/+/909930\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc5a24fdf43f2ec0d2928e58b7741cb6bb7f1f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc5a24fdf43f2ec0d2928e58b7741cb6bb7f1f3a"}]},"branch":"refs/heads/master"},"2bce16fa220f904b07a399533db09a37f1cfabc4":{"kind":"REWORK","_number":14,"created":"2024-03-25 14:52:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/14"}}},"commit":{"parents":[{"commit":"91ed29ca5c0cb552d7da83a25cc3699aea7ee75d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91ed29ca5c0cb552d7da83a25cc3699aea7ee75d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-23 10:31:46.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n- requests-oauthlib\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bce16fa220f904b07a399533db09a37f1cfabc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bce16fa220f904b07a399533db09a37f1cfabc4"}]},"branch":"refs/heads/master"},"13bca2d694f2253846bf424d0b454dc8b589fedc":{"kind":"REWORK","_number":15,"created":"2024-04-05 08:55:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/15"}}},"commit":{"parents":[{"commit":"fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-05 08:55:32.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n- requests-oauthlib\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13bca2d694f2253846bf424d0b454dc8b589fedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13bca2d694f2253846bf424d0b454dc8b589fedc"}]},"branch":"refs/heads/master"},"d99afce8820a0607c854db205abaa9fd2fa65e86":{"kind":"REWORK","_number":16,"created":"2024-04-05 11:52:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/16"}}},"commit":{"parents":[{"commit":"fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-05 11:51:47.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d99afce8820a0607c854db205abaa9fd2fa65e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d99afce8820a0607c854db205abaa9fd2fa65e86"}]},"branch":"refs/heads/master"},"fbcb7e9f935ccb1cebc293ca68d81ffade9a4257":{"kind":"REWORK","_number":17,"created":"2024-04-09 15:48:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/17"}}},"commit":{"parents":[{"commit":"3340d34eda88f5f32e9cae215fb500821ce8ccc6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3340d34eda88f5f32e9cae215fb500821ce8ccc6"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-09 15:47:57.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- SQLAlchemy\n- pillow\n- alembic\n- jsonschema\n- netaddr\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbcb7e9f935ccb1cebc293ca68d81ffade9a4257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbcb7e9f935ccb1cebc293ca68d81ffade9a4257"}]},"branch":"refs/heads/master"},"9427e6cf98a7dd09bd9d5208883f35eb53be29c0":{"kind":"REWORK","_number":18,"created":"2024-04-20 08:54:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/18"}}},"commit":{"parents":[{"commit":"32f354df8ce0db3a36e2a1ad0dc505cb10297b54","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32f354df8ce0db3a36e2a1ad0dc505cb10297b54"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-20 08:54:14.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- Sphinx (\u003c7.3)\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9427e6cf98a7dd09bd9d5208883f35eb53be29c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9427e6cf98a7dd09bd9d5208883f35eb53be29c0"}]},"branch":"refs/heads/master"},"a1e8ca5fb8e42761d5ff15423f4bd088b3179067":{"kind":"REWORK","_number":19,"created":"2024-04-20 12:31:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/19"}}},"commit":{"parents":[{"commit":"32f354df8ce0db3a36e2a1ad0dc505cb10297b54","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32f354df8ce0db3a36e2a1ad0dc505cb10297b54"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-20 12:30:59.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- Sphinx (\u003c7.3)\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1e8ca5fb8e42761d5ff15423f4bd088b3179067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1e8ca5fb8e42761d5ff15423f4bd088b3179067"}]},"branch":"refs/heads/master"},"3176334de5d6c04219b7172ef9d06d4ab9c22d7a":{"kind":"REWORK","_number":20,"created":"2024-05-15 12:23:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/20"}}},"commit":{"parents":[{"commit":"25ea34a8eee299511a30b78c6ad7bf6596469afc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25ea34a8eee299511a30b78c6ad7bf6596469afc"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-15 12:21:26.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3176334de5d6c04219b7172ef9d06d4ab9c22d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3176334de5d6c04219b7172ef9d06d4ab9c22d7a"}]},"branch":"refs/heads/master"},"96d9ffb908eb4e06618b3d10c66478d728633e8b":{"kind":"REWORK","_number":21,"created":"2024-05-23 08:47:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/21"}}},"commit":{"parents":[{"commit":"1da3d304c0a2b31bb26a459d4403469eb6901e84","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1da3d304c0a2b31bb26a459d4403469eb6901e84"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-23 08:46:19.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d9ffb908eb4e06618b3d10c66478d728633e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d9ffb908eb4e06618b3d10c66478d728633e8b"}]},"branch":"refs/heads/master"},"f77715137e81a573724e42559cb755912bf8736a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-05-23 09:39:11.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/22"}}},"commit":{"parents":[{"commit":"128b67fa08fca35922cd3199c62cf0cd321c72ab","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/128b67fa08fca35922cd3199c62cf0cd321c72ab"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-23 09:38:57.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f77715137e81a573724e42559cb755912bf8736a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f77715137e81a573724e42559cb755912bf8736a"}]},"branch":"refs/heads/master"},"81e1bac0c14fc5f838b215c02190cc801356d0f7":{"kind":"REWORK","_number":23,"created":"2024-05-26 20:02:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/23"}}},"commit":{"parents":[{"commit":"2c389e69716f50c78a8dfa02161b9fda1589ee82","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c389e69716f50c78a8dfa02161b9fda1589ee82"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-26 19:59:14.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81e1bac0c14fc5f838b215c02190cc801356d0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81e1bac0c14fc5f838b215c02190cc801356d0f7"}]},"branch":"refs/heads/master"},"3919e5d9e3cb98a42dff20809026bf12ac11047f":{"kind":"REWORK","_number":24,"created":"2024-06-08 15:33:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/24"}}},"commit":{"parents":[{"commit":"15f5dc3483530da0c669d1a755b2991e92249b3e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15f5dc3483530da0c669d1a755b2991e92249b3e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-08 15:32:11.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3919e5d9e3cb98a42dff20809026bf12ac11047f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3919e5d9e3cb98a42dff20809026bf12ac11047f"}]},"branch":"refs/heads/master"},"416d37d7d1065d89abe3f5fd4c9289289ed6f18b":{"kind":"REWORK","_number":25,"created":"2024-06-10 10:59:08.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/25"}}},"commit":{"parents":[{"commit":"92d53f119c5cc00cb6a547f15114218c48cb7160","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92d53f119c5cc00cb6a547f15114218c48cb7160"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-10 10:58:51.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pin for newly added requirement:\n\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/416d37d7d1065d89abe3f5fd4c9289289ed6f18b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/416d37d7d1065d89abe3f5fd4c9289289ed6f18b"}]},"branch":"refs/heads/master"},"d6c48e667c52637adc82fba5c7e67761f29a1cb2":{"kind":"REWORK","_number":26,"created":"2024-06-10 13:23:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/26"}}},"commit":{"parents":[{"commit":"92d53f119c5cc00cb6a547f15114218c48cb7160","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92d53f119c5cc00cb6a547f15114218c48cb7160"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-10 13:23:10.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6c48e667c52637adc82fba5c7e67761f29a1cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6c48e667c52637adc82fba5c7e67761f29a1cb2"}]},"branch":"refs/heads/master"},"317337b6dbe2eabf81894e2079cb1b983b6357f4":{"kind":"REWORK","_number":27,"created":"2024-06-22 21:42:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/27"}}},"commit":{"parents":[{"commit":"79a70ecc812162164a34fbfe89a25822887eb890","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79a70ecc812162164a34fbfe89a25822887eb890"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-22 21:41:29.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317337b6dbe2eabf81894e2079cb1b983b6357f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317337b6dbe2eabf81894e2079cb1b983b6357f4"}]},"branch":"refs/heads/master"},"171d7bd7d399177277519cc2f7954063434cf857":{"kind":"REWORK","_number":28,"created":"2024-07-06 17:33:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/28"}}},"commit":{"parents":[{"commit":"cc8317b9aaca7dbeecd5c6d499b8d28e6e68d07f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8317b9aaca7dbeecd5c6d499b8d28e6e68d07f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-06 17:33:27.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/171d7bd7d399177277519cc2f7954063434cf857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/171d7bd7d399177277519cc2f7954063434cf857"}]},"branch":"refs/heads/master"},"e4adeda9754d3fbe811ecb8a7d9972d1ad3235cf":{"kind":"REWORK","_number":29,"created":"2024-07-06 17:56:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/29"}}},"commit":{"parents":[{"commit":"cc8317b9aaca7dbeecd5c6d499b8d28e6e68d07f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8317b9aaca7dbeecd5c6d499b8d28e6e68d07f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-06 17:55:00.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4adeda9754d3fbe811ecb8a7d9972d1ad3235cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4adeda9754d3fbe811ecb8a7d9972d1ad3235cf"}]},"branch":"refs/heads/master"},"3421a1eed5a3b3bac749631092c605cfd9709433":{"kind":"REWORK","_number":30,"created":"2024-07-07 11:10:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/30"}}},"commit":{"parents":[{"commit":"07de9f9ff6cf0851c405d072eba4be1f3da7a76e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07de9f9ff6cf0851c405d072eba4be1f3da7a76e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-07 11:10:10.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3421a1eed5a3b3bac749631092c605cfd9709433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3421a1eed5a3b3bac749631092c605cfd9709433"}]},"branch":"refs/heads/master"},"6255d7805e23898a607e6e2f763741ccbdee7342":{"kind":"REWORK","_number":31,"created":"2024-07-13 11:30:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/31"}}},"commit":{"parents":[{"commit":"3b2c8ec65999c4ed6836e5b5ce187d29b8dfeb21","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b2c8ec65999c4ed6836e5b5ce187d29b8dfeb21"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-13 11:30:03.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6255d7805e23898a607e6e2f763741ccbdee7342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6255d7805e23898a607e6e2f763741ccbdee7342"}]},"branch":"refs/heads/master"},"13ab0e3711c5a36483994e812a3a763df324be8e":{"kind":"REWORK","_number":32,"created":"2024-07-20 09:51:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/32"}}},"commit":{"parents":[{"commit":"165863af9d99d3493427806803f4f561de321cc6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/165863af9d99d3493427806803f4f561de321cc6"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-20 09:51:07.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13ab0e3711c5a36483994e812a3a763df324be8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13ab0e3711c5a36483994e812a3a763df324be8e"}]},"branch":"refs/heads/master"},"7a19aad0878a97d1e088e17635895a47541d3608":{"kind":"REWORK","_number":33,"created":"2024-07-27 09:47:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/33"}}},"commit":{"parents":[{"commit":"cf005d800feebd797ec92ea3a7cec688e8944a47","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf005d800feebd797ec92ea3a7cec688e8944a47"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:44:25.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a19aad0878a97d1e088e17635895a47541d3608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a19aad0878a97d1e088e17635895a47541d3608"}]},"branch":"refs/heads/master"},"e17c1dc839a2416e766ddcb1468f8067ea033440":{"kind":"REWORK","_number":34,"created":"2024-07-27 19:38:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/34"}}},"commit":{"parents":[{"commit":"cf005d800feebd797ec92ea3a7cec688e8944a47","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf005d800feebd797ec92ea3a7cec688e8944a47"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 19:36:13.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17c1dc839a2416e766ddcb1468f8067ea033440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17c1dc839a2416e766ddcb1468f8067ea033440"}]},"branch":"refs/heads/master"},"4c868d9ea56d6a7dd8c49f8c8efd1b48e1fbcc8f":{"kind":"REWORK","_number":35,"created":"2024-08-05 11:41:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/35"}}},"commit":{"parents":[{"commit":"a3aeccfdcd1cf55311f3f9fae99f63dd4acfb34e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3aeccfdcd1cf55311f3f9fae99f63dd4acfb34e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-05 11:40:44.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c868d9ea56d6a7dd8c49f8c8efd1b48e1fbcc8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c868d9ea56d6a7dd8c49f8c8efd1b48e1fbcc8f"}]},"branch":"refs/heads/master"},"cf9af36232f63dfa435e5f9145eb755d8739403e":{"kind":"REWORK","_number":36,"created":"2024-08-27 12:31:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/36"}}},"commit":{"parents":[{"commit":"6b47104ea8c0666b9894f54454d512085f92723b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b47104ea8c0666b9894f54454d512085f92723b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-27 12:30:21.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf9af36232f63dfa435e5f9145eb755d8739403e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf9af36232f63dfa435e5f9145eb755d8739403e"}]},"branch":"refs/heads/master"},"d942917053d6c846541c63fc016d1e7bea386e04":{"kind":"REWORK","_number":37,"created":"2024-08-27 15:35:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/37"}}},"commit":{"parents":[{"commit":"6b47104ea8c0666b9894f54454d512085f92723b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b47104ea8c0666b9894f54454d512085f92723b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-27 15:33:36.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d942917053d6c846541c63fc016d1e7bea386e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d942917053d6c846541c63fc016d1e7bea386e04"}]},"branch":"refs/heads/master"},"e9fcde7e42e8c402e9f35224f17a353c04684d29":{"kind":"REWORK","_number":38,"created":"2024-08-27 17:42:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/38"}}},"commit":{"parents":[{"commit":"6b47104ea8c0666b9894f54454d512085f92723b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b47104ea8c0666b9894f54454d512085f92723b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-27 17:41:48.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9fcde7e42e8c402e9f35224f17a353c04684d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9fcde7e42e8c402e9f35224f17a353c04684d29"}]},"branch":"refs/heads/master"},"a8f848e9f19921fe8fff8e1faa93be6d1603dee9":{"kind":"REWORK","_number":39,"created":"2024-08-27 18:28:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/39"}}},"commit":{"parents":[{"commit":"6b47104ea8c0666b9894f54454d512085f92723b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b47104ea8c0666b9894f54454d512085f92723b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-27 18:27:46.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8f848e9f19921fe8fff8e1faa93be6d1603dee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8f848e9f19921fe8fff8e1faa93be6d1603dee9"}]},"branch":"refs/heads/master"},"d80f774ba45acf75005bcd820b737abd372027d9":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2024-08-28 08:29:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/40"}}},"commit":{"parents":[{"commit":"8d576ae6e58cfb092afed857593b61538ac2a812","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d576ae6e58cfb092afed857593b61538ac2a812"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-28 08:26:01.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d80f774ba45acf75005bcd820b737abd372027d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d80f774ba45acf75005bcd820b737abd372027d9"}]},"branch":"refs/heads/master"},"13f9e84c4f72652a5a28aa166a794e28e808011c":{"kind":"REWORK","_number":41,"created":"2024-09-27 11:21:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/41"}}},"commit":{"parents":[{"commit":"417ef72285dfeac6cb283a045887c609c5d66d93","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/417ef72285dfeac6cb283a045887c609c5d66d93"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-09-27 11:20:53.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f9e84c4f72652a5a28aa166a794e28e808011c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13f9e84c4f72652a5a28aa166a794e28e808011c"}]},"branch":"refs/heads/master"},"7d331b0f14d4755b68fe0f29b7e14237ddf2d39f":{"kind":"REWORK","_number":42,"created":"2024-09-27 12:18:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/42"}}},"commit":{"parents":[{"commit":"417ef72285dfeac6cb283a045887c609c5d66d93","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/417ef72285dfeac6cb283a045887c609c5d66d93"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-09-27 12:18:59.000000000","tz":0},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- python-openstackclient\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d331b0f14d4755b68fe0f29b7e14237ddf2d39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d331b0f14d4755b68fe0f29b7e14237ddf2d39f"}]},"branch":"refs/heads/master"},"0d85bdc960272c718c91f47401be834ce9645bc5":{"kind":"REWORK","_number":43,"created":"2024-10-23 16:00:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/43"}}},"commit":{"parents":[{"commit":"70e1fd6bde17e3a968eb222dcd2b552028d2548d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70e1fd6bde17e3a968eb222dcd2b552028d2548d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-23 15:59:56.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d85bdc960272c718c91f47401be834ce9645bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d85bdc960272c718c91f47401be834ce9645bc5"}]},"branch":"refs/heads/master"},"1750709dc60e17208ff7f084e9d6730931757c57":{"kind":"REWORK","_number":44,"created":"2024-10-23 17:03:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/44"}}},"commit":{"parents":[{"commit":"70e1fd6bde17e3a968eb222dcd2b552028d2548d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70e1fd6bde17e3a968eb222dcd2b552028d2548d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-23 17:03:16.000000000","tz":0},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n- yarl\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1750709dc60e17208ff7f084e9d6730931757c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1750709dc60e17208ff7f084e9d6730931757c57"}]},"branch":"refs/heads/master"},"2c59d34af78046bd5bdd332b14829a527f3c9a2f":{"kind":"REWORK","_number":45,"created":"2024-10-23 17:16:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/45"}}},"commit":{"parents":[{"commit":"70e1fd6bde17e3a968eb222dcd2b552028d2548d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70e1fd6bde17e3a968eb222dcd2b552028d2548d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-23 17:15:29.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- importlib_resources\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n- yarl\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c59d34af78046bd5bdd332b14829a527f3c9a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c59d34af78046bd5bdd332b14829a527f3c9a2f"}]},"branch":"refs/heads/master"},"58c54a8835346343e0486926741950b33834c613":{"kind":"REWORK","_number":46,"created":"2024-10-24 14:53:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/46"}}},"commit":{"parents":[{"commit":"70e1fd6bde17e3a968eb222dcd2b552028d2548d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70e1fd6bde17e3a968eb222dcd2b552028d2548d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-24 14:53:37.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- importlib_resources\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n- yarl\n\nPin versions for packages that are not ready for updating:\n\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n- eventlet\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58c54a8835346343e0486926741950b33834c613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58c54a8835346343e0486926741950b33834c613"}]},"branch":"refs/heads/master"},"7c7cc35593b15afb38f31c686eb71b8a5c85c425":{"kind":"REWORK","_number":47,"created":"2024-10-24 18:06:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/47"}}},"commit":{"parents":[{"commit":"225d92989ccb76d977f4f2c7bc94c628c29a84ff","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/225d92989ccb76d977f4f2c7bc94c628c29a84ff"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-24 18:04:55.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- waiting\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n- eventlet\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c7cc35593b15afb38f31c686eb71b8a5c85c425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c7cc35593b15afb38f31c686eb71b8a5c85c425"}]},"branch":"refs/heads/master"},"b64ebc7ff5d02eaf8771b507473c36f4a990d5a1":{"kind":"REWORK","_number":48,"created":"2024-10-31 06:19:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/48"}}},"commit":{"parents":[{"commit":"01e6de6a9a9bbd8601b4249bec37ce449096ed00","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01e6de6a9a9bbd8601b4249bec37ce449096ed00"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-31 06:19:14.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- waiting\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n- eventlet\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b64ebc7ff5d02eaf8771b507473c36f4a990d5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b64ebc7ff5d02eaf8771b507473c36f4a990d5a1"}]},"branch":"refs/heads/master"},"0d717348b4e9c6d641c01cbc91f664c1d5519511":{"kind":"REWORK","_number":49,"created":"2024-11-05 09:43:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/49"}}},"commit":{"parents":[{"commit":"ad0ea335a5c9c1d4d701c8894d9ce42030b4fadb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad0ea335a5c9c1d4d701c8894d9ce42030b4fadb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-05 09:42:56.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- waiting\n- croniter\n- horizon\n- netaddr\n- cryptography\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n- olso.policy\n- eventlet\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d717348b4e9c6d641c01cbc91f664c1d5519511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d717348b4e9c6d641c01cbc91f664c1d5519511"}]},"branch":"refs/heads/master"},"5182e2afbcf8308c757ec498d7f70d849ddc7a7d":{"kind":"REWORK","_number":50,"created":"2024-11-11 13:58:54.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/50"}}},"commit":{"parents":[{"commit":"f2b62990bc3bdefdacf40e90ea5b6ba1abe7de0e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2b62990bc3bdefdacf40e90ea5b6ba1abe7de0e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-11 13:58:10.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pillow\n- requests-oauthlib\n- pysnmp-lextudio\n- tenacity\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5182e2afbcf8308c757ec498d7f70d849ddc7a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5182e2afbcf8308c757ec498d7f70d849ddc7a7d"}]},"branch":"refs/heads/master"},"118ea82ec199b5c6675c47d2ac67c8403a924f2e":{"kind":"REWORK","_number":51,"created":"2024-11-16 14:20:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/51"}}},"commit":{"parents":[{"commit":"fb62b7c39bff32f47e273579900bfa935a40c87e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb62b7c39bff32f47e273579900bfa935a40c87e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-16 14:18:32.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pillow\n- requests-oauthlib\n- pysnmp-lextudio\n- tenacity\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118ea82ec199b5c6675c47d2ac67c8403a924f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118ea82ec199b5c6675c47d2ac67c8403a924f2e"}]},"branch":"refs/heads/master"},"3b90b3b5947e64c10c127928e00cc7d2a8e21bca":{"kind":"REWORK","_number":52,"created":"2024-11-19 20:13:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/52"}}},"commit":{"parents":[{"commit":"822be7f4f872298a68823f0e88fc7d98051ce034","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/822be7f4f872298a68823f0e88fc7d98051ce034"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-19 20:11:04.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- tenacity\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b90b3b5947e64c10c127928e00cc7d2a8e21bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b90b3b5947e64c10c127928e00cc7d2a8e21bca"}]},"branch":"refs/heads/master"},"70f3f019eb7fbf1fda431a455f10fa586e50919b":{"kind":"REWORK","_number":53,"created":"2024-11-20 09:58:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/53"}}},"commit":{"parents":[{"commit":"822be7f4f872298a68823f0e88fc7d98051ce034","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/822be7f4f872298a68823f0e88fc7d98051ce034"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-20 09:58:19.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- tenacity\n- jsonschema\n- oslo.utils\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70f3f019eb7fbf1fda431a455f10fa586e50919b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70f3f019eb7fbf1fda431a455f10fa586e50919b"}]},"branch":"refs/heads/master"},"441eda2a52cae484f7bfc5a0b3752523da855cb3":{"kind":"REWORK","_number":54,"created":"2024-11-23 10:41:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/54"}}},"commit":{"parents":[{"commit":"9dad7b3b204417aef0c6475aaa99dbf2aee25acb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dad7b3b204417aef0c6475aaa99dbf2aee25acb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 10:41:31.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/441eda2a52cae484f7bfc5a0b3752523da855cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/441eda2a52cae484f7bfc5a0b3752523da855cb3"}]},"branch":"refs/heads/master"},"258671c850ae95de7f2cbdb438631c68d088d996":{"kind":"REWORK","_number":55,"created":"2024-11-23 11:24:03.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/55"}}},"commit":{"parents":[{"commit":"9dad7b3b204417aef0c6475aaa99dbf2aee25acb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dad7b3b204417aef0c6475aaa99dbf2aee25acb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 11:23:11.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/258671c850ae95de7f2cbdb438631c68d088d996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/258671c850ae95de7f2cbdb438631c68d088d996"}]},"branch":"refs/heads/master"},"9732e595b1a4edcda54f433215ebfa6362504fac":{"kind":"REWORK","_number":56,"created":"2024-11-23 17:50:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/56"}}},"commit":{"parents":[{"commit":"9dad7b3b204417aef0c6475aaa99dbf2aee25acb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dad7b3b204417aef0c6475aaa99dbf2aee25acb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 17:41:02.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9732e595b1a4edcda54f433215ebfa6362504fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9732e595b1a4edcda54f433215ebfa6362504fac"}]},"branch":"refs/heads/master"},"7e940171e6e091cc6ff197f75ad5e9bfea3cc2c2":{"kind":"REWORK","_number":57,"created":"2024-11-30 10:28:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/57"}}},"commit":{"parents":[{"commit":"01063b040c8cc9f4778ce974125846ac158338f2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01063b040c8cc9f4778ce974125846ac158338f2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-30 10:26:56.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e940171e6e091cc6ff197f75ad5e9bfea3cc2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e940171e6e091cc6ff197f75ad5e9bfea3cc2c2"}]},"branch":"refs/heads/master"},"335667da05885e0e96bc679269bef5b62be6c578":{"kind":"REWORK","_number":58,"created":"2024-12-07 08:10:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/58"}}},"commit":{"parents":[{"commit":"29d3b052946316f87d747d1229b42fd36e0781be","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29d3b052946316f87d747d1229b42fd36e0781be"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-07 08:10:17.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/335667da05885e0e96bc679269bef5b62be6c578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/335667da05885e0e96bc679269bef5b62be6c578"}]},"branch":"refs/heads/master"},"d9ca67a5608745b764c345fe18c5792316386fa8":{"kind":"REWORK","_number":59,"created":"2024-12-14 09:45:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/59"}}},"commit":{"parents":[{"commit":"1f8dab92c5382629b1717f9d71d60214f6b39d0b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f8dab92c5382629b1717f9d71d60214f6b39d0b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-14 09:43:42.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9ca67a5608745b764c345fe18c5792316386fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9ca67a5608745b764c345fe18c5792316386fa8"}]},"branch":"refs/heads/master"},"635cb7524d8728f8e0c343a06edbcd82dc77014c":{"kind":"REWORK","_number":60,"created":"2024-12-16 06:29:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/60"}}},"commit":{"parents":[{"commit":"1f8dab92c5382629b1717f9d71d60214f6b39d0b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f8dab92c5382629b1717f9d71d60214f6b39d0b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-16 06:27:08.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for:\n\n- daiquiri\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/635cb7524d8728f8e0c343a06edbcd82dc77014c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/635cb7524d8728f8e0c343a06edbcd82dc77014c"}]},"branch":"refs/heads/master"},"3fe24b4598c5ecb7ab0e8b23306a2b84252ed9ce":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2024-12-21 09:06:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/61"}}},"commit":{"parents":[{"commit":"850a93993f01fc1c01054a20631b0a32a7e002ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/850a93993f01fc1c01054a20631b0a32a7e002ea"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-21 09:05:25.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for:\n\n- daiquiri\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe24b4598c5ecb7ab0e8b23306a2b84252ed9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe24b4598c5ecb7ab0e8b23306a2b84252ed9ce"}]},"branch":"refs/heads/master"},"c9040ec618bcb093375665f4ebc14e36731b7fb4":{"kind":"REWORK","_number":62,"created":"2025-01-06 13:18:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/62"}}},"commit":{"parents":[{"commit":"bb9f9b68ce7190b3e11a91392f8b86c084094123","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb9f9b68ce7190b3e11a91392f8b86c084094123"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-06 13:17:02.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for:\n\n- daiquiri\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9040ec618bcb093375665f4ebc14e36731b7fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9040ec618bcb093375665f4ebc14e36731b7fb4"}]},"branch":"refs/heads/master"},"49139646a3cea3396de45746482db2e5fd2efb5f":{"kind":"REWORK","_number":63,"created":"2025-01-06 16:26:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/63"}}},"commit":{"parents":[{"commit":"bb9f9b68ce7190b3e11a91392f8b86c084094123","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb9f9b68ce7190b3e11a91392f8b86c084094123"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-06 16:24:37.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for:\n\n- daiquiri\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- infoblox-client\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49139646a3cea3396de45746482db2e5fd2efb5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49139646a3cea3396de45746482db2e5fd2efb5f"}]},"branch":"refs/heads/master"},"dddb013bb104f68cd4a5c594a7f3e5d80d6b9588":{"kind":"REWORK","_number":64,"created":"2025-01-11 10:36:36.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/64"}}},"commit":{"parents":[{"commit":"c186ec8383f07b9727bdeae5440a817627eaf059","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c186ec8383f07b9727bdeae5440a817627eaf059"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-11 10:35:38.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n- daiquiri\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- infoblox-client\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dddb013bb104f68cd4a5c594a7f3e5d80d6b9588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dddb013bb104f68cd4a5c594a7f3e5d80d6b9588"}]},"branch":"refs/heads/master"},"37d9251895fe70b0fbe0e1b994d176ce53342348":{"kind":"REWORK","_number":65,"created":"2025-01-11 10:45:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/65"}}},"commit":{"parents":[{"commit":"c186ec8383f07b9727bdeae5440a817627eaf059","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c186ec8383f07b9727bdeae5440a817627eaf059"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-11 10:41:23.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- infoblox-client\n- pysnmp-lextudio\n- jsonschema\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37d9251895fe70b0fbe0e1b994d176ce53342348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37d9251895fe70b0fbe0e1b994d176ce53342348"}]},"branch":"refs/heads/master"},"f0b2f78d93a69ce129de81f5838b0f8df03117bb":{"kind":"REWORK","_number":66,"created":"2025-01-12 12:40:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/66"}}},"commit":{"parents":[{"commit":"d7a9271bc6d4ffa82ee766b045ff6c8d25d2d8c0","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7a9271bc6d4ffa82ee766b045ff6c8d25d2d8c0"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-12 12:39:41.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- eventlet\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b2f78d93a69ce129de81f5838b0f8df03117bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b2f78d93a69ce129de81f5838b0f8df03117bb"}]},"branch":"refs/heads/master"},"493eeb28f3def2420b7b4bf73332baf500a5971a":{"kind":"REWORK","_number":67,"created":"2025-01-20 07:59:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/67"}}},"commit":{"parents":[{"commit":"4af102224975d05a854c4816068182d7a4241698","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4af102224975d05a854c4816068182d7a4241698"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-20 07:58:06.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/493eeb28f3def2420b7b4bf73332baf500a5971a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/493eeb28f3def2420b7b4bf73332baf500a5971a"}]},"branch":"refs/heads/master"},"26f7e188cbb58c48d789b1c02483f9ad285c1569":{"kind":"REWORK","_number":68,"created":"2025-01-26 14:15:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/68"}}},"commit":{"parents":[{"commit":"3ec6ad96acd746cd77679f2e8d668c93394a7fc6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ec6ad96acd746cd77679f2e8d668c93394a7fc6"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-26 14:13:31.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26f7e188cbb58c48d789b1c02483f9ad285c1569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26f7e188cbb58c48d789b1c02483f9ad285c1569"}]},"branch":"refs/heads/master"},"bd4d73b48ea87f3c270709232665fd7baea9f4a4":{"kind":"REWORK","_number":69,"created":"2025-02-01 20:28:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/69"}}},"commit":{"parents":[{"commit":"79dec68df3a3b7df0847419e3dbcd4e9a9059cef","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79dec68df3a3b7df0847419e3dbcd4e9a9059cef"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-01 20:27:47.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd4d73b48ea87f3c270709232665fd7baea9f4a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd4d73b48ea87f3c270709232665fd7baea9f4a4"}]},"branch":"refs/heads/master"},"bf2ccc42063e74665b0d81134fd36a32b2ef0146":{"kind":"REWORK","_number":70,"created":"2025-02-07 08:59:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/70"}}},"commit":{"parents":[{"commit":"79dec68df3a3b7df0847419e3dbcd4e9a9059cef","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79dec68df3a3b7df0847419e3dbcd4e9a9059cef"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-07 08:58:36.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf2ccc42063e74665b0d81134fd36a32b2ef0146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf2ccc42063e74665b0d81134fd36a32b2ef0146"}]},"branch":"refs/heads/master"},"89fedb4a756bedfec40c053416bdc04256562d34":{"kind":"REWORK","_number":71,"created":"2025-02-15 12:46:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/71"}}},"commit":{"parents":[{"commit":"6cd382e413e2498adf182aecd2b2db12b6cbcbb1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cd382e413e2498adf182aecd2b2db12b6cbcbb1"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-15 12:46:35.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89fedb4a756bedfec40c053416bdc04256562d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89fedb4a756bedfec40c053416bdc04256562d34"}]},"branch":"refs/heads/master"},"2565a047022d9def9cf66bbde17ebf3589fa8b88":{"kind":"REWORK","_number":72,"created":"2025-02-22 09:45:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/72"}}},"commit":{"parents":[{"commit":"1742e659fdcfd8a050a0f92580daa9d3650aa542","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1742e659fdcfd8a050a0f92580daa9d3650aa542"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-22 09:44:46.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pytest\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2565a047022d9def9cf66bbde17ebf3589fa8b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2565a047022d9def9cf66bbde17ebf3589fa8b88"}]},"branch":"refs/heads/master"},"4b34158f1e690271949070dd5764fbf506378fb2":{"kind":"REWORK","_number":73,"created":"2025-03-03 04:47:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/73"}}},"commit":{"parents":[{"commit":"35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-03 04:46:01.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b34158f1e690271949070dd5764fbf506378fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b34158f1e690271949070dd5764fbf506378fb2"}]},"branch":"refs/heads/master"},"13353976a1516287c3aa31c54e78432c9869e6ab":{"kind":"REWORK","_number":74,"created":"2025-03-03 05:39:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/74"}}},"commit":{"parents":[{"commit":"35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-03 05:38:42.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n- openstacksdk\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13353976a1516287c3aa31c54e78432c9869e6ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13353976a1516287c3aa31c54e78432c9869e6ab"}]},"branch":"refs/heads/master"},"764fce9f8d1b822818fdc18c7732498c15e990b7":{"kind":"REWORK","_number":75,"created":"2025-03-03 09:35:39.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/75"}}},"commit":{"parents":[{"commit":"35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35ec6cba8ce020f6c0b6bcc7bd11e327c2e4f840"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-03 09:35:04.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n- openstacksdk\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/764fce9f8d1b822818fdc18c7732498c15e990b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/764fce9f8d1b822818fdc18c7732498c15e990b7"}]},"branch":"refs/heads/master"},"0677ab96d641ea14c7877952dc424598aa610cf7":{"kind":"REWORK","_number":76,"created":"2025-03-08 09:18:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/76"}}},"commit":{"parents":[{"commit":"b9c0b51e955da11b85eacf8bfd1dbaea93f7000e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9c0b51e955da11b85eacf8bfd1dbaea93f7000e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-08 09:18:45.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0677ab96d641ea14c7877952dc424598aa610cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0677ab96d641ea14c7877952dc424598aa610cf7"}]},"branch":"refs/heads/master"},"4fd58774eed5376e6e5cd0458f2949dbc3de6b39":{"kind":"REWORK","_number":77,"created":"2025-03-15 08:44:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/77"}}},"commit":{"parents":[{"commit":"c760f7c6b424878b15d3c7328f3afae8d56ef41e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c760f7c6b424878b15d3c7328f3afae8d56ef41e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-15 08:44:14.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fd58774eed5376e6e5cd0458f2949dbc3de6b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fd58774eed5376e6e5cd0458f2949dbc3de6b39"}]},"branch":"refs/heads/master"},"0f65c9caeb3cbc72bce6bfd70ef20abeee31911c":{"kind":"REWORK","_number":78,"created":"2025-03-22 09:51:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/78"}}},"commit":{"parents":[{"commit":"59d917962a99dd65a68cab38b75994b828e8d49f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59d917962a99dd65a68cab38b75994b828e8d49f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-22 09:51:19.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f65c9caeb3cbc72bce6bfd70ef20abeee31911c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f65c9caeb3cbc72bce6bfd70ef20abeee31911c"}]},"branch":"refs/heads/master"},"c1cb0e42b0fc2d3c7e9b12d311dcb4779dcd194f":{"kind":"REWORK","_number":79,"created":"2025-03-29 08:45:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/79"}}},"commit":{"parents":[{"commit":"9a0c7163843466b3375cb333c952e91d0b89eaf2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a0c7163843466b3375cb333c952e91d0b89eaf2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-03-29 08:44:56.000000000","tz":60},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1cb0e42b0fc2d3c7e9b12d311dcb4779dcd194f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1cb0e42b0fc2d3c7e9b12d311dcb4779dcd194f"}]},"branch":"refs/heads/master"},"7bfcf8fd080d64dab058c8f3ad776095c95ada92":{"kind":"REWORK","_number":80,"created":"2025-04-05 15:32:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/80"}}},"commit":{"parents":[{"commit":"fbd9487f0807c501c8ada741eeac688511846fd8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-05 15:29:31.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1_modules\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bfcf8fd080d64dab058c8f3ad776095c95ada92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bfcf8fd080d64dab058c8f3ad776095c95ada92"}]},"branch":"refs/heads/master"},"d9b912bc947444bb0927535089eab60fec425de0":{"kind":"REWORK","_number":81,"created":"2025-04-08 06:37:39.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/81"}}},"commit":{"parents":[{"commit":"fbd9487f0807c501c8ada741eeac688511846fd8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-08 06:34:26.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pyasn1_modules\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9b912bc947444bb0927535089eab60fec425de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9b912bc947444bb0927535089eab60fec425de0"}]},"branch":"refs/heads/master"},"e3d4d92c4b00278a85d1f6dd1a0ab2aad96581bb":{"kind":"REWORK","_number":82,"created":"2025-04-08 08:01:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/82"}}},"commit":{"parents":[{"commit":"fbd9487f0807c501c8ada741eeac688511846fd8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-08 08:00:27.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pyroute2\n- pyasn1_modules\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3d4d92c4b00278a85d1f6dd1a0ab2aad96581bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3d4d92c4b00278a85d1f6dd1a0ab2aad96581bb"}]},"branch":"refs/heads/master"},"63685adfd259adcf7214b4d4f72678fa56488e6d":{"kind":"REWORK","_number":83,"created":"2025-04-08 08:32:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/83"}}},"commit":{"parents":[{"commit":"fbd9487f0807c501c8ada741eeac688511846fd8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd9487f0807c501c8ada741eeac688511846fd8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-08 08:31:45.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- rtslib-fb\n- pyroute2\n- pyasn1_modules\n- typing_extentions\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63685adfd259adcf7214b4d4f72678fa56488e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63685adfd259adcf7214b4d4f72678fa56488e6d"}]},"branch":"refs/heads/master"},"f14bbb241415e22e7ebb09922478aa90da5cac28":{"kind":"REWORK","_number":84,"created":"2025-04-12 09:28:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/84"}}},"commit":{"parents":[{"commit":"90bce089d1c1ded6551ace1b9a2e2546cd996091","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90bce089d1c1ded6551ace1b9a2e2546cd996091"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-12 09:25:59.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f14bbb241415e22e7ebb09922478aa90da5cac28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f14bbb241415e22e7ebb09922478aa90da5cac28"}]},"branch":"refs/heads/master"},"f489552c3dc5b3b4010a7952cebcc97910cb9ba4":{"kind":"REWORK","_number":85,"created":"2025-04-26 13:04:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/85"}}},"commit":{"parents":[{"commit":"6607741cbc3f793d1cdda57fe753a0feeb5ce333","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6607741cbc3f793d1cdda57fe753a0feeb5ce333"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-26 13:03:59.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f489552c3dc5b3b4010a7952cebcc97910cb9ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f489552c3dc5b3b4010a7952cebcc97910cb9ba4"}]},"branch":"refs/heads/master"},"adfa7f1ff9d4cc005521ec1d177022b38c7647c0":{"kind":"NO_CHANGE","_number":86,"created":"2025-04-26 14:06:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/86"}}},"commit":{"parents":[{"commit":"0689e0f594d0d4b14f53ac34f5871fdcbf209f83","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0689e0f594d0d4b14f53ac34f5871fdcbf209f83"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-26 14:06:17.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adfa7f1ff9d4cc005521ec1d177022b38c7647c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adfa7f1ff9d4cc005521ec1d177022b38c7647c0"}]},"branch":"refs/heads/master"},"2ada469253a7b5bfc74cd2115fc1c27b841d48c3":{"kind":"REWORK","_number":87,"created":"2025-04-26 20:17:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/87"}}},"commit":{"parents":[{"commit":"0689e0f594d0d4b14f53ac34f5871fdcbf209f83","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0689e0f594d0d4b14f53ac34f5871fdcbf209f83"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-26 20:16:51.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- oslo.middleware\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ada469253a7b5bfc74cd2115fc1c27b841d48c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ada469253a7b5bfc74cd2115fc1c27b841d48c3"}]},"branch":"refs/heads/master"},"b514f24c5a5e25050c317bbfc0bd6f885bffd8cc":{"kind":"REWORK","_number":88,"created":"2025-05-03 09:30:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/88"}}},"commit":{"parents":[{"commit":"e33a5a8bbea6ff01735972923f88878f44aeef84","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e33a5a8bbea6ff01735972923f88878f44aeef84"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-03 09:29:22.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- async-timeout\n- exceptiongroup\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- oslo.middleware\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b514f24c5a5e25050c317bbfc0bd6f885bffd8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b514f24c5a5e25050c317bbfc0bd6f885bffd8cc"}]},"branch":"refs/heads/master"},"59e906bd12885216ae4046275360a3ab63115d51":{"kind":"REWORK","_number":89,"created":"2025-05-19 06:28:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/89"}}},"commit":{"parents":[{"commit":"ac295e4007a710549889c1b08c51806d2c252222","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac295e4007a710549889c1b08c51806d2c252222"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-19 06:25:47.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.context\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59e906bd12885216ae4046275360a3ab63115d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59e906bd12885216ae4046275360a3ab63115d51"}]},"branch":"refs/heads/master"},"09913d3af8bb6781e11b78c6273146ef098cfd18":{"kind":"TRIVIAL_REBASE","_number":90,"created":"2025-05-19 11:55:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/90"}}},"commit":{"parents":[{"commit":"b7004924922a25135234577653f573b6ff8d3f1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7004924922a25135234577653f573b6ff8d3f1a"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-19 11:55:43.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.context\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09913d3af8bb6781e11b78c6273146ef098cfd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09913d3af8bb6781e11b78c6273146ef098cfd18"}]},"branch":"refs/heads/master"},"22c2be4832821012b3a136f12fd392893bf99318":{"kind":"REWORK","_number":91,"created":"2025-05-20 02:11:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/91"}}},"commit":{"parents":[{"commit":"b7004924922a25135234577653f573b6ff8d3f1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7004924922a25135234577653f573b6ff8d3f1a"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-20 02:10:16.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nAdd version specific pin for:\n\n- click\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.context\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c2be4832821012b3a136f12fd392893bf99318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c2be4832821012b3a136f12fd392893bf99318"}]},"branch":"refs/heads/master"},"9d7b4dc0645233c140c5f546b3e85c21d220aae2":{"kind":"TRIVIAL_REBASE","_number":92,"created":"2025-05-20 09:26:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/92"}}},"commit":{"parents":[{"commit":"ae724f9200ec41703c580185db78c03379a14ba5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae724f9200ec41703c580185db78c03379a14ba5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-20 09:26:35.000000000","tz":0},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nAdd version specific pin for:\n\n- click\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.context\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d7b4dc0645233c140c5f546b3e85c21d220aae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d7b4dc0645233c140c5f546b3e85c21d220aae2"}]},"branch":"refs/heads/master","description":"Rebase"},"b93eb2aad3e60be4f50d01b2722f06321182317e":{"kind":"REWORK","_number":93,"created":"2025-05-24 13:29:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/93"}}},"commit":{"parents":[{"commit":"1f81a38877e9920935067a5164d6f8fabbbd96a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f81a38877e9920935067a5164d6f8fabbbd96a3"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 13:29:00.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nAdd version specific pin for:\n\n- click\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.context\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b93eb2aad3e60be4f50d01b2722f06321182317e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b93eb2aad3e60be4f50d01b2722f06321182317e"}]},"branch":"refs/heads/master"},"fa6ed318ff218d06b5384286d23b51a837c1cf40":{"kind":"REWORK","_number":94,"created":"2025-05-24 18:30:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/94"}}},"commit":{"parents":[{"commit":"1f81a38877e9920935067a5164d6f8fabbbd96a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f81a38877e9920935067a5164d6f8fabbbd96a3"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 18:26:51.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n- click\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa6ed318ff218d06b5384286d23b51a837c1cf40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa6ed318ff218d06b5384286d23b51a837c1cf40"}]},"branch":"refs/heads/master"},"e4592776f34fee5232f4899f61af71c31d29b6b7":{"kind":"REWORK","_number":95,"created":"2025-05-25 09:40:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/95"}}},"commit":{"parents":[{"commit":"1f81a38877e9920935067a5164d6f8fabbbd96a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f81a38877e9920935067a5164d6f8fabbbd96a3"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-25 09:39:17.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n- click\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4592776f34fee5232f4899f61af71c31d29b6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4592776f34fee5232f4899f61af71c31d29b6b7"}]},"branch":"refs/heads/master"},"941c0a69c624fb24694c724dfa4ab6c89cb27838":{"kind":"REWORK","_number":96,"created":"2025-05-31 13:40:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/96"}}},"commit":{"parents":[{"commit":"9e220e3f739e613a03e5cf151d2a3938063d0a1d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e220e3f739e613a03e5cf151d2a3938063d0a1d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-31 13:25:50.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n- click\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/941c0a69c624fb24694c724dfa4ab6c89cb27838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/941c0a69c624fb24694c724dfa4ab6c89cb27838"}]},"branch":"refs/heads/master"},"53e6fc5c1270a3b9b9c4102f53e43f49753d89c4":{"kind":"REWORK","_number":97,"created":"2025-06-01 05:42:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/97"}}},"commit":{"parents":[{"commit":"9e220e3f739e613a03e5cf151d2a3938063d0a1d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e220e3f739e613a03e5cf151d2a3938063d0a1d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-01 05:40:53.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.limit\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53e6fc5c1270a3b9b9c4102f53e43f49753d89c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53e6fc5c1270a3b9b9c4102f53e43f49753d89c4"}]},"branch":"refs/heads/master"},"ffc7f4933eb70c6acf3482e3ae8baf511b15d634":{"kind":"REWORK","_number":98,"created":"2025-06-07 15:08:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/98"}}},"commit":{"parents":[{"commit":"9afdae046ad321da88defca4772f0ffa0d9c35cf","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9afdae046ad321da88defca4772f0ffa0d9c35cf"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-07 15:07:51.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- openstacksdk\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc7f4933eb70c6acf3482e3ae8baf511b15d634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc7f4933eb70c6acf3482e3ae8baf511b15d634"}]},"branch":"refs/heads/master"},"82904e77fa2eff8ef225c54b5fa9a218e4da721c":{"kind":"REWORK","_number":99,"created":"2025-06-08 11:40:54.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/99"}}},"commit":{"parents":[{"commit":"9afdae046ad321da88defca4772f0ffa0d9c35cf","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9afdae046ad321da88defca4772f0ffa0d9c35cf"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-08 11:40:07.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- openstacksdk\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nAdd version specific pin for:\n\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82904e77fa2eff8ef225c54b5fa9a218e4da721c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82904e77fa2eff8ef225c54b5fa9a218e4da721c"}]},"branch":"refs/heads/master"},"8ce8323385ff92b8063d0a07ccb3d0debb13fd72":{"kind":"REWORK","_number":100,"created":"2025-06-14 09:14:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/100"}}},"commit":{"parents":[{"commit":"8d96955558fab1776ef122ca1d984a1d30cdf6e2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d96955558fab1776ef122ca1d984a1d30cdf6e2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-14 09:08:11.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- openstacksdk\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nAdd version specific pin for:\n\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ce8323385ff92b8063d0a07ccb3d0debb13fd72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ce8323385ff92b8063d0a07ccb3d0debb13fd72"}]},"branch":"refs/heads/master"},"6a7685d53662b7133a2763abc765f37db7460d93":{"kind":"REWORK","_number":101,"created":"2025-06-14 13:38:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/101"}}},"commit":{"parents":[{"commit":"8096933b5ce3523c15e3655e64393694b3c0f54d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8096933b5ce3523c15e3655e64393694b3c0f54d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-14 13:38:13.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- openstacksdk\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nAdd version specific pin for:\n\n- scikit-learn\n- doc8\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a7685d53662b7133a2763abc765f37db7460d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a7685d53662b7133a2763abc765f37db7460d93"}]},"branch":"refs/heads/master"},"f9b92a030ca9e619f5e8cdccea015d18623b1b4e":{"kind":"REWORK","_number":102,"created":"2025-06-21 11:32:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/102"}}},"commit":{"parents":[{"commit":"a32424b1a5e48741fbcb2479d9f15aebd0894874","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a32424b1a5e48741fbcb2479d9f15aebd0894874"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-21 11:30:04.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- doc8\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- openstacksdk\n- python-openstackclient\n- importlib-metadata\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9b92a030ca9e619f5e8cdccea015d18623b1b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9b92a030ca9e619f5e8cdccea015d18623b1b4e"}]},"branch":"refs/heads/master"},"3ac4618b375965648e6bf2d76e0b672ef3812881":{"kind":"TRIVIAL_REBASE","_number":103,"created":"2025-06-28 11:20:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/103"}}},"commit":{"parents":[{"commit":"9dcb4ae74256188b81995198b488ceb794b9f00c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcb4ae74256188b81995198b488ceb794b9f00c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-28 11:20:06.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- doc8\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- openstacksdk\n- python-openstackclient\n- importlib-metadata\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ac4618b375965648e6bf2d76e0b672ef3812881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ac4618b375965648e6bf2d76e0b672ef3812881"}]},"branch":"refs/heads/master"},"d325843e95654b31f1a7714f09076b09528144ef":{"kind":"REWORK","_number":104,"created":"2025-06-28 13:14:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/104"}}},"commit":{"parents":[{"commit":"9dcb4ae74256188b81995198b488ceb794b9f00c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcb4ae74256188b81995198b488ceb794b9f00c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-28 13:13:29.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- doc8\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- openstacksdk\n- python-openstackclient\n- importlib-metadata\n- scikit-learn\n\nAdd version specific pin for:\n\n- txaio\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d325843e95654b31f1a7714f09076b09528144ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d325843e95654b31f1a7714f09076b09528144ef"}]},"branch":"refs/heads/master"},"ab56fec5c503707b39b7c0018f5595ee2b6fb77e":{"kind":"REWORK","_number":105,"created":"2025-07-05 13:00:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/105"}}},"commit":{"parents":[{"commit":"3cf5007eb67f8cee1f899f1cbf868710b80d3bcd","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cf5007eb67f8cee1f899f1cbf868710b80d3bcd"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-05 13:00:00.000000000","tz":120},"subject":"Fix for automated upper-constraints.txt update","message":"Fix for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- doc8\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- openstacksdk\n- python-openstackclient\n- importlib-metadata\n- scikit-learn\n\nAdd version specific pin for:\n\n- txaio\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab56fec5c503707b39b7c0018f5595ee2b6fb77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab56fec5c503707b39b7c0018f5595ee2b6fb77e"}]},"branch":"refs/heads/master"},"9d4cefeb3d9152011bbf5e953d5bdd446fe6b96c":{"kind":"REWORK","_number":106,"created":"2025-07-12 17:08:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/106"}}},"commit":{"parents":[{"commit":"fd949a82415ccd22840145107bd7155230ab13d4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd949a82415ccd22840145107bd7155230ab13d4"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-12 17:07:04.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d4cefeb3d9152011bbf5e953d5bdd446fe6b96c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d4cefeb3d9152011bbf5e953d5bdd446fe6b96c"}]},"branch":"refs/heads/master"},"175e79312b76a21ea829feb92fdc76363edf0b3c":{"kind":"REWORK","_number":107,"created":"2025-07-16 07:01:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/107"}}},"commit":{"parents":[{"commit":"0d80d69a96b891c58d1291ecc70eb3a23bb84e27","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d80d69a96b891c58d1291ecc70eb3a23bb84e27"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-16 07:00:38.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/175e79312b76a21ea829feb92fdc76363edf0b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/175e79312b76a21ea829feb92fdc76363edf0b3c"}]},"branch":"refs/heads/master"},"ff601df10ea46bb3ee7694859d7d8c081c97be48":{"kind":"REWORK","_number":108,"created":"2025-07-19 10:15:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/108"}}},"commit":{"parents":[{"commit":"d79a6bba8441dbdc69a1d3011fa0f39e69869ca5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79a6bba8441dbdc69a1d3011fa0f39e69869ca5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-19 10:13:36.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff601df10ea46bb3ee7694859d7d8c081c97be48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff601df10ea46bb3ee7694859d7d8c081c97be48"}]},"branch":"refs/heads/master"},"f1a3dca471563eb773511eaf12a6ff9db8f35fc7":{"kind":"REWORK","_number":109,"created":"2025-07-28 19:12:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/109","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/109","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/109 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/109 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/109 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/109"}}},"commit":{"parents":[{"commit":"19fecc75075f383383bcbf5399f475b919e5c336","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19fecc75075f383383bcbf5399f475b919e5c336"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-28 19:11:28.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1a3dca471563eb773511eaf12a6ff9db8f35fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1a3dca471563eb773511eaf12a6ff9db8f35fc7"}]},"branch":"refs/heads/master"},"3fefd2722b6e29391c5efa93710fc7ac636f9e45":{"kind":"REWORK","_number":110,"created":"2025-08-02 08:44:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/110","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/110","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/110 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/110 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/110 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/110"}}},"commit":{"parents":[{"commit":"9dcdf42b89677db2fc06dcf33181200b3ecba7c1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcdf42b89677db2fc06dcf33181200b3ecba7c1"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-08-02 08:42:33.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fefd2722b6e29391c5efa93710fc7ac636f9e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fefd2722b6e29391c5efa93710fc7ac636f9e45"}]},"branch":"refs/heads/master"},"de87731cfd4473c6ede2b667d0c497cd59bc9150":{"kind":"REWORK","_number":111,"created":"2025-08-02 14:34:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/111","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/111","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/111 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/111 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/111 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/111"}}},"commit":{"parents":[{"commit":"9dcdf42b89677db2fc06dcf33181200b3ecba7c1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcdf42b89677db2fc06dcf33181200b3ecba7c1"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-08-02 14:33:25.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de87731cfd4473c6ede2b667d0c497cd59bc9150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de87731cfd4473c6ede2b667d0c497cd59bc9150"}]},"branch":"refs/heads/master"},"6d1ee745562712f25eb5900b96f732352d5860ad":{"kind":"REWORK","_number":112,"created":"2025-08-09 09:15:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/112","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/112","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/112 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/112 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/112 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/112"}}},"commit":{"parents":[{"commit":"69e27dc7f29907d51c7023a9f78ee0f95fb27df2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69e27dc7f29907d51c7023a9f78ee0f95fb27df2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-08-09 09:13:33.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d1ee745562712f25eb5900b96f732352d5860ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d1ee745562712f25eb5900b96f732352d5860ad"}]},"branch":"refs/heads/master"},"b7df576aa60e90a450350ba44dab7862a174975b":{"kind":"REWORK","_number":113,"created":"2025-08-09 15:49:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/113","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/113","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/113 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/113 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/113 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/113"}}},"commit":{"parents":[{"commit":"69e27dc7f29907d51c7023a9f78ee0f95fb27df2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69e27dc7f29907d51c7023a9f78ee0f95fb27df2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-08-09 15:48:48.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7df576aa60e90a450350ba44dab7862a174975b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7df576aa60e90a450350ba44dab7862a174975b"}]},"branch":"refs/heads/master"},"6826fa4ae1eb610fe62ec3488b8b953d142cb489":{"kind":"REWORK","_number":114,"created":"2025-08-13 13:41:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/907665/114","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/114","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/114 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/114 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/114 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/114"}}},"commit":{"parents":[{"commit":"9b395ab9ddd7651943ae2c84e60e19b3fbef3de7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b395ab9ddd7651943ae2c84e60e19b3fbef3de7"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 10:14:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-08-13 13:41:16.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6826fa4ae1eb610fe62ec3488b8b953d142cb489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6826fa4ae1eb610fe62ec3488b8b953d142cb489"}]},"branch":"refs/heads/master"},"3b9fa9a7f35a27e7fb3833d836cf19ccafda6a32":{"kind":"REWORK","_number":115,"created":"2025-10-23 10:37:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/907665/115","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/115","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/115 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/115 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/115 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/115"}}},"commit":{"parents":[{"commit":"40dac68b6c26a04535a4cdc5572861e17005c0d8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40dac68b6c26a04535a4cdc5572861e17005c0d8"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:34:13.000000000","tz":660},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- boto3, botocore, s3transfer\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b9fa9a7f35a27e7fb3833d836cf19ccafda6a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b9fa9a7f35a27e7fb3833d836cf19ccafda6a32"}]},"branch":"refs/heads/master"},"7394e4124ad01a396ae55ca57380cbf5d0c4c0b6":{"kind":"TRIVIAL_REBASE","_number":116,"created":"2025-10-23 21:31:09.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/907665/116","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/116","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/116 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/116 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/116 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/116"}}},"commit":{"parents":[{"commit":"e6b8663d53b8c5eb3087277a4f89e6d78e7804ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6b8663d53b8c5eb3087277a4f89e6d78e7804ac"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 21:30:47.000000000","tz":660},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- boto3, botocore, s3transfer\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7394e4124ad01a396ae55ca57380cbf5d0c4c0b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7394e4124ad01a396ae55ca57380cbf5d0c4c0b6"}]},"branch":"refs/heads/master"},"fcc4e40756274b73f1996a02ff198b2f7f59ca1b":{"kind":"NO_CODE_CHANGE","_number":117,"created":"2025-10-23 23:34:30.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/907665/117","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/117","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/117 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/117 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/117 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/117"}}},"commit":{"parents":[{"commit":"e6b8663d53b8c5eb3087277a4f89e6d78e7804ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6b8663d53b8c5eb3087277a4f89e6d78e7804ac"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 21:44:25.000000000","tz":660},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- boto3, botocore, s3transfer:\n    glance:\n    - https://review.opendev.org/c/openstack/glance-specs/+/962248\n    - https://review.opendev.org/c/openstack/requirements/+/959560\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcc4e40756274b73f1996a02ff198b2f7f59ca1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcc4e40756274b73f1996a02ff198b2f7f59ca1b"}]},"branch":"refs/heads/master"},"cc3ab957b7a60bfbcbee5db9128f625a8108561c":{"kind":"TRIVIAL_REBASE","_number":118,"created":"2025-10-24 00:28:35.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/907665/118","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/118","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/118 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/118 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/118 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/118"}}},"commit":{"parents":[{"commit":"6918edd52478cf64ff566639f31770a7c467946f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6918edd52478cf64ff566639f31770a7c467946f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-24 00:28:11.000000000","tz":660},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- boto3, botocore, s3transfer:\n    glance:\n    - https://review.opendev.org/c/openstack/glance-specs/+/962248\n    - https://review.opendev.org/c/openstack/requirements/+/959560\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc3ab957b7a60bfbcbee5db9128f625a8108561c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc3ab957b7a60bfbcbee5db9128f625a8108561c"}]},"branch":"refs/heads/master"},"f7822852a2e9b8a564d2890cdebde0eebac9c666":{"kind":"NO_CHANGE","_number":119,"created":"2025-10-24 00:49:02.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/907665/119","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/907665/119","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/119 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/119 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/907665/119 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/907665/119"}}},"commit":{"parents":[{"commit":"bb61e2844b141f27e7cc11b5c67cdb393e0abfe5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb61e2844b141f27e7cc11b5c67cdb393e0abfe5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-24 00:48:44.000000000","tz":660},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- boto3, botocore, s3transfer:\n    glance:\n    - https://review.opendev.org/c/openstack/glance-specs/+/962248\n    - https://review.opendev.org/c/openstack/requirements/+/959560\n\nChange-Id: I6c31420e9dff496e4296b0d7b8af90a78a45873a\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7822852a2e9b8a564d2890cdebde0eebac9c666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7822852a2e9b8a564d2890cdebde0eebac9c666"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
