)]}'
{"id":"openstack%2Frequirements~907677","triplet_id":"openstack%2Frequirements~master~I4ab8e52d4531257628280b14c80b98cda9622fd2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-21 10:18:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ab8e52d4531257628280b14c80b98cda9622fd2","subject":"Block scrypt\u003d\u003d0.8.21","status":"MERGED","created":"2024-02-04 14:41:41.000000000","updated":"2024-02-21 10:19:23.000000000","submitted":"2024-02-21 10:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907677-openstack/requirements/constraints","meta_rev_id":"fc39bb873249c966193039aecbbe614c4cbc8856","_number":907677,"virtual_id_number":907677,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 10:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-21 08:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-02-19 13:37:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-21 08:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-04 16:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-11 23:27:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-02-18 02:58:01.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"}],"messages":[{"id":"6f0f15131999f85f0b4c9408124d8e6ec02980ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-04 14:41:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1d9202401ea31807f546d6b8f14038eed7155c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 16:07:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7db02c85ef9f4dca9d4f87ba06d9afac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e57c3249ec74992a827ef04f34db640 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c2a18966a743948ee7ca291eebe852 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83a3740d07dc467d976c783ee62e09aa : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2548a1e6fe4f4597a0447a243502f34c : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/576ead479d514ffeb16845b293369196 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e8fa22ad86b4644820605d037471040 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a6ea1f97a1c4a04941d57381347bd1b : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/316403a22ab9410db8e1f227e8094fed : SUCCESS in 11m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/50e5f2e23bd546d59350c69860d29429 : SUCCESS in 10m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d7d3b083e6e94a8fa87cd5d1243d96a5 : SUCCESS in 11m 58s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9b8681c7049143bc8708fdd7d79b4a03 : SUCCESS in 5m 04s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6a5726c9a9e247cc9fed3040ac3e09f8 : SUCCESS in 4m 35s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5ffd6ba754e24f918b22d00aa0080475 : SUCCESS in 4m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bef5ba81523e45c189b9ef573d0b4fb0 : FAILURE in 5m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6a646e10d13b43b1b174b80e77880cef : SUCCESS in 4m 21s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/48d708b922344e56b47a1e6c0e66624c : SUCCESS in 4m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/31e4d16d9a5746778e65a4f75de9219e : SUCCESS in 6m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9ad2f5d5bc5742da8445c8fda2ce9772 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1abba50a456b4f6da5bd7286c8918be0 : SUCCESS in 6m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d5881a0f70e24a86a12f5c04cd54e1d0 : FAILURE in 8m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a712e93bc8824cd186b7802bb2492545 : SUCCESS in 10m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/86c83109cb83475f96928b68636f0c7a : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/317ebfb5d9194c9abe0443e2a0ba3329 : SUCCESS in 8m 22s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/243e369ba1204920865acf9bb749fbaa : FAILURE in 4m 49s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2bd9aff019264a7699d76afefd82c21d : SUCCESS in 12m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/4bbbc5beee7b4d1b89d5f6521dee6874 : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f7f62ebfc10f4ae5bdf0f345b1b6e0b8 : SUCCESS in 9m 24s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/56440d9db2a7499a9f2aca4e47c1bab6 : FAILURE in 11m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/44c5eec58f184156a15094aec2e100a6 : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/416ca9c8fb914916aa6cf74da2146ae5 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a042fb51cb5a4d9d9fe78773be80683e : FAILURE in 23m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a547a4655de44c948394b28388ccfa57 : SUCCESS in 4m 48s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/96782f6e9897469da5d64b1c3aa83171 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/745502ec2a2241b895e22f0d702b688c : SUCCESS in 6m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9d25d6b230124c09949a234f0cfc2ee9 : SUCCESS in 15m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5146b4e2d27f4850987a33a03b0e36b1 : SUCCESS in 31m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b51797ac2ca441dc84fe1e83841a1dc7 : SUCCESS in 1h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a97ab68e797909f116689956bc632d8adb06026b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-04 18:22:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"60b8ea87f8441ec8bc1563dab02b694a2c3bb6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 20:25:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efc6847a98c94d929fe31f34ad195240\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f6725c69d84cdab9463b51be451b08 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f464d6406c4d5da2f39d3bc2ef7527 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db94de4fb4404da2a6a6a191842e4cf9 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/784fa74eb9914c8ebb755f6a271efdcf : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bd76bb58d294d388dad8294f86448de : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2257cced4e1f4abaa7b3a7d9e09481ef : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8e92ea5e5744bb0ae03b4a8b2c56550 : FAILURE in 3m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86cdc3b7d8f44612b8d39d574be5974c : FAILURE in 6m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ba2ea9862b04dc88aae837970699ed6 : SUCCESS in 8m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ead342b181fa4c488c5fb410da479472 : SUCCESS in 19m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a698ed0da010490ba9b4dffc101711a0 : FAILURE in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/304b8c08701b4f729a4b10a49ff75e0e : SUCCESS in 4m 51s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c9505cd5de174a7db6772b5bcfd63f3e : SUCCESS in 5m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8dc0429f1e604f3dbbaee41516378071 : FAILURE in 5m 49s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/33ff9bcf47ed448ba2d33d1d1e1813cf : SUCCESS in 4m 15s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cb9c2256b38e4ae09666e75f4a56c44a : SUCCESS in 4m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8d52e2a04e694f029e8fa9bc23cd52e4 : SUCCESS in 6m 17s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0032cfcd04a54ce9b53af45f74e872bd : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61fca6f0b4b34ac0b9750ddab1b81aeb : SUCCESS in 8m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7cad63146c5548d08f63ea79190fd3a2 : FAILURE in 6m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/049d836628744b849df409374a4c6825 : SUCCESS in 11m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e49050214c484d20b92252bd93fcdb3d : SUCCESS in 3m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/84f4824d56774e17832c00b117b75c44 : FAILURE in 6m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/513e328888114f5bb0578182f5a1cd7e : FAILURE in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/871ef7b6a33644798926eb8a06002bb9 : SUCCESS in 9m 32s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/df417706b3e3400aa671447e481ffcb4 : SUCCESS in 17m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/55f7b739415a4752b6e303a9ff05a542 : SUCCESS in 6m 18s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5ee6f17d9f7e410d913a0c221f85efcc : FAILURE in 8m 17s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/887e7587ad204ea39aaa1281cf8ef18f : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6047bd8d266b45d0b4411cd7ca85476e : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8e041c17c2a48eb88efead58b98454b : FAILURE in 17m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/474280712feb4f5299e9190e9ba18395 : SUCCESS in 4m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/dbede26778bf4912bf754b7aa9125956 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfbc40139c3a45af89f2d7143a7acbe6 : SUCCESS in 6m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e3358b9f626649f4b58bab57b0b1e034 : SUCCESS in 9m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8614fb4520aa4ce4b057b3e1e4b9a148 : SUCCESS in 35m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55ad25d46f78443f9f18549d1cf0d692 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a75086a49babc9c7556b39a1a999d23047a98ce8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-11 23:27:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2925754d5672d1d5f5f211ca54feeff8ac97032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-16 20:39:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c0412c74a316f0d7ae80313a011085618113a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:56:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5343b9060394d4f8190dc9dacb8f728\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a33e77b33a4945a6e55068463f8c42 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5bd5fd207949adb4895ecb472f1d2b : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6813f40942cd431583f3ecc2e4f4945c : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/588bf08a503d46bb8bf9575f9b317bcf : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5fc7f31cedab4952aafdbc2f72925f8f : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71c9aa27672d43d8bf6813d7d0532aa1 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/90202c9318484015b2e55611fa0e0b60 : FAILURE in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/07b7fe74f0f34911a10efd22209d49f3 : FAILURE in 5m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b32e5a83245d4343a750cbb7e2fc11c2 : SUCCESS in 10m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1151c410aab2445daf3b5136841972e8 : SUCCESS in 16m 48s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c5548e992c3b4e2ca2d7b8b035076407 : FAILURE in 5m 09s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2b14594ee8d149e089ca5afdfe5d8700 : SUCCESS in 5m 30s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/426d108b58ee4b8893163e93e3e3b0d4 : SUCCESS in 4m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1c8b5d6e18f34e1889e1f888e0e854d5 : FAILURE in 10m 06s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5750e7aa81f74cf0befdd21da5de8aa3 : SUCCESS in 4m 54s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/739f93153c10431a8cc77f92145a974c : SUCCESS in 5m 17s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3f234ce9f57b4236b0e506cf1d3e6d22 : FAILURE in 7m 13s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fdfd7cd5741b489192febed71816f8e5 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/803ab7c5167d4913955b64793db09b9b : SUCCESS in 8m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/57cacfd816ce48dd99e30dc4c12982b7 : FAILURE in 8m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/24669503d88e4df2b176f536c056982f : SUCCESS in 13m 09s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9349a815522b458b9394301c1be1c36b : SUCCESS in 3m 44s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c812953becc74cd191f7cbec99b3f057 : FAILURE in 6m 57s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a803a208da8644158ae461c9bb1003e3 : FAILURE in 4m 17s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/264ca9b9d74540899c29a2e53b3b300d : SUCCESS in 10m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0ce305c4da094e04b9ac46832e8435b6 : SUCCESS in 27m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/319b7a859a284ac39cb9a87285dd754b : SUCCESS in 7m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b004bc267dbd45ffb4c71e09514e1b30 : FAILURE in 11m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5c33721722684120abd01e66243ca9eb : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb16e9a582df4ff8921d1bf253735acf : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84e8ca31e9424b6d918a0ae281367cff : FAILURE in 21m 35s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ba685d4a873b43b9998fdab6df2a41f4 : SUCCESS in 4m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d1a80b040c484c9fbdfd8d6a23d65bf5 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe03eedc73c54238844b4598d059587b : SUCCESS in 5m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5f961d709a4447b6840c91ba563e2d13 : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/478e9fcfb9254358b05851591886b3a7 : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4d80559d670413493953013dd932056 : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":3},{"id":"fd4cb6c6654c6e0f5f94123607cac04ffc731090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-17 09:56:34.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"14ac89704cd5d298c279f494f796c773e87cf45f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 11:07:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90b0501e5cb1470db6884562d1c4b98f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e470cdd587a641e6a9bcf28744c227e5 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7c1920dde24f3cac982697833cc48c : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599c3c0082704755a10f6a4881b705c6 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddff4f1569404859b58a9ff1db280b64 : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e0593c37c7b4134ad5106aa92d65740 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf66be5e933c470f87621137c7799c5e : SUCCESS in 2m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae2a6ebcb7fa4bfe82d464995075c07c : FAILURE in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7dc3653053b491c80f405e79d7d4384 : FAILURE in 4m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dabf98bbb7ed4564a7c8daabfaf5bad0 : SUCCESS in 7m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38c91223d3df46eaa3fbbb7c1a804e6e : SUCCESS in 17m 37s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e19c3b5fff5d48deb1aaa679e6de012c : FAILURE in 4m 00s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/44f623a43375467abc50bfed1279de72 : SUCCESS in 4m 39s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/66532893f2094e2d89506e532806ba62 : SUCCESS in 4m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d605ba484306448e87dd7e2f7e0e6630 : FAILURE in 7m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/957c9640192e43b2a459314e38113ff6 : SUCCESS in 4m 27s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fbf8b7ab235a45e78a178314cd359f1c : SUCCESS in 6m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e3f3b5b722a54e4abfec68adb1e87d23 : FAILURE in 5m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4eef7dbf5cd346ae9bb34cab27e54fee : FAILURE in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fc337d761bb49a0b3b92b7d1327f56e : SUCCESS in 7m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6ecb88aa4af345c99dfcabe82b320432 : FAILURE in 6m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ef627985e4bb4b09b7923f2e78aeef4a : SUCCESS in 11m 24s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/00e73a535ae84ee49efbe44a17a49283 : SUCCESS in 4m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/98806be21bd549338dfdac280b6d0836 : FAILURE in 6m 13s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ef381907675f4cd1bd3d708dfb691961 : FAILURE in 4m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8f20f8ee090c4aa481d63bb481476c4a : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8076878a7ea643ed926fed3df1c92092 : FAILURE in 16m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8f5efde5276e49998e71196e55df2e3d : SUCCESS in 5m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ecf05048fc7b491d8f069cdba1c3c606 : FAILURE in 8m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/dd7174a21aa34bdab1d1307b87ac5eb5 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2841378470274fc192d2d02d8969820a : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3359de8f9ec441ce837a0284a59bde83 : FAILURE in 18m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1044a0aa85be437d9fe045990e62b134 : SUCCESS in 4m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/41b0358da8cd446e8e8169a3a75bee49 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ef2ee998aba47d1af6c63ee191d8957 : SUCCESS in 5m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0242511cceeb4444adf051ed792743bf : SUCCESS in 16m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f59d406b9184ddfb708293bb5518ea4 : SUCCESS in 1h 04m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a43d5eb3e3a8446996bcb5f9ad187941 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":4},{"id":"3d1fd2482abefab2c49c82631c5b4c0efaac1939","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-18 02:58:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"426a86588443c3f717715c7829e20618fbf43e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 13:37:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cbeb8b490518d8b490737758182521ad81736b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:09:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f73be895ab14edc824a1d706e6de9ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69bfd2d49c9a421291d895e70ca5185f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8659d4c1cebd4423b3c2194ee2154ea3 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/642a80f270974c6094711b1df8230579 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1c27bc56a944c549d2203e1acb17828 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11fef7b635254a3689a5267a251a55db : SUCCESS in 2m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7847e9fe8ec411d8a0cc94c24106d71 : SUCCESS in 57m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/080e430b20294a72b0fbb99dd6f91deb : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":5},{"id":"31947343aea53166d74cfd87a13d5be72efbb661","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-21 08:37:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b2656c6d60eb0bb0e8b89f1fc4a9879745dff16c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 08:38:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fb2da4bad8b3028be2efdb2cd3cbbe9c12cbac27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 10:18:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b31f7c8862410c8317b47cc3526597\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab0aff77d6946e5837dde582895e9ab : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfc8554542149f2994c99a191d70037 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c6c62852eb04a87ab1af79ff11925d4 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ac16dc3d8814969ae4820a123494f76 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49579528e4844d57b52a1a1c27d81e69 : SUCCESS in 2m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95d9d7e250ef4f389531718dd0dee7df : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c7dd92790e847e39dba0aef2b177741 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":5},{"id":"3419963ba164ba4b20a8dc98db8750f57dd3e25b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 10:18:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fc39bb873249c966193039aecbbe614c4cbc8856","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 10:19:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55169613f4ae41c8981255ed503b085e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52e7b166cc34118a6affce0b1a6ec6f : SUCCESS in 34s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ca4b5e2a306de6db959bd914a72d818a3b6e4ada","revisions":{"7f4c791c0bd884dacb9d0b658eb8023df7e0b7ef":{"kind":"REWORK","_number":1,"created":"2024-02-04 14:41:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/907677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/907677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/907677/1"}}},"commit":{"parents":[{"commit":"90935d6286263a532a1eee9e5aef8917227c4f0e","subject":"Fix for automated upper-constraints.txt update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90935d6286263a532a1eee9e5aef8917227c4f0e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"subject":"Block scrypt\u003d\u003d0.8.21","message":"Block scrypt\u003d\u003d0.8.21\n\nThere are no cp311 wheels published for this version, leading to testing\nissues[0].\n\n[0] https://github.com/holgern/py-scrypt/issues/16\n\nChange-Id: I4ab8e52d4531257628280b14c80b98cda9622fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f4c791c0bd884dacb9d0b658eb8023df7e0b7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f4c791c0bd884dacb9d0b658eb8023df7e0b7ef"}]},"branch":"refs/heads/master"},"90a3d1b203090d3bf98f6e2618c5ae38f10e6c41":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-04 18:22:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/907677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/907677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/907677/2"}}},"commit":{"parents":[{"commit":"bd9a8b75fdc692d2e0102d64f775388ecef89172","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd9a8b75fdc692d2e0102d64f775388ecef89172"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 18:21:39.000000000","tz":60},"subject":"Block scrypt\u003d\u003d0.8.21","message":"Block scrypt\u003d\u003d0.8.21\n\nThere are no cp311 wheels published for this version, leading to testing\nissues[0].\n\n[0] https://github.com/holgern/py-scrypt/issues/16\n\nChange-Id: I4ab8e52d4531257628280b14c80b98cda9622fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a3d1b203090d3bf98f6e2618c5ae38f10e6c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a3d1b203090d3bf98f6e2618c5ae38f10e6c41"}]},"branch":"refs/heads/master"},"605bdc33e508567bb9c838ef755743036edb16b7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-16 20:39:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/907677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/907677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/907677/3"}}},"commit":{"parents":[{"commit":"4438d2c49a70d22b8d1d0926d00062f5e64433ee","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4438d2c49a70d22b8d1d0926d00062f5e64433ee"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-02-16 20:39:19.000000000","tz":0},"subject":"Block scrypt\u003d\u003d0.8.21","message":"Block scrypt\u003d\u003d0.8.21\n\nThere are no cp311 wheels published for this version, leading to testing\nissues[0].\n\n[0] https://github.com/holgern/py-scrypt/issues/16\n\nChange-Id: I4ab8e52d4531257628280b14c80b98cda9622fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/605bdc33e508567bb9c838ef755743036edb16b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/605bdc33e508567bb9c838ef755743036edb16b7"}]},"branch":"refs/heads/master","description":"Rebase"},"e493017dcccf6c1a28666d1cea07b40bc16c1342":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-17 09:56:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/907677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/907677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/907677/4"}}},"commit":{"parents":[{"commit":"fc66bb90111b905f62bd4f7bb97ba696073a0443","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc66bb90111b905f62bd4f7bb97ba696073a0443"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-02-17 09:56:34.000000000","tz":0},"subject":"Block scrypt\u003d\u003d0.8.21","message":"Block scrypt\u003d\u003d0.8.21\n\nThere are no cp311 wheels published for this version, leading to testing\nissues[0].\n\n[0] https://github.com/holgern/py-scrypt/issues/16\n\nChange-Id: I4ab8e52d4531257628280b14c80b98cda9622fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e493017dcccf6c1a28666d1cea07b40bc16c1342"}]},"branch":"refs/heads/master","description":"Rebase"},"ca4b5e2a306de6db959bd914a72d818a3b6e4ada":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-19 13:37:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/77/907677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/907677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/907677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/907677/5"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-04 14:39:46.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-19 13:36:28.000000000","tz":60},"subject":"Block scrypt\u003d\u003d0.8.21","message":"Block scrypt\u003d\u003d0.8.21\n\nThere are no cp311 wheels published for this version, leading to testing\nissues[0].\n\n[0] https://github.com/holgern/py-scrypt/issues/16\n\nChange-Id: I4ab8e52d4531257628280b14c80b98cda9622fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca4b5e2a306de6db959bd914a72d818a3b6e4ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca4b5e2a306de6db959bd914a72d818a3b6e4ada"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
