)]}'
{"id":"openstack%2Frequirements~907962","triplet_id":"openstack%2Frequirements~master~I03c39c342794b62ba13aec88a833e1bc5bb87eaf","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-07 01:09:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03c39c342794b62ba13aec88a833e1bc5bb87eaf","subject":"Remove unused abclient","status":"MERGED","created":"2024-02-06 08:21:11.000000000","updated":"2024-02-07 01:09:55.000000000","submitted":"2024-02-07 01:09:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907962","meta_rev_id":"5f9f11629ddc99c7e8ec0a9d2292ec57b3353965","_number":907962,"virtual_id_number":907962,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-07 01:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-02-06 13:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-02-06 23:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-02-06 23:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 10:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 13:50:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-02-06 23:51:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2eba9781213caaaee5ab9bed45edb95c1f07a7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 08:21:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44ab30c6cc58739aef24cdbadfc056ef8237e929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/289c2001db014c42986b86f520195f8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbde0ed15c854f9fb0f63e3f46261b5e : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019aa96dd4a243a3b61be44542f75525 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57ea4b3ffeaa437188487dbd7dc08c7c : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/201a051f4ff3474dbffe605de9e3e57f : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69235a0b94c74cd4b195287fffbe49a6 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40e425518a0844dfb83e1aee3002a4a4 : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73aa11aa3fa44f259d0e8a3b0f116e4c : SUCCESS in 10m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c4ceee07ab364e708f1875df2a61d86f : SUCCESS in 21m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/58fae3998fbe45eeae6ef2c1a432d92c : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/675ef31ecf1a4588b767044d1f96a107 : SUCCESS in 17m 22s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/dd3a63d0173a4e2db2018eceb7077b02 : SUCCESS in 7m 50s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3cdd524ea8154e1e9764309ebebf7a7c : SUCCESS in 5m 12s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7cf3591d9967434197147f97d8544fbf : SUCCESS in 5m 02s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/afa008c73cf24c68a4f03fb7147d0497 : SUCCESS in 10m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ee02f5967884444f9742c8cfa37083bf : SUCCESS in 5m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/76a08f8a172f4f9882febf695844c564 : SUCCESS in 5m 34s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/93ca6ce8833a42e2a2b06f2b9630a4b3 : SUCCESS in 9m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cf17e86215194c9e93cabd1bb03cc3a1 : SUCCESS in 13m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d069c504d0404500b07813cfd5689aef : SUCCESS in 7m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fb2c1c223c1d433493c850c7a97e3b2b : SUCCESS in 7m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/355b7e058adb417bb7f3696a93652faf : SUCCESS in 10m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/764171e0e27a460a984e7088aeb5612e : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6e961556a29b472ca5f3e2a0fd5118dd : SUCCESS in 8m 23s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ea5fd5a37e4049d9b5d401df67dc13f1 : SUCCESS in 6m 04s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/6941a6ea921e43deaeef162956adff1a : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f321a2a686c84a9ca3da9af2ab8c7f72 : SUCCESS in 20m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f3606ebf4a0e4f73b23919e95422d9e6 : SUCCESS in 6m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5ae46615f7eb4022a58618451c7f2169 : SUCCESS in 11m 23s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5e0d81827aa242fc9c13848ec1013c0e : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d951295577714c33a2f08d57010a4221 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95192ba2861148d1bb017b020be7b31f : SUCCESS in 21m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/73f5cad6de554f2f91ab0f7e8869f881 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1821163935f143f486a93a1d1906a6a3 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02a747a5ea17453abcadf90bf5c1c96f : SUCCESS in 5m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a055af159c644308b2d3c9096c9c0f6e : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ccbd51f93384029817e477aacac842e : SUCCESS in 31m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11b8d645b6ea4513b66688f5f3c865fe : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"29ea47bdd55d3340e5b84e107a6a9d8576c9aff5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-02-06 13:50:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9efb8dd347fbaac3f2f49020ffd7b9f418e5fd98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-06 23:51:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"258e6f84a87f91c9ee1b3abab0dd3e40705acb52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 23:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d9e054d52428d75e58a03a7cf13aed9e8ba5394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 01:09:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad21ac19db2546638a2803abbe72252c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f5e060aec044dcafa85ad04aee6f5f : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f444402de3614cc380ca646a3b3b4879 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adc31305fa774748a5f755e506dc714f : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e5d05f9db8b4af4a0ba65aa63283802 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86d7018ab3b545a1ad35358c4819dad1 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/715439133dbe49eca1613110f19023ff : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a121fbc44f1b4e8691538cf16c37d22d : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/865d6605795a41828a7de29ea88f1f26 : SUCCESS in 15m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a19d7a8b681469c80c69abaa6c434a5 : SUCCESS in 9m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/86b83b00a6714535883067cbfdcd73d9 : SUCCESS in 8m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0cb553c15b8e4f9a80355d488361b54f : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/697ed418f27d46378965f5b48a773028 : SUCCESS in 4m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8f6c1eb78b774fc09f10fa5d3df9692b : SUCCESS in 6m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/631f47c31e684e65bed8d6b087d1b8ad : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f072e6b9463c4f7fa101fc1a4e3f9b9e : SUCCESS in 6m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f01011554b5b4f65a34f6de49b205d28 : SUCCESS in 7m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d919259c69804294beff17b5ecf38fe7 : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2006f54c6d9245a3a9b2bb16e4d8e91c : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d7ef9982f89748a782baf7bd07d56b25 : SUCCESS in 7m 02s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/171876a8f2bc4d5fb330485e3b9d67bd : SUCCESS in 18m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed15255fcbed4033a01d258129bbefd2 : SUCCESS in 5m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d10baff2fa0549948719d58035f76bb1 : SUCCESS in 12m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d14a55a9e13e4dcf83535776d2177076 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a46e4c93ea5a4ae4ba64a4be6e2400fd : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77f60139ce6f421ea10d1ba944d32434 : SUCCESS in 19m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2e11d0af3aca4aacacc930bf47fd888b : SUCCESS in 4m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7254d502c6e14d749d6daca434152211 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b22b553e16249ac8f831fb53aef7e00 : SUCCESS in 6m 05s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/37e4901123b9483e85fd505c99aac33f : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eaa1a1620b8d4f48baa029ee48dcf55e : SUCCESS in 19m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc598d16abf44f4bcb6026f70633494 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9ca3655ec962243cd2413c5fc7017ec3341dec3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 01:09:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5f9f11629ddc99c7e8ec0a9d2292ec57b3353965","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 01:09:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e6e0cf90c184b7ab94bb79fa5c88c68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7451380fdd4e77bc46a48cbbcc6990 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d8a9890961c44ac9f1b991db61b87baa653293d","revisions":{"1d8a9890961c44ac9f1b991db61b87baa653293d":{"kind":"REWORK","_number":1,"created":"2024-02-06 08:21:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/907962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/907962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/907962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/907962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/907962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/907962/1"}}},"commit":{"parents":[{"commit":"677c43f96725a39dcffc84807692ee0d28d058ed","subject":"update constraint for python-openstackclient to new release 6.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/677c43f96725a39dcffc84807692ee0d28d058ed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 08:20:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 08:20:27.000000000","tz":540},"subject":"Remove unused abclient","message":"Remove unused abclient\n\nAccording to the history this was added to Karbor[1] but the project\nwas already retired.\n\n[1] 1766bb5123c860c48c14a8eaecb02c629d50228d\n\nChange-Id: I03c39c342794b62ba13aec88a833e1bc5bb87eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
