)]}'
{"id":"openstack%2Frequirements~908435","triplet_id":"openstack%2Frequirements~stable%2Fzed~Id55ce5911072e700af2f4e2a305f58852a02780f","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"Id55ce5911072e700af2f4e2a305f58852a02780f","subject":"update constraint for keystoneauth1 to new release 5.0.1","status":"MERGED","created":"2024-02-08 14:43:17.000000000","updated":"2024-02-09 07:16:53.000000000","submitted":"2024-02-09 07:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908435-new-release","meta_rev_id":"0589ccf80fcde8d5a3d9b17baa7548d100d7b93b","_number":908435,"virtual_id_number":908435,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-09 07:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-09 05:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-09 05:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 15:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-08 16:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 05:19:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5d68c794cc47378cd87e09c45d199ef520f49527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-08 14:43:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a92bfd503efe713e658a13b7ce6e8c4f199d7faa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 15:04:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf29674e118b4078ab7a095d30a71c71\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c5c965b18f0341388e1251334dcc66e2 : FAILURE in 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f0e5cbebbf83a32216b77f35466ad6e4f506c149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 16:06:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a486f5906ee3463087867f37e350d6e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2211bccf10140318be56805584ea896 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9ecc00014f4750840f7be7bfc7146e : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0823fd8745340dfa4d087f87630d94c : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6308306d75bb43b18615e04e7c267181 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/520a3ca7257c42cb9faf0222673f46c3 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90eb40c5c4d643379f22facc73a6bd8c : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11ad8b901cdf4dacb749ed6898507ed4 : SUCCESS in 12m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dabc90291b8c4c80ac6c041b3be9588d : SUCCESS in 15m 05s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/b9b5f4cea8de4cd78605649477b9836d : SUCCESS in 7m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/826f578b96104cb18be7c11304aeaf15 : SUCCESS in 5m 08s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/7071ea9745dc4f209617254d24fea439 : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/27e4a0bd73204f2798c2cd7676a9cef2 : SUCCESS in 9m 46s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/16ec7a757e274226a79371b52eb8c412 : SUCCESS in 4m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf75dcb4d2b74164a2142f710ea3f217 : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a53dfc567e8440a086c6371e6c71fb79 : SUCCESS in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/172f3bd1de7649bb85ba17dcf0fa3c0a : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/085af169263d4ef480852229e589224d : SUCCESS in 8m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/764daedc3fe6405bb91061ffd541cbb7 : SUCCESS in 9m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/89ababc434aa4cd3abe8bb96370483c4 : SUCCESS in 12m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/964059be7b7e4f1098ec02cda0eb6ced : SUCCESS in 3m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5c234d2f370343f5b2289b791e9d7397 : SUCCESS in 8m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ad7e9e788e904ba1b642620495655e11 : SUCCESS in 5m 11s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ffa79994986a4dd484f2aceba02fe862 : SUCCESS in 10m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/198448c2c6b240089699fb8e6fa7a742 : SUCCESS in 24m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f0ed391a1fa3419d9e03a1d2012c7993 : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/472c7a26c50044b08cbef9657b82f3a3 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bae5f0adfb9e4bf19d04aec7da726988 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d870d0024d474accb61150ce7cd4ee00 : SUCCESS in 22m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4fe2a3e1eba94f09bcea93f9a196bb9d : SUCCESS in 5m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3bd4384ffb654b028ec5b95a273a71ec : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af7066547d7948e398afebd3ce27a4ad : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5eb9d6f2bb64b45914c459f42af2a37 : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aece3ce3c0ce4a3f9a62f580748ca63f : SUCCESS in 54m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad09f392bddf4587a26941750e4c38a5 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f79ef5b5073cdf391a14f4f2eb3cb88280c272ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-09 05:19:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ac440a95f0a9befe0e9cba92554eba04175a3c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 05:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39174f6160630c494ef203da80e896f490ce1cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:16:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d98824c455b44b0b72d04f8707079a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5ee85692e94bc48de0be7449508565 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f4d280b206947abb32bd56fcae68659 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62c669e717774e35bb364c618f8b7a3d : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b62c553881764e4db55166c1b205503c : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf42f2af5c9e47f097e804bc5fa8ee66 : SUCCESS in 4m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08a3e13872344d24946839ad437c2923 : SUCCESS in 8m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c5761f85ea9e476ebf10d0fbbc32ebcb : SUCCESS in 16m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/19596e4347da4c8184a73d8348e3a1cd : SUCCESS in 10m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a155db8b961e4514871860c2d63edfc8 : SUCCESS in 4m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f8fa06633284667a79f6bbf0f8a24eb : SUCCESS in 5m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3b9628b942914660ae3fddb4a408a308 : SUCCESS in 8m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b0aade48a08049bd8e29be060256c1a0 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/593dbdcf1adf4c64b54b99c41ca0f0b1 : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/acc7f3faf06e48e5a2cd6c8b57b623d2 : SUCCESS in 10m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5a62b695b37b4a4eb9c02f220335e8ea : SUCCESS in 12m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4fdcfb9da2754c2ca45139fb46abaa08 : SUCCESS in 3m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2b01132b2b8f414fa7ede6f79c20ffb5 : SUCCESS in 10m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3730d2c6c42644fa9f8d8cd1404a3437 : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/74dcd5ca69dd48dc986c1872e1f5905c : SUCCESS in 22m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/649110c2e5c945fca2fcad8756e5cdcd : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6af44e75742c454d83093fc71c85b4d8 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b3a3772a6344dc0ba92f5a86e56118e : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ab1956e707d434996ff0fa39441ad30 : SUCCESS in 26m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d9e1e181028c4d4dbc73e05d0776572f : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1aae4b243bda45d3b04f61efd4c1fe8f : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41c22cc699684b7ba2b0456646116592 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2e46842ed6a849df9a2e54adde7bf05e : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad09b62e0f5646b8a78665986236038a : SUCCESS in 43m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f128f9b063041db9d215dc951fb6874 : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0589ccf80fcde8d5a3d9b17baa7548d100d7b93b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:16:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a7188b00d3d0fbead74b8a292fd92bd7b5a0d3c","revisions":{"1a7188b00d3d0fbead74b8a292fd92bd7b5a0d3c":{"kind":"REWORK","_number":1,"created":"2024-02-08 14:43:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/908435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/908435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/908435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/908435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/908435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/908435/1"}}},"commit":{"parents":[{"commit":"4e030edd0f4a687eef4dc3d6aa16967b300bc5e8","subject":"Merge \"update constraint for os-brick to new release 6.1.2\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e030edd0f4a687eef4dc3d6aa16967b300bc5e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:43:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:43:15.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.0.1","message":"update constraint for keystoneauth1 to new release 5.0.1\n\nmeta: version: 5.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I715ae59f2a9c77c3339865e2de106f8fc6327830\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id55ce5911072e700af2f4e2a305f58852a02780f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a7188b00d3d0fbead74b8a292fd92bd7b5a0d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a7188b00d3d0fbead74b8a292fd92bd7b5a0d3c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
