)]}'
{"id":"openstack%2Frequirements~908437","triplet_id":"openstack%2Frequirements~stable%2Fzed~I022cdc9abf3a56cca3324862565eb50a8e5b78a3","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I022cdc9abf3a56cca3324862565eb50a8e5b78a3","subject":"update constraint for python-openstackclient to new release 6.0.1","status":"MERGED","created":"2024-02-08 14:49:48.000000000","updated":"2024-02-12 02:11:30.000000000","submitted":"2024-02-12 02:11:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908437-new-release","meta_rev_id":"7c8a65ca9b289e1569d2d28f9ca8b26e09f2dd5a","_number":908437,"virtual_id_number":908437,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-12 02:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-12 00:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-12 00:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 14:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-08 16:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 05:37:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-02-12 00:49:58.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"879d9f4ccb66442fa7e76190253a6168bbae85f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-08 14:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f171cc789308a662c03f954cb4593390840b4ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:56:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf1750e7305479583a2e0ac6dd6b0d7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f201ff48d4d54410a73b5a2518036afc : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5cd9609739d2c0d254ef57741b0322cfa091ae8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 16:04:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e10cddfd2c9d4b9989880d4ae776a597\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100aeb40388643b69181545cd6eff09f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/411f1899105d42fdaf22bcd2ac9a9b76 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fe266554a4a4de59a57409b3ee7e57f : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ae8efacb431460e9cf837d9b3e07cd3 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e7f3ad8fc19423387278125ae7a2040 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/681e9f6f3385445395f4a48292032b8f : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/043702e979c0485f98aa0b7a28f6d7d1 : SUCCESS in 9m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/21bc5634db5d4a5c9d015ba7651096b6 : FAILURE in 5m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e6062a5ad9094edcab74c9d48e2c2f95 : SUCCESS in 7m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9d356e0221df467290a6ed6912c73281 : SUCCESS in 4m 58s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/fb491e76f4804fe0bff43d1a6e52facc : SUCCESS in 7m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3bc7afa1f4304d2d8bdd04f529fe6f36 : SUCCESS in 12m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7c6e44f8e3ae46689d3ff01431576338 : SUCCESS in 4m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e8ff08ca3dc34ef6a12a40bb85d0a54c : SUCCESS in 7m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3684ef77a1b9466392261c466f554232 : FAILURE in 3m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f7e31e6452ee4abcb745c5e4827595ee : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cfe3739238f43509f5c1e7ce881a521 : SUCCESS in 6m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca460aecd9ca415d934c80eb685b681c : SUCCESS in 7m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b534e2e0095e4a2cade606bb60c28d00 : SUCCESS in 15m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/71c2827eeb6e4d3eb0e275c4e273f17d : SUCCESS in 3m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1306083bb6174e38bec9c430c6ee339c : SUCCESS in 8m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bd8b48bf6afb48df9271d61e22dadad2 : SUCCESS in 4m 34s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/385141fa77274a18b67c00011fc9344c : SUCCESS in 9m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0ba2ee79853f4f4a9511fcdb837d8aa8 : SUCCESS in 20m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d62e720348e6428fa825e7edfa1f3b06 : SUCCESS in 12m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/42d3c64345e34a2fab3b21350860ed69 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d14c9e5da4904e5480ef34b8cee547dd : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c072becc42e744ebbc65b904d97df8c6 : SUCCESS in 22m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/57f11a20d5e84a32baaad0c64a74b9a5 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/59e919bbb1014c9db6e0c3de03f65651 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b69fae2810a4403bf425f70c9902cd3 : SUCCESS in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e69bea1221e144bda51d33cc2487f479 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f82d7f4032dc4e42a582a067ba3cd232 : SUCCESS in 44m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f62e0a923f9c42048b598afb4f1f8858 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1c85e4f602b42c16d7310b98467bd997cb51fcfd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-09 05:37:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e01d0aca84bdbe677b0b0eafb3f7e5fe3003d7fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 05:43:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2510a654ae8a4ff6bd7bc83717447b51\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5037683463b34711b782ce0498a28c42 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9ae0b8ed22c46bb3c1492bfecfeb7e788e2e7d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 06:59:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d5f943eef94fe290798c41063e99ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72f255d7dcb47e59a6593db5c355782 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c6677ff245f4eadafe45e3672ec3dbb : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/907f6aa796e3468a8a3b890b40488b11 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3335ff1eb838497182e8fe886af4157e : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0377f725bc27422990fd1f0238c8ff75 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c69e31f041014b13bbd4886b2f3b1e08 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dde817985b4743dfba307dde0f13ce57 : SUCCESS in 11m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a7223e7575b64e86a941134354a47529 : SUCCESS in 13m 40s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/85f9c7bb36b04b73aee41eb49ab50d84 : SUCCESS in 9m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f2aa3cdbfe7c4606a51219a7dca1bea5 : SUCCESS in 6m 26s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/261e61c261764b44b6331e59b24167ef : SUCCESS in 5m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2bf4c2a894734f789ccea255cb05a5a7 : SUCCESS in 11m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/78ea8de0971b437e896b2cedf7de4a9c : SUCCESS in 5m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/76199cef57da4e6d95c8174923817e30 : SUCCESS in 5m 41s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c90105fa349f4df9bcac80d12d613588 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ffccabc0112349538da725f684fd8c00 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1627409f4581472faf3dbcc66375fbd4 : SUCCESS in 7m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eee9fa66730e48e5b922f234be127956 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e3cc2b44e4434d9cb7d0018af5844242 : SUCCESS in 13m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/560fd7b7c3854f9b9a9f78aefeb9939e : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/20a0fcf2238d4f1e93a3933dd5eecdfb : SUCCESS in 10m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3929c66402aa4efd8b4b3166d17bbd32 : SUCCESS in 4m 23s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/f535edd31176490aabfbb28ba04510de : SUCCESS in 8m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/161962b500c64749ad9906ac2c29f1f4 : SUCCESS in 22m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8c9470f2ccdf46718eb9c61de78cb503 : SUCCESS in 13m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/361dd0b7dc68444286990d2520d9de7a : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/defa72deb65b4efca3d79a677a187b72 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f3719fbf9f342bbab3becce8ef81e06 : SUCCESS in 22m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/40c21a5020e04dd7a8c72e82cec6eacb : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bc5315f62a3440dba32589c1df63661d : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c41198c91d52449bbbc70b39fc804ef1 : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a6f4cd4e433c49dca79e303eaabb104c : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c96d2c4b1e0747008b7bd85b359ffbcd : SUCCESS in 56m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d57639bdd3741e09ed084a7f616aee1 : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5ca5ec09638ca8ebb6ef4f7bad4903308468d839","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 00:49:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee4b4ef5236192abc9e14e0ab4b072b99699679b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 00:50:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba8dbf9861bac6365131b91a7bc2ab462d11bcbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 02:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/766494f0d72a4383b4fea94a24fefd82\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414cc05ff6694eacb1a121be024cf279 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84119fa1685a404e9013e2971138bda3 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55d5e74c2654d9ea57a7632857427c9 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a487d118bf8c4a5ea6b698ff838ff3b3 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f43be3b16f343acbb3afc11150bd543 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/675a876ec25c4f4483cce02b358f1a23 : SUCCESS in 9m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fd566440661349aa8d41427808cdb085 : SUCCESS in 13m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/597946e7bfe54aa6806355ca8737b900 : SUCCESS in 11m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/03975d1ab0204539b3fdf91a28159db0 : SUCCESS in 5m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6b554a23d14d45bd9f233fe38dbc264c : SUCCESS in 4m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/12880bdaf1d643339b5e34b9effb2ddc : SUCCESS in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/28100d7f87844588bbfa7b8c278bce9a : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59733bed389c40b0a6cb08261309b36b : SUCCESS in 6m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33a4f8f4b2bd4479a12b421d1826ea0f : SUCCESS in 9m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2f7a43ac69a04b949ef0e626f388c0a5 : SUCCESS in 12m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5537ef9a4b884351a1abee7eed2e3c6a : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/16eb99fbfe02472485c199e8c8450fba : SUCCESS in 7m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b4b5a70e2e2c40bebf1ef57b056807e3 : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f4e51b56175c49099e0589c8cf466fe5 : SUCCESS in 26m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2aa1f3aaad9642039b781a29736a6e5d : SUCCESS in 9m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b381208f4d184f87bb2cf6736060c369 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd97d1733f994191b265bbafbf2982a7 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81fd6d2abbf7438e8b75bc09211883c7 : SUCCESS in 22m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7ee56bff4ce145a7981a638998ceefa4 : SUCCESS in 4m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4e2e5d8a8a994dd8808d09e77417cfbe : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/634e47523af04ef28b96d9b05307d7e9 : SUCCESS in 9m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3fcf0135bf8543bd89ea7b66e83b284f : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fab5a31766d4b40962cb15ebb2cb500 : SUCCESS in 34m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65443d31e83942dfaaed2a805fb25cd1 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7c8a65ca9b289e1569d2d28f9ca8b26e09f2dd5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 02:11:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a07d5804e5d47b9d4672681ede6242bc03ee18bd","revisions":{"a07d5804e5d47b9d4672681ede6242bc03ee18bd":{"kind":"REWORK","_number":1,"created":"2024-02-08 14:49:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/908437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/908437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/908437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/908437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/908437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/908437/1"}}},"commit":{"parents":[{"commit":"4e030edd0f4a687eef4dc3d6aa16967b300bc5e8","subject":"Merge \"update constraint for os-brick to new release 6.1.2\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e030edd0f4a687eef4dc3d6aa16967b300bc5e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:49:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:49:47.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 6.0.1","message":"update constraint for python-openstackclient to new release 6.0.1\n\nmeta: version: 6.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I08de7eafa4db755bd02cbf51c418216dd112af10\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I022cdc9abf3a56cca3324862565eb50a8e5b78a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a07d5804e5d47b9d4672681ede6242bc03ee18bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a07d5804e5d47b9d4672681ede6242bc03ee18bd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
