)]}'
{"id":"openstack%2Frequirements~908440","triplet_id":"openstack%2Frequirements~stable%2Fzed~Icc8ef9dd94cdbadb5821146bd1a8883a77bfa78c","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"Icc8ef9dd94cdbadb5821146bd1a8883a77bfa78c","subject":"update constraint for os_vif to new release 3.0.1","status":"MERGED","created":"2024-02-08 14:51:40.000000000","updated":"2024-02-09 07:16:56.000000000","submitted":"2024-02-09 07:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908440-new-release","meta_rev_id":"b0ff98ae926644107994482076da37a62d0e7d91","_number":908440,"virtual_id_number":908440,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-09 07:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-09 05:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-09 05:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 15:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-08 16:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 05:19:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"442edab8aece410fa39eccfe5ce8f962bcc60d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-08 14:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1c17cbc45baa643c2faa2681b3a863e33026ba5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 15:04:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16290bc68bc54482a17ae29164e225cd\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d850fef2e3b49d1a591035004302659 : FAILURE in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"19e1dfda0b54f6f9fd9ab216788ebc0564e64d3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 16:28:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2402e918e34cefbe0d8df01e29b2e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af08832766a4e7bbf5842ae7aa5408c : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f61247cb66dc4ea790775068c2c53a6d : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4535990d81e44f35b8acb16d86cc2c8d : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0dabc502891842478bf5207e72a5b619 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f310e5a5f8384842ab119dda5bf0b955 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e3a1cc99c1e42d886b8edc96bf26fb3 : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/925a13a7c85e428cb2ea5dfba767c356 : SUCCESS in 8m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9af17cb6d4bf465bab75fb558e8869c6 : SUCCESS in 12m 33s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/da94f749fbea48ef92a5a8c973efd1bd : SUCCESS in 8m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ffae1825b8694a1893e3afd60046c070 : SUCCESS in 4m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/92cd6a3d6bd8456d9e5e53808352a3d2 : SUCCESS in 4m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8e502e8b89d44065a1eba2362f17a9bd : SUCCESS in 10m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/028699bb73024bd2a40c6c3ba2fc22ea : SUCCESS in 4m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1eb12a849e6543feb381e204b99f349a : SUCCESS in 5m 12s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c3e39915fb1540e5ae60f6702d115ac2 : SUCCESS in 8m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/90365203efc14700be29576c34001122 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2ca7844499f4c60871bd3b3661e194e : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ceb2b5370364fada0b279119232775a : SUCCESS in 8m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e535e6283fc9408a9fac3209446bac32 : SUCCESS in 12m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cc18ccd9c3ed4f4a8d576ac3d12222c6 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f109001e56714eb5aafac6010770c093 : SUCCESS in 9m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/85ec46e0e6be45dbaa739d36ab86ca49 : SUCCESS in 4m 29s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/74a2d118b34240f6a097b07bbb11ba56 : SUCCESS in 9m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cb12399f36d646f683d8664ebf9a84fc : SUCCESS in 23m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fcbacefcbcef4b10b4f6d35be339b7a1 : SUCCESS in 14m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b4df77405f5544f8b4f0b10efc9b48f2 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b51cfc5b61af471398cfa42d32e0283e : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42b71e5d084e4cb98c5fe8b39114862a : SUCCESS in 21m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0a80ee32fb174ad98751b6090cf99be7 : SUCCESS in 5m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cf37f311ff994fb2bad8780a40d4b949 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dc048240e914e789a539cd9a6d50df0 : SUCCESS in 7m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ac6f00381d24dc29b3127dff153e51d : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dabce03ff0214e3e898b2f6ab8c28acc : SUCCESS in 58m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7834d4de800d40d59a04b62ef98a695f : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fd1dcc2ed90c3a00f672734986bc3c00f35b1ca0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-09 05:19:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d642c10ce3123de9c8d195f3fd4069387e4808f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 05:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bdb5fa4550062d2db4c450f30f285c00bda1cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:16:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273ea34d74e541859c2896a7f14b291a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d59e647aef4e46bdbbacd8dc728154 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/441bbb901e41480b903f2fa8bf809365 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0059c58e8d8f4da5996b55b53d4849dc : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fed14ce042124b2aa45d90a9caa025f1 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5ef8915415a4b1a8929e5c1c1481a78 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/26153b00be8b4e2884174ddbe3e1293f : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bbc63854d882401ea5675e9ac8b99942 : SUCCESS in 13m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f9f0933f672642abb87ee43c31635d84 : SUCCESS in 13m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/554bd669c650472a80d92b3cb8709084 : SUCCESS in 4m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ea98b4aeaea44619bc5c10f931a20e6f : SUCCESS in 5m 27s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3615ce7089be46f3b032e1564fd71a35 : SUCCESS in 8m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6be426db65e34aa491a2264fb088bd3d : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4650ef1c33e48d0a1d2bd101b79990c : SUCCESS in 6m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4ceeccc41bc4499b7b08a7a182d71a4 : SUCCESS in 10m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb004705a6d74ea0a415d9653eb7774f : SUCCESS in 20m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0ac2a2e89e9f44469c99ae24c38f4fbe : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f95eac44c5704299a713355f4e338c83 : SUCCESS in 16m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ba9a799b6ac3429aa6299ff2f276ffef : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a7ad222277534f579b6b4351cb17d67b : SUCCESS in 23m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/11b7a5da1c0e41ec9a5f6a9f0d4e076b : SUCCESS in 13m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/97ff137e0d7741ee97d1acb8baaaa085 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/667aa3fe54754e92a602a9aaa8d4a1ba : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/513dc2f3b1e3470faa3460d331ff8b92 : SUCCESS in 23m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5e201c5157ee457d86bd5980ef8391e4 : SUCCESS in 5m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/48a55c9f78b446a4af44fbcfd3e25d13 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cebe573b07eb4f2cb18de6aed53eec8f : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/163c8db471ec4abda5cfcc7988bcc2d1 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3e00099bf064800a05d6fa203c50d1f : SUCCESS in 45m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1726966bbb34814912d1afbd3a5c7ee : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b0ff98ae926644107994482076da37a62d0e7d91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:16:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0e16b0cf417182ecd99ff53ba66818f657fcae5","revisions":{"d0e16b0cf417182ecd99ff53ba66818f657fcae5":{"kind":"REWORK","_number":1,"created":"2024-02-08 14:51:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/908440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/908440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/908440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/908440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/908440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/908440/1"}}},"commit":{"parents":[{"commit":"4e030edd0f4a687eef4dc3d6aa16967b300bc5e8","subject":"Merge \"update constraint for os-brick to new release 6.1.2\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e030edd0f4a687eef4dc3d6aa16967b300bc5e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:51:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:51:39.000000000","tz":0},"subject":"update constraint for os_vif to new release 3.0.1","message":"update constraint for os_vif to new release 3.0.1\n\nmeta: version: 3.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iefde105c62db4c899fc38b89345e120fdf1d6391\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nChange-Id: Icc8ef9dd94cdbadb5821146bd1a8883a77bfa78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0e16b0cf417182ecd99ff53ba66818f657fcae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0e16b0cf417182ecd99ff53ba66818f657fcae5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
