)]}'
{"id":"openstack%2Frequirements~908441","triplet_id":"openstack%2Frequirements~stable%2Fzed~I4cdd56389eb4463ade57b8b35a5d4503c8b1cbe2","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I4cdd56389eb4463ade57b8b35a5d4503c8b1cbe2","subject":"update constraint for sushy to new release 4.3.4","status":"MERGED","created":"2024-02-08 14:53:23.000000000","updated":"2024-02-12 04:09:23.000000000","submitted":"2024-02-12 04:09:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908441-new-release","meta_rev_id":"35ebb0e94c7ac1a908b3191c4e21c134b6c78bf5","_number":908441,"virtual_id_number":908441,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-12 04:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-09 05:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-02-12 00:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-09 05:19:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-02-12 00:50:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 15:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-08 16:29:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 05:19:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-02-12 00:50:20.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"912bac02ad11ea8b3c53eba9eaa29b279f6d3244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-08 14:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"805ed894e62f7d86fdb04ccde45b37c30c367aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 15:17:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499e358f46a54cf388b01bc24b6c48b5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/438f7058303a4f5e9942f839ed1a6f53 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d73760b97046e6d1a59d3afa39a029d673e25374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 16:29:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf2b880cb0fc4be1a24d5db95a8d31cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0a3015e3aa4d4bb8e5e9a6a9d4b900 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e3682434bee4174b94a7851b7064f74 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f97b6cd172dc4946a9de822edd466656 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec5cabe3eb8d48dcaabbe3c4cc36a81f : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6007374f299640e1a0f5f1f1f4fb22b5 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c53a60ed923748fb80601850400a62cb : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd031b48aff643748c3f355cb9c2a65b : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c356d6afd313448a84d56098462d3d61 : SUCCESS in 13m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/4f62aaea3f914154991ebf8009136fbc : SUCCESS in 8m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5e9ae4a2afec40d083f960c9c36a4736 : SUCCESS in 5m 13s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/23455c294089440b9d8232f0c55d6f40 : SUCCESS in 5m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8f6ef5f39368465197e0ff5dfffede9b : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7b4544ca067c4408808472443d1d53ba : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4c65586dea9b4d728bbcb6ea0393c954 : SUCCESS in 5m 09s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/43f0d029c4b74c57abe5381256d9763c : SUCCESS in 8m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cabea0f5bbd9465990c829f5854ae227 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5b6d8c02a86449bb236304b5cc2efca : SUCCESS in 9m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0649fc464f3d4b01b5572ad5cec0b752 : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d5c455a1e7454079a7ef53ed432f7398 : SUCCESS in 11m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/83972c9a2b5e4f768533963c2ea0e952 : SUCCESS in 5m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c181c75efbae44749a108246307d1b53 : SUCCESS in 10m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eeb32f9e311c4fb282aa0dee05491045 : SUCCESS in 5m 50s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/e1f4fac395d740fdab6a221ea20aa500 : SUCCESS in 10m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/178abc481ae143d8980040ca701a5254 : SUCCESS in 17m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/66f93c1c430342ce87b1e155030b446a : SUCCESS in 15m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fdaadb0df7654bf6b1f893b65b3ac9a7 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e24f19de5d9948c59825c9227156c34d : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3078530d6a7e42268aa3692b5c2c50c1 : SUCCESS in 23m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bc365cc8d4f24aa5852cd2b37ec39a28 : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a826ed5664541adb855a91d8488984a : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf286fc82ef8448d9f902d54bfb3b95f : SUCCESS in 5m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1597aa25d3c348ccbf357cfa4719a794 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c24a597c1f23484caaea96426f9fa591 : SUCCESS in 49m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/919061f96dcf4119a1ff388751073fc5 : SUCCESS in 1h 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fc8031d33452378c6fa1a8de3c8cdf9537bbc687","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-09 05:19:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40f62caa3e8c5933e55aac7e004ab814599eafc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 05:20:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f24c27f64fc21adacebf704b2b2bd1e9ecfe866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:16:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2a3156dec9c4b8286ee8edab92c98e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785c07eb4a774ebfa6b332a2aefd01a1 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e356b3c2680d49c892fccd7f66307f5f : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c748df1cec45b08afd2c8f5b25bda9 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05081f09e5924d6fb1a0647fbc0e3ba3 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b65b15b798454fa78289564a64bbc5c1 : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be7d2310fece459195fbc3700f79af9a : SUCCESS in 8m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a38da164d22429b8d82d5a5bfc3d811 : SUCCESS in 13m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bd0e78fb6c1d414abdefd94c2874dcc2 : SUCCESS in 13m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ebc7383b389e4baa820705d4dadfaeb8 : SUCCESS in 4m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c5b8a4277885482caff7673edba0e641 : SUCCESS in 7m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/df49499bd7d24671877a3abd35c7a50d : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/73198897856e421592424eb4d6a165fb : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fc3a6f82d5b4e07a912b43445c1dfeb : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0fb2f688ce1640d1bf85e398f2654ea0 : SUCCESS in 9m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ff2acb45d514a169619b9482af76b32 : SUCCESS in 11m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ff74589ae69742caa9487aef52e13e92 : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8fd6ed79206f4a1bb321267f63d47ff4 : FAILURE in 19m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0831b17ff0744fca8878dec4ccdce2b5 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ba6cefa265fe4322bd5f4829fa9cd78c : SUCCESS in 24m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dcf6cb649c3142aeab701f7da0de11c5 : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/93e072280048414eaea5e6326a2e776d : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bca3404f35c743f087a3c4a2539c284d : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad47d2040a264b5190d2e26ed05552b8 : SUCCESS in 23m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e09f61e0eef8496186da183b6e3dde61 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/22e97dc37605410fa89d9e311afd6615 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0940b74e31fc4411bbe0d96e4971b015 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e336855b139e49a1ad43b6d4f105a8a9 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d9444a9aeaa467f9dbbcb13fec4d31f : SUCCESS in 44m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f6bb11712494a68a5b5bbeb463cc6d3 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3211969bca9bd4741f3d5b246e1e603f14c9577f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-12 00:50:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0702a1fe23e6fbc2b8c2ce266c11db7b3a715f1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 00:58:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec538283dd1643edb95dd406ed76950a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f919d778779e431ca5c61444ab783856 : FAILURE in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"39913ccbf03702bb76c29422dfc38484de1f76b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 02:11:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70fa6e8cb2c4468c8c34e58aa3133c4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c112ff245f5e42108fde60fa288d4861 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf5fc7e88a384e288f9ab200769dff7f : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6616333632924397901a93d0ec2e589b : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60e91e23ec5e429d83deb63bf4a5eeb7 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bfb9ecee6244f11abc789798c29ddef : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05244f0f0b1e4db7a8e70f482ccc5e8e : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9c2fb06681949048e54731731b1c429 : SUCCESS in 10m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3fe10c50c9c848fb980107f9bf53c5ad : SUCCESS in 13m 26s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/628649a7d80e43d8812d4fce58e9b698 : SUCCESS in 7m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cc0cee3e6ba947e590dcba317bb81422 : SUCCESS in 4m 30s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/063ba05b55804d0cadab18d2a596fee5 : SUCCESS in 5m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2d5316349eba40aeb9f80b51b6fd497a : SUCCESS in 10m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d7b59198431f4486bd897a5723b22307 : SUCCESS in 4m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fd95e46ba6954d36825a71fa0bccb3c9 : SUCCESS in 6m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0101eb671b6c41ae989d366891e6ff5d : SUCCESS in 7m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ab3a5f4c0924f0bba29a05058e247bb : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0155a4d4f8b6463c84a8627bbe1d9a20 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b85850766cc04c3a890a291c7be3dd19 : SUCCESS in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9f8dfc9eb03a46619bfc522730b80ebc : SUCCESS in 11m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/af378b69724e4592a9c93533858df06d : SUCCESS in 3m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9e675d6edcba4fc1946d15d65932c6fe : SUCCESS in 8m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7045009abb654725a74744b622b22db2 : SUCCESS in 4m 27s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/8e3086c0be094588991abcf741cb3ea3 : SUCCESS in 9m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/049334414ca240ce82be5a268e6b042a : SUCCESS in 17m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fb32fc7210584decae55884c2ecccf1b : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f3545d8dbcfb43f3836e5c4f651f7e1c : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96a7d4456be3437d8908d5fa053df52f : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/889b64404ba64a799f838fe871255cfc : SUCCESS in 20m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/036b6b4241c948779fe37c4cfe355388 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/422da7840aaf49a8b9ee6861afc393e7 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/797b457501ea4b9f97dd7d34b5a22cf1 : SUCCESS in 5m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1fd6cbfd86de440e9f248179e4435cff : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecdeb1c25ce64c0cad33528a7272df07 : SUCCESS in 54m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40b4411112294be298843a301f4663e5 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"db5ec5f4512542d86886fe97489346a67278ca9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 02:14:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6440d3852fe56e84d8aa13342f8a71f31dce8052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 04:09:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5abf41e1613a449a8f741c9e438fee36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197077ba23914776bc3e8ae866ffea88 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be37aa4bc124a229101de87c40e0fe8 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78d902d113404e8faba1742340b72ace : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8366108fcd9f4d5597ccfacd4581ce96 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bcfac893beb34edc8bf0896a38a3f85b : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ccb531275c04f8a86a669ecf7c96a45 : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b5d80f7dfaae4665b8af868b56be5749 : SUCCESS in 12m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5827a18dd74448181f49c534ce9b292 : SUCCESS in 10m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2084174ff3d24ce196ca369ea4df6f80 : SUCCESS in 4m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9b252801985e483e8a6f9351cbf7d0fc : SUCCESS in 6m 12s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8b208848742a4076a91f90fcbb2b36b4 : SUCCESS in 8m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e2d14bb4f363486dacbb85e173efe80b : SUCCESS in 7m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1aee99d2c2540dbbb3cc96d0ee07b38 : SUCCESS in 6m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/45b8db803280402fad1db045ce3cd5d2 : SUCCESS in 8m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0420c713155546f4a9c1fae2ad6d3e4e : SUCCESS in 12m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8b930039f8f64ec2b19889893831c3ef : SUCCESS in 3m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/65cc1a839f394d288f375c5b31c8707a : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9e96e001a44e4f3c838668cb7673cd0b : SUCCESS in 4m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/33318b2a70fa452ca7905a08b60e73d0 : SUCCESS in 22m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c09d323a50f7405bbbc5dd4d2d19436c : SUCCESS in 12m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6f39224c6fe04d1fbedd22e2d882de7b : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf5e19031d114a8f9e8755783cc596b6 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebd5c8b171304adcbb1271548ccc5ac8 : SUCCESS in 22m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/32a5d6c7a90247d3b2dcd4cb5704073c : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0d83ccb591864c9fba9a54db489d86c5 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6be77b30e3a6414e976d5bd875941316 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0224beb423aa44bdac22ae2a488ce084 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e59fa480c734744982106bdffe07b19 : SUCCESS in 56m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9af6bb95b34348aea482fcf1eacb18cb : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"35ebb0e94c7ac1a908b3191c4e21c134b6c78bf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 04:09:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17caa1a984d685f0ff176dc1419e6aeeed06467e","revisions":{"17caa1a984d685f0ff176dc1419e6aeeed06467e":{"kind":"REWORK","_number":1,"created":"2024-02-08 14:53:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/908441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/908441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/908441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/908441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/908441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/908441/1"}}},"commit":{"parents":[{"commit":"4e030edd0f4a687eef4dc3d6aa16967b300bc5e8","subject":"Merge \"update constraint for os-brick to new release 6.1.2\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e030edd0f4a687eef4dc3d6aa16967b300bc5e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:53:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-08 14:53:22.000000000","tz":0},"subject":"update constraint for sushy to new release 4.3.4","message":"update constraint for sushy to new release 4.3.4\n\nmeta: version: 4.3.4\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id7bd14336259904280b780b67f82b0c26c74cf50\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4cdd56389eb4463ade57b8b35a5d4503c8b1cbe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17caa1a984d685f0ff176dc1419e6aeeed06467e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17caa1a984d685f0ff176dc1419e6aeeed06467e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
