)]}'
{"id":"openstack%2Frequirements~909876","triplet_id":"openstack%2Frequirements~master~Id69e5808a5e37307830751487537c84c8f3b200b","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-27 17:24:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Id69e5808a5e37307830751487537c84c8f3b200b","subject":"update constraint for oslo.serialization to new release 5.4.0","status":"MERGED","created":"2024-02-22 14:40:29.000000000","updated":"2024-02-28 06:22:19.000000000","submitted":"2024-02-28 06:18:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909876-new-release","meta_rev_id":"dcde032f3996bb8ae8c6425327779583948a6d29","_number":909876,"virtual_id_number":909876,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 06:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-28 04:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2024-02-26 08:46:30.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-28 06:22:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-28 04:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 16:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 18:06:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-02-24 14:54:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-26 08:46:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"CC"},{"updated":"2024-02-26 08:46:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-26 13:26:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2024-02-27 14:40:29.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-28 04:04:29.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-02-28 06:22:19.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"}],"messages":[{"id":"aea3996f1064b1262995882e3e51cc4688f823e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 14:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f078693fe255693a36cbaf045f65b8daf35a0cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a735cde87164906a30fcdf7c2d918e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b86edca2ed04444a978df1b2062bb12f : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4f1440bf7147c6b80985abb4ca6ea5 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40da1789bf64295bbdc2c746f62905b : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b16c7d11fc6840d6af30f76a3b4c66d2 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97deb307c914492680aeb69014b25ace : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e78af8d42b564b6484ab99d2f75ca7f1 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c29850df83f44c9b89d0a5d5ebfa07f : SUCCESS in 10m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/48fe8e9c77df47c0b2ed32104babe570 : FAILURE in 6m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/df5e4d15de3e43fc8a10c3ddeaf4a25f : FAILURE in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7caef0da18d949ad9c034b2b63c5bdeb : SUCCESS in 18m 27s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/40c3317666c34369b9c7911303aaa666 : FAILURE in 2m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/10205dd9f7154552b7b08647592927a6 : SUCCESS in 6m 23s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/933191ed1afc4e64b935bb589315f57f : SUCCESS in 4m 05s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/15908e1c2df84788b284daf5d365f088 : SUCCESS in 10m 31s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d56dfdd321594431bd13f2d08334142f : FAILURE in 2m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a7e8da0c68104bfe95c1b13da7bdb1e3 : FAILURE in 2m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3d6fee49dc74473a80e7cc7de446d7c3 : FAILURE in 5m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c0d437a9619b4d3f92572b31cadf8054 : FAILURE in 3m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea8d622bf45247fbac4469b4583fb7d9 : FAILURE in 4m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/00d13d58bef0489c8818ccf65c8b1b5d : SUCCESS in 9m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/19f0c733cb4245bf90e347f44cdd2359 : SUCCESS in 11m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/426d426206404cbe9430808189dea6b5 : FAILURE in 2m 25s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/535e43d7a6ea42a998a36ffbe16b1a0b : SUCCESS in 7m 14s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/55eb99bd1e054fc286f62b5d3ff1e766 : FAILURE in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/df46d507fa9741a1bc35406900cda298 : SUCCESS in 13m 42s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/33a0339cf6b84b179959bf3a80ce75fc : FAILURE in 2m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea9dedfca3a14e7d8061086440d2f71c : SUCCESS in 6m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/253030b83d8e4948b4d9547f5e5cbe5b : FAILURE in 3m 05s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/42937eb261a546a595ead435c0752406 : FAILURE in 2m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9ac0a538d0d4aa38bd8c7f3c4784cf6 : FAILURE in 2m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f90ad03112f649de8a2685e30c21aea6 : SUCCESS in 23m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8196abefa43844ffad25c9f9d25841a1 : SUCCESS in 4m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6d1b3a09abf545589eba9f90f14e7267 : FAILURE in 2m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a45137f29c94cea872645c8f8e795e6 : FAILURE in 3m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d5ed6342baa94417a652d387780b8084 : FAILURE in 2m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3563cfcad9104261b3cb3c23ce2a1768 : SUCCESS in 1h 33m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff3f3b347a9a4d938c6a1ea1c46d058e : FAILURE in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0da73e9924dfef3e031c9d5055b96b31578f5039","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 18:06:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fc59c7f29d20dca30fadd4a2fa17d3becf811d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:54:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8edb33769b3442d2808b92584f5db8a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee15c935ae74e4382405bae24645bfd : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25f4bc288a84a4eaf414a3e47fa254c : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d329e98090314e83a0d8026e599fde35 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76077f3a6da140968b1cb796932fe084 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf15d59fbfcb489384fbdd927de12c91 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5a968ff76ad4f27bedf995fa59fa709 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ef6d25fee704e21a927d57fba0ee862 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17252f424da04f4f8c8f0d0f0417f98a : POST_FAILURE in 17m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c1468fa9a50840499d9b8df3318b7477 : SUCCESS in 11m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7e4cfe1a462f4947a85c44050eb4c490 : SUCCESS in 20m 22s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/105e7b4f167e4f69a7f6fe565135143b : SUCCESS in 5m 36s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/26fefc6737924b4087e9fda05023bca9 : SUCCESS in 5m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/87aec2d562f14826b436935b8c2bb580 : SUCCESS in 4m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/96e16386f7c3497495c9018ed2b3a222 : SUCCESS in 10m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f506224957404e4184264c46df958eab : SUCCESS in 6m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1651d32c03b2429892c78692a5308d95 : SUCCESS in 6m 02s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c87b1ae882214dd2961842894ffc0526 : SUCCESS in 8m 45s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/55726982338a401488d1c221980df5b1 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e223f0fdd1384028b8cbcf5f6df15796 : SUCCESS in 7m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5e92d960398841a4afeaf690911c5f33 : SUCCESS in 10m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9976d30a356d414fb88b4095ef590a25 : SUCCESS in 12m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4327c185a04d4ee9b4bf8b4f53951b25 : SUCCESS in 4m 41s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/477a655d7b9446bfaa61b74fd4eafa68 : SUCCESS in 8m 28s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/cc8dc0038fbf4daab7282cf737eba8b1 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/977a79d4eb1442fd91e6502178843331 : SUCCESS in 12m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f2ba8d6deba044ba9407372acb5968fe : SUCCESS in 24m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b352974f18f40709fb7b1aeeccb331b : SUCCESS in 9m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5c0e48c237314ca5b463b632b3482c34 : SUCCESS in 13m 27s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/aa26903284be4ff39648731a244d6523 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/464c773d09c945a88ec74ec85d8e39f2 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37c9578e3f5449dfa3e8a8db83852e9c : SUCCESS in 22m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b526d6c127af4528b52920fad9ef7425 : SUCCESS in 5m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/49d34739bc294f2ab784498f80ec2162 : SUCCESS in 8m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98ab1053f13c4011b8fa57455dd0a597 : SUCCESS in 6m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/968236aef7234ec08dcc1a61bb7c0140 : SUCCESS in 17m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05853533df984a2cb7e3f0ef2c96e670 : SUCCESS in 45m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbfc7265837d447fa8fd9acdca331723 : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0b0dc9452cfa09501c3dc7100a00ef1498361de4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 14:54:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84724e069f284ad3eedaf17eba15d4f97d2d6d04","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 14:54:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0af796429326d77fc36b5e62fa28f627a05cce5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:29:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6b6689e47934c9595c05cfaffb44ac4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9d5866e5c8404d96ca897b36e8bf76 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4438df51ff80498fad89e8cfe56762dc : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e74289952e3942fe9a3dd1b908d42f06 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e81081012094acb8192b490307cfbd6 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b3277c3112e485ca138f4fdf04595e9 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61e31d32016d4c769dd3ddeeca11097f : SUCCESS in 5m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc55dbcf4a4040599aafcb9b8a58007b : SUCCESS in 11m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bdbcaa7640fd4d94aa7dc076c1c09944 : SUCCESS in 10m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8df3ab6981134b7598d281e404644160 : SUCCESS in 12m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5c66ec9a069e491ba4e11d21ebaa5777 : SUCCESS in 14m 19s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/44bbf9fd2c3c4314ae085b8edb1a0e78 : SUCCESS in 5m 05s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/16404714ba3e4cd4b2bad740b87ea3db : SUCCESS in 6m 02s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b116058077b04c45b1c75284480ede1a : SUCCESS in 3m 48s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/34c381785b3247bbaca3532ba1fc2b68 : SUCCESS in 8m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e2710e07ebc444c6884225d67e809172 : SUCCESS in 3m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6efab338965b44518f8f9909a7144453 : SUCCESS in 5m 34s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cd622abda83e49fb9f8638380855dc3e : SUCCESS in 7m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c1591a214e774925a3caeb95fac9ba76 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6442d6861294bd8953390df9eb44c62 : SUCCESS in 7m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6345b1a0f58845529961fe421aceb712 : SUCCESS in 9m 15s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d74d889264d9444bb740d0aee3e8bf01 : SUCCESS in 10m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/fa2f270f1a26499f8701e42b01764e8d : FAILURE in 3m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/e1a1ec54f9e84541b40b863ee6e01547 : SUCCESS in 8m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4ee78f0540d44373b54329fa809d753d : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/29ef492ddfb24ff4a70141b0c9c24330 : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cf91e47594b04866bd7a2cabe879a715 : SUCCESS in 19m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3c98db9cc164508b3eabe3fcbd0be8f : SUCCESS in 5m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0a7bddb9faa9415a8e5db40fd5746b96 : SUCCESS in 10m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2ef3855270954a4ca6a9cde743d2bee0 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bb0a646cfc94d6f8791127bd842bd7c : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f7b9b9c62da46359cbbbf1aa2d5d5e5 : SUCCESS in 21m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a5d56b3926384538aa0f393b68a9acf4 : SUCCESS in 4m 26s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/28cf902d6d234476a31343e3dfcf951e : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/429601cadf324cde87b74d4f4646727b : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/678530895f884325ac459dfb98afb07d : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b9317a5564747619e2fcb237beb4491 : SUCCESS in 54m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04d47b87dace47ecb78d92ca3177fe98 : SUCCESS in 1h 28m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f5462ef92da4d0d194732bc3d6f86c69b88a4e06","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:57:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d3c8b4c12c2406c9ea02a4191acd0f8\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01f58af52d084b279047e3ae8a6999ef : SUCCESS in 2h 02m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/73323b7eae7b40f5bd6a0c63785e09a4 : SUCCESS in 1h 10m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee1f949e80234282907ca067014495db : SUCCESS in 1h 31m 41s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/6a9d5750c99a4bb1a6a6bda26f194ed9 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"39478a12082a1a960214abfbb868fac1982fc629","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 08:46:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89500cd3b291eefb241b506f3f4dbc0997b157c8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-26 13:26:05.000000000","message":"Patch Set 1:\n\nWe\u0027ve typically considered that a bug in the consuming project (kuryr). Any directly imported module should have its package listed in the project\u0027s requirements. Once upon a time (2014?) we implemented an experimental checker tool to find these, but it was based on calling some internal pip functions and so was ultimately abandoned without the necessary rewrite. It would still be a good thing to find an automated means of checking, as it clearly does impact our projects from time to time.","accounts_in_message":[],"_revision_number":1},{"id":"770527df17c4c950d31cf923fee82a67c5f56e92","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 14:40:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a2915e1cc25d378800152438fe2727a9c2b1d85","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-28 02:33:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fa93c644f1e0c6fe3d257416feccbf738e52593","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 04:04:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"589231a77a2701c99ff75b50cbd7f9cc46e7141e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:27:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524b8ed4b4864db199521ce650abb165\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abc08cc5b01426b806efcd0154a7ae4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5666d6da9224805907e13078f0c163c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5b60228741542c6b03dd2bf1fcfaae6 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acaf37fa806e4734a10d59c4d06b5945 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ade33721bb74d16990aa760dbc38a17 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22e8daee4fbd45108341b761e4d05bf4 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcc3ce0005a7438388097e3d0258448e : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8982878320c940139c06bfd2733f11a5 : SUCCESS in 13m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/49b50dc864c04cdfa4650b9081d9e074 : SUCCESS in 10m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b4fb247c03b4f81b00cb00e8ea49f36 : SUCCESS in 17m 54s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1433595796b841fea5e3aaf8c68808f5 : SUCCESS in 5m 06s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/cac6b934ffb8485c87c78f53dd350c65 : SUCCESS in 4m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2af34e1ba4714d69a22470f75f78e90a : SUCCESS in 3m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/84c2c07064334712aa574fa8def11a10 : SUCCESS in 11m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/504c6cac91c44183aebb46693457d035 : SUCCESS in 5m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d0f53832044d4250ab3cdda3bd539def : SUCCESS in 5m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/04e35e2afd2e46ba8ea13a36aad5e5ed : SUCCESS in 8m 09s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1a6da5816aff4d6fbeed6985ec318b18 : SUCCESS in 14m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27a1e8df1e2a439bae09d63cd927ef10 : SUCCESS in 8m 20s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b1363f6f02b24b5d815806b24d6624d6 : SUCCESS in 8m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0bc54877d3e84961aafbfac97eb41f30 : SUCCESS in 10m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c2e06ff691664a0082a7e469c4cd3769 : SUCCESS in 4m 20s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/31d728f03b924bc2b575592045218bbc : SUCCESS in 10m 13s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/444e3d49e5384b378ea06b487d000c64 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7569738e28ff4e55aab11d68fc6c7507 : SUCCESS in 11m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d986064a98f04804a8ec8942d7701715 : SUCCESS in 22m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/193467665c8c4044a0a76b86b319299b : SUCCESS in 6m 33s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/191f04378c0e4373b5bd38cba377cc3f : SUCCESS in 10m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6f7d9e1dca3e47d3b8e970915b9fdf8e : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ec60de4de9c4d05a3bca4c98848d1d8 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5da592fcf9774f37b3421e0af71c2a55 : SUCCESS in 21m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/fd29e3abf4864dabb4c5c713da2516b8 : SUCCESS in 5m 08s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/176680b6740042c096d48fe109e23e01 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89f6a2cd771b4c43a5358e4162546fe3 : SUCCESS in 6m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/fb13c749eeeb46a3a93c9f10d0705689 : SUCCESS in 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e639f5118c1e48a8969f7dcde908c94a : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/593b5d6d8dab45a48ef0fa15ba0de04d : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bf597be01c78860066cd9ed9833a666a5e049abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:28:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23d554ffd6084e6d28337e3e5c87a21382bdff35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:18:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d1770470aa4388a1f7761432ca8d40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb3b6d39ce1435bb66b923fc834c1e6 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5eb24c2dcab45e59d6fcd2711e39cb2 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7049114977846b980aa3d30d7b3dabc : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fb370b82dc941c986b0eb6475d1e83c : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1903b36faf6b40e693042f9edc404c38 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bede12d150f147e3973c190652ca859d : SUCCESS in 4m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66c07a61df9a4026a2461b45408873bf : SUCCESS in 10m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2fb964a8e9d94c07add05d7a454f354c : SUCCESS in 13m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ba5c482a5954e94b332ee91e6409a43 : SUCCESS in 9m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f4f36689d21842aab9b49e1b36e7935f : SUCCESS in 17m 41s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/37cacef48aad4bfa9464ce0fea12801b : SUCCESS in 11m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d2c9be7e028d401da76311c1975b1cb2 : SUCCESS in 5m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/57959be13a674bd8adeb5b8e98b500a8 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/88614b3e995e4a5baf0f512e26bc2317 : SUCCESS in 8m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/936c9766bd6044679bd623eb8cda7041 : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/974e666623934d35a0d62d89c9b65b91 : SUCCESS in 7m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ba03130456cd4204b11f2b4e53701c90 : SUCCESS in 9m 56s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5d0b34d6067c4237b62be545d50d9fa8 : SUCCESS in 12m 27s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/eae4b1278fea4f109fbb4e72686b32c4 : SUCCESS in 4m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b0f9a51065434863a7c9319380dcc031 : SUCCESS in 8m 54s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c0a7592bfeb24459a4d6317979952b07 : SUCCESS in 23m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f581ca2ae8db4464b126b84051a237e7 : SUCCESS in 8m 45s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4b40cca9ea274d8ba6260b6b26b9eff9 : SUCCESS in 12m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3bcc008592d34d40bf7a96aaedf517d3 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16430cefb3a242afa7ad9774daabff59 : SUCCESS in 6m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40deaaa9f6dd4e1a94db232414b51e6b : SUCCESS in 21m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a44910510d85498caa36fef769eac0c6 : SUCCESS in 5m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1873fe35bc914708822af3a25cb196de : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6e454528c5248aaac3e7cd35a2ac6c2 : SUCCESS in 7m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0716bec1b2ae4f9b9fb313126625ebef : SUCCESS in 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f613b8f63a1c4ffea1217c606c3c5770 : SUCCESS in 1h 01m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cc1083a85d6489f9612b1d46edd841c : SUCCESS in 1h 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0f9ee1c1cfb34c03d678d71af10ad46ae5e07c2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:18:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"27d03d95549e4b9181bfb0bf9969d0a0af6d10a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:20:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3751fc6225547edb627078ff5861f45\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e20a3034a546e0aaa59f2c134fd885 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"dcde032f3996bb8ae8c6425327779583948a6d29","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-28 06:22:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f628446109aaf9b00cc4ea92f75f8335eb07d95","revisions":{"6f628446109aaf9b00cc4ea92f75f8335eb07d95":{"kind":"REWORK","_number":1,"created":"2024-02-22 14:40:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/909876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/909876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/909876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/909876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/909876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/909876/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 14:40:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 14:40:28.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 5.4.0","message":"update constraint for oslo.serialization to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I39bf1081f87bd34bb7fb0fedce94cb2316c44731\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id69e5808a5e37307830751487537c84c8f3b200b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f628446109aaf9b00cc4ea92f75f8335eb07d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f628446109aaf9b00cc4ea92f75f8335eb07d95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
