)]}'
{"id":"openstack%2Frequirements~909892","triplet_id":"openstack%2Frequirements~master~I88b6e8554299d6ebe92275cdcf105d51662dc677","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I88b6e8554299d6ebe92275cdcf105d51662dc677","subject":"update constraint for oslo.i18n to new release 6.3.0","status":"MERGED","created":"2024-02-22 15:17:41.000000000","updated":"2024-02-22 21:14:28.000000000","submitted":"2024-02-22 21:13:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909892-new-release","meta_rev_id":"93bb5b559df29031bf26f255ecbb0cacb5127272","_number":909892,"virtual_id_number":909892,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 21:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-22 17:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-22 17:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 17:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:27:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"662f035bc9405660c10624cba1b2649326e249f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 15:17:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3beffefb4ef2630531267673bc83199d1c27c360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:03:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a56b7eb1c424311b37dcc1425b4eb69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0235c737aded4eeb931a26bc991ff55c : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ece702167c4d1590cfb59eec82df32 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eef3e4e714449708e1b576a64c36b8a : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e4a17def98c4f95a848d8e7f50416bb : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2aa69c46b33546e5ac3f41c7e8c29ff7 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ba9787d8eb8492d8742e16c95d667fb : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73e2b50a7d044d0c889669a7bcd4d259 : SUCCESS in 8m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cd58f5b4f124253a51c0ceca319525c : SUCCESS in 16m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/727c185c76834d4d8f9df4c937a98137 : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cc1ac496ddb84f0bb0e6ea3aede4c808 : SUCCESS in 18m 09s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e1e221d923f044a99f15839aa9e9d67c : SUCCESS in 5m 21s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a06023429a384a7bbab37c22884ed8c0 : SUCCESS in 4m 53s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8a3a36ae04a340c0bd4ccac81fbdf3ff : SUCCESS in 4m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e03d2ca9879d481cb4481ca2325c0023 : SUCCESS in 10m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1bb1a5c2bb374a4798da150a3bbc451c : SUCCESS in 4m 40s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/98f1f0b175dc424ca373b228cac224cd : SUCCESS in 7m 04s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/38ac3784d5af470abe879933f78fd798 : SUCCESS in 7m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0a3c7adaf675456abbdb77682ce3d2f8 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dd5815b206e40d7b42bf7c6c7c4e67a : SUCCESS in 9m 20s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/80d1fcb2ab934944994a0eeb3c92bb44 : SUCCESS in 9m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0708fc40a8134525a9a3d6034c1a6db1 : SUCCESS in 12m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7de6f084077e4a21bc10a26ae3013ad8 : SUCCESS in 4m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4defaba1b7eb4b2f8b35d34d7a0408b3 : SUCCESS in 9m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fa207f1df19642d795beb1b20f4544ba : SUCCESS in 6m 13s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1ee1f95b50594ef8b986e8e26d8310ae : SUCCESS in 12m 17s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d98f3f73c7534287a44eb55b0584ff6c : SUCCESS in 23m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5ff6dc588fc342908c7751ac3ec6e931 : SUCCESS in 7m 06s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9de10e4a1b9140f1afcb15b071edac7d : SUCCESS in 13m 45s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/96284c491cbe45c39cefa9720cb5e299 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0815504d9f554e6780cfae5c66007dd9 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6501674d7fa245659e5d90d6aec0dad3 : SUCCESS in 20m 55s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/47a671ce71324ac79ae7798e24f4bd52 : SUCCESS in 5m 04s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c44769fa79044b999f8be0eee5aea47a : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2728c87b77c5450db5fe9406369bee7e : SUCCESS in 8m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1da02ce07ada44c3bc7ef791b9f5f1b9 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1aae15fe2f0d4520acbd5655147bc806 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/646866c436bf41a680fb097c96e60370 : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":1},{"id":"501c8abe1475a0185423e86658ae4c4ca5ffd033","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 17:27:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42d804e65163a2972658085db9734c2cb512b23b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:29:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8598d21672d2397119e477b2a8730c329de3a28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb464dfeba944399944623c92059c4cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6f4206657d492ea462ac9402a1f641 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b81d86bdd7b43759aa7b9e2e727fa31 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2e562e4a584bbfb0320e644aef0aa4 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91be6171ab484722b43e11f29f85a557 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65db7bcfbf1d4a1b81ad33bb22f20595 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90864c97b11741bdafe274f8f623e9e6 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3c5490e2ff64e918ecf0a458b6a4808 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/23153e02e07b41788ba5deffc4522d2b : SUCCESS in 14m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bb70aa26d4e047ae90a758964efbf7b7 : SUCCESS in 9m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f2c15c9c1d55405381a672e3c6ad5c92 : SUCCESS in 12m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f16ee04e26764bf8a83d4c15b8aa0a66 : SUCCESS in 6m 03s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b2248f1bfb5145fcac94a5f2ea233854 : SUCCESS in 5m 21s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7761cc842f624202a6f5d0f8ea6a3fae : SUCCESS in 9m 14s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b6b18030b21941f48139b32ca2fc6b26 : SUCCESS in 15m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78a8705c4340479692faf850efa53cfb : SUCCESS in 7m 15s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/57148ddaef0d4fe1b541e0a860df08b1 : SUCCESS in 7m 29s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/41069a0813134fd7bf2e4e572e907e1f : SUCCESS in 11m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/058c21a700aa4e90adb70c8426c3f938 : SUCCESS in 4m 59s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8268e55c1bdb4a429445efee913e0a85 : SUCCESS in 8m 19s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b26f34f69e54448faecc90aa76126d61 : SUCCESS in 28m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3478272f5d5e45539eda14a5e57ea89d : SUCCESS in 6m 33s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b5b8514c8922428f953e7b0f553973f3 : SUCCESS in 12m 14s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/34160275374d4b63a8cf73d4fa0ca446 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84e5882b96304162b8cf9c7d46a85167 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73651135842d4cbab84378821a8259f8 : SUCCESS in 20m 09s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e63bad74801b4c619e9281a821da326a : SUCCESS in 5m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c4a97553594f491bb8fa9dd81ccefa87 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/417c438c88f640df9d736e1b6e802ae0 : SUCCESS in 6m 23s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/13cf4afd6cb348ebbb0cd281d852e8ae : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/032c4f6aff1b4f4fbae4ad68ed7077cf : SUCCESS in 52m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d29aba2ab6348ddb84de5f7f60a7eb7 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f8dc1417bcce5a488a65e7ed2a7019feab77c5db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:13:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93bb5b559df29031bf26f255ecbb0cacb5127272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cddca6dfa11844e5ac5af0e7dd2f9b83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76093515c55b46eaa84411933f880246 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"341d87417b629e6eb0fb396665176e4b5dfa97f8","revisions":{"341d87417b629e6eb0fb396665176e4b5dfa97f8":{"kind":"REWORK","_number":1,"created":"2024-02-22 15:17:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/909892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/909892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/909892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/909892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/909892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/909892/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:17:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:17:40.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 6.3.0","message":"update constraint for oslo.i18n to new release 6.3.0\n\nmeta: version: 6.3.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibb5031552bcb43fb6e9876c93ee9e62d4edfcda1\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I88b6e8554299d6ebe92275cdcf105d51662dc677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/341d87417b629e6eb0fb396665176e4b5dfa97f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/341d87417b629e6eb0fb396665176e4b5dfa97f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
