)]}'
{"id":"openstack%2Frequirements~909902","triplet_id":"openstack%2Frequirements~master~I047e89e7f12c09c76ebf8c9aee2c961b2e8b5a82","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I047e89e7f12c09c76ebf8c9aee2c961b2e8b5a82","subject":"update constraint for openstacksdk to new release 3.0.0","status":"MERGED","created":"2024-02-22 15:39:37.000000000","updated":"2024-02-24 06:47:39.000000000","submitted":"2024-02-24 06:46:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909902-new-release","meta_rev_id":"16a596e16b92b2268c0e01a1b68dead0a1ff8b46","_number":909902,"virtual_id_number":909902,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 06:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-02-22 16:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-02-24 03:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-02-23 12:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-22 17:43:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-02-24 03:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-02-23 12:24:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 16:09:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-02-22 17:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:43:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-22 18:06:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-02-23 07:48:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-23 12:24:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-24 03:43:36.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"48e47bf276e3e8e547e3207442c21dbb211dd568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 15:39:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c26ce13d71178298f76dfea05ad3638b2494b69e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 16:09:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7a52d6b06db94b825778ec1d04778355806a8bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:12:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6db0139aa00f4dcead0f411e586c5240\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67929cf4ca7e454ca72f203afb3d1265 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc7ec423fa24437bf2043be7aa76bbe : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1318fe08ae44f08a86252db46828e10 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e1cf7cf586d42caa70e660bf36141f7 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fa789336bf442209ef3737195e1c254 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99483fddd6224ad583383c34ac36a40d : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fba136fea5114075bdb1a3d39024ed7c : SUCCESS in 9m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3d97fad4ac744e4a8d186a881ff45f2a : SUCCESS in 14m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4ec49734b85465f9dc8f129fedcf25e : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c0813181ae524e28b45622e2fae0452b : SUCCESS in 18m 31s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/96330dcbc7b449e78e19ea76fb1336b6 : SUCCESS in 6m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/dc42f05e2b4d45eb86d5ec257632509c : SUCCESS in 6m 25s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/82e3cf1b2ec34e628ac90694eeb402c5 : FAILURE in 2m 37s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/aa8a52f0107d4eccb612978ac4a85914 : SUCCESS in 11m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0ad9771ed8ab4d20b1cee7c2bad7c607 : SUCCESS in 7m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/755fd3ab5767408eba06bf8408991ae7 : FAILURE in 3m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9050458ed88241b68a7424238f44a6bf : FAILURE in 2m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c08872d4eecf46538e7e8ea669db2c29 : SUCCESS in 13m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62a8cf683b2e442ebfd4ee6557231fb8 : SUCCESS in 7m 06s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a2047cd590ed4e24b00b2d63854ff246 : SUCCESS in 10m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/922af3f91d5041f3968a45e6df384c6b : SUCCESS in 12m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e2d7be3a51ed4810aad8f19a4904d8ee : FAILURE in 2m 24s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f323f3aef40c490899e0a4f6c3e3f221 : SUCCESS in 8m 32s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/cc7b0920be2147049d28da3c740d611b : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/38d01afa4a25497f97a1c76491c1dc02 : SUCCESS in 12m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3d210d6820644784853457ea369468d3 : FAILURE in 4m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/00786a37ad4544b6aa4835990fe5a347 : FAILURE in 2m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a110315c5fbf43cd9e267726c7cdea82 : SUCCESS in 12m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0d876e79926c4eec8a5274bed172e068 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/823a2ec94a5441328df0cc3e38ddc999 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/876d8c6baf354f50a7b22e98b320f6d7 : SUCCESS in 24m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d0d1a14f50f148fc839d3ef483fa4d2f : SUCCESS in 5m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f8f242d682844387a139a9b46e11dcfa : FAILURE in 2m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef84357eb7d34bc08ec719553910c9b4 : FAILURE in 2m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f209612888504f4986ab058ff6b95117 : SUCCESS in 17m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba5533ced8e54dbd816ed19a3980cd26 : SUCCESS in 58m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3bedc3270e044a2972c33338073963d : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8cf44def2dc8156039fa0be4a7ce7456d2c8fd22","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-22 17:43:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef33cdcf1a7a8b3a316e1094839ed44eef3e150d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 18:06:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e48f8cd9250ba4c9411d9e703fcc7bfe769b24b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:15:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f009e7730c024f7ba046620849a23e3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58890ce96bb4c6ba957ebec0c6cd687 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0e589f9e9a425b8fc599a37639539f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22ee73fe6ab0461e9b04122ba909dee5 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2543c61c01fa464e977b8cf9d1555e4c : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/024623f62083439c92643abdf79e9f44 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6730acb3475247528f0c5bbe4c0bb25d : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f6c2cb1f0ec4c35af471b555545b788 : SUCCESS in 8m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/72d8ca1355414342a76d02309b15cf80 : SUCCESS in 13m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed40a9249c94477a81024f24f58b8365 : SUCCESS in 10m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ae03fb383ca4b098ca1ecc6b4b7c08d : SUCCESS in 15m 54s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ed0354edaeb04c588fdad5288d70e63d : SUCCESS in 5m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0c80b7856b714841b551df66eee5535a : SUCCESS in 4m 58s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6fe47ebf6f694181a11ce68e36434a0d : SUCCESS in 5m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e5231cc3a6c34917b283f83b82a145dd : SUCCESS in 9m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/97fdc32e8c9d46478682e334be7d110e : SUCCESS in 5m 41s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/60029b0b2e754c99b59360de2947ff7e : SUCCESS in 5m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/737e5ba67d2f421c8830773e942ca31e : SUCCESS in 9m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/3d864eb212cd424490429cead9dbb47e : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64536cdd47aa4775a9b263ab77ffd59c : SUCCESS in 9m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/18741d7afb4b43d1b935f5a42beec004 : SUCCESS in 7m 48s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d5b6d1615cbb48209c1280a264a80f06 : SUCCESS in 12m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/31f33965f2eb44208fd8b13b8782d483 : SUCCESS in 5m 20s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6e46ff6e7760445b9f662d572bd1c3e0 : SUCCESS in 8m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a61144829ce74ae0900a292674844a49 : SUCCESS in 5m 17s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f049563ad9524a218142834fd82029e0 : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e0e9fc5ffd1d44238dbd599582c54f37 : SUCCESS in 25m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd91886f0dea487e83dbd0ac9c15f970 : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/194e6da4e09e4e4fa76ca77e07984138 : SUCCESS in 10m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d4a0a061f5a74d69aa141c6361c1badb : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a76c03f343e41e1915d3d623701f234 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89b366e89b9c4ce180f1b2b8e6dc79dc : SUCCESS in 20m 05s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a71f0dd4234a44c9aa9a1b8d86877841 : POST_FAILURE in 5m 58s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c4e24eacadeb4504bc315507b3fca06a : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d5fedd8f572449194332d54feb38f0b : SUCCESS in 7m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1bcea97b2a434b0785aa28d93e0462bb : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d309092369864ba48f7842a88c83a06a : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afc9c2c6db1f4442b39baa376e7a7bda : SUCCESS in 1h 48m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cc9c9b9031729a88600a411ace327d18ea5d8cfe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 07:48:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d7f14089cdeb45953c6d326dde852618b9ce459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4db6a3bdd54c9ba2b80b48a6c51a15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d55fcd8b484a4abf03b722de4d66e5 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a0b93681e14ea58e8fcd6dc21b5cc4 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/462b9a4fc0174fde859455db30318d7e : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/636821b2e6da467797d710fc7d9a3973 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b15b77e2c19d473ba023147d58b23fbf : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f9096a7319f40e1910dc68ad4b13496 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c27ef96d22404fbd981033e76a1209e4 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f456d7c4e0cd4f0784ab7961bd4d6cb7 : SUCCESS in 12m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f657109d19904e188df1bf812ecdc750 : SUCCESS in 9m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c7e3ad26538c4bb296f687dc78ae5abd : SUCCESS in 16m 50s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5a5a987aa4b94162af09976b28078caf : SUCCESS in 5m 05s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/29f1f6ec2f6b40da91a74cdfd351ddc0 : SUCCESS in 4m 11s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9b30f66d64bc48f691ca65c65087d998 : SUCCESS in 4m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/31e3ed3339ae430395f04fe7493f64d4 : SUCCESS in 12m 31s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/963a9520ea4f457d8a83b7aa25760ab1 : SUCCESS in 4m 57s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6808fff09aa1462295905aa1beeed4c6 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/47b89ce0608b4591be0e4f50343dd0f5 : SUCCESS in 7m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/222f1ad838dd400b9c5cdaf406e30a0c : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f38867c291b43acacb48bb4c16d67b2 : SUCCESS in 7m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/122253900cbe45f296c0732153955689 : SUCCESS in 6m 54s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/06ea331f0cb1488ab8a1fa83ad9fc097 : SUCCESS in 13m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ee6fa27750aa4b4a8494217db3ecef53 : SUCCESS in 3m 57s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/06a80a5c056f489dbda6cf1419cba897 : SUCCESS in 6m 38s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ef8ef637134b448b9392e8cb6c753d04 : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b7defb41e531495b9b48574298ee9dcb : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ae0d05e74b284f079bc15dbc219aeca2 : SUCCESS in 18m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dec0ba25c1af4f5898e1928919f9d0d9 : SUCCESS in 6m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9f095a92d46f4ac98e5d4f91bd492882 : SUCCESS in 9m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8d605bbc55394ea6a4a0d5d4ea339351 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c77866db08541baadc5619e2e421073 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68515efaf46a433db7e76912948f159c : SUCCESS in 19m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8e8d86676e2f4d329f02bac2bd8e2057 : SUCCESS in 3m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/bb4fcd0cecc4448faa2ded2c68eb0941 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea2ffbe34f1a46e9a7858dc008abc636 : SUCCESS in 6m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/09dd27d2cb1249e58a977d0306d3b890 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b7603b5a21444918019c6b30d192f83 : SUCCESS in 1h 02m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/931db804621d43628f3930f5be97da50 : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5f0e04b6dfddb96c8cf536ba7c3587f932e6902e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:24:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8584643ad25fa06b1929cc1525df4b65bfad20bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:26:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0039ad14090711d8330dff8e394013f1abd1044e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:34:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20b6daee68814ce3b5e4a9e0ac6c16c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083e600b68aa4a80a663586c6e58c375 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a471d0f545473a93bf7c9f20d1b650 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a97be3bc6c425fb4ba62b4a0a1b582 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f3bea68f62547ea80ddc894640d1d0e : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/091d228bf1314524967d69191b9c1605 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eff73df1ae804007a65e8541d5fa2438 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee9e4fcf208443488b4fef7bbf252736 : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e135fa85c20447a9f86884d076ef07c : SUCCESS in 16m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a419102dcb3e4d28aba8d7362089182a : SUCCESS in 10m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/481a033cef84440dbe2c6ddd08046806 : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ca1a3ca0909e4fa1946ef317539a88cb : SUCCESS in 4m 31s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3131c6a13ab446e1ba1bc4d1d7303c83 : SUCCESS in 5m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5d35b412812d44098960b4c072a7684e : SUCCESS in 8m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1908722dc3e44652a34bce42d34ee777 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df57f42bf34245518325bf19fbbdb783 : SUCCESS in 7m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cb387274a22e4ae486d3e8785d1f5675 : SUCCESS in 9m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bd9937b0186e424b9b44dc0c236030f4 : SUCCESS in 9m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a0663800f9f64ad7a64b5e9f4a45cf0e : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7bc85bc959d54735ab800bd1d9c60bf2 : SUCCESS in 10m 03s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3f665bc301b04655b0ca44720f16f1a0 : POST_FAILURE in 26m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/880ceeb2581045a3a87b7aac8d34c0b3 : SUCCESS in 6m 54s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/888beeff9c5e4d57848da225f9dce166 : SUCCESS in 12m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/220622bc9be1498fb372baec8b9b4597 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6cfcd12042ce4e079ca1714b168d1998 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d0275781f0b4ebb9cbe8709ddf68931 : SUCCESS in 22m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a198d6f672504027aee1e8546cd69e7c : SUCCESS in 4m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f17e5f95ab8a44a8ab2d15b7acf47138 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2e522dc35a24655ac78a85c33bf4a2e : SUCCESS in 6m 49s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e41d9e6852b8434990d3101c2ed4ba42 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e06cf45595b24f6ea9776ccb830c0217 : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9e7bbc1f7d645cea1fb6e2431d793da : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dd5e1b4e11945e34fafde9973e649e24e446df59","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 03:43:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92350fa841d8ccd30f3c071a69b5b8ce90516b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 05:29:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5914b7c263a34ebf8373d1854cd452d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363976e685af402498a9272e20000c6a : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff6504ba14d4e20ad17bac0b78351b4 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76f0143966c746c59e18a09b4e4bf5ed : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2d89bdc43a34fa2961201b6ff56bcd9 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3133b837275549128eae68d2803c205d : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/723c3cb2e79b41f7a2d94f938158cece : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c1734211136457c9d67e068f974860e : SUCCESS in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ed55f7c60c341c2a7a4f5edb17f4097 : SUCCESS in 14m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fe67ee1443ca404281501f465db8d89b : SUCCESS in 10m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59192fa202ca4cb1b199cca0672701cf : SUCCESS in 18m 32s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9f29b02e39144fa2b7ecc4ea9584dd11 : SUCCESS in 4m 12s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8f777d914fbc477b81e29844cc027399 : SUCCESS in 4m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a0bb5a66b77b48d487dea6459bedda10 : SUCCESS in 5m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/215b99e96dfc4a5487427985f12f4d4f : SUCCESS in 10m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/db32dcdae2024a069f5b640555394753 : SUCCESS in 5m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9def7940adb24c85b94eb4e95ecbe44c : SUCCESS in 5m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1e320a4e82f94d3bb65b13d94a918f9c : SUCCESS in 7m 35s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6e67fa59101a46fdaeaad91dc40914e2 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f4d62ddc97a406d8547a49f66ee6ffe : SUCCESS in 7m 16s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d839591ff4134f0eab94957ee5b13534 : SUCCESS in 7m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/848aa37166084a3784bd66fd4939edb6 : SUCCESS in 10m 29s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/462f5b7815434e39a4414af87fcf3607 : SUCCESS in 5m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7fad31ca4bee4e0796742773a27b074e : SUCCESS in 8m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/eda1eccb69c54d81a9c807a6fe8f3948 : SUCCESS in 4m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4c632127d67f4563adafa289bb76de42 : SUCCESS in 8m 47s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/41eb55d085eb4857aad08b9f4a789a90 : SUCCESS in 19m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c649b34f64b7462b81fe8cd8cde9affa : SUCCESS in 7m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6d2ce90180f74bb696245b65e22f3df9 : SUCCESS in 9m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/71a5623388ac418c88f72a0a7f6be602 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0b8b211f01740fdb5cf9846ca1273c5 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/384d3b80b4ec4c4480362421c92ea1ab : SUCCESS in 22m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1b98178da20444b989b4490cfafca90d : SUCCESS in 3m 54s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/33ac05ce138e46aeaa4e2b7e79371aca : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0598fdb5f554d88aeaba39f6bdad675 : SUCCESS in 5m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f950cfa7e5b04eb6b5fcbc17512c385a : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f106ae0e1d3544b0a4e624b724ec47b4 : SUCCESS in 40m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5400841a1ff54468a9817061879a45e4 : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d04dc1d068cb974505af4e6f35dfdd734c9dcb39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 05:29:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea403b18acbd125000d57817b7d0e00c379e834d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 06:46:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f10a29ffba504db1acc99ab960557245\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268b1bb2504d4be4bef8a880ce3db84f : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ab6a9aa0744ed1ae57ff29d75c95cf : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d6c2bb3d294edda6683310ecf53c74 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d049e60164c54983ab77570555c5ee30 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/851084b3556a411ba8d95f464136460f : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d600031d5334e97a268456580bae3c3 : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a001a3d33399422d8d3bc1e58b93d481 : SUCCESS in 9m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69441a273ddf44b3b8833795f4163aee : SUCCESS in 15m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e38bd9b206a847a19172b2ba4b979f3e : SUCCESS in 6m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6267cf6b086d4696817b2c46291aebbc : SUCCESS in 17m 05s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7261842a2ff44a6bb65c72cdcdbb599b : SUCCESS in 9m 34s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9510165183d14ce58e7054f7771fe61f : SUCCESS in 5m 23s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cfb85858f2d144e2a03bb8a3de23839d : SUCCESS in 5m 35s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1e84bfccc437482d89a3dcadfcca918a : SUCCESS in 6m 55s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a3886ba376e1474cbf61059525f92a50 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c9c3b68698247f69f3d55804019e0bd : SUCCESS in 6m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a590b939f9a9454c815877d07b83062c : SUCCESS in 6m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/80a21cc47ad044349fbf6b2eb62fc2c2 : SUCCESS in 8m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b43e777b23024f18ab56cc9cf620047a : SUCCESS in 4m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7fc9276fa32d446f8c813c08ed40d838 : SUCCESS in 7m 06s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/94fbd73fd2964acdb8c4707aa1a2457b : SUCCESS in 16m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a185e5e0f0e428db44664801600bdfd : SUCCESS in 4m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8af79480f3b04609ac9bed71f0596447 : SUCCESS in 11m 04s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f0c27daba3af4c9c9a24750e5939fde6 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5acee130e42d47938127ccb8a91f525a : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e00d1960051d46e899107213aface399 : SUCCESS in 17m 29s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/86a70572c64647399ba03c30ffeaccde : SUCCESS in 4m 15s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c266dad565934b31aa4ef03b8dac1b80 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62500e8e45db46eda18716e95ed96f97 : SUCCESS in 4m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/63fa2765ceeb4aaaa4aa7a201419f322 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aeafe628ffc04da68b3f0dbeba32c671 : SUCCESS in 1h 07m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/070d99fba0164eb5b6d49792ece1fd25 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e8b3a3e6099dab07eca714b6fed290edf2bf26b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 06:46:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16a596e16b92b2268c0e01a1b68dead0a1ff8b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 06:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bebcc9ddd6848deb160a7f79d978de9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58060788a018485d84f39ac40905d2d8 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87c8b89124d5a947f7cd3e31d16a6c844dda422b","revisions":{"87c8b89124d5a947f7cd3e31d16a6c844dda422b":{"kind":"REWORK","_number":1,"created":"2024-02-22 15:39:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/909902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/909902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/909902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/909902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/909902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/909902/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:39:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:39:36.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 3.0.0","message":"update constraint for openstacksdk to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia79667ca74811535e3a088f606b86904eaabc879\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I047e89e7f12c09c76ebf8c9aee2c961b2e8b5a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87c8b89124d5a947f7cd3e31d16a6c844dda422b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87c8b89124d5a947f7cd3e31d16a6c844dda422b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
