)]}'
{"id":"openstack%2Frequirements~909903","triplet_id":"openstack%2Frequirements~master~Ia9087f3e7269dd5b22b9934ddb3df3381889eb56","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia9087f3e7269dd5b22b9934ddb3df3381889eb56","subject":"update constraint for oslo.rootwrap to new release 7.2.0","status":"MERGED","created":"2024-02-22 15:41:12.000000000","updated":"2024-02-22 21:39:01.000000000","submitted":"2024-02-22 21:38:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909903-new-release","meta_rev_id":"14947e1997704f7f110be3bcb4709ea9efe806b0","_number":909903,"virtual_id_number":909903,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 21:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-22 17:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-22 17:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 17:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:27:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"409105deadb75c0258ab83a5e08a505a6de0b071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 15:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30dd350795885f8913f2f0e4897edd3e9289e369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:14:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f5716b8d7a464db8356364180b9691\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83066e9aa6e4ef1af07661ecac92a97 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f05592a638748b29418a7110f331f52 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3efc8c5529a74a9abde3bcab333d2275 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d536e500fc6d486ba488b6156d2aa493 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a5f89a5a04643c7a431111c913ca3b3 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/489f4b054927448884412e94cc67646d : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/503e3557b50d499598cac656019380a8 : SUCCESS in 12m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c7b2c021e0c4b8db157aa5ca82e15ba : SUCCESS in 14m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/830c8ecfb37b42099f458bd33a10d901 : SUCCESS in 9m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eb6192c995854538a4497a2a69ffe579 : SUCCESS in 16m 52s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e9047f71970d4b94b8992a7e86f6c7a5 : SUCCESS in 5m 52s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/78b81da277f042d0b3ae3eb5dc492f8d : SUCCESS in 5m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/03b78f1fb9084e72af707695d6b2d031 : SUCCESS in 5m 11s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c07437c8c1904dc9a458a4e68e234d4f : SUCCESS in 10m 11s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/87c0c5c5963e450f8e9af8a5ba8d615a : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6fc12433659b4af6bfcae930c64f792e : SUCCESS in 4m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6b9444d3cfc44ea7b2d5cf0a193e3d32 : SUCCESS in 8m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/50ed92b88ec9452ba1c080edf6e36078 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce743ed81c764f7997ec7153cc96a841 : SUCCESS in 9m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0427754e12f748b5b639d6eae194ef37 : SUCCESS in 9m 27s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fead2a3753d14b41abd6b0dd59e1b98b : SUCCESS in 11m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2d80d331d5b74d298ac72e773dbfdcfe : SUCCESS in 4m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2a81d967879044aeba59489ad8aff482 : SUCCESS in 10m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/10ab3b762b364d4fbbd219daaf207984 : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a830cb25109e44628a971e94ec282658 : SUCCESS in 13m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/efa06182d06f46ab93ba74bab44473a1 : SUCCESS in 20m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/143cabe410134e85936bb9e4122163a9 : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4b09a64cdff04eab87f33710c086be93 : SUCCESS in 13m 11s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c86410904f324dab9d47cb35c44b7d97 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a4572a15d9f400597e21a429252a6dd : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16fc73396640466a8d2cc3f8e65704f8 : SUCCESS in 19m 27s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d6155b8ec4864a2eab257df1c43fb803 : SUCCESS in 5m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8a0d51412394443cbd4656d9a32354da : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fcfa9182271e4ce0bc45f866b88a27b7 : SUCCESS in 7m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b0e4baf80fd144e1b1d28437a5c6e8b5 : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee2af26d303c4f508c59572897637f02 : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76444f7015934debb69736bf1f051178 : SUCCESS in 1h 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cd99afd6bb38a732652e8d39ebc74b858b9e1057","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 17:27:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d002e1e5ae478f77deed70359f84738f96c42b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:29:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"688fdea1cceeab9a27770fca337f597e835ba02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:38:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2a409004084cd3b3407213c28b96a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52965bd43fa45c38ad1a1f70294dc1c : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e3ff3b7d4d49aeb77f541f13ede516 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70fa2689e6b345029cb81f3a82e66aa8 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2e45f449ea4402eb1bf9fe61180a8f1 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9e076f91fa44b7e85f62691488ace5a : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00a63858932f4c3999ee6bc3aa6c4101 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/992affae7bba4913a1947ccebd0de787 : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08e6b61a121c4dca962e6b22f0fb1b25 : SUCCESS in 15m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f1039b3b62840eead4c001c441aa8bd : SUCCESS in 10m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/09eee27d67614bf5bad7f518bdee826c : SUCCESS in 11m 02s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/654a6f1c67164e12ac90b6a36bac994c : SUCCESS in 6m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a3109b04250b42428641886d26c32b1b : SUCCESS in 7m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fcd2a83494fc4376b3cc1ba7c6f9555a : SUCCESS in 7m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/99ad72b4adea4ec192c948608983192b : SUCCESS in 14m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bfb148d6f40414490dfcd61204985b5 : SUCCESS in 8m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/482689d563bb49e5ac243168d38e5562 : SUCCESS in 9m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0914210ef9be4d00a853efe1e5b2774f : SUCCESS in 10m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cfd3cd24335a4b20b816aa2d01884f6f : SUCCESS in 4m 59s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3edcba7505bc4a07bbe02992a284903c : SUCCESS in 8m 42s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cc90244a6cea4ea8b7dfaaf3b23ff3a6 : SUCCESS in 20m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/87e7ee15d83e4389a014646b51f2690e : SUCCESS in 7m 02s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b7fd9258d73b45bca69533cb99b87472 : SUCCESS in 13m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/63e2b9012de74adeb0677aec98d0eb20 : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ec1dc0e6fb914eaa9844a1e843e91c51 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5aae241697ba4ae1b7cc3e7386f00f2a : SUCCESS in 19m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/bcd21b77d671431db203597faff7814e : SUCCESS in 5m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c5657d05bad74734b59dc56bec4edba8 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74cb4ec229ca40f3ba7c5327ef778d97 : SUCCESS in 6m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e7e201e5e6f04db48713bdc71110dba6 : SUCCESS in 15m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8565505820a84d50bfaac1fe0ae129d0 : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf09db18b54944b4816cbe9fc6edf307 : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bb5fa28c1d27a421354c2d8c5441b786f3408f35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:38:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14947e1997704f7f110be3bcb4709ea9efe806b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:39:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c355271503f64339b665c7df6c81d491\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a005c24c2c6c4c128d55d21ca88a80f2 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"634b079e39d7a0f8b00eb7518c2e1308e79d807f","revisions":{"634b079e39d7a0f8b00eb7518c2e1308e79d807f":{"kind":"REWORK","_number":1,"created":"2024-02-22 15:41:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/909903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/909903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/909903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/909903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/909903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/909903/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:41:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:41:10.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.2.0","message":"update constraint for oslo.rootwrap to new release 7.2.0\n\nmeta: version: 7.2.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I66cbf8b7605638795ac7bfe2325594e4b30b25e0\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: Ia9087f3e7269dd5b22b9934ddb3df3381889eb56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/634b079e39d7a0f8b00eb7518c2e1308e79d807f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/634b079e39d7a0f8b00eb7518c2e1308e79d807f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
