)]}'
{"id":"openstack%2Frequirements~909906","triplet_id":"openstack%2Frequirements~master~I7bfbeae6b849d490185c6af7e90e1122d468c797","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-27 17:25:46.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I7bfbeae6b849d490185c6af7e90e1122d468c797","subject":"update constraint for oslo.messaging to new release 14.7.0","status":"MERGED","created":"2024-02-22 15:44:10.000000000","updated":"2024-03-25 13:30:43.000000000","submitted":"2024-03-25 13:29:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"909906-new-release","meta_rev_id":"9652910b91fcf0b232f775232dea2541ea9f1b1a","_number":909906,"virtual_id_number":909906,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-25 13:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-25 11:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-03-25 09:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-25 11:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 17:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 18:06:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-02-23 07:58:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-26 09:22:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-03-25 09:27:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c8485e2e6910cec72554a63340b478fcc2eecedf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 15:44:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"487ddd30664476e9ce32f4b9827364eb1ab95564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:20:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be4933fb853645fa8a6b4651f484fa2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3605ee72df5243cd87870082ef5a938c : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1671e00607343a9ad77158c0ff487f5 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7bce96ed96443a5978e83b24f1003d5 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a97c0287666c4611b30a09287c390695 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f76324f3c484977a8f4460c7e5ed9f4 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/937b10f2434b4441ae0889ff648869c7 : SUCCESS in 5m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03519c54b23a4e15bba9cc39e83f6f2a : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b3ca19004c44d39a4066eb7c374fb2d : SUCCESS in 15m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/223ca2ca42984959adf96e8144ecea0b : SUCCESS in 10m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6de0fc4ff00c47019416c566e78142c1 : SUCCESS in 17m 55s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/23df5671a945468c8584bb13a8d3ef88 : SUCCESS in 6m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/22c7005b051f413f81cfdf8c7afffe02 : SUCCESS in 7m 52s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8b55675056274dc08fd3d65035a96385 : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c9cb68bb9ef840918f666422d5cf45a2 : SUCCESS in 10m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/94c582693a104d8e8f32496f2b539397 : SUCCESS in 5m 00s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/12ffa2c8e2154786937d5096c4ab49a3 : SUCCESS in 5m 09s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/616726326b5b44919be36a94b50d1865 : FAILURE in 5m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/715ae2cd1544439cb10dc0ae42b52046 : SUCCESS in 14m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f74f5ee8e6d6439eb919142d2b4167e6 : SUCCESS in 7m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/96a6f4ab2d8e4bb2b47c787dbf2e3de0 : SUCCESS in 8m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5745b5f14935411fabaa7e6e73790357 : SUCCESS in 12m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/5b935e1d7f9b4cc5ad5a5b31ce3b1130 : SUCCESS in 4m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/186d9750229c4b1ca49e219280c5283d : SUCCESS in 8m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/334ecd7c0e4b445dbac3c593e7f1e12e : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/caa1c3b2408e40499430d1c4dbbc0b59 : SUCCESS in 12m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/01986808d0d34b85836212cd102090b5 : SUCCESS in 24m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f163287d22541779cf7324b3dd9e4dd : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6b36e64c53b84774864404651778fd0e : SUCCESS in 11m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ade9d3f857044202a0e95a73b0f56394 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78781bdfc1694a5eb29cc0565242c09f : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eea633257d044e328897bb6edc199ce0 : SUCCESS in 23m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/aa9237d600ea484eb34a3a59f719177c : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/560adad59f7845f48d2f7b0d24f93a08 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f842f6bbc674e3396625a9ad938489e : SUCCESS in 6m 12s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/82ceda83cebd4de5b2f32f987531f835 : SUCCESS in 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5641eab3db64b8db8ebd2becd165935 : SUCCESS in 58m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb9f837dc31e40eda0faa5d3cc63c933 : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bc9836830bba50bee0a92a8d9a45bccc434bb55f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 18:06:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb1e5454dd561931419889c130d0310bf0a85403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:38:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cc64709e9d4430facfafed15577ec2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa10a7805a74d81ab8da90a2a99fb43 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1658b9fd331045848211276697b3d07b : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f817f17df264015a2b31ec63fba35dc : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff04ce978ad249f3bb86d6db83eb7339 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc68c34ef48c47b5b7ad7803ad6ca238 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e731ec62c0e14df9acd009a44dcfd228 : SUCCESS in 4m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/264808df7b914969b7dd7ef8272738b2 : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c8f1ea86129422dae3dfc5e193d4a76 : SUCCESS in 17m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3517a6f0ec3c462bae5daac8bd2e0727 : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/01bce309bb684a26ba68f1db6fa5dfbe : SUCCESS in 16m 07s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/49a1d66fabb84bf892dcc16dae2c03fa : SUCCESS in 5m 09s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8140c1bd7b7d4ec1b6829f22938b1583 : SUCCESS in 4m 31s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/36b9d0177d5a41b1b6d04e86425c009b : SUCCESS in 5m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8ddfe9f88e53447c906732af8a823de5 : SUCCESS in 9m 48s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6858dc24f5db4ae388cc99dc085795e9 : SUCCESS in 7m 08s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fe8f33bd2ee14df4a2856ae2ba04cfb1 : SUCCESS in 8m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d5f69f7b726f447dbbe282a2715a560f : FAILURE in 5m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a92d93e25b5a433b910b9c45bebf3fad : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79579ab33e0c4a2a97e44f4a5820c11b : SUCCESS in 9m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/dad9bbd70db445edbd2e104907caf575 : SUCCESS in 9m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f1f44f555ea94a43ad6e0f310ed91707 : SUCCESS in 10m 55s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/159b1cc8271145e2b5cdcae511c9bb5a : SUCCESS in 7m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4688ed85163a4e149b77c57e9e372fab : SUCCESS in 8m 08s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bbe32adbfd3248cabef6d32539862ff2 : SUCCESS in 5m 57s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ece4b6b0dc014ac2b5d9cefe538ae3f5 : SUCCESS in 14m 01s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0d0830684eed41d3b56ac7e05b1eb956 : SUCCESS in 24m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8849536fc1c4883a919698b18c9ea69 : SUCCESS in 7m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/13f23a3edd8043ca814f318fd551c643 : SUCCESS in 11m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/046d79f163064128b2c382cad42760ad : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98677630037645c0bb63e95e5fdafca1 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec5222ddc74c45519449822af3d8ce8e : SUCCESS in 22m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a46c490567514d83b197d2c33cf65738 : SUCCESS in 6m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7d2819fc82bb4c3dbfb0931fdb8fc1da : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a4e3b3cea18e43a59aae83884957d59a : SUCCESS in 9m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a580854afe4e4cb296dcdec2a8270cb7 : SUCCESS in 17m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2602f88ee174bfbb8f9826cfc2c1f47 : SUCCESS in 40m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/585288657d0f45ce8c95ecf67af5e775 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3c9c24d33c8305a27d2a69c4d5b680d325efcd66","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 07:58:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d1be21dd50db34badc2f96aa235c9b5602a1b96","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 14:55:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83d2b028227524ce3e34c4ae90e4ba0a3a7100f7","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:48:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45dc109860f446ba134cbab0d54389f\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c60fede2f7542ec9971f992489a8730 : SUCCESS in 1h 48m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/77abead90b8d4c90bda21167625dce28 : SUCCESS in 1h 11m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a363d5fa3bff4d448a65a4ea46911b32 : SUCCESS in 1h 35m 37s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/53d5545152f84168a35e0b7ebbf5def5 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3382c6903a9f21da1d7507ca280ca7b8c78b7130","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 09:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0afa99b1daf9c01d69cf67513ab30b00e55b610","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 04:00:58.000000000","message":"Abandoned\n\nNew release needed","accounts_in_message":[],"_revision_number":1},{"id":"53e6f996e5ef06d4cc1cb7d2a89cf63b513bf298","tag":"autogenerated:gerrit:restore","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-23 11:02:53.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"47bf1a11cf4d66478618a97f48e85083a212242a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-23 11:06:24.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fddcf52f6e485aefa3ddde4fb55183e385030fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 12:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9397bf3847994ffe91e488e51360e7bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7769547cfeb469a969b8e1ee8de6899 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/357e898d4fc7404fb790a648b2896272 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4647b4f75504d56bfae8704d0701cb8 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/180d59b9600341dcb824dca9e674dfaa : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6aa1aa61410c4c4f8652a26dc8dca7c3 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8510fe7a97c465483715d41c9129f57 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33c0e048e04344b48a163352936c5a50 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a78b40f01c8c420db1835d8ddde10ab8 : SUCCESS in 7m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8751cbb9a72a47848705f6a44e4547c9 : SUCCESS in 10m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/875b4ce98e3c49d595b5a17c6745c5d6 : SUCCESS in 6m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c0e90f8b9ad84181bc46c6b0f9ef05ba : SUCCESS in 17m 34s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/bfeedac8339b49529688bf8f6b857d54 : SUCCESS in 5m 16s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/40f21f94e3e54b63be96baa40f5790e7 : SUCCESS in 4m 58s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ca12e9ee93e14336b15b448fc946daaf : SUCCESS in 3m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2bb515a30b044ed4b5085e522b8fb37f : SUCCESS in 11m 51s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b1206464720a477da442ad27f6e9d2c8 : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7cc5a929eebf4d2e87d2726c682fbaaa : SUCCESS in 5m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/378dd730dac34669b8088a1258af6b5e : SUCCESS in 8m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1e9a2e1cb87d4a7b8e1185082af6c0dc : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f3121909b6745bb8f6cf84695087389 : SUCCESS in 5m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/242c3f5442924cb6b01c84a0a98da35a : SUCCESS in 5m 31s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f56237a6955846b6ba01f659e7289d3c : SUCCESS in 7m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d81d4191dd6346e9abfb8f831905b5fc : SUCCESS in 3m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/cc07ae9e47ed4e84b84482ef7b48264b : SUCCESS in 8m 36s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c569ce2921104f59bbfa6d0ed1097549 : SUCCESS in 4m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/90696b4fde15458380aa60f7830fae3e : SUCCESS in 4m 11s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d94224031fef4b5e9812726ab8eb1312 : SUCCESS in 8m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2d91662cf4ce4168a27b0aecf5e1a090 : SUCCESS in 22m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ad9cfbe91e7a4b8b8e78f09a0f26c715 : SUCCESS in 4m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/241b70d51bb7464f9a2e1210898d995f : SUCCESS in 11m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5a0f8db84ece4c51b426254ae19f7a48 : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57c161fb0c9e45b3bc40152e11511d87 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/464cbdf4b8ab431aa0bc0a65042e9a51 : SUCCESS in 23m 51s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/88719129608c48328e881fa9f075d484 : SUCCESS in 4m 28s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6515e79afcba4af1a3390df243204785 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21738955af584e2c93ec23edc23f6cdd : SUCCESS in 4m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/52fa56b827164615beda9f062af3a1b1 : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51ee93bd4ccd4836941dd40664d83c6c : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6cb6a8e34ae4926b5143b5eb86cddce : SUCCESS in 1h 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7d0643d7e1b8b7bd944b4cefd583778960d244a5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-25 09:27:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6edb1a23baeccdcdfc46e416aa8b651cc90fc140","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-25 11:49:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c50543b9a96954900336a64155dff8a8150d58f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 11:50:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a50278cad8d45b369fd51bc1d63685dac9f3b180","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 13:29:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dbdb85930324746a7d419591980212f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294ff968f8654ccd879211cdbcc51cec : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1534149859a84a7886ceb27a334e0e04 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c05dd4066ab4cdb87679d409142abcb : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f41993dd76a64e038e68cb0a1f22c299 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94412b58276549c19e05de0187159771 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a3f0ea0b11a495aaaf2d411409a065b : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be486a843d944cf1948af8de469f971f : SUCCESS in 10m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cfa17d12b726408384d66acfd772c91a : SUCCESS in 14m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0982cde4836542dc8364ee0ddee548df : SUCCESS in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/94790e5cc64b4032883aff06f1b8110e : SUCCESS in 20m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/66bbeb60881f45628cff6a2d11f168cd : SUCCESS in 12m 02s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/65617ad14b28441d9c1b1653cf0478bb : SUCCESS in 4m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/be793472aae14dd88c87b352786c60a0 : SUCCESS in 4m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2e0428a08dff41fb95f89e71fe0fd709 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/3d2ff5cdcf79491293a1df01cad9f8b8 : SUCCESS in 13m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a8966a1bcd541459853712efa733d89 : SUCCESS in 12m 04s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/896a017ffc644088a9878ef867f91725 : SUCCESS in 10m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/841e86d5e785497faa443eddf5ecb390 : SUCCESS in 12m 50s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b1b74990f1e74ec2abb808d2af0b6ee7 : SUCCESS in 4m 17s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5471be4402c843f5a391da163498fead : SUCCESS in 8m 58s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6cd163c8453a4582abdebd24a04503d4 : SUCCESS in 23m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/342d74428eda41f6ae62c246bc642770 : SUCCESS in 6m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3ddfb75a99c84fda842822a5d937e3fe : SUCCESS in 14m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2b1d16bf037e44aeb2b563e9c5412fdb : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe5a5ae6aa6349d19776ed15a4aff376 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16a3a905edfb4fbb8ad48e6845608ad9 : SUCCESS in 24m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5cf4cc4e60414502a5587f5180c1a4b6 : SUCCESS in 5m 07s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ccafbb1f378e4463ad168778ce04f51b : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ace87a55326437ab26ba3c0c2919409 : SUCCESS in 5m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/84295e95c01041a8bc01e11e8b33984c : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b38598ae16b64fbd9cc348027780a9bb : SUCCESS in 56m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f767ee4b7a6649c99ffc3628c4779ba3 : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cc70b1095ad40426e24b98e989d3d59127e520e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 13:29:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9652910b91fcf0b232f775232dea2541ea9f1b1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 13:30:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cda5f6d64584cb7a58cf11486eb7e6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfe53d7de0b40ca9d37739e7c4d7e29 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4845af8877b78a81a66ba7108cde6be5b7baec2f","revisions":{"4845af8877b78a81a66ba7108cde6be5b7baec2f":{"kind":"REWORK","_number":1,"created":"2024-02-22 15:44:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/909906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/909906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/909906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/909906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/909906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/909906/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:44:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:44:09.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.7.0","message":"update constraint for oslo.messaging to new release 14.7.0\n\nmeta: version: 14.7.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I40015ebb455a8d2e0491a56b2fd8bd795cf33e28\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I7bfbeae6b849d490185c6af7e90e1122d468c797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4845af8877b78a81a66ba7108cde6be5b7baec2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4845af8877b78a81a66ba7108cde6be5b7baec2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
