)]}'
{"id":"openstack%2Frequirements~909908","triplet_id":"openstack%2Frequirements~master~I256b843084e8af00f34dc1c12bf3e407ecf779fd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I256b843084e8af00f34dc1c12bf3e407ecf779fd","subject":"update constraint for debtcollector to new release 3.0.0","status":"MERGED","created":"2024-02-22 15:47:19.000000000","updated":"2024-02-22 22:18:56.000000000","submitted":"2024-02-22 22:17:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909908-new-release","meta_rev_id":"19ee532f7963cb2044e443f26eeb8664d08d6ae5","_number":909908,"virtual_id_number":909908,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 22:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-22 17:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-22 17:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 17:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:27:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f361165504219bc68e69f85bfca7cea0e64ccd74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 15:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ffd26bb89f3a102c3c05c7dfc78c041f083ed59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/074092aab7d24cd8bf718b58898652d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f062ac615b354b9c84366b9dd7633c56 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ad114efe6c4764ae944b3bffac3ec4 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a68c7ba82964f7182b9fcbfd03356f9 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c19002d886004ce8848f29df1e9f0dc8 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e7792d9177544c0b8184b2a9311fbc2 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92bd373584864bf5879dee8788bf991a : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03c66047e57744e3a09c49119557a5c7 : SUCCESS in 10m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/775844ee9eab4328bda454ba240bfb83 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ff06f94fc9314cccb765910909512ed0 : SUCCESS in 9m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ce2855b60d7946569b5ec91420cd0009 : SUCCESS in 16m 53s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ccbce029012c45d581075c1067ce698a : SUCCESS in 4m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b1f5ef1d4c9443d495e82bbe1f5e97af : SUCCESS in 5m 41s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7d02b54413e7414a91f58df225d5902f : SUCCESS in 4m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/dec0d26080914c05b7ada34f6a1fa935 : SUCCESS in 9m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/914031eb0d804ce0be31f99e50614b3a : SUCCESS in 5m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/435bf74488144167a187207c298204b6 : SUCCESS in 5m 54s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6b2f484b7e9149c793d2886dcded9ce0 : SUCCESS in 6m 07s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ce39feccf72a4c9b8aa2c66bb26c68ee : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/627b760d61f7416ab53f525d3725574b : SUCCESS in 8m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e1f7cf63363a43e8be8911b28de68485 : SUCCESS in 8m 14s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/34afdbcdf66c4660b6255c9f4a6ca980 : SUCCESS in 11m 54s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9649495b84a7438698b6296f7e1d0dff : SUCCESS in 5m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1cf5550bd8204e37a2a69600e76df8c9 : SUCCESS in 9m 42s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/8a5407aa78c846c1a054b1d45071ae7c : SUCCESS in 6m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2ecd9f67818b4f508ff9fc125d713924 : SUCCESS in 12m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9e0e6a738d4645949d57090cbaa55a89 : SUCCESS in 30m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/61e5220c6836401fb1a5c4cd43ca86d2 : SUCCESS in 7m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9b3f24daa9a04f89be1c172b12decf3c : SUCCESS in 12m 27s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/56b3b1a469924805b8a3224bd80f7dbf : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/06c7e4194c434ff7badf43c4b04c2bd0 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c99ab929aec043a6868dacffe81dbffe : SUCCESS in 21m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ee01c19ea2584faf8987bceb28faeefa : SUCCESS in 5m 55s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/925110210c2f439dbe6d1a5f9a39fc52 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6834a8ff1c9e482d9d41cff94b62b6f0 : SUCCESS in 7m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b6d82c30534446b9ae0979e39ac23594 : SUCCESS in 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88b0f2e136304551bf273abcba1753f5 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6837e2e00104f60bda73d4005858033 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"12753e65fe1d9b0bb57b11741218d9ca6e7b058c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 17:27:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f1ce7b88df52aae8498db41471f9f8cd36c63ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:29:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2c8e6082c4230cd8814bf8987bf6ee848a57db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:17:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90cb2f612e7e4395b01d2dc558b6beed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fefa506b4649729fae8e6fc5b0ccb2 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2783fa30b5490ea19325404020588d : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcc7e2a7b81d4fe093e39295c2e4df1e : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d98d0ed16f84250ae6a34729546c8b6 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d371d1d03484714aa92ad4c000203f9 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/914d3445e4c142fabdf5451e751e5218 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4d9b65f177124bb2b826f60d172e7ec1 : SUCCESS in 9m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/391aab3b5b874733b93abbccce1094fa : SUCCESS in 15m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ab62af76d5fb435f8b2cc8823549455c : SUCCESS in 10m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5ae98155dc734218b920976033e5ca1b : SUCCESS in 11m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5542d6af8794464cb6c063fb030f75e1 : SUCCESS in 6m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3d24aa3491df45acbc23ec6b7378f805 : SUCCESS in 7m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3b9af92353314ff090b7b7554d7af2a8 : SUCCESS in 7m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2b42d64fd2104e1e9a47d6c861172b6f : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ca13649c56d4ce3b0f55ea333c1c28e : SUCCESS in 7m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/960eaf055caa4a0f811d2c883abf202d : SUCCESS in 7m 36s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5bbc058284114d4ea843b52b2209597c : SUCCESS in 17m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/5fa38f7d5d9c49deb36c4c7e5c505e17 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/84603bba340245318a8ca49d865609c0 : SUCCESS in 8m 22s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/77df2f0cf45045b1a0bbe4c8a45259c3 : SUCCESS in 23m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01fd12d31a7d4e6289a485b29d22171f : SUCCESS in 10m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ba4e2ea53ac8402f8cbf09b0ea35dd46 : SUCCESS in 12m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e7b1fef82bda431087919b8ea2fb1586 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7534a6c7164147f0a370590e7e925e52 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89c2fb3cc2694331831c4bdbfba0532b : SUCCESS in 32m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1aeead88fdd048b7a7e413224d06de19 : SUCCESS in 4m 31s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6ad2da729aed46bca6f234fe133030ef : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89df8786792f40d9af129a28bae9c884 : SUCCESS in 6m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/768519dbf03e4f86923f496dd7cbb70c : SUCCESS in 16m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19d952cb7f654e919868ad213c95bc9e : SUCCESS in 48m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03cf15086e574dcfa782b243893c6025 : SUCCESS in 1h 47m 27s","accounts_in_message":[],"_revision_number":1},{"id":"030b52be8d6df1d9ed7c02add6ac44d7042c1d71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:17:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"19ee532f7963cb2044e443f26eeb8664d08d6ae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:18:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec52a92e2b9b4e6a984c2b77bac004d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ceeeec99dfd479486328b5cd7d76f8b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1619129ecfacae94c5d3ee063a5943de9599887a","revisions":{"1619129ecfacae94c5d3ee063a5943de9599887a":{"kind":"REWORK","_number":1,"created":"2024-02-22 15:47:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/909908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/909908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/909908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/909908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/909908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/909908/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 15:47:17.000000000","tz":0},"subject":"update constraint for debtcollector to new release 3.0.0","message":"update constraint for debtcollector to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I1d577412ca03c9883a2c8ed0ea442809985ca762\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I256b843084e8af00f34dc1c12bf3e407ecf779fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1619129ecfacae94c5d3ee063a5943de9599887a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1619129ecfacae94c5d3ee063a5943de9599887a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
