)]}'
{"id":"openstack%2Frequirements~909923","triplet_id":"openstack%2Frequirements~master~I993dc7b71f2629d04a10229e429f613bde134965","project":"openstack/requirements","branch":"master","topic":"downgrade-dnspython","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-02-22 23:29:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I993dc7b71f2629d04a10229e429f613bde134965","subject":"Downgrade dnspython from 2.6.0 to 2.5.0","status":"MERGED","created":"2024-02-22 18:55:08.000000000","updated":"2024-02-22 23:30:33.000000000","submitted":"2024-02-22 23:29:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909923-downgrade-dnspython","meta_rev_id":"905719eba36eebf89b5fa4237298ecc4d20ad178","_number":909923,"virtual_id_number":909923,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 23:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-22 19:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-02-22 19:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-22 19:01:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-02-22 19:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-22 19:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-02-22 19:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 19:01:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-02-22 19:03:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-22 19:11:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-22 19:16:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-02-22 20:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 20:56:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"c6e73ad14984000b1d7be1bd37b2120afee8ffde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-22 18:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e75eb470ad352d5d2fd7e1b7f3769215c3cfca77","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-22 19:01:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"095aee09a195e50f3feb6f83d9c1b71563566f3e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-22 19:02:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcab3d6a0ca80254d304e1a842a3f5ee592cc066","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-22 19:03:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90dab9946a5b89b00c73b841b3b37ebd01399cb4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-22 19:11:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c59ea12905316f4c934ec3548cff20ed0f8cd021","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-22 19:16:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"002539b59f73784e458f318720d1224a912bb8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:16:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c79933e99eb2466a8e36532c5818b8b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0513b493013a49c59221fdd7b54f3b00 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7314ac33e06c46d8a4952fc30e18885e : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa4df72e68d54b3791fa2a064a28d355 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f742375eedb424688a7f611623041a3 : SUCCESS in 2m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75b99e5e2c464c089a46c1d4bdb1ce56 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bca7bb3cbf9423fadb9af6a6c266d8f : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a316c117cf64567bdc22487d0d85e8e : SUCCESS in 7m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df82fed8ab424a2f86a794cd797e9fc5 : SUCCESS in 13m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/22a18bd1fb094b7ab487fbd33cd941e0 : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5664db6bae8f4a9a906a931480d0f620 : SUCCESS in 12m 11s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/082fc8316c3445a5ab2121ab451d51a9 : SUCCESS in 5m 12s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8d71644dcdde429094bede204d8a8d1b : SUCCESS in 4m 01s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f0165e65a6fa4cf7a4488d31d13057b4 : SUCCESS in 4m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4952444a2d22437593c2de78a7e53889 : SUCCESS in 10m 48s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/34a2e1b8690f46b28dce2e8b325e778a : SUCCESS in 4m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3f5fae5659074621a7590fe7793e50de : SUCCESS in 4m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/10ea53124a414092b2d1e56c31ce7679 : SUCCESS in 5m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/bed0a07fbb0c46e38440155e13115d95 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81188225880648059f58d385ddd04d92 : SUCCESS in 7m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2527d03f67f842d1a6bbf31fda697021 : SUCCESS in 6m 49s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b0566c443e1544aeae83df540004b654 : SUCCESS in 10m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3e41710766b4411dae5d9073ddc63129 : SUCCESS in 3m 22s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4ddfede97a13433aa70d319b812d083a : SUCCESS in 8m 32s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/84e6e3f21a5040959797a5d7c7d0eda5 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/fcc20a38b6394bd4808d5ef2cc86dabe : SUCCESS in 11m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8ef929f7110f4ca0a4c24fb381c909ec : SUCCESS in 17m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3770c85b178d4e86ae3bc8a37d5014fa : SUCCESS in 5m 11s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/fe6d4d3245874a1c9dbdab175a7e1609 : SUCCESS in 11m 00s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/269fbc2e3c1e44d696646f15522aed2f : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72a5054d453749b7b96339cf41817507 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd88a518f4484475881059bae9640525 : SUCCESS in 21m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/70d5fc6de0f746c7ac7d85e6eff6810c : SUCCESS in 3m 52s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/013f3a1ea2584b6982ef65819d852c07 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19b9606b8812416fa73cb8e181835b4d : SUCCESS in 5m 29s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cf8c6ebcfdb443df9a7d9cb553bec076 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/247b04f552934067a125afb2591a1485 : SUCCESS in 47m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/041f0b868cb6482098f1f71a94492d9f : FAILURE in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"718511699fc6d36fc0fdb1a8dde355b69f342366","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-22 20:56:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a399dccccb1b9635b9390ca373ff967b0787f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:30:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3834b6702a1249e681979b372f624183\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d215d1b66bee42b789814357ea013464 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10d66a856754e2eba7910e930e188a8 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/143a69ab80e74416ac018ea7bb896b07 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab98a60527bc42a998ddd62955757e2f : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/756d2537e5d94aef9907e367aca3c9bd : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd739721ee3f4a3d8cfdb8f85edfa469 : SUCCESS in 4m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af622972d48e401ca1fd3ed5a3e2504f : SUCCESS in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1fa32d900fe04844b409c78ac63b47c1 : SUCCESS in 15m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa0cfc34e18640adae915685d320c6ae : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/625699b72e72496085c2e5c0c3a752bf : SUCCESS in 19m 02s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/85d4d97ef710474cbbfe268c113ee3f8 : SUCCESS in 4m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/f69efb69e50b49089e27ade4d42b4986 : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5a301a0db5d74f2782821e82c8591347 : SUCCESS in 3m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4bebc905b246418a843276384fbc3381 : SUCCESS in 10m 24s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/57756bf27634491384e4b34660370b71 : SUCCESS in 4m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1e7edb54ecaa4ade986f5e3ebe886077 : SUCCESS in 5m 02s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cf6f4ae724084cefb011853075fa6025 : SUCCESS in 8m 33s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/01c4320ee2cc4098ba7ab2839104fa8e : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e41d97df0614a06a6ddee2a194f5ff4 : SUCCESS in 6m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a11cd4d74be54a8894ad683cf087999c : SUCCESS in 7m 32s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6cab3e534f2749849ae0a0c0afe76859 : SUCCESS in 9m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f4e8b43dffa94390852a0b70261ed233 : SUCCESS in 3m 44s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b43e351d15c64d9a8c987072a67d0187 : SUCCESS in 8m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/8b9afb484e4743f79179695cf0b9bc3a : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2c4f84fc1c244b1695f7f24d5f61a6b3 : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0824faea390240868f3568b86235387e : SUCCESS in 19m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/43ab6bc5116546628e76a081e92b3854 : SUCCESS in 4m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/79d300f69d64440eab64ac484710f089 : SUCCESS in 11m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/998ada056170424ab4b5aa963c47cbc9 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a0859d53dbf46d2b3e8fd40fa4ff14b : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fcc3d20572d4ad5a1ade3050734ea11 : SUCCESS in 26m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5d8eedc55b4846caaf0eacaf7cfd6c45 : SUCCESS in 3m 55s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7855e52ebcf447a8bf5376f162a46bda : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0bc48f3a2cd04695ac68c3689038ae98 : SUCCESS in 6m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b3cef89f66f24f9e8ece2677cd4ec087 : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0c78d1ff37344a9a01c1d24f37f4248 : SUCCESS in 48m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c09271264acf4dd690ac2debbf3b010c : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2418c76a3c8f3cca276ee46f99284912cf8f3405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:30:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9533ab1e0731eab2d816d934e0ce0f9d72b765f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 23:29:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40d76001d55846ecae08d9ad8c3f4733\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bff75823df845de825e476a4b2cecc6 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1ffa99f9a94b5a9e18348cfd2fa5e8 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0ca2eb348e0495fb6153a156ae7cc6a : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4b816afbb9a435eaa571d897d6d8c24 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34c0089db813486691e920cef21d6c17 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df75875003114b69a57392ca999705fb : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1d3185c485e84b40806afa6b8dfb49e1 : SUCCESS in 9m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b82626fbd6324bffb080d6d55bb9a22c : SUCCESS in 14m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/15ae336087bd4b4babb1c008fb56533f : SUCCESS in 11m 18s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c25c05bc82c445b291ff9c6fe4d6d4ac : SUCCESS in 8m 00s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6c00b30f5d44471d947caf1ad403c1b1 : SUCCESS in 4m 58s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/078ab53e324045069b5605593c1cac6e : SUCCESS in 4m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/690285abee444235a5be7ba0ba14aeae : SUCCESS in 6m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b60438de00754604ab6f01f8848d6559 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/009888b1bbd4429aaadc863e67d8c1cc : SUCCESS in 6m 33s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/637ee773168e4d229389eed4afb50807 : SUCCESS in 6m 56s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a7be40c5b4f941038e4aa4b2c4318b49 : SUCCESS in 8m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/67523b06d10a4e4fa6435a07f9b3668c : SUCCESS in 5m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9f666ae3d5ef4caa8b148649c6856e07 : SUCCESS in 9m 17s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2d6c379315c845ceb76f7eee25cf88f3 : SUCCESS in 21m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b1e21b35d5944f7482b0a0bc825e90b3 : SUCCESS in 7m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5b7f6290d32942e7a5025f2340782aa9 : SUCCESS in 12m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b8eeba3bd20640e8a66f6ded79734df3 : SUCCESS in 6m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea0d54befab44e0785568c4528e7d7ae : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6dc0aebb71f94d9fa298980854a36152 : SUCCESS in 22m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c9df5616469e4f4a843921a2c0b5b799 : SUCCESS in 3m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3664a51d2c7049f4873ebe0786a329b2 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/738d4dc19f0d489089ea47ff083fe521 : SUCCESS in 5m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7c630e92a2ae4f2ab221867dd56e274b : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2d2db9b564a498e9287cd6f619bc089 : SUCCESS in 42m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7aa8a21c7284a278837f1dd7525f1a8 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"db5e7030eede366c3fc601c4c15fcac26f04e9cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 23:29:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"905719eba36eebf89b5fa4237298ecc4d20ad178","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 23:30:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74fc5de6acf84320828bfda02d0c7f92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9733b64c6734dea9d62da2d86b72edc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac406a7c7d8fb17f31921f8093d9b9f9005e148e","revisions":{"ac406a7c7d8fb17f31921f8093d9b9f9005e148e":{"kind":"REWORK","_number":1,"created":"2024-02-22 18:55:08.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/23/909923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/909923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/909923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/909923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/909923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/909923/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-22 18:53:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-22 18:55:04.000000000","tz":60},"subject":"Downgrade dnspython from 2.6.0 to 2.5.0","message":"Downgrade dnspython from 2.6.0 to 2.5.0\n\nAs stated in an upstream bug report[1]:\n\n\"Dnspython 2.6.0 has been released. This may cause software using the\nlatest dnspython and eventlet to fail with timeouts, as the monkey\npatching in eventlet has not yet been updated. The eventlet team is\naware of the dnspython release and a patch is in progress. In the\nmeantime, affected users should pin dnspython to \u003c 2.6.\"\n\n[1] https://github.com/rthalley/dnspython/issues/1050\n\nChange-Id: I993dc7b71f2629d04a10229e429f613bde134965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac406a7c7d8fb17f31921f8093d9b9f9005e148e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac406a7c7d8fb17f31921f8093d9b9f9005e148e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
