)]}'
{"id":"openstack%2Frequirements~909926","triplet_id":"openstack%2Frequirements~master~I8850519c3ff886f1fee03b61b76cc10a9dc0e1e7","project":"openstack/requirements","branch":"master","topic":"downgrade-dnspython","hashtags":[],"change_id":"I8850519c3ff886f1fee03b61b76cc10a9dc0e1e7","subject":"Bump dnspython to latest version","status":"ABANDONED","created":"2024-02-22 19:17:30.000000000","updated":"2024-04-05 10:09:47.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7ad6ead68a48981ba578f5a2564b0c6e92d1998","_number":909926,"virtual_id_number":909926,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-23 13:53:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2024-03-01 07:30:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-02-23 12:16:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 20:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"339de1dfa887f51fbc2f1c2d2b00229eb365430e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-22 19:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d118e4cad6208a4c39fe88383dfacc326979e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:46:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52313b4b5a4409bacc88139390dec78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51a30f86c514f298fa53f35c0f2e48a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ba9a07a0974434bfb001b1cd428b87 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59cec2f7f0234d5490b4d9454ffac72c : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c03021f0a2a44526af855fcef844158c : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e43fd9a0ea484fe29702bd6e23d899ac : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdc323d9798f40f69f8de055ec43e80c : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fab16e810caa4e0694ad077280fbd6b3 : SUCCESS in 11m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a68317964640427796f332487ab815b6 : SUCCESS in 13m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/66a07244f2d24d2681f1e0b1795e25ed : SUCCESS in 10m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f0e2d954000f44758378a0b9261a07ed : SUCCESS in 13m 46s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0a635a17185e4bf2b6882ce0d7daa889 : SUCCESS in 3m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ad3462dde8554d91aa58c25c74f262ba : SUCCESS in 5m 01s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0eba65588ad041da9b3a658829527965 : SUCCESS in 4m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2cf0ff8a108e432daedc8231b594e0a3 : SUCCESS in 11m 29s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8e15d14a769f45d3a9e02effa2a0b7fb : SUCCESS in 5m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e3546535f44448989bbd8b64818fdf7a : SUCCESS in 7m 47s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8db31ba5fa8541f79b415ca808f63e52 : SUCCESS in 8m 36s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2d431acc768549e0a6cb2584da8777a6 : SUCCESS in 13m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79e61258fe1f42a49ca37fbc5129bd91 : SUCCESS in 6m 28s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4f2c902d173a4ecc931b159a9e7c9760 : SUCCESS in 8m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8ba1e927f74144199305dba9924ffd73 : SUCCESS in 12m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8eb26efdb94c47088be58ed2b81deb0e : SUCCESS in 5m 07s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/60ab4fe785824a07b78b8dfbb5f3879a : SUCCESS in 8m 40s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b5e7061ce5754bce8f6c3955de5f3c9c : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f730f440ba864897b962b70330b829b9 : SUCCESS in 12m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/76ba974acb2e4921a3ddbbf916ebf271 : SUCCESS in 22m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd01f815d6c045b9aeda9db97b617964 : SUCCESS in 5m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/43d73062d8f7472690fe943519ec55b6 : SUCCESS in 12m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7f37f0c1059c4f0f9c84606d333157fa : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a638f42cad34fe3aed551cb08c8d15b : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1c52b9816674235a7e99e3a212945c5 : SUCCESS in 23m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a4116942b1ac42fe824a4640a11d2f8b : SUCCESS in 4m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c681de8bad0b492c9fc4feeaeea4203c : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6125a69555f2406ab60888e29832823c : SUCCESS in 5m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9c37ee4096814e37952209db1348eea6 : SUCCESS in 16m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cf2944340ba40afa4ead61140f9ec9e : SUCCESS in 1h 04m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf72a78a250f4ed89211659ac5c7eb2e : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"906441ada6f244ec2aa356c3140e578f0a13bcd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:16:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d1b1ba5b5d63a0a2fbf886f5b022d3e31b2a7d51","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:16:30.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db63e9918da164555e75748ea9d691153fec247f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:53:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c26f559273467ba5d59bc259f358c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fc404631344f6980dd2773c547d3c5 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bdf8de6dd64f51bf0383c2e19afdaa : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05944941838c4a8581ec8b762f4e505a : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bc6cafba1d94c81a2150e3d3ae3488e : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85f3e0c17b714aa89a11ba2ea704500e : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9676017cc3ba41129fd3f5df47dee823 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f164b783f19e4c8dbe5a2b1f9be94f4a : SUCCESS in 10m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b38bedae17f54de8903962c3379de803 : SUCCESS in 16m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/01172859d0b54471b39f7866a694164a : SUCCESS in 11m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c4d39367f754cc382cb3f40fe15634d : SUCCESS in 17m 23s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f0303ecdd67142ae9b319a0e6409bef4 : SUCCESS in 6m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6361e5ca255e40dfbddaa1d99c519093 : SUCCESS in 4m 51s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f99a62a8a6504f72a681596597e5b83f : SUCCESS in 3m 39s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8327fede46804843b08625b3f792c377 : SUCCESS in 12m 40s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/378fba167df7484e9c101061d8ce4c8f : SUCCESS in 5m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0b86b9db2c4a4219939ce88cf4d76724 : SUCCESS in 5m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/80148a1deb0648fcad6a6464fb447b03 : SUCCESS in 7m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/58033b4775db43a9b86c086a30b255a0 : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b3bcfe7a9354d3ea28030b060f6bbe5 : SUCCESS in 8m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7cfc30fe8d92416cbb2f11053f16e35d : SUCCESS in 11m 58s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/beaf64aadf734305a4386adcb46fa5e7 : SUCCESS in 9m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9dae45fb35094b7d955bd99e99545942 : SUCCESS in 4m 28s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/986c6a2b1a1d4cd1a0e20cd75917d8b9 : SUCCESS in 10m 10s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1e6fdd869f1c4b87b6bcd84382973662 : SUCCESS in 6m 54s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d6345f1ebf7648d9af3eaab60211d270 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6dc48273d5de4acda3bf4ce4b9c9149d : SUCCESS in 21m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/31147742627d4ac280a84992ed240824 : SUCCESS in 6m 32s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a7718bcf2e77493cbdbdf990d7b0f869 : SUCCESS in 13m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/68214548fadd455183613ccd8951f016 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e1893b8da474964a682e80ebbf0bdee : SUCCESS in 6m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d6c9df245ba47ef97ec2269b4cdece0 : SUCCESS in 20m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c500ae5bc4d0447198d895785f9c7ce8 : SUCCESS in 4m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6d47a7d62c474db39250197206ab3945 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15dc704e881344f7bd75f85e390ba314 : SUCCESS in 7m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c150e5a55d8941eaa471c62fb99e5d4d : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f919cbd149ab4cd9bc6c7dd07d97a440 : SUCCESS in 54m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1112970ea8d448eaa7d10fe5b9e44cd1 : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":2},{"id":"75526bf85abdbc256e2c7d6b2c878e37f00e0e2a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:19:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9337b7265b44fa78db86932cb4cf45a\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca3b3b66255e4c66bfb2febe73b08598 : SUCCESS in 1h 52m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/f521b86431324285b296603763b4f9c6 : SUCCESS in 1h 57m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5504762a530c4df7b58e16b211813ee7 : SUCCESS in 1h 19m 00s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/856055f38f8f4a32bf2d363c07bdd79b : SUCCESS in 1h 46m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a58049a16c8009d1c1703d84b476bcda88f123e9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 15:15:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b4983627df556df222eac1c4947b4529263e1c0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 15:15:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb424f21518330cbed9865c4b177f05e2c68fd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:52:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b4ec1deb3a418b839db635fe8d286c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d1f10a103e44a38df1894c3d96b9c5 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50339899381b46218023fe7cc7306ff8 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c15ecf086384bc18f2b32ca46cfc219 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d0e373a75b54c5ba0450ff8a87130d6 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e045605f9a314a64ac347fce7a83c2ec : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c779832150d040a79fbd763b860ce199 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/099f7f02ac0e401985e5be949071f55d : SUCCESS in 8m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8420a00a1c16491d866d02a7f8f69bbe : SUCCESS in 15m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/55d83d73faee4fce832a29dec50efc5a : SUCCESS in 8m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c2d8484405f54fafa7f13971870c3bfb : SUCCESS in 14m 35s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/890600e8a47241b29c444a942c45971e : SUCCESS in 4m 27s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0272835732734b2e9a4899d3f029eec9 : SUCCESS in 4m 41s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0d004bbd58064f548b189f6facf001ff : SUCCESS in 3m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/50018e3d24414a4dbc03c06164ff5755 : SUCCESS in 11m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/219d8244951d4ae19faf01193ce39a62 : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/68afa63c96b94a73b7b045da96ffe133 : SUCCESS in 4m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/32f30a39241b4a4c8023c0313e931e33 : SUCCESS in 5m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/99787c3afd09430ca58bc397ceaf0b95 : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dbce72458cf46b6bf1ea47f2910246c : SUCCESS in 6m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a77f5f1877e6463b802762539e9e117f : SUCCESS in 8m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bde757ec94ae4bcd8f0d0f81476b724c : SUCCESS in 7m 26s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4b71e1b81cb04fdf8a65ac54324f4f8b : SUCCESS in 5m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/27e895425cda4cb69e8476696dfd8823 : SUCCESS in 8m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/28d1caeff5ba473798b9bb3b9b103c3b : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3fb8c0cb18264fb283243efe59c37288 : SUCCESS in 11m 00s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/02bda455f87b46de877f19513bb04d5f : SUCCESS in 16m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9372f7c111df4d1e960b341570650498 : SUCCESS in 5m 55s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c4f9a39120414613b505870992fae9db : SUCCESS in 10m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3600bb453d8940dfb86221073239fde5 : SUCCESS in 3m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a96b540beb4045b89feb8433ccbc5cda : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eba2a445a454c23b30990896c79d3f9 : SUCCESS in 25m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d7834412ced74023a94e4bdef8c1e894 : SUCCESS in 4m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/080c7512927546ef99cbbaabc8c5f35e : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4e9e21b088a4462ad12dfa5263655c6 : SUCCESS in 8m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/331659cebcfc4cfd97576bcc4b40f573 : SUCCESS in 18m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/514ceb0d13d74b81a3cdc0560b197735 : SUCCESS in 1h 07m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c80fbcebd49444489a5a94446e6ed0a : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6e013d552443971e0dc3213054d558816bccc59f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 17:26:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ebdd65399749259e310149892e08c4\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2536a64c38994004831b7029314f612e : SUCCESS in 2h 03m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/20ffa3f4f6d4452d952d08eeac3936b1 : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/383b202347e04bbe9a85aa8b4830649b : SUCCESS in 1h 33m 12s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/001bede983294841875096bf262c5825 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f0826f445ac50ad1512743ffe114037c799b3852","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 07:30:44.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7ad6ead68a48981ba578f5a2564b0c6e92d1998","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:09:47.000000000","message":"Abandoned\n\nUpdate was done as part of I72a777f3a553ba720f80fd66ad7a488b55de51f4","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55536172e3758110219b43e58b9bca67e2b61d44","revisions":{"839312cd788c0be28d18f2f4cda7bfde7397169b":{"kind":"REWORK","_number":1,"created":"2024-02-22 19:17:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/26/909926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/909926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/909926/1"}}},"commit":{"parents":[{"commit":"4e7b7b8283c99488b48d38b8ea498267dd7b7e6d","subject":"Bump eventlet to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7b7b8283c99488b48d38b8ea498267dd7b7e6d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-22 19:16:51.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-22 19:16:51.000000000","tz":60},"subject":"Bump dnspython to latest version","message":"Bump dnspython to latest version\n\nChange-Id: I8850519c3ff886f1fee03b61b76cc10a9dc0e1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/839312cd788c0be28d18f2f4cda7bfde7397169b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/839312cd788c0be28d18f2f4cda7bfde7397169b"}]},"branch":"refs/heads/master"},"55536172e3758110219b43e58b9bca67e2b61d44":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-23 12:16:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/26/909926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/909926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/909926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/909926/2"}}},"commit":{"parents":[{"commit":"9655292bc14de1e06216e3b09b9dbf4f22fd8516","subject":"Bump eventlet to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9655292bc14de1e06216e3b09b9dbf4f22fd8516"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-22 19:16:51.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-02-23 12:16:18.000000000","tz":0},"subject":"Bump dnspython to latest version","message":"Bump dnspython to latest version\n\nChange-Id: I8850519c3ff886f1fee03b61b76cc10a9dc0e1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55536172e3758110219b43e58b9bca67e2b61d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55536172e3758110219b43e58b9bca67e2b61d44"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
