)]}'
{"id":"openstack%2Frequirements~909930","triplet_id":"openstack%2Frequirements~master~Ia9fbb2f337097ac5a70cb72c5e2e1f3adc459204","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-26 10:02:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-28 12:24:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9fbb2f337097ac5a70cb72c5e2e1f3adc459204","subject":"update constraint for oslo.db to new release 15.0.0","status":"MERGED","created":"2024-02-22 20:16:56.000000000","updated":"2024-02-28 12:25:56.000000000","submitted":"2024-02-28 12:24:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"909930-new-release","meta_rev_id":"ff9186284b8d5323e7864a8d82462d2c7e3cf248","_number":909930,"virtual_id_number":909930,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 12:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-28 03:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2024-02-24 14:55:53.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-28 06:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-28 03:56:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 22:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 07:27:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-24 14:55:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-26 07:50:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-26 08:27:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-02-28 03:56:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-02-28 06:22:59.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"}],"messages":[{"id":"b85244ca0cc701e04936cd8eed466d0cabeac80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-22 20:16:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a099fe633005e275a58efd44b578f6a1db97d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:05:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbae5021b7ea4cf2814d63f2979aecb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b55d38974e34a70b4a24f5e1e77c986 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a7f00521964a9dbb2e2d5473e4daa2 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69e745ec9126438f8d0f843271df4635 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19f7601819664e4987fec56cf4125de2 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61fb96dac8d14aa093e77c82faa532e4 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1d90a75094e491aa941ef77fc4a2bea : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40d04ca1c7d34cd29b662671dca23088 : FAILURE in 4m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d8a7b990de874fc19a9b93e7a0f254da : FAILURE in 3m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e96aabd15ffb47fba8e5e0298ba99c4f : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5df5c8b85850410497c6c58300e3dd3b : FAILURE in 4m 31s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f7ffafcb5487492ab2ed40258b17f83a : FAILURE in 3m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ff13b631308140d194b6891e29eb34b2 : FAILURE in 2m 14s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/54b0ad1c9989429c9882b8ea4ed986de : SUCCESS in 4m 09s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b4134d01bc7c472ba25d68d33762bb25 : FAILURE in 3m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/dcd7f59dd28d4cc3ae1795314a93ff48 : FAILURE in 2m 12s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/07e3e4ced061413389662a9b9253e99d : FAILURE in 3m 50s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b3fca04033ad4f52b5dfc9702fdda574 : FAILURE in 3m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/48180709794249b9ad8a9c57ce0ad715 : SUCCESS in 13m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3f1e704aac74a728fe2b75273bbcfcf : SUCCESS in 8m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1ab88e0d1b434190a2ba1359de4b39d1 : SUCCESS in 8m 24s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6c0e339512a6414187a9319cce8fa1a9 : FAILURE in 2m 58s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6eb71a56ea8a4a29b2d82ca9568faa95 : FAILURE in 2m 25s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/60ae28b2a756493b93c79b27beeb207a : FAILURE in 2m 58s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0ecb96052f4b484cb24eccfc7e75d590 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/12011f3aee734a84a348f12afff2d2ca : FAILURE in 2m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3a2dee0099ae43b69f897e89eec5adb1 : FAILURE in 3m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a5a5195f67104b54ac547c35f0debf30 : FAILURE in 2m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e4d60452db624c549db26a2ec3841a22 : FAILURE in 4m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/501f89c1460c4e9aa37700ee5cc225ed : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f6924a63d744f6da082187dabcb2421 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e4041090b4b4099bdb97f5b72f8ab4e : FAILURE in 3m 09s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7e362ad5e01f48e7800ec20503eb6d15 : SUCCESS in 4m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5eca18b15eaf4ffb92c601c966c841fb : FAILURE in 2m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aff55e01402e4bd0945e168cd66acb5b : SUCCESS in 6m 10s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b51449d3063c4e9cb77e61fe911b25e2 : SUCCESS in 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1aee265cab674fa9b461ed0bb8c5bd2f : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94cd5afc2aaa490e90647f57ebc675aa : SUCCESS in 1h 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"11a86ff4519e1b40d4ca7f43ab4a09a84d9d9350","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 07:27:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e17abaebe67bdefd4cc1372181b0964c361787b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:13:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb600ea2e225445fbbc8649ba1598ab8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24070688d7fe493897faba3e277dbc92 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7ae4dc42f04668b13b1516f4380325 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9dc59c0a4b84eb8a4510b66ab4339a0 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05d33ea9a3e44e7f8eee25c135915777 : SUCCESS in 2m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efc1a528595a4466b4d5f91535f8b0eb : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b0e48fd61a748738c81deef353bd77d : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d02fb784c824f5ab266206aa45ced7b : SUCCESS in 7m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/82c0a93d54c44ae1bd3aa009f9dd1c01 : SUCCESS in 16m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54d37056fc81480b90dcdf15d1a00a59 : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ec6e9aec4ca4b8ea49c8e768620cb7a : SUCCESS in 16m 21s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/154d3bb3644440c890aa40d6a67963f5 : SUCCESS in 4m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/fd411f159c2845cb8dc191866855df57 : SUCCESS in 4m 43s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e74874031668435b9fa24d10214bb96a : SUCCESS in 3m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/71204fe8b181441f91cab7e577f6637c : SUCCESS in 8m 00s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/82b7c0c678b5493bbd75d12277f7c7ec : SUCCESS in 3m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8720b1316b8f477a8a9ef845c93e364d : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b2eabfb00462497d99fd80ecfdbce7b8 : SUCCESS in 7m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/dd6f617830ff4a34a9698ee73c24d130 : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f52f71b2a3654cf2a3a1a70b49103ecf : SUCCESS in 5m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cca03cd2361f4615965423f9ca3c56f4 : SUCCESS in 8m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/35c0da1f15444d71b367a306e296f639 : SUCCESS in 11m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0a891171eb2b4463bb8edda6d92d1f72 : SUCCESS in 4m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4e19cd749ce949788bd153218e9a10c9 : FAILURE in 3m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/97c2321f475c4e3ea2786757c55f4b42 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a831242876eb411c9c360b65c8c030c2 : SUCCESS in 11m 17s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/89097b12ca594ce1b7dfae6e4c8981e5 : SUCCESS in 24m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a5aee1f4be1341779c37c3879de86122 : SUCCESS in 4m 41s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/87475a34cda0433789432cec0a7040a8 : SUCCESS in 11m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8acddc71aec149f1901e7948e5dbeeb1 : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a13f44651fb44db8468ebd280e109f9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff1895f80936467784873aa0e1672a81 : SUCCESS in 16m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8d36a3903bf9499d8ba2b5cb682040d9 : SUCCESS in 4m 21s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/963ab325f8d4480d9bf5efdb67597426 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e9dfd13fae24891bdb17e9ffac3879d : SUCCESS in 6m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b0ac1b73ba284f2e971d7289e3c7e6c7 : SUCCESS in 8m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2ab8c7119074e9790d8052712de0709 : SUCCESS in 41m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf9082cbd0247099aaa2888dd77f8f0 : SUCCESS in 1h 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6c55f0e5f39e9a3e0787a929a8fabf8d9fe2256b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:20:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30e524386dc08c308810dddd1d27b13219a9da07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:20:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"0a0f3183bd175a66cc0286bd38d40bc4b485a841","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:20:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d9e600e9c88397878e234f0b96d1d67a5ba5276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:18:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6eb5e9dd9674a7096f55077a9fd9362\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4956729fdf841a584bf360e1b952e69 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04530399d1d24f1bacae75b5d50e3a6e : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47195e215aa941e6bb0b78659464ef9e : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5cee5649c0545699c57d6f1f8d3eddd : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccbc5d1e2d364f5c9b240598b0691cdc : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ed411f7fd3f4f998a8da9dd20c403fa : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e598f22310c7478e939cb423545b6ea5 : SUCCESS in 9m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a50ffc731cae4ec09d8c66359661375c : SUCCESS in 17m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8485700d40bc4942a2eed92e02686676 : SUCCESS in 9m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a28ce506c3cf4756bd9b37759fde83bf : SUCCESS in 17m 51s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e67a402ed0524e279708ceae458b7298 : FAILURE in 4m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c7ac6355acf0455eb14d2b77d0a7a725 : SUCCESS in 6m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bdb1531c752140029eec69333dab8865 : SUCCESS in 5m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8740b742d15e46449090a53c63433569 : SUCCESS in 11m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6d57842293c64a1489ef1471786f97c0 : SUCCESS in 5m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a588cc4a06134343991002ca4be5f834 : SUCCESS in 6m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ef0a1e1de138492e8e8c4aa60c70adb2 : SUCCESS in 9m 33s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/501935a6a71648e1bce8fdf89fb5d787 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c308eb17d17d43e4800d1b01fecf942c : SUCCESS in 8m 05s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c53b61923ef349079ef51c94dcd1de15 : SUCCESS in 9m 48s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0953a8b2abd5421e90cae0bad0cb0260 : SUCCESS in 11m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f2abada508524167a7630c7639cd917b : SUCCESS in 4m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2c4c5c116dbf40bc9189705d3695e8b5 : FAILURE in 5m 58s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6e1286ee426f4dd5a8b9f96172f3728b : SUCCESS in 5m 25s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a768ccfd2b804996b8401d2991023785 : SUCCESS in 12m 40s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2b1fa9b8b48f4260a2b25bd9dc5f6012 : SUCCESS in 26m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0953231fddc48798809d675b78f8f45 : SUCCESS in 9m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/35e96adf1e814088b99c89f472c09478 : SUCCESS in 13m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c7223307b5e54dc6b6597526a28ce218 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/76407b79c7c84e1f82d600398ab4fab8 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92915646898b4c75b7b310a36073da24 : SUCCESS in 23m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/777e5199a35545b5b9ac01ecbdd90866 : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a5df44dbfb9f4a639914d37679c4b839 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5023eb07f0cb44bdb97482a67e1e242c : SUCCESS in 7m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/87aff700971248a9b341086c55f3b0b0 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4af9a309e7514f04843d12efbd036d5f : SUCCESS in 1h 07m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e56bc572290c4b5fb43a51204364ef09 : SUCCESS in 1h 46m 55s","accounts_in_message":[],"_revision_number":2},{"id":"de9d6d5e8e78d599d0d53742f7885c51afe4b3fe","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99306ddf31414fecaa03b8bc3fc99dd6\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed0bddfeaf9f413e8d3677045455d016 : SUCCESS in 1h 48m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f7aa5ecb7d0432eac6227983179b986 : SUCCESS in 1h 12m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9e8d609475a47818e0c6e23d9de87c4 : SUCCESS in 1h 32m 19s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/91c60707b2c24ec2974997d5233d1e01 : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":2},{"id":"19b6e4dbf37bf46ff678f5e151eb7ce71f975fcf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-24 14:55:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce1904d95f0c3fd79235891baf5039b1dea1c868","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 07:50:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e784ee212e6d6292b4a9ce51af6c40346be1ae74","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-26 08:27:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4209ad5ff2771764a0e92d6f75415848ffc4ac26","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-26 08:36:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"010d180e87f4af996c6f2f2e2c45cc479139f38e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 10:02:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"549a6256c62ac19f92016060eb9550255b6fe36f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 10:08:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4a785aa84addc8706c9de25da09a1d1dd2ff41b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 16:44:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3914c395ff2ba06cea53b59431dbeef70028ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:32:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e54f0540139145f79afec1d476ac437c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61b39b90baf4e439766948ab511cd0d : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56ff03f6ca246a09f5c1b173c66f1a5 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a7bdd4dc0e74d9ea208ba32f95c0cc8 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41eb9252f0e648e8bd031f069f884336 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd9f2be4f25d4de48dd707ac3b6ce366 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d9a060119394fcfb408f98944696fb8 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/711ce6f86b014f39aa841fcb340b83d2 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6463a0fb31ea4ff68c27ca9304b43cab : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9641c1b8d3544820b7974672915ebe59 : SUCCESS in 9m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7af18161dc2c4e00b54391826c5dc666 : SUCCESS in 20m 05s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2de46f24560b4dd4b9c541de58aaa78d : FAILURE in 4m 21s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/461db40ecd454b7abc0d24a81f1e6644 : SUCCESS in 4m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a168755596e84739b2bc5a28beb45dfb : SUCCESS in 4m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/effeeee32b9b4ccc9c37eacbf5f94a77 : SUCCESS in 11m 28s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/779d175cecae42cba516d7d91023dedc : SUCCESS in 5m 08s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/07d655367436423c967e2c1e26cd127c : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/17ac26c2dfcb4bf8bdfd14d4eec83e4e : SUCCESS in 7m 47s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/49d954c8cae6423f93c843fae847835a : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6997051008d9433d9c2d5f092afa4625 : SUCCESS in 6m 45s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/90e30c6cb1234597a6e4ae765d1eea62 : SUCCESS in 8m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/81120521b6b74551856f1b7277122cc2 : SUCCESS in 13m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b66efa09bd364292ab5c064a56085af7 : SUCCESS in 5m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a4be840e295c4d2fb95d83d247a9a923 : SUCCESS in 10m 00s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/17ea7a8062f74fdab47ef48a304b4897 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/706b01e0c59446faa156d6b1f84d96ef : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3dd359dfd848400ca2390764541e0d6a : SUCCESS in 26m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bf21790b5f2e4337b8580e5ab206f8ca : SUCCESS in 6m 41s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c9af216e1dee4ec2ab4192a903dd7b91 : SUCCESS in 13m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c7202901877241ddbbafc971bce09cc4 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1018eb7bd8754fe2b20db0d8f0713a6c : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9e8dc1ec00e47eb90672a2e34257cbd : SUCCESS in 24m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/01156a4ff0ec4e86a5fe41f5bb515d2c : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f87058902f2244a892dbe396bdc68396 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24294817ec494727b956f528c7441e9c : SUCCESS in 6m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1521acd5fbdc42acbd6b6ecf7b26a783 : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c02e46011ca411d9c8a5b85a887f0de : SUCCESS in 53m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57602e0d63b44ee188a38984e28e125c : SUCCESS in 1h 41m 05s","accounts_in_message":[],"_revision_number":2},{"id":"419dfe43770fa0fafb20d74dd044c730a77b9fb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-28 01:17:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1996365242e50b8cdf35cb53a297172fd721ee08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:06:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9fa486fa9f74218b67228336e390786\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d81841813a45ad8db6e00e31853520 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45872a0614bc408a8aefa802a7ded1d4 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7774379625647458491889f1145d572 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e79ef1ef346a4907a07e035ce6fb5872 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9af0462247d4ce08be3dad13ead95a0 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f6d91ab265b4a9a931a37c8b059a7b2 : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d49eb36835d4049876832a44754ddd4 : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/578a22d148a94c0ab7e40f1e239780a0 : SUCCESS in 18m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/66eadf77c2d34410bc87c1e15d3927c4 : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7692fb14d4a146c2bc57582ebc007235 : SUCCESS in 18m 05s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/4d4b0c3f615a46aab4073cfddc7b14e5 : SUCCESS in 5m 13s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c02a3e2da9d24b299f2fb4e23e3015b9 : SUCCESS in 4m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/be94fb973e1c43a5a892f0810a2eb1b6 : SUCCESS in 4m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d8cb8b198b934f7b90d2246ab9b8874f : SUCCESS in 10m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2c52f53e44e0446091c38e5aaa058e91 : SUCCESS in 5m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/83ff73747acb412a8295994080949d1c : SUCCESS in 5m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/032bd017ed12443eadf241ffcde6f7fb : SUCCESS in 7m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7068364073fa470d91b47796a5c775f0 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06dcc4e779b743f599e43cfb5d10b10f : SUCCESS in 6m 41s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e80481d72d25482999c4ccb3f864a95d : SUCCESS in 7m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/98acdbbdcf77493f8297a8c80b056ea0 : SUCCESS in 10m 55s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8706cecf7bbc4672a71eb37fd5ee3d00 : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/04639e224548463a87fc3f7fc72884e5 : SUCCESS in 9m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/52c0fe461bf2496fa354da435d6ba39f : SUCCESS in 6m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/180bdb28188d475689ee2bcbde43fcf7 : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8b1801a07dd84eafbf8d819095d798ac : SUCCESS in 27m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c78756ce977b463c9522f1afd372098b : SUCCESS in 7m 47s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a64775e5e92a411a8eef2242c1cf326f : SUCCESS in 11m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/46825bbfa5244fbaa99d84daddc9d0e8 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3bbaf32b39824b1ab451b3d02a47df10 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92088412c2224a01804e1dd85483d7fb : SUCCESS in 20m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6307cc968b1f40af94b41dfa4911048e : SUCCESS in 4m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/88527013133e4d5a8e16c905a7bd9ddd : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/839e68fe95c64a088b1a6eff3e156b85 : SUCCESS in 6m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/042e8ae968ba4800a4161dc46edc112c : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/391bc30e0acd4597898973fc28607431 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2664dc8af58643b19d4bb2091a723117 : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":2},{"id":"df43cef9a01268a7dad46e299ddec8401f1dba44","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 03:56:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a92db4d1fe34d25c369de086d67d76ceac3b416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:57:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5cd493bbf6a7eb3a757fd234b642e236981eef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:18:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43dc580bc2ab47ce923c1c8a09875c23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c785f59b53fb496786884fe4a6a2345c : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853d576dee1d4afe9fe48ea2c0da1a30 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/841c2af7fa2f4e33b8f1dcb9d279c4a3 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b090a294739430ab8fafc2474e8ae12 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0dcb77741a54c3f911a0e317654c5a2 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5d5b44704f244989c568cc022243137 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f2f6445aa9e440683f13706faf50886 : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f1b6ed51bd2a4fd88817c00037d72092 : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9ac63687991453aa063280981b5b888 : SUCCESS in 13m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a417f6be642e4ec991ac213f528c107c : SUCCESS in 19m 35s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5554f639b01d4948a0db9da96d9f3ed6 : SUCCESS in 11m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/896135bcbc0d496dabaeaed6f1a2f77a : SUCCESS in 6m 00s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1df05a91859c4e00afc5b99463646c8a : SUCCESS in 6m 51s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/4042a2517afd4d46b77d244c965f33fa : SUCCESS in 9m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9cadee45989a43fbb778dc001a665613 : SUCCESS in 15m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86cba0bfd6ed474aa1b78775ad5ddcbd : SUCCESS in 8m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/af245f5123be4076b6dc3f3cb2c5b6fa : SUCCESS in 9m 34s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d9c0d89623bd4b64b59cf20215359955 : FAILURE in 11m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3e54755926f741288661c55059a37280 : SUCCESS in 4m 30s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bf2bed7c81664cf5873d960fe619cfef : SUCCESS in 10m 25s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/89c86d3ec15440b888b5f8adc51e8d78 : SUCCESS in 23m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7d195c413a594e6b8e2913d3321b4abf : SUCCESS in 7m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/024561708a2648f583663f25ed60c508 : SUCCESS in 12m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c625c9a51a77421fa767c9c666a76913 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2b2973b677a405b95b2fdb84b155577 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1dce91288fc54fb0bb13c81de09262c1 : SUCCESS in 26m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6b6140fb4479409daa23ec69ee8f9c6f : SUCCESS in 5m 34s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9e072ebf794143bc9b24b1f233e2ae1b : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a5cf482d0034bab92a254b04ac5fc79 : SUCCESS in 7m 20s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/64da1fe724b24bf888e7cc6454ec978d : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f39c2a37eca14446a0337c2cff57b714 : SUCCESS in 59m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1aea03187a548b4ab0d032d8f43f246 : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":2},{"id":"30344c5873b96aab9833faae277403d2e0f58e2e","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-28 06:22:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7c146dfce3643d1424cbc70c0516f7d62e12ee5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 07:27:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1065f67b69d6cfc81d825b58f87369a9fc735fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:57:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc18f54ca75443bca60bda639f4c84da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f666a8078dc4c889c8ba355a7003376 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41db4907bf543388e7fe1dc323fea7b : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/085524a5f8884be48c564e31edd707e0 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09428a39523346aeb1bdee96a587107a : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38575684599c4becaa23b4373982448c : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fa07f468fa149f6abd0f4362794b813 : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/36868a83bc784856b73bbd20c8c37c38 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/61394c3fafcf4ab190909d9774f6c515 : SUCCESS in 14m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ef75126099a4f5296d8b6a6aad7e4e5 : SUCCESS in 8m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/53f2ae4bfe384afc8ed7c0d8829f41ed : SUCCESS in 13m 04s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b2597a3ea80c435bb79afdbd9349a826 : SUCCESS in 4m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a706c93eecfe4e2599f1897369d06d43 : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/428ac105d66e41eab2b05d9525037d7f : SUCCESS in 3m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/888bb4fe9b364f208130469a146c51d3 : SUCCESS in 9m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/39aa7241324748beb130f34306cc6ae7 : SUCCESS in 5m 14s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e31972c2365042f594c98ddf9b005bd9 : SUCCESS in 4m 50s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cab3ad9416024657857c414b588282d0 : SUCCESS in 7m 29s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/499fc821e09940c3811d03634103ae14 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af5f67d759ec492fbf21a774dd040a7d : SUCCESS in 6m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/aed365cd001c4595b66af7a35eee5503 : SUCCESS in 7m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5d50b671feff410e8e187e97e6f1991b : SUCCESS in 8m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d91d20a4a52843b4aad6fc3feeb14799 : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2d1378e409b743fdbed816de6d7bf6c3 : SUCCESS in 7m 42s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/41575c2f3d394b2c9d5012f6e20de9b9 : SUCCESS in 4m 03s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9944e0b14e374d9681a6528a92073f5d : SUCCESS in 11m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f6b84da8eafc40c4b649e70ca64e195e : SUCCESS in 18m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b9333cdc50b472aafc869804d3c6867 : SUCCESS in 6m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c86311617b834708a728bba2eb385b26 : SUCCESS in 11m 00s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/abae06b84a534a1ba4f1f85da3262996 : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90caf6daefa347efacc56df920ef23b0 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3f8f52782eb4b31aa84c0e290227136 : SUCCESS in 26m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/70cc071e6bfd4be9afb319a90fa7069d : SUCCESS in 4m 27s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d913ae41eb8047938e5949feb9e05648 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e519ba836cab4945b6b7c7a7d7c0263c : SUCCESS in 6m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/645081a671054b56926e4850bc511e64 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34b56b27f7c84e61a656c03575d74a12 : SUCCESS in 51m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0d37fa18b864a25a902ee0e094ace4f : SUCCESS in 1h 29m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b5e121705dff94a51cb1c73853167a822a487385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:58:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd0474881ac5e1f322bcc87661416eef866f7465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 12:24:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c8eb18a3f1f4a3f985fadad012b87fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a652b4c725684e7796721be75cde23ff : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7618a3f49b1b4afe88bd45421578a9f1 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb7e6e4510645858ddce6235d23682f : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe13e8d3051540eebf3d3480195e40dc : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8547e862c5fa448397989faaad85d32d : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/566b0801060241f488ffbe004998ab0f : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08c0a51b0674464c8536ecb81152339a : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e65da87ea26a4016af204fee322d0200 : SUCCESS in 19m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e152320889204a898676ee2d7886c802 : SUCCESS in 10m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/72bd49a099864ad29efaf087d3c5ca4f : SUCCESS in 22m 37s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8a90375834624c32b1dcfd5e8ec087ad : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/746f0278ab9e42238d3592cc4b98dc4e : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b8bd81ac396d4e519f8d369bb0f69293 : SUCCESS in 5m 31s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/db8a896de5604ca3b6a14d18c16bc785 : SUCCESS in 8m 16s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/40e80a0c0f4049e89cfa12854066ec72 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9eadb92003a4faaa6a64eff1b2a82ad : SUCCESS in 7m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d6422ac375a84263a91342720838ca18 : SUCCESS in 8m 20s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f377d102e90f4341a24332e22e613c4b : SUCCESS in 11m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/eaf40e97e8e04912bf73ab8be3abf8c3 : SUCCESS in 4m 26s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c6a8e6e6cafe45469acdf54f43d5a0c1 : SUCCESS in 8m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/17c110ce5e5a491d9e0ff0bb4d4b2d5d : SUCCESS in 25m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc3f462a2a00426e9a2c98b59d153135 : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a36c41ffc8064f2f9d3cb0ca793a7c07 : SUCCESS in 12m 05s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2e28c38e44c04bf184d658c01ae7d20e : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c283de4f9c144488b9cde82d878b75d1 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/beda5c8962d1468d8ceef389736b35f9 : SUCCESS in 20m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a62d9126e8694bfebb198ae9430e2f60 : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4784c5ce48334d5089d2e52b5745349d : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c217d1865cb4872b7fb9737a91f6ff0 : SUCCESS in 6m 02s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/229225354a0844269d31d7cbab9046ed : SUCCESS in 15m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0da3a13755d64e4bb1b69e9393fec486 : SUCCESS in 1h 20m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63117324196b40cfb2f93099c30e79cb : SUCCESS in 1h 51m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5c97d609117989b62a4d77638c62921a2640e9ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 12:24:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff9186284b8d5323e7864a8d82462d2c7e3cf248","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 12:25:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b35b033f5842e590876acfe8bb2f4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa1918882f449e09ce483ac2e7c1092 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d361c50e933db1ba1f291c2289a20fe2fb240ac7","revisions":{"f0b3d1ac6d77e2793b51ab63ff21cad3b85b5e4f":{"kind":"REWORK","_number":1,"created":"2024-02-22 20:16:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/909930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/909930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/909930/1"}}},"commit":{"parents":[{"commit":"330559fbc75e42160825e949d2301c1886963f55","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330559fbc75e42160825e949d2301c1886963f55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 20:16:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 20:16:54.000000000","tz":0},"subject":"update constraint for oslo.db to new release 15.0.0","message":"update constraint for oslo.db to new release 15.0.0\n\nmeta: version: 15.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id3ffd09fd70ff74562ff5f6975448e7d97bfdc8c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia9fbb2f337097ac5a70cb72c5e2e1f3adc459204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b3d1ac6d77e2793b51ab63ff21cad3b85b5e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0b3d1ac6d77e2793b51ab63ff21cad3b85b5e4f"}]},"branch":"refs/heads/master"},"d361c50e933db1ba1f291c2289a20fe2fb240ac7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-23 12:20:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/30/909930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/909930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/909930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/909930/2"}}},"commit":{"parents":[{"commit":"f910188f4977a5a17996ca9db7d2ebc83cefbdd0","subject":"Add more experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f910188f4977a5a17996ca9db7d2ebc83cefbdd0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-22 20:16:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"proposal-bot@review.opendev.org","date":"2024-02-23 12:20:44.000000000","tz":0},"subject":"update constraint for oslo.db to new release 15.0.0","message":"update constraint for oslo.db to new release 15.0.0\n\nmeta: version: 15.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id3ffd09fd70ff74562ff5f6975448e7d97bfdc8c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia9fbb2f337097ac5a70cb72c5e2e1f3adc459204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d361c50e933db1ba1f291c2289a20fe2fb240ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d361c50e933db1ba1f291c2289a20fe2fb240ac7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
