)]}'
{"id":"openstack%2Frequirements~909956","triplet_id":"openstack%2Frequirements~master~Ia7a84da5e4b6f42df36b76ec5760fd6b51e008b2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia7a84da5e4b6f42df36b76ec5760fd6b51e008b2","subject":"update constraint for mistral-lib to new release 2.10.0","status":"MERGED","created":"2024-02-23 08:35:48.000000000","updated":"2024-02-23 18:29:36.000000000","submitted":"2024-02-23 18:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909956-new-release","meta_rev_id":"c423a902ba204af1199fc096b0d42ab91bfd0d11","_number":909956,"virtual_id_number":909956,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-23 18:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-23 16:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-23 16:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 09:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 12:32:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-23 16:41:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d7c01d7dd2c06e349ff67f737ad8adae7c03b03c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-23 08:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86209a37aec4e2b8bdfa77336da14d917fbc2af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:55:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bc8da39f8a54705b6771d255ab7a23a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb76165fb03348ab95bc3018ed943ce5 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6a6d71d5c84f0eb749e5f3366983ab : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f08bcdda91924417ab0e19da5a1aa5e1 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc449144147543919f860fedc4225f65 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/173b84f8ebb849a399ba79dcc22f9621 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3e8e4a61c5b45e59e912f1d80d4ad53 : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc7895349eaf41d6a7364bb704ef7180 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8bad850243c3456cbaa00283f2963323 : FAILURE in 4m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fcf2d78b93694fbfb8d4081f6c7e5a73 : FAILURE in 4m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7322e6ff8d974705a7f5b9af1beaf109 : SUCCESS in 15m 11s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/00b117f99536496fa741f7e1f63946ca : SUCCESS in 4m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/878adf883f6647af8394079c729d519a : SUCCESS in 5m 24s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/af6478db20ed433b97220ce4649efbb2 : SUCCESS in 4m 30s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c4a293cdd0b441cb90f41c4bac113db0 : SUCCESS in 9m 31s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e90984fd3bdd4bac8c08d4aae5d16124 : SUCCESS in 4m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ce5bd36e54114852b01ac74ba3bd3500 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7b651f8e32974405a2880d6e54f75abd : SUCCESS in 7m 47s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/86aa6c5412714263a6d09309f2ced8cb : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d0c205891b64dbfaea66cb95a3d17d6 : SUCCESS in 7m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5cf6952ab97c4aa9a963cf83fa375bd1 : SUCCESS in 8m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/070d5b9e6faa4e6d96c13e98656adc04 : SUCCESS in 13m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a0f78d87de2f4f9ea503c0fd4eedd165 : SUCCESS in 3m 37s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f844b4d7d9f6402bb2053495620c80a7 : SUCCESS in 8m 15s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7f6b2c2dd4814514ae36a56726206c3e : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a506b9bfe0a74a5f84cb4ea77293e0ca : FAILURE in 2m 52s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/11c212a18bbd4051b257cd9afc11d3de : SUCCESS in 21m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3aa6ea3cf9de407db145a63059aa574a : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f0f9b0f959f94a1b830e8b14cece45ff : SUCCESS in 11m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8fa292a7e9f645eea8b72f77132b6d49 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba90e164512b4b6e958c0662aa98972b : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ac15be5530d40fb9439f59614b968d1 : SUCCESS in 20m 26s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/de5b238244cf4fd9abaca6628fcfd89d : SUCCESS in 3m 56s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/102985a802c34cd2a0b74fe8674d95e9 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f4ac5cfe62b44f1baf290ef37206dc9 : SUCCESS in 6m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/08a7fb86233e4c26922f18e6765dda37 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ddd0d2591af4fc8953c0343514c722c : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94dd082ec56a4c07b10272bb017bebac : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b0ff2615903d2d73941e6fe5011246a0f90aaa14","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e29e30f0a81ff5807654972c723f8b5813d7f2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:15:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ea1e9b8fc7430daa6b1af94160cf09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c832c007e842e4af40611f5e330275 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4680ced421a6492e9030c58d33d75913 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac11b2b01c794945869b454689f0382f : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15b0d15a71ef40dfbbd80e3dc1a17891 : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27847a08b76c464fb0d21d965de840b0 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2559f5c968324f18b98eccd6484d1293 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aff15ae1af6f4b418e2dffd3afc6a250 : SUCCESS in 10m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/141e5f094d8c44d7b908a4111c247b0e : SUCCESS in 14m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6b5a4effd1194da8b27bf1679127d049 : SUCCESS in 9m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1062779dc1924e5cae6d41e114d8a800 : SUCCESS in 20m 09s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9309e4d65af04266a744641a44886434 : SUCCESS in 5m 20s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1fd25510c8fc4943a1fe4756b8d8302f : SUCCESS in 6m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8e61513997a64df6a65e800e2bd458cd : SUCCESS in 4m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0491788c5e8a44f79afab5e8943da0f7 : SUCCESS in 11m 22s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d3fdd0ba8664422d8cf6c9194db2db24 : SUCCESS in 4m 53s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/aea89e00d1d648feb1f8da31301f87b3 : SUCCESS in 5m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fd8f28b657f84f179c6c1c57a786d19b : SUCCESS in 6m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5a27946ffceb43aca165f9c20617f196 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2067b61ccc314adab0c4cf0e15570dad : SUCCESS in 6m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bad97695670d4d42b78b81027d94aade : SUCCESS in 9m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/dbf31a3ffebf42fdb331b29938f70a1d : SUCCESS in 12m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/05881662da0a43a1849b845e78f5301a : SUCCESS in 6m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/29e7908ccfd946f6b93cab826daf2bd3 : SUCCESS in 9m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ad29e4061ec94677a88e1392f0c8c621 : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/5e1e9fd193724c40a2f822b22240033f : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bc71789443604d02b854e54143822d32 : SUCCESS in 21m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed3f1e92429f431db7739c141f3470c7 : SUCCESS in 6m 54s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/006cb999e35b4fa0a1a08d06dfb8746e : SUCCESS in 12m 08s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c6ac30ac55204752bdd136f8d79fdfc6 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25601cb2bba44ecc84e4e1e8ca97462a : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ed8ec502f6a4aabad299a637e4523cc : SUCCESS in 18m 48s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6c99ee20390d4757af24c6bceb28e6fb : SUCCESS in 4m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/864266bb8cdb41c2ac9d04a74a3e9df8 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/922caccee1f045ffad3b671d8a1488c8 : SUCCESS in 6m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2dcc011ff83c4034aeb4a770073b1005 : SUCCESS in 15m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/454d99f24a9e45f1bde0086f61a4d8ef : SUCCESS in 1h 00m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2508494380548bd9fbebf186745b764 : SUCCESS in 1h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ac7d74848b66e22e8ca028e48ff3e42a1a832a6c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-23 16:41:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3395fe560d468b24c1b61c05675b813d8afadfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aadb606b2e80d4e3ba73bf13502b99f859e4154b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:27:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42276a68e8f048348dde14641d8b1c3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b34d7ff1514c0582f60204e2ca1909 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2046ffd8ab2453580639184607c509e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef7f6d625d34272a77cd8d8e00803ff : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e55586e7c0424a05b41838bd25a8cace : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/189d349ed72c4ddb95783236d264d6a6 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd3803bdbc83421a9d68c96eca0ee4b5 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a4b6f712e984208af70468e29cad2db : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/63f04342e7c547b1b415dfeae567139d : SUCCESS in 17m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cf5803f47884436ca0081dd238057089 : SUCCESS in 9m 00s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7df388bf618c4e8a9e99f4ca8486b378 : SUCCESS in 13m 38s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/cc61e40d495343fbaec7d98a7109894b : SUCCESS in 5m 21s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/15fbfa8a79774b3f859288872bba894d : SUCCESS in 5m 22s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c071977d5d5646e5911114a93f061ae4 : SUCCESS in 9m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a51378d65f024535a7d2d52d011ad7f1 : SUCCESS in 13m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9add640957c74849acbc8d3ff0b96613 : SUCCESS in 9m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/92b98e43dcba4b01abd7d1e057a942a6 : SUCCESS in 8m 14s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7d2069fd3bc243b9a37b21968d7eb378 : SUCCESS in 11m 16s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e5edfe632bcf4aa0bc39deaf8ba04cc4 : SUCCESS in 7m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f615fa6c7c3040ddaa399f811651aa5e : SUCCESS in 8m 01s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cc779d1427c245508a5d0f13d8191471 : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/281880d7a3ed4717aba1cbc63d61ab16 : SUCCESS in 7m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2ef6d9c2ee9440dd833739e904ffae53 : SUCCESS in 11m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c576deb8ad7743069cbed65cad0282a6 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79daaea4b8ba4f52b336908779c30434 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51e4c4d59cf440f499d4f5846add0556 : SUCCESS in 26m 00s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/499e8c96e8e44149ba6b8bcb0dfc5c6d : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0acea48d588b4e49b53556cdfe420086 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a700861cfde4765bc5bb68da97bac6a : SUCCESS in 5m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/31342843a6474aacb116915d8b3680af : SUCCESS in 14m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b001b143aa3f450c92c4b792bb6a8391 : SUCCESS in 1h 18m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33d6b260568641e4bfb572527fc4f602 : SUCCESS in 1h 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d490ad90ecbc9798f9947fd65022876be5f9606e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:27:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c423a902ba204af1199fc096b0d42ab91bfd0d11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb3ceb9e8a84475a19f09db6ab7eb73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca76399776e34e91aff185e4fbd40815 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"614026cf861b4e85c0a0db29ff46aba70865d628","revisions":{"614026cf861b4e85c0a0db29ff46aba70865d628":{"kind":"REWORK","_number":1,"created":"2024-02-23 08:35:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/909956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/909956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/909956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/909956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/909956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/909956/1"}}},"commit":{"parents":[{"commit":"d05abcdd76e78de64d628d18ee5975ca08d5ce6f","subject":"Merge \"update constraint for oslo.policy to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d05abcdd76e78de64d628d18ee5975ca08d5ce6f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 08:35:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 08:35:45.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.10.0","message":"update constraint for mistral-lib to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If33cf5ce8b21595543cf5d478cd76179963d86a5\nmeta: release:Code-Review+1: Axel Vanzaghi \u003cavanzaghi.osf@axellink.fr\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Ia7a84da5e4b6f42df36b76ec5760fd6b51e008b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/614026cf861b4e85c0a0db29ff46aba70865d628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/614026cf861b4e85c0a0db29ff46aba70865d628"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
