)]}'
{"id":"openstack%2Frequirements~909963","triplet_id":"openstack%2Frequirements~master~Id1e0ff2ad72417c493bf28209ddc8f64fc75d5a3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id1e0ff2ad72417c493bf28209ddc8f64fc75d5a3","subject":"update constraint for ceilometermiddleware to new release 3.3.1","status":"MERGED","created":"2024-02-23 09:06:16.000000000","updated":"2024-02-23 15:03:36.000000000","submitted":"2024-02-23 15:02:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909963-new-release","meta_rev_id":"6da417640d5e7604290a5ddc2e16e4c17aaaaf11","_number":909963,"virtual_id_number":909963,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-23 15:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-23 12:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-23 12:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 10:44:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 12:27:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"969c57a9ba0df5a01b67d71cf64dcc22a59572eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-23 09:06:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca38f18e078f35627315386080267ea6ede2e0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 10:44:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1225df10b8cb4df6abc64f8687cbb468\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341a0113b01f47808c4052a4034bc1fc : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e12ddf41644493fa59384e472764201 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a4a9c592c294b13a597bb0137a6591f : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f5ec49ae95c4dcaa61cf443c6187802 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae2414b23b9c4b87b3283db512f1c041 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/122a6421d9fc4310bcaf3188e39528bb : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8428c8d47c4048bf8f48e66ff3005778 : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce4b59431b7941fea82dcddd71b34885 : SUCCESS in 13m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b882b9c9141d477c937c1163377e7c0c : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d8ae0b70c2a14ff0b397054f9e511ff9 : SUCCESS in 16m 25s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/840a10718a154492b84e866cb2896627 : SUCCESS in 4m 41s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e0ed4f7221a74c5c9c04730cd1fb3b31 : SUCCESS in 6m 09s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/927369d1051246e3bafa5860c1282c7e : SUCCESS in 4m 14s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f43f9eba0fe54610948f93a2e76f1d58 : SUCCESS in 10m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b3c55a5a301a438ab8c85bab11947a75 : SUCCESS in 4m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f51e1188ce6c452ca144847218b63d39 : SUCCESS in 5m 09s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/53d479e98b8e4de0b90a5583472f6c92 : SUCCESS in 9m 20s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/869e6e18c2eb45dd9732c1559cfa9e36 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1791cd541a14c21a2b4a708d04cb4ad : SUCCESS in 7m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0ba49a13d03a4583b7568d79d0f56bc5 : SUCCESS in 8m 48s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ecad51a27b2347b4b36c49375c6b101d : SUCCESS in 11m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/eab8dd46820b40d391e8ecd683997449 : SUCCESS in 3m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/98205f5017fd441b80a05fdf5750b441 : SUCCESS in 6m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f70ce0dcec564d3181bcdc91ca741242 : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3b5181613b494b7fb3b077a21cc127d7 : SUCCESS in 13m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d6cb3fa1df554e118a8a884204d90150 : SUCCESS in 22m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/398802a3ad3243dfa687312b760d54c7 : SUCCESS in 7m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/86fd13959e0a4f999402f3f68b358495 : SUCCESS in 11m 54s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/41ce415887dc40b7b84daf2826df78f2 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f9241c43b004744b48b987e0c01c3df : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cfb769e0f0404c43b82696b592005c26 : SUCCESS in 21m 30s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b345607725224db79ebf8a25c9d9d63c : SUCCESS in 4m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7e23eec91ea4404cb5327c531ac1bbd2 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9da8af88ed4a4ecf9fc04f8013d9aaaf : SUCCESS in 6m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d7ae5c1c3a24440fb6d585617f05c7d0 : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a090192254046e6bc098083a9bd3f65 : SUCCESS in 40m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f921b2acc78f41ea89c0bf9122eab2c6 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f32dcbc6964b405782c23454bfee47a6aa8e37f4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:27:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"911206cf3824579984770ea3a22eb31a03f31627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:28:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a68d2975edbcf582272decea4fdfddc34181efaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 15:02:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0defe9c202d4aaa80ddfcff4c37cc13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6748c2fed12f4c47b4be175e44019a14 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af433da2b4a4e45b7d911ae56731072 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16302c7b839a41d992e4150398a171d4 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbd64312628a4a35959cf5059e2caa99 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/746448360cf44e36adcde1010b50ae9e : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/813017d8dcb1449ca470b455c4098fd0 : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/725a88ceaff64f9c836b260b462dd3f8 : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38a339eccbaf46d68a5a0d9099d0c679 : SUCCESS in 16m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a3a294394cbc4957b0513a744aa34c1a : SUCCESS in 9m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e54682d6798349b48e7069ea00c02d1e : SUCCESS in 10m 53s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/41c5643b882d474d9b22ef92b8d9d454 : SUCCESS in 5m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/65c077a3d7a84f818fb10c4738bb3807 : SUCCESS in 5m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/950b0838a2774661a70e63407b81d71c : SUCCESS in 9m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c57cb7c68d47428b8fe9a4863443861d : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e06a7bf9b8ce4bbab14c984fee418be3 : SUCCESS in 8m 34s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b521e515162c4fd09dfc33f56304f697 : SUCCESS in 8m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cd2962f812b34fb48b2c0e0367cd6573 : SUCCESS in 12m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/33e2d038e9d54537835be7877611457e : SUCCESS in 3m 40s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5d312b230cf8464489b252e089f3517c : SUCCESS in 8m 11s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e20923748cfc4b17b8bad020f21a28c3 : SUCCESS in 28m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/275bf5fd7317429aa104e5696c6a6a9a : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/57ac813c10f447c8a3ca8fbdb1d01fc9 : SUCCESS in 11m 30s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6204d018208b4626a274d4b5551406a9 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/654bd956fafc44adabb776d266a4d1d2 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7f14711105a4a17996538cf712497f0 : SUCCESS in 22m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f97849a57b364a8ca38031da093064b1 : SUCCESS in 4m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c127557070b74ac9aac461bb74ae8cd7 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce1d0c4caad44efa9a51245b3243e9d7 : SUCCESS in 6m 19s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/80d5a4edd9734576b7fefdd22a559400 : SUCCESS in 15m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae1dd9d6f2a9460f8c5b60399e200e74 : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4c0ef2877d74d9fbbe26c06d220e073 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3569da5077f169b553fff206cc66f19d07faf081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 15:02:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6da417640d5e7604290a5ddc2e16e4c17aaaaf11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 15:03:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a05635b92f54b8593a5cbbbb616b64b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2546ad55574682ad281441d6426c77 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"524104b092c9843b021428727361218587928849","revisions":{"524104b092c9843b021428727361218587928849":{"kind":"REWORK","_number":1,"created":"2024-02-23 09:06:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/909963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/909963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/909963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/909963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/909963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/909963/1"}}},"commit":{"parents":[{"commit":"d05abcdd76e78de64d628d18ee5975ca08d5ce6f","subject":"Merge \"update constraint for oslo.policy to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d05abcdd76e78de64d628d18ee5975ca08d5ce6f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 09:06:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 09:06:15.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 3.3.1","message":"update constraint for ceilometermiddleware to new release 3.3.1\n\nmeta: version: 3.3.1\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I11c3d3dcbcf5ea1aff215937a1b2e68aecd1f081\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Id1e0ff2ad72417c493bf28209ddc8f64fc75d5a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/524104b092c9843b021428727361218587928849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/524104b092c9843b021428727361218587928849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
