)]}'
{"id":"openstack%2Frequirements~909966","triplet_id":"openstack%2Frequirements~master~Id9540ae53a98a9a782ce753c92e31b6ad6e09534","project":"openstack/requirements","branch":"master","topic":"more-experimental","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-24 04:56:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9540ae53a98a9a782ce753c92e31b6ad6e09534","subject":"Add more experimental jobs","status":"MERGED","created":"2024-02-23 09:37:12.000000000","updated":"2024-02-24 04:58:13.000000000","submitted":"2024-02-24 04:56:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909966-more-experimental","meta_rev_id":"045396ef7d8299e03fa74e62ab2e3821cd68fea5","_number":909966,"virtual_id_number":909966,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 04:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-23 18:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-02-24 03:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-23 17:43:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-02-24 03:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2024-02-23 12:15:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 09:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-23 09:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 17:43:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-23 18:50:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-02-24 03:46:42.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"0a48a0215cd6166493d262d3bf22aad96508e8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 09:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2426ac0586ff03cd93799d64fdfd4613047e1bc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:37:34.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  Job nova-ceph-multinode not defined\n\nThe problem appears in the the \"opendev.org/openstack/requirements\" project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n  ...\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"e10efcb375dfba1fce8df5f3141a6deb5da042e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:37:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  Job nova-ceph-multinode not defined\n\nThe problem appears in the the \"opendev.org/openstack/requirements\" project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n  ...\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"0a3050ded0e546cb72d55c9706899021d87e3eb3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 09:42:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ec95b150499a55790d0e0a729738c5e19928b8f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 09:42:52.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  Job nova-ceph-multinode not defined\n\nThe problem appears in the the \"opendev.org/openstack/requirements\" project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n  ...\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"74e80f6cfe79d06fb0bbe3ccfd3e620ba4bcce62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 09:44:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"42d00bc13ee82ea08ef718fff6117a87d6e8a2ee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 09:44:48.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9448f8da086d18c4ea26d57e3dff937694c7949c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 09:45:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"103e08f1eeaf2d3da561a4250a04faf1636ee306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 11:18:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84a5388ad4e54e69a8c45949cafb9362\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f19d2d4b0f4a12b7b361a46d90a201 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c268d20a66154c2eae0659996d453812 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb6a0cffe05e460495fcca293f3a347f : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fd0395f22f14d91958c3f291ab6a4ed : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a75aa62f5284db7ba4cb5082c85c76b : SUCCESS in 4m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ffcfc9f7801449bbe58faa1783eed32 : SUCCESS in 55m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21d78dd4282b4f36962d99a27c458fad : FAILURE in 1h 31m 48s","accounts_in_message":[],"_revision_number":2},{"id":"024ffc26e7770e7d6d42d0085e440d231b044648","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 11:49:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1020d7a6a4ba476ca923321d29c4f124\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c23e5fa8a704e74bd7fc989fb7bd9d2 : SUCCESS in 1h 51m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/30a47996cf664df98da3f3a60439d962 : SUCCESS in 1h 55m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c8778f645a74a2595a0d18b02879f5b : SUCCESS in 1h 06m 28s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/cc1d91f3356741e48c5d7d659b8b3cc7 : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9d1013ff6ca8eac3602c0de8591c76ee019e1bbc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-23 12:15:23.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea362eaede7c5f031a50fffbe12269a6ebe85966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:54:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06422181bb47451a8d5f418386d76904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f300bbae2fa34bc99218dd9d6a810d9a : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b517153db5843fbb416173601df324a : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea57027eeac46bd937253eca16d23d7 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dac67a52bd94110b234dd2363258800 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43320f51daaa4dd596e37a29eae5667b : SUCCESS in 3m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/546b1c121e03403abe341797c6b6dc89 : SUCCESS in 1h 03m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af405864274844d3b8fe94ec1286b2f9 : SUCCESS in 1h 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"fed8b942a16829cbb6da2f3754fff8e445a3e31c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-23 17:43:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a74f73c232cdabbb204b5cd50ace0ed58de2c19d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-23 18:50:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f9a9d36ff93a2a27411714d04f2c3039154ba600","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 03:46:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44c7364dd6216fe236d598a8ac6ffd6d21262965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 03:47:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a351c371a1d3ee5b1cdce1617ec760a86a1fac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:56:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a67c7b45eb4023a14a581ec93a843a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce2595713ab4423b161672d7f8dd80b : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d7cf21e8d448659749698662b8aa5e : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a678df1f5e547de9844ae33fab96a1c : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed88bdd2e7d644e0af7239cc3ad9e4e7 : SUCCESS in 2m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f34149a25d964e5ab70d5c945b4da307 : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8667e753bce947caa6f10722246ef204 : SUCCESS in 43m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3845dfb099724d82a71c375394d448eb : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a34761b59b2ab769f8a51f4982d97f78f3616f75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:56:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"045396ef7d8299e03fa74e62ab2e3821cd68fea5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:58:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9862ccab6b0745d58cbc5db4e66279a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6747c6fddc424aaf8c32a3567a6ecd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f910188f4977a5a17996ca9db7d2ebc83cefbdd0","revisions":{"76f19d367cb6061e1686dbecdf2a4345cca8364e":{"kind":"REWORK","_number":1,"created":"2024-02-23 09:37:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/66/909966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/909966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/909966/1"}}},"commit":{"parents":[{"commit":"d05abcdd76e78de64d628d18ee5975ca08d5ce6f","subject":"Merge \"update constraint for oslo.policy to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d05abcdd76e78de64d628d18ee5975ca08d5ce6f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-23 09:35:35.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-23 09:35:35.000000000","tz":60},"subject":"Add more experimental jobs","message":"Add more experimental jobs\n\nIn order to allow better test coverage for version updates that are\nknown to be critical (like dnspython or eventlet), add some more jobs to\nthe experimental list:\n\n- nova-ceph-multinode\n- tempest-integrated-storage\n\nChange-Id: Id9540ae53a98a9a782ce753c92e31b6ad6e09534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76f19d367cb6061e1686dbecdf2a4345cca8364e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76f19d367cb6061e1686dbecdf2a4345cca8364e"}]},"branch":"refs/heads/master"},"f910188f4977a5a17996ca9db7d2ebc83cefbdd0":{"kind":"REWORK","_number":2,"created":"2024-02-23 09:44:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/66/909966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/909966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/909966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/909966/2"}}},"commit":{"parents":[{"commit":"d05abcdd76e78de64d628d18ee5975ca08d5ce6f","subject":"Merge \"update constraint for oslo.policy to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d05abcdd76e78de64d628d18ee5975ca08d5ce6f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-23 09:35:35.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-23 09:44:05.000000000","tz":60},"subject":"Add more experimental jobs","message":"Add more experimental jobs\n\nIn order to allow better test coverage for version updates that are\nknown to be critical (like dnspython or eventlet), add some more jobs to\nthe experimental list:\n\n- nova-ceph-multinode\n- nova-next\n- tempest-integrated-storage\n\nChange-Id: Id9540ae53a98a9a782ce753c92e31b6ad6e09534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f910188f4977a5a17996ca9db7d2ebc83cefbdd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f910188f4977a5a17996ca9db7d2ebc83cefbdd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
