)]}'
{"id":"openstack%2Frequirements~909996","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I0c0d191604ccc693d6cca9c557aa4fc681dd77b8","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I0c0d191604ccc693d6cca9c557aa4fc681dd77b8","subject":"update constraint for os-brick to new release 6.2.3","status":"MERGED","created":"2024-02-23 13:49:00.000000000","updated":"2024-02-23 18:38:01.000000000","submitted":"2024-02-23 18:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909996-new-release","meta_rev_id":"fecc905fd9b0eb0bda52a85d8cca8aceb3c0016d","_number":909996,"virtual_id_number":909996,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-23 18:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-23 16:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-23 16:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 14:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-23 15:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 16:41:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f3d12ee6c945cb2b128c8f4ab495bf8e4cb5cc6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-23 13:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d24199bd951d6cca3bae6af0c271e7c922428b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:00:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4770b9f224584b60b9d285363cd1b001\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de5e37baa4e44bcf9d31e231bdae3810 : FAILURE in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5a4a9b0a0dd70da46d4ef7630d3f41ed9e1662b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 15:18:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e7bbf5691e4b10bd81af0a31eb359c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee56570d0b6a4ce4b943a5f83d6dd7ef : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5897d343d1d4cb483d7cdb3ec1fe732 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c93ea435c124688bf4f564f17812273 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20ea4a81bc8a4e24938e8a8b56075c85 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/916a1e08c8ac4ea6b141482a5eb8c0a5 : SUCCESS in 12m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0ae336403d9b4bd9ba4399333399bf47 : SUCCESS in 13m 25s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7bb9a5da57ab45ec909a6f717c200b70 : SUCCESS in 7m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3f8aa0dd43d54f2c9363921457937328 : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0bd694ad85ae4598ab8c8e23a145c899 : SUCCESS in 4m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9e0d4a6095ce498483abf34c5eb4a25a : SUCCESS in 11m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c8f180a8e6ce4f85bb15c8f7ab1dfd8b : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/65713d0b83a7408aa5dac6f3e9666ebd : SUCCESS in 4m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/078c6cd902e142c7a130bdded2c34784 : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/28c0988180e444b280b3873245d3d2fd : SUCCESS in 8m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70517e2658c244689fc9655cbfd93a5d : SUCCESS in 7m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4739f3173730468a94df9898e47ef727 : SUCCESS in 9m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/625942b7b8054e0e9be0c9bed590d298 : SUCCESS in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/014fa004cb6749bf886b8543150cdc09 : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7b49ddb041de4f4eae0751597a40720a : SUCCESS in 11m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/192c55deaf4746ad9e6fccbe2e22b095 : SUCCESS in 4m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2d3389263a0d472c91d602b8c6e417fe : SUCCESS in 10m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0621e6542a13495f85c557ca138dfb9a : SUCCESS in 19m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4b415fcf88494ab3b26b676080f0eac2 : SUCCESS in 12m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/79649fb98fbc42f8b9fee4a68a988f52 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca46213105254d4ab4e7a96ef2234839 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7bd03f46c5b14802926b53092faf86b0 : SUCCESS in 24m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dacd3b0a3ebe4437ba0ca82e06e4c226 : SUCCESS in 7m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0cd7efb247bc41239d3f19dc6e5785de : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54726596880842fdbb90773eb08cbad3 : SUCCESS in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2175857288ae455c8726f0cfe1180686 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76be9f616ff34abb84aa3873e7e90df1 : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b7063fae09b4e28be824c94e1fd90c5 : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ce44f42e749e161c11caa63e5e55fe408e810075","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-23 16:41:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8222f5ac08313f0f1ebff7880770e42d22a5023a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:42:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"937063529e730f82354204f34e9168d12734726d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:38:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b30c053447d48b1862813ed94022d67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bc82cd875b4725b69d229ecd8d40b9 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07c67a27756643578c296c6ecb9fa56a : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50901ab55c1d4b9ea55f929a9b7e6acf : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a78c2655e0c64c5bbb92b7804b28f681 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/55b599d1c9bc47768994ea9fb5312ac0 : SUCCESS in 11m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/34a8d388b2554cf1a6c2844fdfa1d269 : SUCCESS in 11m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3357b4514e364cb7b087043beece4176 : SUCCESS in 10m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2bfde9af0149411995875fc0d68a0312 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f2acc4f3a41745e48f023613e060e050 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fd171d95a0b4405eaee75cc65d7363a0 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b400190d310a405ab5e612748372e149 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d2badd387c742d6a99cd32a7ab9c50f : SUCCESS in 10m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6904bd050a7647b19630a0c346b7f545 : SUCCESS in 10m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bdeb5524559e4d4890c7b46ae5ad9c54 : SUCCESS in 14m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9ea28b91b86040e7af4a0c28bfc0000d : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/61699ec700ca4a8cb12eb6bf9d398c4b : SUCCESS in 9m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f3dc3342c1674960a139760d1ba1aed6 : SUCCESS in 6m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c7ec3e0b518c4cd592653dc5effb7f18 : SUCCESS in 18m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/79beb7e5551a41b7b8002afb7037cf01 : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d4fd4bf0e60943aeb1ca3253860b9c81 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/844712b6fa344ae59fb4b8ee42b64ef5 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f63d93bdae8747f1b7616a49b46c1d5c : SUCCESS in 25m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8004748a1a054b9e80ec8508d4565d19 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1fabfbe216104e87b08d7e612df78835 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a116e4df27044154ad56ece6e7a0fc83 : SUCCESS in 5m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a5fdb8fa8d04417db49cccdb77ca77f8 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d3c6c20bc114efa8d131d6f4f08cc87 : SUCCESS in 1h 05m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e84c133206242ffa514d3ee293d68be : SUCCESS in 1h 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fecc905fd9b0eb0bda52a85d8cca8aceb3c0016d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:38:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5f1b8040b8f9c3c086815350049706454645f20","revisions":{"c5f1b8040b8f9c3c086815350049706454645f20":{"kind":"REWORK","_number":1,"created":"2024-02-23 13:49:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/909996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/909996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/909996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/909996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/909996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/909996/1"}}},"commit":{"parents":[{"commit":"5191edd7a65378a214f678feadb3a3347e4e3d85","subject":"Merge \"update constraint for glance_store to new release 4.3.2\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5191edd7a65378a214f678feadb3a3347e4e3d85"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 13:48:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-23 13:48:59.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.2.3","message":"update constraint for os-brick to new release 6.2.3\n\nmeta: version: 6.2.3\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I4a63b56f7f77d95d33fa9c2bb5ffc24c378ac10d\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0c0d191604ccc693d6cca9c557aa4fc681dd77b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5f1b8040b8f9c3c086815350049706454645f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5f1b8040b8f9c3c086815350049706454645f20"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
