)]}'
{"id":"openstack%2Frequirements~910040","triplet_id":"openstack%2Frequirements~master~I72a777f3a553ba720f80fd66ad7a488b55de51f4","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-08 22:57:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72a777f3a553ba720f80fd66ad7a488b55de51f4","subject":"Updated from generate-constraints","status":"MERGED","created":"2024-02-24 08:12:14.000000000","updated":"2024-07-08 22:58:05.000000000","submitted":"2024-07-08 22:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910040","meta_rev_id":"6b3dc3ceb3d4bd73dd73ba3ac70fcc8f4adedad3","_number":910040,"virtual_id_number":910040,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-08 22:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-07-08 16:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-07-08 16:52:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-24 08:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 07:28:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-08 16:52:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e85b1207625c5565090b57232f6719afb2d84a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-24 08:12:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb192def08872ebd9df160c3b7aa01de0ee1c474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 08:50:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d59bc3c3ec524b9c9119f967df662153\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db54a0c8d0e544feb6662d760ac3d234 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b0908e6f5445d59aa1644f0096c3ce : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1673174b36bf4bc58fe6f4ee00e11400 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf77f00c6a834391850656d4820b43b0 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73be3baaedf148bb803e059da6a1d668 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3602c6b2ecc64c0c84a6816d7925a6ce : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1471cf4321fa4a2a8aea778a1aa6452c : FAILURE in 4m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc41eb05d4284e56bf500392b6ebbf0c : FAILURE in 6m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6f249b9ad9b6453082de63dd417bb134 : SUCCESS in 9m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8f999c87a9c04f5b9c99b8eba7ebc9e7 : SUCCESS in 16m 28s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9fd06b1efa2b474981b031a3f2b07f3e : FAILURE in 6m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2359ed78e36043e88420acebccef36fa : SUCCESS in 4m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/547ca33f95cb4a50a3dda21f9065d9d4 : SUCCESS in 5m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1a156dbac53d4b5588a13dea77df68e3 : FAILURE in 13m 00s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b8c4457f6770464eb5f929dee72db658 : SUCCESS in 5m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/db5e4a30ed6543748f428e9b1d37fab9 : SUCCESS in 4m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6b6c6e911d564366867231367177e189 : FAILURE in 5m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/22b4410eaad3446a9a451239e239e615 : FAILURE in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7718f520ca3641209d354305f086ab0a : SUCCESS in 6m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2c85ea736c4345159be73c2e12668ccd : FAILURE in 6m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/94041d297f9a443ab7a7c62744ef1aef : SUCCESS in 10m 16s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2fdbcf2a1bfb4c12994933bd23b3a778 : FAILURE in 3m 25s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/216011e5a48f45c58a85a1498dd48bc1 : FAILURE in 4m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0ba1d80546654b3eaf896e5764899426 : FAILURE in 3m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1807ea25fbc846d0ab3f90815f383344 : SUCCESS in 12m 19s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/28ac3e71a79547ba992315671f28096d : FAILURE in 16m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1158516047724206865ccb5c3e5af80e : SUCCESS in 5m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5e428efaa63a4b98bfd807b8393c814f : FAILURE in 8m 14s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e19eff5edaa246c8968916ede557c83e : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bea331aa26a544cdb4fd49e9a8fdac1a : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6364a9f974b407095e96cd0091b577e : FAILURE in 17m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8ff04f2f60024a149c90fd7c37909e2e : SUCCESS in 7m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d7625bfb2d0f476981c5742f9c93f4e3 : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5502030d1c0d45f09357690f358c066e : SUCCESS in 8m 20s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5998ad1181b64a48b928c7cbb977975e : SUCCESS in 8m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78a842e8d84f4b59a1a9178004ad70a5 : POST_FAILURE in 17m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80713f2b83ee41b9a97419d42101af71 : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1bd0f30a2912f7ee2e043a4af6e1a8c17724607b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 07:28:51.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c91c2e4dc497bfeaa63f5db4268ec7103a46ee70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-02 08:17:00.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8b54ceeed89844914c3e73a8e8d0f2dfa281b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 09:02:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/847c6e0c784a4f2abefe2dce8eef2390\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2adad169dced45a8857d0a0559ffa7ae : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466aec80fa4542f893f19990e1e5d449 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a11cd81ba64fb2b92d311475f82d98 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dcbffe889624573ad556dd84bda6f29 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e70e88d32114d4a8bea76a2be477950 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75256912b927498386fcb7faf4d2b5b3 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a42a3be91574ac8a0266229f2bb771d : FAILURE in 2m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f5f0f11a5983426f800e438654d47616 : FAILURE in 4m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0e6a21ab826542b0b0e2a7fb5c2de169 : SUCCESS in 10m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b6737b3d834746cdbfa707f76c1fea17 : SUCCESS in 16m 52s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a59684a4ab99458eb33a18b094097f0a : FAILURE in 5m 27s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3762eee8b8b44473b74637adc3fc20b7 : SUCCESS in 4m 26s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bc66078023de4e6bb7beb1f51ad5c7ca : SUCCESS in 3m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b07aab5f9b804f72aabef638e7770bac : FAILURE in 9m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/07a0082cf6494f18b3b4b626cc41da2d : SUCCESS in 4m 24s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3f532a00d3974d1c80a900e5f4cee231 : FAILURE in 4m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6ddc03df6b2e42fda7b7233ca9a73552 : FAILURE in 6m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/959974c291cd45cfb0f5efc93274f01c : FAILURE in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a526f425f3cc46afa08f75ab77124839 : SUCCESS in 7m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ea05de99fbbe4d188b4ee53645233c89 : FAILURE in 6m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a7c48c77f69249e8b2d489a3eb39d0bc : SUCCESS in 9m 22s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/68321bae407a486b8756610cf566cd5b : SUCCESS in 3m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ed04f883ff624379a2e21c735b851107 : FAILURE in 6m 51s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bce8f554e8e1475ea727560a54b8c70f : FAILURE in 4m 02s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9a9e58cc3207495281a08363c3497c55 : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/59b09f56a5d045f295bd5fe05910cf60 : FAILURE in 18m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d21521003e8b4fa29d43914ff14cd88a : SUCCESS in 7m 35s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/841508e800ec480e9a5c7baf1cae2e8a : FAILURE in 9m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6d9f4861161147ccbd730b00a64d9ff8 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f0b8f05648d243898340a6456c976ef2 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/579992a94c114ce08e6beeaaef04dcd5 : FAILURE in 18m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3d6a20094b4d496db59703c2f781dd07 : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/bed4b113bcc04cdea423e0bbb61740f6 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37bb9b089c804ef1be88cde1ca88235b : SUCCESS in 5m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/17f483802850414fa75d99fec784cb2d : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bf03d7b50fb4ffaa51d4bb36df67f97 : POST_FAILURE in 28m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdbce9ce635f46c9aab54267220ff167 : FAILURE in 15m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ebd11a429aaef304b0933c6219449a2d0dc78db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-09 08:18:58.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5141036d30e4c322b8cc87b0e8143f8f0667084e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 08:56:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d90f64029d749ba9c23024ea6f1813b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b00fdd68ef44424a48afcf578627131 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48430c6b387d4ba2b1fdf81d5e78b2e4 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d206fe9b7a441f5a4050394d2cdb294 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce5658d64f5846cba6cc12ee3ef3a0ab : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6446d89a63664f45ab0236a502377732 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d3cd0b67a644b36858567213d4c3c77 : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ba97b51d30044d08affa0f9d7c04578 : FAILURE in 2m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f00b5810e0c24fbeac566b3445f42ac0 : FAILURE in 6m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9b159fb1ed774e30bdcb61484ea0541c : SUCCESS in 12m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/63639e3d793f4ff7b9331a7078af03c0 : SUCCESS in 16m 08s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7dfca6def9ed4dc7974f40032771dacf : FAILURE in 5m 08s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/06e0aaafca2e43cab726534bddc689bb : SUCCESS in 4m 14s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2aa14b88ae1c4ff58554ef754f364584 : SUCCESS in 3m 45s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e708d44174964e1b835698fe8a2d3ccf : FAILURE in 11m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e48c30056427492da9fce0101905e11d : SUCCESS in 4m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f635580aec28479c84f88d7bf4cc1395 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e1bd8444dc7944818d87d59811a4f62a : FAILURE in 6m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2132d91b0c0442478060db64984f4ee5 : FAILURE in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7704a73442214b47972e6caed8a8ec7a : SUCCESS in 6m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0179ecd320fa45029085826d2e1537ec : SUCCESS in 7m 58s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cdf055c8e9ff4316a1d17b080d8944b5 : SUCCESS in 10m 58s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d7cba981a081402e821c2417cad1375d : SUCCESS in 3m 57s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8b7e616866044e008a70d0278e82073c : FAILURE in 7m 23s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/37c8e0f6e2dc46f2a8e8c02372602e4b : SUCCESS in 4m 59s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e51053007f2a4a30a8196c0006431b44 : FAILURE in 4m 33s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1d32b6a0c34e42e89ca6bca915118dcf : SUCCESS in 10m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e7f9c25bb9d54ce08c6f95796a035e9e : FAILURE in 23m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b9a09ff282f14f98b25282c57dab07e3 : SUCCESS in 5m 14s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ff288fe1390b4414a13aabe73c34be36 : FAILURE in 9m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/69dd65a01e204f5f9ab3e873e228a05e : SUCCESS in 3m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eeb15762e3ee425599bb00b487fb6675 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63e6fecae5f34a5592349291889f5e7f : FAILURE in 19m 42s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1e447da9821f428ab54a1eb426bdb4b7 : SUCCESS in 3m 43s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0b90fa7267934cff8e4529a15e774673 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b02047f63c045cf85814d7aa2f5374a : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/15cae6d130434d30882fabfe49506db9 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aea2e9f6990648e8a02095638d133914 : POST_FAILURE in 28m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7867cf26d40045bf99b72eb82c5bf5f7 : FAILURE in 17m 50s","accounts_in_message":[],"_revision_number":3},{"id":"724554649b6f6ae7ea294f881eacf545c8f92736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-16 08:22:38.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"54644332971bd77acebb8eed2e87d6e8ea8f16ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 09:14:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6812b4cb9d04490a82c499726a2ddeab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb37024238a4ed19f746bcb6d4fbd0b : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346a420bc73b45b1b95368f7de138c4a : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ce5725ef99429893b9bb726aa513b4 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a17b6a62215e457cb791c05aa71b118f : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15a482ba21664dfaa5fc4bd1e18eb6b1 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b7f786c4f634c7fafdfab675dfd75f1 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b275e041962a4735b3fc802480162ee6 : FAILURE in 2m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/292267c813964d9caab45b64ab0530b1 : FAILURE in 5m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0afc53a3f267476d85f8d7793f1aae72 : SUCCESS in 11m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbec362a5c8541bfbb11afc6df0547ba : SUCCESS in 17m 23s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/57271dbe95994b5884de2047a77fb761 : SUCCESS in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/4f9aae0a90d2473881944911d06599aa : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1e659e788746418f91ea948b0c5475bd : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/281aff03b2e44d7095c50211152eeb41 : FAILURE in 10m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/89fc231d465d47a4a1f648e0975cd573 : SUCCESS in 6m 44s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4221020d5a3648698bfadb2ce8dd7deb : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/26cebcb63e314a8abe14bdcd2bd80f7e : FAILURE in 7m 25s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/129bee77361e4abb8ab0a0b02d63e9a3 : FAILURE in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09420950983a4c1193f3c53ed62c2b8b : SUCCESS in 9m 07s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c6dc11af302e464db60cfe7b1feee615 : SUCCESS in 8m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/361171e79783403f937b4e1c2f9ea070 : SUCCESS in 11m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/dbd23c00bd514984b38d337a41dbe566 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a7c83d7a3b7245cdad8f73f61b7a1b2f : FAILURE in 9m 03s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d8e99b0ff0ab4dad83a864b66c1a34a6 : SUCCESS in 6m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/712d148e39a64c35a589cd5615fd4125 : FAILURE in 6m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b47346704d634b2987c4305dd2ddebc8 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/87a2aa4e2e464a16829871d51b965850 : FAILURE in 24m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f9196a95fa14000ae5a14c4c9c2d51f : SUCCESS in 6m 27s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/95475c23fcf140dbbfc61cbad13db8d3 : FAILURE in 10m 10s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e2ca5ebf75bf43808af7ef88bad87e15 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c39f23caee2345af9bc15480177b8d9f : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cef90956237a4a1db487cf80ce6658a8 : FAILURE in 21m 29s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a58146b6e2dd40c788c302982c891ddc : SUCCESS in 4m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/aecea5c28bef4ddaad0ae91932109238 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e9c2ccaa8e34beaa18cbe9eb84b37d5 : SUCCESS in 5m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3be7b23773e1436896279eceaa4d230d : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68306167b3ff430f926c526e021c614e : POST_FAILURE in 27m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f702ec18dc241e8bdc687ffa6deb309 : FAILURE in 28m 52s","accounts_in_message":[],"_revision_number":4},{"id":"301dfdbbd45f0e9df37d6c1fa235204b277f21df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-23 08:11:35.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"543f79cce93d34c2cccac01affb79fab02f88543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 08:44:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57d4004385a34385aa59e6c7a9ff52f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d727a2db61b14a92aa099ca07d1b0dac : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac751f8c9e144309962517e94a185a03 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d5df4a81b846abbcc5f7130802cd75 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a3d4bcf3f99457da04897e76ecf280a : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01657623c70f40d6b747d02360fc55ef : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/001a19661c8244f8a6380a8072e1b9a4 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b00661ed2ce4745b939c31980904e43 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8cc6ae6df2d543acb82b2649515a2130 : FAILURE in 2m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c9293112ced44a1ae9c3e00980d66c5 : FAILURE in 6m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/68b5f9ba350248d19d0b261497c2e1d8 : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/41593bdae7ce497eab3556246781e5b2 : SUCCESS in 14m 06s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8afa4ef760e841628661685a00578512 : SUCCESS in 4m 56s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/53d96e0e034f4e548a11067b5b3f91ce : SUCCESS in 4m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c2272ed03bec44b39383ee1988cd19a7 : SUCCESS in 3m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4e3f0fc09be34de688bf2d9e3b2523de : FAILURE in 10m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/987ea44bb395489c944b3e9ef49b4dc4 : SUCCESS in 4m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b2630603f3ca4e67868c106f8a8e2209 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/220ecdcb28cd4562bf4435f8717004c8 : FAILURE in 7m 25s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7404a7a64f054c25b585b7c342bda259 : FAILURE in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ba601431d0c4093a2ed88a54b334257 : SUCCESS in 7m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/52063fe1935d4401ac2896611f2de4cb : SUCCESS in 8m 56s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4836a91085964da6871db297d21dc3d9 : SUCCESS in 10m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/39cc0fd694674500a1ebba50b4fa985d : SUCCESS in 4m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3f30600d608f431eb761090d8e8361af : FAILURE in 8m 33s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b53cfb3723c04e8e82f267e9fd69ab01 : SUCCESS in 4m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/52bc74dc2b9e4995ad07808665ee5269 : FAILURE in 3m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b8d7d815e74e4f5d91c0d99306e3e569 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2bb75bf4535e4e5086bf5ad645defeef : FAILURE in 26m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/714e513dd6ea4dafb061a7bb8c28d44d : SUCCESS in 5m 16s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7ff550fa06fd45b9baa1c07450795e62 : FAILURE in 11m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/eae2ad09d3ac4bea9dddf8714dd41a5a : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3464a68acbe14194a68bb3be5e43a860 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/887587de1dae435480d47507b8824cf6 : FAILURE in 21m 37s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/abca795674ae49518697e38029eb8f6e : SUCCESS in 4m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6da2fa6efb4e456e9c2f4e8a91326814 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1caabbb5794e43f6948a367a12aa64a8 : SUCCESS in 6m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/73b1f219ab7d464fb824902c9abdb186 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5415df6ef083424e9fb4c64814abe0ad : POST_FAILURE in 23m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2359af234a34dcbad399f6c001c13fb : FAILURE in 21m 10s","accounts_in_message":[],"_revision_number":5},{"id":"ffd7dfdb3f0a561393dc1429510a33b6003aac35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-30 08:28:57.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6040d8881385e45135e6fc29356fb72d0ce523eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 09:02:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5beda88c63314fe0a71a8c05dbb1e264\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fde895bd204c978e2cfe4e4d8d502c : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5778e6c593ee4cc2b63912da40c3b785 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a41effcf62d4614b5b999c16582d49b : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75f7a22c1be645c99848e228fe3e45a7 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92eb86820a9b416cad4bf3c608b5eec1 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15ebdc4243f3403ebc0551dda67c112e : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/163df0f497ac49b6838e7981026f9eb6 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/52971ddaa5f2405c881a3892e821cbf0 : FAILURE in 2m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/19e7921a29ce4490ab3e73c273b0593e : FAILURE in 6m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a475795c5e5400d98e9884996ce3d6f : SUCCESS in 11m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b3e491ebd5ab4883bef6c8739b5a733b : SUCCESS in 18m 24s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/cc6b723c2dad42ed81d7a1a4f5905154 : SUCCESS in 5m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b8867b2c94b246e59b4b6b86f8818cc7 : SUCCESS in 6m 57s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/658bd6e01825482993c93d77ab406964 : SUCCESS in 4m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a784358b430c447180b2071fc2357dc4 : FAILURE in 10m 38s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5401e4ce6e314b4a8c2cdcf3f0ed2b4b : SUCCESS in 5m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bb19b88e7e294832a9fa554e2a3ffd75 : SUCCESS in 5m 21s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c4afee956a7248d79fb53f59b4f18e2f : FAILURE in 7m 39s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0c01520de9cf46dcbad5db39f35a3a7d : FAILURE in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad41e50eb87e4e99aee183399cc03c0b : SUCCESS in 7m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/907c5d66d117402786d0b36ea90e2e8f : SUCCESS in 8m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/aa888fdb0e1c46ae8943fe954cd640bf : SUCCESS in 9m 15s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7af375e3264449cba084dc5c384c4043 : SUCCESS in 4m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2e7d09d5bfd04d31933c7dc94648caf2 : FAILURE in 7m 09s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/5f83ca92a3dd482dad8cbc095af027a0 : SUCCESS in 4m 48s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4373f2a1a9d3482cb0f46c4183ed211a : FAILURE in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a5ee17fec4ae4df78888e74195f88e78 : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d1ae73b2215e4cc59ab5f03d0f75a78c : FAILURE in 17m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9310032e790b4212bc1ffc49b211ab90 : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1ec91a520fc740adb582d74bf910805e : FAILURE in 10m 07s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/777ba7ac5d0e42e6b959318e9bad13b6 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86432a6b673a4179b7f216f017115332 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e64780571f64029abd7bda5ecf20a43 : FAILURE in 21m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/9c2eccce10a24a22aae5692b10492b62 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/00de068369b44adfbd329ece15de714c : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db2ddc8948b7404bab7a870e793d99ab : SUCCESS in 5m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8141821f07fe45adb72ba9ae7c4ba8e2 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/806925bdd34642ef8138ae51fcbbaf60 : POST_FAILURE in 23m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2cf627e7ba74c869ae8ee92a4284dc4 : FAILURE in 21m 58s","accounts_in_message":[],"_revision_number":6},{"id":"0bf7059fea21818e31f4092a37e7040490b270f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 08:55:52.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ae02ca6b0f91ee38359358be594b2a1defedf008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 09:43:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71528444ebb44598947c14dc1cbb1c64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e16ada752740e3a16421268127211d : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a40697f686948189b1d163674309c66 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e031d0efd84e378d514ecca01b835f : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fc07f24b240467cb52caeebc57aa348 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d426d316cfb42fd997f0c96d5054e97 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efeda3b105574ff382918865071e6fc5 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c12219e361294b6291a56c4456df4d58 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bfe1a92d7d424fbeae4a0fa8775c2caa : FAILURE in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ac7e93906b4940ff993391d8cd501e15 : FAILURE in 7m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70e24b0c00db4631872aa946c8b5c181 : SUCCESS in 11m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e91c4e8f3a5d4fd3800a75e0001972ee : SUCCESS in 16m 45s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/34527fa01d4e4c19bfbd2553d6b3c1d0 : SUCCESS in 4m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e7497f0f17064c798eb7eaf83758e0de : SUCCESS in 5m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ccb47cd2930340cc872ac84d1612dc86 : SUCCESS in 4m 13s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8c6bdf9409354e749ac57662aa83e039 : FAILURE in 9m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/975df664c63e481ea17d448925bdff25 : SUCCESS in 5m 54s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cfbe47bdc5a84cc68eb5e6f8d84837b5 : SUCCESS in 4m 17s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c816d30f9b8b42e19c619bcecaa34903 : FAILURE in 7m 14s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/eef4a3faff204133b63e81d8a9e72cd4 : FAILURE in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66f6eee6febe4a67ac640ec0965aa90f : SUCCESS in 14m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c3a5df94a77d41ec8c5aa6b0e6302717 : SUCCESS in 7m 44s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1a8c2ad3fba542428a1215f6102fbef2 : SUCCESS in 10m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ce6616f1df524c24b0d4ce19658ad97b : SUCCESS in 3m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a3cd8e05cb5c4c55a7cf708cb650e6ce : FAILURE in 7m 35s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a58469fbbeb14b33b91ce87ca69595e0 : SUCCESS in 3m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f02f9ba956e443328f23e86767770930 : FAILURE in 4m 42s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/76bbcf6e564b4dc39156bb376db69645 : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/857041348cb14a4e8de0d81b2b291bd6 : FAILURE in 25m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8ccde954a2644a34b42bf9db4b342992 : SUCCESS in 6m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8ff39146f268422989986a3d748b0f20 : FAILURE in 11m 12s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f6f05cd5c7674d9ea1a77dd1b19b5c32 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a938602b9984408f98d2e69b8b7b1e19 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e923d8c4a86d41d196594155c10fc46f : FAILURE in 31m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/19622648256c448b8348e65960884708 : SUCCESS in 4m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/53608a4389a347af8753b85f4f21adce : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c164a69a8ec24fa39bdffd5dac96ca29 : SUCCESS in 6m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7c006193175c4a85a9343e7730ff5808 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0af7082202604f17ab86754fb8a64732 : POST_FAILURE in 18m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29e3d05b5ed1401c9d9be6be909c61a8 : FAILURE in 26m 38s","accounts_in_message":[],"_revision_number":7},{"id":"dc6482202d6a783c885f35d31e93838dc4de9e25","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:05:07.000000000","message":"Patch Set 7: -Code-Review","accounts_in_message":[],"_revision_number":7},{"id":"90d51989bff7fd1553deddf64ab786d7a7051bd3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:07:21.000000000","message":"Topic xy removed","accounts_in_message":[],"_revision_number":7},{"id":"3e88c7263b893946e12d643eddf13307d1c2a2d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:07:32.000000000","message":"Topic set to openstack/requirements/constraints","accounts_in_message":[],"_revision_number":7},{"id":"3313a807478864a2dbbbd9a4d712bcde298ed08f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-06 08:19:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"689b2b0f1942a449e4cb11cbb4b637b2e9276940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 09:21:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cdaa757813646bea8befe6c9bfff5b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4af1223c8d44c9b23b6be3341b84ee : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af249cb43b424df090009539b58b1302 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f3b6900e5e74cdc9a9afb727d4f3c7b : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd5104d507914117a5f7807e7146bda8 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6c3927c675d4fcc945b3bd68149125a : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c91f31778fea408fa8d12f0251abd521 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c58298fa0661469e91ab367f612e3798 : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7a47cc12b6646aa92b5f0dbc3257a9c : FAILURE in 3m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f77b310adba14ed68d6d9bd682669baa : FAILURE in 7m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b8b3762eca18434b8179a20e2b6de6fc : SUCCESS in 14m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ede143e6c3aa45809838231607e8c641 : SUCCESS in 22m 18s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e722f793e0d54028b7cc458d593ab510 : SUCCESS in 7m 13s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3ad541688afb4b04ba5535bc4fe70c30 : SUCCESS in 5m 17s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c61711b5b697437c90a8fcdd29267370 : SUCCESS in 5m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/002b1776d1c84d878e06da7b6181e14b : FAILURE in 9m 55s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a478e79fa5c842a8b719a8f78623831f : SUCCESS in 5m 52s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/db1654ce92ad458499180efc8a6c558e : SUCCESS in 4m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/21e957b12984430ab73f08add0849089 : FAILURE in 9m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e710e7ef3c074a4c86a5280612bda5e5 : FAILURE in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cf049a8bb0641b48bbc5171a5d88b07 : SUCCESS in 6m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8c5d3a47226741d483fa9e20cc415ca5 : SUCCESS in 11m 20s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fade1d381e154e6d9e1147240e3c20c4 : SUCCESS in 10m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4a56c894d7884cafbc96ea81f66db9f8 : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bf20f2f56def4aebbe8f3f6b35b96cd1 : FAILURE in 8m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/902e089e48a24b12a3358af691cd9e62 : SUCCESS in 4m 13s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bf4b5f2024a443d38e1b5a2e506023d4 : FAILURE in 4m 36s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/98a4093631774fe5beee7f29ae044a4b : SUCCESS in 11m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/29481e7bf5d443fb961833821335b4d1 : FAILURE in 21m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/06411571b6d3477ea5252fa805141af5 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/269cb7d5e47d46ec971c6f6e39962ecd : FAILURE in 12m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/20d567450f3240a5a1f2422b31d6b8a8 : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e450c17f710a4eddb6ad516ed8f46165 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78a3d4f2644f4383b407c0635fd9432e : FAILURE in 23m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f3f7c3b257ac46c1aeb753c908d08582 : SUCCESS in 3m 54s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/22904494be79413a8ff0c16a5367ceb5 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7f95bb75d044dc2aa9fae5bbe54a429 : SUCCESS in 6m 06s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/35d314e11c424c95a056f7576a40d308 : SUCCESS in 14m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a5f11cdd3c3438bb48566d1add011f0 : POST_FAILURE in 22m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8aa7039c13246b8b6b2c144356142e0 : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":8},{"id":"60d1bf71b3e61ac94bd4547f39c65f31717f78e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-13 08:18:41.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d62b1dcba6a07ff582aad6177da45e72720be5dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 08:44:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d206bcb4802546ba8d9f9c6bb902eec0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d299402b9b4568ade24628d7fb4c2f : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b542c1f47047a2b3282f6337cee270 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1eb54c1c25b487ab0bffbdf1c29c630 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef6d6030d1254333a625d52f3ad5a930 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5d7731f7cca4931a16974eb00d8544c : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0accd46c983a4446985745508342668f : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd80779cacf945c9adcc47f75d5c668d : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8fa8c7e5c4e0447e8380e516c01ee9e6 : FAILURE in 2m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/18d21c87fefd400799a244cb2578b2ed : FAILURE in 3m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e6d77dd8caf74bd1bd1b1ba4afd3462e : SUCCESS in 14m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d44b066aab3b436588db3a952e5a63cd : SUCCESS in 19m 05s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c172f9c2336d4119a70a45b9dc92e7ed : SUCCESS in 4m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/7eb6377c212643ab8a48d23c1f015af9 : SUCCESS in 4m 57s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d4283ecedc8b43c3a02ce176d520faa5 : SUCCESS in 4m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6af4afdcf57241338bad133111b2daec : FAILURE in 10m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2dfe2820309745dab5d3ec80e0f8df23 : SUCCESS in 4m 42s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b253a3a23402433994b03701eb985acb : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/149616eb7f1e4a32b699550bbc476b0a : FAILURE in 7m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5b4d6d442e614c2e92308b257229fc8e : FAILURE in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8decb48d043d4c21bd31b28df5f76a22 : SUCCESS in 7m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9e0fb036cf614173b070e961eab35fbe : SUCCESS in 8m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6376fb93d1474037858771bc4deed7b3 : SUCCESS in 8m 54s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/54a93b00a6b04b76b709c737b231d561 : SUCCESS in 5m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/317e808a83d5405abef3e03cb75681bc : SUCCESS in 9m 19s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/497fd3550b8e4e3897500105e9d1165c : SUCCESS in 4m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/32e06d1f9d824241a0b30f1d57f34fe4 : FAILURE in 4m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ff8c792f850d4656ac58a93856c6b134 : SUCCESS in 9m 54s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0501a8f511364bf3ac78933b75e46767 : FAILURE in 21m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2af9052305c7465c9d0d4582401e837c : SUCCESS in 5m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9d31a5fe18fd4b0cb155de54dc3e0126 : FAILURE in 10m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/86b0aa40ec08407eb00876336602dd0e : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b9fe51de91c45d58d043020e347275e : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/080ad8e4f33244249a620eb05c998b9a : FAILURE in 21m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c5058bb721a94bc6b1a35e454ec126bb : SUCCESS in 3m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/45005b4425e64e299ac5299188d9f52d : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/874c2da4247b4287bf237e0da0b68f5a : SUCCESS in 6m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/559a1944f73240d3a37b42c2bad5e230 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04dfaf5bc0c640bf9830af8b0558a546 : POST_FAILURE in 23m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ea17ef0eaba42cd90f2d2de723e28bf : FAILURE in 22m 39s","accounts_in_message":[],"_revision_number":9},{"id":"8bd8f118ef0ef04e8146fb55ced96cbcc545601d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-20 08:19:00.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"7dc29877e6e51b9440f4c6b1147ffca3560ef8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-20 08:54:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f95b61d69b0eb117c23485e155022b3478b7525b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 09:47:51.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65dedaa825564b07980312e709c8245a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2b1a78c5134b798b2370ee9e9f8af3 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecba0eea3bb14d5bbfd89a9d421af224 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69dead0fdc454d2984fca3b77137ade0 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d8e2d6914764ac383b6db969671b2d0 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58243519bd0b41d391372ac2f6d6fecf : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34d46ef68bd54b9988b30f818547467a : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b6b26fe0dda4b6fb6a8d06dc7b6934a : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37a402923b92481986433468fd17b40e : FAILURE in 2m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08cbda5d713c4c48a0c4747560110068 : FAILURE in 5m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eca0c3df657043e08034712491b479f3 : SUCCESS in 13m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6da677a8b65d4dcda3fe18b8bd77403b : SUCCESS in 18m 34s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e526b9e7105b4611969950564c242a4a : SUCCESS in 6m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/cce02ff927624a08a85ed7c25460d0a9 : SUCCESS in 7m 19s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/91b9d433612d456d982f5147bb7de357 : SUCCESS in 5m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ec23acbabc4a4cd288b19f483216c16d : FAILURE in 10m 25s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c4a57bcac9c944d789e51ddbdca94c37 : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cc8cd743bee54ce1b7477d0615211001 : SUCCESS in 5m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f71258eb14084a0bb5a1f4b1a34ca90d : FAILURE in 6m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b47e22b0be554367bbf08a50893c5da6 : FAILURE in 13m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c77af6bc5c224ff3b5483a937e22d850 : SUCCESS in 7m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/12631dac21fc493984720f96446eec41 : SUCCESS in 8m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/dea5d5e775b9415a93ea919f64dcc848 : SUCCESS in 10m 24s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c922ecac70654edf919d9afe9d3a6ac3 : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f9a14bc496d54f7195edc79b287626db : SUCCESS in 8m 26s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/6c30b5642f3c45fc8da1baa131f703c4 : SUCCESS in 4m 26s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ae4b2132997f43a18e0e63506c995463 : FAILURE in 4m 10s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/45313ca472c1430bac59ab9aa9fb4b56 : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bcc0808528e44aff8966559183563896 : FAILURE in 25m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fd21fa2fcbb0417fa3f085ae68ce963c : SUCCESS in 6m 51s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b0add8569e4c4bc3b5785e6784955048 : FAILURE in 11m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/acf3d29114fc44ffb6c7ec9f11adad4e : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a32a500867844c4835260dcc5d00d42 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25351adb7d0e45ee9f5a9cebba4b6da9 : FAILURE in 20m 38s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c4dc8e405ba34495aceaddc5771ea247 : SUCCESS in 5m 55s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f387b743a4e14908b1022f72a639ae80 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1604250429245cfa7b8885143408e21 : SUCCESS in 5m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/45d3cde233394072b5a4007db7aebf7d : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/817c2481f3874588b891031a89fd48bf : POST_FAILURE in 31m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cd9f5f2b71f4d46b978b5431a51d001 : FAILURE in 22m 54s","accounts_in_message":[],"_revision_number":11},{"id":"58252e2d02e9922b8044f18a2220dcbd30703700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-27 08:18:40.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"02f66c19d0328b5654f02ed1bc7df6d1430b8f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 08:54:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/330753f72c2f4a888b0d334e9e494fd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632afbf3c0664af59194615fd3de4911 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9371006e4547dd9c411f3dd3016340 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c31195617a2a4ec8b01de7e730cfdebe : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6db30450ed4b4cd5814df0a4d2a3e72d : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68157f6ba7f74260a55afbebacfee307 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98c98cb4de904c409b339785e59d8fb8 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a9e22a4f7514b78a97424cc6564aa15 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9f7d69e3db4844e0ad584622408a3a0c : FAILURE in 2m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1c26efa0dce14599a836a2b9aee93e49 : FAILURE in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d19f22bd2e494d0493b52d411f93894c : SUCCESS in 10m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b154abd53cbf4a0aa84298918db517c3 : SUCCESS in 18m 06s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5f5879e29ccc49c583a11634ef7e9b1e : SUCCESS in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/bb8383c002fc452eaee279a031eff796 : SUCCESS in 4m 33s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8b5bfaef728142e38c0ac52064e48a9c : SUCCESS in 4m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/fb362031cbfa4c6aac81be6c72974d83 : FAILURE in 11m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a18e37571ff5491e8e224a58b9522d67 : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c02daa6b79ba426b9cf4ebd97e15df27 : SUCCESS in 5m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e77a9211cafb4449ae543ab2e9816295 : FAILURE in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/58a09debe5c24d66a56aa3f9b30c3c14 : FAILURE in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02db2fe1cb22474f9601a10b3e271beb : SUCCESS in 6m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a45bc2204cea4f189a348f473470eea4 : SUCCESS in 8m 29s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9789e960abb34230bea3a22547c97385 : SUCCESS in 10m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/62866a570be94b23bcaf21fd79b4db83 : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2a6e786b2143492b9da546bd7bab08f6 : SUCCESS in 7m 43s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4b11d4413e61456696e809580d1044fe : SUCCESS in 4m 22s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/23cf5d83ce7141dfac0a0870dd02a3bd : FAILURE in 4m 12s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/33c797118b734a80af0e474d84d37d63 : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/335668d1f8d2434f990080a185b99498 : FAILURE in 20m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1657a48542e14a84ab20bf642d4513a6 : SUCCESS in 5m 45s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5dd1e89d87184235a2a5e2ab8824043f : FAILURE in 9m 32s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/637de8798a574d7e91873e1220e09233 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c8ed93e8ddb4b08bbc2fd675150dcb9 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cd1bf6e04734ebfacc54758080d1d8d : FAILURE in 22m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ad5df0fa53024bd3a5e5359b691b9c76 : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a297ba75ddfb49fda82555d6a805b3d8 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc09d7a2a349473ca6247a15179d6a76 : SUCCESS in 6m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5843e9528e4c4f7487f438dde9299f85 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35849556016948e18acda44591142d72 : POST_FAILURE in 18m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97d0ae4982044785a27aa8873a6c6b47 : FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":12},{"id":"2bc079bec9be799823ea95690f667323b8fd7769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-04 08:14:54.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"7c9a9874ff50763a5583b49f6e4069be4f4236b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 08:44:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94981f167cbf43a59b72683bfce17ac9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7cb9068d2a430ca94a44dd4845914a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5083c7fc99d04aca8c240c3750726ef7 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70b90cc3bbbd4ed884b02581dc32404d : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd69c035b0954d508daf49fd363cfe1f : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/222b74ce8ec741679b31e896e188cd15 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8bbdfcab9bb4c6491c884eb21a1a2b6 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f83a8bdb5b140fb9f16a8b09628c602 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e439ab34a3f4d61918873b67dc52074 : FAILURE in 2m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9987f2788d1e44428d3b25cf8cbb6a51 : FAILURE in 3m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/91d73c2cb2624d7981587a71f17bc02e : SUCCESS in 10m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a4e8f3eff739472b827ed24c56d51508 : SUCCESS in 23m 42s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fe81e07334614437a34dbc8b78d83300 : SUCCESS in 4m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e56473acbef44dcfbe3821422270af18 : SUCCESS in 5m 09s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/49360a55717b4a4da6500d4510c1c5fc : SUCCESS in 6m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/66e11447e19145f088d730fbede78ae6 : FAILURE in 11m 02s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c0d52e07c9d349949743cbd49ad4735a : SUCCESS in 6m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/77eefc9094f1449aab168816e535fd35 : SUCCESS in 4m 59s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8981b9050e904308a669c5483fef5bb7 : FAILURE in 9m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/54f9b97677874befb3b2496611c5bcef : FAILURE in 7m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e0dbb4d29964a98bec89dd62aa0bd49 : SUCCESS in 9m 05s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/20e881087103428eaa03ec729444026b : SUCCESS in 8m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9c1528d197e04dbea27d9941e32e99bd : SUCCESS in 11m 29s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/512185ed30eb49789a923c12bd29d907 : SUCCESS in 4m 37s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c6d553eca91b464781c61d38286780d0 : SUCCESS in 9m 11s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/51ca2eda1534485dade75b10be48767f : SUCCESS in 4m 25s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9cc2a04fa77d45869f265f086476f5de : FAILURE in 4m 29s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/faf4d68a26b14d73b6532b26de849c8c : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/49a616c1a17240449ef4e5ec55d2c655 : FAILURE in 23m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c98024ff7c7341eaa8e67d13b4116b51 : SUCCESS in 6m 07s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/74b597f6e602481fb62210c402defe42 : FAILURE in 11m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2566ee0c03a447a3817c0b5f8ff2218f : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3bed8cad6f3e45be94014b2dd8063fdd : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24bb968184ed4e8c9f7eb92ff9a331e8 : FAILURE in 24m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f8cfbaef01cf445ca8a6b10f496e706e : SUCCESS in 4m 06s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/537f26494e2f48a78a979dba45ca47c1 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ff804514044466fb8c8458f654ebbc0 : SUCCESS in 6m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/938822146d214710be4fd78aaceaa49d : SUCCESS in 15m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f023958bf7784a2ba869e6eb5e053918 : POST_FAILURE in 18m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b66a6ae14dba4d4da3902029c3754439 : FAILURE in 22m 09s","accounts_in_message":[],"_revision_number":13},{"id":"05c5488173fa2b5b012548ef6a1467e4ac5b7e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-11 08:17:17.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d5515f615812b26cdc5a706da876c7a2e8624d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 09:15:54.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b79bdb47df34cc78f8ae34526381bfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78515329c2704c41bdfefbb3a6c63ee0 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007c3838184f4e76aea68ba57b96a377 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0db8b8bd3f4fd59dcb4c1ef310e4e5 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/761bf74c47474dfea495b394eb436e02 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89af53a9cafc428a8d8e8f2e1550b426 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/514e26c4ca0d420eac1e2a81df9d6e5c : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bfc6dabc6dc4c52aefbe32f153d3a68 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b0ed772f9ff454aa5b45d117892f3da : FAILURE in 2m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a90668c2250f4e4bb9779f613fa77019 : FAILURE in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/118a61949e5548c189a9af2a1483ea31 : SUCCESS in 10m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8383d91d14854a51b2d20246addfa4c1 : SUCCESS in 14m 37s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/be8a39708b84465ea7b11b7c4a7e69c1 : SUCCESS in 4m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/149bb73a8a434b058e978cfb916565a8 : SUCCESS in 4m 54s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/25c3c517d1164b03adcd99831b9278bf : SUCCESS in 3m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d81c0a4c3d124a2ea2ba8360ec89d279 : FAILURE in 18m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/551de5d847b440849ec6381c2180c258 : SUCCESS in 4m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b8e44749647045708f2c2ce8eb6bc309 : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/551e1ef1c2d143b995aa68e94b09e232 : FAILURE in 12m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6d0d24d754e84d36a5054a30e7a28efb : FAILURE in 7m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c7fcfbf45014b3dab460672ff9314cd : SUCCESS in 7m 16s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/59a83a59dad049acae5ed224b3f1c10d : SUCCESS in 7m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/37228c8f02484c41a3c40717a1b7a350 : SUCCESS in 17m 51s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a014f8b3896246708087625defe4a7b4 : SUCCESS in 3m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/083ac836678449d1b6826bdf5d0e60e0 : SUCCESS in 9m 46s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/768cafe0d88a455790a1cc042d857e5a : SUCCESS in 5m 43s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7e0fe92cbde84c13ac4333824be4c9a6 : FAILURE in 4m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ad19b20da35a43babc6375ca4235f5b7 : SUCCESS in 10m 27s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/384bf6c76b9a494fa092590471f71acd : FAILURE in 26m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fe912df97bfd4c74b498f27d3468367e : SUCCESS in 7m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2687bb731b6047cabd5b524751296f0a : FAILURE in 18m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f99d03789dc746cf89bf9374fcb6ea14 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2db911c7cf94b98b4977380e9b1af39 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e3784f0c9b8446a9681891111e754f4 : FAILURE in 28m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6611911e3adf493fb884296ad3dd8aed : SUCCESS in 4m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4324d3a1f5be40b38cab63185edfe0aa : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4133389bde6145f4b4d3144e2399a595 : SUCCESS in 6m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b75ae160361c47fa94c029af39e74b60 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fd7f9ccbbef4b8793e69ec3bc9cf4f1 : POST_FAILURE in 30m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ca235e3e54c4ab1a21e7a9ef3b452aa : FAILURE in 36m 44s","accounts_in_message":[],"_revision_number":14},{"id":"044e959f199a57198e2e347499f73052e123fd84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 12:23:52.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"11ed190af83366ab4b251e098a61576c952f51c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 13:02:31.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c15774b68f341b8a9efc019b3ece147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6847505ecfd64a3f94992027010f8133 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13759ac0f7bb43e3a5be3baf6ef6056e : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f67b0861d7094a10afeaaf7a11c1fe82 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fd54c48363f4ce8ac23d8b10f19894f : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4518684d4a6b479db3a835e11849d403 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78fba88bac4a45579655714d39f59c1b : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b1a5ecd987d440c84c38f419d4af17b : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff0eee7a65cf4c7ca1c302bb21d68603 : FAILURE in 2m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c25377d2a25745f2b228a74d77931e72 : FAILURE in 3m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/727a9d78c97945ada430c06fa0faa50c : SUCCESS in 11m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/24e25be1dbf141bfbbf39e14637008ad : SUCCESS in 18m 09s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a51519c874e04d33b0b47763721ad3c7 : SUCCESS in 5m 08s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/acfd065b8c034b7788c3a4c8a7fff3f0 : SUCCESS in 5m 03s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9eb3a497b85b4a3fb22c4c3803af8270 : SUCCESS in 3m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d5a3251103244a9187ee9a6825b57d82 : FAILURE in 11m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0ed115436a4d40e0a6822ff1a83d9ca1 : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0b9c97a1e64048cb9cfc9e48ee7ff73e : SUCCESS in 6m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0c9d84a8b52f4ad8abd3d9db6b3fc873 : FAILURE in 7m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1ec156227e964403abbb306e05000348 : FAILURE in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efa38e3ceab44a72ac72d29d11415dea : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/11b4e112ac994741930e2ddd9b093aef : SUCCESS in 10m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/baa7430d948c4594928d1f1293468548 : SUCCESS in 9m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a405642e03074628a25ca3f7c389ebfc : SUCCESS in 4m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5805aa2435aa4dfcae45c2c419a1efd1 : SUCCESS in 9m 16s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bb3ded1f9cd04778a56251b1f8462fce : SUCCESS in 4m 57s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/dd2c3ac0855a495eb41d62c3b2f28000 : FAILURE in 5m 13s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1fd1b0675c594769b9d3777395ba69dd : SUCCESS in 12m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cd1b3c430b484b1794bf46ba5cb41c82 : FAILURE in 27m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/94776ae4ac394d3498e25cf647c885c4 : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e23ce1b7b4d24a3689089a87498c6418 : FAILURE in 12m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/10182a2edeec4fa8a25e4e29114d539b : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de0e36867de04429ac2c1c5076b86abb : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86863cf4544342438484b019b9076c80 : FAILURE in 26m 13s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/08ecebcd5b2140c587cf10c30982f8ae : SUCCESS in 6m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c49e92ec88924b8f99ee11718d4a1383 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53b669fb12b0425b95a4ff556650b5f9 : SUCCESS in 7m 33s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ae83036089d54adda9910c681919041a : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8a7e2973dd5409cbbeb6593235507c6 : POST_FAILURE in 30m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87a9f5783a4045ddbbf4e84da54c95e4 : FAILURE in 21m 15s","accounts_in_message":[],"_revision_number":15},{"id":"f760f9ab8603fcd99020879fbaa49241a0cb4d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-18 08:12:20.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"02ed962832f40a466bb58d37af328b6ca028541a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 09:24:15.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23c5cb6ec5ef42ad959372ca0b6695ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04be3a0b4ea44f5b90c053bd7b0397d9 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f6c509ad7a44598ceaf5a7f09427e6 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e7ea0a169c34c9bafe82680c316708b : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70d8c68c433c4813a99ab314957c5ec9 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fada2f6486b4ef3ad7062d1ca3878b1 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cdc1571109c499294b06f6fefdef01e : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c6a73e620304d5fb2355ad54ee717bd : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb4fabd57e0a422c87599bea512a9a45 : FAILURE in 2m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0997c9442b644133abba04b4ed5d0957 : FAILURE in 4m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70a79215565347d8872c83ec848ab8db : SUCCESS in 13m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/90fae862906d48c9be7460d1ba70cb0a : SUCCESS in 20m 20s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7653b91d1cc24c4098dd292f19baf76d : SUCCESS in 5m 18s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3c18c5ef9bcb46f1a39e5db3cf43c1f4 : SUCCESS in 4m 36s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0d17dabe4cdc4311a03c530bbcd8110c : SUCCESS in 5m 40s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/158596a582f543628172cac56ef9460f : FAILURE in 9m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b6d2588b842a458f8b6c221335b7f6cf : SUCCESS in 5m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2f213037580f419683251864bf175a57 : SUCCESS in 5m 48s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cceb996342e542f2aa0513e2fc8374d6 : SUCCESS in 9m 27s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/55c2dfc015c04f15886d64f8a2dc281c : FAILURE in 7m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9a3903cf51c4268ae70a75cfa044b9e : SUCCESS in 7m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f329d936679e4559b46d90912c329fe1 : SUCCESS in 10m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ec0961e329bd481cbb040d5ea34c4a54 : SUCCESS in 11m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/64c459873d8f4a6d85d8b6736d2a830f : SUCCESS in 7m 15s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b61e013f7b194ab3abd0a6ee70b88955 : SUCCESS in 9m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/2c8f611baa05499eafea34777156dc53 : SUCCESS in 4m 41s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/19604643984f4392adc8f726cb26018a : FAILURE in 4m 35s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/081740e0c2a84c3fb22502e392497507 : SUCCESS in 12m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6856caf565514abfbf46a16d33d4fdab : FAILURE in 24m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/39dec48d988c421787dfc4a96750e566 : SUCCESS in 6m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1867550bc16647c9b0d09d0508480565 : FAILURE in 11m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/60207b3e57744fd69ea3adb3f6d6b74d : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e8ee5404d754ce3bec4d6a5063f987a : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0ce301c33174a04927df5d270b99a4a : FAILURE in 24m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/df8ba8c377814b47b68d8af255e8c2a2 : SUCCESS in 4m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/88b0fdc4691f486ebbf03a5906097eef : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3c3f9a6627b42658f56edd043c07252 : SUCCESS in 6m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/71e812ab68914648ac67d4093921431a : SUCCESS in 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9232ed1c8a234f448c21a3cc192a63fb : FAILURE in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fda5a7889014f95b054a0c049e23773 : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":16},{"id":"d8c3d156adc3566560839ca8a1ccc5a8bf39d0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 08:47:02.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"cef79029f4aeafd9cfb2f3016717649bdbaf3b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 09:39:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"72a4759b4128120876f0e9e591c85e7722a3cf0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 11:13:18.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58d8711a26de4e61ab9bf94a41efa818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b890e69b7cd3423ea0e5aa26c65a9d46 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6a0ed5b1c64b7fb647127df3a5af35 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e941d88e66bb434c89f4db5c697d42ec : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35b727f54442415cb37028bba6a3329d : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9957c3904069497ea27bd48a01537daf : FAILURE in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cf65f9e28d5409787f328557c1cbd18 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1f6f922724e44468ab94f0781ef7dfa : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb7c15e999e44c16becbdec2f91b97fd : FAILURE in 3m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e2a96724425a4d7498624cc2c57f1742 : FAILURE in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fefa6bce085c44e2a9a7e5ee38ecb9fb : SUCCESS in 8m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/71caaa65db744c728b32443ea836f0d9 : SUCCESS in 17m 44s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/047dc91bf1fd40de8dd68a6e36563657 : SUCCESS in 5m 10s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/af9ae0fe1eec4b69a5a1af56a72cb7f8 : SUCCESS in 4m 58s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/eab790e06470490ab8d0f0172d572bfd : SUCCESS in 4m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/dec497f2bfe648209ecded68645dba6e : FAILURE in 11m 02s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4db39b7109664a8f98fc285a35108120 : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dba65d2adb8c46cbae7d8bd53f64bf79 : SUCCESS in 5m 10s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6b924f6512e4459db6b1401ec4c91354 : SUCCESS in 7m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/409af0299947422d98ed8117fa2dd3b3 : FAILURE in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29c07a6872a54943bbccfb9e63ea5305 : SUCCESS in 7m 05s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f5094d52f3ee40a6ba70f432cd70a882 : SUCCESS in 6m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2cef1a533de2437cb93b70e857a07432 : SUCCESS in 7m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/05c5df44c3d74a9389e3159efb067f92 : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f4a7a527501a43e0957c9a68ce1c5f7f : SUCCESS in 6m 59s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/804a806744ab46ffaa229fa2c7632b8d : SUCCESS in 4m 02s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fefebebe7a574bb48135e4b98b056a88 : FAILURE in 4m 10s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/99ba3fbd1dcc464586e1046832922c34 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/527886ce57c541cd961bb7f5e7671b11 : FAILURE in 17m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9eb85b05d27a432f991565b89387bdf5 : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/870738c4eaa7490788162cda7cde8651 : FAILURE in 8m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5f27fdd9e0f4416ba11ca2ed244ec78c : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2560cded2d840088d72e0071706d5a8 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6cb89b30f264df8acf25cae51d6b02e : FAILURE in 23m 35s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e18b6a6fea154f3d9be851d5c297413d : SUCCESS in 4m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f25db76da0ab490a815ee954b30c92e8 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c5fab108d3e44d5ba20978957cbb19d : SUCCESS in 5m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c7d2043fde0647fc976bbf64f8e5b161 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a35b4cb0ec0c45c79975f0daac398ae4 : FAILURE in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afe362a91dd847f789c24491b687d252 : FAILURE in 1h 19m 24s","accounts_in_message":[],"_revision_number":18},{"id":"1aac34d4ad2f3cfaa56a44bdaf39cb745428123c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-25 08:33:40.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"304637c137cea531a7f321040f66e34bdeb98453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 09:53:18.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9270f7605bfb4bd48d9188c34f381e32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661e2ec03cfc466da7a2b23ad145eb67 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d417270dbf546848bf4a909dfac5533 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c1a0b37867c49b8ae7989a62f4ea3b1 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb04d382505b4a789139ce085296837f : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00848acb2d984aadbecf10a1d836f6b6 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a48ed2a81dad4360a4be9998fe28c3aa : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cefd9bb82fe483f9b8d4def47157cb6 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/009c3eb48d6f433b8c79ac8b65f39365 : FAILURE in 2m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b92acce608524f88b3165b7b81f18951 : FAILURE in 5m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5fcdf4565b9a4e9f9608bc861497dde9 : SUCCESS in 12m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59a2dfe574f6491dbf44768f7bcc4aa3 : SUCCESS in 17m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ec27a79a1aa74a1abbe266c8fc659009 : SUCCESS in 4m 31s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d7c49b216833428dbdf3b733846620c9 : SUCCESS in 4m 33s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/55459d4c5b75416a91bdc0b03705d0e6 : SUCCESS in 3m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a2b0a147037d4a02acfc7c7cfadcb0f2 : FAILURE in 10m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/535c839535dc4672a5b69eff7e27753b : SUCCESS in 5m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b0a2d7494f3946d8a58e55953b81f824 : SUCCESS in 6m 40s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f09facc2619c4634b43e00cbc1b83070 : SUCCESS in 8m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c95c064825da4f2ca6b3cfe00124f87d : FAILURE in 7m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51424265792347bab325044cbca30345 : SUCCESS in 7m 33s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f276977291a04beabe7beaa646284910 : SUCCESS in 7m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b487e850ba8c4c4f848a0b602a98e1c4 : SUCCESS in 11m 39s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2306472d2c814663a34a985c390bb0ca : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ead90162a6e448c5bdb8a12e218a706a : SUCCESS in 8m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/670f2323eb6c48b48fc2e971c7ae5636 : FAILURE in 4m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/21e614f982c34e1f96e31134caf57c63 : FAILURE in 4m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f1ac73439c7f447c86fd5903db4242be : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2cb90a0e43534e0bbba2b77972c86d4b : FAILURE in 17m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2e1431b5f6824cab8606824146fcb217 : SUCCESS in 6m 25s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5950491cec5c4a30bf943ec54c1e0051 : FAILURE in 11m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4c5df65d157f4161a5b03705d2295570 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/185263692ba54c69bc6bf28586346c69 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9b601a798b449a085a2b6609eeed056 : FAILURE in 21m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0d162ed8406a4350bc426543f6dceea5 : SUCCESS in 7m 07s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/73b81c751213451a990070248c1ae9d5 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a668d3435284503bb7c577982a374e1 : SUCCESS in 6m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2b1d39e98c6344d4bcd6ca72f26b547c : SUCCESS in 14m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64ccf3d5e21042c581fd0881caf4f3bd : FAILURE in 50m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24934f0695f644a9b0df7ee0d5bac258 : FAILURE in 1h 17m 25s","accounts_in_message":[],"_revision_number":19},{"id":"6b8b6a8da5c4db93192315788bb8499b56e28a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-26 20:02:49.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"fc6c0dad1ea9393890e7c28b2600ec17a260a980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-26 21:36:06.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bed61d8c80645718a8b56a050c1ee81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f479180b6c07410d902d134691698f91 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3764a35437b94daf8c276c18038d62b4 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d00d78566db49b493be2af484293839 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6690010d2d0f49e8bb7bdb805e9d7cd2 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdc449b7dde84cd29dc04a376fcdcda3 : FAILURE in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3eb6b6bc0e145ac9fc78b38bd867c09 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/896ed5876be74e6888d1504b82eebdba : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79ed758e7fe94f34b75963f4c52647f6 : FAILURE in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/938bf95d4248491b99abcd7b8c74f807 : FAILURE in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81c6bb62e24b4cd2a8f4f4d8b298d8f4 : SUCCESS in 10m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b92f08c5fa44482b2f13f0f22b7706d : SUCCESS in 14m 41s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6c521e86cb694a54842a5ce54e297c53 : SUCCESS in 3m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/cf9f8b0927ec4878acaef80045b47af7 : SUCCESS in 4m 09s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0d0fcf6b62a4408cb3bb95bdfc2db0ac : SUCCESS in 3m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/80f8950efd064449b9bfd7b3a68ca014 : FAILURE in 9m 28s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8cc70e8540b84453ae9d3e47537d6cc5 : SUCCESS in 3m 54s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b7138726da584197b492b895be1332b1 : SUCCESS in 5m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/15118da354f04db3a39c088cc28aa471 : SUCCESS in 7m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7c6666b1555f44cf9bb6946f3aa802cf : FAILURE in 6m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f850c758d76f46b98cfe61b419babb44 : SUCCESS in 6m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5dd95f8e097442bf8bc0b065c71a1fc4 : SUCCESS in 8m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/74281717b6fb428b856de2038a782462 : SUCCESS in 8m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/08de9590ca5349f6b5ca3d53cd2d88e1 : SUCCESS in 3m 21s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2b12b738e9eb4a7e984680edad424dc4 : SUCCESS in 8m 40s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4d2a9c8b3ff14e40b58ecc8a2e817ebe : FAILURE in 4m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3dab60b5dab540e4baf4b10d9555daf9 : FAILURE in 4m 29s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e806a9d0fa57404bbc10fbc7d7f721f1 : SUCCESS in 8m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/dbe3788844a540868797aed1f5fde673 : FAILURE in 17m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cdb012598fc84bb69a7330fc4be33c24 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/06e5c5c52e8343e8b9d55baa248eb181 : FAILURE in 9m 30s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/536dff3948a643e499b8c217055561bb : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e82d7b5715c5498b9086cd745e512df8 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a195ff40a6c8409384cd24b4a9854ef0 : FAILURE in 17m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d6deecb2a3d44e8f91f7dd43f70c5e61 : SUCCESS in 4m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3848e9c9a48c49539fdce13df13d5618 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b18257a04cc94ada92be40dd944bebe1 : SUCCESS in 5m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/53c4dcb612164fd59522a623b87320bd : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4233f43bdf0d4f67b0c3556290ddee8e : FAILURE in 1h 04m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96efa61a46584feb9cd8c2d4ab8dc077 : FAILURE in 1h 27m 36s","accounts_in_message":[],"_revision_number":20},{"id":"0127a9d9bb680b16f9902a0747905cfb0e1c054c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-06-01 08:29:25.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"fe2780a2d73acbec384a76ff35139637c52cc0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 09:45:31.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e98bd8229bf544c497fa34e3b9e47955\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c53dae7b4344c58d98a8625382be4c : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d5dd518fe1410d84d1c4b8cda4d728 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/698d1bd0e90a4124876c8bf8ad967a00 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e6fd31336c349b49382a5a0410e235d : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27d4da902eec450a813c94d516aa5ace : FAILURE in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d82f1b55fe7c49c5bc8186dbbecb5bff : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f2398447f6e4953870df400de0c2359 : SUCCESS in 2m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0699d27e777413799f7ebce8d4575f3 : FAILURE in 2m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/841eee07c3cf42e5a32b4018a823da68 : FAILURE in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/017f23f9676240e8aaca64df18ba61d0 : SUCCESS in 11m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0ec4b454b6b24ba592992b21f531bdba : SUCCESS in 20m 04s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7ebf310fabeb422e92207bc9c13c3407 : SUCCESS in 6m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/aabca098eed04fe19e000c411763487b : SUCCESS in 6m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9ebd539ddaf140f2ab7940d2e4eced04 : SUCCESS in 3m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7b544eed586b434783148b74cc852b9e : FAILURE in 11m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a2a537bc3e084331a67f0d948b0d7c9e : SUCCESS in 5m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8badf3b64386475e969c72b45c657e05 : SUCCESS in 5m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e5f0435ca0aa492fa19d80e365277cf0 : SUCCESS in 8m 38s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e0640732e2fa4ce88cf5f5c6f059b5b5 : FAILURE in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e31cf235499f4c3dae990aff9ded6667 : SUCCESS in 7m 06s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4a6e9e83601c4656a403cc7d65327694 : SUCCESS in 7m 24s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/124475d1fe3d4f57967cf901fd4e4a0e : SUCCESS in 15m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7196089961b04e9e849cd72befee6e89 : SUCCESS in 4m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/987869c2e880483db34197359be471d6 : SUCCESS in 8m 04s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d59f616e219f4ab1af309d84263d2752 : SUCCESS in 4m 21s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2c6da36eab7d4b8db56a9827bdec8d8f : FAILURE in 5m 44s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3917a080b1b141d0bdcafe28c24b19e5 : SUCCESS in 11m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/093bf31188a145f1814ae0961a40ff04 : FAILURE in 29m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/52aeabe8c3b04cfe91061374a07533b9 : SUCCESS in 5m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/dd6228af05c841f886b42c70c6dc834e : FAILURE in 14m 01s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7cde90289c704283a7e8f93c8c864657 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98ad483a5a3d44d686de8710050b3eac : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d229ddbea9904d11bd828fcebffa7053 : FAILURE in 19m 31s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4885074ec49c4137ac83d6429ecc7e1a : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6ba750a58fc94e839d95b87bc43dfad4 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83dd96e7e653424a947a6adb2a82e648 : SUCCESS in 6m 24s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/abb8a3f1533c4ddb8d2b2c2d69be191b : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb6f17931f34b7c90bd0c5edf8390f8 : FAILURE in 1h 03m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f6fd7e3f5b74a99b9b14dd0f0a43677 : FAILURE in 57m 50s","accounts_in_message":[],"_revision_number":21},{"id":"2a05ad258e67e07966b6f1f05a4fc770015d5b79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-06-08 08:17:35.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"13b6f3de37ba4e372c5f8dc94359eca8374318db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 10:17:48.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97ce395ed93543dfa16814eee3d22390\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382aac8cff8349c5858681db46aca1c2 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f60972c52524b1f8b363113b14d2b8d : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ff6b681408e498fa164e55e966f5850 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/149eb703193b42bcbb2881d63d652d94 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db8676e1e87246aa9aca3cf13304fe9c : FAILURE in 6m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3132bc0cbd224218b8533db0db0dff06 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13b72361750b4b97927fef5439ea8f55 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2bf60690fef14a1a8cdd081292903deb : FAILURE in 2m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05e95270d800471cbb02c30c1d47321d : FAILURE in 2m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bcbc755d1be2458a9480d1ea2a316ef2 : SUCCESS in 12m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2dccfb128bde476f86b475f145990993 : SUCCESS in 16m 28s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7638edfb7430403f95e7d33e86d4fa20 : SUCCESS in 4m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ba10300251b2407983ecc49c69c4c6b4 : SUCCESS in 4m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/49e8f25e662e45c2af34b71cac2468ea : SUCCESS in 4m 46s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f614f9bc423042739840f34f3c4a07c8 : FAILURE in 10m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/04ead6e08ea94a80832d480eaf48c163 : SUCCESS in 5m 44s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c429f78cf83d4901902d51ff7ab7c437 : SUCCESS in 5m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1fefefb0285c4b77947c6d017ee60a9e : SUCCESS in 6m 56s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/15b237dc6ac242ffa32f017650109739 : FAILURE in 7m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68c52071200d423587557e62be0aebb3 : SUCCESS in 7m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8d91a9c96b6948f2afd5d09a81c07216 : SUCCESS in 8m 00s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/50febc982e6a4d41b4bbbba719db8976 : SUCCESS in 10m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6849cd7a4638447d8788baf0d543e92b : SUCCESS in 3m 55s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3bd69c3f7c134073b41aba8f0bd645a6 : SUCCESS in 8m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9b66ec049e4d441ca30aa36163cf6efb : SUCCESS in 4m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1e3d2d39f0d24ea1a2ecc7fe9c17efa8 : FAILURE in 4m 26s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/cef0a06099624d70835a22d155139d1e : SUCCESS in 11m 40s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/32958395033440bea8bf9b8d7229bb4f : SUCCESS in 31m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3145f28b7af64752860592b89d016a0e : SUCCESS in 6m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ffc6b641bf3c44fdb9a83d8c59b9cfe9 : FAILURE in 11m 43s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a628d4da6b35428588d6e461f4174649 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c71a2abef60742b28b7006aa61352950 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6cdf41f58574a66bb39a96671605d91 : FAILURE in 24m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/54112423e57f446a9356f18064bbd125 : SUCCESS in 3m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/132908e16d8c4e5fbdfc0678b0a9d9da : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ff6d54418e948e5846d4fe83a844fbf : SUCCESS in 5m 24s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ee420785f44447e795562c96d3f3d09a : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5f953ffa4544a6dba7bce2cbcbe141b : SUCCESS in 1h 08m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfd4d309c7224dbda198845cf576b571 : SUCCESS in 1h 53m 23s","accounts_in_message":[],"_revision_number":22},{"id":"c3ea426463595b508c513cc4efa30a87daa618f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-12 17:52:10.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"d7321ec293772a69b10a3f7db7b0eb3e1d0799a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 18:37:55.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d719944de1784d969d3c2257f011e4d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc2d76b04cf4c5984b680d481126d7f : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87035a0184e149eab14de78dffc30e8d : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa42963adcd24642a63de5e5d4adb287 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da0b5a43047a486db470be8fc562a286 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2970b16903a44ab9ac2a23e9efece55 : FAILURE in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de47b5b6feb643b5adf95bf39943f3b1 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c2ede634a9349e483e9fad3681a0430 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91ddd45ad8aa4d9b97f1ab1de43eb038 : SUCCESS in 9m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3a79c73ca705461085a3864e43aaf83f : SUCCESS in 16m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a171b7de930847a0bb3ba847dcc96dfb : SUCCESS in 11m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ddcf0a7d9210408699143dc32dc1711a : SUCCESS in 17m 17s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fcb522f3021445febec706b13904590e : SUCCESS in 5m 21s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5a5b041427374bcb93ee20defe5d71f0 : SUCCESS in 5m 42s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2bc02b86d6824577b95f1e709db67ca7 : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2c725635acb6422c8d550d2831827319 : FAILURE in 3m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f976111a3a2b41f18d5161ea0bde75ed : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a4aa94800d0d44d7b8ac06a4192a0b37 : SUCCESS in 5m 00s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1db8c92f9f1d490d8687cdccabf21f65 : SUCCESS in 6m 29s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9f1aa16f43604d34ba6ea492f60da4a3 : SUCCESS in 13m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/816d86bece7a41a4a871029192a399cf : SUCCESS in 6m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7465ddbee98848708ddc9d4d3fe1d7a9 : FAILURE in 3m 49s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/476d6489b6d34c9590969be85e2463a0 : SUCCESS in 9m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ff6f63383ec149a6a6e520b6f1a461cd : SUCCESS in 3m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9c981798580544b98aca1a0cdf3fc106 : FAILURE in 3m 52s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ca55dabcd12d4ca18c5a80f053802d06 : SUCCESS in 4m 15s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5ec786208d3c42458b764ff2a654fb38 : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1fbbe2c4217149b4ae66ad00fe54897a : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/dfb74c51e7a24777b787c24cab6a5b06 : FAILURE in 4m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/25de92c387c448fb86b193eba6750251 : FAILURE in 3m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/dc63cf7cbd0f4147a7baed038c2b5ec9 : FAILURE in 3m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a7d3d729ca464c6c8bb58b49b9df6582 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2059ec5d2eff46f489a6c0eaa30302c1 : FAILURE in 2m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/038e26db9fe74d09a6f2ea254bb32aa5 : SUCCESS in 18m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8125329139304a0db7080456347407d4 : FAILURE in 4m 51s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ebb6a276f8bb45b281dccb98692299b6 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6905428e985419f8b0d0453451b892c : SUCCESS in 6m 24s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/afff5bc71ef24182ab6b3f17360e5661 : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f8eca79908f4a1197d8a1d96228c834 : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":23},{"id":"8d9b07db2d84b5de0032746857e0b0bd4bbb71de","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-13 12:24:42.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"d4cab568f6afee8c580361af091a53ac76589ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 13:47:05.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f4ea84bd8d542f7a15f155746135bae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42988b090084663952d1d4a35882e7f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d249539651d4de3bb5f0eb4d80c5229 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f69242293cad4722b14e56c4cc2c967d : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3abe4c8ab1246e29dbac1ddcb024aab : SUCCESS in 6m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b03f861ac1854fa7b8e0b5b422cf57dd : FAILURE in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a15a735234847f6bad2aa70dbf6ccff : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10734ae8c4df4770aa178a64f3544662 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/de81b681828e4bda941d28393d867860 : SUCCESS in 11m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a89d42fc73a34e01bd50342b4b3fa39c : SUCCESS in 17m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fdaa97234ee34ec4ac5a83b2b7e68ee5 : SUCCESS in 10m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fe96b15149e74a8c92b45314cc2438ec : SUCCESS in 18m 15s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e2e612634b9e4a1197dc46f25b652a68 : SUCCESS in 5m 13s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/393fd8290b184c3187a68e3935c2df56 : SUCCESS in 5m 02s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/525100e7f409495ebe3e4ead2b29eeb7 : SUCCESS in 4m 50s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e3471773a6804ed597262c07c41da3d9 : SUCCESS in 12m 16s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/bf0a553b66174cb3b98084208eeffc5a : SUCCESS in 5m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d7643008f6344eeca90eee20130113eb : SUCCESS in 5m 21s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/14f0dab160614e07aec90f22dca93bb4 : SUCCESS in 7m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8109aac0f5dd41788143ee540067c76f : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d11e4da847c40aab548aa86232f9ede : FAILURE in 6m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ec60e46893e34a0587c1a20b28766a26 : SUCCESS in 7m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cac5bc044e5f4220bda44b21aa19a19a : SUCCESS in 9m 56s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0ba10f5274f04e4b8101f4cdae814e04 : SUCCESS in 3m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/dbc43c96ee1b46ff8e41ebe6939d181f : SUCCESS in 8m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3de42054d5fb4d37bec4eb1f8c5c103e : SUCCESS in 4m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1dd4c95cfb1f4a1fbad2a2a67679e1d2 : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/feca9a34f1cc490fa55220910f68969d : SUCCESS in 12m 43s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/395d1a225de7444eba0bb995da88565c : SUCCESS in 26m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/40d6dce077854d6f8e2f7f9cec4e16df : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/358ed0660d05413d976d97db8ca4fbcd : SUCCESS in 11m 32s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9a02ff68a9944ce3844032e39aac0aa6 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bdf467e9b5f7411ca5f3c49aef25a915 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc22fcd670ce4f8e907593d2c7f0e52d : SUCCESS in 24m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e6f5b056d4434952979a851e231818af : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3d894907d1754df78c9807800ec3c4d8 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f64ddf8fae24747be9d2cfc59c9f0c0 : SUCCESS in 4m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/34eba3b46c794d1e93832d256bad6c86 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97bcd1c393f445a18b14dbb3a0deefad : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":23},{"id":"8c8af12ced5fb045e6fbbdc036955362bccad609","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-14 12:07:56.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"1d8dcf23cdadd83e8280f7643d1fb69a500a0138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 13:04:29.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3735295d5ce644e6b5cd6e29e9c5fd08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3eb59ca786546bd986f8f5f539e1dc8 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70699cb5d68145388812cf5d8354d120 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25236278e1c74016a143d2ce575739c9 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4183f0c380545ac98db1d2a55ad08a0 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70a0fc13d2cc437d950d488875f1800b : FAILURE in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2235bf4b67dc4da0b6ac69cc8f384715 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a548862d9ff4138bb1500e27ee845c9 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7645374a50cb4d29827718741a1a209e : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/474c498f416b4f4fa71ad9584a3564be : SUCCESS in 15m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0540f85bd0904cfe90bee4fbcbb78425 : SUCCESS in 10m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c012e96d1d094278a162aa890f6e7766 : SUCCESS in 17m 27s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e6dee00cfe1d4b918f95271a31e2255b : SUCCESS in 5m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/cb53c058d99a4adcb60ecd53f1b724ec : SUCCESS in 5m 07s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/60e243d3c0074b9a83b11776fd738496 : SUCCESS in 3m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1c05d7ee4f854fdfbd5aadd7f28abe1b : SUCCESS in 9m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/97c047b520084c01ac595d15443f1f6a : SUCCESS in 4m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1eca7a788f3d48128bc6d6fa144a08fd : SUCCESS in 5m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e610e6d329df40a488e28afcb09ff69c : SUCCESS in 7m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/681afa7760b24424a98dada31935303d : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71d990ac15f64cc0b0130d8afa32dece : SUCCESS in 9m 15s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/548f636ed4b448808d29e21aa767af6b : SUCCESS in 7m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0e487e004d0d41a9a8fbcce524aa494b : SUCCESS in 12m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/56d90136fba642ae85f92d89cfa6ec28 : SUCCESS in 4m 15s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9a23216504234318bec1e2f3e0d67f7e : SUCCESS in 7m 28s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/986c4d0f06544f729f0231060825788c : SUCCESS in 4m 44s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0603e5346cdc4146b64721ec3abbe2d8 : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/68c8af8a4a6e419ba202c7cfbc9c9efd : SUCCESS in 13m 05s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1c34f2dd101d424a8a97c26c7e676d1b : SUCCESS in 21m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b21ae7790aee4653b1e1210ac6b2489b : SUCCESS in 7m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f869451fef3c43b491bc0f858867b593 : SUCCESS in 11m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9c677a69b9e040f09dd1be42a7a12114 : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ddb6fcb760447c891b32bcc7fc99c85 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f646fb09213467d82a34dc0b06ea843 : SUCCESS in 22m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/aaa69594760143659135c39320a1e06a : SUCCESS in 6m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/77541feab7b649e4aab1a74ed598324c : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d66a68ab33704efeabfb4b8ca5dffcf5 : SUCCESS in 5m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c633bf3b2e61481b8daa2933318bc484 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78dd96079ef9481b87e4b96a914bd139 : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":23},{"id":"3c5a23ec2a316efbf054b51d7d5714c33dcd6c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-14 19:11:42.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"1e34e8c5690ca51705c3b4d5138eb0eb2c4be1c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 20:17:59.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad6a38b2e6449a580cd87b72820304f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d0a5cadc0047279752465712f5fd54 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b452ab12a864810a1eaca4487b95796 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bee4f96ae3ee4aa09a0282dcf14b8227 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3155c0695a2f41bab54b14ad8478e854 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3d40513c6eb4741a75a33b1cca5de6a : FAILURE in 8m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b133bc521474b479a3fa4cb40e448b5 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/405c2b502cb8470ea69465d5911720f0 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c641c2e16864b7384a4ae0e7a726b6e : SUCCESS in 12m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/daecf0f98038427d87e347d1192961c7 : SUCCESS in 16m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/295262e8fb6a4dc18c08885bd77228cd : SUCCESS in 11m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7386bd5a0826427d86875541a781be97 : SUCCESS in 16m 45s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6f749df0abe14d1dbce5522ea469d073 : SUCCESS in 6m 09s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0485d1210bb3455dbf53ed8189b72dfc : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8d8c0616f5294ada980ae1689a00495d : SUCCESS in 4m 08s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8d58b31b2a224e69b15de5ccbfcb9b72 : SUCCESS in 16m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9b9661afbf6c4203ae7f6125f3c14b8e : SUCCESS in 5m 50s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2b402b14b5fe4314857638352482caee : SUCCESS in 4m 35s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b454233547b045d6b7b984e09a853ec7 : SUCCESS in 7m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0012a2e1a9d6476f9ee4af80de9bc347 : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51b52dd673d8466f822902fc76317011 : SUCCESS in 6m 42s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6372810797ee4e4d84e3566f4bbb6592 : SUCCESS in 9m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0cbdec372aa344d5817f6e6384b08fed : SUCCESS in 10m 09s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2d28a17cf88d4724aac51b0fd6cc1ecb : SUCCESS in 3m 26s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/716e12b761034d95ad6f955a28dbeec2 : SUCCESS in 9m 39s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/87c4ca5804ae489594356ea8eb62affa : SUCCESS in 4m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/78958b836b4247eabb5c15a6b986edc3 : SUCCESS in 5m 31s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4b193486637b4047a6143015c61a4f43 : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/85af20adc2604c85bae9d086540496e0 : SUCCESS in 19m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7518ce0580a848e7804e08cb3501b89b : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0e62d7d053ac439cb996533115a4bc87 : SUCCESS in 12m 06s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/31cc98bc2e8f41cf9dc1842ee1028704 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de44753866e7484899c953c524b154a8 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e3bb3c90cff4916ac7e789259b1e9a1 : SUCCESS in 22m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/29e8e0fd11374a4e9d2981bf0e1a40f2 : SUCCESS in 4m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c7a6b929fa0045afb0842832beb116b3 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a86a4381701249d7acbf62c7b0c2f057 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1f1372535f964ef3a8a7d52024d18f2c : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded5fe00b2414ff987e527c5c1a8ae16 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":24},{"id":"37e2f9a7580c91a908b4a6279d4dfce042cfc111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-06-15 08:22:45.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"837acb632a1f9002ae8d710031b59e72d27c2d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-15 10:15:26.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/206a1178737d4728a315ac021c2ac998\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2fe01938484418873df3a7e3f3aae3 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689f4c8579a44cfa9dd1ca2bad27b619 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c95fa69aad784e92b631ba309f7b855a : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faea279e628647c08cb4a967cf54af1c : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d57d350efa0444f18728d0b7dfef612e : FAILURE in 7m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50e4fefc4f294102ba5323cd6aa1d40c : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6fdb001e48b4ae5b58923da15ad907c : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd19e65fc89e46e5932a7b88b6d98148 : FAILURE in 2m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d958019c735e4f60a60aba646869c917 : FAILURE in 3m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5fab8bcbc81a481abb463fde789f219d : SUCCESS in 11m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eaa66905eb084b25b4f44e56746f8e1d : SUCCESS in 18m 04s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2ed0722d4c7e41a9a28954892f507a82 : SUCCESS in 4m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/637dd4cd02fa496bbfc7c111e1566e86 : SUCCESS in 4m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/46b0d9bc351d4c10874e8849d37c6667 : SUCCESS in 6m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/146c70c840764e6ea74b9a07582a6ab6 : FAILURE in 9m 16s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/11f6b26a9edf42e19b360b783e2d05e7 : SUCCESS in 5m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d3d420ca8b7a481093ea2c4e033fd26f : SUCCESS in 4m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/eff8a46738084625b811a0cf4bfc357c : SUCCESS in 6m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ae9cd495763d4b1b8cee6edbfcb6febd : FAILURE in 6m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c4dce39acc94c6eb685fef73a522824 : SUCCESS in 9m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b3f2601d00dd49b987b5c36b458c3ef1 : SUCCESS in 8m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/149e2a05145b478fb7acbae8b42625b9 : FAILURE in 11m 26s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/899632f6dbc64042859075558000dfd8 : SUCCESS in 3m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1c5202253087495d90e284b1868795bb : SUCCESS in 8m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/dab96ea0c3284983990963da54c9b98b : SUCCESS in 4m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6695ebc4e5b64a04a5c5ff9e527c623f : FAILURE in 4m 18s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/58105f28b3494209bdfa8a26acf5a3df : SUCCESS in 14m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/35df3e34185a4b958ca056c76dc68ebd : SUCCESS in 20m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/22b8eada9b5d44ad9c60b7bbc2927059 : SUCCESS in 5m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/76f823ffe9254aeea221bb15ceff12e1 : FAILURE in 10m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7de71a379f0344ada7ed0cb23a528724 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c53fadcd85a8401c9a1368a7b8ea6a21 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2aca518b65ba46ebb2cef8e0ad9f80cb : FAILURE in 25m 00s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f6f351135d3d430a9e9e9594817c227e : SUCCESS in 5m 11s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/55c5c105d1df49299dfb06dd061a81c3 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac90136cab4641938b01c65543e8bcc6 : SUCCESS in 5m 45s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8bf7c047f1174c2f801e66a777e0b7e2 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdf3e479576c41d6acd96cc16f04f6f1 : SUCCESS in 1h 08m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6169335bb4a043a89ea872a45616e159 : SUCCESS in 1h 46m 55s","accounts_in_message":[],"_revision_number":25},{"id":"4295448a8ef1a91609d3dadd47c98ddf9bd80735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-17 15:34:58.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"381b0cb8a81feb6e9c91021c3b3c93eed7837d80","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-17 15:35:37.000000000","message":"Topic openstack/requirements/constraints removed","accounts_in_message":[],"_revision_number":26},{"id":"c6a8f718b855f5112cb244fc2bfa711847a47ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 16:55:31.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63e07312534b4224b1366f7b0564bdc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401a752849eb4cdbb1aca7763e25c4ca : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe0e567d6584d0395f46812060f06e3 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fec880219a6e4d168a098e6eec3f0c63 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01b9c7cb9c3a4b6ca34f90e6fad6f94e : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eecab02490bc4d4a99e9a16a4d024e93 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9bb202c12254867865b29d26363fd89 : SUCCESS in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bff0fce8325e42eca2f7b242b4b958b6 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7e4f6ccb6c1445bb923787cd17307ed : SUCCESS in 12m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fa2578db8f6e4cf6829b5ac44eaa0bb6 : SUCCESS in 17m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/199d4e10abc340039f47087bb872019f : SUCCESS in 12m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ea45e17f7ba045cbb483b192e06e1d5a : SUCCESS in 17m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1963734f284e4a9eacb67788fd3ae9e8 : SUCCESS in 5m 03s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/fc561891d0ba47b982da1d544e313ef2 : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f764871e30994a038d0c3d737fa660cb : SUCCESS in 4m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e44b77d38cca4dd5be0941f11753dd4c : SUCCESS in 10m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a0fcf7e6f3094d66b43f6549624f8ccf : SUCCESS in 5m 11s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/16727ff13cf3464e884358da30dbf0fd : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5d727bb1b11a4a99a4c64c1f38a67959 : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/19da20aa11dd4870891e5e64e17932c7 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86a1089d23d7444b86f80a63b7b37fd1 : SUCCESS in 7m 08s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3bb69fd68e914142b35b5f0ba1307faa : SUCCESS in 7m 44s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9e41d6284e52491ab26d358d64209b99 : FAILURE in 12m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a36099a31bbb4be39dd808a67083b8b2 : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a566dcac361d48248c64416c7c177e7f : SUCCESS in 9m 07s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/41f2273d5686410abdd2cb5a0d917d26 : SUCCESS in 4m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/526aefbf60784c8988c17cd55bf37dc1 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/721e9273a3e04e4fb20a7b82285647ef : SUCCESS in 12m 08s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/102b172a777847aeb1dc5d54347623b7 : SUCCESS in 24m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d1480cd59c72461198a3ef919931c91e : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e4b3df5a1942412297f47e9b4a50e90a : SUCCESS in 12m 14s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d081ffc04b094418b90aec9e1fbf1af8 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/424e5eebca18499c92d3a574d9d5e254 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e75359552fb0472b831f6eb5c9b9f0e1 : SUCCESS in 23m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/344808e800004479be0c01d9c1feff81 : SUCCESS in 4m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8726c44a91c64a358e6326e83e5c6998 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a22bd7a62a84bc2a0418583f2542ab7 : SUCCESS in 5m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cfa53a39c3f049b49bfbcc33875c7790 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ebb81ffe069449681cefe3a9e98a4ba : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":26},{"id":"2205004c01bc27e62b3e2efd7e1bae12a8bd46ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-18 06:03:44.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"e9eb7067fbcf79beb841495ac90e0b424515aa8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 06:55:36.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/322ea1270d164ad3a3c3067144530c81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44914063c6d04eb5b88547907a957a51 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c1eecc561a4761ba13e08e0c713b10 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2e5cd9ba214794ac16a295bfeef161 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbcb039e794346c89f711a9454115843 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8acf7fb1d20c4c9c95d3af4c270ec374 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ac7635bd1c442a9b356ea0957d379d0 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a984126727c24352864347d0f456f1f7 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a908a3d179e44ec9b0dd7e9a43694d0b : SUCCESS in 11m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b8e2cb6fc25b467d97f47c0547223959 : SUCCESS in 19m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1cb554e020e042afb49daf851d54b97c : SUCCESS in 11m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5394f5e6d8994f1a995beff9cc6d18d7 : SUCCESS in 17m 40s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/d020bf6c6f714854aaa1390d247a256e : SUCCESS in 5m 05s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2897b9683a854023af63d1abef9dd11c : SUCCESS in 4m 47s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/069d7d5f42054fa79d937825e2cd839c : SUCCESS in 5m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d91a39750dc145909021e13454ac04ed : SUCCESS in 8m 55s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f6eec5d4da8344019cec4cfe58813320 : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/78d8f76c06a3453e84109472a0b37632 : SUCCESS in 5m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b7f7c263d5ac438a824a77bd52092751 : SUCCESS in 7m 20s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ba1cd15a78534569b284caa870975881 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/296594dda67149728725dd164fa55a78 : SUCCESS in 6m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0d3fd9054b8c4de6a11ef543cbdc3f73 : SUCCESS in 10m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f6e2a0aab59e4e92b134954aa4198453 : FAILURE in 10m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/922c0c85d64e468482aa2a51fd643a87 : SUCCESS in 3m 49s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/81d91489cef94a8db7558a7d6a74bc17 : SUCCESS in 8m 08s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b71fee62a74a4064a6bcc49c4e590f80 : SUCCESS in 4m 29s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/786a37cf392d448788fa33279e15ef8f : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/76061423617e4f33b3b495cc27ff75e9 : SUCCESS in 9m 32s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e4cd140d125946db9adf3160a5c83e4b : SUCCESS in 19m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0614e547d5e64c14b0bdf0fb93998fa9 : SUCCESS in 6m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/809fcac713b9457ba4b3918070f61b97 : SUCCESS in 11m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/acbe78485915474991449262bc090f99 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6cb644c63d35471092e93124d9e28b45 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06dd6657726e459882a4c63f96b552cb : SUCCESS in 21m 44s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/87ba115cfde74949bd35098f420e6f74 : SUCCESS in 4m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1ffc79e663dd4bc9906c136042861910 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d47594f2a3f145968451e4af55c08865 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/df0e64de543742a4ad017c94f732bb56 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37626e5546654d7d8543b3e9b99cb58b : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":27},{"id":"a60edd2059684a44864a77463526387690295316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-22 21:31:47.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"1985db0925471e4970e3ce318f5c48fc139b71a7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-22 21:57:27.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"1e17d5831921879fa05f3a72a7fcdeb2bbb1e174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-22 22:57:44.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61ee6c705d504e4199381aa29eaaf64d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acdcc800dd24c1aa5d17f3eaa2092d9 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abaf673a374c4671b1b5b88cc2d24efb : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dab75dacd14148b085636c4769f0a932 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/116b0844f2bc43439ee3d22f6a28f462 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/016a6a7c558c439f99f13dc3f813a2e4 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/361836ea88a24934ac63ec63edc4d680 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/901f205325784b8b86c5b7fc0f527274 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa33e7faba2a429d9ac77637d521be55 : SUCCESS in 9m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af73c1ff59de45eb8683b14c3e7bd3a4 : SUCCESS in 13m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce908634bd1e43a4af5ac914d882d4a0 : SUCCESS in 9m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/203d2f64070c47ce90cf7c5fcb9dd307 : SUCCESS in 16m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c294ff3201ac44e8880902a4bb5af56d : SUCCESS in 27m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2e6b1f3e95cc43abb8b19d647f4cb326 : SUCCESS in 4m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/9b975b3ece8d4e6a92be22cfe8f7d631 : SUCCESS in 5m 21s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/95fca75d40f345bf9949127b2d56d554 : SUCCESS in 4m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8ee747bdbf324d50ae545f9d660f9b6b : SUCCESS in 10m 05s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/433a07188c14483eaf494e6208dd14e5 : SUCCESS in 5m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dc7168ef92f24540b919858ca92ad5d8 : SUCCESS in 5m 14s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cff6919c34bb40868cd33d1e441eeac0 : SUCCESS in 6m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/df423e30f35a4499aa33c90e3954b325 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7305ee9c840641a79ca84c095efd8daf : SUCCESS in 6m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/25eefbe0c79d41c1983323fb4def1968 : SUCCESS in 7m 44s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/319555a7f7b84e5d8daaafd7d2a2bab6 : FAILURE in 10m 27s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/386a85c47c2e46429d2b350a03dac6fa : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b14676bec3b244219642334314a46602 : SUCCESS in 7m 32s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/05883d975bcf4e0787aedefe919d13f8 : SUCCESS in 4m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d832725dca404ed68f9dcf5d92844d3d : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/93d943c1d27f4b51b10e12010fb08aa8 : SUCCESS in 11m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/52227082420748fb8e32da7094934237 : SUCCESS in 19m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1ab967eef16e4855bb12859ab504fa87 : SUCCESS in 5m 57s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8e36ee6646fb4c77b2d34e3094d9de18 : SUCCESS in 14m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2583d8b4b55144108b7ba58405c26d22 : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae01ce8e7280428a802a41706757d73d : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f72113e5281348b4b1b84a2841e8e327 : SUCCESS in 23m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/379dea5ab83b4823bc47630a229fa43e : SUCCESS in 4m 21s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/587e5ed8711b4f629df6108cf29bab95 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5cbf0980e1fe4bcfb44c876602e347ef : SUCCESS in 5m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9bb93d5baa4b4e17a17b9829ebe07ef7 : SUCCESS in 9m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5d79c0c5a2a40f28a28a623cabc369b : SUCCESS in 1h 01m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbbd9f7f7ef742389337264d6094992f : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":28},{"id":"48a1606a2f82b692c3653197827ba0c8f4101c5b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-22 23:46:09.000000000","message":"Patch Set 28:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969565ecf5cb4e448255fd0c77a08ebd\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5ff51cb942546558ae22aed60e8ab06 : SUCCESS in 1h 48m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/a173c921354544aea1605cba7fca4433 : SUCCESS in 1h 32m 20s\n- requirements-tox-py38-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/04b0c7a2e8f949fb88ddeee93184ab02 : SUCCESS in 16m 12s\n- requirements-tox-py39-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/14a862505e994bd49dc86c4e85c33353 : SUCCESS in 13m 59s\n- requirements-tox-py310-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/59750eaba5cc45c698f160c998f494bb : SUCCESS in 13m 45s\n- requirements-tox-py311-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/6d552fd37bf0484e8c22891eb2e1a48e : SUCCESS in 15m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87bc8dc6f3cb478db7128421b48308a4 : SUCCESS in 1h 08m 59s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/033d2adede994e218af92dc9dbb2b7c9 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":28},{"id":"b0df287b7c3d9086b44c207dfe2851673ba5b6d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 13:15:35.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"7df1f10f844f3e2e18814603fa2bf615bd5c3251","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 13:16:22.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"b28a1e8f14e8601b2feeac6da9d0acf009c550d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 14:44:48.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2fa7edbf20f4454b36fbeea7ad18e33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9d3454fa9e46d1a2e0d0d9076cc5c4 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa4c06988e143bdb265599f665e9bc6 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29837ccce5174fd18863bf3e691ed4a4 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/925b82d07b574d51a18a3f80c974609a : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1a70716fa9b45658ec5f26ad2755778 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac607f2b56b74acea89949591b2470c8 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e916e6242cb144b2b88cb355e0e7a4ed : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a8960155763450f8eb0e9e9ab4c8688 : SUCCESS in 12m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e8c6e132ec4443ab9b221777c3f707c5 : SUCCESS in 13m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8f40cb9ffb2545d48d150288fdb10865 : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5e0440c534d544c2a46ef36b1ea6434a : SUCCESS in 16m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1f682f01e7df44dca67f7e8e7eab42d2 : SUCCESS in 24m 50s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8e8601c4fb07473b85e6e747c3d5dd8f : SUCCESS in 4m 32s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0a6104f423f54339b5d74298f9dee80f : SUCCESS in 6m 40s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/411575e48ae24162b60cb97d6797e317 : SUCCESS in 4m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1cd75efe10af4a97950d9d424077680e : SUCCESS in 10m 51s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4a4eccd4e1d64f89a4c7d14c97c44494 : SUCCESS in 4m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/49389a7b1763461aa7560b6d375e7d36 : SUCCESS in 5m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2a4ad5100030422db3587f727034f728 : SUCCESS in 8m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2cfa7fe0d58c466c815c9cb633aba373 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8682b44f91f746769f98ea311a069a7d : SUCCESS in 6m 42s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ef68348a2cad49798710ffd870d795ea : SUCCESS in 10m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/29b884fb886b4602a2ceab4d9f36b04d : SUCCESS in 9m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c4fe50684f5b4b3897fc53a70a824e62 : SUCCESS in 5m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0b0af90c31724156bae8d821d70a0544 : SUCCESS in 7m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/2e371036829b4c47aecaf742deeace73 : SUCCESS in 4m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/815d33d80e5b4770ae1d4bceecd8ae29 : FAILURE in 3m 53s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3d8671901fda4bbebb50c01662062d64 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ae4c9d94698e471398f30a496a454023 : SUCCESS in 17m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71e056c7143e4dfbb0d2b6a90a1c542f : SUCCESS in 6m 14s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/35feead6099c480b9763c4c46cb0bf9d : SUCCESS in 11m 27s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/98cea10d09b048be948376a08e534536 : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2048e47689148cf89250a8a91d711f3 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e13543a88134718b08fcddb9d5e79a3 : SUCCESS in 23m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b47f8afa7b484e5e908d59f40c3db12a : SUCCESS in 5m 31s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/35f1205f95f344f3b9aa7ebd3959459c : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfd56491a6cf401291bbda0fe3e34ead : SUCCESS in 5m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3f84f2a87e2643249905f28044a2999a : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c29fb12ac6475c8c7d118ba89940eb : SUCCESS in 43m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a621a50096204154ab647bd207a1ece2 : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":29},{"id":"3a17b43ea7e52da3f33e2c1540d7dd1acca90f44","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 15:07:25.000000000","message":"Patch Set 29:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b93520a37651492bbe8f7c4a3210124e\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/834a79291fbf4f4584541b378cc42f91 : SUCCESS in 1h 33m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/6d3673f4456746a9a64bc14def1ae027 : SUCCESS in 1h 32m 22s\n- requirements-tox-py38-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/94d1f26819d24475bf40eb53c110a4db : SUCCESS in 15m 26s\n- requirements-tox-py39-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/7174d11eb743474493f48baf7e77036b : SUCCESS in 16m 35s\n- requirements-tox-py310-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/5ffc92d5ca1847af8bd8fdb71e67443c : SUCCESS in 16m 12s\n- requirements-tox-py311-check-uc-no-wheels https://zuul.opendev.org/t/openstack/build/bc1b43d4b7794d2fa75271073e9f9406 : SUCCESS in 15m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ecbc0732dbc4c699ac4b2b55a0d2388 : SUCCESS in 1h 44m 59s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/82ab456ad5e24cb7bc87ecc6b0feab7b : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":29},{"id":"6bbac90f7ceeef507e130571d6bb067f8180c744","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-08 16:52:35.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1\n\nship it","accounts_in_message":[],"_revision_number":29},{"id":"4c5023d4fab2ece3f67c580c4204816177db7a17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 16:53:20.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"702ef8ae49f9c8957317c356ebe6ede5ec78619b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 19:37:04.000000000","message":"Patch Set 29: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46867ea2be543f380299dfc052da280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b6317bc4a94495a59a2e2c59248ae5 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0337467a97aa47a9b9e0792fb2c8d830 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9429c6a146b48838fbfc1f47d48a611 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0507d34e739463e876c07e5fba26acc : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ecab247b901740a69e0a81db674f05d2 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c69344e7e59349f5883b268b76994540 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5d140cc7b14402ab703bcd628582909 : SUCCESS in 10m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/edcae8f58fa84298b00bc17e6e6ccdf6 : SUCCESS in 16m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f6bf27240e4149148b0a670f0fb406e0 : SUCCESS in 10m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/211284dff84d44c88aa96fdab0ea1f5e : SUCCESS in 16m 32s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b27223d786f24c42a94a77dde9bf5445 : SUCCESS in 11m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5d67bfe407c5400883acdebe6bc80bd4 : SUCCESS in 6m 39s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a0395437c4b44d01bc95685abaf6be0a : FAILURE in 2m 51s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/385e2146832247bebd793b5f83e5eb4f : SUCCESS in 7m 01s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/91d2a702b6c1476db7c6702813e26584 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32a249bad8c44a6ca86869fa0432b370 : SUCCESS in 5m 45s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9a1a82316f594545a1bc509b3008eae0 : SUCCESS in 8m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a86637f9d977400581026de8564520ef : SUCCESS in 11m 46s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/342a6b18b433457e973efa2b97013a8b : SUCCESS in 3m 41s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/347f4db0f8634da5a67ca77f40f436af : SUCCESS in 9m 00s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c48ef7dc087a4b1caacc6064afda2a72 : SUCCESS in 20m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/93661708515744a8b40461764c73731b : SUCCESS in 8m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b818f72a12fb4fefae22ffa0a111b0f9 : FAILURE in 3m 10s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/eb85c3b2dace408c9f943ebc61d4e3f6 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89b6fb8178e947e986070ea890b6f503 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c36178eecbf548a4a75ddd9bdd5b1aa1 : SUCCESS in 24m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f5e1d8bba7174368a51e488d4e6fa580 : SUCCESS in 4m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3c91ec3e21d647e5963e4c470c457eb6 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d47c304e8aec4d58b3f962657841f0fa : SUCCESS in 6m 19s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/74216a6bbaff446689b03b1086499809 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23a2028b05314f41b6702b030778a07f : SUCCESS in 48m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b54e61c488fc4a29a2d70816d5dcff35 : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":29},{"id":"1259a206779521bad353d04306094946da0bec9f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-08 20:01:00.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"2b301f073907c7b73c5788b5d4350f5a8f3a51df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 21:16:41.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73bea8bc9cf94d678cc91acde1cc7224\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da34b478b4c4d508ff83367e9b2e018 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a05fa59e91437cae6bc5694ddd53ca : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4add08ec456a47dbbc14a88a8964eb85 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebd02d9e63a24361a5734701e2aa1265 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e91d591c072d445ca2c0ff4beb4e0a5d : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65b54f922c2840ffbb3898ae1371c579 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c39651daf5d84abd8ccc05b68ac47eb6 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d87ef72de48346e499bb97af7e495d90 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7322d51fa497462d8fadcb873b9367e6 : SUCCESS in 14m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/388e2437593d4094ba2e93ed071a24cf : SUCCESS in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c4f9054644614f06bf8e8f4b8d8b8b6b : SUCCESS in 17m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4c461d5768554c5f8f33c3c7dc4ccf5c : SUCCESS in 27m 58s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/344dde87eea14891abddb24235f9d606 : SUCCESS in 4m 39s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/9a57d802b21f46619b22407e4eef03c8 : SUCCESS in 4m 54s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d88f00b41e80476e87824131ec408762 : SUCCESS in 3m 38s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b547cb8636bf47d4b01c769345f2b0f7 : SUCCESS in 10m 24s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/db9cec7467c54612a075cce6e1c4d3ce : SUCCESS in 4m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b114bc2441af4be8bd609e0a57e999db : SUCCESS in 5m 04s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/970444ea5eb14dd7acb2b424489cdaab : SUCCESS in 7m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/26308b5e1ec3454585f28e1d5abe5f9a : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13526bc2591c48518c5bb83f8d36f174 : SUCCESS in 6m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/71c9532692db4eaf8a595355ec231e23 : SUCCESS in 6m 31s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/65d8fdb0357e4de0b02d1364e64bd1eb : SUCCESS in 9m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e2a53a9842d5401faac92635418266dd : SUCCESS in 3m 36s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6f07c758eb6740d597e996be9cf38e44 : SUCCESS in 8m 20s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bc5ed86a429e4df4ac38ff1c07f86b3d : SUCCESS in 4m 32s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b25e67aede52433bbc85ece8b0bcbde5 : FAILURE in 4m 07s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/08145d71362943f1bd28efbadd61d035 : SUCCESS in 11m 38s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c724e5e825134450905223b56b27840e : SUCCESS in 19m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4248918a71bc4e6e82c7ccb438f0dcf9 : SUCCESS in 6m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6c04a43f40b44c67bba017d941fef486 : SUCCESS in 11m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c91c3412bebe494ebdb2a58c67c15470 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/377c113ea64140e4ade7af72ea11d614 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0b245c38dd04966932cca8d18556413 : SUCCESS in 23m 17s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6a0eaf713d3943169a6e170a12b77c6d : SUCCESS in 4m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/56d30bed9bf0410fa1b824c92133400f : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c87b029996d14ceb9cd9892b71ab45c6 : SUCCESS in 5m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c609c0f889f84a0daba2a2a6998ebaa4 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84adc4c5dde64f1c9e051b806c29fbdd : SUCCESS in 55m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc4bf1d410dc416fa1df9192bbf93e42 : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":29},{"id":"60e37697201de80de84f771835f752293ce9b1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 21:17:18.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"0299f6f3e304590b5f477d0aa6499f9423e72265","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 22:57:08.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6a0b6556c94f5499a35907ca21bb5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267107df9d534e779c208a6e28009721 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39bcf8907154884a9cd544a06e98f02 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6625950965214c77a2f159d3fad46c3e : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd7efa6ac2a3477abcd972d7de555e2d : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9413169146f4531a90487af2647d5c5 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bf853cc1cfc4092bb1636d6d2f14155 : SUCCESS in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b96c61bd7704dc486def6a8b874fcb6 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1e282888244844ceb5012bc82a942ee1 : SUCCESS in 17m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f5bc7885e644f79bca3e6de3a93b413 : SUCCESS in 7m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/34dcd2b008084366a45f5eb4d4b3ad6a : SUCCESS in 12m 27s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/293a61e45f0343868249d1e0a45fd77a : SUCCESS in 11m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/22118b96ad5d4cb0955b28a5edbdbdb7 : SUCCESS in 4m 46s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8dcd5f6f1a154d7d999a50a6f59a8213 : SUCCESS in 4m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ed2a4a0d8e9d43a7acf2b71aa1b959da : SUCCESS in 7m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d117448197da45d688d14ba018a6870b : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49da8708776147eaa4bf897f5ba5852a : SUCCESS in 5m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9ef51892badc4b69b280ded83179dcfe : SUCCESS in 6m 35s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/3e48bacf31cb44c6bc5716fe08acc6b9 : SUCCESS in 8m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/53f0466dc3f64cfeb254e14e8cf8bb6f : SUCCESS in 4m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/500cb4ff03604f8e8d52908cf5050ee2 : SUCCESS in 7m 25s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/19e9ad941e1b496ab1f52828801db174 : SUCCESS in 18m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8ae669f8dc214765b49ecd5352e35a10 : SUCCESS in 7m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/59bc8490f6f24830a731061f455116eb : SUCCESS in 12m 55s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9619274ee6a24266b9788a878c4a840b : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96314da217b3454e9b848cce29823e94 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92fa5295e834487b80edf2a86e3571f9 : SUCCESS in 22m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ccb31360a29c4f91b27295bfb6e0df4f : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f74a9c37973b435fab32819647455816 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/464670e7abde49fdbbc1fbbc8431c97e : SUCCESS in 8m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/235531e7795847fa937572e4db651097 : SUCCESS in 20m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cee1adb331e84e05a737e7565b41b2b2 : SUCCESS in 53m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edf9caf05fe84c57ac5da0efb2cfc0ce : SUCCESS in 1h 39m 45s","accounts_in_message":[],"_revision_number":29},{"id":"430d99e6467210e083697783448d281e9be49d3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 22:57:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":29},{"id":"6b3dc3ceb3d4bd73dd73ba3ac70fcc8f4adedad3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 22:58:05.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6f8aced00c411289e530f0ffa9d8f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f380051a42451b8eabdcb666a08438 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"d0c6e1b2210a000b82da36ba14d7c6b108131fd7","revisions":{"46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26":{"kind":"REWORK","_number":1,"created":"2024-02-24 08:12:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/1"}}},"commit":{"parents":[{"commit":"94fa4434f50558964cea0b7271eb816ff9d19253","subject":"Merge \"update constraint for openstacksdk to new release 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94fa4434f50558964cea0b7271eb816ff9d19253"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-24 08:12:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-24 08:12:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46f186d1cf1de4c1f6a66ecfa3a1dbe5c7b34e26"}]},"branch":"refs/heads/master"},"824f8de088599234de86f259444de75f2daa6151":{"kind":"REWORK","_number":2,"created":"2024-03-02 08:17:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/2"}}},"commit":{"parents":[{"commit":"bab2a0f9eb63b73757b5c3e06790d8d0c7e73ced","subject":"Merge \"update constraint for python-manilaclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bab2a0f9eb63b73757b5c3e06790d8d0c7e73ced"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-02 08:16:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-02 08:16:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/824f8de088599234de86f259444de75f2daa6151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/824f8de088599234de86f259444de75f2daa6151"}]},"branch":"refs/heads/master"},"fa1ad57de8d480b21a9a758d38e62ec3b8e2574d":{"kind":"REWORK","_number":3,"created":"2024-03-09 08:18:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/3"}}},"commit":{"parents":[{"commit":"450cfd75b8c95c1b413429c5ad5f85854bc142bf","subject":"Merge \"Add cross-magnum job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/450cfd75b8c95c1b413429c5ad5f85854bc142bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-09 08:18:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-09 08:18:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa1ad57de8d480b21a9a758d38e62ec3b8e2574d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa1ad57de8d480b21a9a758d38e62ec3b8e2574d"}]},"branch":"refs/heads/master"},"fed4d1284a51230aed1e58eb29521dd9a4a74685":{"kind":"REWORK","_number":4,"created":"2024-03-16 08:22:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/4"}}},"commit":{"parents":[{"commit":"99f60d6329d545215561b198c680e79daf690192","subject":"update constraint for yaql to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99f60d6329d545215561b198c680e79daf690192"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-16 08:22:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-16 08:22:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fed4d1284a51230aed1e58eb29521dd9a4a74685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fed4d1284a51230aed1e58eb29521dd9a4a74685"}]},"branch":"refs/heads/master"},"91ed29ca5c0cb552d7da83a25cc3699aea7ee75d":{"kind":"REWORK","_number":5,"created":"2024-03-23 08:11:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/5"}}},"commit":{"parents":[{"commit":"be28fc2a88b8edb7dacde28564905dce0bdc7be2","subject":"Merge \"update constraint for python-openstackclient to new release 6.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be28fc2a88b8edb7dacde28564905dce0bdc7be2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-23 08:11:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-23 08:11:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91ed29ca5c0cb552d7da83a25cc3699aea7ee75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91ed29ca5c0cb552d7da83a25cc3699aea7ee75d"}]},"branch":"refs/heads/master"},"b353a97ed17e000909e119bcb18234bcf7ff09f0":{"kind":"REWORK","_number":6,"created":"2024-03-30 08:28:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/6"}}},"commit":{"parents":[{"commit":"93a3e51432519a14a378a729554b5273675e945b","subject":"Merge \"update constraint for oslo.messaging to new release 14.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93a3e51432519a14a378a729554b5273675e945b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-30 08:28:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-30 08:28:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b353a97ed17e000909e119bcb18234bcf7ff09f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b353a97ed17e000909e119bcb18234bcf7ff09f0"}]},"branch":"refs/heads/master"},"fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29":{"kind":"REWORK","_number":7,"created":"2024-04-05 08:55:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/7"}}},"commit":{"parents":[{"commit":"cb43fae47eff6b5fb36ca07fa344ebb87a88d903","subject":"Merge \"update constraint for tap-as-a-service to new release 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb43fae47eff6b5fb36ca07fa344ebb87a88d903"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-30 08:28:56.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-05 08:55:24.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fde6c7aa6d5daddab3e88ba4c861b2be46fdeb29"}]},"branch":"refs/heads/master"},"3340d34eda88f5f32e9cae215fb500821ce8ccc6":{"kind":"REWORK","_number":8,"created":"2024-04-06 08:19:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/8"}}},"commit":{"parents":[{"commit":"b5b63fb81989d23346145a0b1e33db71619fea94","subject":"Partial Revert \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5b63fb81989d23346145a0b1e33db71619fea94"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-06 08:19:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-06 08:19:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3340d34eda88f5f32e9cae215fb500821ce8ccc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3340d34eda88f5f32e9cae215fb500821ce8ccc6"}]},"branch":"refs/heads/master"},"0b8398af74f9a83c1f330a8302ab3e5d39b2c0c0":{"kind":"REWORK","_number":9,"created":"2024-04-13 08:18:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/9"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-13 08:18:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-13 08:18:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b8398af74f9a83c1f330a8302ab3e5d39b2c0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b8398af74f9a83c1f330a8302ab3e5d39b2c0c0"}]},"branch":"refs/heads/master"},"59ef7717168e17166440951af54d6ef11676350a":{"kind":"REWORK","_number":10,"created":"2024-04-20 08:19:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/10"}}},"commit":{"parents":[{"commit":"b2f1bb17383c044c47d254b10e141a62de969b7a","subject":"Merge \"update constraint for os-collect-config to new release 14.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2f1bb17383c044c47d254b10e141a62de969b7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-20 08:18:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-20 08:18:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59ef7717168e17166440951af54d6ef11676350a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59ef7717168e17166440951af54d6ef11676350a"}]},"branch":"refs/heads/master"},"32f354df8ce0db3a36e2a1ad0dc505cb10297b54":{"kind":"REWORK","_number":11,"created":"2024-04-20 08:54:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/11"}}},"commit":{"parents":[{"commit":"92c8b4c52b3160df1db1768e8891fc19eb704424","subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92c8b4c52b3160df1db1768e8891fc19eb704424"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-20 08:18:58.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-20 08:36:54.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32f354df8ce0db3a36e2a1ad0dc505cb10297b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32f354df8ce0db3a36e2a1ad0dc505cb10297b54"}]},"branch":"refs/heads/master"},"2bc97ba7892330c8e6e61fea8a08a416fd51969f":{"kind":"REWORK","_number":12,"created":"2024-04-27 08:18:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/12"}}},"commit":{"parents":[{"commit":"5c1abc456deed2db69dbf4470958932e017961dd","subject":"Merge \"update constraint for os-apply-config to new release 14.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1abc456deed2db69dbf4470958932e017961dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-27 08:18:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-27 08:18:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bc97ba7892330c8e6e61fea8a08a416fd51969f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bc97ba7892330c8e6e61fea8a08a416fd51969f"}]},"branch":"refs/heads/master"},"3679bbf59f7b842aa84639b3c3d2d1ecf9b0bd50":{"kind":"REWORK","_number":13,"created":"2024-05-04 08:14:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/13"}}},"commit":{"parents":[{"commit":"1e07b8cf6e8a84d18987d1ac4ba7b0fba1ebdca7","subject":"Merge \"Remove pyinotify\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e07b8cf6e8a84d18987d1ac4ba7b0fba1ebdca7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-04 08:14:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-04 08:14:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3679bbf59f7b842aa84639b3c3d2d1ecf9b0bd50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3679bbf59f7b842aa84639b3c3d2d1ecf9b0bd50"}]},"branch":"refs/heads/master"},"d990eb6f70b28c3ae2355b3b29af22186dbe43f2":{"kind":"REWORK","_number":14,"created":"2024-05-11 08:17:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/14"}}},"commit":{"parents":[{"commit":"091035a0319fbca632e5ae0fcc4f249fa6f5792f","subject":"Retire TripleO: remove the TripleO deliverable from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/091035a0319fbca632e5ae0fcc4f249fa6f5792f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-11 08:17:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-11 08:17:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d990eb6f70b28c3ae2355b3b29af22186dbe43f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d990eb6f70b28c3ae2355b3b29af22186dbe43f2"}]},"branch":"refs/heads/master"},"25ea34a8eee299511a30b78c6ad7bf6596469afc":{"kind":"REWORK","_number":15,"created":"2024-05-15 12:23:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/15"}}},"commit":{"parents":[{"commit":"a8ef1881af9cdb048dc0c5e3f814aaed80efd7a3","subject":"Merge \"Bump testrepository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8ef1881af9cdb048dc0c5e3f814aaed80efd7a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-11 08:17:16.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-15 12:16:14.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25ea34a8eee299511a30b78c6ad7bf6596469afc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25ea34a8eee299511a30b78c6ad7bf6596469afc"}]},"branch":"refs/heads/master"},"656c21f5d7ecf9674a6238e1a026068840a15d94":{"kind":"REWORK","_number":16,"created":"2024-05-18 08:12:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/16"}}},"commit":{"parents":[{"commit":"54a294000965c829f8c348a5ef086cd260d173c3","subject":"Merge \"Remove the Murano deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54a294000965c829f8c348a5ef086cd260d173c3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/656c21f5d7ecf9674a6238e1a026068840a15d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/656c21f5d7ecf9674a6238e1a026068840a15d94"}]},"branch":"refs/heads/master"},"1da3d304c0a2b31bb26a459d4403469eb6901e84":{"kind":"REWORK","_number":17,"created":"2024-05-23 08:47:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/17"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-23 08:44:25.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1da3d304c0a2b31bb26a459d4403469eb6901e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1da3d304c0a2b31bb26a459d4403469eb6901e84"}]},"branch":"refs/heads/master"},"128b67fa08fca35922cd3199c62cf0cd321c72ab":{"kind":"REWORK","_number":18,"created":"2024-05-23 09:39:11.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/18"}}},"commit":{"parents":[{"commit":"26e59920a7bbdc47118845704322dcd32510a2d9","subject":"Remove ansible-runner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26e59920a7bbdc47118845704322dcd32510a2d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-23 09:38:55.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/128b67fa08fca35922cd3199c62cf0cd321c72ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/128b67fa08fca35922cd3199c62cf0cd321c72ab"}]},"branch":"refs/heads/master"},"28c8ea1db75bff325c6691b1e647cd268ad85e80":{"kind":"REWORK","_number":19,"created":"2024-05-25 08:33:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/19"}}},"commit":{"parents":[{"commit":"f9099ed1b695dfcdcd9fbc682f3225e68b9b0cd1","subject":"Merge \"update constraint for oslo.messaging to new release 14.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9099ed1b695dfcdcd9fbc682f3225e68b9b0cd1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-25 08:33:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-25 08:33:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28c8ea1db75bff325c6691b1e647cd268ad85e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28c8ea1db75bff325c6691b1e647cd268ad85e80"}]},"branch":"refs/heads/master"},"2c389e69716f50c78a8dfa02161b9fda1589ee82":{"kind":"REWORK","_number":20,"created":"2024-05-26 20:02:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/20"}}},"commit":{"parents":[{"commit":"fd1228c681d745ec6ace3b620cb419cc087c0d31","subject":"Merge \"update constraint for os-brick to new release 6.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd1228c681d745ec6ace3b620cb419cc087c0d31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-25 08:33:38.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-26 19:55:02.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c389e69716f50c78a8dfa02161b9fda1589ee82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c389e69716f50c78a8dfa02161b9fda1589ee82"}]},"branch":"refs/heads/master"},"18a16692f5ad74d1cc18fb420ff6615e9f5d2071":{"kind":"REWORK","_number":21,"created":"2024-06-01 08:29:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/21"}}},"commit":{"parents":[{"commit":"fd1228c681d745ec6ace3b620cb419cc087c0d31","subject":"Merge \"update constraint for os-brick to new release 6.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd1228c681d745ec6ace3b620cb419cc087c0d31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-01 08:29:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-01 08:29:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18a16692f5ad74d1cc18fb420ff6615e9f5d2071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18a16692f5ad74d1cc18fb420ff6615e9f5d2071"}]},"branch":"refs/heads/master"},"15f5dc3483530da0c669d1a755b2991e92249b3e":{"kind":"REWORK","_number":22,"created":"2024-06-08 08:17:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/22"}}},"commit":{"parents":[{"commit":"c1131bbdcfbf778816dd5bf838cdae9d84c17674","subject":"update constraint for neutron-lib to new release 3.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1131bbdcfbf778816dd5bf838cdae9d84c17674"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-08 08:17:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-08 08:17:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15f5dc3483530da0c669d1a755b2991e92249b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15f5dc3483530da0c669d1a755b2991e92249b3e"}]},"branch":"refs/heads/master"},"42d1087290056141859251526d569e4447c23863":{"kind":"REWORK","_number":23,"created":"2024-06-12 17:52:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/23"}}},"commit":{"parents":[{"commit":"450d0553b0c75649cab08af7bc51ed4255886e78","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/450d0553b0c75649cab08af7bc51ed4255886e78"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-08 08:17:33.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-12 17:46:03.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42d1087290056141859251526d569e4447c23863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42d1087290056141859251526d569e4447c23863"}]},"branch":"refs/heads/master"},"247a883bfb316d9b00c8b852ebed27c719d6867b":{"kind":"REWORK","_number":24,"created":"2024-06-14 19:11:42.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/24"}}},"commit":{"parents":[{"commit":"3b66a9865b02a17b7633385ed1fcbbe1ddd9544f","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b66a9865b02a17b7633385ed1fcbbe1ddd9544f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-08 08:17:33.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-14 19:10:13.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/247a883bfb316d9b00c8b852ebed27c719d6867b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/247a883bfb316d9b00c8b852ebed27c719d6867b"}]},"branch":"refs/heads/master"},"c67b7f1d1f3096c60c0c66097f1860bc8d408643":{"kind":"REWORK","_number":25,"created":"2024-06-15 08:22:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/910040/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/25"}}},"commit":{"parents":[{"commit":"fe3cb64fc8ccbc62d2f61c1ddad639f241eac5fd","subject":"Merge \"Update pysnmp-lextudio version and dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3cb64fc8ccbc62d2f61c1ddad639f241eac5fd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-15 08:22:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-15 08:22:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c67b7f1d1f3096c60c0c66097f1860bc8d408643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c67b7f1d1f3096c60c0c66097f1860bc8d408643"}]},"branch":"refs/heads/master"},"120c6c69f8e525969ef11279857aed2a1c285309":{"kind":"REWORK","_number":26,"created":"2024-06-17 15:34:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/26"}}},"commit":{"parents":[{"commit":"3b66a9865b02a17b7633385ed1fcbbe1ddd9544f","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b66a9865b02a17b7633385ed1fcbbe1ddd9544f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-15 08:22:44.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-17 15:33:30.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/120c6c69f8e525969ef11279857aed2a1c285309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/120c6c69f8e525969ef11279857aed2a1c285309"}]},"branch":"refs/heads/master"},"8cf77496806fae10f12a43e4bd8edbd0661e14c8":{"kind":"REWORK","_number":27,"created":"2024-06-18 06:03:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/27"}}},"commit":{"parents":[{"commit":"af402e5ab515eb716fc6e77e1d485d005a2f112a","subject":"Drop pylxd from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af402e5ab515eb716fc6e77e1d485d005a2f112a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-15 08:22:44.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-18 06:03:24.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cf77496806fae10f12a43e4bd8edbd0661e14c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cf77496806fae10f12a43e4bd8edbd0661e14c8"}]},"branch":"refs/heads/master"},"a047db1e747efca219f516fb46b122925ac9c133":{"kind":"REWORK","_number":28,"created":"2024-06-22 21:31:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/28"}}},"commit":{"parents":[{"commit":"a7bf31fe5c7cadab519f36b6be43dfedf0d8951c","subject":"Merge \"Fix pkg_resources exception handling for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7bf31fe5c7cadab519f36b6be43dfedf0d8951c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-06-15 08:22:44.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-22 21:30:24.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a047db1e747efca219f516fb46b122925ac9c133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a047db1e747efca219f516fb46b122925ac9c133"}]},"branch":"refs/heads/master"},"d0c6e1b2210a000b82da36ba14d7c6b108131fd7":{"kind":"REWORK","_number":29,"created":"2024-07-08 13:15:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/910040/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/910040/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/910040/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/910040/29"}}},"commit":{"parents":[{"commit":"516f66364c0671e541bfc731682879353d008adb","subject":"Merge \"update constraint for oslo.versionedobjects to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/516f66364c0671e541bfc731682879353d008adb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-06 08:17:02.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-08 13:14:23.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes added to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nAdd version specific pins for updated requirements:\n\n- voluptous\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- tenacity\n- jsonschema\n- pytest\n\nChange-Id: I72a777f3a553ba720f80fd66ad7a488b55de51f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c6e1b2210a000b82da36ba14d7c6b108131fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c6e1b2210a000b82da36ba14d7c6b108131fd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
