)]}'
{"id":"openstack%2Frequirements~910143","triplet_id":"openstack%2Frequirements~master~I30072518c9a4fa9e04bbd55477383b150de59ba8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I30072518c9a4fa9e04bbd55477383b150de59ba8","subject":"update constraint for oslo.cache to new release 3.7.0","status":"MERGED","created":"2024-02-26 07:43:29.000000000","updated":"2024-02-26 19:11:50.000000000","submitted":"2024-02-26 19:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910143-new-release","meta_rev_id":"961a224363eb26289e9d753d051ae64ac9bfee89","_number":910143,"virtual_id_number":910143,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-26 19:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-26 16:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-02-26 16:40:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 09:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 09:57:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-26 16:40:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2120b75defeb0b2d74bb18f9eb8f7f1a5b2d5c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-26 07:43:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a8a77cfd112a63a3b5cbfa41b917c4a6b0a4261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658b385391ee4425a992767b7b9b130f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d142eec8e89745d895fd5b3e71e4b1b1 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85942c05acf48e195053cc405104c49 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a823c3253ee74a57bda11352c59c43c7 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd4cc83ee8934960b5d37125ef4c5760 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e41eb56647f47c69d6859f391096d9b : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97ab1d069c354817a0300adf6d3ba3bf : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2deb8ed1bdd2436699ecebae4826daf7 : SUCCESS in 9m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec3ae16f24ec4ab192b0151f64871186 : SUCCESS in 14m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8505433ef1514598b3c78e75c1414dad : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eea7780238624b3a9ee8ea72a3d67c8f : SUCCESS in 17m 24s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a7162020adb7414888e823b9c4228216 : SUCCESS in 5m 00s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/16b63193d0e14498a0aba5940c40ba37 : SUCCESS in 5m 00s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/59098dde65d746639fa43344b19dd7b1 : SUCCESS in 4m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/18f3c96973cb491bac85a109f4c3f513 : SUCCESS in 10m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/18f80a61dac54ca699890bb548b2d5fa : SUCCESS in 4m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a69fd5b90e1b4f6a8da182264043f99c : SUCCESS in 6m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/eccf0597272441148631e804974a8122 : SUCCESS in 7m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/42bc6346e9fc49d7918c1a2d63b45fb1 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec173280d68948a78f2abcd3dc42d601 : SUCCESS in 8m 41s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f67b9ba4ee114a50ae3ea28c423d1d81 : SUCCESS in 6m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9655e169e5834c00920306bc7c8dcedc : SUCCESS in 7m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c5168fd694b5495bbfc5e57c386e9d93 : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6d0e6854b94842fc83a77c697045a912 : SUCCESS in 8m 44s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/c2dea8e8313146ada5292026ddfa7156 : SUCCESS in 3m 59s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/fc5577a772284256a99369fc84ce6956 : SUCCESS in 13m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/7c7c5708c3424c27b841ffb9ca485197 : SUCCESS in 21m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/313eec70acfa4077b45cc4d72881bb7a : SUCCESS in 5m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b0a5f109b1814709a3446128b5ffec89 : SUCCESS in 11m 53s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/38752c0b5ffb4c8b9bb760fb3c300ef6 : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed7ba744a8b8418594f52b7944f5e122 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4b6afe2812b48d2aecaeb0d8c270f8f : SUCCESS in 23m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a14d2147bff946c78b011b52afea2dd1 : SUCCESS in 4m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/fd1cb37ede554e1dbe0a1b4d492657bb : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b69c66fd4e8f4168835d023d573b8379 : SUCCESS in 5m 46s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cbd31997454a4c6c83837612dfad936f : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cbcb64a81ea4cd88621c9c3d8303d7e : SUCCESS in 50m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bfdbce21ea345409c73bbc20a532e57 : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bcf25741628b8f80be3d11cca1aadafe2f695669","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 09:57:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5065a243472f61e7f936bbdb264cfef329d5078a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:03:10.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a90bc9d512546ab95a2dfc633576a31\n\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acf36acb85754545874776e07fa59231 : SUCCESS in 2h 01m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/8ddb847c647a4833ac3961912a60472d : SUCCESS in 1h 06m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe5826b3c73c4684b516ceb23c448fe7 : SUCCESS in 1h 31m 51s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/9149688bd9da4977bbc43f49be62ffb3 : FAILURE in 1h 23m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d8ce9d10472369babbd68c5de9c8c414d7b9b8d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-26 16:40:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"538aedcd218a0a2b099261e30fbc41b38ca36db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:41:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84b12584e5ae3513cb9ec9c5baad57579d55dc27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 19:10:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c44e3adec384c6d9ce250e366ba2de8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6dbf98744684163ac7e4cf7595703be : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0879d6edb82d4cffa1ebba1955d2fff8 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1952113e927c43afac923add547462fb : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28e75a2e73c54b82b2f4dadda4d9a60f : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/809ea26be1e54be884ebf7466f9034eb : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb3d265eeda24e9eb785e32313661d99 : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/504c45e58a7941ec841286861fc17544 : SUCCESS in 10m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5a2b8e8c04b9418292a09b6d9c835ab8 : SUCCESS in 15m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e395a54dde3544fd97aa94af7403dc53 : SUCCESS in 9m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3e6e2388106940d9868ec02f7701e9c7 : SUCCESS in 16m 40s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9eb8063ecf744c9aa51f1defc0a60d2f : SUCCESS in 12m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2cca9e8c76344524be59a5d74ec68a8b : SUCCESS in 5m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7845f0110f93445d903fbe0bdfd9862d : SUCCESS in 5m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ca7c80bf96c848fbbbb7b11d44a2317f : SUCCESS in 8m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cbccbd082a2a4589b04959d655350a19 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/679d2903202a45c2a66a25d8a30beacf : SUCCESS in 8m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/285283328b504fd3a3463dc739550958 : SUCCESS in 8m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/27293f218b5647049b21a42694713c3d : SUCCESS in 10m 55s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/de5cdba8352c4fd699ac1d1e8fea4e2f : SUCCESS in 4m 29s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d36bfe59a78a4218807ca4564a2a4b41 : SUCCESS in 9m 56s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f3ceab943d3342188ed22b6896aa4669 : SUCCESS in 23m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d74d480d2344c0cabe4241ba42c7128 : SUCCESS in 6m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1a8b6a5d7e604b2fb0f221698e419bd8 : SUCCESS in 13m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e07bf2108946409b816930e044f69209 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/306d69b27e654b81b0199656606f815f : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e24c00138040419bb43ef1bd45144a0f : SUCCESS in 22m 38s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5302c2f162754e96b430dd44676d35eb : SUCCESS in 5m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f347a730106b46d0a80da35e9175dc48 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6c3239f7a224d7c8ec07940c338ae53 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/19437f27e52d499eaa2d52986fed2f24 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0190f4e5ad549f58b286e2190665149 : SUCCESS in 49m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c2bc5e3d9ca4c0e8d75f75bc493631d : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":1},{"id":"58b163fd72c673b2e3d812e61582c0e413eb9837","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 19:10:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"961a224363eb26289e9d753d051ae64ac9bfee89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 19:11:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd18c76d50f4d769f16c3d4c6fede56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8f7c9fc3b3463c94f14d54414d49ae : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea550a20b9605159e5e63a6bf838238419547996","revisions":{"ea550a20b9605159e5e63a6bf838238419547996":{"kind":"REWORK","_number":1,"created":"2024-02-26 07:43:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/910143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/910143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/910143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/910143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/910143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/910143/1"}}},"commit":{"parents":[{"commit":"94fa4434f50558964cea0b7271eb816ff9d19253","subject":"Merge \"update constraint for openstacksdk to new release 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94fa4434f50558964cea0b7271eb816ff9d19253"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-26 07:43:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-26 07:43:26.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.7.0","message":"update constraint for oslo.cache to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If4d618b3dd4a917778583e36626fd129521b523f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I30072518c9a4fa9e04bbd55477383b150de59ba8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea550a20b9605159e5e63a6bf838238419547996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea550a20b9605159e5e63a6bf838238419547996"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
