)]}'
{"id":"openstack%2Frequirements~910242","triplet_id":"openstack%2Frequirements~stable%2Fzed~I4bc5a74c3b26834725c6f7841e5558ab2fac08c5","project":"openstack/requirements","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-02-28 01:01:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bc5a74c3b26834725c6f7841e5558ab2fac08c5","subject":"Drop buster wheel build check jobs","status":"MERGED","created":"2024-02-26 16:37:20.000000000","updated":"2024-02-28 01:01:05.000000000","submitted":"2024-02-28 01:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910242","meta_rev_id":"ab8a6ffa503ae1a7edfb29402326ca137c29ae35","_number":910242,"virtual_id_number":910242,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 01:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-02-26 21:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-02-27 22:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-02-27 22:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 18:12:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 21:44:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-02-27 22:58:49.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"37f13180db93f4b4d69587309cb69718e8daba97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-02-26 16:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d155cd337243d5c5469ff4fdbf242e4c83ac55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:12:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79da68d0ae7e4ebaa0c06824b2e9ba16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fdf667abff64059ac5697614c4013ca : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8343490baf4197a8da3a488f85af65 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cd7d1800b3648a48db4ae00d234f855 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eeebaed8a5b448ca1f6699edb807fdd : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d48bf7b0a54462b8efe3b41535717d0 : SUCCESS in 3m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/716a63d72ecd441bbe312c68bfda06f6 : SUCCESS in 54m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e62f3552a0b24eb788dbc149ff4c25ac : SUCCESS in 1h 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1a0203757a5975c298e5f9803637f47d39c11df6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-02-26 21:44:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24f4af51048342678d77ef536eb3664f94a9aa97","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-27 22:58:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e4481795cda5dd4a9129b6ba4c52b78dbab979b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:59:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3298712f6ae2af44fc746869d50512b5cb62e253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:01:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbf460e80a5496fb16ba4696d5a1173\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743884455d7148b8ae5da6e1afd9b5aa : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2c34515db8743b5a8440df51762a655 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73f3c0f2ed134f6d8c501d6285e51058 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ed81443c740405695ad1e54f0b80122 : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d4e9bdfc8da407b9c1ac2092e7846be : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1135d42a77de4956a87382aaf42279da : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ab8a6ffa503ae1a7edfb29402326ca137c29ae35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:01:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0c7e89a8e2607cad63c7e1664a24b507fdb0664","revisions":{"a0c7e89a8e2607cad63c7e1664a24b507fdb0664":{"kind":"REWORK","_number":1,"created":"2024-02-26 16:37:20.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/42/910242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/910242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/910242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/910242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/910242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/910242/1"}}},"commit":{"parents":[{"commit":"02fe1706f90d99b6c3bebb95186c4e8b24076f99","subject":"update constraint for python-manilaclient to new release 4.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02fe1706f90d99b6c3bebb95186c4e8b24076f99"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-26 16:34:27.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-26 16:37:03.000000000","tz":-480},"subject":"Drop buster wheel build check jobs","message":"Drop buster wheel build check jobs\n\nOpenDev is dropping buster nodes. The buster wheel build check jobs run\non buster. We need to drop the jobs in order to cleanly clean up buster\nusage and avoid zuul errors.\n\nChange-Id: I4bc5a74c3b26834725c6f7841e5558ab2fac08c5\n(cherry picked from commit 872feac390d3b7dfb35b5ead5ae4f15ec25c46c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0c7e89a8e2607cad63c7e1664a24b507fdb0664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0c7e89a8e2607cad63c7e1664a24b507fdb0664"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
