)]}'
{"id":"openstack%2Frequirements~910341","triplet_id":"openstack%2Frequirements~master~Ie8470178a3c59da80b2673d684e8197bbbb87ea2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie8470178a3c59da80b2673d684e8197bbbb87ea2","subject":"update constraint for neutron-lib to new release 3.11.0","status":"MERGED","created":"2024-02-27 14:11:46.000000000","updated":"2024-02-28 07:00:10.000000000","submitted":"2024-02-28 06:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910341-new-release","meta_rev_id":"ca9393534e81b5539d9caef9f469c908dc1e9cd2","_number":910341,"virtual_id_number":910341,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 06:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-28 04:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-28 04:00:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 15:21:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 18:03:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-02-28 04:00:08.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"46297d3e32fb2e2f41183c224133597ab75a8ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-27 14:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88b0c3bde24cac753b1c4e33b08444cfe417c005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:21:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2d328522d654fafa82dafa82e6a3176\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b353f483104108aeeb44df2025dd8d : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a54e91314a44638781ca19451ef8b4 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/864db257e25e4b578d2b7476a9b396e9 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ea66d3690e3471bb6ebb37119df4466 : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/276ea42a19d84e15bbfa8bb605196fd5 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/805499bdf0604dd9bcb0b51b240b4b0c : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fef9bf34d0384a1d9d75b13983cf2041 : FAILURE in 2m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2cde175efbeb4b09b9f3765a2b83ce16 : SUCCESS in 15m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/719b9ad9aa9849f598c45f17eecb0794 : SUCCESS in 10m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff015598f7f84f5091c4387f4bbfbaf3 : SUCCESS in 16m 57s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6b92d78197554473adbc866e34cd1fcc : SUCCESS in 4m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/52251cdfa1374852b3bd0c338417ccd0 : SUCCESS in 4m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1e0b01e9d4334dc8b9b3ca970abd16dc : SUCCESS in 5m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d8468b26c7984386814ea11058ba9e69 : SUCCESS in 9m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9f23dbdfeb2d4e5abaf4e22520dabb52 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9b6bfdcd1ae34af99f1a6d620dca08e4 : SUCCESS in 5m 26s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/907eb6c9bbdb4339b6301490242a0b2d : SUCCESS in 8m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4d43af3933c14f7898e107be1e45f272 : SUCCESS in 13m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73decdc00bc143bba8a0c0647b4d6fda : SUCCESS in 7m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0464d00572cb4e9c88c02419cc51be58 : SUCCESS in 8m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/3a1d1a219ebf487fa55e71896a82df0c : SUCCESS in 10m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/13ffcbdc60114369a9543f7c79b470f2 : SUCCESS in 5m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/fc54aa0ac6be44168577d4fdc9662bfa : SUCCESS in 11m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bacbfbde86de44d387a4f4e655185d53 : SUCCESS in 5m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9eb6a5fe54bd481fb08ef238067b0cef : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/85f21bbdd50147f89e76f9ab96e0165e : FAILURE in 2m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2bc8cf924daf45ab8b36da507828d1f4 : SUCCESS in 7m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/058fd61007ce464694f55fdb3f6faae6 : SUCCESS in 11m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/54649fde40634a7e9967ada8b3212340 : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9fec277d11f94f0ab734c1beee43ded3 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b09a2d950784493cbd5f64cab289641d : SUCCESS in 25m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8a51cfe2e96148649e1a8c982a60a065 : SUCCESS in 4m 08s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1420433487c846e18ee7edc378165fe1 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4b4ffafd0f149639dd041960bbdd90e : SUCCESS in 7m 28s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/6b2dd3dea88445319e179b0c0f80e04f : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b68ace214ec94cdea40ee33bdcb73600 : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b14b238f8d7f4cd49e37c4ba615f9f5b : FAILURE in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1d3082522ec22f83efbffd62597d831511bdbe36","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-27 18:03:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cc7dd37a2cb1d36756d9b287fa2a93a6d1cd3c2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 04:00:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"633bed3eb33192d75ea8cf9a29cb7de9f75e9554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:15:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27bad7dacf8a4438ac1d0b14f2a9816c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6169e3e49f604409b6d7743b92a85d1d : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1012b5c11194620823fae53189de6cf : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4f6fff2ba11439ebe12add685d74328 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c84a3703e9a4b95a74e2ec25858d4af : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c9be4fab42e42249cdc8b5e6f129341 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3debcc2979c24f348ba32d60434513a1 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/984eba0126b84a03b002cb4c909e53ac : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0796c44380b84c8eb595a17a6c7cd945 : SUCCESS in 14m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b2c7951873b14896b810ed61508abb6b : SUCCESS in 9m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6a9d010d332545e4b327fd47eccf273e : SUCCESS in 19m 35s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/121dd56c293f4a7ca2b6837cf39c3cf3 : SUCCESS in 6m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/132222cf06244dfda7a393d32b01a548 : SUCCESS in 5m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/34551ef0938e42208bb81b3caf3216ca : SUCCESS in 4m 39s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/3afa2539ce5c4568ae3f9d4542af366d : SUCCESS in 10m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a3e7b452e94b4d4b8a171af41dae5ca9 : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f3c2ee5f854847cbb4a10b246a7dcc53 : SUCCESS in 5m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dc9f34dbb9994fdbb669d046ff0de6ff : SUCCESS in 7m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/be0f6fab86cc452790d5ba53cb6f818c : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c059c6acd0d41309d143d070278c16d : SUCCESS in 8m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f614653bd43d48429d310d3e40889e42 : SUCCESS in 8m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/63fbfa17e8ea42d495c4c42d3b09e8e8 : SUCCESS in 9m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3843146608644e3b860735dcf85e72a5 : SUCCESS in 3m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4c78e1ba550d46208f84f5f241edc236 : SUCCESS in 8m 59s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4a1cc429e9c04d8bb7c57a8baf3e62c2 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/51f3f013a1124ca883a6f050212ff302 : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9349d713feea44bdbe6452a60419d568 : SUCCESS in 21m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f449bf9957124a39aac309c665d0d825 : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/503c40856cb843538fcfbecf148806a5 : SUCCESS in 13m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4569ab367c284af99fc224e8f4fd4dab : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b31c9c8f35f549ffb45c5b326a668c65 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19b4706aeb2a4eaabbccb29e85da1e7c : SUCCESS in 22m 18s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e069ea4bfdfa4060b954656b4ccabd48 : SUCCESS in 4m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/bf288f8537504e389f2db622202ba7ea : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/885937d81aac4e959c3ae747d062a42c : SUCCESS in 5m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f236c05bc50643b78b3255aadd344698 : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3d388c64d6a45a8a3cf8ac25929a1fc : SUCCESS in 58m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/091b1221a28a498fb1170bdc0b3dd407 : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9284c305d833561125f9e1c85fe0c816dbc64cfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:16:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bdd35cc34f6449205a6b5175e2333a45f57258f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:59:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0107126d6394ac3ab3cb60f849cd4e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4ea2143ec6471f8aa6e13dfe4e87b1 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/761f0157e52d4e69acf76ddc2a9aa38b : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bc871f861a845e99b5c0e8a7c03735c : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a429e4039fa47f7ae6f1d924bae9bc9 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2959609bbe984462ba024f2a10952d79 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8caa33a38a49439281789b4f12603b6a : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11f59aa234c040ed948f17e842f44804 : SUCCESS in 11m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d14616f8c45f4fcebd332b6c2b0f0f60 : SUCCESS in 12m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/88ec6d76be0b4aebad90032e852d0a0d : SUCCESS in 10m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3cf909d8a52464589ef62771f5c603f : SUCCESS in 20m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/65943c2013fd409aa5b0bca3a828b3ac : SUCCESS in 10m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4a80e4ef6e424c6d9e0b4abaa5792f64 : SUCCESS in 7m 01s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c55e949ff2074e0d9a70f5000e683705 : SUCCESS in 5m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/32d548bbeb38456b80710045e1fb2b63 : SUCCESS in 8m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/57e6eb6ba2394e25971d179bf7669ff3 : SUCCESS in 13m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0af91d75220c4d59870a435c33f2329f : SUCCESS in 9m 20s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/87d57d938e4e44bf9d78038af99494a6 : SUCCESS in 7m 14s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/16aac120f7cd4c6aa720a0f53f1f1e33 : SUCCESS in 10m 54s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3602dfec178a4af1b263b2bb00dfb038 : SUCCESS in 5m 25s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/752d2a197227445c98d814a24556b49f : SUCCESS in 7m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/590e14ae57f44ddfa7427483d6c8bfcf : SUCCESS in 24m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/29b5dcfd7eb54fa4b1b500ceb47ab77f : SUCCESS in 7m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/053c1c365ed2416dbdf4a9cd06b0b4fe : SUCCESS in 11m 11s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7ce1db61e02b48a5bf4954f828ab983f : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8703a555dd194da390b7059932e72d1b : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b65cb0e27e14cccb0249aab2600808f : SUCCESS in 23m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a6b8bb8f1cd1403080f4ef28c0338fab : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/129c138f15a44ea5bf63275438c60ddd : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71d15e11f4324f56ab2616b724853190 : SUCCESS in 6m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9cbe1dcceba6466eaf05009001862e1d : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afb171e3dfe34175a2de2683efa72ad3 : SUCCESS in 47m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2fbe567f82f47a585fd6e36ddc7a7bd : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d6dadba915dd9f8df1b227db4292e4e728c413ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:59:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca9393534e81b5539d9caef9f469c908dc1e9cd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:00:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b2b56fb4bc4e7096418e3e6036b998\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c94a938cb8a4a4e8057daec71aa376d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d46ff8815cb778af86d258870361a5424abca651","revisions":{"d46ff8815cb778af86d258870361a5424abca651":{"kind":"REWORK","_number":1,"created":"2024-02-27 14:11:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/910341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/910341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/910341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/910341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/910341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/910341/1"}}},"commit":{"parents":[{"commit":"c99e7a228ef89ad68a37107dc526bc701554fe90","subject":"Merge \"update constraint for oslo.concurrency to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c99e7a228ef89ad68a37107dc526bc701554fe90"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-27 14:11:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-27 14:11:45.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.11.0","message":"update constraint for neutron-lib to new release 3.11.0\n\nmeta: version: 3.11.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ib4190e845bdf0371cffef3a78e1c1236539e9db8\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Ie8470178a3c59da80b2673d684e8197bbbb87ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d46ff8815cb778af86d258870361a5424abca651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d46ff8815cb778af86d258870361a5424abca651"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
