)]}'
{"id":"openstack%2Frequirements~910573","triplet_id":"openstack%2Frequirements~master~I9a9de6c5cf38857769d761aff06ecac9b6542941","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-03-08 19:17:34.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-08 19:17:34.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-08 16:45:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I9a9de6c5cf38857769d761aff06ecac9b6542941","subject":"Add cross-magnum job","status":"MERGED","created":"2024-02-29 07:52:00.000000000","updated":"2024-03-08 19:19:31.000000000","submitted":"2024-03-08 19:17:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910573","meta_rev_id":"735bef53dbcb719286f1728609629aeb441ed1df","_number":910573,"virtual_id_number":910573,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 19:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-03-08 16:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-08 16:45:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 09:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 00:41:19.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-03-06 00:58:51.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-03-08 15:53:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-03-08 16:45:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"026736e40992660bdde654436b65862b8cd2e1b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 07:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c85c9cb6ff209c3c8067df93c6df9a58698d25ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 09:29:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4582c78f988041189c4549a49c144bc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a69713e00fb43cc9fe3c12100d33ef0 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00784f19acb40379dc908bc081a1078 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da69cb1668314a518d6ef6189c943f00 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/492a6fcab38e4d3c912d962abdf895e5 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8998b39d1e624a01a96ea1efd43bee6b : SUCCESS in 3m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ed9567aebe3842ba93cecda904ef532f : SUCCESS in 5m 26s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1c086dd28d354920b412943a677139b6 : SUCCESS in 5m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ea4d485bf7c646c5b691acb4f7ea7bc2 : SUCCESS in 4m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ae518b33ae9e4dd780edc73c239a521d : SUCCESS in 10m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d5887389764e4c9c8dd286f5c3bb8794 : SUCCESS in 5m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/518aae4c7ae34c2683e54bfbf8c57cd2 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/670674b73f684d22aa505fd2f006edce : SUCCESS in 8m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d231ed438b5949c4a314bff86c18ea65 : SUCCESS in 13m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf93f3de9fe04110808b0dd207c33696 : SUCCESS in 7m 28s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/676b0beca5be414591c2b88f5b041db3 : SUCCESS in 7m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/3e87b74d5f914dbda010c1d1d05fa952 : SUCCESS in 11m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2d3aa74975e44deb8d6b7973f7c51ed8 : SUCCESS in 3m 28s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/81929034871647bf9e54850e177ba419 : SUCCESS in 9m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e14b06acf2e54531a82b17f09f89515e : SUCCESS in 6m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e341931f899d40beb6b943823935d140 : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d6c8478aed1a476f81a04b5ba9c3987d : SUCCESS in 24m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8086db6295a4251aab4c8efa825727c : SUCCESS in 6m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9c369c1282714dfbac1754d622ffe13c : SUCCESS in 11m 01s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d29f5a1d3e1c4cbdb51ebf4f2521cc4d : SUCCESS in 4m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47a5010657464ff0a480c63399ad97bd : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9b0042eff2a4c4bad85141a2eca42ab : SUCCESS in 24m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/988e1794be674e00a0b9f6682f026f58 : SUCCESS in 4m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0ae3d55a34e74afba022295573541296 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/220b8243c5bb49fc8a80ec2b73b0e9a4 : SUCCESS in 6m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7de5958617b04707b15736dfa09536a9 : SUCCESS in 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c46f0010a00842a9b81af7510d628d05 : SUCCESS in 57m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b8785ae8d8d4f068f5e0b3390accd4d : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"30102b97ceb18983c87534c74bb63a87e1425ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 09:36:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f90c87ef1d03885e82d657ccbd45601a56351a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:20:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/008f3e3fe4ba4c62b1f7517690321f41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74671fee9abf4b42bf0b188e3d33a6c6 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29865cebd51c459cb6e7c714e54a441d : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c12d723ea74c4fa4dfb0bc500119f9 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20af5eb586964ca5bb6af3c9c1d5e151 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afaf6debb38a4ace9789cc65f470cd76 : SUCCESS in 3m 07s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1da73d36fc7648afb6a7ae1479cea6f4 : FAILURE in 4m 27s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/71a27cd9603a46dd9fd85f73a18a75e7 : SUCCESS in 4m 52s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bbb4518cbd774fed95313bf5aae879ef : SUCCESS in 3m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/279234bc2e0843d88004cbdb3ddcb818 : SUCCESS in 9m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d251560978004d33a12649ecd13ec17d : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e1730853c4a0473d81ee61564f83e22d : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fc8ca9d29d7c488ead46c72228bbe6b0 : SUCCESS in 6m 41s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4bae3d9126374f94a79cb4d00a913b1e : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/911b0df9ad7d49dfb4223cae289e0756 : SUCCESS in 8m 44s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/41cd1ff24b83401a8e6dfaca36dd36d3 : SUCCESS in 10m 55s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7dc719ca5afc49cea366657b81b87568 : SUCCESS in 10m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d127afd06eb34b4e996ee237a9a46f91 : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8588948ffc0a447fbbbd59fcdd268096 : SUCCESS in 8m 21s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/23192ca72e644693a700c4dafe7f2ed8 : FAILURE in 4m 22s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a44c1ee56c584c8c97fedef331fb1119 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/abe3c19ba6b54662a23001eea962b30d : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/82e82701500d4b69addbe2dd3ed2e0eb : SUCCESS in 28m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e751e29435754621afae11b60847e536 : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7b08c48c8c184768b4539190ae74e1b9 : SUCCESS in 12m 10s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/eeb89fbddbd04116befa1abf0144071c : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/544d4cf4f26b4155bba7e819f88bbd4c : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e491a4b915f74d75b890a6d7ebfd7d63 : SUCCESS in 22m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e68bb5b4a1e74cda96b05fda38d930f8 : SUCCESS in 4m 11s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5a56deda40884a30aad4d88d9d342dfc : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fbff359235e4b4ebc9765f4e772b9ea : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/82ff99c9d3a14913925940e80e408146 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5efdd8ec176448ec95b150a4f57520e8 : SUCCESS in 1h 04m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2ffebfceb2a4dbda10d6c9465f91121 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":2},{"id":"57687c738aa1ad7cc70d3875ffeff73129f5d10a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 11:28:58.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88555fb16c97712ff19e3afb0a39edcb87bf8ade","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-05 15:54:15.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a954ead6c75bd22c4abe8dee85b4020618053adb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-05 15:54:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f14ddee1c31cd8e063edda5af5aeec1c0f0bac8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 17:38:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30bfbf2939cf4a0f9901b73dbbac2901\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3dcb2a76d84118968615ec1907139e : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719bc6636d454c0fb5d38f42449e6fff : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecafc80c82b94fde861ba5ae351386e0 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4a3c610ca4747ea8f6fd360d7f3d478 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f23c69474484e4097ee8bd4dae6cea6 : SUCCESS in 2m 57s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/69d13d3106904606bab895802bad81d4 : SUCCESS in 5m 14s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5da9d11e18d6428785c37e46dc2dbd29 : SUCCESS in 5m 28s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4b233620430a48aea7fcff2753407218 : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/476ecb02ae3a4e8e992a89e9edb67983 : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/30c5eb3a64bd455f9861bd0925a6dac9 : SUCCESS in 4m 57s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/70a82c4360d44716bf6fe1b07efbf5ff : SUCCESS in 6m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/11593d5efe284250a0d75e1cc1df7e12 : SUCCESS in 7m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b5065b3b17534b7eb513eec666bdb415 : SUCCESS in 13m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cab19e5c6e7a418e944c800c24f3d854 : SUCCESS in 34m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d7bef5b5a41949f9b943defd6d6f1db9 : SUCCESS in 10m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c1dc17d0cf114c70a7e11374a546bff5 : SUCCESS in 9m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/63ff317ab18249cfb5b4f43f93bc09e1 : SUCCESS in 5m 22s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/41ee447a1d79404f81c86d711eb48532 : SUCCESS in 9m 32s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/585bdfca7e0842e9b041db02e8b20207 : SUCCESS in 4m 37s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e66fcfe7e0fb4b56b84d28270433edc6 : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9646f565782c4e81a9c4d716f89c1f0c : SUCCESS in 10m 17s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8c21c96e51dd45888ccc98c8499ef37e : SUCCESS in 27m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/68588c200cd745c49b2d1dbb18b17f01 : SUCCESS in 6m 24s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6026b9bedcac43f986996c2f5740040f : SUCCESS in 11m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3c3a9f7e74e34b12944fc3dbfac4ed3b : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a309e4dce9a74690b455b2985106e326 : SUCCESS in 6m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b935eac20ca4280a70821d17fc01e39 : SUCCESS in 21m 02s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c8ade03cd88c4f438a1bb1b1ac4922c5 : SUCCESS in 4m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/73895a9420dd45749944f5af6d7c97a9 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74c7b349ee6d41529e2ec8e49ee67f58 : SUCCESS in 6m 47s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f9c6f8c3578747f6abf6c9c3a49a44a0 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc685bc87b24051accf2db8a5ff9506 : SUCCESS in 52m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b271dceadda945f8992a570dfcbbfa36 : SUCCESS in 1h 35m 52s","accounts_in_message":[],"_revision_number":2},{"id":"378c7a461e858f48f3bfe1545ff5b246a0f3d1d7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-06 00:41:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19f42c2298266bfe099b6939bceb5805c534bda2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-06 00:58:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9739195df9f0077f999442faacdec18f594b9cfa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-08 15:53:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56b77cfbe8724b754b1399e483f8d46639599bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-08 15:55:04.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c260bb75d3a4633a04d836ff7d5b0d7bbdf946d4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-08 15:57:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96ac35084e67ef55907413e23b8df5ce29f3e8bd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-08 16:45:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9bd61e82b5033dcc44b9f711a053722805b61969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:36:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98107a90b4cf4839b5b8f0df4c60a817\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dae82adb5994d38938e5917146d3f07 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d7b140bd4b24dfc9af80d19e8187e0b : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fdfeddac25649acab1f2264e46077dd : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4364d424885148fb9fb76c3e00b6a423 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea62b484b8e84116b77fcea8349f4d55 : SUCCESS in 3m 47s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/4991b4e8b0a649ae89b7280f57b34286 : SUCCESS in 4m 21s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/4d23a39aaf614a82aa3b0ecdeddc5fc8 : SUCCESS in 6m 19s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/46c96ea276004c5bb61d00af017f1127 : SUCCESS in 4m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7c0d9b6e75264179b25d57422922dcf3 : SUCCESS in 11m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7239be843109464888de2ddec3d8aa24 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8eb8af3093cd4415aec245b7a94c7eef : SUCCESS in 4m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/df6cd336d0484b729c595fc40bc57c45 : SUCCESS in 9m 12s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e6d9b18e45264afdaf696316dfefe8da : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6b59e5d796d4c95b648bc99c4c40b86 : SUCCESS in 9m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/87e22349d5464e14b13b5e444a886176 : SUCCESS in 8m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/48fed161d34441fbb04b3ffd7e6bc247 : SUCCESS in 11m 16s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/07de93c294b2445795f370439f213227 : SUCCESS in 3m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3aa4a497e8c040509df7d9932812779a : SUCCESS in 10m 11s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/5d074f8df6524fad9f1310d57d6b5e4e : SUCCESS in 4m 37s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/344ad22b1b6d49f3874da6f19c12760b : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/bcff3d32359e47cab121bf5a91eadb8a : SUCCESS in 11m 15s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/096259f9848f4a47b725950d9291b7e4 : SUCCESS in 24m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6ec249dee3434bab820f5c681350a026 : SUCCESS in 6m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d86d907bd26740dd93d582956f896e00 : SUCCESS in 10m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/60f807d0d9ad4896a83a63ac60cce5f6 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e33a554632f433db6339d2bc346ee09 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/697cdae7d7d74e0a88cea0787b94e660 : SUCCESS in 24m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/81ed8eeb21084da88d76f4a05c38eda2 : SUCCESS in 6m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5af1870998354a56af634aa48322b805 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f22c5cd5b5e9449a9f181548b0dedd12 : SUCCESS in 5m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/6145b080e9d240dd8bd2d465f47cf03c : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb1880cf044c479a98410cdbe5c98e91 : SUCCESS in 1h 07m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66a82c5e5339406e9869cec440118776 : SUCCESS in 1h 32m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8eeac0ee8c9b8d6228c8caa8c8aad8ba7284bbbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:37:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0deed8e332adf6273653e1fa9d76400cf14321bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:17:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/488044263d0246769f290d40dc19cf2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc267f4f1c4a48d1972c0f84274ee70b : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6a2368a447444b92db483fb12bcdc7 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c14248729fa48f18e6cdedd4720d7c4 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2262e63785684c53a17fd8b1e7ce4cd0 : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d73153200bd7411ebae20bea2625f80f : SUCCESS in 2m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/45871c8c63274218b4ce009113095109 : SUCCESS in 8m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8da4328e0f39448ebe99a77208f14c84 : SUCCESS in 7m 36s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2e18dc1af85e40989dcb99c2a50ea344 : SUCCESS in 4m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2af61e06ee7546c3829fd28136d5e50a : SUCCESS in 7m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cce728e0371c4f0d8caca6bd9ec2f4e0 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb9531c9476e486e80153e24827891fa : SUCCESS in 6m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/78eb267880a3430a93eb3ce68bbd5b7f : SUCCESS in 7m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/874b7f98c4494cb4a510f733115ccb53 : SUCCESS in 11m 13s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1f70c5ff8da24cadb17cdd47151782e2 : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/054ebcd138324ebfbcf2853257c53213 : SUCCESS in 8m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f5ffeca6913f4e138f3fb0d398c02fd5 : SUCCESS in 21m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/80226b46727a4bbd9609789892fd0be7 : SUCCESS in 6m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a072e0581394d34a70267a234f91e6a : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d0e50020c3b440c96b7007ee4eb5808 : SUCCESS in 21m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2eab8e898192408e9bcddfaa7bc84525 : SUCCESS in 4m 18s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d9cf34cd01f142ebb22990dceb1be704 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a5af543bca4473ab35e409824127d0f : SUCCESS in 7m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/67eda5a001504db1bd3f11958239fff1 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ddfda1f695741948f14be41e9a4e5d9 : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9010e3f59b7841eca78c3ac041b3ac83 : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":3},{"id":"ff27b4c1396e541b38dcd057fa37bf5ee1987812","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:17:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"735bef53dbcb719286f1728609629aeb441ed1df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:19:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cac016cd33043a684fa74a7d03a7a52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbeef4db2324ff9b8b71d0e102f29ff : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2041edc6729b78ac4d6715b9b0fdfb4f213bb869","revisions":{"2ab7e90194adc57b04de870c89d74c4eed1b711b":{"kind":"REWORK","_number":1,"created":"2024-02-29 07:52:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/910573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/910573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/910573/1"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 07:51:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 07:51:58.000000000","tz":60},"subject":"Add cross-magnum job","message":"Add cross-magnum job\n\nChange-Id: I9a9de6c5cf38857769d761aff06ecac9b6542941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ab7e90194adc57b04de870c89d74c4eed1b711b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ab7e90194adc57b04de870c89d74c4eed1b711b"}]},"branch":"refs/heads/master"},"8beb193759ae56c71d516efc91bd03b51306e7f6":{"kind":"REWORK","_number":2,"created":"2024-02-29 09:36:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/910573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/910573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/910573/2"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 07:51:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 09:36:49.000000000","tz":60},"subject":"Add cross-magnum job","message":"Add cross-magnum job\n\nChange-Id: I9a9de6c5cf38857769d761aff06ecac9b6542941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8beb193759ae56c71d516efc91bd03b51306e7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8beb193759ae56c71d516efc91bd03b51306e7f6"}]},"branch":"refs/heads/master"},"2041edc6729b78ac4d6715b9b0fdfb4f213bb869":{"kind":"REWORK","_number":3,"created":"2024-03-08 15:55:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/910573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/910573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/910573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/910573/3"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 07:51:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-08 15:55:04.000000000","tz":0},"subject":"Add cross-magnum job","message":"Add cross-magnum job\n\nChange-Id: I9a9de6c5cf38857769d761aff06ecac9b6542941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2041edc6729b78ac4d6715b9b0fdfb4f213bb869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2041edc6729b78ac4d6715b9b0fdfb4f213bb869"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
