)]}'
{"id":"openstack%2Frequirements~910586","triplet_id":"openstack%2Frequirements~master~Ic09bcffe2fb301d082d5545ec97ffad0207a1bf7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic09bcffe2fb301d082d5545ec97ffad0207a1bf7","subject":"update constraint for python-troveclient to new release 8.4.0","status":"MERGED","created":"2024-02-29 09:56:02.000000000","updated":"2024-03-01 06:15:16.000000000","submitted":"2024-03-01 06:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910586-new-release","meta_rev_id":"496865b35e8f6b417e50a339c3402737d02de2a2","_number":910586,"virtual_id_number":910586,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 06:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-03-01 03:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-03-01 03:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-03-01 03:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 11:07:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 03:16:18.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-03-01 03:23:59.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"fcd73429844418c37e85795e23f1c93f478819f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-29 09:56:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f8fdf33b0a67bfc24034744a1e1f42202b2213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:07:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08db04780eba4877aeda7558e5f02a1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53420d9130a94343ac89efb2d980da4e : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5110ca70009348ee855a9843eac666f2 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d94e30ac19224d3c9af7938ae13078a4 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31109ce3d4754921bd54a4169381a832 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83096df15a944066a8852fc6ea1b5c49 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/455bee81b9d9439288f53b2a9f3cea7d : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bec5bca3b7384b3fb933bcc6a9cec201 : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0410aa584f3a4d64ba0948ad66ed8541 : SUCCESS in 15m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e46ed00141144e94b297b66a0e790497 : SUCCESS in 9m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8977ccc2f68a4094a003e64f89f0b929 : SUCCESS in 17m 10s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/781af91313ae46cab4393953dd74e628 : SUCCESS in 4m 56s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/82f090e9f2154ed68959d61ff8b7d7d2 : SUCCESS in 4m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6c2552d8e68e485692a581a6e048cf12 : SUCCESS in 3m 47s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6b41f2139c584ef096318d8cdec5e23f : SUCCESS in 9m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2cb68d89c6b140c7a2f3fe5bda2e14dd : SUCCESS in 4m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ce91ecba51b54c9eab68edcb08752398 : SUCCESS in 5m 17s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/4dc2ee3b939f46edb25330a8870770cf : SUCCESS in 7m 52s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a56197efdb50494f844f99bb8e93b4d3 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cf7214da35f4c9f918b78dae7e14c50 : SUCCESS in 6m 57s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5938bd9f07274eae80d46ab3331ce4a9 : SUCCESS in 8m 35s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5884a62a588b43109d54ca811293e4e4 : SUCCESS in 13m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1eecc188815e46249674a4e2d3f5eda8 : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3fcbeb5a2a6e44ba94c353973caea2e4 : SUCCESS in 9m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/814309b22a9f4c85b2bc92957a5211b1 : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/298231fc6d894b379be538d6d2d4dbf2 : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/367b4d602b9e416a958dfbf7b469849a : SUCCESS in 21m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b5d19fb057b477ab5bf45ebf8f81ad7 : SUCCESS in 6m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d658c7b6fc1a47e4916ec488d4dd2817 : SUCCESS in 12m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b3906a0fb8524f41b6d5272729789a76 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4cc04c8926ab4b86943ce632ec9d031c : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fb7a7705c5d44d695e7b5c44d34a7b4 : SUCCESS in 20m 03s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/bdc0286dc049405eaf96e2cd7ba2ee4c : SUCCESS in 4m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b1937acbe3944cdfb5276b726f315ddf : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92aaef0fa4b64019a1366d36343e078c : SUCCESS in 6m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f5403b6942ee4bbcbf97597c8a52bffd : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39680a493c1c4a7d96e385aae17bae0d : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a45d551975cc46a2bcc286c3388967a8 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9d5641941196f40e26f47ef3a4430ff5182fbbc3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-01 03:16:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e096764810e71841336532828649bf9e875a39d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 03:17:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38efeb395a3c36c0bb2c287ac1bbf5f2c770e5a9","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-03-01 03:23:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f96385ee18140bf4cb81356ce3d8d54f81be497d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:13:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9645f8a0ae4409b9ae4a29a921baf52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1225d9f694472f9dbb6f9b00c10be7 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7dba2ccba04fb38c00ccd9ae22d1bf : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/973de3ca233b4be7821a463001c8a77a : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/687c460cb3174c00b632153004ae7b5b : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b66c4809611f44489108f37f9229ead2 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28046328fa5a4610a46a4e397ea59b19 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/909497627b59478288d37dc0d6759133 : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a6428045ab5c4e0dabc254feaaa1dd22 : SUCCESS in 15m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e1bddd1d42ba42de95cb380f38553bb1 : SUCCESS in 10m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/735c8fe94ec7428b83b13c36c55fbf7c : SUCCESS in 14m 27s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1352d0ff58e5424f9997373b233d3717 : SUCCESS in 10m 24s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/dda5655ec12a4c51b71e0cf7c4d840b1 : SUCCESS in 4m 58s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1aa4d9d0579f487e80c30eee38b10fe3 : SUCCESS in 5m 26s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2901badaca024e9e80f6ef62b2bcf9df : SUCCESS in 7m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/376d4858cb034acc884a1d66892efe00 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99974f7a8b90486593d6953335673cd1 : SUCCESS in 8m 49s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5cf62cf18bb14a6ea61d9bbec759e0bf : SUCCESS in 7m 36s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/24a777656be14ff08a6c558a5b9e50c1 : SUCCESS in 9m 55s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d0b10fae3bab490fb01e19fdb43fba54 : SUCCESS in 5m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a74896db6a3342ae952737615ea0cc74 : SUCCESS in 8m 51s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b865acba1ca641ef8d3e42e37e8e8eef : SUCCESS in 23m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b39650085e2b4199b0eb481e9138be81 : SUCCESS in 8m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e215f4d5f93b4215804f563667690fba : SUCCESS in 12m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3cb1e86d19a0463b8d5d4b377b82bc29 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbc145e862bb49789c84485a043eb8ef : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9237c0e9be794a798bbf131f923d0723 : SUCCESS in 21m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b7a32efce9054900ab54a9b1cc66c6fc : SUCCESS in 4m 43s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8a32560e007440c78d3ea5dbf296a4ab : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2929771f480a4984a69628fee561d810 : SUCCESS in 6m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/618fa2f68f6447559e4e59a0f0f86306 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0b0a82625c44a6ea21cf5fc61066920 : SUCCESS in 45m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d9b4c5c43f044a29028209a9eabf1c3 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0f2bfbd5da8604ba28fdda0ab48936959ae1c926","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:13:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"496865b35e8f6b417e50a339c3402737d02de2a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:15:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab5af6efd08465aa5800116066857e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef50b3d5da74b8284b123c7fbeaf1bc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5ff8686ab07612103bf384085196a1d7a543662","revisions":{"d5ff8686ab07612103bf384085196a1d7a543662":{"kind":"REWORK","_number":1,"created":"2024-02-29 09:56:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/910586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/910586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/910586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/910586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/910586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/910586/1"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 09:56:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 09:56:00.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 8.4.0","message":"update constraint for python-troveclient to new release 8.4.0\n\nmeta: version: 8.4.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia57fbf93315315e547ef7c67d7f388ac47ef76e3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: wu.chunyang \u003cwchy1001@gmail.com\u003e\nChange-Id: Ic09bcffe2fb301d082d5545ec97ffad0207a1bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5ff8686ab07612103bf384085196a1d7a543662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5ff8686ab07612103bf384085196a1d7a543662"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
