)]}'
{"id":"openstack%2Frequirements~910593","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I8bb79cc7f65874d70992f64bdf1685024dac012d","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I8bb79cc7f65874d70992f64bdf1685024dac012d","subject":"update constraint for aodhclient to new release 3.2.1","status":"MERGED","created":"2024-02-29 10:51:13.000000000","updated":"2024-03-01 00:52:55.000000000","submitted":"2024-03-01 00:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910593-new-release","meta_rev_id":"770ba811ac91fafbba46d079cf911cebd14c8ef5","_number":910593,"virtual_id_number":910593,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 00:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-29 22:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-29 22:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 10:58:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 12:15:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-29 22:55:53.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"e96bf3191c7c37cca2059b995e14321f3a73d913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-29 10:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327526f45c00f32edaf5b6c2d0a33a6b89bb70a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 10:58:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8daed6b0e38041ac8d5d34f2075a1a7f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/98d3821f6ad342678286f7cf72389c36 : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"435f476442e82ff91e25978b9452c4645bf2884b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 12:15:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b112f838de44c02b97134e2f39cacb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ac630a2fe248f585668034bd05bbd8 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b7003396cf45bfb673d5d75f3a9efb : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5feacc0c41944bb482a72115674056bc : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e399c9bbdb84f0982268dc0331b3e87 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb208e186b1849f092e67b348347ed74 : SUCCESS in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ac6c0e1cec44c73bdcc9b3cc73ebc0c : SUCCESS in 13m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/92c43fab78da427ab5c133f5b0dfa0a8 : SUCCESS in 7m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3a17996e7ec84777861fc4330d3838f9 : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4d154d63033d47888698e377628d4afa : SUCCESS in 4m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/755e553eff7e4892a72cd87ad4174bf2 : SUCCESS in 14m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2c63d1c6a5544f67ac9138ef58cd0f84 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/956ba3338921483e99fbf64329e09f09 : SUCCESS in 7m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/df7e808ac7a143cfa777e916aa70935f : SUCCESS in 6m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f30b657008254df48d1683c0633b20e3 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c867347677a4bfcb4fb9986be974e25 : SUCCESS in 10m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1d96b727819949a4a11bc926c90efb11 : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/637036aa1914467b87d2e716c89d12cb : SUCCESS in 13m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dede66cc779a4b0ebf92bcdce1921b81 : SUCCESS in 5m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b783e20576de4b838d6c43e41676ec14 : SUCCESS in 13m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9988c26d9c254183afd9ab09c5957524 : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4fefdc2bd41b45d8a80bff02575e460c : SUCCESS in 12m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9bdabed2fdb1409b8f455df26e89c904 : SUCCESS in 18m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a0c19fc848674bd3a368514900f76c86 : SUCCESS in 12m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/513261f0553746ee921e91b453658a8b : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1b26e583613483986ddca5428837470 : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/077f6641378b41929c8e6c2fbe951de3 : SUCCESS in 27m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/99f4f525cb564685965010d1dcd21bda : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f5a228fb250548399e8b0c45e897917f : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fc444d03bea4114abfbc80dcd8ab9d3 : SUCCESS in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b09bb53a2e4c4a27a6594834deddc9e7 : SUCCESS in 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a9c8082095948fa94a8a666e9dd0b64 : SUCCESS in 37m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f80e14989d024733afc5615d3a744267 : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7b755d8625072cf1a905038ee4528e03db8b3053","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-29 22:55:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d52f626ebc1516790315b187a8e72d908d676d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 22:56:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58b4605f2eb4295df529ed5b9b38ceb858815c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:52:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b282e1814e944babf192b7a7c1ee623\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e1495a48594da5af90ce9dea965391 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b603fa32bcfd41df895a3391433963b9 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8122d7213e4140199e2d87c7ab299734 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/888fb25f878f4b2b9ca1a5ed2b187044 : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d58902e0bc704d758007dafed062613b : SUCCESS in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb6357cca06846a5b96f89f0dc7c1639 : SUCCESS in 11m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/983c95557c964b0cafb351c5908103ba : SUCCESS in 13m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b12aff36cb3d4a0bb654fe4acd875a83 : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2ce70880ca9f42469820d5cf45a20dd0 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/81ddff5be54549ee855644196787fa61 : SUCCESS in 7m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f78fc73fcf4448b6b0fdb62a12ba3a02 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/301083dfc70a4c51bef7eb7e473da02f : SUCCESS in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bc6be24e30314692b9247560cc7c0f2b : SUCCESS in 10m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/687b0e4c8d9f4c42af8afa43fa3b36cf : SUCCESS in 15m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d1ec7f39d4c849b9a590083ab79ff2c5 : SUCCESS in 4m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/377476ef3b22479ab365be3281228f0d : SUCCESS in 15m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d60cd6db55e7470a8e596659bf073c6d : SUCCESS in 6m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b7f340d9d94f4be3878a364182e140ec : SUCCESS in 19m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ef1fb2626f77427ea35f6f28f11a0d54 : SUCCESS in 13m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8c2207aaae7a4cfab340c18a10898814 : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79ee65afc66a4792bc9f7bab8cd8cf94 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c2c0abafcac42f3ae39876e157648d7 : SUCCESS in 28m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bb7871cae1ac4042880a1fdc1d9817b3 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2509635e3cfe420a888f3ead09d84283 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d9fefa638d54e6693298ebfe59e46ca : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6bbd6fe0c15f44a498248b46b6a6bbc7 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61b63db382914ab9b64b4f0df695ecfb : SUCCESS in 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5883f45b9d2248678562fced44fff31e : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"770ba811ac91fafbba46d079cf911cebd14c8ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:52:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3ea54bd5476bb1a19c230d7993b820271758f6d","revisions":{"c3ea54bd5476bb1a19c230d7993b820271758f6d":{"kind":"REWORK","_number":1,"created":"2024-02-29 10:51:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/910593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/910593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/910593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/910593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/910593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/910593/1"}}},"commit":{"parents":[{"commit":"872feac390d3b7dfb35b5ead5ae4f15ec25c46c0","subject":"Drop buster wheel build check jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/872feac390d3b7dfb35b5ead5ae4f15ec25c46c0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 10:51:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 10:51:11.000000000","tz":0},"subject":"update constraint for aodhclient to new release 3.2.1","message":"update constraint for aodhclient to new release 3.2.1\n\nmeta: version: 3.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta: release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta: release:Change-Id: I143266a39d2a7dd0184932b625f9299d1c17e0b7\nmeta: release:Code-Review+1: Erno Kuvaja \u003cjokke@usr.fi\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I8bb79cc7f65874d70992f64bdf1685024dac012d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3ea54bd5476bb1a19c230d7993b820271758f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3ea54bd5476bb1a19c230d7993b820271758f6d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
