)]}'
{"id":"openstack%2Frequirements~910599","triplet_id":"openstack%2Frequirements~master~I1eaa95c19045213a7cc7f801ecfa18673b2376e7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1eaa95c19045213a7cc7f801ecfa18673b2376e7","subject":"update constraint for python-swiftclient to new release 4.5.0","status":"MERGED","created":"2024-02-29 13:17:15.000000000","updated":"2024-03-01 15:32:21.000000000","submitted":"2024-03-01 15:30:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910599-new-release","meta_rev_id":"a7ab4769b8cdeebc5956a26b41566e42201c8c19","_number":910599,"virtual_id_number":910599,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 15:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-03-01 11:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-03-01 13:44:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-01 13:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 14:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 03:21:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-03-01 07:53:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-01 11:44:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-03-01 13:44:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5466d8b0cbb9f97e34e9bb397cab2bca0363b9a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-29 13:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b68976e1febd9ab6960421fe3e1715f67de78a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 14:52:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf8bad55c41d49f7979e2f9e5c9a4b45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e4ee44a6a24250b2d227feb78efa2a : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa3f4158a2541bb8a0edef3e51ccc47 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dc64cd5068340ffa014cd98813bbe01 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e2ef85f13f24ebf9fff1f7d14347a31 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce9e5e41e1584b61aaca24c8b9b6d4a9 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0d43cd17e8245c2a0c53bdad23b048a : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f325b75f1dc4ef2ba7eedd385e8b753 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/abb0e44e65ed48008c1cd5487eb221ed : SUCCESS in 14m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d066df65df6b43339e4506ebcca283e5 : SUCCESS in 9m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/18706b6a70314af7a241103356a2ab09 : SUCCESS in 16m 57s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8040edf098fe4525aa8e8c0a83edd8e1 : SUCCESS in 5m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d8cc6b19fc7a41a79e12a23dde541a13 : SUCCESS in 5m 21s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bd87de88e2464a01aa706617efbb578b : SUCCESS in 4m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/75d3f344eb0444b784892c5c089513c8 : SUCCESS in 10m 41s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f60b4a914226449ab64a5b9aba72ff6a : SUCCESS in 4m 44s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/554c2c1361034d2585e3fd7b8d4c7c84 : SUCCESS in 5m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/159e8c54aa7843d180c86dde0af180d9 : FAILURE in 2m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/efff9abfb7354f3cb7a1b1696ee31de6 : SUCCESS in 14m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/181f4af6fb444ba09602290044d2e72a : SUCCESS in 6m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b127d46d24cb4cc7a66c3e9555e1393e : SUCCESS in 9m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7083006402d4478dbdc8e7df1b3137c1 : SUCCESS in 12m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/09149afa7df64c9ca6a668fc723c80d0 : SUCCESS in 5m 26s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ca62cac9c6994482ae008af97497fd0c : SUCCESS in 9m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2ad3ae7ece234a4c899da11e2a7ebf45 : SUCCESS in 6m 06s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/feae51e76fcd432bb5efe528f3186bcd : SUCCESS in 12m 29s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d57a49d8e2d944729a6961a9826c3c00 : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fd3a6f94f61440f3b101365be493dd4a : SUCCESS in 8m 13s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4e3c070000ac42bcae9b7eacb6abd9f4 : SUCCESS in 13m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/440393bf5c994c8caf1cdae459d5f6c0 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/012f5e153d9448f5bac97fa91bdbb70f : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0218f181237241439b39afaa9cad7535 : SUCCESS in 23m 03s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a5b5045a3ed64216865b4c684b52085e : SUCCESS in 4m 28s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/68349fbc5ec54ca7b566ffccd5e7a19c : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5af40785b3974eb5a307ef47ca4952de : SUCCESS in 6m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0ee30dce1c7d40cabbe8f8cdee0f4a46 : FAILURE in 3m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c256889cf3f4af98fc7c0b5874902b9 : SUCCESS in 1h 02m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15bf8bd8f30e411fbeccb62c35963a4a : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"87f014839e3a8003b0a3701d0add7dce96703420","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-01 03:21:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50d182f8932e22b44496e645c165e32b3ec13ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 04:58:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d111a2b2d7914634a01a8d6999e43926\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f410c547f84149a2bf767be1a9b361 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2af6bf39675441d82e6d9b745beeca2 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/229c2f413a984b40b406d0e2e2725270 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/213434743b244786bb39139aad7a7995 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a26f915c9b854613a1342e1afdc10a01 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9165f49e11bc4c3f92c9260f59a209b1 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69a72395d2e0410d802eb4c7857405c9 : SUCCESS in 9m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c688e7e0c2834a94bd26a4a1c33d46a0 : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/570fe59ff85346d1a985fcb5fa6d53d2 : SUCCESS in 9m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1e5e9586daaf42d9a6cc2693cf26197e : SUCCESS in 20m 32s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/06c0bcb2ae7542a2b1671b991a4fac6c : FAILURE in 7m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/44607b09f70040eaacf594b8b1ceb0fa : SUCCESS in 4m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bb3f4f9a51e54ac4b0b00b90c5b0b17f : SUCCESS in 4m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b94e19ca56aa4f54a1c48f26332ecf69 : SUCCESS in 9m 41s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a579e7df35194676a725b38620372715 : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b8fb97b2cd4d4026a25e151ea3f7b80a : SUCCESS in 5m 07s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7bb4d2623baf4048963254376ea3bd68 : SUCCESS in 8m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2332769a9d404abdad1b677df55714a6 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37de4cb873eb4d778ae6536a36c18dbf : SUCCESS in 7m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ec0670cc94224697abaf9d7c53703571 : SUCCESS in 7m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/db8e9f83c2544992b680ac624888a49d : SUCCESS in 10m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/99acc38f2a8846c0a2ad39487fa80a1f : SUCCESS in 3m 42s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9479dc86eeef4a2db266a78747848920 : SUCCESS in 9m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ce33862120ac4b3fad772f756bb8efa4 : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/134c9160b1f64be289826d8adfb22943 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b50613a1468d44eb9fc14e94f47488cb : SUCCESS in 21m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2bdd03e04da04c3498d79ae8d4aa300c : SUCCESS in 6m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9295f0df5bac4d4b916ceba1cf7e14ad : SUCCESS in 13m 02s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5a6ab9068e7f425c828e6fa4fe568a69 : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1142a56f44846b09eeb5081977d46f7 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/236afa87a0df434a9e6873806c0a656c : SUCCESS in 21m 42s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/bc0f514ab6444d3bab9ffd96cfa690de : SUCCESS in 5m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f8d567d9a2034de1823483287cffbe19 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b10471ded6445d2a922a5006a5e32a4 : SUCCESS in 6m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/dc4168e69a094b9795a5abac3d71935f : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a755950a50ba430cbcb47c9ec8bf64cc : SUCCESS in 58m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/652d79d4a7a24bfd8b38cb4f86daeadf : SUCCESS in 1h 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b0278892abd772111adf67d05b631c4f2a5f867e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 07:53:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c55a83ce7b04ed9a979eaf84010dd5b0a33086c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 09:34:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/003506de601f44458e87a9366589a8a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080127e31f16434b963198b593904142 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c87efc45acf41b0a0a5dc51e72ae77e : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3485ae13d9b54f9f9e3a9a2a4fa894d7 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e9f4c7ce02e40d78d80cff8e20dc8a7 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf3c065dc23d4a209ebc81d8029f6a11 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a38f84440b9745c3a6c67f429180c277 : SUCCESS in 5m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8da8af9f8da84aaca2c8b2319c2ed4ec : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f2a05904d2c142bcbf7c0c0a58dc5fdd : SUCCESS in 15m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c26ff7dead584f24b042822628088e21 : SUCCESS in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1be4472c8afa4e27b54d83066993f19b : SUCCESS in 17m 19s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/56033b6563644463880f22d565e92bd3 : SUCCESS in 6m 33s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/aca8cdc10ff8489c9b702741894f56c3 : SUCCESS in 4m 45s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/3a753ea39e29423faa6cdadab42ccc21 : SUCCESS in 4m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/27560da097364e78ad5cf0f8c3accfed : SUCCESS in 8m 55s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/3dedfc53d52a47f591d62240407d138d : SUCCESS in 6m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/321baf144674486bb339632ff3b37286 : SUCCESS in 6m 15s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b57828aa844941c2b43dfe794df7fa6f : SUCCESS in 7m 38s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/021c3d24641745e5b6efbdaaed0b32da : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbbdd663e2f449aab2cddb939427b96e : SUCCESS in 7m 43s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/18110e5dab974025ba263a72ac6ed34a : SUCCESS in 7m 15s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1e956a4cacad4cd09d29180b71941f32 : SUCCESS in 10m 23s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/30c5ecd98c8c4a7d9200170a28d331ee : SUCCESS in 4m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/166634e9881140a0bc6060290a9be759 : SUCCESS in 8m 32s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b20edb23cc254f4ab9f92a7057a29254 : SUCCESS in 5m 59s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7a3f2256ccd74dc182b6ea7c58cf8dcc : SUCCESS in 11m 17s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/aec7d28993d84894846c299779f9cded : SUCCESS in 19m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f6a2bbbba08b4931b6a637ca8fff0ea8 : SUCCESS in 5m 27s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/585f61affc9241fdab5eda9deca997f6 : SUCCESS in 12m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0a1647a9f2c24c85a2647af82b2f5c50 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bed73a6341644257b3669d51c7b31da9 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3ec774c251e441fbd05c85e922a816c : SUCCESS in 21m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/04e73de6d7fe47ed972ea97ea20a6ef3 : SUCCESS in 3m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/df234954754548a895d73ff9f07c47ce : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88cc0b368c7a4e9a9cfa2fd2942b7e16 : SUCCESS in 6m 05s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/555269e6829841f5bfaa33d9ff5b03c0 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6778d25e71d847c8b5fc05be16cbcc5a : SUCCESS in 1h 14m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28472372af024c23b1391bce7b6c9253 : SUCCESS in 1h 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"58fc11cb12b0dc246b29ebc6b2ffea9735edce51","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-03-01 11:44:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53a0618d8f983b02dd804870cdbe0ab6ad81ae99","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 13:44:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b9ccced1db0d1bee0391e9a9bcc885adf8cdccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 13:45:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d29fa2f7d6cc0d97640759d7db73733bfed4e62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:30:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2e9397ac0a43369963d6077e37041f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5de1dba9607478c80bce01d05a823c7 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1c719c24b24f80bcbd2144822ba133 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/955d715421f3412292632153a6097d67 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfada407405d499d9b464f4e5ff7d9c8 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18bd4625802542ab9644c6d5269baab8 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac8128f2d31f4261ab4ba56cf5828c47 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3431e2daaa4e49aaabcda829412c451e : SUCCESS in 8m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/50311f75a11a430998087caac8c988d6 : SUCCESS in 17m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f958285288894521971db489cbda3611 : SUCCESS in 9m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/251db9e1c72841939798dab571c7a3b8 : SUCCESS in 16m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8bac74dee92c472595f21102f4d831fb : SUCCESS in 10m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/12cf352fbc854c61abebcf80d943372a : SUCCESS in 6m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9b6233b8cf574daf85a157bcfc7d5bd7 : SUCCESS in 5m 59s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6eea2e25bb4042dea7d7c422cb09e1aa : SUCCESS in 7m 41s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/831b72994ad94005bd559881815efd07 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d012cfb5263472c89907a40db5ebc33 : SUCCESS in 8m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d2b4c1dd19724a09bb6f8a648b0be5ec : SUCCESS in 8m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/295a4e5eab8548adb37c85930cdc18f5 : SUCCESS in 10m 53s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/dc109d79a5d84d5fac50d69db42a506b : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f9dc6ab9422849bb853e78b355ee9afe : SUCCESS in 8m 56s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/15977bdebe9e4d9ba173f573a1adafd3 : SUCCESS in 22m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/809b16c87d4645568a72181480d3f820 : SUCCESS in 7m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e7f8d5e351304cd5a7e2496cb47d3a78 : SUCCESS in 11m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/364f429b4f4e4c0f9c698ac30895a02c : SUCCESS in 6m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58205b43bc6b49648409eb3604a0307b : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3189beb25fdf4a708df4c51fea42bb7d : SUCCESS in 21m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/860e3f514b9445faab9501d2b20c54c9 : SUCCESS in 3m 58s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a7e0118874f5451c8d85b4408eba1499 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76d738535789406fac6b5f23c5ca5905 : SUCCESS in 7m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0d1d3fc03d0a44c5a4ab1684287daa9d : SUCCESS in 15m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3a486d2362b4fc798402e8b89036258 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70449ed4f6f44e1c881faebe4bd5de7e : SUCCESS in 1h 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dc24a4737b4469814bc8f2fdfadc43bffffef9d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:30:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7ab4769b8cdeebc5956a26b41566e42201c8c19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:32:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2dd7bf07def4b33be71cc866dc74233\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db38357f7aba4bb1b1b8e39622eb3c81 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20e1f1704aed8c152f3641f54c47c8f8f67a1cb2","revisions":{"20e1f1704aed8c152f3641f54c47c8f8f67a1cb2":{"kind":"REWORK","_number":1,"created":"2024-02-29 13:17:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/910599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/910599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/910599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/910599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/910599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/910599/1"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 13:17:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 13:17:14.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 4.5.0","message":"update constraint for python-swiftclient to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta: release:Co-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta: release:Change-Id: I511fa86c21ae2087129876804fe326b517ba1226\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1eaa95c19045213a7cc7f801ecfa18673b2376e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20e1f1704aed8c152f3641f54c47c8f8f67a1cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20e1f1704aed8c152f3641f54c47c8f8f67a1cb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
