)]}'
{"id":"openstack%2Frequirements~910639","triplet_id":"openstack%2Frequirements~master~I81a712de5d1b51c9ea901b825f248340ae8c6332","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I81a712de5d1b51c9ea901b825f248340ae8c6332","subject":"update constraint for python-octaviaclient to new release 3.7.0","status":"MERGED","created":"2024-02-29 15:34:57.000000000","updated":"2024-03-01 06:50:58.000000000","submitted":"2024-03-01 06:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910639-new-release","meta_rev_id":"1a4813e6f7b97c433f2f50702ce60ba09982c7a4","_number":910639,"virtual_id_number":910639,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 06:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-03-01 03:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-03-01 03:22:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 16:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 03:22:44.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"b880de3cd723c94e74a3c30a15ee0f36a17a4151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-02-29 15:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9054fe95864baa6a4a83a39dad4789ad039551df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5faed3aeb83a44daa61c29afdb4421ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61fe127d64c0419ea692b2deb4c9405f : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6785d8c5610428ab32ec89c52ddaf72 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f305d335f064396be866692ae813ad7 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea7298e586af4f7b8a53a49b0d96d2a6 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffc31772640c4f2f93669a2296536d6c : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c34802c1f564382a5079cf7679b5196 : SUCCESS in 3m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85ac5300fd2d4c5e8d49944eba451bdd : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d5203a6834bf4fbbbf43577461a37be9 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8aba96e6bb7e4980a0f5f5fefb48d7e3 : SUCCESS in 12m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/786b2009920b447a9af62edcb3d79f09 : SUCCESS in 13m 28s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a0d71f37422446da913eec931dae8a74 : SUCCESS in 6m 36s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/385e4deaf5b142ec8d00c96e7f409efb : SUCCESS in 6m 15s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/54d04d7986cd453aab97af58398bc70c : SUCCESS in 5m 40s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/65923f4052614ee3bf4d7a54f8ffbca0 : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/83847d98c1514517adcaba2bb73e4e56 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ff733fb649b44c78ad079facb1fb68cf : SUCCESS in 5m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0ea13d25d71b4fb8bafe3e28d0a52d54 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/512de37000be418c8ea18dffbca89b69 : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09718ba142a6453ab295b1386072fc18 : SUCCESS in 6m 55s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a709564c39c042a1a1183c7792aec388 : SUCCESS in 8m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/429193605673432b929f6f4e813154c7 : SUCCESS in 10m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7f9810d0c39e4169afc0c823e6baf3f6 : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9ae72e5d2bcd4ceaae0e54925a0f665a : SUCCESS in 10m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/21b8eb6c3e24476f921c324393d2c67a : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b8f341729b2443698bdc8d2b4242b888 : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/639f34d1bff04517b7f99050fc194298 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/12311dac40d54abdac9921901d811fd2 : SUCCESS in 7m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/720dc2cdf8c64d0887f0ba5c69e298a1 : SUCCESS in 14m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8bbdb08569f4491295c972fbb2b3f8f3 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/703daba5bfd44bac8e79d27b4c925dd7 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54dcd36bbf5846f084e7d63dfd19bc33 : SUCCESS in 23m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/72374071dd6d464eb6cb1c7499595d76 : SUCCESS in 5m 55s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2a2af999a9664c72991986725c4d7df0 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b6bf333ffdc4b1095cfa76aef5724a4 : SUCCESS in 6m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0db8e60484764631a22d5319e446ab1a : SUCCESS in 14m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb84872be819457bb39cdc5606e6cbe7 : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97dfa96b65af47788df7d8d712558054 : FAILURE in 1h 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8a9cb29aaefc22a092f1a08ae605dc067b315063","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-01 03:22:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d2b4a3d5e6c347c83607f4310a8836015e98603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 04:38:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa96db97ab546fabe9c1810011f9dfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff6c67bbe1f48e9aea5f7a0b653fb79 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db188635b933433494b7d594aafdae65 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21d25d9a31d14f9d8e0b5f49a1f8fd92 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4d4fc2f0c294ee7bdb8c43573402f7e : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cd0710b21de4044bf630c754c72d7ad : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95a2f07c329d45c4bf1e2bb0b81f4071 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a8a7e901eee48e1ab0507c3df6637de : SUCCESS in 9m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a2739fceabf545f3ab9512140a61b6a7 : SUCCESS in 17m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c590fc17c60d4197acfdbb9b894483c4 : SUCCESS in 13m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32a81903bdc54806bdf70fdca8b94d72 : SUCCESS in 15m 57s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e25ddb56d37c49d390f56d23d803aa59 : SUCCESS in 4m 35s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0faebd15ec474d5aa57490a1ada93bd0 : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d7c6f374bd9442fc921cbb08c490cea1 : SUCCESS in 6m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/98c4dce1c4b440e9b6bb2536b0eeec79 : SUCCESS in 10m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/06488c14ea1e415db054424864b0f304 : SUCCESS in 6m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/de2a14aaebe14211aa9cba092f6d0259 : SUCCESS in 5m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e730b47109da481f8857131674a21f72 : SUCCESS in 8m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/eb6c49cb304b4be79cb3dce38347e3eb : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8a10abc718d42ec96333846dde9ad58 : SUCCESS in 7m 42s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2972dc5ee842489c987a0e236bf8b3bf : SUCCESS in 11m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5eafa0be59de453a8fec47699d170963 : SUCCESS in 12m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/fe81cee6e75e4ac483915392d7080780 : SUCCESS in 4m 06s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/fb2fc1c289a74a6f951bf3d3a8ca87d9 : SUCCESS in 10m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/51b61ff3840244c7abe1628dd13b88b5 : SUCCESS in 3m 59s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/dd081ae1c52a4304bdc5123d0684802e : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b2da23cba2b3423e844690ee83ea7974 : SUCCESS in 25m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e3f0f253c304fdbab2f8cef39bd750d : SUCCESS in 6m 07s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8c15ffbca84146b89c708dccb4ece11d : SUCCESS in 13m 43s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3932a24d9d844248989c9dc6b22430e7 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d66b6d59b234b4eb76d81287503cb38 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3cb75aa40554ad5b9112ee075e191b7 : SUCCESS in 23m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ba9d49a2bcc7419298a9cf78d2639f62 : SUCCESS in 6m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/91b31165827747ed80d6b1dc384fb9ed : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0635b2a2c3e443a384a6a2c4896e9291 : SUCCESS in 7m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1f2f85ce8fd146e2bcfe7f6cf58ff957 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6de2c177ea6e47d1b67b84e567948249 : SUCCESS in 1h 03m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7dfc942b8914706ae77c614470f9296 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"db23921bda6df6493645d8c75bb14c2c737a2f1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 04:39:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e349e61c4ec84a087dee627d65b2c2307781d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:49:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faa374f5406d44eb94e88de88da93c81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29df01de29a64522bbd7030aac5e5ce9 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ebf420648d84de6af4704adcf805392 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d3d57af41544ea1b3af598d9d71ce79 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3be5d5aa2fb44a96a9a638527a896354 : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43d5be830fd244009d78043bc1eb7691 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccbc84ab1ecd47cc85df0b011f0b9d22 : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0eef9de8bca484bb0654af96c548582 : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/54d65dbb8eeb4c948f743276e33aa551 : SUCCESS in 14m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e787f900f0d64d5cb541a7139be6bde8 : SUCCESS in 11m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fb156deb3087416f88a64aea89de86be : SUCCESS in 18m 02s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4b0599140cf94cd3974f6928bca669c3 : SUCCESS in 7m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/14b3566d665e467e8897429c9331e190 : SUCCESS in 4m 42s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fb6c1d9d98854c8e8b9810b7b484b98f : SUCCESS in 5m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d6fadbc1b10942ccafdf9dd005767e1c : SUCCESS in 6m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0ebe74083a2b486cae7c9a03f6e8ac70 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e471d66d51454838a76a5c54562a69de : SUCCESS in 7m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5f09f629a2fb450bbd76b0a90616fc24 : SUCCESS in 6m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ce87dd8673af4dc8ba461157af7f6e22 : SUCCESS in 11m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/24f1198d8d28432b9b392803e9ece912 : SUCCESS in 6m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/33bf721ffa334e1993a3763ceeaf7e5a : SUCCESS in 6m 31s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/fc3d79f99f004fe5ae68b6a88e784cd6 : SUCCESS in 24m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d1b63f0356fb478bafece8f8f0f3569a : SUCCESS in 5m 59s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/84ab6b93128544a8b96fd01ab5f0454d : SUCCESS in 9m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fc1e3213ae134aefb02a4fdcc00714bb : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c4f38a79d8c4abbb50f774ae3f09bc8 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67d302403d634462bf49c9461ff3f5b8 : SUCCESS in 23m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/af96fac43456482c900234612b8cb598 : SUCCESS in 3m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2b62ac11eb49446a81680d3437999bc2 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e984de3470814193bfa744f913754892 : SUCCESS in 6m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4f89d1de79914d36948e429a125d2057 : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/332093bb63b94bfcae742d2c3d8f3ccb : SUCCESS in 53m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/569212a3d20c4951bfe0da31b77d1bd0 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"472ddffe7166f84174734064edb43249a8d6d91a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a4813e6f7b97c433f2f50702ce60ba09982c7a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b44d59fb9740ee905ddca09b0d2d58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb1700ca1ed49b9aed6a460c40083e4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"484dbd107a3cb03fc067d46621491de63d5d0c4b","revisions":{"484dbd107a3cb03fc067d46621491de63d5d0c4b":{"kind":"REWORK","_number":1,"created":"2024-02-29 15:34:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/910639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/910639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/910639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/910639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/910639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/910639/1"}}},"commit":{"parents":[{"commit":"30a749770e423696dc53764d65c6fb9ac3d0a78c","subject":"Merge \"update constraint for oslo.db to new release 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30a749770e423696dc53764d65c6fb9ac3d0a78c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 15:34:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-02-29 15:34:56.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 3.7.0","message":"update constraint for python-octaviaclient to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Change-Id: Ibb756b43e8bfda236587aaf35a69e060e815e8b3\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nChange-Id: I81a712de5d1b51c9ea901b825f248340ae8c6332\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/484dbd107a3cb03fc067d46621491de63d5d0c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/484dbd107a3cb03fc067d46621491de63d5d0c4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
