)]}'
{"id":"openstack%2Frequirements~910709","triplet_id":"openstack%2Frequirements~master~I51e9168626365eb1e31e5bb805fb77ae32f24303","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I51e9168626365eb1e31e5bb805fb77ae32f24303","subject":"update constraint for python-heatclient to new release 3.5.0","status":"MERGED","created":"2024-03-01 09:21:46.000000000","updated":"2024-03-01 15:22:49.000000000","submitted":"2024-03-01 15:21:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910709-new-release","meta_rev_id":"28f378b88ae4fa12f491673726a5e8c4babfad8f","_number":910709,"virtual_id_number":910709,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 15:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-03-01 11:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-03-01 13:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-01 13:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 10:37:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 11:43:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-03-01 13:42:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3a70f09266a4957ab250808720f4ac131f7df1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-01 09:21:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"016d9138127278f1cdaecf641b31c6998fdace0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 10:37:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/229c51f1f29144bfa04ed971183d952d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedd93957cc64d56a485d24e81d146be : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b207e3133316410085e631d0d444734d : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fddd3c6115949c694e4b4211b243cf5 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f772b657d2c3403ca2f4fc40776f6ef3 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/befc6c1d797a4fb5ba9215353246f4e9 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb1baccb167747849cabef938bdabf70 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7812ccc3788d4a5f9a8bcd7b18ce3096 : SUCCESS in 8m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/52d5f1f26b0d44d78166dbce52bf206a : SUCCESS in 18m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bb2a97c3f9db468dbf7999e7a19b8e59 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/784a4bec09384b42bc5cb503674e911c : SUCCESS in 17m 26s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5ab7d5676d294d9da2e5ce365b292824 : SUCCESS in 5m 04s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/85fac5ad61cf4e3f90fb619ff4f14d5e : SUCCESS in 5m 06s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/790504e076bf4e90aa5e012a6ff51ba6 : SUCCESS in 4m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7418e88deb9f4e77a95762faa20e3d1e : SUCCESS in 13m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b6bb57bea25d413faac6288dd3b33e7a : SUCCESS in 5m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5110dcac74174fe4863c4e81b323267f : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/bef353050af34ad79a5dbb9ee39f368c : SUCCESS in 8m 13s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b02752d68e994cbbafca338bd320c778 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40e25623035c48428abf860f407f0f57 : SUCCESS in 7m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c2e2f82b88a94577b6b24d7275628dd8 : SUCCESS in 9m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e9435a012f1e420d992798b754751140 : SUCCESS in 11m 24s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/997ab25f9df44e839f2f315d6ea25d3a : SUCCESS in 3m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4cf58defa5444a85829dc83e5a43d0a3 : SUCCESS in 9m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/73237512ab474869a253b7078206df9a : SUCCESS in 6m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a9acf9c4c6d5481c957bb6ef4ffcce39 : SUCCESS in 12m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c262524a576a49afaf27ec328a460ea7 : SUCCESS in 26m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d545a77221754698a95d394c257825bd : SUCCESS in 6m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/cc1a58153d934bbc93145409915057f6 : SUCCESS in 11m 52s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c5970e9971744eadb2ab83f50ca9e8c6 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52c010dcad704690988cea331cb71b06 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eceed967f6354d86a748a8f5a93c5028 : SUCCESS in 23m 26s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/43cbc4461cc8463da6581ed829bac600 : SUCCESS in 6m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3eff4e7d363f4821b2a648c4f342457e : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/857a4fc221a44ed9bdcab84ece99a807 : SUCCESS in 6m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d83e2d4c4c7640d6938074dca5a10508 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c1da90afc846e18da1cf42c94b4a6a : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/996ee8b9e8534509b8cc2c8a4f2c1750 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9f58a998fab495473fdbc576f29080572b174a58","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-03-01 11:43:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6ea2a984415a0cb5d306b9fc977329596ea69ab","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 13:42:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"466d90491a95f8bac15bd94e6460a8010cafaf88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 13:43:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f041ad147ec70df9ded35bba229b1f8b3c70d02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:21:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d9b91855534acdaadd77f5ae59ed8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ec3a3926c449c2967f251e505c3984 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9b94c3943744a38af0f14075a6dd88 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5527a510bd7c4a0b9c00cc9ea1593cbf : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e30e05343f814243a8b3dcbd83b947c3 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f7228a181544ec5bcf24d7a47c9e591 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8d03a0cffb64dd3bce7813728d9b663 : SUCCESS in 5m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f18a1f6663f4531a9d8ce889cfed01f : SUCCESS in 10m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a4c871b5360d42e082a487e5f24bd54f : SUCCESS in 14m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/53c628e58b74475f90ee0344667b850a : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9cdd1e800a56490a8df45c24e6a95457 : SUCCESS in 17m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9f79b75120584c59965e76535a8d9289 : SUCCESS in 13m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/66b21c6ef3c24ac98100552a116d4ad1 : SUCCESS in 7m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d656a32a6fde4b5cb7f49a6fa3bd7588 : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a9b5380234f34506b447927ee86748e2 : SUCCESS in 6m 44s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5a1149f60d8b4cad90fddc55f11325f9 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff5cd25f29ae413fb012a8c09b8b6d88 : SUCCESS in 7m 52s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/24806cf522474cdaafb9f9835ac12f18 : SUCCESS in 7m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d22bd5aaad5b4ff4a4eb11343daf9411 : SUCCESS in 11m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d5c46afe71464ea49acbfe1f45c5a337 : SUCCESS in 4m 26s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6940c4b6348e4ca69f59777d2b3905a2 : SUCCESS in 9m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e7254db1bdfe4879b4f6f6907d518a9a : SUCCESS in 20m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ce3fd3e1223546cf93b248055a2bff9f : SUCCESS in 5m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e08b25a1c30341cca5c6b1bcf32a150b : SUCCESS in 12m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/eba24ab6b4174305a55727ff65dfbf2c : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5cb76dc281d4df7af128f136d53ebfa : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/928524e0ec8e4a0dbba9385cbb7ccc83 : SUCCESS in 23m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/56f4a127938241a49827b4a2ca12784f : SUCCESS in 4m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/72e24f662928416fad1c82432fa2b190 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a0fe7bcdcc1463fb195d597381ae24e : SUCCESS in 6m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/fd91ccf24a4941c693a1c11b2e0fe719 : SUCCESS in 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4720a941099a42aaa1ddb2cedfbf243b : SUCCESS in 1h 00m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/146c974a8133413d801c198189df4376 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":1},{"id":"846f14ac9c21d614b92ec1878100f6eaf0849947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:21:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"28f378b88ae4fa12f491673726a5e8c4babfad8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32d76da899b34c9294a9db7dd8d72af9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6188213e96cb4bcab96373ce7e2e201c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2aa955e2b711eaa98b3083dac3f1388c945f9112","revisions":{"2aa955e2b711eaa98b3083dac3f1388c945f9112":{"kind":"REWORK","_number":1,"created":"2024-03-01 09:21:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/910709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/910709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/910709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/910709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/910709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/910709/1"}}},"commit":{"parents":[{"commit":"f4ca2053bddcd19e2025dc4f28dddfe045cd7363","subject":"Merge \"Remove Xenial and CentOS 7 wheel cache build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4ca2053bddcd19e2025dc4f28dddfe045cd7363"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 09:21:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 09:21:45.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 3.5.0","message":"update constraint for python-heatclient to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0c3e084a00046b335cacbf81ce18cc987aa21927\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I51e9168626365eb1e31e5bb805fb77ae32f24303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2aa955e2b711eaa98b3083dac3f1388c945f9112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2aa955e2b711eaa98b3083dac3f1388c945f9112"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
