)]}'
{"id":"openstack%2Frequirements~910713","triplet_id":"openstack%2Frequirements~master~I44b1cab5d4a05558a40c3d80e47d327aecbd0ce2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I44b1cab5d4a05558a40c3d80e47d327aecbd0ce2","subject":"update constraint for python-manilaclient to new release 4.8.0","status":"MERGED","created":"2024-03-01 10:03:48.000000000","updated":"2024-03-01 22:32:21.000000000","submitted":"2024-03-01 22:31:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910713-new-release","meta_rev_id":"9682eb43d5907f4cf91dc238cf57d8c76653a6e4","_number":910713,"virtual_id_number":910713,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 22:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-03-01 11:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-03-01 13:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-01 13:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 11:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 11:42:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-03-01 13:43:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b540347552d5991488ea9cebee41886a62d4a6de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-01 10:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70faf439f7288116b8cae70d262ec9926fc67ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 11:41:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d2a9275315642ca8dacff035bdcdfbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a797bf634a3f4890862974f7086ae102 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad5d780a5e848b2ac7ed2c033de8b51 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43f6c963b40c4972a811b36ac02fc347 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c32a365bdcd14c51b0b85cd74079a6c6 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89d4d039458845079e23710d206dd56b : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d44b66d30eb14ea2bb4d97fa37bc2efd : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1d2a68a7ccb942f3b5781f0722c03d29 : SUCCESS in 9m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d1c15b7c5c4d4e03bff75282e7c7016f : SUCCESS in 15m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3323754e92e34bf18d3283c848d5aad4 : FAILURE in 2m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/10c8dbbedd384cba8bb88496e8f8e805 : SUCCESS in 17m 57s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ff09e673f7a546bc96e94e33a4f9d637 : SUCCESS in 5m 09s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/72bbe1eb5fc74c72b218901a71a70207 : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/45274afe60da4fadb966a036090b417c : SUCCESS in 5m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c778bef692fa44d7a02f241be579b34c : SUCCESS in 9m 53s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c6c720956a974da88a2916ab8e8e5479 : SUCCESS in 4m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fda0b507cb5e4bf0ae0e243fa3c21290 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/82ee1feeb0b24899a0fc6ec53574a132 : SUCCESS in 8m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/69fe181de04245338fdb5bedae2cbf26 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d90fbe413114ffc8da6ffd9fc1fdef2 : SUCCESS in 6m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0b7978d3f69e48969fba09b9136f4aa4 : SUCCESS in 10m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9d910dbbbca0446abf02581252fd4b7c : SUCCESS in 9m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a76c0654391a4d8e961a796053ffd3ea : SUCCESS in 3m 59s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/72a5ed7bec8e4cc6b503eecef78b3c3b : SUCCESS in 7m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2d4a473c7e5c4f0d90439233d6fd5311 : SUCCESS in 6m 37s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f7e7ad9916ac47129176b9979e29d3d4 : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a834e2f8a553453cac0062b80eb27c6b : SUCCESS in 19m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e9c46becc6144d879a6726f95a27b8dd : SUCCESS in 7m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/98d444613cb24875be8ec74065de5c0f : SUCCESS in 10m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4a580cad8fac44dc8a3aaee3b512a8c9 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/969bc1722f4c449e8b310cf23f618167 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42f848c35e914b4c814aceb6d9efe602 : SUCCESS in 23m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/98c0ac25b5bc4d9db11b9baf333792a7 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1044372987a64aa2895039c47c41242c : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc36311744884f258dd36579499e4436 : SUCCESS in 6m 02s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/06d669d5c2de410fabe73323d0169b9b : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47231dd906c4400d8bfa48491778da0e : SUCCESS in 1h 07m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b907a6c510fe43929b59216580529dff : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"aa105ff5e7c27859d5bd4e6904d76106439de392","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-03-01 11:42:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1bb140d92309b6ed4ff991a96b940c835da73ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 13:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e937a39bb994eee9e638096b6bf8390\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28375bd20d344e1cb01c9794b690a49b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30090e214bb44d7daf5a9a2b757800e3 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95028105fb1448889571661d68ea3b1b : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3259b9fb898a4e3fa8967fdc35bb59de : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aa1a1e045764a2595e43013a0ce78dd : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cab8d323c4564e3886fb371d4e287a6f : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65502807b2cd44c9aed09165d5909fa5 : SUCCESS in 11m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4968821679c54d31b50b97c5a7cf1a0c : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0287fee033ac4bf8bb4ec642a007f4bc : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38bc1d451cea4b11bcc17994fca6c9f3 : SUCCESS in 17m 15s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ec28dc8d631e4ee482608eb08ce26373 : SUCCESS in 5m 21s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/94b07c254732401e9bb674b7094f49a7 : SUCCESS in 5m 41s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6cf2c617044a4d28b33afe7dd23682a3 : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/cf7d0c69e0e44026a027b722951c1458 : SUCCESS in 10m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/73e46de6cc16434db16bf07981831c11 : SUCCESS in 5m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/62634db1fb6d43c29ec51783d32c83a2 : SUCCESS in 4m 47s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ff835ad40ec94d8ca6c22072b65ae719 : SUCCESS in 8m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fdf49459df3a42098fa32838e9299924 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19750e39339c4ef999eafa1d471a413d : SUCCESS in 7m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/40d25ebde71847c69a3293f463174cfc : SUCCESS in 7m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f6a626bbd44545e5950c89d7e2ac314e : SUCCESS in 9m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/11b0b4f7b6df4c28a8ccc9416043de02 : SUCCESS in 3m 57s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a1903ba74bc748fbadcee36bd950e4c6 : SUCCESS in 8m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5e1a77f0ae884c1f80a9a7148db90235 : SUCCESS in 5m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b1397cc092d648f89230a7cc982eb285 : SUCCESS in 9m 47s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c66af7c1dce14cd7957273a64d9a5ecc : SUCCESS in 23m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92ba68ccfc3d4fbda4ccd8d8285daef2 : SUCCESS in 6m 00s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ec5ed0d7595d47acb42f68250c705f47 : SUCCESS in 11m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d710f26269294f379c2eba61014f1843 : SUCCESS in 3m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80796681c1b84e408c8535774126c3bc : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/950492cf038f44b580c57bf5d0edcbee : SUCCESS in 19m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/71e9540ed073440cb799b4a203606051 : SUCCESS in 3m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c2f36198cc1d4bd18ab13f227122c1a3 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17b9326107c347299ededb1b460828e2 : SUCCESS in 5m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f30c1d1845cc40699792295eb5819787 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79de727f1dda43e7ae636a769bb4485d : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e123dc051611491b8a94cbd9f9de26a0 : SUCCESS in 1h 37m 31s","accounts_in_message":[],"_revision_number":1},{"id":"be3cde5a5991f29c1e47e1c9ccf3fdd88e74c577","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 13:43:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7037700cc9d9229feab7642804656d3dbe4d0445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 13:44:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f697c813f29b30d520d571e7ffef9c5cd76f512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:21:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/feed789842404e938398dbcffcbc3da8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c69d944634f4f7384e590939e15ae2a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f883227aec4d7cb26ba5686149c36c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c2841821bc64111995c38e9bf07ae8e : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d19a19d15315463188b750fcce3e36aa : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11b01b20638249e182d3c04923090d1d : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb01680e76ad46c7b367b33bf9557593 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/092507b41e21434e92268e13c562712a : SUCCESS in 9m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05e3550609d144ebbc69b06c5cfd6a34 : SUCCESS in 14m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fbb5a38a93ac43fa91c18562e19c744b : SUCCESS in 9m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0e3abd4d027462782dbb34bca3133b3 : SUCCESS in 23m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/65c726add7754e588420f51ad13d3791 : SUCCESS in 10m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a8ae324149f34832b33235b4aa217cd1 : FAILURE in 6m 14s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2e7003df604f45839c5881e1ef15f41f : SUCCESS in 6m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/84dcc44a2fa2486b93b305d4b8910fd3 : SUCCESS in 7m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4e99773b647448d79d7d7662eb1ecfde : SUCCESS in 12m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d380183968d445099c19fd01782ccb3c : SUCCESS in 7m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fcfe1fa7488a49a1aafa0872361c3f02 : SUCCESS in 8m 10s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8e998fcb998b4128a48884e0c25150b2 : SUCCESS in 9m 51s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/99578db8e8f14fff8cac591bd2fae318 : SUCCESS in 5m 42s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ea2caa38b05144eeaa3ef311e70ac265 : SUCCESS in 9m 57s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/439bc5c49b1449e8866df40f956c1c27 : SUCCESS in 31m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e7cfcdcdbe6e4aec908818aeb857cadd : SUCCESS in 6m 32s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a12c73c2372946639abddcac61595eee : SUCCESS in 11m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9b1dfc16c0114178828f42c43940d1e8 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98064854eda54a7fbd476b16ec058e28 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4eee2bde0698431081ad8ed334e5da32 : SUCCESS in 21m 31s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b3896450d4d448ffad8a09090c719a95 : SUCCESS in 3m 52s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7382471b48a94ad9933f7b0acddcdd98 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fcb7bc17691e46e8a5f3b446980287cd : SUCCESS in 6m 19s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b0935986af43420f8b75aac406834e74 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9957038bde64afbbbf7c14d024ca830 : SUCCESS in 56m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0913a840e50a48c0958239a9477339af : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"515e25666786c0f823552af2e5f68b36483eb979","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 18:24:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04b398fedbcf28be1ee4c49e529989c1eeced4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce90194e18849ec95a98f6090ad81c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cacad8200442dab213d8ccd3d40589 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9928b6b1690145fd8a36ad9f01cff256 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de46e4c89b8e46e0bf166e7e42a46d7a : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ea16e4ed7cd43baa3b19246f74e679c : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d646b08c7754bc8b565cad833661907 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/256eb4f4bf5744d2ba8ffc48dc6da5d3 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2602e05fa06a41cba9c3527ce16c5c75 : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9821f04e9b14cfd85a9eefa6f4739c0 : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/690eb5cc64b244c68b7f491c1a301e22 : SUCCESS in 9m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbd95a59e762490f8f52af842fd0d6ab : SUCCESS in 20m 40s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/46fdb8bade6f41bf826e1847430c9935 : SUCCESS in 4m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a8b03ec472de45eb90437ffffea8094e : SUCCESS in 5m 06s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a7b62ddecaa54ed783d937b22fb63c99 : SUCCESS in 4m 28s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/70af6c02bae44d5bb82eec7b5b958ed1 : SUCCESS in 11m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/57b0ecf5ce4b4a36b5417689e314c372 : SUCCESS in 4m 58s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/57801b8907a541a285bf16e23ee14429 : SUCCESS in 5m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1d718f8cbf5b463f853dea87882fb2e7 : SUCCESS in 9m 10s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b8c8c2c2f9a846b0a0ec607c8f6dfe0e : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6e5a579fb6d43898c153531d82e2d12 : SUCCESS in 6m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6d2dfe2161f44eeeb111d86169e4258e : SUCCESS in 8m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4cae90b80ff8488f946d520b3e3dc300 : SUCCESS in 9m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4e67a6bf9cd54057b818cf103a286d9b : SUCCESS in 4m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bbda72b195f4441caeeb9fb3d612d325 : SUCCESS in 7m 54s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7e4e818a438f48408cd056b1d6712afd : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2adf00afffbf4cf9ac05448c9f1c8623 : SUCCESS in 9m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6a1ef8e1a7df4158930c3f4da66d8ee0 : SUCCESS in 23m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7f0ce2a7427147d5abddd6b262430bd6 : SUCCESS in 6m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b6f9e56f186947b9af5e5d1e8a2244ae : SUCCESS in 11m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/51faf0a1c68d44059396c94377fa9e83 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aeac30faa8ae44af8494cba555e88265 : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56cab145d7a04a0a9ea4c42810b885b6 : SUCCESS in 25m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7db0f09c1deb4cf08ed80096efa6dd3c : SUCCESS in 6m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/81c76cc016d34d0cb7e8437119cf9053 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d88ac83a68e5476a9a3ca43fb67c43ba : SUCCESS in 6m 11s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9dedfac881d540ceb1df6f9ed3ffc1a4 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d7261f30aa4c66af377935a52ccee6 : SUCCESS in 48m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a7886e8bf8941829b06085f9bb527f3 : SUCCESS in 1h 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ccbffce5e0dc98a9a4ef2cf4fb4b3645521e83bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:05:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52b8cc7d0032d8e242cebe43a0378796d6480273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:31:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221aedf0d7d7408c96202b2c78099241\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1363b7a4a7443da8d7774a4f6efa8e : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43953838f2e541dd868f4eefb381bb2b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79cc796a3d8b4e9cbd66bc2521cb49cb : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bea020e0757047c3bf39c74c0a808a69 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8dc8105058794735b4c246c283359890 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e5d707842054c6184a50296e3838b04 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38fdfae7c60c434ba00a5b07c71bc029 : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bfdde9e8e61245f9b200a71c7a655bfc : SUCCESS in 14m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ebaf63c53ae348828b91ae95d0d62e1a : SUCCESS in 8m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/409ae9bbcef74af8b6c58ef3d9ba2b4e : SUCCESS in 13m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/870c495c931f45048202f87b36cffec4 : SUCCESS in 8m 28s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6cf45bb2e8d4419290f74c9520f03c6d : SUCCESS in 4m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a5cc79f1c19f4c528a1f47e2be229abb : SUCCESS in 4m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3f39bb5e2d7441c9b1d9fed0e7cb17b2 : SUCCESS in 10m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4c68d392138845fc8a7787a4056e38d0 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c16f0e3782b249fe984b95423683e1dc : SUCCESS in 7m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1508b93031a04c079f52822b24da7eb3 : SUCCESS in 8m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0184d07b9545445ca3ed93ea0da8583c : SUCCESS in 8m 46s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/43208eefcd434c8f994317d93ab768d5 : SUCCESS in 4m 21s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3496fd24a1ee49ff8ee2bc52669055eb : SUCCESS in 8m 55s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/181535c3aeaa474cac3f71364740863c : SUCCESS in 23m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a01b3cf47f4474bbedd1c4bda090e84 : SUCCESS in 5m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/91bdfb08746a4de7b461928af8c638d7 : SUCCESS in 11m 02s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0f6b8e54113a4448aac3600b0c30472e : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01e8ca009dd14318b4d281febcaa258b : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24af26fcef8049b8a657dff964b7034f : SUCCESS in 22m 02s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6ae75f487439426eab3cbb0f239505bd : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7a23e092f2644878a01b9053877245e4 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7755550efee34e61b24147202d4ddd30 : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0ca5dcc67bea4d129212e4a2c2609738 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c478cf2e42041859e526ba5b4ec0ec2 : SUCCESS in 52m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcd90593f46d491c946d0236a789918e : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e07684ddb1fc88ab2fc3d1274874e85af087e96e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:31:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9682eb43d5907f4cf91dc238cf57d8c76653a6e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:32:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748182d64f8643c0828cae92745c5db6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce61680f050d49b48d02232b817cd948 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc6a791c40f366af2978c4dc307e29d0c530f041","revisions":{"dc6a791c40f366af2978c4dc307e29d0c530f041":{"kind":"REWORK","_number":1,"created":"2024-03-01 10:03:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/910713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/910713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/910713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/910713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/910713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/910713/1"}}},"commit":{"parents":[{"commit":"f4ca2053bddcd19e2025dc4f28dddfe045cd7363","subject":"Merge \"Remove Xenial and CentOS 7 wheel cache build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4ca2053bddcd19e2025dc4f28dddfe045cd7363"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 10:03:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 10:03:41.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.8.0","message":"update constraint for python-manilaclient to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I9d16001590547ac49645bd9afc98b83da361849e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I44b1cab5d4a05558a40c3d80e47d327aecbd0ce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc6a791c40f366af2978c4dc307e29d0c530f041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc6a791c40f366af2978c4dc307e29d0c530f041"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
