)]}'
{"id":"openstack%2Frequirements~910749","triplet_id":"openstack%2Frequirements~master~Iec136d09030046050336c08af5349b8c16756d42","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iec136d09030046050336c08af5349b8c16756d42","subject":"update constraint for python-vitrageclient to new release 5.0.0","status":"MERGED","created":"2024-03-01 14:35:24.000000000","updated":"2024-03-04 09:36:31.000000000","submitted":"2024-03-04 09:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910749-new-release","meta_rev_id":"5300369b802cf4fa47c47e71766956d5ec8e1007","_number":910749,"virtual_id_number":910749,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-04 09:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-04 07:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-04 07:55:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 16:18:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 07:55:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b6f0cf1df395e9a6ed337463b43fa51a2c44479c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-01 14:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60b10f7b4f1a9af3fb77ff88d68926a0b93db1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 16:18:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfd90f425ff4ff2b4ad21fc32bd4a00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fb95a9b4f24343a181b5b956f8e4c1 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404f6379536749a8b056404a246a4df3 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b98858c1b5a406c83bb98863f472314 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2aa0f810a9c24d66abab162b9fa46437 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b548dc26f95a414bb797ec3ebd8d33a4 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a9dcc3ad24f43d7aaf0836f128cca8a : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2331739089dc4da682f2f347b91abe5f : SUCCESS in 12m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0fa0488f43804ef2980722e5973995ef : SUCCESS in 11m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2523e0c46eb042958ef4c15ddf23fecf : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ec1745def9f84303a37ed31051afa843 : SUCCESS in 17m 32s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fd9f488cdf734a8d8abab8b4d52f0705 : SUCCESS in 4m 23s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/dae79233ab0043be9876fd40921c6746 : SUCCESS in 6m 28s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5ff5ec37b1af479788677726161478b3 : SUCCESS in 4m 32s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/08e81649f4954ae383bcdf4c651081ce : SUCCESS in 12m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f473a4e78fd8469481290ac770f9e943 : SUCCESS in 5m 07s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ca27facd2f4f48f78334649653711ae2 : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/4b6b9002700a44e49642b1692a83020b : SUCCESS in 7m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ab468e28c3914242b5c80228addbc740 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e674ef5f402145d58e56ec75a86d6e1a : SUCCESS in 8m 16s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2bd087e9acef4b24911f92e40084d9a4 : SUCCESS in 9m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fd5e33156d9a4dc48c400fb3f1aeefc8 : SUCCESS in 11m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cd9dcb45c3404ee6ac8f834614b5b989 : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/22f65b5c431a4d1dbaff390c7694b682 : SUCCESS in 7m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ae8787f3e63d4afa91e24fee5b743452 : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8d10d33679f3483e8cb3a774f114f7fb : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/79683d8d0cfe498193a0043b089a5bfe : SUCCESS in 21m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f3765a4f1aa047658d6b50b66deacabf : SUCCESS in 6m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/41c4324a6a4e4513994e7cc3134bc2e7 : SUCCESS in 11m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e648088eb51b4d52b9a5a2d9c6a5682a : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c69aee2a530439793270f4596bb94cb : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bcd48ee9cf04ff883c24a54a0ce3221 : SUCCESS in 21m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/cdcfe62169f34dc481b7ea95a7335e58 : SUCCESS in 4m 28s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/34f2cfd08f044fccb013e486653fef8e : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad614d8768264a07a886802132d3a77c : SUCCESS in 7m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4faa02bd22f649ac856d2e3f199c9e48 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c99da5cb9b0d4d7595a214784728e587 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f7acb44d373430cac9a1cfcfd8624c9 : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e28e401ff987ba231f38f2ccfbbb51729792fa9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-04 07:55:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8590db8a048fca2b601607badbd8fb6836e4e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 07:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9ef85dec72cee9a664e58dd785f946cd5c62248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162857d3ad11484eb450c3b3e389b5a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfaa1df44f284bffb53f192097c60046 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5d9ebcc6644de8bd51a8ca3ff91077 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38ae562d849e42f28594845cf429b3d5 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7aa5ecd5e5644411b67d8a05f8345a62 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae56f2ec2aa1406fb89095d577f2b9a5 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1fc7104abab46db870e920cd8a124f9 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aeff6488b0114c27bca4087b45c7750a : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/096ea349445045f68775c9a61cd0ecc4 : SUCCESS in 14m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/116fb6075ab64303a05f3fb24ed74500 : SUCCESS in 9m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/68a80816c5704f1f86cb72542cff8878 : SUCCESS in 12m 48s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/626ffc8b0ce94009a12293cc88cb7026 : SUCCESS in 9m 09s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6af064dfd8664d8a928247ac0a4cef97 : SUCCESS in 4m 39s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/38a6eef318384de58df5632607630b62 : SUCCESS in 4m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c53ea1e3342244f0bd860e8c6dac799c : SUCCESS in 6m 07s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fe761a1763db43e0b065a00d671890b2 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/116d6c754e7c44f8b03b670a34e2f25a : SUCCESS in 6m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9b004e9f3a7944c98ce5239950caef62 : SUCCESS in 8m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a46dccbaf3f64c66ab1447898645ae05 : SUCCESS in 11m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1875b51cbc914970a9a521b83825c6d1 : SUCCESS in 4m 15s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4f4de43f1c824a6ea9b974db05cbb36b : SUCCESS in 6m 49s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8e545028def84b5ba68004320f3ac660 : SUCCESS in 22m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d34198b632d045bfb6244ec3b361457a : SUCCESS in 6m 46s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d28025ee56ee4e62bda5990ce5267990 : SUCCESS in 11m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a5100530cb264a52be9bdc6fe7b1d2a0 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df320730a62a46aaacb68049f9bfc911 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2340c1f23573402bb7ef7ed323c9a65a : SUCCESS in 21m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1c3234f8f18d468abcb2e43033c88e90 : SUCCESS in 4m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2124692706d742b580bad8112c0863f7 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d286ab98e034fdc8b05fe48058b7528 : SUCCESS in 6m 04s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d53cd9ca3695417e9709a2d7b5d04ce8 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bbf3eb775a64a3099631a31cf3116d5 : SUCCESS in 1h 11m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5887a848fd749ceafa978a3b8422e93 : SUCCESS in 1h 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4961d5bf3c1175d259b15cdd64dca72680823222","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:35:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5300369b802cf4fa47c47e71766956d5ec8e1007","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:36:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68e66d062e748408cf7a490337837e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107e3e832e2442059f992f66a26cd56e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0591b669e57e8313a72a74527a3af95eaf97aff","revisions":{"f0591b669e57e8313a72a74527a3af95eaf97aff":{"kind":"REWORK","_number":1,"created":"2024-03-01 14:35:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/910749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/910749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/910749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/910749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/910749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/910749/1"}}},"commit":{"parents":[{"commit":"bc9d875f8057c9d8d5fabfa6b0c81302fccf4a48","subject":"Merge \"update constraint for python-ironicclient to new release 5.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc9d875f8057c9d8d5fabfa6b0c81302fccf4a48"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 14:35:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-01 14:35:20.000000000","tz":0},"subject":"update constraint for python-vitrageclient to new release 5.0.0","message":"update constraint for python-vitrageclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I63246d2403a82aab8742304b4af3a096f4756109\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iec136d09030046050336c08af5349b8c16756d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0591b669e57e8313a72a74527a3af95eaf97aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0591b669e57e8313a72a74527a3af95eaf97aff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
