)]}'
{"id":"openstack%2Frequirements~911053","triplet_id":"openstack%2Frequirements~stable%2Fzed~I0835e34d45152d02688050136eb19dc118d14465","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I0835e34d45152d02688050136eb19dc118d14465","subject":"update constraint for horizon to new release 23.0.1","status":"MERGED","created":"2024-03-05 09:52:31.000000000","updated":"2024-03-07 13:28:33.000000000","submitted":"2024-03-07 13:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911053-new-release","meta_rev_id":"9ea3318e8d827040aef8b991aa6851e1b68cf6db","_number":911053,"virtual_id_number":911053,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 13:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-07 11:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-07 11:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 10:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 11:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 11:38:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9275ff49854ca82374d6759dbabe7488c31f0a8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-05 09:52:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfb27af1adcded45f33540ff7c3dad737e3432dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 10:14:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c611ad14f934e4d99b46f298604c254\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/16ee23e87fa844c79206edc2ad766d94 : FAILURE in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cf5000684d78ed00373aab5fa443b0499fb09ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 11:25:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d465ab9fa3c44522b32805d60eced8ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d2df6a8d9a4f25bdb06a202c6710dc : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540b2644eb014f33921eff0a76b2101f : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32def8acdae049bea43c3d8ab039a290 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45e57c631c0d4bdda1c1598e61750152 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53177dcff6da4d02b8209797f00ac792 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f3a14a6990644c6bb24f7e99724e634 : SUCCESS in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28600f82722847c292af34519f668ebf : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7e926ba7b4874ed78ff74be70ba24b13 : SUCCESS in 12m 46s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ecb53fb601e84c1ea62196a5412050a9 : SUCCESS in 7m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c9bc9e859a7c46cf9fc3cd9c33e05d03 : SUCCESS in 6m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e8ef494dcf354e27b5de6610958d02e2 : SUCCESS in 7m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/85d3c7b8274143a68017ee64d8872c64 : SUCCESS in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fdd9c06fce584d6d80325d951900ed41 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7d6d38dc338d408a985a62ca3df7850b : SUCCESS in 7m 45s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a3be86fa3c7c4ad6a3ee2a61cef63fc3 : SUCCESS in 6m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/68047c0449fa421290a8c531be7089a2 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa7ee1b719e1427f9c76370a65069d80 : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/768b7c05ef124d268a4dd59b6f462d45 : SUCCESS in 7m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5395176650254229ad00089ae99b918f : SUCCESS in 13m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/67eb216f53334786877b5ed52d5edfe7 : SUCCESS in 5m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/165810f56e204070a3b2296008a07700 : SUCCESS in 9m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e366d2660ea6477c8193651df80d9481 : SUCCESS in 5m 49s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/1986e54447434769aab3378d9f0a780e : SUCCESS in 10m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/64e2eaac3f6b4af1ba242bea6fb0ef7f : SUCCESS in 34m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a239c677873418e82a9ed888e7eff80 : SUCCESS in 13m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eb3c2ce8f4914f9d9dd280b3fabd11c7 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ba894966a544e8da480d87a9a0affa2 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/183dfa8d78514ed7ba919af40ac75bee : SUCCESS in 24m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/616b9b0f4edc4086ad21eeeaed5edd1e : SUCCESS in 7m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6f961857a5b24df2be924f4a70c44c40 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8b6052383e9461f963d4b7b7f5a41b9 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/730ca5d84b0742b79098e1c540c0780c : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b41ce92bfbf49daa3b76b023e920528 : SUCCESS in 57m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd02e72daebd4538bbb45cc263fd2d38 : SUCCESS in 1h 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5f420e8d4f8daef3c989314bd937b0734e861b36","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-07 11:38:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a66b64d656a628371a0e6d401ec7b109eeaf492c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:39:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7bfb5eaa9057f596b23e698dbdcd0b61a7f8250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:28:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f9699453b843d09104e6e9eb14017a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361ddb039f23475c98595ad3ff564890 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29a22c7ff9b64d8f87d678a00b575b85 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dc02ae1371748dbbf0eb16d95f9fd9a : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35fcdbec1d6e4fbf8b4ee9ccc6f0687a : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43a7957c18284729b0828d7959a48aeb : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/134ee02e41274e498299e26020f6b419 : SUCCESS in 11m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cdce2e43eb8549e4884bcf4533a6f95f : SUCCESS in 14m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/590cefdd1138453291c34612cc4dd9e1 : SUCCESS in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5fdb29671cc74906ba77aa14fe3d49fb : SUCCESS in 6m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4d81d0a4f6f443c985fa8c511d57aabc : SUCCESS in 6m 05s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a660c52956ba4d579e45e01ca1e4f445 : SUCCESS in 7m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6731ca4536df49b5b3dffa232b9d0946 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5496d28d75fc41deadb23305d38609ae : SUCCESS in 6m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d3fab339871d4c65bfb06d74a9d86165 : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/77e0bf96e6954c248664a501fa9decbe : SUCCESS in 11m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2ec5f08648524795b9ceaaa785533977 : SUCCESS in 3m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/145c565ccfde41f595b81930a3a29ff0 : SUCCESS in 8m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f4aaf3b7e5664c7d80582d2b986fdc11 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/10e7f04f16234db09d4717af3cfe686e : SUCCESS in 18m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dfbf0a55acaf4361a0567a5d17a2b93f : SUCCESS in 12m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9c52227da0aa42248b6cbbaac2bc9536 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58ed3a23a73c47c18013e72c23be0443 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12cf3eac6a484e56b6c3524729b1a9e3 : SUCCESS in 23m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/929885fa09b847ce80c21bddfec3d675 : SUCCESS in 5m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2f12389b87f14bf1b1dfbf338aa838ca : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f9c05584bff4724a55a127cbb7999cb : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f9dd193067914836b15b474c3dfab56c : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b79019319c7410eb063ba1f72888341 : SUCCESS in 49m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1690308347c44187a84fc15f947232f4 : SUCCESS in 1h 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9ea3318e8d827040aef8b991aa6851e1b68cf6db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:28:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c379d80fceba448eebe6c3121184435b49c48ec","revisions":{"9c379d80fceba448eebe6c3121184435b49c48ec":{"kind":"REWORK","_number":1,"created":"2024-03-05 09:52:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/911053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/911053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/911053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/911053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/911053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/911053/1"}}},"commit":{"parents":[{"commit":"6c7a236da8e3db710247499f3bb94e702e6c22d2","subject":"Merge \"Remove Xenial and CentOS 7 wheel cache build jobs\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c7a236da8e3db710247499f3bb94e702e6c22d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-05 09:52:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-05 09:52:29.000000000","tz":0},"subject":"update constraint for horizon to new release 23.0.1","message":"update constraint for horizon to new release 23.0.1\n\nmeta: version: 23.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: branch: stable/zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic2aa34f9d07c8fe8bbc659843fd932f9edef09f7\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I0835e34d45152d02688050136eb19dc118d14465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c379d80fceba448eebe6c3121184435b49c48ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c379d80fceba448eebe6c3121184435b49c48ec"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
