)]}'
{"id":"openstack%2Frequirements~911059","triplet_id":"openstack%2Frequirements~master~I2eabd32145286a1d81309c4731500077683d0966","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2eabd32145286a1d81309c4731500077683d0966","subject":"update constraint for castellan to new release 5.0.0","status":"MERGED","created":"2024-03-05 10:06:36.000000000","updated":"2024-03-06 17:39:00.000000000","submitted":"2024-03-06 17:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911059-new-release","meta_rev_id":"da9ea91ee29ef7c87a84c060e880b57dcecc5a39","_number":911059,"virtual_id_number":911059,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 17:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-06 15:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-06 15:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 10:14:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-05 11:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 15:57:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"af6c7a51e4e93137d28ef9a8e02a057337dc55dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-05 10:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"618910e74e82db4b090f87acbae51dc5cf59a835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 10:14:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"03b28669646b8b90c6b68c717b6086c137b70b1e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 10:14:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e82b5ffda8433eb4023758bb20bf0630ba59502a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 11:46:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8367e08dee394b21bc1437a138f08941\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e9ef68af0a4a4d95b49b7e2b0352ac : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df20b98331647379987b72e5879bc5e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2974c80e9dc9489584e77de09a707bef : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09224aeed70e41308932ba7fd3437bbb : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3aea6664337489c89eec9e2080bfc77 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abfe274126874cd983678e1b41a54542 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32c50f4db5a74e7baa617cfc183f1942 : SUCCESS in 9m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ba6ff6cb7d8c404ebb707145efb19bf3 : SUCCESS in 16m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3b01f781002642eb8a776e193331b7a5 : SUCCESS in 9m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2790e899a2d54b4191f5be962071f6e7 : SUCCESS in 16m 28s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6fde55d8328d40e9aac6dc8e8d5eeec6 : SUCCESS in 4m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0baa17f40e0e4a2dac746903ba26fa20 : SUCCESS in 5m 07s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/dfecb87b23ee40b1b5e19256eaf126b1 : SUCCESS in 4m 35s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/82dd7ab39fea4a2db13fb18423dd6d2e : SUCCESS in 11m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/254ce47012e94779b84115aaf36a0072 : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/72d51393df6f48eebf360c6a502d6c49 : SUCCESS in 5m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2d70ff99200546e28c86ddb076ac1000 : SUCCESS in 7m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a6253123572d4daab781497d192fb61f : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3a61978489b4401b2f31889c697aea5 : SUCCESS in 6m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c68b85768e5043dba4efd166bb64ba2c : SUCCESS in 9m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fa4ee04a38014520aada9b8978bc2cfc : SUCCESS in 9m 39s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9f97880c4baa460e9278538a952b2c95 : SUCCESS in 4m 24s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b279950065ec4477ab9770111939bea9 : SUCCESS in 8m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/45604884a47944a989b37f3039077a48 : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/423e33c8882047deb137b44c2e27cc16 : SUCCESS in 10m 31s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b43368155dbd4a1d9cbf4918e25ad62b : SUCCESS in 21m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/99dce1182f5d4854ba3a53c29030f298 : SUCCESS in 7m 00s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ba805ed41bf140a4b2b407e2feac10b1 : SUCCESS in 11m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/52da09046bed45c38a2ed4d00bbb7241 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ffd712ed032417c81a3dd18d37534bc : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/601f5e11fc054ab7b6eaa482d7ebc237 : SUCCESS in 19m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1191987ec5484090bf0026338034e0af : SUCCESS in 4m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/66ea7b50cd2d4ac3973650ee435e2a93 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b2a9b15fbb94020becf13ff2bd1d739 : SUCCESS in 5m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b1acaa7d98f54ef19d2b2111f5c2860e : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d051764a916e464687792ba85c599955 : SUCCESS in 1h 08m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fcd93172b404e8887efc7136f0f3fed : SUCCESS in 1h 27m 18s","accounts_in_message":[],"_revision_number":2},{"id":"83d8bda71e15665189cc0030ddcaf2689f7fee8d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-06 15:57:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68389db2abc6b4ac1f873f851e5bd1f3981a6c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:59:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b71a549a8a62915a6e7af3cc903e1995251144c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 17:37:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5f81193b7240bc81bef16da41d69bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f34f710b61f40348655ff6beff3cdad : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b7ef04d10242eea4077f7a74b19780 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77b36cc2e331410fbabffd6b189b2ce4 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bc8d48ff75943cabdbeb93d46b631f1 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c54dcbd9da2f4364aaaf26fefe877450 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec437196d2294a9dbb05967f518fabbc : SUCCESS in 5m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28397ce2fdb94e6eae1b530c64e81e2a : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a0e85a9a999e4dd98aad7b12a5d2aafd : SUCCESS in 14m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e49eb3528794080ac7e8dd22dbfcf38 : SUCCESS in 11m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/88cd3b460ed4485b8f06b009df32cc91 : SUCCESS in 18m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/44359312dae54e98a8761328e32c4e9e : SUCCESS in 12m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/eb9ef891aa8643a3b44b88b157e519bb : SUCCESS in 6m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6885655d97b44daa8026d7830e5af8ea : SUCCESS in 5m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f9c353c011e749119f1e2c47bd8f9b3f : SUCCESS in 8m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1b816a08198a41f0a64c75e201892aad : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c266b9edc4949289690b0ead0869db0 : SUCCESS in 12m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7b8cc42916ca42fb8d2818ff081dd8bf : SUCCESS in 7m 36s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8fe08210ec534e009a1430dd8bf5aa04 : SUCCESS in 13m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f8c14708775f4fcf83f387058aeb0b72 : SUCCESS in 4m 55s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5d8821466ac94070898fbcce64302502 : SUCCESS in 8m 58s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a5d16867861649618376ba3770851e2a : SUCCESS in 25m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f3dbf9bd2f3b4dff9286800762aa32a0 : SUCCESS in 8m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/adb53d7ab33a4e22ad146aed4f498356 : SUCCESS in 12m 04s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/80f68500e5db480390bfbccc9e1263a7 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b809d45faafb43d9b33b54e6221d8b4a : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2db41a92bae44af196056bbba50e4e43 : SUCCESS in 22m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7d8cc6923c7e495a9de49b44869b1f7b : SUCCESS in 5m 25s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/96f3e1dc2d0948c4ae2fb8fcefeb0c17 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00ff6fcedeec4497a36a6ea94d974df8 : SUCCESS in 7m 25s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0c45600c369245769fe10e480703101e : SUCCESS in 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/949d616c719c4e28a9ce06976e95daba : SUCCESS in 47m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c82e1ced7a044fa9aa7cf91480e88bd : SUCCESS in 1h 37m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d58e644e2604b137b9a2e83e43fa6a3ed42657a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 17:37:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"da9ea91ee29ef7c87a84c060e880b57dcecc5a39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 17:39:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ef07b0d2e34a0eb60901e2484c5434\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3fede1cfc14fdb8582b0f7c4414d66 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6230793a37d1ce67a7565d4ac876ffd3b6138352","revisions":{"97bfff855d13a720e6765d51491e1a5dfa4bd175":{"kind":"REWORK","_number":1,"created":"2024-03-05 10:06:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/911059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/911059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/911059/1"}}},"commit":{"parents":[{"commit":"02836711bde7b324e42910110585f84d1624cf1d","subject":"Merge \"update constraint for python-designateclient to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02836711bde7b324e42910110585f84d1624cf1d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-05 10:06:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-05 10:06:22.000000000","tz":0},"subject":"update constraint for castellan to new release 5.0.0","message":"update constraint for castellan to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I01be64a2b7e060903bdb3edb182478743ad482b1\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I2eabd32145286a1d81309c4731500077683d0966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97bfff855d13a720e6765d51491e1a5dfa4bd175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97bfff855d13a720e6765d51491e1a5dfa4bd175"}]},"branch":"refs/heads/master"},"6230793a37d1ce67a7565d4ac876ffd3b6138352":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-05 10:14:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/59/911059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/911059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/911059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/911059/2"}}},"commit":{"parents":[{"commit":"02836711bde7b324e42910110585f84d1624cf1d","subject":"Merge \"update constraint for python-designateclient to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02836711bde7b324e42910110585f84d1624cf1d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-05 10:06:22.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-03-05 10:14:02.000000000","tz":0},"subject":"update constraint for castellan to new release 5.0.0","message":"update constraint for castellan to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I01be64a2b7e060903bdb3edb182478743ad482b1\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nDepends-on: https://review.opendev.org/c/openstack/nova/+/896100\nChange-Id: I2eabd32145286a1d81309c4731500077683d0966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6230793a37d1ce67a7565d4ac876ffd3b6138352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6230793a37d1ce67a7565d4ac876ffd3b6138352"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
