)]}'
{"id":"openstack%2Frequirements~911407","triplet_id":"openstack%2Frequirements~master~I05faa1be554c98eee69c4247e5fd9c0dc7e0cc28","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-16 08:20:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I05faa1be554c98eee69c4247e5fd9c0dc7e0cc28","subject":"Revert \"update constraint for oslo.db to new release 15.0.0\"","status":"ABANDONED","created":"2024-03-06 07:54:00.000000000","updated":"2024-03-16 08:20:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"87062426e56280f95084ccfb470dd880eb95744a","_number":911407,"virtual_id_number":911407,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-06 09:21:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2024-03-07 11:37:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 09:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 11:37:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4cea278490971d0aba901be0e328c09e95b44c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 07:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cfdbc8e043abacc0de0276fe6a99bf8ab9b6171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 09:21:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451713fb0feb4988986650feecccab22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d04b334d4c4b32a4958179246917d0 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4589749094cc4486a64f63152e79f57a : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c73932cdfa40f985c56da805db4fe3 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8abc781c2c4c4045a2a381ae5e9c4a92 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2c7d05a133c4f92a4c152ffcbc5604d : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c30bc4e4cf7248888860988659bdc2a0 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7b36af1e6f4402dba69e37e862cfd88 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9d0d63de052d43dc8c012d5f641f9136 : SUCCESS in 14m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5081d3bfdbe0496dbeb4ce2c0c1e8f90 : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a028dc26f95e42678172bcebd07c1ebb : SUCCESS in 15m 56s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0a733b035c67444aacbdafd65c461156 : SUCCESS in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/89c11762d4e948c3a436652a9079cfe5 : SUCCESS in 4m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e6e1c777f9a449bcb7efde74d9410437 : SUCCESS in 3m 41s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/12c5b407738e47bd82b670cdeb028adb : SUCCESS in 10m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d651eb9606554d518b2a58c44815a0ba : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e41f83f5b59a40acbbf8fe605938efe1 : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c8f876fafb3a4e558c60996e681e8d4b : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b7bfab03c6054e37a4be54ba2cbd0b17 : SUCCESS in 13m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4e04b9f36ff4656a4920d2cc0de3cb0 : SUCCESS in 8m 08s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2b7ff354efd54495835b90170f1639a2 : SUCCESS in 6m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f4a28b47f0464c30975d5b7c3b675945 : SUCCESS in 9m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b9ef2b0995ba48a485f12e290b2d8664 : SUCCESS in 4m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/18548010eb564f0597a9d4b0513f7a4b : SUCCESS in 7m 27s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6ffaafd5b3dc494585e3c472f30421dc : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/c4ba0051343c4a6ea7706c74e97fd185 : SUCCESS in 12m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bf97c9fad2f24dbb8b180be1657d5341 : SUCCESS in 23m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5932b0650842480aacde05adebfff74e : SUCCESS in 8m 00s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/80df917f253c4e69851d08e6450b8cf4 : SUCCESS in 13m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f322037d3d2d4c08ac87d82a60793a43 : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c062ef61717d42e68a820c611c83af51 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce736f6ae5fe4864ba2c6bc089c7332d : SUCCESS in 23m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/fbad756a681f4b3f8447a1e1be88fdf2 : SUCCESS in 5m 25s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8d8307fd51e84ee0a2eed815b5ae4bf1 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbca20f8713b4b1f9cef57af65026d03 : SUCCESS in 6m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c297df102e5547e49e1b19e711ba0de2 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/608cea8a268541598545e092c7c05978 : SUCCESS in 54m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5093e8d271d740938617af11a0e9dca5 : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f224cb7e0eb6a3f6cd4c769508efbcc5f4dd7034","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-07 11:37:43.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87062426e56280f95084ccfb470dd880eb95744a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-16 08:20:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81eac1a70789ddbeed31fd4478f47e50335d299c","revisions":{"81eac1a70789ddbeed31fd4478f47e50335d299c":{"kind":"REWORK","_number":1,"created":"2024-03-06 07:54:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/911407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/911407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/911407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/911407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/911407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/911407/1"}}},"commit":{"parents":[{"commit":"02836711bde7b324e42910110585f84d1624cf1d","subject":"Merge \"update constraint for python-designateclient to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02836711bde7b324e42910110585f84d1624cf1d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 07:53:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 07:53:49.000000000","tz":540},"subject":"Revert \"update constraint for oslo.db to new release 15.0.0\"","message":"Revert \"update constraint for oslo.db to new release 15.0.0\"\n\nTHIS IS FOR TESTING. DO NOT MERGE.\n\nThis reverts commit d361c50e933db1ba1f291c2289a20fe2fb240ac7.\n\nChange-Id: I05faa1be554c98eee69c4247e5fd9c0dc7e0cc28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81eac1a70789ddbeed31fd4478f47e50335d299c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81eac1a70789ddbeed31fd4478f47e50335d299c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
