)]}'
{"id":"openstack%2Frequirements~913489","triplet_id":"openstack%2Frequirements~master~I9bd77902b809ae4bfee955bd04bc629c3bebf365","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9bd77902b809ae4bfee955bd04bc629c3bebf365","subject":"update constraint for horizon to new release 24.0.0","status":"MERGED","created":"2024-03-18 07:11:29.000000000","updated":"2024-03-22 16:09:07.000000000","submitted":"2024-03-22 16:08:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913489-new-release","meta_rev_id":"1497419cb8192d6f53116c664026f6eb51130510","_number":913489,"virtual_id_number":913489,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 16:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-03-18 16:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-03-22 13:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-22 13:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-18 08:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 16:39:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-03-22 13:39:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"98b05e1d66cd213b862c3de5f767d0704cf8982c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-18 07:11:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7da0bbfabf2ea0403168f44cc2d286bbb5b6d1f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 08:28:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6361e5a7974792889424d7021bd9de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f5017de9d843e8aece5cd1479c9e39 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb7966b19d154cb9a5581eff3caca7ca : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d214aa03a94507ad8f9a893a54da9e : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41e97d8ca7a94d8499f801e58739d492 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dee1ee988fc4bd182e71769ea32759c : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6030630dcf744d338b4138d5faaf4516 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09db8101cd34403fb536b4266e387139 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/174b7531064146a9b73c979aa683de66 : SUCCESS in 14m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/deb39c8409d84f7fa2b3b7379f947915 : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b70080faa3ec4c998ac78e7d4b092e35 : SUCCESS in 18m 16s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2a0053d456ef4ccab80173fa885b2e56 : SUCCESS in 4m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a8f193b447974201b583e96d41d07c99 : SUCCESS in 6m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/03b56688ba084f9c8e68d239cdfed394 : SUCCESS in 3m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5f3fb3d56b9d45e89691954e5fb5401c : SUCCESS in 10m 14s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6c118c8849fd45efa727b9ee7794d18e : SUCCESS in 6m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1a704b7591c64d16acfd99b0a7787397 : SUCCESS in 5m 35s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7c4e80c1037b42168dcf566cc3ebdf5e : SUCCESS in 7m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/76613ede6efe4d05bfdb2bc0cf7ceba2 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7ae645426ea40e089ef92b57bf9ab25 : SUCCESS in 8m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e5b2f018fe0d4098b389c8aa21a9c6ec : SUCCESS in 9m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f2b66217575e49b797cc8158672009bf : SUCCESS in 11m 07s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/fade75d733494e1e8714935c3c941ce0 : SUCCESS in 4m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3f896b052396409ca616fb06801ac49a : SUCCESS in 6m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ce967a109621441398f6f0e4ef6c3fa9 : SUCCESS in 4m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/403254519ff9486ca2a84c16505a2ca6 : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ed5d08bc8703453fb3d152689758f7b1 : SUCCESS in 9m 04s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6ec1a59bc6854202ac2109c5ac2feec9 : SUCCESS in 17m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4dd0ac111ca24ca7af1fcb7d609d044f : SUCCESS in 6m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6e994378fd6049e1bbd11ee074101bc0 : SUCCESS in 13m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4986452708554a768b80858e0405bed2 : SUCCESS in 3m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd6f1fe346b34ce4a2dd9f229a084e5d : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f64b03379a5435b8cb10f1ad1843ecd : SUCCESS in 22m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8c2dc09fa05d464da8e18bb774a451a1 : SUCCESS in 3m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3abd366cfc8f4efc96f3ee36031128a2 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c51e8684b604d5cbcc564c3af771383 : SUCCESS in 6m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/147c0909c58041da8bf206db9e20afd5 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77e4c0a7a4784fdca2c8ea2fd4283bc4 : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af3e3120247948978989dd40453017de : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"59fa7f06d59d2025af7c511b6d8897f09407426e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-03-18 16:39:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"809f503f17eb2c7cd6d2b2d9b27c66f5497dbdaa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-22 13:39:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddbcc65b070c9f772d194c73517a29c4380c6d31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 13:40:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5bd2268ecc6b30f28506cab44fd02ae696b75c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:08:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adbf7f501412414f823348a7f8c34ebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95f2f69f54845f2ad470e0457af864b : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b759ab4b6b14331a7dd55279cb992bb : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24f1bf3f73024377b04788f8d89d7079 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab1ffd6162af45ca806f74d86b8e65c8 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d150b621d95f41b8b01ebd5b872d0cb6 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e794c9fc76e4263a62eb9ae91040041 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5681cc0eb9d346518bfb90d59b65b97f : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df29fd422dfb47d88f60c88690ba4aa4 : SUCCESS in 14m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/95bde58a829a4ff89f8c62eb510a390e : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2d6e0848097b45b999257d0c410695b3 : SUCCESS in 17m 35s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9524d17d19514293a8ec96f3d8447ec8 : SUCCESS in 10m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6b1e46a6c7524925970f76c43160e7c7 : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/18f1128531f5414590c6caac0bdeae4f : SUCCESS in 5m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/382b518eee2a49beaf9cd5fd647b288f : SUCCESS in 7m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/da93914115594b279a39e1c867d29615 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/146c3adc78f849389e32e47637694676 : SUCCESS in 7m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/dca68f9d72b841f2b83fa54703ddfb16 : SUCCESS in 8m 31s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b917ef7d18174a9ebf241c7203a794e9 : SUCCESS in 10m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3fc707c431e74fce812670f20b34b03e : SUCCESS in 4m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8e4ad57e4912422ea714d3bcfad7ca76 : SUCCESS in 8m 33s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8ea97b4031a74553a7baae0506458587 : SUCCESS in 24m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6ecfb42c53794e5283506c59f15b20be : SUCCESS in 6m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/701a1fb32bc34eadb351a52085db06ca : SUCCESS in 12m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a30885da18974394afddf11f623847ae : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7fc7379eaa4049758af0f34088402f6a : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42e492c73dc740b2b99fb86bc00d96ff : SUCCESS in 21m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3d1733b5f62044dd906cf2e2b44efdb8 : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8efd99c4a0da463390df282e5ee2853c : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0345a6d5cbc9452eafd91727ab5ff898 : SUCCESS in 5m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4954aed1d5e14f6cafcd71f0c46bc930 : SUCCESS in 14m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de891a7ebd554267b56fd1172ed7a69b : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e281c02f45c47ffba22d7ffd88ebfd4 : SUCCESS in 1h 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"75d0e017a76c4164586acd16ceab4138712b379e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:08:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1497419cb8192d6f53116c664026f6eb51130510","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0786db36d6fb4983869b0694cb46311b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54cda88585c74660b369fc3f6aa47eb4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"490982c836927694db91de84a9fba5e643ae3dbe","revisions":{"490982c836927694db91de84a9fba5e643ae3dbe":{"kind":"REWORK","_number":1,"created":"2024-03-18 07:11:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/913489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/913489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/913489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/913489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/913489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/913489/1"}}},"commit":{"parents":[{"commit":"99f60d6329d545215561b198c680e79daf690192","subject":"update constraint for yaql to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99f60d6329d545215561b198c680e79daf690192"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-18 07:11:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-18 07:11:25.000000000","tz":0},"subject":"update constraint for horizon to new release 24.0.0","message":"update constraint for horizon to new release 24.0.0\n\nmeta: version: 24.0.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I0996ef7d2ce6fdefcd5137c7c8b16f155999b7bf\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9bd77902b809ae4bfee955bd04bc629c3bebf365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/490982c836927694db91de84a9fba5e643ae3dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/490982c836927694db91de84a9fba5e643ae3dbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
