)]}'
{"id":"openstack%2Frequirements~913567","triplet_id":"openstack%2Frequirements~master~I029b6bc3ca8e23861595eb67b23334baa4d079c6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I029b6bc3ca8e23861595eb67b23334baa4d079c6","subject":"update constraint for python-openstackclient to new release 6.6.0","status":"MERGED","created":"2024-03-18 14:15:02.000000000","updated":"2024-03-22 21:08:10.000000000","submitted":"2024-03-22 21:07:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913567-new-release","meta_rev_id":"966aefa921bb6b072511d4d1572c72b68d901cf7","_number":913567,"virtual_id_number":913567,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 21:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-03-18 16:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-03-22 13:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-22 13:40:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-18 16:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 16:38:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-03-22 13:40:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"54e68b1f486c787e0fd34e7cb7f594b859a5750c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-03-18 14:15:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7516e2ba098f5c27e3e39ba26adec28f13190ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 16:36:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0aace5e5d84b51bb58d9e4889e1811\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b0246639464087855c442929351106 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2272c25625ed42ec8328e2c6f261d7e1 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4a49439360440e0924fc2b04aed96aa : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9491670e81324793953d48b492af9b32 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a32a29c327d84f2683d338af839680a0 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/962e9a6a7a9e4fcda0e8281a0ed830fb : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e52973300424908a26b3db1a7d4a81d : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5cc265a492b44265925edfa3328d6800 : SUCCESS in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3de6de85fc244edea008241c003dd49a : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f5f8095cd26c4368ae7411ae3f8b86f4 : SUCCESS in 13m 10s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/221302e9c90c43e49f6a62bed3be820b : SUCCESS in 4m 53s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a160e52c262b4e319e95144106c9b52f : SUCCESS in 4m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ef20b9c2f06f4767ad7809820039b968 : SUCCESS in 4m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b79483d796e14565a7ec03905ac1020d : SUCCESS in 10m 25s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7ce8d58f057c4d8aafa430fa9af14f0c : SUCCESS in 5m 11s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c95eed42f9054d21815abcdb4367b266 : SUCCESS in 5m 26s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9d01362c9b8441e6994b568a77742557 : SUCCESS in 5m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b4dc7c8e37e34130bb64dad9ee96f329 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b66fb2423beb40deb36f7d91e54cd23b : SUCCESS in 7m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/88c24364e6134f7ca30b6a73f6ca12d3 : SUCCESS in 7m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8cc7af043bbf4b81ad0600fcd7ac8549 : SUCCESS in 10m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/159610a942cc40b8a1e0238bfa9795b2 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b95575045e934a28b0b022455bf24770 : SUCCESS in 10m 44s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3924320b31204cd4a38083dd76d59a8c : SUCCESS in 4m 31s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/32749fc2bc244091bbf64a9f5cfee8fd : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d61d02edf8754ccc8bac2a14627c8de4 : SUCCESS in 12m 25s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/166ec76f76f04b5e9a432055cfc94854 : SUCCESS in 20m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d95c1742e3ff4e789e9d457663501896 : SUCCESS in 6m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f4cac2c362c240e3a84093a92866e9c6 : SUCCESS in 11m 27s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e0e67ce17e0e4a13877d0ca066f7ca35 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24cbe8577fcb4bf18053d93cc5cc2c60 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eaf8c3a1477b41af829cf897922a736d : SUCCESS in 22m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/806c4bb0449946fd87600ae7a3f6f9e7 : SUCCESS in 4m 28s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6bd728220ded433ba192b08495031241 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1707d878dcfb401f936eeb54ed8de499 : SUCCESS in 6m 19s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/124f1788b05d43e39a9a14c657e91fc1 : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f24116dff5441aaa218e7006d6b9fde : SUCCESS in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19df6dad3aa14eeda86e4409e09331d7 : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fe41750427bc4b286052901343b84ba7d27ab9d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-03-18 16:38:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9cd5d0ffa3b5d4cd984e573b16cf419cd55c08ae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-22 13:40:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c40941056ad5415f9af340f1f44db1f1ca980c73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 13:41:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c794821c789f43786aa7d96c38f19d1ff6cb6f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:25:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7bc6344d3543f092956cbefe80a38a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec593e3937704b5da5ef00db512c8f5e : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b458983f3c14a48b40877158f0f5ed8 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5afb48b9d2ba41f1a86a1ab33cff9b99 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91203b17ff4848b58e239e4ea8fa3f22 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ba7fa88c7f44ecfb157c7c1d5cc8dac : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5d2a3417a164a258bd49a8e7a263c82 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e844d94d703a4f669ad20a5c4d3079aa : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/03dbae7ae9b1440c807a070211a4fdf7 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9b95be2bdca5498fa07575605a4285f5 : SUCCESS in 10m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0ded7e0e00bf43c0811a35052a3be4c6 : SUCCESS in 19m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6ce57fa9317144718f4fbfa667a47da5 : SUCCESS in 11m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6598a72c2ad64c16bfacf3498c92d602 : SUCCESS in 5m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6ef7436bd12f4981946e6904a3bde1bb : SUCCESS in 6m 34s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/be0ecb7c122945028a38aa72cce7c293 : SUCCESS in 7m 55s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0a25e74f33ff496e92796d8aa2a6a6db : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1627e1ec35c46318aa1940948e01ef3 : SUCCESS in 7m 28s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/01035c06c73a4305b79b9f369f79eb4d : SUCCESS in 9m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ef42dd51976449dda88124acaa963bc8 : SUCCESS in 10m 48s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/bbeafd4a8bc14c4184d741d45fa3b59a : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/31d4396cd6454fca9979373a52e1d08c : SUCCESS in 8m 17s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8e364efa2f39473aa06306bcf3f4e5f2 : SUCCESS in 22m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea86aacfa13e4203855e309d8bff0a45 : SUCCESS in 6m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/90c252360e924582aa0648d5f09d1e1f : SUCCESS in 12m 07s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/aa6a2d272a914cf4b89a6955a59c8c28 : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4991ae3cb231498096a32b3dd6482807 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0574ba85751f4134b1fb157394d27db4 : SUCCESS in 22m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d8859e7816c247d9b491ded895b1f199 : SUCCESS in 6m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3eaed9f067e541489d1db15fa858f0fd : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ded008aa1a0f48268f40cfc7881088dd : SUCCESS in 6m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/872c4053b16443daae0b9f83e99df351 : SUCCESS in 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8f44185fae34df3941f7f4ac0c61651 : SUCCESS in 49m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a4a8fc66196496bb46520459cc259d7 : FAILURE in 1h 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8d151e2f71ba439b433474302161379bed5db2b5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-22 17:02:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f25c5f128b2f927223bd650d83b05c61083d6aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:42:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e0b191d282a4d1aabbde012ed7d03b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b014cf2a8b89409883159e55e6e38dd1 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cce1d7be6604e5a80a9bca47a02d298 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad93577daba4caf8ec697b2f98309f0 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58b933764c0b45139de503a2ebfc9682 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b09e8470f6140bda04f7c53443ba889 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d689ace59eb4eb4afa47e8de9011fad : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98b3cc5ac26741bb8dc8f1eb8078a70a : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03bd54d7aa9f453cb9002007d19cf921 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c9b3e3d13aa840da86e1df70ca366e2c : SUCCESS in 12m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/95c19ee7f63e4ff19e418418d7611247 : SUCCESS in 11m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/80975bc7ec7146949876828afdd918fe : SUCCESS in 15m 03s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/84424a18359a4ee6ab2ac0d3079a7403 : SUCCESS in 4m 52s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1af47e8115ad4cf09a0783b927b2044f : SUCCESS in 5m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/3db2bc7d68fc4e698da30d5bf9d431ee : SUCCESS in 4m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ff170b126b8744c1b9ee7c31627f4f1e : SUCCESS in 9m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9326f91b6d164ab4902dc390ff7de606 : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3e37bfb7f8004d01ab1275a2559cf7c7 : SUCCESS in 4m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0683b30e10474fe4a4a724bd35e049fa : SUCCESS in 7m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/228741f6aa6b41bf87713a37a40a073e : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/347bdaf0266e45b080052e0bf1621846 : SUCCESS in 7m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/745b0bc08f414ff4935ad916f816ab1d : SUCCESS in 9m 11s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5827f70b83a14454b550339689b0121d : SUCCESS in 10m 56s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8cf19fac13cb4344974cb6f5c854d2cb : SUCCESS in 4m 08s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/193bf163dc6c4825bd05bb9e38fc5121 : SUCCESS in 9m 40s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/932653944a9e4aab82126cedb5e52f37 : SUCCESS in 4m 31s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/48a9d6bdc8654561a824b9849f8cf653 : SUCCESS in 4m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/874bc1b441a0435fbc98a4244cbb22e5 : SUCCESS in 12m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/90e4fb05ff5149f0986651edabee7185 : SUCCESS in 23m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98035eefdbaf4a488b06fadeaf970410 : SUCCESS in 6m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0ff34f6154664c40ab734703d4a3316b : SUCCESS in 11m 54s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f42cc61e6080446c900d7f9e05d3e56d : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2b9c03faf4645b69ce2bd4f7fb7be5d : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdce7e4517514247adf2337ebaeb018f : SUCCESS in 23m 32s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1c870f92464e4a7498866441dd1669d1 : SUCCESS in 4m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2d677c0124dd47ef87805b654d7849d4 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5013d0e224847c48df17c48e0cd9e06 : SUCCESS in 6m 10s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/48402ed8bbb5458880a33a9559ec3544 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dce0ea812154790a0a6cfcedf2a4349 : SUCCESS in 41m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17e2884e45cb4d78b89d5f07e1a34780 : SUCCESS in 1h 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ff9fc482d4a081aa6dfea53a4fa74465a6746bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:43:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5a4d391b6eab31cc8308f22c1d72c0a4ed2c183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 21:07:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e73aacbc03e4872a2c70111642f0d7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d99db583a594caf9c12a6db69cfadbf : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42887f830274337bce8b97d3a380669 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/231a9866055b49e2bc39d452a469218a : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f843d16e556477c9f52f33e2965cb77 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec616f14b5ae4f91a4340c2a7439227e : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/918fe4f332d74c60aaa2d50c5ec4e9e2 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4b089987b9eb4f8094f17d4d276e8c5e : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/35119ce0bec149a097fc4f280afb5dcc : SUCCESS in 15m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f2388880efe4430b2e06aa0938fad12 : SUCCESS in 9m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f72a25c523a4125a017b23efa1f8313 : SUCCESS in 16m 41s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1856d3ed8570438b85c666da42ce35ad : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e1c4561bdc0d48c4810868ea1edd236c : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9d014ccbf7a84eea962e931f7f1abb12 : SUCCESS in 5m 28s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/082876f6cb44407a90e4a8e7695681a6 : SUCCESS in 7m 55s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/847e642783bf4e0490eff05690f39f10 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b123e1d15574d80ae413275b5974ba1 : SUCCESS in 7m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bddfcad0bccd41fcb2f94d1ebb37eb91 : SUCCESS in 8m 52s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f13d4b3c15b445089b041537d2222b46 : SUCCESS in 11m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/276c61e547c84d849bfe7d6408278e49 : SUCCESS in 4m 17s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/464fb573c3a4427c9e069e7ce8adf539 : SUCCESS in 8m 39s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8f7bfebf16dd459ba808670ebce7cc60 : SUCCESS in 24m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7ddf726d94c648659242df80bdf86869 : SUCCESS in 6m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/85ba112d0a314bb5b69a58301084be6a : SUCCESS in 8m 06s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/259bd9cfa8314967b13c073ce0c4d459 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8baac8ad4b734ba9a03c7c25fc935b00 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c992947804a42ae862cf7e01510c29d : SUCCESS in 22m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e04a304f48734dd282812b4c242e9081 : SUCCESS in 3m 52s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c0c8df6db1954c69be21f0fdeeabd9d8 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9831acc6e8741009ef8f1305a7bf4be : SUCCESS in 5m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ab0d861d713545429e2cc2545c2856a0 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01132cab43e2460bbb4531a21e1a6cee : SUCCESS in 59m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d7b49e7bc6e44f7866296e0aceca255 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8e1f3e37ac11b62d261ace50cf451038b4fe9b6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 21:07:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"966aefa921bb6b072511d4d1572c72b68d901cf7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 21:08:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea2d78779bd746f8a236ba50bf4f3030\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed96bbe2c77b4f989ac024d82471983b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"354076486e8de2ac2cadab7d658bd67bcf8a2cf8","revisions":{"354076486e8de2ac2cadab7d658bd67bcf8a2cf8":{"kind":"REWORK","_number":1,"created":"2024-03-18 14:15:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/913567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/913567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/913567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/913567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/913567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/913567/1"}}},"commit":{"parents":[{"commit":"99f60d6329d545215561b198c680e79daf690192","subject":"update constraint for yaql to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99f60d6329d545215561b198c680e79daf690192"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-18 14:15:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-03-18 14:15:00.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 6.6.0","message":"update constraint for python-openstackclient to new release 6.6.0\n\nmeta: version: 6.6.0\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: I2d2021d4ff5d3fe74fd89c68da0af71b5e7d2a06\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I029b6bc3ca8e23861595eb67b23334baa4d079c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/354076486e8de2ac2cadab7d658bd67bcf8a2cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/354076486e8de2ac2cadab7d658bd67bcf8a2cf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
