)]}'
{"id":"openstack%2Frequirements~914852","triplet_id":"openstack%2Frequirements~master~I1c924f7f608abe77248f950787869111cad3081f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1c924f7f608abe77248f950787869111cad3081f","subject":"update constraint for oslo.log to new release 5.5.1","status":"MERGED","created":"2024-04-02 07:56:43.000000000","updated":"2024-04-03 08:25:15.000000000","submitted":"2024-04-03 08:19:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914852-new-release","meta_rev_id":"d5aaad9ae744f5953ab5fcfc232ebc188388d9c0","_number":914852,"virtual_id_number":914852,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 08:19:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-04-02 14:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-04-03 06:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-04-03 06:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 09:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 14:19:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-04-03 06:23:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0218ccef4bab3635229514e4cc8131b4e176f7d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-02 07:56:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d4889984ebb68ccffcf38a9e1c2440a93c31690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 09:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16999fde08a44f6b302eb1d7f8fd53d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb406ccd35e148fa90f6e188cdf057d5 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8cd2589a0943d3a6f2b01e50a0fd7d : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83234d5a70d4b458667ca41805a1b9f : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0c2cc23db574b48a23d6133a76bcf9c : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db909290fb62468c9f1ee284d0b03369 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79b9ce4f9f5d4dc29172d2c77906fdf5 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c42b8e82acb84000bdd21b9ea0b8fc34 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/736a88ef28af416c9a343822affd789c : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e1639941647348a39da63bb40f5a7899 : SUCCESS in 16m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5606a3524d6b4b689946d45b568b23fd : SUCCESS in 9m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d78e36e7e76e4c8f90f70bfc93bf446b : SUCCESS in 17m 22s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6170ed7813df47c1a32360a906cb3251 : SUCCESS in 7m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b754d18d493c483b8ba6223056c016a5 : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/91f5f890991d46d5ae0dc94c92450eef : SUCCESS in 3m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6b8534e2a5f34ee6b5ba8b04cb5a75c7 : SUCCESS in 10m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/046cbdcf7c06407a80b9aabf2699a651 : SUCCESS in 4m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0f33ff9202e347038d2079229cbf158e : SUCCESS in 6m 36s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dddcb421afaf42d0b853e03836e38f20 : SUCCESS in 8m 45s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b6240d6c228642829edd5b324684b05a : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5aa37ac9efa4ceab1992322529a30eb : SUCCESS in 8m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d546d08900554cdaa455f839973d0eba : SUCCESS in 8m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c43b2cbb02e14b19b10a521e2ff48198 : SUCCESS in 9m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/163f685b835747078710685519b05be8 : SUCCESS in 3m 32s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/090c0f5e87c94f41a57c5268e2397622 : SUCCESS in 7m 32s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bce2568d46e846928d49577bc003a648 : SUCCESS in 4m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b497dd5794ab4a9da179226e3f0b8e5c : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/973171b5b8b34067aad8350a50700cba : SUCCESS in 10m 30s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bead75d280344ba1a47cff2c218c0785 : SUCCESS in 25m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7ee4cf72fb35466eaaf7a945498fc0ee : SUCCESS in 7m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/dd541573704b4da2b744fdae7e6eb526 : SUCCESS in 9m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d44d6bfdd3b4432c8aa34f015d333725 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1f76d4bbcc94b919802e6f05a80c1cc : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25fd40e7644e48028d4aa4f13e5596ad : SUCCESS in 23m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dd36fa5cccbc4728b6ca6b95fe2f825b : SUCCESS in 6m 33s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ca4ad63b571b415ab06bb3f47e0d0fe3 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0df6d58c36c4192b08f7fcbc2bb3218 : SUCCESS in 6m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/fe402929bb27400ea89f4d3577767de6 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/359ce06519fc4cada83d1c9b8f95aeba : SUCCESS in 59m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7b50f671a0c4434aae2a3e727f85cca : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":1},{"id":"25ab220c9bb9ef9189740e8462d823bd294f8e1e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-02 14:19:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb2fc2f859d70d31a3d8b4c1066a3c006c484606","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-03 06:23:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e7a613cc67500c52fb23839dfa31eeb522615e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 06:24:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1593d75b8b6083c25fef3d3c172b76c8e4d5e37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:19:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/994ad9523c664ec9862c1f864c630421\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f5dc40f0b84d68a30adb55caca5ec9 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaec022747e6413197ae20b459b8b62f : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f46f3dc8ce74b19bcb6ce1ba911c1fc : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01d7871fdda14f35837e2ffd7580deb3 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e80f1a4708854a93a24844b1e0adfb1d : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0dc965a207fa469fb290a324ed84129c : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e981c6ada5c942618112ad63419ab729 : SUCCESS in 8m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d519e64b3a0946d59030b1ff36e0f332 : SUCCESS in 11m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d77de287505459f870e7a133c8b2553 : SUCCESS in 8m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c8347df258f4b59bbb01ed06423623c : SUCCESS in 11m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2d2a4e84d7c24821a464c0b9f6333109 : SUCCESS in 9m 47s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/94ce03632d874630ab144e198f119089 : SUCCESS in 4m 53s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e59446f654a843edbfc3d77d545e6fb1 : SUCCESS in 4m 31s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0f937468a7d949cba183466e9b14bd4d : SUCCESS in 6m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8eb5dea7cfbb495aa3f499c60c772e9e : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9828ea48cf464510824747f4f7a31bca : SUCCESS in 7m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8a5fdfcd90c44561a1f8bd030d8934c8 : SUCCESS in 7m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/db27263ad6b543febe35da1e11aa45c1 : SUCCESS in 12m 26s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/55cad3d938824801869daaaa31bc8365 : SUCCESS in 3m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f7d22593876e4ced983caa7b0c806c3b : SUCCESS in 9m 10s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/4cf034dde3e94da8b69c90cfff4fb912 : SUCCESS in 24m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/caf13615c6b7442e96d596b243315cd5 : SUCCESS in 6m 09s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d35f5d03f95543ee978281586413cf77 : SUCCESS in 12m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9ac876435c7d4f8f97a916e2954dffe6 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69ab7b4b55bf41218174539bf5a186bd : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39e6d3cd7b614db9a786d19e4f27a482 : SUCCESS in 22m 28s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5df747f2fd404ba7919c9ab56598c8d1 : SUCCESS in 3m 57s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/efcb76ac52034a1183976e26bda1f381 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c6ea711c1724c268cda09cb4a5c3f51 : SUCCESS in 6m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ae16888250c24aea9962c879410ff974 : SUCCESS in 8m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51b0ec6e6ad5410d9368afde63d1ab9d : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb3d9f97ed814a5396ccb0b40789b75b : SUCCESS in 1h 37m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b0e5e46cce54ea0318bb5f2e2e8fba6aa85b5cdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:19:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d5aaad9ae744f5953ab5fcfc232ebc188388d9c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:25:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb15f107200a434cbd22b9d6d9503dd7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6017568a1a54ef0b1f28c7fe7a18d10 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"849852467b96056237e5b8888d13a5190fefb9ce","revisions":{"849852467b96056237e5b8888d13a5190fefb9ce":{"kind":"REWORK","_number":1,"created":"2024-04-02 07:56:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/914852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/914852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/914852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/914852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/914852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/914852/1"}}},"commit":{"parents":[{"commit":"93a3e51432519a14a378a729554b5273675e945b","subject":"Merge \"update constraint for oslo.messaging to new release 14.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93a3e51432519a14a378a729554b5273675e945b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-02 07:56:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-02 07:56:43.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.5.1","message":"update constraint for oslo.log to new release 5.5.1\n\nmeta: version: 5.5.1\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I1f6a3f080a56166688f4efbfda0a34eae8305262\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1c924f7f608abe77248f950787869111cad3081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/849852467b96056237e5b8888d13a5190fefb9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/849852467b96056237e5b8888d13a5190fefb9ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
