)]}'
{"id":"openstack%2Frequirements~915166","triplet_id":"openstack%2Frequirements~master~I4ea0ca2e4df21446996951617d0d000951069bc6","project":"openstack/requirements","branch":"master","topic":"cleanup","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-04-18 06:54:25.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-17 12:29:12.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-18 06:54:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ea0ca2e4df21446996951617d0d000951069bc6","subject":"Remove stale, unused dependencies","status":"MERGED","created":"2024-04-05 16:10:31.000000000","updated":"2024-04-18 06:55:23.000000000","submitted":"2024-04-18 06:54:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915166-cleanup","meta_rev_id":"02bda3d0dd7c14257e22ce29de3eeb0b2113e6ec","_number":915166,"virtual_id_number":915166,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-18 06:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-04-17 12:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-04-18 05:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-05 19:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-05 21:08:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-04-07 06:43:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-04-17 09:52:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4a6e2ae1b4028eda291853e0f78df6f034534780","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-05 16:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b39acfffa9d505b4336ddc81b0706e6b49dcff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 19:27:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755160b759d84357833b226cafef8ed6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bcd5aac3a547cc8cb6eea2236fa3bf : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb3d135c0984c989b1004011eb982dc : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f248938f282243db838c5eba39a9429a : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85f09410e5404ee4959aa44f38998c2b : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/249fe25fd7ff463d93167b321e5e6aa7 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2615a83fd2734bfcaa70d91a2ab7853a : SUCCESS in 3m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f7573b421504fe8a792e5e1a92d634c : SUCCESS in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9d330e11ba644e9bf8f9622129eb5c5 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"134be4af6c898f010c4e810600f0602a0e7a77b8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-05 21:08:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73d21b68e7ed8f8cfe0a7733709328ef16922813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-06 09:09:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b19b8ffce75c40715ad89ca196a00acbbd6c54c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 11:16:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd17cd4e9b634c75bbf0de8ae7fc05a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe85be208ca48a087c560f56529083a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ead5272c714c9a9a46f2b3578b48a6 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7814b311c1cc4038a601a572f231b3f6 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35edcf9ce05d40bab20be0a9f8937a1e : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f84704b1bb1467a9bebe1808569a309 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/312245ca897d451db11c74b76f5216c0 : SUCCESS in 2m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d9c2e02f9914c49bc9125b304ee2385 : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71242d29cf464364866c9fdd167ce6f1 : TIMED_OUT in 2h 03m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3a75fb6cf954c5285ca66d39d8dda9c0a2faba79","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-07 06:43:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1f03695170aa6c88f0b170b849b98314fca51d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 17:37:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206cffb36e4b47be894459bae4c004d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb67bb40a0574d91982f5ff12c24e416 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4dae95143fc431d8ce3c8af139e4ee8 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbb6b57f79cb40838dcf5b174f24a22c : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17657211a7a646bcb0692f1bd1cc23b1 : SUCCESS in 6m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8dc2ca9f4f8d41b4b770bdb6604be03e : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e40787af4779427293702580dbe01ae2 : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36cb78789bfa49d2a7b01cf718b59d8e : SUCCESS in 43m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64577e9b9d7344ec96bf76972f5bb809 : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":2},{"id":"35ae6cc63e2350f1ead8f267bdb879b9e290b4bd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-08 11:45:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81636d11fbe63fafba97d73746a51f4b6d22dedd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-08 11:45:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ae5cf314bcc34de327e4e84aaf87b01b23b407d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-08 12:58:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"955e771bedcf4de6ec9de9b34e8fcd5a06101831","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-08 13:21:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"074ab58f4d6d34c61f11bf18e1c050351430e41c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 14:38:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc5289b9b2f4c1f96b91155c78014c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29573df96b804f6599e417fa1841e1a8 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b6d23133d6496db7535b1d57e25fda : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe87c00aae740f6983d9bb85e1f1509 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69a7b62ca4594b13b37e6f7e93b4e79b : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11a14ebf793d4a0783bb3afe1899f819 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c039bf17b294ed3bae5e7c0810a4d00 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c12cfc8e4234f199a3203eb0484788c : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83a19ab9b2304178a16b582e64223806 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/559597312c7a459190a4b0e887c01710 : SUCCESS in 12m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/16e4149671774481b0e1548a6a154bb3 : SUCCESS in 7m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59c19a6e3024413d91130e5366b98178 : SUCCESS in 14m 39s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/224b9c8d73c447adad17dadff3f40833 : SUCCESS in 4m 53s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/149873e50ec1422a9a9a0bb4be8f4668 : SUCCESS in 5m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6b79c98ba5954b4598c21d92b03e9379 : SUCCESS in 3m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/655951d10b0c4d50a8c7f792d3a9ba48 : SUCCESS in 9m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/bff54bb54df343cd81c0af443c4be774 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8b1ff6359c744019aad3ef427fb6d4e9 : SUCCESS in 4m 31s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/043cdb6e24a041bb82c6d22887c8a365 : SUCCESS in 8m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/352a40f5aabd4b88a51bcccabae2df50 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe62dcbe6d7b489bb919b2a857e7a908 : SUCCESS in 7m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fab7af19d2934a26901cb452e3eeb83f : SUCCESS in 7m 49s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c629a5f41f444c17baa4693168884edb : SUCCESS in 11m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b12ccc9a5e0c4be4847f769dad7c01ba : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bda6c6b1d78945449d0b7e06eb9abca6 : SUCCESS in 6m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ffbff8aa9661444e84fb5ceed5343e1d : SUCCESS in 5m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ee3d4711e62144b8a5feeda4394d2081 : SUCCESS in 4m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/71479cdf3e1c479fbc74f3489de70e2d : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/09e21d6b1ca44b35bc764a4051abc054 : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/663947bb52924ab1bc8858b8632b4ad5 : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b9884a5481214b899eedbb0ff4c707bd : SUCCESS in 9m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4923ce051bd04fa798df7d1a5d6748e4 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d31dbc5ab28f443093c226f06b1dd373 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5d4c7e0cf3047af8c70e18b95a6c808 : SUCCESS in 18m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/014cc2cbbe1b4b92bf89517adf1aa189 : SUCCESS in 5m 18s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7cbfa45d235b4261b038dadab219919e : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16c4fc695b684d84af169575e87931b3 : SUCCESS in 6m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/646980ed3b4d47f5b10905cc1e78f101 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8276de117e6433ab26b8903e4d8a637 : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b4a3b84ae244636bdf35cbda70d8a3e : SUCCESS in 1h 33m 10s","accounts_in_message":[],"_revision_number":3},{"id":"cc86efbd1dcf81571a7e212c6b0d5032ffdd05b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-16 16:52:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b20a66a50e18e20b3fa17076d140eb840641c424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 18:08:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d6b7e3b0f6404381aa8839b0e118e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cf1c412d6d4517a1dadbd58760e2fb : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f217229892cf4c5e8cb04a1a847adf09 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16ffde633e244f9cb4eee1db7424970f : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44f975dbb7a54f87bc9530cea89f5b8e : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db8d41262c0b45c4935563afe0dd6e83 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7ea6e1e4ce34b779a4394cfa7d709c6 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c99189d9900458ca0268383c5a17197 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d26100aaa1224586959df93b80f2468e : SUCCESS in 16m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e435440fb8df486dbf02edd9d17049da : SUCCESS in 21m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b8500a55333842c5abf8cb3699b2b518 : SUCCESS in 13m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/57260318b87d48f39e02477cf58eef69 : SUCCESS in 22m 11s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3e23de1cc4a544928d22841f8c36a5f4 : SUCCESS in 5m 03s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e511ce49e2d140969064c5951aea2569 : SUCCESS in 4m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a73d3c9505884244bdd85a5e24af2216 : SUCCESS in 3m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/48ab78d719c64699a01a0fac9b58acda : SUCCESS in 9m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c85a15e63847488883e149caed4454d3 : SUCCESS in 6m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c2249fe1978543c69085f7f146ed7134 : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/09f129d89688407cb711b79cff3110b4 : SUCCESS in 6m 25s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/bdd0361003fe41058be2ff864dc69544 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/484d0fa5250c483ab6d7d031ac434d41 : SUCCESS in 8m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/15dd8277ce504510a271346f4362fd64 : SUCCESS in 12m 10s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b9ce51487c3c4b178affa0fff8eb759f : SUCCESS in 7m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/99f1755b4bff4b869bd17cf47e910abc : SUCCESS in 4m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/626ce9ef84f74655bbbac872f72e271d : SUCCESS in 10m 14s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/88ebe35db66c428cbe03e7de22c2daca : SUCCESS in 4m 49s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0ea3662c00044448b4ba1822b50a678b : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/78a3749d9e214c11adf2a9ecab2f3050 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e4da09b4e1a24cb6a443bd2406377aca : SUCCESS in 17m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/48a4b3873743470e94ef66640af8b97f : SUCCESS in 6m 38s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1724dccd6c1943e58fa2fe8d4308a6b1 : SUCCESS in 12m 31s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ecaeca8ff7e145bfb2847426c1265b3a : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1beced0f4ddf4702b0699c73312b3383 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9f0ac88382c4da29bf83f470db5e08b : SUCCESS in 22m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b680936433b448e2bdc83317179acfe1 : SUCCESS in 3m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/778e7683b61b4751aca9ec2062d4fc76 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75bd4f5875ec4711ba72096c603d43c2 : SUCCESS in 6m 12s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a7427d3385b84d84b8d61644ce1d3311 : SUCCESS in 9m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb40a80e20fb48e98fa4961c01f59ee7 : SUCCESS in 54m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dc3e342bec843ce9c563d4a874790ab : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":4},{"id":"06355be1aedfb39b1099d3420c022e7cae57c266","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-16 18:17:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"717e08c08e11891bb4ad843e20de658c11a6085a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-17 09:52:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"65ccb9710964d5386ad316c2fa7c149d6cba1af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 11:24:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3e736f576532fc02d4d68d1ee922e86996835f2e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 11:25:11.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"165f28501ec24a8d4f96a038e8fa343403b1dd0a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 11:26:44.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"728b351a3418b4b1664a48f1bfa8e7afc5ac540c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 11:27:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"97d970312902b65198c9bf87ad005c55ce5aac87","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-17 12:29:12.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e072f8deb9f6e5295d58746fc58c4b5150e8f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:55:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f387049761ce45a1a06f33cc2732d431\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73439d44e030484e97d833fe2c97ff2d : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430a4e28e9ea48f680238ef57ed1f450 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8fc6950c13845948cfb3532e376dbe9 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b61f7ec5a42d4814906e6b2853ad01da : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2fea3dd08024b7d82b8851fce3bd472 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8cbc681358348d788232c852e89ac7f : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba48d618c9e04e6d932cc23b8c44f4b8 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a4f7353f3f22402481a272f883069620 : SUCCESS in 11m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c0534276a414328a0b30f9813dd6df1 : SUCCESS in 17m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/daaab8dd02f44dafa65a0bc9a88e7912 : SUCCESS in 10m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5fe52080f7d94065b49d5b11550ff174 : SUCCESS in 26m 24s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1d1323bcdcae49c9b295d9bcc7d5aada : SUCCESS in 5m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0f70756643a04a38a7ddea7c0aa9c38b : SUCCESS in 6m 23s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/701c54249182458a8e17fd4c3e3168e9 : SUCCESS in 4m 03s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/843a8d43982d496cb4f99a1e9299ebe0 : SUCCESS in 12m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/86fc055e311c402cbd3358acfb962fa2 : SUCCESS in 5m 12s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bbde70ca07fa4d90ac60dfb71fe3f0bb : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7fb0c8fac86f4676a605ce94a86f6526 : SUCCESS in 8m 56s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8a7a097d194c48b89f79b510170b896e : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/042d5253bca143bcb043c7edc2287a76 : SUCCESS in 8m 44s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/aa1c29fd7ee84d21af4e4d32c5d96655 : SUCCESS in 11m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c7993014a41a4f578cf11a4c5750d3c2 : SUCCESS in 11m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/be25ad3f23c0455491b37380b2426094 : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/147cc6a450e942a1942ad9689dfbf650 : SUCCESS in 8m 09s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/01922fb5893348cca4ef23426134b93b : SUCCESS in 5m 07s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/22c2a69c97774134acabdd692f66d603 : SUCCESS in 6m 16s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/15765e740af44de18d817372db0479bb : SUCCESS in 9m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e8c88a39cd8949bd87740a347056a702 : SUCCESS in 23m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ef67a5f9c7704be88fd9ffbd61e91efe : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4970c81df6dd457dadc2b16de7122683 : SUCCESS in 11m 20s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/82713a76334a4e77952ca99b264034cc : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0e92017e29f41b6aeebd1a3a79b1e5e : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7206ce2eaafb4f97a261c87ca532441f : SUCCESS in 25m 03s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a48e364c0b51412ab50ca32308af4dd7 : SUCCESS in 5m 34s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c06514f38dea48ab8fd2481b023366a7 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ceac80b0ca945b58d5e6153c42b7205 : SUCCESS in 7m 10s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3d6923d4ff5847acbf638bc59abb4edd : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/918d38ff890241a982b1fedc15d6fec1 : SUCCESS in 51m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb9583f0f57c4c529d98e6a193803d0a : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":5},{"id":"6603b8be893bbf6f201fcaf4da97fc97b53765fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-18 05:21:46.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a04f3f036ebc96027b51f8b48362c447d3bdc511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 05:22:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c019e6fb567f5bb3d9a2070d7c53009b1621739b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 06:54:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27514e532acc460b80e9d8b785f1d3f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eee1eeb23374bf2927c82e76c89c1d5 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c0e65132904379a5516716df4ed64c : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a09aaea5f3ed43e397bd24b77379160c : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a997234f1ba045379cb8dc9186e632a9 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a46f29aa547a4d2d84127de141440926 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b2e4d18d8d54b4095ff33a9602fcf2c : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/652a1eb4ff4540e49911933629ab0a4f : SUCCESS in 11m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/62569a9f8a46479aa298ccaf6ca469ee : SUCCESS in 17m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/951ab68992fa497ca870c42a9ddac8b3 : SUCCESS in 11m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eac6b38d8f6d4f0c8441eb469c2d1e68 : SUCCESS in 17m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/913b2aada4ad4068b2391010f7d593c6 : SUCCESS in 8m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2d6f706e92e74aa1b11a3f42a8a27ec4 : SUCCESS in 4m 24s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/edf82f8b2c514e6ca6338bc30b68cb64 : SUCCESS in 4m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c88b36b44bd044018a601cd06fced0bf : SUCCESS in 5m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/478c980c32c24ee4a7a89e351dacd062 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/036cb382d594486498415380f9eaaacc : SUCCESS in 6m 38s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/639fab4db25b44e7a92c1be1e316d8ed : SUCCESS in 9m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/49e5bd74b983450b918510465426964a : SUCCESS in 7m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/31d3a3a0c2644eb6a21ada032457dd37 : SUCCESS in 3m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0f7268b808604fd39635fd0e2ed665ae : SUCCESS in 9m 02s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c9d1ffeb4a374b268ff507ca9f45501a : SUCCESS in 15m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79b24f93a560493da688578d638ddefd : SUCCESS in 5m 06s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/68a60cb0826b4588a341f3462b3b29e5 : SUCCESS in 11m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d42fda7a76ee4bc3b15cd5d3cabd0ba5 : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f4a7fab3d6e14228935148af7fd4034e : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d591adcff26a417085142f7cdd7a79f4 : SUCCESS in 22m 54s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a981c3af22904b198e46392faf746dd0 : SUCCESS in 3m 57s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ffe7dc4d81b143be9fa48fc5f59000d8 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0df2402c19a43d69a4702d55f98b5dd : SUCCESS in 6m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f2d345fff6a445beb133a105e99ac57b : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1945be939dbe482eb4e28cbf10b8fdaf : SUCCESS in 1h 02m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/684913cc3695409cace1f1ac6869240c : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":5},{"id":"25aba110f22e49d1103922d2b37a7cebb616091e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 06:54:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"02bda3d0dd7c14257e22ce29de3eeb0b2113e6ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 06:55:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd61da0fb50645a19a6e3149314a5e9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b25421a0b414cebb81381b65b4237e2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"07d44b297bfe3d30685327890729d82d504850bb","revisions":{"4c9b00c0200ae2324aa117f6bc8528b3209ca6c5":{"kind":"REWORK","_number":1,"created":"2024-04-05 16:10:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/915166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/915166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/915166/1"}}},"commit":{"parents":[{"commit":"d7212e13aa564d8302c8aaba9ec8904a8cfd2056","subject":"Remove Cinder volume-driver specific libraries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7212e13aa564d8302c8aaba9ec8904a8cfd2056"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:36:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 15:49:29.000000000","tz":60},"subject":"Remove stale, unused dependencies","message":"Remove stale, unused dependencies\n\n- couchdb\n  A client library for couchdb that we indicated was only used for\n  tests. Trove references it but doesn\u0027t list it in requirements and\n  instead seems to install from apt\n\n- django-pymemcache\n  A django cache backend that allows you to use pymemcache. No users in\n  openstack namespace.\n\n- kafka-python\n  A client for kafka, replaced by kafka-python3. No users in openstack\n  namespace.\n\n- marathon\n  No users in openstack namespace.\n\n- oauth2client\n  A client library for oauth2. No users in openstack namespace.\n\n- openstack.nose-plugin, nose-exclude, nosehtmloutput, nosexcover, proboscis\n  Various plugins for nose. No users in openstack namespace.\n\n- pyinotify\n  Python wrappers for the inotify API. Unmaintained since 2015.\n  Used by some oslo libs until recently but no longer used.\n\n- python-rsdclient\n  Abandoned client library. We no longer depend on this in\n  openstackclient.\n\n- sphinx-testing\n  A testing package for Sphinx. Replace by the testing module in Sphinx\n  itself.\n\n- unittest2\n  A backport of unittest features to Python 2.4+. No longer needed.\n\n- zake\n  A set of test utilities for kazoo. An optional dependency for tooz but\n  not required anywhere.\n\nThere are a couple of libraries that we should work to replace, but\nthose are not tackled here since they need more work elsewhere.\n\n- influxdb\n  Deprecated and replaced by influxdb3-python, per the GitHub page\n  https://github.com/influxdata/influxdb-python\n\n- jsonpath-rw, jsonpath-rw-ext\n  Untouched in 8 and 4 years, respectively, but has a few users dotted\n  about.\n\n- nose\n  Ancient test runner that\u0027s been untouched in 8 years.\n\n- os-client-config\n  Unmaintained. Users should switch to openstacksdk.\n\n- oslosphinx\n  Unmaintained. Users should switch to openstackdocstheme or another\n  theme if not OpenStack-affiliated.\n\n- pydotplus\n  A library for interfacing with Graphviz\u0027s Dot language. Super old.\n\n- pykmip\n  Used by barbican but unmaintained.\n\n- pysendfile\n  A test dependency for Glance. Unmaintained since 2014 (!!!).\n\n- pyngus\n  A dependency of oslo.messaging. Unmaintained since 2020.\n\n- python-consul\n  A dependency of masakari-monitors and an optional dependency of tooz.\n  Unmaintained since 2018.\n\n- requests-aws\n  An optional dependency of ceilometer. Unmaintained since 2015.\n\n- sadisplay\n  An optional dependency of octavia. Unmaintained since 2018.\n\n- testrepository\n  A test runner. Unmaintained since 2014 and replaced by stestr.\n\n- testresources, testscenarios\n  Add-ons for unittest. Unmaintained since 2016 and 2015, respectively.\n\nChange-Id: I4ea0ca2e4df21446996951617d0d000951069bc6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c9b00c0200ae2324aa117f6bc8528b3209ca6c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c9b00c0200ae2324aa117f6bc8528b3209ca6c5"}]},"branch":"refs/heads/master"},"d3988163c174cb95bbe84a390788b656606b3611":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-06 09:09:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/915166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/915166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/915166/2"}}},"commit":{"parents":[{"commit":"15c40d03b2178bf1db5cccb069cf608724998ac6","subject":"Remove Cinder volume-driver specific libraries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c40d03b2178bf1db5cccb069cf608724998ac6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:36:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-06 09:05:47.000000000","tz":60},"subject":"Remove stale, unused dependencies","message":"Remove stale, unused dependencies\n\n- couchdb\n  A client library for couchdb that we indicated was only used for\n  tests. Trove references it but doesn\u0027t list it in requirements and\n  instead seems to install from apt\n\n- django-pymemcache\n  A django cache backend that allows you to use pymemcache. No users in\n  openstack namespace.\n\n- kafka-python\n  A client for kafka, replaced by kafka-python3. No users in openstack\n  namespace.\n\n- marathon\n  No users in openstack namespace.\n\n- oauth2client\n  A client library for oauth2. No users in openstack namespace.\n\n- openstack.nose-plugin, nose-exclude, nosehtmloutput, nosexcover, proboscis\n  Various plugins for nose. No users in openstack namespace.\n\n- pyinotify\n  Python wrappers for the inotify API. Unmaintained since 2015.\n  Used by some oslo libs until recently but no longer used.\n\n- python-rsdclient\n  Abandoned client library. We no longer depend on this in\n  openstackclient.\n\n- sphinx-testing\n  A testing package for Sphinx. Replace by the testing module in Sphinx\n  itself.\n\n- unittest2\n  A backport of unittest features to Python 2.4+. No longer needed.\n\n- zake\n  A set of test utilities for kazoo. An optional dependency for tooz but\n  not required anywhere.\n\nThere are a couple of libraries that we should work to replace, but\nthose are not tackled here since they need more work elsewhere.\n\n- influxdb\n  Deprecated and replaced by influxdb3-python, per the GitHub page\n  https://github.com/influxdata/influxdb-python\n\n- jsonpath-rw, jsonpath-rw-ext\n  Untouched in 8 and 4 years, respectively, but has a few users dotted\n  about.\n\n- nose\n  Ancient test runner that\u0027s been untouched in 8 years.\n\n- os-client-config\n  Unmaintained. Users should switch to openstacksdk.\n\n- oslosphinx\n  Unmaintained. Users should switch to openstackdocstheme or another\n  theme if not OpenStack-affiliated.\n\n- pydotplus\n  A library for interfacing with Graphviz\u0027s Dot language. Super old.\n\n- pykmip\n  Used by barbican but unmaintained.\n\n- pysendfile\n  A test dependency for Glance. Unmaintained since 2014 (!!!).\n\n- pyngus\n  A dependency of oslo.messaging. Unmaintained since 2020.\n\n- python-consul\n  A dependency of masakari-monitors and an optional dependency of tooz.\n  Unmaintained since 2018.\n\n- requests-aws\n  An optional dependency of ceilometer. Unmaintained since 2015.\n\n- sadisplay\n  An optional dependency of octavia. Unmaintained since 2018.\n\n- testrepository\n  A test runner. Unmaintained since 2014 and replaced by stestr.\n\n- testresources, testscenarios\n  Add-ons for unittest. Unmaintained since 2016 and 2015, respectively.\n\nChange-Id: I4ea0ca2e4df21446996951617d0d000951069bc6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3988163c174cb95bbe84a390788b656606b3611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3988163c174cb95bbe84a390788b656606b3611"}]},"branch":"refs/heads/master"},"d91a23478e9391155806996c0cff2e6fbf5f6524":{"kind":"REWORK","_number":3,"created":"2024-04-08 12:58:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/915166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/915166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/915166/3"}}},"commit":{"parents":[{"commit":"5c041106f8594211eec0c11201a3222dcdb95bdf","subject":"Remove Cinder volume-driver specific libraries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c041106f8594211eec0c11201a3222dcdb95bdf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:36:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-08 11:44:11.000000000","tz":60},"subject":"Remove stale, unused dependencies","message":"Remove stale, unused dependencies\n\n- couchdb\n  A client library for couchdb that we indicated was only used for\n  tests. Trove references it but doesn\u0027t list it in requirements and\n  instead seems to install from apt\n\n- django-pymemcache\n  A django cache backend that allows you to use pymemcache. No users in\n  openstack namespace.\n\n- kafka-python\n  A client for kafka, replaced by kafka-python3. No users in openstack\n  namespace.\n\n- marathon\n  No users in openstack namespace.\n\n- oauth2client\n  A client library for oauth2. No users in openstack namespace.\n\n- openstack.nose-plugin, nose-exclude, nosehtmloutput, nosexcover, proboscis\n  Various plugins for nose. No users in openstack namespace.\n\n- python-rsdclient\n  Abandoned client library. We no longer depend on this in\n  openstackclient.\n\n- sphinx-testing\n  A testing package for Sphinx. Replace by the testing module in Sphinx\n  itself.\n\n- unittest2\n  A backport of unittest features to Python 2.4+. No longer needed.\n\n- zake\n  A set of test utilities for kazoo. An optional dependency for tooz but\n  not required anywhere.\n\nThere are a couple of libraries that we should work to replace, but\nthose are not tackled here since they need more work elsewhere.\n\n- influxdb\n  Deprecated and replaced by influxdb3-python, per the GitHub page\n  https://github.com/influxdata/influxdb-python\n\n- jsonpath-rw, jsonpath-rw-ext\n  Untouched in 8 and 4 years, respectively, but has a few users dotted\n  about.\n\n- nose\n  Ancient test runner that\u0027s been untouched in 8 years.\n\n- os-client-config\n  Unmaintained. Users should switch to openstacksdk.\n\n- oslosphinx\n  Unmaintained. Users should switch to openstackdocstheme or another\n  theme if not OpenStack-affiliated.\n\n- pydotplus\n  A library for interfacing with Graphviz\u0027s Dot language. Super old.\n\n- pyinotify\n  Python wrappers for the inotify API. Unmaintained since 2015.\n  Used by oslo.log which is dropping it but has not completed this work\n  yet.\n\n- pykmip\n  Used by barbican but unmaintained.\n\n- pysendfile\n  A test dependency for Glance. Unmaintained since 2014 (!!!).\n\n- pyngus\n  A dependency of oslo.messaging. Unmaintained since 2020.\n\n- python-consul\n  A dependency of masakari-monitors and an optional dependency of tooz.\n  Unmaintained since 2018.\n\n- requests-aws\n  An optional dependency of ceilometer. Unmaintained since 2015.\n\n- sadisplay\n  An optional dependency of octavia. Unmaintained since 2018.\n\n- testrepository\n  A test runner. Unmaintained since 2014 and replaced by stestr.\n\n- testresources, testscenarios\n  Add-ons for unittest. Unmaintained since 2016 and 2015, respectively.\n\nChange-Id: I4ea0ca2e4df21446996951617d0d000951069bc6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d91a23478e9391155806996c0cff2e6fbf5f6524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d91a23478e9391155806996c0cff2e6fbf5f6524"}]},"branch":"refs/heads/master"},"689ae847ab680f31b61366921b1e4295e5097c68":{"kind":"REWORK","_number":4,"created":"2024-04-16 16:52:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/915166/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/915166/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/915166/4"}}},"commit":{"parents":[{"commit":"0a72885e1063303245e9f1fc2243d4b601722b14","subject":"Remove irrelevant notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a72885e1063303245e9f1fc2243d4b601722b14"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:36:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-16 16:45:43.000000000","tz":60},"subject":"Remove stale, unused dependencies","message":"Remove stale, unused dependencies\n\n- couchdb\n  A client library for couchdb that we indicated was only used for\n  tests. Trove references it but doesn\u0027t list it in requirements and\n  instead seems to install from apt\n\n- django-pymemcache\n  A django cache backend that allows you to use pymemcache. No users in\n  openstack namespace.\n\n- kafka-python\n  A client for kafka, replaced by kafka-python3. No users in openstack\n  namespace.\n\n- marathon\n  No users in openstack namespace.\n\n- oauth2client\n  A client library for oauth2. No users in openstack namespace.\n\n- openstack.nose-plugin, nose-exclude, nosehtmloutput, nosexcover, proboscis\n  Various plugins for nose. No users in openstack namespace.\n\n- python-rsdclient\n  Abandoned client library. We no longer depend on this in\n  openstackclient.\n\n- sphinx-testing\n  A testing package for Sphinx. Replace by the testing module in Sphinx\n  itself.\n\n- unittest2\n  A backport of unittest features to Python 2.4+. No longer needed.\n\n- zake\n  A set of test utilities for kazoo. An optional dependency for tooz but\n  not required anywhere.\n\nThere are a couple of libraries that we should work to replace, but\nthose are not tackled here since they need more work elsewhere.\n\n- influxdb\n  Deprecated and replaced by influxdb3-python, per the GitHub page\n  https://github.com/influxdata/influxdb-python\n\n- jsonpath-rw, jsonpath-rw-ext\n  Untouched in 8 and 4 years, respectively, but has a few users dotted\n  about.\n\n- nose\n  Ancient test runner that\u0027s been untouched in 8 years.\n\n- os-client-config\n  Unmaintained. Users should switch to openstacksdk.\n\n- oslosphinx\n  Unmaintained. Users should switch to openstackdocstheme or another\n  theme if not OpenStack-affiliated.\n\n- pydotplus\n  A library for interfacing with Graphviz\u0027s Dot language. Super old.\n\n- pyinotify\n  Python wrappers for the inotify API. Unmaintained since 2015.\n  Used by oslo.log which is dropping it but has not completed this work\n  yet.\n\n- pykmip\n  Used by barbican but unmaintained.\n\n- pysendfile\n  A test dependency for Glance. Unmaintained since 2014 (!!!).\n\n- pyngus\n  A dependency of oslo.messaging. Unmaintained since 2020.\n\n- python-consul\n  A dependency of masakari-monitors and an optional dependency of tooz.\n  Unmaintained since 2018.\n\n- requests-aws\n  An optional dependency of ceilometer. Unmaintained since 2015.\n\n- sadisplay\n  An optional dependency of octavia. Unmaintained since 2018.\n\n- testrepository\n  A test runner. Unmaintained since 2014 and replaced by stestr.\n\n- testresources, testscenarios\n  Add-ons for unittest. Unmaintained since 2016 and 2015, respectively.\n\nChange-Id: I4ea0ca2e4df21446996951617d0d000951069bc6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/689ae847ab680f31b61366921b1e4295e5097c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/689ae847ab680f31b61366921b1e4295e5097c68"}]},"branch":"refs/heads/master"},"07d44b297bfe3d30685327890729d82d504850bb":{"kind":"REWORK","_number":5,"created":"2024-04-17 11:24:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/915166/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/915166/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/915166/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/915166/5"}}},"commit":{"parents":[{"commit":"0a72885e1063303245e9f1fc2243d4b601722b14","subject":"Remove irrelevant notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a72885e1063303245e9f1fc2243d4b601722b14"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:36:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 11:04:16.000000000","tz":60},"subject":"Remove stale, unused dependencies","message":"Remove stale, unused dependencies\n\n- couchdb\n  A client library for couchdb that we indicated was only used for\n  tests. Trove references it but doesn\u0027t list it in requirements and\n  instead seems to install from apt\n\n- django-pymemcache\n  A django cache backend that allows you to use pymemcache. No users in\n  openstack namespace.\n\n- marathon\n  No users in openstack namespace.\n\n- openstack.nose-plugin, nose-exclude, nosehtmloutput, nosexcover, proboscis\n  Various plugins for nose. No users in openstack namespace.\n\n- python-rsdclient\n  Abandoned client library. We no longer depend on this in\n  openstackclient.\n\n- sphinx-testing\n  A testing package for Sphinx. Replace by the testing module in Sphinx\n  itself.\n\n- unittest2\n  A backport of unittest features to Python 2.4+. No longer needed.\n\nThere are a couple of libraries that we should work to replace, but\nthose are not tackled here since they need more work elsewhere.\n\n- influxdb\n  Deprecated and replaced by influxdb3-python, per the GitHub page\n  https://github.com/influxdata/influxdb-python\n\n- jsonpath-rw, jsonpath-rw-ext\n  Untouched in 8 and 4 years, respectively, but has a few users dotted\n  about.\n\n- kafka-python\n  A client for kafka, replaced by kafka-python3. No users in openstack\n  namespace.\n\n- nose\n  Ancient test runner that\u0027s been untouched in 8 years.\n\n- oauth2client\n  A client library for oauth2. No users in openstack namespace.\n\n- os-client-config\n  Unmaintained. Users should switch to openstacksdk.\n\n- oslosphinx\n  Unmaintained. Users should switch to openstackdocstheme or another\n  theme if not OpenStack-affiliated.\n\n- pydotplus\n  A library for interfacing with Graphviz\u0027s Dot language. Super old.\n\n- pyinotify\n  Python wrappers for the inotify API. Unmaintained since 2015.\n  Used by oslo.log which is dropping it but has not completed this work\n  yet.\n\n- pykmip\n  Used by barbican but unmaintained.\n\n- pysendfile\n  A test dependency for Glance. Unmaintained since 2014 (!!!).\n\n- pyngus\n  A dependency of oslo.messaging. Unmaintained since 2020.\n\n- python-consul\n  A dependency of masakari-monitors and an optional dependency of tooz.\n  Unmaintained since 2018.\n\n- requests-aws\n  An optional dependency of ceilometer. Unmaintained since 2015.\n\n- sadisplay\n  An optional dependency of octavia. Unmaintained since 2018.\n\n- testrepository\n  A test runner. Unmaintained since 2014 and replaced by stestr.\n\n- testresources, testscenarios\n  Add-ons for unittest. Unmaintained since 2016 and 2015, respectively.\n\n- zake\n  A set of test utilities for kazoo used by a number of projects such as\n  tooz, mistral and taskflow. Unmaintained since 2015.\n\nChange-Id: I4ea0ca2e4df21446996951617d0d000951069bc6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d44b297bfe3d30685327890729d82d504850bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d44b297bfe3d30685327890729d82d504850bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
