)]}'
{"id":"openstack%2Frequirements~915411","triplet_id":"openstack%2Frequirements~master~Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff","project":"openstack/requirements","branch":"master","topic":"types-test-requirements","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-04-29 14:38:17.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-04-29 13:13:10.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-04-29 14:38:17.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-29 14:38:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff","subject":"Add some more types packages","status":"MERGED","created":"2024-04-10 11:17:12.000000000","updated":"2024-04-29 20:45:43.000000000","submitted":"2024-04-29 14:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915411-types-test-requirements","meta_rev_id":"93997c00f5b25dd7b7e9ce1e6c0b6a6be7a268a0","_number":915411,"virtual_id_number":915411,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-29 14:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-04-29 13:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-04-29 12:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2024-04-29 13:13:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 11:49:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2024-04-10 12:15:23.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-04-10 14:52:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 15:02:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-04-10 15:19:13.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"state":"CC"},{"updated":"2024-04-10 15:19:13.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-04-29 12:12:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-29 12:12:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-04-29 12:12:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"9077ace729b154fbd1cc2367389c5e978bb75142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 11:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8c9f65a94926d0e2c7cdde02351201266155591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 11:19:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ad9a30bee0995cc91fdfe90274b8c585b5275654","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 11:25:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"af57171a54a95dec159a8f00ae8356a724b139ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 11:32:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4ba940d044df00f9dfded69b28f8309eec17b911","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 11:34:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"db1a234dd89dc50b965112d138c5de035ea7d3f4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-10 11:49:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee7c38f465da76c58b9c5737027115bad1684a15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 13:03:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fc426fe8a4a7d393e3c64a4e8d93fba27bf1b5ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 13:16:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b5fcda4d8949717136bc7d896510245fbf5a5fd9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 13:17:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55dc53e0105e0fc2e50d427189a238b9fcac7ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 14:52:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d25b61159314b4f917672c7a5ab1cf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb09879078d41c294ace3d64bbe7251 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd678ae3dddb4a9899fd63c6b0804f56 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce8d0e8318846058f9cfd27012a7d8a : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64c0e4d445d74d70abfacfbdbbed5330 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d839b7f2edb649298dca1780d4debb3a : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f17c7fc0b02048ed8f44403715eda31e : FAILURE in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1307644590374eb69939a4003f6792f2 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ebe0f232099e4c64828a81e753b97ac5 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a653366a8fd3447dac9546e27cb25a68 : SUCCESS in 14m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/27ed2cf358764aaba66c90c96d41420f : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a9ed2d8ea7214572b3815cd532326054 : SUCCESS in 17m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/28c230e5e85546a4a35842657c992907 : SUCCESS in 5m 04s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b38172e4a5634de299432b6e2e8b3926 : SUCCESS in 5m 01s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0681efb9ac1a4299afb78e01e11fa0f7 : SUCCESS in 4m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ce88177b7bd44616a3ed25924b1e5155 : SUCCESS in 11m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/07139159d1b14152b68d8a429c3663cc : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/76081e903ba44a22812884d68bf4b03d : SUCCESS in 5m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/988a0b4d25164b41b735570709d9d12d : SUCCESS in 8m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c8e0303a37a94a289cef76204a4c8590 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2653ef45c9ec464c912a3ee9da3eaf78 : SUCCESS in 7m 15s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1eb8939bc0594387923be31bcc38fa8c : SUCCESS in 8m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5348c7741b9f4f088c72960511f44853 : SUCCESS in 10m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b1d416651f784e8f9eba197ecf5695ad : SUCCESS in 5m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/33ddb3544ec54d7fb6216d6070db256e : SUCCESS in 8m 52s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ac59d342bc894218bc81397cc82949d5 : SUCCESS in 4m 11s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f74be929d2c44fdbbd0890a677ce7ae3 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/c77c76bf95734c71a9babe745d573d43 : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d5d4c3e383334d329a50f4c99e7fea1e : SUCCESS in 23m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1a5c0a14543c414eab0e5490ca2c6822 : SUCCESS in 5m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f951c598f5ed43b9abb33e7a80fb1ced : SUCCESS in 13m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a0cbfa1e9932451cbac4997fac6f5e7b : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2acffb04ba32483da0f8aa49f7c0076d : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4aa1f204a16d478e862d9e1cfde78c65 : SUCCESS in 18m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/eee89f42edd74d7092a8baa348254f28 : SUCCESS in 4m 27s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3670543c88df4675b0ec25d285451c7a : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37dd4747416b4c31b670de4ec1efa095 : SUCCESS in 5m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/bbc14dea0caa431f868d982bfa23aff9 : SUCCESS in 14m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/738b043535b14785b298c1a761e62787 : SUCCESS in 1h 04m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38563fadb4a948bf842bd3b95d6bcb85 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":5},{"id":"542f462aec1e8a043ee2689257cd98bd7c7e61bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 15:01:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bc8394bded85f6e4f612336ba138cb624089da28","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-10 15:02:58.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f74fd0eefa84380043d699ecb05cc80c7a861ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 16:24:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24496992a68d407c899a4dbbbc927f28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2646d1e7d04669a5b77e3464fc78e0 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b61c8b6628d43078d7c2c0171074081 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5748d51914f4c18bc6544a1937c40de : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4138363edd884b4a8ec1082dc8cc1f59 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0f7efc11d6e44b2ab636e54f1210f0b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9512c1e3288f49b59c932f971ef160fa : FAILURE in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9bc1921f7ec414db2b5bdddad568630 : SUCCESS in 4m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2842f1177a6e419bb418071a96150202 : SUCCESS in 10m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/609f9e0551644d30b576a2416d8320e5 : SUCCESS in 16m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9f397b5393e4450a23ab98aa202dda1 : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/70b6010713774d37af26dd668a9caa71 : SUCCESS in 19m 36s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/54ab62d16cc34604a984bd2261b9bc6b : SUCCESS in 4m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/129248db156d4cebbf496ce268a2078d : SUCCESS in 4m 43s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1b0bf451e1ca404a944bf7319a449ca2 : SUCCESS in 4m 14s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b9e5f8cd3d7846668226f946aec56999 : SUCCESS in 10m 52s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/fe53acad381f450cad49b358f01915c8 : SUCCESS in 4m 31s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b10441b9b626471c89b73d1d6ead8e54 : SUCCESS in 5m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a4a191d72b204ddabb57ad912362a322 : SUCCESS in 8m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/22e521b87ea94e17b0743de2f71a4892 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bf60081f20147d6be138de4b2a0404c : SUCCESS in 6m 41s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/640e583ac64040bbbb91dccfc761af89 : SUCCESS in 10m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0f93f7c61f9a48b78e34b1e9710ead3a : SUCCESS in 10m 51s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/02d61b58e9db43988a78037d35d54694 : SUCCESS in 3m 30s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d1039a11980d4f958d86232c7788b954 : SUCCESS in 8m 12s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/d84e4a15ab3c4ab9b20b81c2678b3a7d : SUCCESS in 5m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d9dac043fbd84e83b570f10679743752 : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a71f10aa0f9a437fae46e409033fe23b : POST_FAILURE in 12m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/4397499416204889bb85ab5c30163494 : SUCCESS in 22m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0316a4dfeb8e4b1685a98285c107a7b1 : SUCCESS in 8m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/70aff3dd99e9411cb6cbe3642227aeb8 : SUCCESS in 9m 23s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/154ca4bfe9084031bbdf514e813cd3aa : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0cf968246db7465596db0ffbd9d16775 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2a975833a7b4131acec9c384633748a : SUCCESS in 25m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/458a807332f14a51ad29ee2160203ca9 : SUCCESS in 6m 21s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9b28d367120243febe3acf7f8342c242 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/499fae0a727148d4ba69998d0e64015b : SUCCESS in 6m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d269af9f9c7e4fad9c9811d59540f33a : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e5d1087a1684e998d0c5764387e1873 : SUCCESS in 47m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13169d18ce1b4da9b8c54e49899ed459 : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":6},{"id":"9cf714dacc7091ca68d59b45ddea9fbc0e19ade9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 16:32:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"0a842aceee64560cc0ca3081a4ad615253abc3f7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 16:33:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e305cee7519bf59d253c3572a06f1cfcd10de174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 18:12:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee13bb818024ed7a129ce34c78906e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc169af473a49db9751a76c860186f0 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96d9b3ef3b8454ab4171c5c0024aaa2 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f2c0472102f4415a97242f2ffc043f0 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a584d71dee3a44d3b1fe4bb7b87437ac : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93717182edb949da8ab8ec7ab700e96f : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c0de2ac2cd94a448da58a9a723fbee9 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/397643bbecda45ca9d335134f7499cb2 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a132811c829845efb45921a3db707ede : SUCCESS in 7m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a6584b5792247949a8205d64d6f663a : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a2c5c3fafb441caaf636546ae88d0e6 : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d3732a3ede114d0eacf670d5ca0702bd : SUCCESS in 19m 41s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6ac77057f2fb4fd2accf52624dd072a6 : SUCCESS in 4m 12s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/19bf2a98da61498a9a54db4704fa2bd1 : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/179b69c99e2d467a8d2c3af589b4f9bf : SUCCESS in 3m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/59f0a274a9b9465499c786cfccf5f1c5 : SUCCESS in 11m 38s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e3a4c52c8f6b4b93b0b5dae22ad2b0a3 : SUCCESS in 6m 48s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/65d65b13528349a6b605acde7d8fb2df : SUCCESS in 5m 06s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3ec6fdafad15467fa93f974b30bb77dc : SUCCESS in 9m 35s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/68c0e7dfd53749c3b02fa967ef45ae61 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a041773609ed41329b01948f3199297a : SUCCESS in 7m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/88a3785d9bff41d996c6634ca7b3b349 : SUCCESS in 7m 56s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4d8c81a562494c539f3bef197c87aaad : SUCCESS in 11m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e99a7b9ec34048f093641485b2da66d7 : SUCCESS in 3m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/af7b6f2574c9436d9cf69cab8c9ba6da : SUCCESS in 9m 08s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/361f6137b53a4243adda68f35003f43f : SUCCESS in 4m 36s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2824d08fb47c472997d52f87a2542b7d : SUCCESS in 3m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/748b7d8ec1bf4cbea532bfe43bff5530 : SUCCESS in 7m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/85aacc29dc714878933aac798503e299 : SUCCESS in 25m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/869e526197bb44eaa90a9b1788598ef8 : SUCCESS in 7m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/defc49227c6d4ee8adad1298b9e84258 : SUCCESS in 8m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/85ee42ef3b0c4298a07adeb0a7a4ba0e : SUCCESS in 3m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71c41cf1fa8648ea96b06e6ea034c0f5 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80dec764b6f34b099616322185686930 : SUCCESS in 23m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/49797d46814c4b34ac75c74fc0e64375 : SUCCESS in 3m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/357c865f8be8451fa775529d622cea2a : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaa27bca385c40178d4feb405961c5cd : SUCCESS in 5m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/659eca30777b4593a7d4ef08d13fd27c : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59256d9af1d24ff29938600783f89527 : SUCCESS in 53m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6911757f615740d68e9e20727d53d43a : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":7},{"id":"0d0cb72cf978a2fd2606d67ac0b009671404a38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 12:30:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3e2bb4a63949a334f77ddf826756f05ea20e8a31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-15 13:04:52.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"59b61c98d28b5366aef3f9a7cef378e7bb3d5a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 14:14:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7009cf7ba64542a189fcde018b43439b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de73e1b8c97e4e54af59567138282586 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f60079791114dc5bb943d1b8673ce54 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c64e67f23c294b0ebc4fd29b19a0ef52 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c930549b61974b1ea1db3372b9068154 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/203531fca4ae48db9ef684380ee863a3 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd7cfb4f09814f6cb5e2716637f58749 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02474e316b9d4e439ac347629ee6327e : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/410ab4533290459c9b30a7b67dac1b92 : SUCCESS in 16m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/961cb8fb47364e09b6ee14dc44dc2a35 : SUCCESS in 16m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/471ad068af894f78aef033dab21c5103 : SUCCESS in 11m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2f3a5a4d5b56444f96a8a069f714043e : SUCCESS in 15m 55s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/afef5976925945a88b24185ef3e94879 : SUCCESS in 5m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2c39162ee8444b6f986a749b7f84ad5c : SUCCESS in 5m 01s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e7e12a57049341549743954adc799484 : SUCCESS in 4m 06s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1c5b50ece81e44449241ddd21071ae8a : SUCCESS in 8m 09s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/02a3285b334749938c4c88ffac4ec98f : SUCCESS in 4m 52s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/acb08e84469442b59f4aa1628bdabf88 : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dc3f1109924647888f2387a9f8bdec17 : SUCCESS in 6m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/58e7b85a23c44b23ac33acd50683e3f1 : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75d3a64bf1fa4545b2f03739e8ec8c50 : SUCCESS in 7m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/33124dc0d688467f91ffe3f484d79b52 : SUCCESS in 9m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4a9d1936d19c488a83d27778e92d9e25 : SUCCESS in 8m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/faf1ea40b2314b52bc536a3dc12fc25a : SUCCESS in 4m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/99df7ca951944d448fcec3000a125e62 : SUCCESS in 8m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/55e98aa18f0c4b87bab6135a1be3e257 : SUCCESS in 4m 21s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f62cb8406c14467c8bb793bc04213595 : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4851fa34191f480bb1986810b1645eae : SUCCESS in 12m 10s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0d9a2fa70ece4f38a368779290503ab3 : SUCCESS in 16m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e3023a6dc0d4eea9e3644a36fc2116c : SUCCESS in 5m 11s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3bc1f021d29e46e6b85895f386554acb : SUCCESS in 11m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/92b03f3822444e5b8e98436048e64507 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/212ad08a43ef4572b3f342b55e9dc1c7 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3135e53bfd544e0aa59f06d12b88fe5 : SUCCESS in 21m 55s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d027adaaac2d445eafb633d28fdd421b : SUCCESS in 4m 59s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1859b91eaaf7448cbc10e0d21e761dee : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/20fed47f3e7746bbb241be4786d448d3 : SUCCESS in 5m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/782f2e069b764edfad31766ade4cc7ec : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cec6ee9df9d4fa3aa7b722084b21acc : SUCCESS in 1h 03m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84dccc9179d14a7e84f540eb612c281f : SUCCESS in 1h 42m 29s","accounts_in_message":[],"_revision_number":8},{"id":"bf08e13936d1da9f40e96ac405dc052b1ae34a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-24 16:21:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7582f6184b878bb173085c501441960fb2a5de8f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-24 16:22:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"51a4d6973153a73a8173bed7a63b4817712ea0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 18:10:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f5501643e0419ea1e5198245ba8ee7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cca976c65c4422aa1b039c13add8294 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cda8c4a05314e75b91fb7cd3a838184 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5986aebb5424f0cb5bb5e47de51f0b4 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f9ccf45791d41be912592ed15b1923b : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80c9beb00ded429e9b3c236e0d448c73 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94d0c71da3f849e1884c448426159ca6 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b586e1fd05c4717b90beefe9675534d : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/724de0db1bb148c88696e62220878174 : SUCCESS in 10m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c123c43d77cc4ef8ac5d7e2e758d642b : SUCCESS in 21m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/07035e27e9bf495d8f577be398a2da58 : SUCCESS in 11m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2889f48227144ec083d2b190d603d42d : SUCCESS in 19m 47s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e8e14b5a702743d6bfbc0a6c35cb61d8 : SUCCESS in 6m 39s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/67c1c634dc8544eebce3fb982cf7f371 : SUCCESS in 5m 51s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8656b94f80754f3b950f5b1be9914570 : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0899fc37fa4541a3aa47ad4b285eaadf : SUCCESS in 12m 03s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0a34f1105a6f4dbd99722f3d413fc5a7 : SUCCESS in 5m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/027c68058cba4cf3b534c3ff6941a92d : SUCCESS in 5m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b300705efac6487e989d59d6b00ca12e : SUCCESS in 7m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/898a53f86b5d4aadbd2dff6088be1be5 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9c44b079c624835bceda57173341de4 : SUCCESS in 8m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/32ca379f83d7439e994439b1f3ef6eee : SUCCESS in 10m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5bef56af56a04a448164bac02a90a80d : SUCCESS in 10m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/29487bdfde204342a5c73f43b87fd9c5 : SUCCESS in 4m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/94c1f0c9cd2c4dfea0ac063d12f35d1f : SUCCESS in 8m 48s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/1d4a0c1c35094215b5e1e5fabf15ad45 : SUCCESS in 5m 43s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9483e13c8e514b06adc37d9be3961a0a : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f1d971a193624152b253fb80a7083a5c : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9be3cb209e554b8f8d8ce7431c72e73c : SUCCESS in 21m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9661a578e6024490baf67b317d8192b4 : SUCCESS in 5m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/09b5140d323e4c14bf0ba6c7385c48b3 : SUCCESS in 10m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4d866b4a6c164c98ad89addc2190ba3c : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/808b81e089e64683af36336401db2898 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/334240ed9b52425588f4c9c42a8c6984 : SUCCESS in 20m 54s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a4df723431cc4259894321877fe3a730 : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/806954b475c2458b861dbf1f7cba5d8c : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcc5ebcb03d44a47a79f3f94cd0557f2 : SUCCESS in 7m 05s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c933d7300bb849ea8d6af70088154640 : SUCCESS in 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51f5f6672a144ccdb7db19fef15b1a4a : SUCCESS in 1h 16m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62120fa55fcf4366a519787833cd27b0 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":9},{"id":"81c5e2ae9e6eb62d5a82091a77fa098135c7b6b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-29 12:12:49.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a94da88704475b0ee993f840bfe2a6c0846a7e6f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-04-29 13:13:10.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e22f2e3ead1d9513aeb2dec7e0fff1e73fcf2f59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:13:50.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8ed1e24bb3e6210485a37e400410f8b2b5b6566c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:38:17.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a85129b08e4720b05e0650a91ac825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd55b3abf24e4fab89269950f209ccec : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2419d354c6114339b664840bc7137d4c : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea68df180ac44c20b30850193260c7c0 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/032446d7abc9470288e07b1867f1fe38 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af96f3ace9714e4db157833105bb98da : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2b64bb3969e462dbc893cb2dae3696e : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c017d68c1d84c30b5b99e383e2365df : SUCCESS in 10m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bc9398473bd04a0fbc0f34716a00cc37 : SUCCESS in 20m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/697ce1395b2740f9b3bac9f0a9f1451f : SUCCESS in 11m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/613c11ed02884effa73673f90fb84770 : SUCCESS in 23m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ba0b2bb06d244a38b3c2d6e012731f7d : SUCCESS in 11m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/605e6a6d264543178408babb2be3d010 : SUCCESS in 4m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/06abcabdbc7b43d1809d6531869b4221 : SUCCESS in 6m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e4557b6a447f4520b23e8f8a903da59b : SUCCESS in 6m 29s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/036ce7d877084b1f96cb510f65f72d80 : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50f8c2369ac8442d801f35e99fbd60c3 : SUCCESS in 7m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f82c754bf1344ae8a00b2a67a7bed3e5 : SUCCESS in 13m 43s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0fd9ccf698754852ab169bc490a9012e : SUCCESS in 11m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c13581c5606f46d5bc26f37f01fe084b : SUCCESS in 3m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b899b67f328744c48e0218b623096580 : SUCCESS in 9m 43s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8ad84e3d0d7b4f0fbf35a6d53b83b8d9 : SUCCESS in 24m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac55c39bdc7f4e46b804adcb239a1435 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1efdcb5fb1284c30aa5c1e94331434fa : SUCCESS in 12m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e461e7e70d994f52ab7f7cea490026eb : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d771294b17694186ad7ab25563449ed3 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37651ecf202a4f00b6d40265b9372687 : SUCCESS in 22m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3103259c84614654b51c491e37b3c82e : SUCCESS in 3m 44s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0e2a3e412401492395f40c074127482c : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a22e530d355c48c9ad99b040b1eed406 : SUCCESS in 6m 17s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7288dce8d7c34c2a9e505447d4e74345 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeda2ceb72e445ce8df44f84b2458e6a : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea969308e06244fba34fb716efe79f3a : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":9},{"id":"bd5efc9f480d059db2e626f5252c6f77c3999818","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"dff31c7db173e454ca41bdc50d9a2ee6c75b948d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:39:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9708b21e2dbe408cac77c9298d750028\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93daffc6344e4727a06fe318f076f0de : SUCCESS in 36s","accounts_in_message":[],"_revision_number":9},{"id":"93997c00f5b25dd7b7e9ce1e6c0b6a6be7a268a0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-29 20:45:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f624135ca4c4ba5ea0a119ab89ef7818ea9a4c18","revisions":{"24cc7426c44266e7e3b92dd1318dc7d45e4ab068":{"kind":"REWORK","_number":1,"created":"2024-04-10 11:17:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/1"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"subject":"Add and pin some more types package","message":"Add and pin some more types package\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24cc7426c44266e7e3b92dd1318dc7d45e4ab068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24cc7426c44266e7e3b92dd1318dc7d45e4ab068"}]},"branch":"refs/heads/master"},"223437e8118341909e2ba3c4f08946054bb70fc2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-10 11:19:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/2"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:19:06.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/223437e8118341909e2ba3c4f08946054bb70fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/223437e8118341909e2ba3c4f08946054bb70fc2"}]},"branch":"refs/heads/master"},"6feb872e47db640e38d814b50f115488fa6f42f2":{"kind":"REWORK","_number":3,"created":"2024-04-10 11:32:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/3"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:32:30.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6feb872e47db640e38d814b50f115488fa6f42f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6feb872e47db640e38d814b50f115488fa6f42f2"}]},"branch":"refs/heads/master"},"3248145d43b4a191ac6a2e0831d6a5031d734235":{"kind":"REWORK","_number":4,"created":"2024-04-10 13:03:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/4"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 13:01:41.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248145d43b4a191ac6a2e0831d6a5031d734235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248145d43b4a191ac6a2e0831d6a5031d734235"}]},"branch":"refs/heads/master"},"2f36f0e73872b39ef84ccdcdc9fda7456f462e50":{"kind":"REWORK","_number":5,"created":"2024-04-10 13:16:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/5"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 13:15:46.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f36f0e73872b39ef84ccdcdc9fda7456f462e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f36f0e73872b39ef84ccdcdc9fda7456f462e50"}]},"branch":"refs/heads/master"},"93f345cd2e6b102aaba198b2c41cf9ca9510e887":{"kind":"REWORK","_number":6,"created":"2024-04-10 15:01:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/6"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 15:01:41.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93f345cd2e6b102aaba198b2c41cf9ca9510e887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93f345cd2e6b102aaba198b2c41cf9ca9510e887"}]},"branch":"refs/heads/master"},"c0d7180b3f6a5fb3570af1b07f2bceccc8fd61f9":{"kind":"REWORK","_number":7,"created":"2024-04-10 16:32:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/7"}}},"commit":{"parents":[{"commit":"6116d8fed4a5131eaa45038d5f13bac876f29f1b","subject":"Merge \"Add jobs without the pre-built wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6116d8fed4a5131eaa45038d5f13bac876f29f1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 16:32:05.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0d7180b3f6a5fb3570af1b07f2bceccc8fd61f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0d7180b3f6a5fb3570af1b07f2bceccc8fd61f9"}]},"branch":"refs/heads/master"},"1f97af70a0c967798920f89f4189af289cc09a32":{"kind":"REWORK","_number":8,"created":"2024-04-15 12:30:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/8"}}},"commit":{"parents":[{"commit":"0e0061bc8bd303892836c0ecfe60e473e405f092","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e0061bc8bd303892836c0ecfe60e473e405f092"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-15 12:30:32.000000000","tz":120},"subject":"Add and pin some more types packages","message":"Add and pin some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The pinning of versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f97af70a0c967798920f89f4189af289cc09a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f97af70a0c967798920f89f4189af289cc09a32"}]},"branch":"refs/heads/master"},"f624135ca4c4ba5ea0a119ab89ef7818ea9a4c18":{"kind":"REWORK","_number":9,"created":"2024-04-24 16:21:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/11/915411/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/915411/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/915411/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/915411/9"}}},"commit":{"parents":[{"commit":"5c1abc456deed2db69dbf4470958932e017961dd","subject":"Merge \"update constraint for os-apply-config to new release 14.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c1abc456deed2db69dbf4470958932e017961dd"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 11:07:56.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-24 16:21:09.000000000","tz":120},"subject":"Add some more types packages","message":"Add some more types packages\n\nThe added packages are used in [1] to allow mypy do proper type checking\nin the Octavia project. The specific versions in u-c is needed in\norder to avoid false positives due to unaligned version of the\nrespective types-* packages.\n\nThis is somewhat related to the discussion on the mailing list about\nmodernization [2] even though I started working on it in Octavia way\nbefore that discussion started.\n\ntypes-requests\u003d\u003d\u003d2.31.0.6 is necessary because we\u0027re still using\nurllib3\u003c2.[3]\n\n[1]: https://review.opendev.org/c/openstack/octavia/+/879749\n[2]: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/4V63CHMZ4GPC4IYN7JCJPVKHLZAHN5BL/\n[3]: https://pypi.org/project/types-requests/#description\n\nChange-Id: Ic8b557a512c29ae76a3ba5e805f7b6f5d118baff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f624135ca4c4ba5ea0a119ab89ef7818ea9a4c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f624135ca4c4ba5ea0a119ab89ef7818ea9a4c18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
