)]}'
{"id":"openstack%2Frequirements~916059","triplet_id":"openstack%2Frequirements~master~Ife587a0e62e18f2bf07434c837424f824705d982","project":"openstack/requirements","branch":"master","topic":"cleanup","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-07 14:57:17.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-05-15 06:09:17.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-15 06:09:17.000000000","reason":"Change was submitted"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2024-05-15 06:09:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife587a0e62e18f2bf07434c837424f824705d982","subject":"Remove kafka-python","status":"MERGED","created":"2024-04-17 11:24:41.000000000","updated":"2024-05-15 06:10:13.000000000","submitted":"2024-05-15 06:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916059-cleanup","meta_rev_id":"a021e18800dfc63506cc30aa652ac41fb181d580","_number":916059,"virtual_id_number":916059,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-15 06:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-05-07 14:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-05-15 05:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-05-15 05:03:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 13:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-18 17:49:06.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-04-29 21:26:56.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-04-30 01:47:32.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-05-15 05:03:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8cce803302565688112e40283e5b0ae6bcf548e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 11:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1936825d79fd82be6226b4c0948068fa4fa57ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:09:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5828c54da6a04f90a0d6db98da4250fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b9f3e062384836a06764b318a4f4b8 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963af2bedf34436a95bde6229985d4d1 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97172c72f51040138456a1c850dc1923 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa4c4aa7b32a49d4b46e3dd569445cca : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d85d368cd8d4bc8890bf2ddb70ee0a1 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6facca01800b4f8da82cc7d09909ee8c : FAILURE in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de3901e544c8459ba73f2750db110a92 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc0d8b60eb734e9491654c9907ed352e : SUCCESS in 11m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8e479e18f60c4b9ca8b7b1fd9e679d3a : SUCCESS in 16m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a85d27ff08234736b8e27b1129301bb9 : SUCCESS in 10m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3b315abb70a45f5b4492a4aadc7b1b1 : SUCCESS in 17m 47s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9232b5f3a6714df09eb1ac969598ca0f : SUCCESS in 5m 43s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a238d1474e904733b12c0a3322cb7a39 : SUCCESS in 6m 40s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8a4fb6e47181412ca66d74bd3be2a505 : SUCCESS in 4m 09s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d64d30963f90431da797ca0797868687 : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d54f2c0e9052433596b7a19150d19fd5 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/031e437deb2d4a93a8833a119295324b : SUCCESS in 5m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9bbc20191aa4483b8c4533d24610d63f : SUCCESS in 8m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fecd76ffaf0741f2ac9486a286b69ff3 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3642c0c2882b4075996deef87d5a94eb : SUCCESS in 7m 53s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8badb50ba84f4fc780f501e893354637 : SUCCESS in 14m 58s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5ef3d62d35284623ac14bc3f23142421 : SUCCESS in 11m 14s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/69dfaa36dbf54b8881d7d3fb1a3c8068 : SUCCESS in 4m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/55eb7a5aa71948459cda14d636189071 : SUCCESS in 8m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/0f827bf8ec954ddab651be658138f70d : SUCCESS in 4m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/90b8fd1c3e3d4273b45483cc571fc4da : SUCCESS in 5m 49s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/186cd90822de44d7ae5a4cdebf065b07 : SUCCESS in 10m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0e2c4bd455e0461cb8adfc66c942e77d : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/61bc48561ea649e1a56b8e52368ba2fb : SUCCESS in 5m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f3b15d678e074b9182e5bfbb95f8774e : SUCCESS in 12m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ede9c7700c294748bae0e5b9ce5a0f84 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6f1dc17b3284c019640b8da8a729a61 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15d961a5dfcd47858da9077f307feb4f : SUCCESS in 21m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e16139d6846d48538ba1c80e6b7127f3 : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/26316d7d6bab4989b9357ac936871cfb : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10951961ce8f487cb5d6caf049340679 : SUCCESS in 7m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/31ff20ab1a3e4b91a2d782129dece9e2 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b2cde7d636a4eeabd954e8a34c8c4a5 : SUCCESS in 1h 04m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bbdc805b9d445a99a6339fe149d9d00 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e23e943aec17b11b1510c6a17f7b21a9dffbb8b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 14:32:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"171b4d2a2d508591be2f805b844849707c121394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 14:36:03.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/916054 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"f960cd761a6715331a4701d5b0387c50ad9561a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 16:01:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"989612ab769d4b94bb791079325a770cdc4013da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 17:46:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e4072427764521a0d61d9bc6b38307\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15406388b974378a06f3f0313394425 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b7152d4bb747749eb7330c8068e5d7 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf8816dbc253456e80146bfca910dd44 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aecbfe7dfff54cb787802727f5666d35 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c1f3dcc5887482b8ef6a6501694adb8 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7bb17ebc6b14f3b9be96f725cd7f277 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/caccbbee33274a7288d8239a6aaf59dc : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fcf25a6ccca4ac6b319bbbc94a4dd20 : SUCCESS in 9m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b1b2ea9999e486b8b48ab33540f5b28 : SUCCESS in 16m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0ec35a9eff5147a3a424144feafbe4c7 : SUCCESS in 11m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/955153882f8345f496d26d0b8c82e3e6 : SUCCESS in 22m 49s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ff4e86909da74496b84c5631ba1fffc9 : SUCCESS in 4m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ce49ca58f28d4173ae67b043b749a81f : SUCCESS in 4m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0a21862d9e3a4249810c50f7b2ced082 : SUCCESS in 5m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a831e052ba84449f90a568d16d92a8ed : SUCCESS in 11m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c511dbd1aca94181b2f597c048863fa2 : SUCCESS in 4m 54s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/964a7c656f574a5d9c0976b0fd0f5b3d : SUCCESS in 5m 35s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/29df1744ddcc4f8d977efbad9e53372c : SUCCESS in 6m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9b7fe3aedf104f85b543595595986c85 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11d08bff4202408cbf2f169aca6123b2 : SUCCESS in 8m 45s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2893c9179c37408a8100e3c6ee7bbbdd : SUCCESS in 12m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/06a2d797c594491da4c301c60797d5b4 : SUCCESS in 11m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8aee10e3a6c34390b53446d8176a610a : SUCCESS in 4m 28s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/415f02ef6a654eb3b356711dccaf001d : SUCCESS in 7m 45s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3095eb92f6384a6eabda7c99c4785ce2 : SUCCESS in 4m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1f8e1b43493241049609a5fdd7506805 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9da1d72004fb42bbb6a748e115870f19 : SUCCESS in 13m 27s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6bfa6fb61e7742d9a1f7ad8f72577ccd : SUCCESS in 18m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0489b371a03449a990f34677b32ce59e : SUCCESS in 6m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a1d032bfbb66492096500ec238699ff7 : SUCCESS in 11m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d4e23d3d4a084284a4a2b6b4da120af1 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5666476f6a7f40208ffed64b3f4a0e4f : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/722870f81c394d2a8bcad1f2b3c47d68 : SUCCESS in 21m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1f8005ea3e5f4a4dbfe1c8aad4cfdefa : SUCCESS in 3m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/45df84ca7b034c50b253af19a9e2ca2e : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0367c34d3cdc4c9ba4a6c1602d78beca : SUCCESS in 5m 19s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9341928543954c579d07072d809c9244 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c754f8e77ae4d9cb9020227c2d15659 : SUCCESS in 50m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69b7bd2ff54246699c5c4091af8a53c1 : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":2},{"id":"129597226ac6f40a83858c4c0fdea18c6810f047","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 17:49:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f8bfa9e1526d64b4c2ed1a0a029dd8a6bf9d52ff","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-29 21:26:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"101e5e5df1d76837dad8697b4bedad952137608a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-04-30 01:47:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"88f67d8049f6ada698f0b9e3143b0c72d2860766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 03:46:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3350673204d1722ba406d66a92dc1b2958642385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 05:00:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ff178d3ffc4e0c8f7ad75d2365670f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8356b2e5df9844918239d407cb72511d : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84603560b2f641618edb61034e7992e8 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ede58b1e75447e9accc89c07f91aa64 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cf3d8b0096541158b17b08973cbc074 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64d059f8d10b472b8978f58f7b63a324 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a842b401d7a4d4f80337885422086ec : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24970dc4e5b44bd0bc266a8bc18eda1b : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8d6429814c2491c92253030335374ba : SUCCESS in 9m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d5aa5e61fb704dd5b5eb18f9842b6813 : SUCCESS in 20m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9721152c109415fb5f2f0348ee0dc8f : SUCCESS in 8m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c9e56630a2514a609f113a2600ab02c0 : SUCCESS in 23m 34s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8f811b0401744509ae2a2eb37558b202 : SUCCESS in 6m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/76ea28c3f5df457b9495d6b79d458afc : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a8cbb742de2d4379858b79fdc5c2598c : SUCCESS in 4m 13s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f92cdc65f85a4b96aec3321742aa7c81 : SUCCESS in 10m 41s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/af12d3f66ef644e8857ab044d1f00696 : SUCCESS in 4m 35s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/385eb5f9b160402190802901361f9ec5 : SUCCESS in 4m 35s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3a370ed2dc0f472bba0fc67ebc8f65be : SUCCESS in 9m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9e7204d1011741b4a6c9aa646e3a3b1a : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae525f1b50bb4e7588c30964e28a1797 : SUCCESS in 6m 38s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b2fe6b7f1f9345338043781f81fd34d6 : SUCCESS in 10m 00s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/493b392800f844b592bec7572f811cd3 : SUCCESS in 10m 43s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/99c080f0a92c431fbe3bea50022fc4ac : SUCCESS in 3m 58s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3f8999e1ee88484489e13a38db1bb88c : SUCCESS in 8m 44s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/52d5630eb23a417d9692f565949e2161 : SUCCESS in 4m 37s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3747407c74b4478f8c095299be91fcd9 : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/420cd576960b41e4a2089f1206ecbbcd : SUCCESS in 9m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f2ff7b228f6b472abdafc8ca183e2256 : SUCCESS in 17m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b17c79c539940f6a9201ddb7cbd732c : SUCCESS in 8m 17s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/34abc045f8594b24b416a907d5d7a5d0 : SUCCESS in 8m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5e8077e2fa38401799d47c5f29143373 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aaefcdfe44d746d7a656581c34f5536d : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a99d301c1dc34a748a3038ca60e7f5c3 : SUCCESS in 24m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6239821a225e439c8fd01e49a910b473 : SUCCESS in 5m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6e93230aa99447a38fa8dec8fb8f29ce : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd78697fe9384713b8e395a747dd98b3 : SUCCESS in 5m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/43d3de646af0415383251974281687d4 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eac5231592f4ab9b85767609a6b6fb9 : SUCCESS in 55m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08741178620d42c38319ed351d947510 : SUCCESS in 1h 13m 11s","accounts_in_message":[],"_revision_number":3},{"id":"70cd07edd10e989f259e3b832137d0a68ca1ceaf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-07 14:57:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"403229675530a76a9b393a2457935b60c1872631","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 05:03:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9ca22781995b799cefcc7bf47b9bb3d80d1318df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 05:03:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"392a8d67cb0b52cea91b12a99b78e1e5b49cd708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 06:09:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2cacf20ffd4e3093ebf791a94a9a4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bbffbd2df840109c065356151dee0d : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e070e1abce74043a3d04c2b0b2f6630 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/877425d69219448f9e09ad787bc2c158 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6f0334ce81646b9b73df99ddad43f8f : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea518cb79e1f4c7d91860eddd37b02f2 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9be2db7acf1d4e91a028e18db9848c56 : SUCCESS in 2m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aced32f322a7428b90996b1a39183e87 : SUCCESS in 7m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5422b0576abe4a75b5193652a3fdaa63 : SUCCESS in 21m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45549c3930da4c1e8165185baa4a7445 : SUCCESS in 6m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8da0d88f4cd543d787a0e14ca76b29a0 : SUCCESS in 14m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e0ac76c1186c433d831759f984995be8 : SUCCESS in 8m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8d7ffab5082b4e9a8ca1978ec0f69747 : SUCCESS in 4m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d93339d4c7034c41a850b129182c308c : SUCCESS in 5m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/994c54664ad5413f9cd21ee09154f208 : SUCCESS in 7m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8d526c0f47ca4b508657e444532185fe : SUCCESS in 13m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29b5e68ff1d249b9b01aef96d0666474 : SUCCESS in 7m 10s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ba09041cda7a4fd58627fe4fdf5b42bc : SUCCESS in 12m 16s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/00689b670bcc4ab0a865c66b07206045 : SUCCESS in 10m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c517a9ba277d4d3c91584797cd213134 : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f12e5ef06d814680ae9c93cf90a4442f : SUCCESS in 8m 06s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/22e6c7cd104b4abf99c20c60c561a8a0 : SUCCESS in 26m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f3901f5167349fda7ac75a443bdfb10 : SUCCESS in 6m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/80b0340580034eb3968fe86beab94a8d : SUCCESS in 12m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3d32f51824ed41908068aa4648c8f035 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b948d2ba4ab04ad4a1e96b5722d419a7 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bb8af78a693483db9107ce9fcaf428b : SUCCESS in 21m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/94f74f59cc3948c0bb2f5c35bc5222e7 : SUCCESS in 4m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7aa6d7a726a34c48a1f1a3e1e1c6176e : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b10be9bbcf9245898d2ed658429d7152 : SUCCESS in 5m 55s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/16374ff4a9a4456a9429a51b5359f553 : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dc6ecb8cb6f4591b33da007b4002ab4 : SUCCESS in 48m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc13edca5e564b4685fb61a72c2d085e : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d6150c45acf9ad749d3eee1f67866a170a6a0f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 06:09:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a021e18800dfc63506cc30aa652ac41fb181d580","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 06:10:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493ab518b1484867a95ee6912b426b3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a6c24564294821bb0a3a9b4e0a7be7 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8f2ae54cbe7f745fb42ccd90c3164beeb5953389","revisions":{"856004bc45ac8aee87ed4df920be66e5c8f3dd98":{"kind":"REWORK","_number":1,"created":"2024-04-17 11:24:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/916059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/916059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/916059/1"}}},"commit":{"parents":[{"commit":"c9560a1ea285cbe18e351e1d7c13dceef40a9ade","subject":"Add ability to strip old excludes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9560a1ea285cbe18e351e1d7c13dceef40a9ade"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 11:13:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 11:14:53.000000000","tz":60},"subject":"Remove kafka-python","message":"Remove kafka-python\n\nThis is no longer maintained. It has been replaced by confluent-kafka in\noslo.messaging.\n\nChange-Id: Ife587a0e62e18f2bf07434c837424f824705d982\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack-plugin-kafka/+/916054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856004bc45ac8aee87ed4df920be66e5c8f3dd98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856004bc45ac8aee87ed4df920be66e5c8f3dd98"}]},"branch":"refs/heads/master"},"bad850c4d2414ae5a4c87dc5e48e07016f72f3f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-17 14:32:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/916059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/916059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/916059/2"}}},"commit":{"parents":[{"commit":"314734e938f107cbd5ebcc7af4d9167c11347406","subject":"Add ability to strip old excludes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/314734e938f107cbd5ebcc7af4d9167c11347406"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 11:13:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 14:32:35.000000000","tz":60},"subject":"Remove kafka-python","message":"Remove kafka-python\n\nThis is no longer maintained. It has been replaced by confluent-kafka in\noslo.messaging.\n\nChange-Id: Ife587a0e62e18f2bf07434c837424f824705d982\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack-plugin-kafka/+/916054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bad850c4d2414ae5a4c87dc5e48e07016f72f3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bad850c4d2414ae5a4c87dc5e48e07016f72f3f4"}]},"branch":"refs/heads/master"},"8f2ae54cbe7f745fb42ccd90c3164beeb5953389":{"kind":"REWORK","_number":3,"created":"2024-05-07 03:46:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/916059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/916059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/916059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/916059/3"}}},"commit":{"parents":[{"commit":"1e07b8cf6e8a84d18987d1ac4ba7b0fba1ebdca7","subject":"Merge \"Remove pyinotify\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e07b8cf6e8a84d18987d1ac4ba7b0fba1ebdca7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 11:13:52.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 03:46:23.000000000","tz":540},"subject":"Remove kafka-python","message":"Remove kafka-python\n\nThis is no longer maintained. It has been replaced by confluent-kafka in\noslo.messaging.\n\nChange-Id: Ife587a0e62e18f2bf07434c837424f824705d982\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack-plugin-kafka/+/916054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f2ae54cbe7f745fb42ccd90c3164beeb5953389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f2ae54cbe7f745fb42ccd90c3164beeb5953389"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
