)]}'
{"id":"openstack%2Frequirements~917786","triplet_id":"openstack%2Frequirements~master~I90ac679c1745a31474c6f24ec7953624ad056d79","project":"openstack/requirements","branch":"master","topic":"inclusive-language","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-05-14 13:21:38.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-14 13:21:38.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-05-14 13:21:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90ac679c1745a31474c6f24ec7953624ad056d79","subject":"Improve terminology in the requirements tree","status":"MERGED","created":"2024-04-30 19:18:00.000000000","updated":"2024-05-14 13:23:51.000000000","submitted":"2024-05-14 13:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917786-inclusive-language","meta_rev_id":"a53203ca66e4168278e3f09917c7aface1805861","_number":917786,"virtual_id_number":917786,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 13:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-05-14 08:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-05-14 10:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 19:50:45.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-04-30 20:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-01 16:25:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-14 08:17:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8a4ab764d1c7f9d9f4ac12ac1dd82c95a3f821bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-30 19:18:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd0860b5fc427840ef6be4f1d0a44c2dab839f7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-30 19:50:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0684c68bd9077b7f2e7b3a92a87cd4b9f3c3f9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 20:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a264e69edcfc4ba8b250ecc07f581bfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13888320a4c442295ec978c77e8333e : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587476f1b53f449fb6d8a0f2bd9f78ef : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edeb457854994614a26d6ea68b2ba76e : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e6dd0fbc3cb45319de05c288146360a : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bea16b170de54faeb550ae342f4cc61a : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4896a7763a74726958f5cb9a5245a0e : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e4ca1b78de54c9aae58d405669d77fe : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a669e632e3c047368d0c02d51cd4a641 : SUCCESS in 1h 05m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c9c6e2b43b1483aa8d3345bf053dc2b : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bd07390f1f5963ac5c707a03fa1c98d907704860","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 01:40:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3f792536bedc4298e67282ea60ad67dc16a4757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 01:40:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"39138ab97a404fe0a56fdde679b02824ae36e70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 03:13:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66847172aee42648bef45ca7d6168e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e464a2b9c741518375b3e2a0f27c85 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b48387a7e5493c9d46626487bf3c7e : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2426071941ae414e896e56a8f729ad47 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf0ea98ed6dc455aaedb1c391035a5b1 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33fa3cb7a6be46d3b976365260c4fc88 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b2a594abd554a77b42bcdce16cf84c2 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba3ac08527744a2e92ca065c66537a8a : SUCCESS in 2m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/641a167811554cfa94b5d872373ca8d1 : SUCCESS in 1h 09m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/449178df12104f45923055f23aea1445 : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":2},{"id":"db400dd436860393309316b6df47ee149f03181e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-01 14:22:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d88fd6169dac998a46667b1ad014034ac16ecda","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-01 14:29:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6819a279c5e9fe625955bec49c740f940396008","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-01 16:18:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77567dcf134b4077637570a186c81dcceca90f05","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-01 16:25:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ccfa066a98a14a9f8707295a50ccd4f69f5f64f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-14 08:08:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"109cadc33745ea4507e1d0cc63909c4d75d49433","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-14 08:08:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22a226fb62a576ae32b82a5d7a12ab8bd48d1a5f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-14 08:17:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dbe080cf8ae5b78a7b4a9d75d0391cb884ebf41c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:09:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe279dae73147529586e69eb09c5cf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a8f1ef753e40259bfc418d65730657 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5399ae2fd743ea883efa2cb5ef2bf1 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/802310554c944d9ea5b6ec54fa549aec : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/698d19af13484aabb017017de959a48d : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a18e89144ad34a9f812873eaaa0da550 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f94dc043f5924057b118fb2eab83c22f : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f5ddbbdee0f4635930af96a9491af55 : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fd0078c1f674ac78135ba0a6eaf6752 : SUCCESS in 48m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cc9d31087df40a0b80e6c3700d5d814 : SUCCESS in 1h 26m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ccf52e21b6876be1df471fe64e5bb3a08e47ff5d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-14 10:41:57.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3da4ff4784737043136726bf215a482b855a82e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:42:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21a5199b181a35b9125aa0c45542aaeeadea6e29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:21:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9617c83e4924ef39d59bafefbb17ab9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd03bac49e36474a9470f2a052dadff5 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7f30056eed480793e88145fd17a936 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c766dd0dd0d34f6e93d23684cce71ed3 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f2f0ea8118743f6ba2fa1604be5ab53 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4df38ecb504345f5957386e1aa61043a : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d80703ec1fd4e1588c3846e7f25e2f4 : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71a8f97f4d0a43a9b74cd1a80b80f00a : SUCCESS in 1h 00m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1138aaf8940c426f85446471d8965edf : SUCCESS in 1h 37m 33s","accounts_in_message":[],"_revision_number":3},{"id":"0e977eb77493de56890b5c9e8475d87144c26aab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:21:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a53203ca66e4168278e3f09917c7aface1805861","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:23:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7871734d4904519ba4da564945bfbb2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0723551d3ed45a6a6cc7e93eb729f48 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea345875cd43ef95c26e46199956c82cd8132ad2","revisions":{"afc0b1aff0484336c3c3222940dafa08073bc6f8":{"kind":"REWORK","_number":1,"created":"2024-04-30 19:18:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/917786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/917786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/917786/1"}}},"commit":{"parents":[{"commit":"fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89","subject":"Merge \"Add ability to strip old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-30 16:34:26.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-30 16:34:26.000000000","tz":-240},"subject":"Improve terminology in the requirements tree","message":"Improve terminology in the requirements tree\n\nThere is no real reason we should be using some of the\nterms we do, they\u0027re outdated, and we\u0027re behind other\nopen-source projects in this respect. Let\u0027s switch to\nusing more inclusive terms in all possible places.\n\nChange-Id: I90ac679c1745a31474c6f24ec7953624ad056d79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afc0b1aff0484336c3c3222940dafa08073bc6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afc0b1aff0484336c3c3222940dafa08073bc6f8"}]},"branch":"refs/heads/master"},"0c7c5eca24f612bfb453dc9dd3c81eb7337a1f32":{"kind":"REWORK","_number":2,"created":"2024-05-01 01:40:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/917786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/917786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/917786/2"}}},"commit":{"parents":[{"commit":"fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89","subject":"Merge \"Add ability to strip old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-30 16:34:26.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-05-01 01:39:47.000000000","tz":-240},"subject":"Improve terminology in the requirements tree","message":"Improve terminology in the requirements tree\n\nThere is no real reason we should be using some of the\nterms we do, they\u0027re outdated, and we\u0027re behind other\nopen-source projects in this respect. Let\u0027s switch to\nusing more inclusive terms in all possible places.\n\nChange-Id: I90ac679c1745a31474c6f24ec7953624ad056d79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c7c5eca24f612bfb453dc9dd3c81eb7337a1f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c7c5eca24f612bfb453dc9dd3c81eb7337a1f32"}]},"branch":"refs/heads/master"},"ea345875cd43ef95c26e46199956c82cd8132ad2":{"kind":"REWORK","_number":3,"created":"2024-05-14 08:08:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/917786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/917786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/917786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/917786/3"}}},"commit":{"parents":[{"commit":"091035a0319fbca632e5ae0fcc4f249fa6f5792f","subject":"Retire TripleO: remove the TripleO deliverable from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/091035a0319fbca632e5ae0fcc4f249fa6f5792f"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-04-30 16:34:26.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-05-14 08:08:34.000000000","tz":120},"subject":"Improve terminology in the requirements tree","message":"Improve terminology in the requirements tree\n\nThere is no real reason we should be using some of the\nterms we do, they\u0027re outdated, and we\u0027re behind other\nopen-source projects in this respect. Let\u0027s switch to\nusing more inclusive terms in all possible places.\n\nChange-Id: I90ac679c1745a31474c6f24ec7953624ad056d79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea345875cd43ef95c26e46199956c82cd8132ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea345875cd43ef95c26e46199956c82cd8132ad2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
