)]}'
{"id":"openstack%2Frequirements~917831","triplet_id":"openstack%2Frequirements~master~Ic631bf666290d08a7214b07c7c863ee0d070ef9f","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-02 15:59:38.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-03 01:00:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic631bf666290d08a7214b07c7c863ee0d070ef9f","subject":"Remove pyinotify","status":"MERGED","created":"2024-05-01 07:07:36.000000000","updated":"2024-05-03 03:46:55.000000000","submitted":"2024-05-03 03:45:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917831","meta_rev_id":"1aac221d8f8f9787ec4eef524492bd85bc0ab186","_number":917831,"virtual_id_number":917831,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 03:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":2,"date":"2024-05-02 16:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2024-05-02 15:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2024-05-02 23:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 08:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-01 14:45:18.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-05-02 16:28:27.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"728d6b6a29ab93381bdf35085a98093976385f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 07:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc64b7bee0de19e14c89af7051a936a8121a75df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 08:25:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05e096df22e4bc3a1219418c3b31719\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fcc9de3ef0472f860a7660d4824d44 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4bb775a61a414c87229a7994149c16 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9207ba8deccd496c82587b8e647e81c2 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73746baa0d7c430fa7817d156f0aeb1d : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9fd3a6e628d4129b510d8cd77b71552 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5199cb0dc3d24a67bd16d6dae914a63b : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/728ea44aff9645aaa2f63f5cc60f968a : SUCCESS in 4m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8164f9a87b5a41b8a097bc1d3ce42fc5 : SUCCESS in 7m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/48713f2d50f14896b02e3f8ae988ed75 : SUCCESS in 19m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a18870823424407c937d9d0e6aca088f : SUCCESS in 11m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a9ac2f2773b645ef957adcb791ed7b06 : SUCCESS in 21m 52s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8c1fcd45c42d42f3b00b4986339ee7ab : SUCCESS in 4m 00s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/14cf0c6a8399434eaeb7ed762296896f : SUCCESS in 5m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7bcb00836ec3495b90b90f59da8fae3d : SUCCESS in 4m 11s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c7ab3e4f770543e7b7861b9333dfe127 : SUCCESS in 10m 28s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1bbf24e6c2ce49c6ad501a0052266e6c : SUCCESS in 5m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d34b703d412f4e148a20a866e2230100 : SUCCESS in 4m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c30070f7a992406bbf373f087036c866 : SUCCESS in 7m 50s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9103b95ab9f148a4beb0b06a1451d97a : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35713fa840404a3f8ce1a81dde08aa53 : SUCCESS in 6m 18s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/aee8883a036b4e33ad79e6ff541b9d33 : SUCCESS in 13m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b1e66426688745bba1b605c3b63bc831 : SUCCESS in 11m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/31e868fea5404044a84a627122946760 : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d7a7d1f5f8804d7b825473edb23fd6da : SUCCESS in 7m 53s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a856d9dd50fc4367aca63df2c931634c : SUCCESS in 4m 11s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2b1fc47514d847b7b068fbca68578fef : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/dbf9d109b3e64a868b7caec572626524 : SUCCESS in 12m 05s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1b8c13cd848a4962a7fa289668ed6003 : SUCCESS in 22m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d6cd97104f942038f2bec777ac25337 : SUCCESS in 6m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/622f0b50ca5f44fe93e2d1be05777919 : SUCCESS in 11m 00s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fb297f5d51704192bc03c68765d775d3 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f515b4d70d0c4ebdbddd47dce346bedc : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cdcc9ac19feb4d2c84c213dbdb654faa : SUCCESS in 28m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6085cedc22994ff7a35b4f4b37ec02eb : SUCCESS in 6m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/67611a8d187d4cd58c8ca8bcbf233e8f : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5009bd73005455bb28533c873f107d4 : SUCCESS in 7m 47s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ca3d949398464b19a9c75bb369c8131f : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5f9330f1e044b71a147de198081910e : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af8188c1a17f4c008ffb5c4a8cc323c6 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"213b4af434b590bedf5e4a85dbf50242d93b03ee","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-01 14:45:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a18a0d244de957a34088dfe4d74ebdc87437d348","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-01 17:54:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4a8f01612af08d31ce7beeaa52d6f5f4bc6daa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 01:12:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8cd698e2d0721b5c793c37722b7378d86e6ef8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 03:06:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7608c175fb984705b74bfc0f5b5b6423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb642b93c2f241f6a6465bd581584f47 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb94acd1e854205bca96f563fc9ab72 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0340ca2254444418cd60b29299fb046 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d8399da915f45c8b725e940ba0a0891 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd507dedc5c94fc59beb511a5816d2d6 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ac555a14f204129a0a36de04397aed6 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecc404c496914ff481e230667d5559d2 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e03d855d67324fdb8251b6efa5caed68 : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3754a533d8a54fe1a93028078b4ec5ac : SUCCESS in 18m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7d6f0611227f4b79bf32000d44931550 : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e8fd6a5fb66d4a928f791373d8131dc3 : SUCCESS in 21m 58s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/12d15d01c14d4798beafe48517b72b5d : SUCCESS in 4m 12s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/eca5c2eb1d7b4a14a7adf79cd9cf4b30 : SUCCESS in 4m 32s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/644f149f20ed47739b515fbebff7b4e3 : SUCCESS in 4m 40s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7b857507a7d44ed9bf027818ad217b80 : SUCCESS in 11m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/232071f3b61e46898557e310669c2fee : SUCCESS in 5m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cd4c536d1d3b4c44a5a5127668895f05 : SUCCESS in 6m 54s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/77f5879ebec1460f981e9eb320a65c3a : SUCCESS in 8m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/23596e37ab254772a17c61f583f9b5c8 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56426c031d344ec687d9e6a350f6c9c0 : SUCCESS in 6m 40s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d7a70f48eedf4b5b8642f9ad8051dfcf : SUCCESS in 12m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/72955b6854c04fefb92554393342758d : SUCCESS in 9m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1ffe1f7539784a4f8d99ca73599fc893 : SUCCESS in 5m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8706209b17664c279905e1ae3048bdd3 : SUCCESS in 7m 36s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e73bb98952494f9cb8efdae34b357880 : SUCCESS in 5m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/19a91e72e663425da365f3989c454d28 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2ab290aaa9fa4775aad62c103def976f : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8f770ada480d486d8df2727dd4584457 : SUCCESS in 22m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7c3fbdd021f45c2842b56eba1fde802 : SUCCESS in 8m 43s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/860509f478964a5c9d498b5ee7e642d1 : SUCCESS in 12m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e5adb8117e0740f69239b6d8f922445c : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/472a86915dee41e4a907475c3e49ef5f : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01632b4b604a427c8c0f986c04fc3653 : SUCCESS in 23m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/23466280fc434c05aed9924d9bd1a05e : SUCCESS in 4m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/cec915bc81fc4be98e6fb9c29eb4e01a : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/674f0876424f40feb946bee294df9d8c : SUCCESS in 5m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0085ce866675414ca7dcd200e578b205 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d9d7f363f34462fb39b510412db93d8 : SUCCESS in 1h 04m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e096baae9003433fa040750bb8831cbd : SUCCESS in 1h 42m 45s","accounts_in_message":[],"_revision_number":2},{"id":"88ff6b6cd121708834c3cf147b839cd0cba09ca2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-02 15:59:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"71a2344202a5081cc0616947eab459d42c8eb87a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2024-05-02 16:28:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b815b36120e81ad77c32ae576e0a7f2f12527f92","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-02 23:29:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d624a3b8dacdf1804b889e44b87d49616603d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 23:29:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19f15cf3f19d7060e828e50194d8a7ec33e17be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 00:52:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/855bf701f095439694fc46a40947fa2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e928dc57d444592bce7e896ef065f00 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a214b29dd824f70a886d73579e4452a : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fd02da1cf5341d3975c7e3e646c7f13 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/199f14d20d254bf2a18854c05d3bc3a7 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed42fe9202324a079947ed436c4a43b6 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed0e3ac534d649e79e7bb734f76ede5c : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83a87ff929a04e46a0c76109fdb23fac : SUCCESS in 9m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1f51670cb7d440b7b948c63f1b1db7f6 : SUCCESS in 18m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81bc1d0885724315ac092903aacddbf1 : SUCCESS in 7m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/46ce77a9be904bdd8aca87068e7919c6 : FAILURE in 3m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/cfb88bb6fab74cf980f79148331fd632 : SUCCESS in 7m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e3692a03ec0f438993a3b5cb1ca94003 : SUCCESS in 4m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5ef0ce3e88104e53a2d73ea67d8c04cb : SUCCESS in 5m 36s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/726c900688b440ae8095622c50601d89 : SUCCESS in 8m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/349fa34de74b40a89b9a2c97f6708422 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f1a8a6224074f74a148ed670c413566 : SUCCESS in 6m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3e10d80adca94b5da132d7193fac0afd : SUCCESS in 12m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/eecd84b17dd24f1ebedfa62f7ffa07d8 : SUCCESS in 10m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f123bf95acc5439ea515a92e136217a4 : SUCCESS in 7m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/085d341b88ae47f79f37f8c6f9aaac59 : SUCCESS in 9m 18s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6e3ac067fa614fae8818998030564dfb : SUCCESS in 20m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e2619efb9d85466fbb66cb50e5920ec7 : SUCCESS in 7m 39s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bb7dc0c58ea14e6596540b012fb08add : SUCCESS in 11m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d7418f880891477594afe8b13f0eca6d : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a24b22a5f6904e6b97d3319ac1d3e969 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68f48b107d74472f8c27360908f38f8b : SUCCESS in 17m 20s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4e2141ce795449bab59f20641ac98879 : SUCCESS in 4m 25s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/43507f6595e143719075db8cce7a0b9f : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46a8363b543a450393fddfe923db6b55 : SUCCESS in 5m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2ad31f165b204799a83bebe4dedb2de5 : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d11bb37f3bec4caa85c5b8dbf692b8eb : SUCCESS in 1h 05m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/601306eada704fb5ba40b201f620c024 : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6d610e01705382851364df4cef17a0c24fb25d62","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-03 01:00:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b91f5e5fcee0e118aa74ebb9cdeaac1eab58d14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 02:32:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e5631bef8ba4318a200063de26a3a96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a72246b367048009dfad8c176464309 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae8bcb3a6efc4047bd08496772752217 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cd7f33a6b6d4bc8ad78df7a3e262116 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a3acc1d4eba4b5e8a4a219c28e19cad : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a942f27df509472f967f7bbe6e217bed : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/742265467f84472c9990c89179860e28 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8e64d6de4e14e209abbf6861489424d : SUCCESS in 4m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ac1bdd66e54496f9811207f4e07e110 : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/141dcf35b3d94554838253907cabaab0 : SUCCESS in 14m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7cf8682755d34a91b4f64986db5d6ca4 : SUCCESS in 12m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc51e502cb3d44c29ac379a981df9031 : SUCCESS in 22m 00s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2aa2951e8afb48d9a264e7f49845d657 : SUCCESS in 4m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5e127e8d700446358418f9bd5c2e5fa7 : SUCCESS in 4m 39s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2c7f8844194d478697393c13140c05e8 : SUCCESS in 4m 18s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/faa542efe94b42ed97f2ff856672bdff : SUCCESS in 10m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6e2b6b5d16a04d07be5b0e8c4c28192e : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3ee5497364194c1cbdbf2790cc78430d : SUCCESS in 5m 06s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/64406c2e81f74cdcbdd75a8b27f982ef : SUCCESS in 7m 36s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a722443ab9a145f5a5e7528ae8cb7bac : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/569cea03e51c4772b29104dfbd0c692f : SUCCESS in 7m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/51a41435db3b482bae84b5fb3e7d2464 : SUCCESS in 12m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ad7193bed73b4230a83a513db2b280b3 : SUCCESS in 7m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0682fc5b20174995b8e322eb8b81a47c : SUCCESS in 3m 33s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/927c54b6fe1b433ba45ecb63a852bc72 : SUCCESS in 7m 11s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/cc587c895945438890ae8fbd564b86ae : SUCCESS in 4m 37s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e9509724a3144ab1a06fa097c62c3cef : SUCCESS in 3m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9e0a043971fe4a03a4f99f156939058c : SUCCESS in 14m 59s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e4cae03e78df48a0a70943b0d34c00c8 : SUCCESS in 28m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/73de4c6b483849b899dc0f97283202e5 : SUCCESS in 6m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7107bb73b62e4856a00e32b061f24bd1 : SUCCESS in 11m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6589b3e174924f76813c55b73a4e2e04 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2166b31175af4c2cac1b6d295a1093cf : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c272c726a4684015ae4654c5c5b58206 : SUCCESS in 18m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/286c522c87f04f16829d0fd10a4d4375 : SUCCESS in 4m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9d5dd6649adb4a80a36fcdb72c9e0193 : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4c54fc111e24e4c89b4f3eecc337b5e : SUCCESS in 5m 07s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2706c40b26d14472baeee0eb02dea577 : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ad6c34bc0bc44daa86a6fa0ee3678a4 : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/013e76bf04cf4d1fa8a8c386daa9c5c1 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"75325ec4fed418c1ef807ffdaba52990ffba5250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 02:32:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10ef7f508fd681d72356f36f987c9d8e3cc4d62f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 03:45:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1993b92070284e27a0fd3d701be5aae3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdfd39c318c14515a108e996a30fe863 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce53be2f90994c8f9a8eae7cf8e71607 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35e9e11482a42e9a3fe2272bfc5e733 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59c4fb1299494cf5962e5affb5992b67 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a044677d137e42618d7454914267cbf3 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49b013f80ced4572bafdfa5903630a11 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d4d173fff7a4b3f86b5e2e53e317230 : SUCCESS in 8m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d5de274b270c43b8842bcf4614ac4639 : SUCCESS in 24m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a3045c1f45054531965bd33ae9ffc7bb : SUCCESS in 11m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c954468ebc4147c18fe0f4c1544067e3 : SUCCESS in 22m 38s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/abd0500a313542a3a7bf356e027af438 : SUCCESS in 10m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8be8464e229043e7b781e76f099e744f : SUCCESS in 5m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bb78b68f795e48ea8ef58ae08981bd9c : SUCCESS in 5m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/854622e9de3449988e37c424d1914d8c : SUCCESS in 8m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c509a0bbc89d408cb3054540a0441abe : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9712d00b6f0e4c7d8e04796d468e03c9 : SUCCESS in 8m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a67610af7e784289bb04e0ae758dad5d : SUCCESS in 11m 36s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/077103b6a4dd4c67b0b42d5f25e2908b : SUCCESS in 11m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/928642cf844d4c3e9756d5351c3439d4 : SUCCESS in 9m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c88a85455d12477d9869568bf2098382 : SUCCESS in 9m 37s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e3345b528b4943b3b2f1e0c07d35c425 : SUCCESS in 22m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79222d65d7c14334b8ce3a77bfa197e8 : SUCCESS in 7m 02s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/13b703a89dff4af19f73d0072d1ed150 : SUCCESS in 13m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/bbe37e94c0d3439b981990f292235e99 : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a02a4c5c956644e7b9a5c0765d5aa63e : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8fcc9a59ce74b519dfe8393496e984a : SUCCESS in 21m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dc2b8e700d21487aad5ab1d13228e020 : SUCCESS in 4m 33s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/48ab40026d7848daa6f898e90b518450 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad5bf67fa48c427f8cb0abb44a036808 : SUCCESS in 7m 22s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/dba4770aeb82466ba52d9162f04a77f9 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f24a955e9d7466ead5034cf650039b0 : SUCCESS in 1h 03m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01888a99874b42119809c2812a6d3469 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a69b444f08471c1020de60127fa49dd0f36a764f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 03:45:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1aac221d8f8f9787ec4eef524492bd85bc0ab186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 03:46:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecb7bf5985a4de08b3e7f635181daad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b294153464e341be9ea2394d0e6c1abf : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1cef7381ce65c88a14b0ffe4c2795c64cbe0857","revisions":{"c5a08f5ca0f6cd9824d87e38ca1e78d5633ce735":{"kind":"REWORK","_number":1,"created":"2024-05-01 07:07:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/917831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/917831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/917831/1"}}},"commit":{"parents":[{"commit":"fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89","subject":"Merge \"Add ability to strip old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fccd0b4f8c3ac37dcbd7cf9da2f916fb8ab8bb89"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 07:06:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 07:06:31.000000000","tz":540},"subject":"Remove pyinotify","message":"Remove pyinotify\n\nThe library has been unmaintained for long and is not compatible with\nPython 3.12. We decided to remove the implementation dependent on it\nso it\u0027s no longer used now.\n\nDepends-on: https://review.opendev.org/c/openstack/oslo.log/+/917808\nChange-Id: Ic631bf666290d08a7214b07c7c863ee0d070ef9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5a08f5ca0f6cd9824d87e38ca1e78d5633ce735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5a08f5ca0f6cd9824d87e38ca1e78d5633ce735"}]},"branch":"refs/heads/master"},"c1cef7381ce65c88a14b0ffe4c2795c64cbe0857":{"kind":"REWORK","_number":2,"created":"2024-05-02 01:12:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/917831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/917831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/917831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/917831/2"}}},"commit":{"parents":[{"commit":"cf1e5cae4567808949498b22587a74cab6c526e5","subject":"Revert \"Add ability to strip old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf1e5cae4567808949498b22587a74cab6c526e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 07:06:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 01:12:41.000000000","tz":540},"subject":"Remove pyinotify","message":"Remove pyinotify\n\nThe library has been unmaintained for long and is not compatible with\nPython 3.12. We decided to remove the implementation dependent on it\nso it\u0027s no longer used now.\n\nDepends-on: https://review.opendev.org/c/openstack/oslo.log/+/917808\nChange-Id: Ic631bf666290d08a7214b07c7c863ee0d070ef9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1cef7381ce65c88a14b0ffe4c2795c64cbe0857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1cef7381ce65c88a14b0ffe4c2795c64cbe0857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
