)]}'
{"id":"openstack%2Frequirements~917870","triplet_id":"openstack%2Frequirements~stable%2Fzed~I61af5056c3bf25c52712ba559c8e99cffcc123ca","project":"openstack/requirements","branch":"stable/zed","hashtags":[],"change_id":"I61af5056c3bf25c52712ba559c8e99cffcc123ca","subject":"Update cross jobs to be explicit about which branch to checkout","status":"ABANDONED","created":"2024-05-01 17:45:50.000000000","updated":"2024-05-02 15:00:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb3638012c4fa5f500421c8163d0547e64eb102e","_number":917870,"virtual_id_number":917870,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-01 19:07:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 19:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c24119b6b926f6b735b94daceea09bb74e11507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-01 17:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abdd05022b3e163cfe0787b17d1efc8c68973733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 19:07:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2d59270d55436a8bab6b7719d6295a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f14bcf96fe4e5a92d5aaa0a412b0fa : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/015c7f1217d24bfcb1932f08c43a3086 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe2b597d1d184e5c91369fcbf73b0ea7 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e809790dd7b347818748c93e72fb25ca : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fb365d7a5e64a69b0f72422129fbb5b : SUCCESS in 3m 08s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c05f07712ca54d4b915cb47b14c9b896 : SUCCESS in 7m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/00ea702903e04a3088377955d3c8becf : SUCCESS in 4m 40s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3f5852b8dfd24e9599f790fc346a6951 : SUCCESS in 5m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c32e5e9cbe0d4734bfaa04c60b79029e : SUCCESS in 8m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b9bae9dff22d412ebb8895166e4f4ad6 : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7403ac63fce342f7a7fad2df7fdbe173 : SUCCESS in 5m 12s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f0a9e47bcb8f4dc3a093b28f6c0cbc20 : SUCCESS in 7m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/89909c425ec5423bb9c31f424b609844 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a8788d03e864c569ceb7f1f9c76c178 : SUCCESS in 5m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49a7efbbe1224104871e80afeef8ca12 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0ed1f9a1c52e43e4a174c029a173c4b2 : SUCCESS in 11m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a23b8393663e45d88d6cd4ee56521633 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/035bd2bd8fb04d68a5257b46c367a61b : SUCCESS in 7m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9c18e2c7eeb74466b893ff05b94f4acb : SUCCESS in 5m 30s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/19035e676e6d4c28bac8f899c040d6b6 : SUCCESS in 8m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bba40fdfb01c4316b72bcde0cd59e148 : SUCCESS in 21m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/546ed0db27ec4c5ea53be48b7817e640 : SUCCESS in 12m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7adcf887effd484bb073a4695e61b289 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7cca1a7fd4e14ddda15102c8488bdf72 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a21b20fe06c743de82933a1456adf7ad : SUCCESS in 19m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/021f3cb0d8a049ff862eb0e46f2bebc8 : SUCCESS in 5m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fab458931fa74871804cf83d6254dfbb : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac6bc718169846a6b55d7c5e8723523e : SUCCESS in 6m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a477b0e838724eab91b89869f583150f : SUCCESS in 11m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdeb4775bf9e475daf558a6edba8b150 : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fb3638012c4fa5f500421c8163d0547e64eb102e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-02 15:00:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"062c1b149749e8b62f5b35f203c08446ec3e2291","revisions":{"062c1b149749e8b62f5b35f203c08446ec3e2291":{"kind":"REWORK","_number":1,"created":"2024-05-01 17:45:50.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/70/917870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/917870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/917870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/917870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/917870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/917870/1"}}},"commit":{"parents":[{"commit":"9c379d80fceba448eebe6c3121184435b49c48ec","subject":"update constraint for horizon to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c379d80fceba448eebe6c3121184435b49c48ec"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-01 17:39:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-01 17:39:48.000000000","tz":-300},"subject":"Update cross jobs to be explicit about which branch to checkout","message":"Update cross jobs to be explicit about which branch to checkout\n\nAs some projects create unmaintained/zed branches and remove the\nstable/zed branches it becomes slightly difficult to gate on these\nprojects.  At the moment there isn\u0027t a good, documented, way to do\nbranch fallback so as a project removes stable/zed we end up cross\ngating against master which isn\u0027t the intent.\n\nThis change switches from the required-projects from a simple list of\nstrings to list of dictionaries.  Each item contains an explict branch\nto checkout for the gated project.  This is correct at the time of\nwriting but may change.  If a project has both stable/zed and\nunmaintained/zed I selected the latter.\n\nAlso an unrelated style change for the files list in a couple of jobs so\nthat the file is self consistent.\n\nChange-Id: I61af5056c3bf25c52712ba559c8e99cffcc123ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/062c1b149749e8b62f5b35f203c08446ec3e2291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/062c1b149749e8b62f5b35f203c08446ec3e2291"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
