)]}'
{"id":"openstack%2Frequirements~918733","triplet_id":"openstack%2Frequirements~stable%2F2023.2~Idfe615ce9b08fa7d051a4a28ab6c0ec9adc9837a","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"Idfe615ce9b08fa7d051a4a28ab6c0ec9adc9837a","subject":"update constraint for sushy to new release 4.5.2","status":"MERGED","created":"2024-05-09 13:13:41.000000000","updated":"2024-05-22 18:42:56.000000000","submitted":"2024-05-22 18:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918733-new-release","meta_rev_id":"bcb07a4d56a03648278520ff44c2d8ac3e33488d","_number":918733,"virtual_id_number":918733,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-22 18:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-09 13:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-22 15:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-22 15:57:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 13:19:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-09 14:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-22 15:57:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"144b36fb2af1df9023eb5adce49ec4587efce43b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-09 13:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b9b80d64884fd4b0a75e869e3a7ddc3e0849c55","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-09 13:19:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"245e9886d9749703295f03219d96ecb8208757d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 14:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5555f11106447de8513db0c5418b189\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67a1961c2fe49be936ec8ef67b2624c : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bca4c87c230248cba9743a1d81dafde5 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/879aeea890664a17aa071e171054ee16 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b23d129f4a2340b781499ab3245b2642 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c121660a38ff42f1a1e919ea567af357 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4cbc2c9fb7f44d78e68854c07be23f5 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb6e264e995d44fabbec03ff0e12bae4 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4ac5baec1464363adff277d60aaa526 : SUCCESS in 10m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/52d1983fca6443cdaf19a4aa2cd7d537 : SUCCESS in 17m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b825161f3d5342e389380e8775c298fe : SUCCESS in 13m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/30932c3d084d4c9583271050312ae777 : SUCCESS in 11m 54s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f308aa729c29444dac7764b2988a5875 : SUCCESS in 7m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f3a5e9243195438bbd406c93e8e4cea4 : SUCCESS in 4m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4d3a8434dfaa4a3c8835bdb59bd679b3 : SUCCESS in 4m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ce3949a5c7ca401d920d7b82c3c6230f : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f71781cf48214a9c82a07649bb381b0b : SUCCESS in 6m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/07de5424775f449d9e4268ce76054818 : SUCCESS in 6m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2a0fbaedc6e641f1bb0dc790b3ac44e8 : SUCCESS in 7m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/574ee0164e404b738a022af369ea11c4 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e175f56a76ed45a5976804453b9944db : SUCCESS in 8m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fadb132ea15347eca1e2ff793333fb9e : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/43587045c2ec47e0baf26ceeb13babf3 : SUCCESS in 13m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/153428e8da164b5f8ba7023caf64f9b9 : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b403d716cd4247a7972d15f8dc1986c6 : SUCCESS in 12m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f4dc3be0ad97488da3853e3321047264 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c7bcd0aa6af647188193edc6f5ebc920 : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7472e5db562b4c7d93ffdbbf14410c91 : SUCCESS in 28m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cb0fef9b43e441e6a0750633f1ff7f7b : SUCCESS in 12m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/834bfda42c6742369cca7e4be3b538d6 : SUCCESS in 12m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/79b3b7c3d66f4ef3b4085f56303c1384 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d488e8fd1a6c4709a987bb189129950a : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20aac9b6a0e74167af663677366928be : SUCCESS in 24m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5efc57fb44d74287b6e28d4231b953eb : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/10ca9a19e9514c66b3f3161395859b3a : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddb2c305a5a647a7998682a7e3bca1f2 : SUCCESS in 6m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f379b614a9524afd8e56cd10b5dd7b2b : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81114e79302f438aa68496894c1c8996 : SUCCESS in 45m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31526173e5b843e085d035236f731c59 : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9602de8737e440dc91a4b1f46557e40437baa4d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-22 15:57:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"828e183c7d7e7c44e080a5fbb06933e5e80ab4b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 15:59:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6ad11a33a26700875b2d566fc79fa812075c970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:42:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eea622b11d64c07b095e158658047dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2098adc94d844201abeff99fa2b4b22f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef3c77cf12c4f578bc5b4696c773ada : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99f4d9cc43444b4991bd7e7b50369ab0 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52f97f34e0a64944a21bf2570a61a947 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f2b51c198ef49b098bf3508489b8c3b : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d6baf3419014b6caa89e30da3f86afb : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fa0b3aeadac4b68876a85aec648510a : SUCCESS in 9m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f8b72cabd904726a6878f18df4bf09d : SUCCESS in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/faa773fe4bfa44ad94c406e18e489937 : SUCCESS in 10m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/97df8f1e688141ec8af9c7e4b33cb741 : SUCCESS in 12m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d397d0a8e4a94aeea44bf0aa2066f9e9 : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6dbd70552e3d495890fc35d9550fc29d : SUCCESS in 7m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ff2b87f7c7e242d6bd0b0dca6c503e86 : SUCCESS in 8m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/78d71cf80b6b48978ed06f7e3d2862f1 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01a10de3ea374f089a2bfa83c64649d3 : SUCCESS in 8m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/81ca0e8eda8942549a13d1c8b05fafc1 : SUCCESS in 6m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7b9355ec1ee64c1cbf4711854374f00f : SUCCESS in 12m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ecbcb900af3a46c8908fac06764aa94c : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8ec4b8ba1c08439cb3429ec86cb0c596 : SUCCESS in 9m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7abdf98394c141fd8a1e18c077661c9d : SUCCESS in 19m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/34c3e4e1b5414758849312dc3f7c5789 : SUCCESS in 11m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/afa40cf83f2f464e89ea94f8a6c99afe : SUCCESS in 13m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e7ece60eafc2495aa03d19bdd6028bf9 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67b4b4dc59fc4b79851d840f7fae4f31 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/802cd75ff5174d6ab150883b443dc951 : SUCCESS in 27m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/229f76f4fe9740df9add990f9c88c1cb : SUCCESS in 7m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/82f1dbaa34de4a789717317a8be9a837 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6afb046d451b45e59ccbc46115986b91 : SUCCESS in 7m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ff257277c8fc4da1bcb0361fc83dbb5e : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc98c507532a44e8a185637ff71af6b6 : SUCCESS in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/211c178b968b40df94a91d733243f0c3 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bcb07a4d56a03648278520ff44c2d8ac3e33488d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:42:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cc454b4496dce640893a0b25a654d91ca3a8542","revisions":{"2cc454b4496dce640893a0b25a654d91ca3a8542":{"kind":"REWORK","_number":1,"created":"2024-05-09 13:13:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/918733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/918733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/918733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/918733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/918733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/918733/1"}}},"commit":{"parents":[{"commit":"8a1f278a535c2bf665521fa93f40177ded111750","subject":"update constraint for oslo.log to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a1f278a535c2bf665521fa93f40177ded111750"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-09 13:13:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-09 13:13:40.000000000","tz":0},"subject":"update constraint for sushy to new release 4.5.2","message":"update constraint for sushy to new release 4.5.2\n\nmeta: version: 4.5.2\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ifa4b8bee01ee632e06e8b135c2e18376f87601e2\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idfe615ce9b08fa7d051a4a28ab6c0ec9adc9837a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cc454b4496dce640893a0b25a654d91ca3a8542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cc454b4496dce640893a0b25a654d91ca3a8542"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
