)]}'
{"id":"openstack%2Frequirements~919648","triplet_id":"openstack%2Frequirements~master~I7a3f58df1b00cc4cc5aa2270cf4160764635dde4","project":"openstack/requirements","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-02-18 13:18:01.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2025-02-18 13:18:01.000000000","reason":"Change was abandoned"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-18 13:18:01.000000000","reason":"Change was abandoned"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-02-18 13:18:01.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-15 22:06:42.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-02-18 13:18:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7a3f58df1b00cc4cc5aa2270cf4160764635dde4","subject":"DO NOT MERGE - for testing neutron cover job only!","status":"ABANDONED","created":"2024-05-15 13:20:35.000000000","updated":"2025-02-18 13:18:01.000000000","total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"revert_of":879743,"meta_rev_id":"02ccd5f0b8d72b80df6aae4fae7ded35ee78b186","_number":919648,"virtual_id_number":919648,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-28 17:37:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":-1,"date":"2024-05-28 16:15:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"state":"CC"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"CC"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-05-15 13:20:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 14:30:19.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-05-15 18:16:54.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"294cc942353f6cf7f8b3a29c4833794b0d173a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-15 13:20:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc31889eac9568e7b0d0d94eba6b266065fb249","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-15 13:22:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5c1f3b3cbcd9f0a5f4976606e27ca51cc3be34df","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-15 13:48:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35763478d16880801f5b8f37f88e45f2a1d3b0d2","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-15 13:51:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fd6ecdf3826fca6dfd52925630a7804c5d439f1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-15 13:59:58.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92e6ebd6311bb5363861f52d3b78b12b95bfcc77","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-15 14:30:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30a672eaeb20f54e403303e9727ef52b93a20dae","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-15 14:31:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b62711ca40c150cc9e2b57f48ba93651c9c88564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 15:10:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d08809cc0b244c19e9ba5af92901503\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6d0270ff6042528a9b85c76ec2d5f5 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914f9ee665f84207b9d73a32b7adbff6 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/141a22d449a24b91b1a1829150c5b7c1 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40cf21715f784234a0d93f2b26aff9d6 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddef762cf43a44c49f8acf6318b452a1 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34dcb2e2d1ca4e059a22f88892fd56eb : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aff92df16b5649a19e974aed6412f985 : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f6fd36215684156a1b130a4f3bb6e9c : SUCCESS in 7m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d4e9f5e9987248baa45181d06170bc78 : SUCCESS in 21m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ca564a782334f949f1879e1aa50224e : SUCCESS in 9m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/447fac45802b41a2ad91fa377bab3da2 : SUCCESS in 22m 12s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/84a5978e45b0453a82a0c88ca895dfad : SUCCESS in 5m 23s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e656d625e7444ae1a25a7ed2c279ac3c : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/161c4d0135c84f47b22546881237bff7 : SUCCESS in 4m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9bcccdf54f114ad7b2b0d4794283dc3a : SUCCESS in 11m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a84daefc536f48d9a206e1981799977f : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/71d77f6ebcab4daa8ba9a1bd2b794f2e : SUCCESS in 4m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/37d9d0a99782471c8699764887b52ac0 : SUCCESS in 6m 56s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/822b7fd8b1d749d19142ec72a0ef52e7 : SUCCESS in 13m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79c94dc446ec4b05a3ea98b06783c561 : SUCCESS in 6m 44s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0538ec188c654d69a1775c87dc98b5bd : SUCCESS in 10m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fd70c8a1301146239c3a3a889ea0866b : SUCCESS in 9m 41s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e805842b85844a7bb9bfd3c8a4986395 : SUCCESS in 5m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/feb25d22552e4de9a6a7bb10f3044311 : SUCCESS in 8m 43s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ac28d16ef9aa4994b6b72725a5a82f88 : SUCCESS in 4m 52s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a0703abefe88462c90cc607ff0352f9a : SUCCESS in 4m 13s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ec229ff1101d4f058adf106151ae3fa8 : SUCCESS in 13m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/92a75d9ffddc4001a8c72f0ef112f322 : SUCCESS in 21m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e9a413e606784de9be93fccd5f0af6c5 : SUCCESS in 5m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2ab929f135f749bab4ae2b6046fc4f2c : SUCCESS in 9m 17s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9e1754f102ca461cbd3579fff9dc196e : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38e156972d94432dbde260008ba84914 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ce718f346dd47faace7f9b9bcb16bf4 : SUCCESS in 20m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e5d4912be44f46cbbfbd36c6bd261145 : SUCCESS in 4m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/dce3e0a6efb64513a6be1e9e1cf7699d : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2c80e9ea1c540c9b7082eecb1c2cc99 : SUCCESS in 5m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d7327aac72804af0ac3babe50f2a463a : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f93aeb0447f94c4490b56597cbf0dd92 : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af4ed3fa43ab4743b901a3ebff9b0e63 : SUCCESS in 1h 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"64672c06a6a09b9c11d2ee0a3af9358eac10e9ec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-05-15 18:16:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"223c5d7ee69171094eb5aa141896eb4f662fa6bc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-15 18:38:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed4fb43f5e8c1bcdf871b36191047d425aa70762","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-15 18:39:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"263903bf62d6cbb3077413ff3c75a60537ebfb75","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-15 22:06:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"116c601e9cf9f60a5fa3af90141fc69f85885c66","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-16 04:31:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2101cd4c9949ed8e7971ce18e9d8482381d1f84b","tag":"autogenerated:gerrit:restore","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 17:15:10.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"1c3d36bce5e279f8d049e325928c4cb6fb8fa255","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 17:15:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a501d7e43e0f2d969d411087ff151590770c0c10","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-23 17:18:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1a7b7e67d746834de040aaf42dfa6ae018d9c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 18:33:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a43e88556dfd4188a984582762bf5aa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7eae4cd0d074f6ab574562ade4e02c8 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518a2320d85e487db2eda347912b9ea9 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f42f922cda984e8093a98f885fecca22 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d69a7d01538041d78871ce244507bf36 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4e1a5666c634024943117867c687cae : FAILURE in 7m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e671b94040fe45a6b9eca3a6d73e0692 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e605d486e16443d38143461c1bc3abac : SUCCESS in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a649886ff02a49c5a01db7e977efaffc : SUCCESS in 9m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9d34fe190184c149b238be9287e6ae6 : SUCCESS in 17m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4c17a69b59db4f9782e27253acebc1da : SUCCESS in 10m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8ca13b543d234585a11f25c4cce8a745 : SUCCESS in 23m 43s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3360d09a7e5e4217aa09f50650d6ff79 : SUCCESS in 5m 02s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d45ce5f55c4945629b8bc33f815885e8 : SUCCESS in 4m 31s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/db83c42adc384c64ba1238a57545b292 : SUCCESS in 4m 45s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/aafa199052564874b1574acdde33b5ff : SUCCESS in 10m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/820574fc8af542ffb4168d491358f181 : SUCCESS in 4m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8ff43b9cbb454028b4e75405ab22b3ee : SUCCESS in 5m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9744a091d54c4ff384784f638d926461 : SUCCESS in 7m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/12c184ab909f42fc808e928c0d6d92dd : SUCCESS in 13m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79687785b7c54875a0d044b9f31f05c0 : SUCCESS in 8m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bd0809b5fd504491b5e58569548c6b43 : SUCCESS in 13m 52s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6b615f53a62e42509536d57418789f61 : SUCCESS in 11m 22s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/750d6ca8a87c454e968f43bb50ed964b : SUCCESS in 4m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1590faaebcef4cb983e5d45b99e0316a : SUCCESS in 8m 21s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/85641d13083442f6a578f47f111ac5d8 : SUCCESS in 6m 15s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2125b0e1aed44aa29eb275bf85bc7229 : SUCCESS in 5m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4569c42522a54c59b9ba71643d66b941 : SUCCESS in 12m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2d974902be5447ef81a7ecb5ae487b18 : SUCCESS in 24m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fc13a1d601e54054a6bc7198335cd6dc : SUCCESS in 6m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ff1024d32c794e31bd13a584c732bfe2 : SUCCESS in 12m 11s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/75970487014d46fb9d42fc29f0221675 : SUCCESS in 7m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91025e41436842eebbc4ddcd6114f3fe : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a49ac5c22db403397d5a13572512fc1 : SUCCESS in 24m 02s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/739b165fc43a4746bc76dbc5df5c0cec : SUCCESS in 5m 59s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/77c49880afd84f518f93eb256984a35f : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec9e53b63a5e4fd59a1e13360aef6c86 : SUCCESS in 5m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e97d13495f034ace9f7f31e9062d3118 : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b5b4e01146b487395a204b3dd64c8fb : SUCCESS in 1h 04m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/723a720fddaa49aa86b051af7d4d97ac : FAILURE in 1h 16m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9de1c5caf9a41769e417b0d94acfc83d01d6f3be","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-23 18:46:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3077088151a8322892214aac3a76952fa830eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-28 16:15:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b15c9f1b09b69d508a74ab8897f50f195888b2b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-28 16:15:33.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"c70b4fa19e2619229ed24e345048fc4615c628bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:37:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc17d49dd99643ce862089d899655188\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4725e6f60a1f44b4a6bef360894dd2f6 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f57ad4a56842de844a39f4fd55c97d : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e49cbda25c4ed5aaab3470e11e1614 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a429260c18cf4d5c9223a7798f14412b : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d235248d85e44049812d53b85c02b91 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f7bd640e72e428cb8688175c7fcc4fd : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23ac10c9e8ad426ab8c06117f4038084 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d075566e82c746f088cc5cc9096e56d5 : SUCCESS in 11m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d76ff8ad958c4a239222aad1b2cca120 : SUCCESS in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/56f7c427058b46dd89ccabf25e10aec0 : SUCCESS in 9m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/01b2ed352adb4bae9751011feeab232f : SUCCESS in 17m 12s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fd79397c22844331b21bf6f90a110ec7 : SUCCESS in 5m 03s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ac1781f3e6ed4b61a716ed284c8112d9 : SUCCESS in 5m 08s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/32989c14909b4da1afceadd287a4a84a : SUCCESS in 4m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bdeb109806bb487eaafdf214f3c76c1b : SUCCESS in 12m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9ee34d21fa074f71924ae4107318467e : SUCCESS in 5m 40s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4551af72d97642e9818b6f3e321b0232 : SUCCESS in 4m 50s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/102bd4d0653d4b3a9c0b2e04fb24eff3 : SUCCESS in 7m 21s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fdd897db02df4cdc8bbfab0a3025c767 : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/479c3e592d404d0b881f9687d9dc987e : SUCCESS in 7m 18s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0a832b4358aa452488c26b005f0eb05d : SUCCESS in 8m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fcc82a9c57304ce5a9c0efd587d7718a : SUCCESS in 10m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d834c439c26d425594f54a326a058036 : SUCCESS in 5m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/73a2e1b5eae54ac4875a8c7aab6930e3 : SUCCESS in 7m 28s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/020f93f1a0a74e628d4b58952a24a9e1 : SUCCESS in 4m 59s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f87443e56a324a3ea6e28be87263b7cd : SUCCESS in 6m 04s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e6ab1d56dd714e2d8b5afea328c03276 : SUCCESS in 8m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c6898c92344a47f6a0194f000ba9c767 : SUCCESS in 18m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b537df84c7b4588b99157706f48c36c : SUCCESS in 8m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/37517184f92345a2ae3d97d0f48d989d : SUCCESS in 12m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b99a5f14544e4356ab6deba9144fc4b8 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e2d3028352c4673af073d36823895f8 : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7ab40605b294f32b44c8c1a73ef63fc : SUCCESS in 23m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/36c411b99eba45a7bfc450a8e7c9a84c : SUCCESS in 5m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/55fbf4a24b044596a840280a3e016549 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e94913459b4f417c8f3afc08d29cfaf0 : SUCCESS in 6m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5377393988e74435b4894133235e3ea9 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e0164b501148b2870c0e48da944df4 : SUCCESS in 1h 14m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14a99e7594f14f29ad443ae6116c2ed8 : FAILURE in 1h 21m 12s","accounts_in_message":[],"_revision_number":3},{"id":"02ccd5f0b8d72b80df6aae4fae7ded35ee78b186","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-02-18 13:18:01.000000000","message":"Abandoned\n\nhandled in neutron cover job","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f5b50d6d4c7846ce49729033959420049767e46","revisions":{"52d23fdd264954145b3a37944375beecab7c4992":{"kind":"REWORK","_number":1,"created":"2024-05-15 13:20:35.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/48/919648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/919648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/919648/1"}}},"commit":{"parents":[{"commit":"92c8b4c52b3160df1db1768e8891fc19eb704424","subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92c8b4c52b3160df1db1768e8891fc19eb704424"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-05-15 13:20:35.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-05-15 13:20:35.000000000","tz":0},"subject":"Revert \"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\"","message":"Revert \"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\"\n\nThis reverts commit 92c8b4c52b3160df1db1768e8891fc19eb704424.\n\nReason for revert: Neutron fullstack jobs broken, investigating.\n\nChange-Id: I7a3f58df1b00cc4cc5aa2270cf4160764635dde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52d23fdd264954145b3a37944375beecab7c4992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52d23fdd264954145b3a37944375beecab7c4992"}]},"branch":"refs/heads/master"},"cbf7fd7d5477f90c7e065b70f5ef7258ec0b4dbe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-15 13:22:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/48/919648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/919648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/919648/2"}}},"commit":{"parents":[{"commit":"92c8b4c52b3160df1db1768e8891fc19eb704424","subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92c8b4c52b3160df1db1768e8891fc19eb704424"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-05-15 13:20:35.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-05-15 13:22:24.000000000","tz":0},"subject":"Revert \"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\"","message":"Revert \"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\"\n\nThis reverts commit 92c8b4c52b3160df1db1768e8891fc19eb704424.\n\nReason for revert: Neutron fullstack jobs broken, investigating.\n\nRelated-Bug: #2065779\nChange-Id: I7a3f58df1b00cc4cc5aa2270cf4160764635dde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbf7fd7d5477f90c7e065b70f5ef7258ec0b4dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbf7fd7d5477f90c7e065b70f5ef7258ec0b4dbe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2f5b50d6d4c7846ce49729033959420049767e46":{"kind":"REWORK","_number":3,"created":"2024-05-28 16:15:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/919648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/919648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/919648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/919648/3"}}},"commit":{"parents":[{"commit":"fd1228c681d745ec6ace3b620cb419cc087c0d31","subject":"Merge \"update constraint for os-brick to new release 6.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd1228c681d745ec6ace3b620cb419cc087c0d31"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-05-15 13:20:35.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-05-28 16:13:41.000000000","tz":-240},"subject":"DO NOT MERGE - for testing neutron cover job only!","message":"DO NOT MERGE - for testing neutron cover job only!\n\nRevert \"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\"\n\nThis reverts commit 92c8b4c52b3160df1db1768e8891fc19eb704424.\n\nReason for revert: Neutron cover jobs broken, investigating.\n\nRelated-Bug: #2065779\nChange-Id: I7a3f58df1b00cc4cc5aa2270cf4160764635dde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f5b50d6d4c7846ce49729033959420049767e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f5b50d6d4c7846ce49729033959420049767e46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
