)]}'
{"id":"openstack%2Frequirements~919786","triplet_id":"openstack%2Frequirements~master~Ie52f19c272f2968821372894c9a432fb337aeea3","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ie52f19c272f2968821372894c9a432fb337aeea3","subject":"DNM: test with capped pytest","status":"ABANDONED","created":"2024-05-16 04:57:15.000000000","updated":"2024-05-23 09:02:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e73f0e433a7ce06d10acaa6335046fdf2edd094a","_number":919786,"virtual_id_number":919786,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-16 06:07:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 06:07:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0be30e11b28c2055040795487f90bade11738f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-16 04:57:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb2f4148fc045cf7f30017badae1f5beba0dfa0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 06:07:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a78845c7f101448eae2fadadbbd57c98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b45f3102cd35423cbaca69ae7ea21296 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1be53201f0f4ee89e5efaef74c85fe4 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fd0c2eba8ea4d04a07f68be5fb8f8f8 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a15ea48906cf40d1b16f5406eac80a1e : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cc92d74a3204104b761adcd0a093603 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a692cb571134443a7a20187f2a1668d : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b515415154ac42b8a64b212d7448bdb9 : SUCCESS in 2m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/565ae76a1d56458ba3a1d919ee679599 : SUCCESS in 10m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c19aab5c236e4a9f876565f4fed99342 : SUCCESS in 17m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aba6e6e6b53e49f890f3ac3c5b922e74 : SUCCESS in 10m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0f8da815ad648b680dea825c086899a : SUCCESS in 17m 35s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/aa4a4128bc264621a16c101ffacccdb4 : SUCCESS in 5m 01s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/eb8ee7406bc3405babecfafd3f974508 : SUCCESS in 5m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/cb924317d8274d1ead226f740d626b55 : SUCCESS in 4m 07s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d1b553e535ef4dbcb84bbe5f21e3e95f : SUCCESS in 10m 25s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9f5302615b774c9984cd5ff774f9e163 : SUCCESS in 5m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/88d28164bd3547c7a6302d3d1a05a37d : SUCCESS in 5m 33s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5cc21096d33e4cbd8ea5b2f33c2564be : SUCCESS in 6m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2d1abea8de9d47d3a2789de4dfc73856 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9afb7f758e14460abe79eeffedb3bb8b : SUCCESS in 6m 41s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8204ec26a3644e21b3bf1640c6ca52d8 : SUCCESS in 9m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4d8e205f9ca8403fb2265efea4ddbefe : SUCCESS in 9m 14s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/df0d0059bc76487b814af9ca3329413b : SUCCESS in 3m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2d7531b96de1430e8ca08153a3ee4f50 : SUCCESS in 7m 13s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/844b5890de3b4b14a6ea475daefe270d : SUCCESS in 4m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/46010f9315774164b306f9e2bcbb40d6 : SUCCESS in 3m 44s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/5aa198f9574d491c8e3962595ae5b189 : SUCCESS in 9m 07s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8403b440a3554383994c3a29b4e40ea8 : SUCCESS in 26m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f52efd502fe4ca79e026fe08ce3bec0 : SUCCESS in 6m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/15343b2d7d2145a5b8f67c5644a731a1 : SUCCESS in 9m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/457918e71f72418c9887e2929fb6ca48 : SUCCESS in 3m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0186dbad59ef4d78977d55af54a9e355 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c86636fd37a54870a11dfbf1920eda38 : SUCCESS in 23m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0f7adb036ebc4e1b82b8261f409d4d10 : SUCCESS in 3m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/fc20f01c61524858aef10b553982de1a : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4465cf5a4b2a4fc399bce6b0a8579acc : SUCCESS in 5m 02s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ece916be130045ad99334741202ba939 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a82dadd03e94b61b6e87510d9ccff07 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e73f0e433a7ce06d10acaa6335046fdf2edd094a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 09:02:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c4def99714853aea60d3ef91d6ac9581a722757","revisions":{"1c4def99714853aea60d3ef91d6ac9581a722757":{"kind":"REWORK","_number":1,"created":"2024-05-16 04:57:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/86/919786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/919786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/919786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/919786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/919786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/919786/1"}}},"commit":{"parents":[{"commit":"3176334de5d6c04219b7172ef9d06d4ab9c22d7a","subject":"Fix for automated upper-constraints.txt update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3176334de5d6c04219b7172ef9d06d4ab9c22d7a"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-16 04:56:52.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-16 04:56:52.000000000","tz":120},"subject":"DNM: test with capped pytest","message":"DNM: test with capped pytest\n\nChange-Id: Ie52f19c272f2968821372894c9a432fb337aeea3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c4def99714853aea60d3ef91d6ac9581a722757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c4def99714853aea60d3ef91d6ac9581a722757"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
