)]}'
{"id":"openstack%2Frequirements~919839","triplet_id":"openstack%2Frequirements~master~I31e906d0c9189db08b78e07a291199bb32acb109","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31e906d0c9189db08b78e07a291199bb32acb109","subject":"update constraint for python-tackerclient to new release 2.1.0","status":"MERGED","created":"2024-05-16 13:15:44.000000000","updated":"2024-05-22 18:43:05.000000000","submitted":"2024-05-22 18:43:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919839-new-release","meta_rev_id":"6443ccb9474dd35f6d175dafb4d357be505e0fc6","_number":919839,"virtual_id_number":919839,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-22 18:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-05-22 16:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-05-22 16:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 14:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-22 16:04:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e09967e5588b16d212f399f3eddf67ed354028a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-16 13:15:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e48e06ce04bbd01367605b3618a6677516b05b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93ac7046a064edea10828a83ceffc0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d7b184c181419eb6db7e0cc6a88e32 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccffea08ef74dcfb1a25445c2de91bf : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d5791aff37747c0bc992e389d9f5d29 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd81fd0b81ba4a3db487fb3286b44874 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/241d3e1169c14301a5216803843bf46a : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31148112f2ef43de8edf9eec33fd06c0 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0129263509604a3492297f8ba2e3b2b4 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c650d2e08214473f9708c46a9cadfc11 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/58951d89063e4e5b8524c0e11b1633f4 : SUCCESS in 19m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6a22361eec884383ad2845c389402c0e : SUCCESS in 11m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5ca3f3e019314560a15c4e45f3bd2553 : SUCCESS in 24m 27s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/626f5c0e77c94ec483662534c677ceae : SUCCESS in 6m 20s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e41ab6cd616e4c708d0a30ea75101ee1 : SUCCESS in 4m 42s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1e05c9f8445f461a80fd556278da295e : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/33af259f3af042c581c6a5bd40a287c5 : SUCCESS in 9m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9a3a2f3633614b1dba170f0a6c7e4184 : SUCCESS in 5m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4c4231a6ce484c8e8d894cf0a5f7b500 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cd1ce469a1e84db1bd077607f39fb564 : SUCCESS in 7m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fcce81a9fcd744428d73472291bc5716 : SUCCESS in 16m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/565ebcd030e4403095f58f956bf9b820 : SUCCESS in 7m 17s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5c3db11d22504047bdb5309a378423c1 : SUCCESS in 12m 44s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/48c659fcd37248419e650d62aa01440d : SUCCESS in 10m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/684bf878fe3e4e0db0ddec02f6477643 : SUCCESS in 5m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6b01035be9eb441795c78e2e0b96a61e : SUCCESS in 10m 21s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/144ab2c0b2874b20ae3baef9d6a856d7 : SUCCESS in 5m 03s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/278ff2d5cd43451ea3f2172a83f4639c : SUCCESS in 7m 00s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/6a5e4ea70adf4dbcbab27fe872cc1af5 : SUCCESS in 10m 41s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3fa233a6629f44e7874bda6b30eb7c9a : SUCCESS in 24m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f7892c51befc4ec3a0c247e6867e3668 : SUCCESS in 9m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b17993f729a041e69825bfcf80a91d4d : SUCCESS in 11m 11s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9e93ad44fc694baea0c6abbef9beeef1 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6723552fa73b403b98c79d4e38cb6c5c : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b747ffdb33df4397a5f111677664748a : SUCCESS in 24m 05s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ba1cf3fc230f4131aee4d41f9af8093c : SUCCESS in 6m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7e09b9d35c6a411fa8be24370fb86f4a : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/635aecb7fca54f428ed46d76e3f11c91 : SUCCESS in 5m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2d0648cd20c24357819b24fe03e78ab2 : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53b7d413e12240da9ad3c3bf0c9b9158 : SUCCESS in 57m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/033dda1b13db4c258c21123cd55ad1ab : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ea33fcae836e6f779cc6e8d67ffc8720378a0374","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-22 16:04:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ae6c02b7b7188127065bc3af7c39dcf90f45330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 16:05:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dceb9b6a0079bd572ec92c69ddb482d35bc2aed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:43:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa7c53c9b124f41ad35b7060ac21e19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d60c6d432a41ac9d6efc53a4412f6f : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482b8f747f014c188e177a8b1a2ea553 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3934c2b1954f4990af040c781fcb7642 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3418dd1b0f2a47cfad9c6830e708c260 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ea74ed76f414585af093ebbe0bfd6c3 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0399110267d24075b79750c0a0f53993 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d54a3ec10b0348f59472e5d691481c38 : SUCCESS in 9m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a8f286297b364372a8211751fba2c098 : SUCCESS in 19m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/88eadb7249844539be316795a5922c06 : SUCCESS in 9m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/20cb1d335c274097b371566087017108 : SUCCESS in 21m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/127911e8d4f14c1481669b5dda21d7e2 : SUCCESS in 9m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f55ceb901be64eaeadd087c186205db1 : SUCCESS in 6m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/558e8728512d41a5864b7f3d2b8af6da : SUCCESS in 6m 17s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/78fce708a75b46bfacaa7cc729ed5f01 : SUCCESS in 8m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9050bd43b859445797dd7372baf89c81 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc026f9f39f2460c9599b86fb5c0c3ea : SUCCESS in 8m 53s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/50b299bbfaa04dbbb9f38d457a7853ea : SUCCESS in 13m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/364072d5fd37473aa518ffffa0b84d47 : SUCCESS in 10m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/304b20df2eb64f5d811cc4f77352c58a : SUCCESS in 4m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/49daffaa92d24ad289b5c17b7bb5acf5 : SUCCESS in 9m 25s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b88361ffbe46486a9a4eb4337e368b8d : SUCCESS in 24m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91e4e99f9e6445cf9bef1de05c641a85 : SUCCESS in 7m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/be5a549625c3409ba10bccbba2389545 : SUCCESS in 11m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/03b79dafb9134d4ca6c654b17677f612 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b64abf74f814e7e942307f5fcf2d69f : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e20e550a741484d883861163bc93d50 : SUCCESS in 20m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/756f5ad5c9754e8cbcd1379220a84f38 : SUCCESS in 4m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1421b1133ed84461b124bf2a0196211c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9efd9853b3c54e9fa07e5e380e55c8a0 : SUCCESS in 6m 54s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/630e1e0b74a54d1ebe6f6ade694757b5 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3698fff5d2ec42278e3ab66332718b8b : SUCCESS in 47m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f8526a769d44b509891e8e9da6a1d3c : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6443ccb9474dd35f6d175dafb4d357be505e0fc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:43:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27c1ef23c4c1706729ab368fbffbcc0b269c3bd0","revisions":{"27c1ef23c4c1706729ab368fbffbcc0b269c3bd0":{"kind":"REWORK","_number":1,"created":"2024-05-16 13:15:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/919839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/919839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/919839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/919839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/919839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/919839/1"}}},"commit":{"parents":[{"commit":"a8ef1881af9cdb048dc0c5e3f814aaed80efd7a3","subject":"Merge \"Bump testrepository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8ef1881af9cdb048dc0c5e3f814aaed80efd7a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-16 13:15:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-16 13:15:41.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 2.1.0","message":"update constraint for python-tackerclient to new release 2.1.0\n\nmeta: version: 2.1.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia14a2e1ddc500899204fcbee8c3f0cd42d102af9\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I31e906d0c9189db08b78e07a291199bb32acb109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27c1ef23c4c1706729ab368fbffbcc0b269c3bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27c1ef23c4c1706729ab368fbffbcc0b269c3bd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
